)]}'
{"id":"openstack%2Fneutron~794130","triplet_id":"openstack%2Fneutron~master~Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41","project":"openstack/neutron","branch":"master","topic":"bug/1930432","hashtags":[],"change_id":"Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41","subject":"[L2] no provisioning block for internal service port","status":"MERGED","created":"2021-06-02 05:30:00.000000000","updated":"2021-09-14 18:41:26.000000000","submitted":"2021-09-14 18:39:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":18,"has_review_started":true,"submission_id":"794130-bug/1930432","meta_rev_id":"ae13814ebe87c519b7db469097056580f0a7a900","_number":794130,"virtual_id_number":794130,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-14 18:39:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-07-20 11:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-09-06 12:01:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-09-14 16:22:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 06:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-02 06:34:29.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-02 07:14:59.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-06-02 07:25:51.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-06-02 07:38:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-06-02 07:41:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 13:45:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-06-08 02:07:16.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-22 07:49:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-07-07 14:02:56.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-09 03:19:04.000000000","updated_by":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"reviewer":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"state":"REVIEWER"}],"messages":[{"id":"baeab2ec1004ee168b64c095ceee9aa00544643c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-02 05:30:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7217dbf2d2c047fdab78ecdcee27fca446d8e6ce","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-02 05:30:51.000000000","message":"Patch Set 1: Workflow-1\n\nWait and see how zuul talks.","accounts_in_message":[],"_revision_number":1},{"id":"762e2c0e75868f56c99347a35f265e2cb754fc25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 06:14:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/578f0251f2f3412f9f0462744bbc2302 : SUCCESS in 43m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fbf59b337a8746a0afa39e24260d536a : SUCCESS in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ec0278b6b7657e2da534d8df52b4a7ad87b8b83","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-02 06:34:29.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/794130/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/794130/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"36db2eb99c70e2d866500d59bb5f3661fbe59a80","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-02 07:14:59.000000000","message":"Patch Set 1: Code-Review-1\n\nThe bug says \"provisioning_block may cause some timeout for resource setup\" - do you have examples/benchmarks for this case? Provisioning blocks look rather lightweight so I\u0027m not sure it\u0027s worth additional checks overall","accounts_in_message":[],"_revision_number":1},{"id":"634208de601a4f0a24049c61d1780b4152761566","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-02 07:23:28.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e The bug says \"provisioning_block may cause some timeout for resource setup\" - do you have examples/benchmarks for this case? Provisioning blocks look rather lightweight so I\u0027m not sure it\u0027s worth additional checks overall\n\nI\u0027ve looked at this bug [1], it can be a good example, one of the case shows that the router interface did not set to ACTIVE was because there was no provisioning complete action.\n\nIMO, it\u0027s not lightweight, it will insert a record to Database, and remove it after provisioning complete. \n\n[1] https://bugs.launchpad.net/neutron/+bug/1930401","accounts_in_message":[],"_revision_number":1},{"id":"6621f5a491a9698a397efaf8e361c0c06960451d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-02 07:25:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794130/1 : SUCCESS in 1h 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1da9de2ca90b0b91df979a44c8d9f5989bd53016","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-02 07:31:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fee8e57ee2bde65f1020c36dc6822d78c4bfe41","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-02 07:35:21.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e The bug says \"provisioning_block may cause some timeout for resource setup\" - do you have examples/benchmarks for this case? Provisioning blocks look rather lightweight so I\u0027m not sure it\u0027s worth additional checks overall\n\u003e \n\u003e I\u0027ve looked at this bug [1], it can be a good example, one of the case shows that the router interface did not set to ACTIVE was because there was no provisioning complete action.\n\u003e \n\u003e IMO, it\u0027s not lightweight, it will insert a record to Database, and remove it after provisioning complete. \n\u003e \n\u003e [1] https://bugs.launchpad.net/neutron/+bug/1930401\n\nYeah, I forgot that PBs are stored in DB. Still I doubt this is the reason for port not ACTIVE issue. Seems it\u0027s an agent being slow: did you see agent sent port ACTIVE in time and it\u0027s just server who\u0027s slow to set it to ACTIVE in DB?","accounts_in_message":[],"_revision_number":1},{"id":"c734c0e75978a4873bb0a2c0ab98b6b62259d945","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-02 07:38:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"790a39bf5c119573597a571a428186f859a4ec4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 07:41:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/572026b033ca4d728971365bab6d1ab8 : SUCCESS in 58m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/71623bf49cd9474b997488fd353edeba : SUCCESS in 2h 01m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2f4ecb1de6fe49b28a838d0132dff2b4 : FAILURE in 1h 44m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fa0c5d24cd34583b7782eb32dc4ca0a : FAILURE in 24m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/142d9cbf04204919a9a41c527ca82432 : SUCCESS in 33m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e9b5587d37474cb024ce916577ff95 : SUCCESS in 13m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a7fc8b1d8a1481395cda94f0af91de8 : SUCCESS in 35m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51ca11dd13464173984e0ee576228c0e : SUCCESS in 22m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/375d172f5b7e4d0c96d2be86c38617a9 : SUCCESS in 26m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198388b331984884864b62df2635a1a9 : SUCCESS in 11m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/26f94f65a4f14faeb064495649dabe25 : SUCCESS in 49m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ba0c9104fc144c298e73bb10a754e79 : SUCCESS in 1h 22m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8af98afe532249dc920eac91d1bf7f19 : FAILURE in 1h 34m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fab9a6da5c09434f97f7cb0a74000905 : SUCCESS in 1h 01m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/020cba317f5245aeba8e72bbb74facdb : SUCCESS in 1h 00m 34s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/27fcda938f304ea68233a6c434f1b2c5 : SUCCESS in 1h 26m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f7bb7f4ad3d04cbb80d4ebb6f8d511f2 : FAILURE in 27m 23s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a5ba55a07e3492fa6544bea120398c7 : SUCCESS in 1h 39m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e38ebdf40a9b47e4a40da4bf1adc6b46 : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1a6cba3f2dddd35ae0896e887b15a1c6eb71931b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-02 07:54:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bebac14ff1215bdd63fb53c2cb0faab655cee0a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-02 07:58:38.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e The bug says \"provisioning_block may cause some timeout for resource setup\" - do you have examples/benchmarks for this case? Provisioning blocks look rather lightweight so I\u0027m not sure it\u0027s worth additional checks overall\n\u003e \u003e \n\u003e \u003e I\u0027ve looked at this bug [1], it can be a good example, one of the case shows that the router interface did not set to ACTIVE was because there was no provisioning complete action.\n\u003e \u003e \n\u003e \u003e IMO, it\u0027s not lightweight, it will insert a record to Database, and remove it after provisioning complete. \n\u003e \u003e \n\u003e \u003e [1] https://bugs.launchpad.net/neutron/+bug/1930401\n\u003e \n\u003e Yeah, I forgot that PBs are stored in DB. Still I doubt this is the reason for port not ACTIVE issue. Seems it\u0027s an agent being slow: did you see agent sent port ACTIVE in time and it\u0027s just server who\u0027s slow to set it to ACTIVE in DB?\n\nSo I checked logs of one failure from https://bugs.launchpad.net/neutron/+bug/1930401:\n - port 45ed2855-2935-406e-87a8-83ac7cc08a3b failed to reach ACTIVE status failed at 08.20.28\n - L2 agent provisioned port in time: \n   - 2021-05-27 08:19:36.646 138296 DEBUG neutron.db.provisioning_blocks [req-c2b1dbf8-4d37-4ae0-8006-e0cc28097338 - - - - -] Provisioning for port 45ed2855-2935-406e-87a8-83ac7cc08a3b completed by entity L2. provisioning_complete\n - provisioning was not completed by DHCP agent, looking at DHCP agent logs I see they stop at 08:19:28.266:\n   - https://b87ba208d44b7f1356ad-f27c11edabee52a7804784593cf2712d.ssl.cf5.rackcdn.com/791365/5/check/neutron-fullstack-with-uwsgi/634ccb1/controller/logs/dsvm-fullstack-logs/TestLegacyL3Agent.test_external_subnet_changed/neutron-dhcp-agent--2021-05-27--08-19-12-515519_log.txt\n\nSo the real issue is that dhcp agent was hanging or stopped","accounts_in_message":[],"_revision_number":1},{"id":"e2ccf3ca79bd5ed70f474444df65476516ec0638","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-02 08:59:36.000000000","message":"Patch Set 1: -Workflow\n\n\u003e Patch Set 1: Workflow-1\n\u003e \n\u003e Wait and see how zuul talks.\n\nOK, not bad, now we can think about this, if it is really needed for neutron services port?","accounts_in_message":[],"_revision_number":1},{"id":"b0e8e75930d6e43b8ea6473d57d45cc6431ce97e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-02 09:02:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b510d4258226a798dc7d95e5667d812e3a1acdda","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-02 09:07:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9aa761bc10ab9d498e4bd38e9cb0e386b7ab4c67","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-02 09:30:00.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e (1 comment)\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e The bug says \"provisioning_block may cause some timeout for resource setup\" - do you have examples/benchmarks for this case? Provisioning blocks look rather lightweight so I\u0027m not sure it\u0027s worth additional checks overall\n\u003e \u003e \u003e \n\u003e \u003e \u003e I\u0027ve looked at this bug [1], it can be a good example, one of the case shows that the router interface did not set to ACTIVE was because there was no provisioning complete action.\n\u003e \u003e \u003e \n\u003e \u003e \u003e IMO, it\u0027s not lightweight, it will insert a record to Database, and remove it after provisioning complete. \n\u003e \u003e \u003e \n\u003e \u003e \u003e [1] https://bugs.launchpad.net/neutron/+bug/1930401\n\u003e \u003e \n\u003e \u003e Yeah, I forgot that PBs are stored in DB. Still I doubt this is the reason for port not ACTIVE issue. Seems it\u0027s an agent being slow: did you see agent sent port ACTIVE in time and it\u0027s just server who\u0027s slow to set it to ACTIVE in DB?\n\u003e \n\u003e So I checked logs of one failure from https://bugs.launchpad.net/neutron/+bug/1930401:\n\u003e  - port 45ed2855-2935-406e-87a8-83ac7cc08a3b failed to reach ACTIVE status failed at 08.20.28\n\u003e  - L2 agent provisioned port in time: \n\u003e    - 2021-05-27 08:19:36.646 138296 DEBUG neutron.db.provisioning_blocks [req-c2b1dbf8-4d37-4ae0-8006-e0cc28097338 - - - - -] Provisioning for port 45ed2855-2935-406e-87a8-83ac7cc08a3b completed by entity L2. provisioning_complete\n\u003e  - provisioning was not completed by DHCP agent, looking at DHCP agent logs I see they stop at 08:19:28.266:\n\u003e    - https://b87ba208d44b7f1356ad-f27c11edabee52a7804784593cf2712d.ssl.cf5.rackcdn.com/791365/5/check/neutron-fullstack-with-uwsgi/634ccb1/controller/logs/dsvm-fullstack-logs/TestLegacyL3Agent.test_external_subnet_changed/neutron-dhcp-agent--2021-05-27--08-19-12-515519_log.txt\n\u003e \n\u003e So the real issue is that dhcp agent was hanging or stopped\n\nI think it is an environment issue, not only the DHCP provisioning_block is not released, but also L2.\n\nThis log\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_73b/793141/2/check/neutron-fullstack-with-uwsgi/73b08ae/controller/logs/dsvm-fullstack-logs/TestLegacyL3Agent.test_external_subnet_changed/neutron-server--2021-05-27--10-49-03-097753_log.txt\n\nshows the following blocks:\nTransition to ACTIVE for port object 129ac768-a987-4a6a-bda9-d5c08ca38b42 will not be triggered until provisioned by entity L2.\nTransition to ACTIVE for port object 129ac768-a987-4a6a-bda9-d5c08ca38b42 will not be triggered until provisioned by entity DHCP.\nTransition to ACTIVE for port object 47a1f7fb-b24f-4359-a262-34e303c3b95b will not be triggered until provisioned by entity DHCP.\nTransition to ACTIVE for port object 47a1f7fb-b24f-4359-a262-34e303c3b95b will not be triggered until provisioned by entity L2.\nTransition to ACTIVE for port object a5a86928-6476-43b9-bf91-8ac1bfe7d415 will not be triggered until provisioned by entity L2.\nTransition to ACTIVE for port object a5a86928-6476-43b9-bf91-8ac1bfe7d415 will not be triggered until provisioned by entity DHCP.\nTransition to ACTIVE for port object 0eafe3a5-e311-4289-a9f7-bdafe8092e9f will not be triggered until provisioned by entity DHCP. \nTransition to ACTIVE for port object 0eafe3a5-e311-4289-a9f7-bdafe8092e9f will not be triggered until provisioned by entity L2.\n\nBut for complete action, only it only has:\nProvisioning for port 47a1f7fb-b24f-4359-a262-34e303c3b95b completed by entity L2. provisioning_complete\nProvisioning for port 0eafe3a5-e311-4289-a9f7-bdafe8092e9f completed by entity L2. provisioning_complete\n\nThese ports\u0027 attributes are:\nReceived new resource Port: Port(admin_state_up\u003dTrue,allowed_address_pairs\u003d[],binding_levels\u003d[],bindings\u003d[PortBinding],created_at\u003d2021-05-27T10:49:24Z,data_plane_status\u003dNone,description\u003d\u0027\u0027,device_id\u003d\u0027dhcp4b6768e7-5d55-5150-a44a-086c7cfbc0d3-b7ffe55d-22c2-436f-8544-3ad9d8800525\u0027,device_owner\u003d\u0027network:dhcp\u0027,device_profile\u003d\u003c?\u003e,dhcp_options\u003d[],distributed_bindings\u003d[],dns\u003dNone,fixed_ips\u003d[IPAllocation],id\u003d129ac768-a987-4a6a-bda9-d5c08ca38b42,mac_address\u003dfa:16:3e:fd:94:de,name\u003d\u0027\u0027,network_id\u003db7ffe55d-22c2-436f-8544-3ad9d8800525,numa_affinity_policy\u003d\u003c?\u003e,project_id\u003d\u00277d2e79f4-153e-4c55-8570-18f0e895f73a\u0027,qos_network_policy_id\u003dNone,qos_policy_id\u003dNone,revision_number\u003d1,security\u003dPortSecurity(129ac768-a987-4a6a-bda9-d5c08ca38b42),security_group_ids\u003dset([]),status\u003d\u0027DOWN\u0027,updated_at\u003d2021-05-27T10:49:25Z) record_resource_update \nReceived new resource Port: Port(admin_state_up\u003dTrue,allowed_address_pairs\u003d[],binding_levels\u003d[],bindings\u003d[PortBinding],created_at\u003d2021-05-27T10:49:25Z,data_plane_status\u003dNone,description\u003d\u0027\u0027,device_id\u003d\u00275d37f0e2-6fc3-447d-bc60-ba8465184bde\u0027,device_owner\u003d\u0027network:router_gateway\u0027,device_profile\u003d\u003c?\u003e,dhcp_options\u003d[],distributed_bindings\u003d[],dns\u003dNone,fixed_ips\u003d[IPAllocation],id\u003d47a1f7fb-b24f-4359-a262-34e303c3b95b,mac_address\u003dfa:16:3e:70:f6:b0,name\u003d\u0027\u0027,network_id\u003db7ffe55d-22c2-436f-8544-3ad9d8800525,numa_affinity_policy\u003d\u003c?\u003e,project_id\u003d\u0027\u0027,qos_network_policy_id\u003dNone,qos_policy_id\u003dNone,revision_number\u003d1,security\u003dPortSecurity(47a1f7fb-b24f-4359-a262-34e303c3b95b),security_group_ids\u003dset([]),status\u003d\u0027DOWN\u0027,updated_at\u003d2021-05-27T10:49:25Z) record_resource_update \nReceived new resource Port: Port(admin_state_up\u003dTrue,allowed_address_pairs\u003d[],binding_levels\u003d[],bindings\u003d[PortBinding],created_at\u003d2021-05-27T10:49:28Z,data_plane_status\u003dNone,description\u003d\u0027\u0027,device_id\u003d\u0027dhcp4b6768e7-5d55-5150-a44a-086c7cfbc0d3-314ddeac-6f67-42b3-b371-eb0175a22091\u0027,device_owner\u003d\u0027network:dhcp\u0027,device_profile\u003d\u003c?\u003e,dhcp_options\u003d[],distributed_bindings\u003d[],dns\u003dNone,fixed_ips\u003d[IPAllocation],id\u003da5a86928-6476-43b9-bf91-8ac1bfe7d415,mac_address\u003dfa:16:3e:78:fb:cb,name\u003d\u0027\u0027,network_id\u003d314ddeac-6f67-42b3-b371-eb0175a22091,numa_affinity_policy\u003d\u003c?\u003e,project_id\u003d\u00277d2e79f4-153e-4c55-8570-18f0e895f73a\u0027,qos_network_policy_id\u003dNone,qos_policy_id\u003dNone,revision_number\u003d1,security\u003dPortSecurity(a5a86928-6476-43b9-bf91-8ac1bfe7d415),security_group_ids\u003dset([]),status\u003d\u0027DOWN\u0027,updated_at\u003d2021-05-27T10:49:28Z) record_resource_update \nReceived new resource Port: Port(admin_state_up\u003dTrue,allowed_address_pairs\u003d[],binding_levels\u003d[],bindings\u003d[PortBinding],created_at\u003d2021-05-27T10:49:28Z,data_plane_status\u003dNone,description\u003d\u0027\u0027,device_id\u003d\u00275d37f0e2-6fc3-447d-bc60-ba8465184bde\u0027,device_owner\u003d\u0027network:router_interface\u0027,device_profile\u003d\u003c?\u003e,dhcp_options\u003d[],distributed_bindings\u003d[],dns\u003dNone,fixed_ips\u003d[IPAllocation],id\u003d0eafe3a5-e311-4289-a9f7-bdafe8092e9f,mac_address\u003dfa:16:3e:c5:fd:70,name\u003d\u0027\u0027,network_id\u003d314ddeac-6f67-42b3-b371-eb0175a22091,numa_affinity_policy\u003d\u003c?\u003e,project_id\u003d\u00277d2e79f4-153e-4c55-8570-18f0e895f73a\u0027,qos_network_policy_id\u003dNone,qos_policy_id\u003dNone,revision_number\u003d1,security\u003dPortSecurity(0eafe3a5-e311-4289-a9f7-bdafe8092e9f),security_group_ids\u003dset([]),status\u003d\u0027DOWN\u0027,updated_at\u003d2021-05-27T10:49:29Z) record_resource_update \n\nNone of them are compute port, so I may ask the question again:\n\"\"\"\nprovisioning_block should be added to Neutron internal service port? Or should not?\n\"\"\"\n\nFrom our cloud operators, they always say it\u0027s very annoying to inspect the DHCP agent issue for a large cloud with tremendous resources. So, locally, we had removed dhcp-agent. 😊 And for our community, we will have a dhcp extension to replace it.","accounts_in_message":[],"_revision_number":1},{"id":"0c7d469e91f11fa69bbaeacee21422ff5fa43159","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-02 09:59:07.000000000","message":"Patch Set 1:\n\nPlease check my comments in https://bugs.launchpad.net/neutron/+bug/1930401 - the underlying issue is not with provisioning blocks, but with privsep making agents hang","accounts_in_message":[],"_revision_number":1},{"id":"488660680fbfa4483017e33936d360f2dd2d55fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-03 07:50:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f74a7ee42c385e545b6e214e755db695ce5e3a91","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-03 08:35:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd4a679989d105fef82ca289e9b8e242cceb413d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 08:38:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f4c4284c11d4abb995eee3948fc4719 : SUCCESS in 35m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf0bd200a2864ebbabe6fcfc08482605 : SUCCESS in 42m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36f5b1c1501fb8db0e8a719daa0594cc2a9e36ed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-03 09:24:46.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/794130/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/794130/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7c11f33a341b56a10de74bf4ee5e8364c8d9546c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 09:27:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794130/2 : SUCCESS in 1h 34m 57s","accounts_in_message":[],"_revision_number":2},{"id":"9722608eb41d56d74ff0398d8ab053449eb39f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 10:14:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5c5b8a8c4af24491aa90c3e9c9e7143d : SUCCESS in 52m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0ac5f2bb39c042208d3c1dea61905a9b : SUCCESS in 1h 50m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/32e50f9871544b1393fa32987db0e5cf : SUCCESS in 1h 33m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b380cf116e04e048a6e29bc9a4e1a49 : SUCCESS in 1h 20m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cf8661158893414187f49bedc6f61510 : SUCCESS in 2h 13m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bcb32492a0049f8bd79b625975011a1 : SUCCESS in 35m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4eea5b224414fc6829032d75af84c87 : SUCCESS in 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ca84369dc648548a916302a153db8b : SUCCESS in 14m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3343140cce9a40ad8800d6b3ae41900d : SUCCESS in 27m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b962f4926874f358634826479e3a9a8 : SUCCESS in 20m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/200a28dea2e94cf7b0ac5344866cd077 : SUCCESS in 30m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/456f9a47206f41288f4cb4241d666cf5 : SUCCESS in 10m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9755a556e7f841b69e0cd5dace3ee71e : SUCCESS in 41m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1fd1c581dc8418e89b606762188b6bf : TIMED_OUT in 2h 15m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bc6c2e61ad724c69b1102e70818cb927 : FAILURE in 1h 26m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98b2043a041f48578c9c8da6ba8aaefe : SUCCESS in 52m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dee910a072e04263831545290faaeb18 : SUCCESS in 59m 24s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bab96698e2ae4d1799d45fec9b1c0522 : FAILURE in 1h 20m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a904d406c87f4117af4cf38788af34e4 : FAILURE in 24m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb650ccdd99e4e05a67ddd794fe3edbc : SUCCESS in 1h 33m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2904704d041f449fa1aab448cac6c118 : SUCCESS in 1h 04m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7a08489df51d46b4b72b09d0891bb3c4 : SUCCESS in 1h 07m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8dc35198dd674bd38267afbed8037501 : SUCCESS in 29m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/afbc8b06f9144d50aea5bb1a9b1935c0 : SUCCESS in 1h 25m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d471690bcfbb445cbcee0b283674c8ff : SUCCESS in 1h 18m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7921a02e98c312c503fdd3e515dce7f2b714c500","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 13:45:47.000000000","message":"Patch Set 2:\n\nI would put on hold this patch. The issue with the DHCP agent has been identified. There is an issue with privsep, when the daemon is spawned. As proposed in [1], a timeout when spawning the daemon could help. If we retry this command, we\u0027ll skip this problem.\n\nAnother question is why the daemon is not spawning correctly.\n\n[1]https://bugs.launchpad.net/neutron/+bug/1930401","accounts_in_message":[],"_revision_number":2},{"id":"bd9f7f91ed3f97d11615ca24fdc9a46e10a8b370","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 02:07:16.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/30/794130/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"6adaa990c8831ce8dd5483066141fd1c29e73f43","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-08 02:40:22.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e I would put on hold this patch. The issue with the DHCP agent has been identified. There is an issue with privsep, when the daemon is spawned. As proposed in [1], a timeout when spawning the daemon could help. If we retry this command, we\u0027ll skip this problem.\n\u003e \n\u003e Another question is why the daemon is not spawning correctly.\n\u003e \n\u003e [1]https://bugs.launchpad.net/neutron/+bug/1930401\n\nWhether this patch is merged or not should not be the real concern we have, the main problem here we should think about is \"if it is needed to set the provisioning block for internal service port\".","accounts_in_message":[],"_revision_number":2},{"id":"19d726cbb9d0d8c1ef63bd4a0fe859f87c9fbae1","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-08 02:43:20.000000000","message":"Patch Set 2:\n\nPlease think about the question: provisioning_block should be added to Neutron internal service port? Or should not?\n\nFrom the CI logs, there is an exterm case is Neutron sets DHCP provisioning_block to network:DHCP port, this is really not necessary.","accounts_in_message":[],"_revision_number":2},{"id":"3e00a07e7cc535782502dfe78fcf24307287ca71","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:13:15.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"dc41de57235714d81d21455969be22bb9fa9744b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-08 07:38:17.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e I would put on hold this patch. The issue with the DHCP agent has been identified. There is an issue with privsep, when the daemon is spawned. As proposed in [1], a timeout when spawning the daemon could help. If we retry this command, we\u0027ll skip this problem.\n\u003e \u003e \n\u003e \u003e Another question is why the daemon is not spawning correctly.\n\u003e \u003e \n\u003e \u003e [1]https://bugs.launchpad.net/neutron/+bug/1930401\n\u003e \n\u003e Whether this patch is merged or not should not be the real concern we have, the main problem here we should think about is \"if it is needed to set the provisioning block for internal service port\".\n\nIMO provisioning blocks for internal service ports are not needed and hence this patch makes sense. The reason to put it on hold is to not mask a more severe issue with privsep hanging - once it\u0027s fixed this patch is safe for merge","accounts_in_message":[],"_revision_number":2},{"id":"2e96c9b5aad9ca4b0e0a4d5bcfac4b0e29f50f23","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-08 07:38:48.000000000","message":"Patch Set 2:\n\nAlso please check my inline comment","accounts_in_message":[],"_revision_number":2},{"id":"66b3954629714e690d87a31d3ea2a6b27bd6def0","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-10 01:10:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"331d8c6d2542fc0020a63564e1b0fa7469088770","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-10 07:12:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49d662fc6e1b02d2303a0f6de6536b5d7f4ef8b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-15 01:08:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"920541c97d38fc19a27451c7df51640f4ffcbd4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 01:48:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f76e057ae60f4a72bdbdda7fc43c6281 : SUCCESS in 34m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46753e0701e04e1992aa792a3142ef39 : SUCCESS in 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad7e755a011be1c9476c62061791acdc53e3f64d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-15 03:00:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794130/3 : SUCCESS in 1h 51m 57s","accounts_in_message":[],"_revision_number":3},{"id":"724eb801b8b6d9c41b0d755d93d27e4871ab6d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 03:23:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/844dc2fc9b8742579f3ecbe6c2f28965 : SUCCESS in 40m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1c09258e21c94d4cb5b61f27f40a783d : SUCCESS in 1h 43m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/22e1c0f1ee54441cadc2e876c32f9929 : SUCCESS in 1h 52m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/72501be4fa224d93bc33e3ef65011c0e : SUCCESS in 1h 17m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/482922c91b724f3a937851118acd27fc : SUCCESS in 1h 41m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58c861476eb14a549f969b2a167db86c : SUCCESS in 33m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb88d364b4d7460884bd64619b25a84b : SUCCESS in 34m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b72db983ba443593c9e32657c56fe2 : SUCCESS in 13m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bd83c3c277f438a9547cb16833f73ab : SUCCESS in 36m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d70acec2aadb46f98b6344f1166ff748 : SUCCESS in 24m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36e3ed3b615b47fbb34a52831e253655 : SUCCESS in 25m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92cc7d2addf64b049b1ca4db20f77e4f : SUCCESS in 9m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aadfa3d9acad45f8930e00f5a71a1162 : SUCCESS in 46m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/23a34c8cbf5742f0aef00c70e75cf112 : TIMED_OUT in 2h 13m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b62fa876e24e48ad8223ef3695ac7a88 : SUCCESS in 1h 10m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccabe6a639f540948bde89d509a08f04 : SUCCESS in 57m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4eae31d9abaa4d42a82f62fac91e6bfd : SUCCESS in 1h 00m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/819d23ddf4c4491b8a54e46da9692c50 : SUCCESS in 1h 11m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4b6517b694444363a97764bb3f8ead09 : SUCCESS in 1h 24m 24s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e93c9beca58c4eb2bdde3b4a55bb0e16 : FAILURE in 1h 18m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fb7ee421581448684f852716a2c27df : SUCCESS in 57m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6bb492ef8423431ca37fee6d32e98940 : SUCCESS in 1h 05m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/371446b137834370a680abe586e3dd78 : SUCCESS in 30m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/24f0f7e27dc74b3a994e8ba81f71d0ef : SUCCESS in 1h 10m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d8265e1a215f4240b2ba513c443f7434 : SUCCESS in 1h 10m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"baec41c4cf9b01dcd1ea504a3acbb12f45ae9fcf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-15 04:41:39.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/794130/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/794130/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"72fdeedc2abb80c9ac9e9decd83932a3e026982c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-15 04:55:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2153a9e328dc13c9febbcf1efd32be90b44fb61d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 05:36:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c096d9eff8f4caa94b40b249731d97f : SUCCESS in 32m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0becb7918c954deaa796fbe4a7bdc854 : SUCCESS in 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d3a12940d92e12ddaecc0901dbc66e0e115bf93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-15 06:01:18.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/794130/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/794130/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"1e77135cf96ac965a2e7c59cf4fdceef5da84403","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-15 06:38:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794130/3 : SUCCESS in 1h 41m 46s","accounts_in_message":[],"_revision_number":3},{"id":"80334fa986408f30ee8dc72b2666c6ddd38e6433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 07:32:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ec4bccad16a4408e961579c1f788b076 : SUCCESS in 43m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d4875a2272c641e7a800d6ece6076e9f : SUCCESS in 1h 47m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5d29c30f5a6144b49288dd01eda0ef12 : FAILURE in 16m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b09454582ef34ce1b676067a22294f60 : SUCCESS in 1h 35m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/40c994b1b56f4e38a241149c9a0c77d1 : FAILURE in 31m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb981c5cd14a46fcb9e7bb2ef41bfd70 : SUCCESS in 1h 22m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2fc066c31f0436aa774de76c3216fbd : SUCCESS in 36m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3380fedd48cf4edfaffa99c5382d4a2d : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8c71d4b4c934c5ba6645a1a406bc97a : SUCCESS in 34m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e50d831bc73546389ada323ba1087fcd : SUCCESS in 24m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/500285ce6e3f4de3b2172df37a2d3c75 : SUCCESS in 28m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c9438c35334af9811c6552f7c806a0 : SUCCESS in 14m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/24a43d3280e247d1844a901f72fc90ae : SUCCESS in 54m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf5250820e214c6ab6a3e95ec79f41c7 : SUCCESS in 1h 26m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/715430cdca534b1ebf404ac86b0318fa : SUCCESS in 2h 35m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce2702adaf0a490ebbaa7ddfa604f702 : FAILURE in 41m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/974ca36f100e4685a56274a01a9b6f73 : SUCCESS in 1h 11m 00s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a4d965b728234464a1510f3397decd01 : FAILURE in 36m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/529d02d52aa9415e9807f2ff47f5f5af : FAILURE in 1h 06m 21s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af6fa3cda4aa44498b4a610cd5764e25 : FAILURE in 32m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b24096d4a7043398349333e0fcc42df : SUCCESS in 59m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d859e1a382944ed2b8a56849d44d7e24 : SUCCESS in 1h 13m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/11a42c504c584791a2ec04fb606173a6 : SUCCESS in 32m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7cc2f967441b424881f1d38b89a17aac : SUCCESS in 1h 29m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4127decd34c345099f4f97c543d364bb : SUCCESS in 1h 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d5b0828c20be6d2d63923aa7b74fba57d8dd6a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-22 01:53:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c2bfb8aee5bae0a5eba5bdde2a3e61bcd8fc1314","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 02:38:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/047860cbecd34e21861c7a6a5c9dd7a2 : SUCCESS in 39m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdcedb3978944e7f95da9489fdf52f33 : SUCCESS in 38m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"289f136c70ec8b7d87e4a533ab5b0ef885b168df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 04:06:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bc946e7ddf864ae4b77bbec5e32d0513 : SUCCESS in 36m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ecdda4125f6546b09dddee1a85f17d84 : SUCCESS in 1h 40m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2825af4749a943ebba178f7023779a57 : SUCCESS in 2h 11m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8de152fa70c34def864f516f393282de : SUCCESS in 1h 21m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e24d906b2bd84023877b61c412c2e666 : SUCCESS in 1h 34m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54dd0f9cbfed48968939f12611c29c5b : SUCCESS in 31m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a04dca9c61244e2beb286c925b683ab : SUCCESS in 42m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89034051506b4f5ca70236c3a336bdb1 : SUCCESS in 13m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2d8b10a8eee4a8d9bed1980f9e65aee : SUCCESS in 29m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af359f12adf54d20a1b75ee44bce8f0c : SUCCESS in 26m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93ee6af756d74ce985e348bcbd24a40d : SUCCESS in 28m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd5177e9729447148600e08b17b841ea : SUCCESS in 8m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e1cf301d14745509b2bd08ecea26d70 : SUCCESS in 39m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ad842e6fba04b679d6b2e69a0629651 : SUCCESS in 1h 21m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c1d2b05b0fc446cebf35aafcc6c24966 : SUCCESS in 1h 08m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/745beaa2573e47da816cade68fe2b242 : SUCCESS in 1h 08m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/525bfea104c54deb811c912f07b21612 : SUCCESS in 1h 01m 09s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/04ddb70822ab43aba4b4a9e13cd7d54b : SUCCESS in 1h 12m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5afe32495f8942aca87380c0b6efc69a : FAILURE in 1h 14m 47s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e4d312782d1439eb2647791289cdcb8 : SUCCESS in 1h 20m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cabb77eea1fe413bbc0bb386e35e1b57 : SUCCESS in 42m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/445a83a2665a4667b04f1c130d08ceb0 : SUCCESS in 50m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6f9d8f6c10214e799e38cb903588f2ac : SUCCESS in 24m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7296a3fd76c0425992df55c2055b8035 : SUCCESS in 1h 07m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c7608ba6d7de4b7e86f69400a1719f84 : SUCCESS in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"45fa8bfb29193de917e314b6a9054ccae4a1face","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-22 04:10:39.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/794130/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/794130/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c5c6d4b6779a7698731bc215142cb7d503b7bba4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 05:10:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794130/4 : SUCCESS in 1h 54m 11s","accounts_in_message":[],"_revision_number":4},{"id":"97db25444b8afa3a0cedecc513b89e0b26f6ad91","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-22 07:49:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)\n\nYou should also update documentation in https://github.com/openstack/neutron/blob/1ad9ca56b07ffdc9f7e0bc6a62af61961b9128eb/doc/source/contributor/internals/provisioning_blocks.rst","accounts_in_message":[],"_revision_number":4},{"id":"e630eceeb2e404aef5af6ebffc98c082bb8790eb","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-22 10:31:37.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6818e709fb274652d3a53a7134b7933dc7f4442e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-30 02:18:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"76e2344f32e27e6db620b1d8211049ed4c338d87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 03:01:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccb536e79d034c37b53cc8722d3841fd : SUCCESS in 41m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e40d00af8af463db10194261e7f91d7 : SUCCESS in 41m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3422add4be2599644587521c73b4b49dc0bd0af5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-30 03:58:41.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/30/794130/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"056d09a0fd10ebb0267e5cb5d85667c3cea7c523","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-30 03:59:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794130/5 : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":5},{"id":"7d05274bc6373460f5bfdaacf6b8dd6d0185cfdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 04:33:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a875268612f7446aa4293a951ec696bd : SUCCESS in 39m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a9d0e4dafa2d4848b8b05433ef801163 : SUCCESS in 2h 08m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/36ac7b2a613642d1ad67647a5ac1f566 : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/742e479ca50240edae089896b7750c7a : SUCCESS in 1h 20m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4efcaa83d9724a88aa9401a2fcdc856c : SUCCESS in 2h 13m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e732c4d00ccf413cb44345e095839dfd : SUCCESS in 30m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0be79efb276a4ecda3c55525cc88e34e : SUCCESS in 29m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8303fe3f9f814b80b6eec4f61ec1d11c : SUCCESS in 34m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47c656b37ded44e2bbd31e10ecbd33e9 : SUCCESS in 13m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/183f6e58a865436cb431b2782662de48 : SUCCESS in 23m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87c64f98579d43a6a79777e37ebed131 : SUCCESS in 27m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14e6e74bdcf044dba50852747088f3ed : SUCCESS in 23m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbbb9c5ddb434f03ae72b783af0095c0 : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fdb7853cb7e498aa05c52f3f899c471 : SUCCESS in 45m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9cf9413ce774cb99ffd2868ffc0ebad : SUCCESS in 1h 24m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1471b2ac255e42a68096c228d9fa6b24 : SUCCESS in 1h 03m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/52c075f0bad4417ebeb56234ec8ee766 : SUCCESS in 59m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7b73e56b37c24c79af18fdfeba56e31b : SUCCESS in 1h 07m 58s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad3d360687024dda824887773a8b57f8 : SUCCESS in 1h 06m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1697cba8d51d40d08a7127de249a7421 : FAILURE in 1h 14m 16s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/3283a7455ac1458cb1074e280a9288a5 : SUCCESS in 1h 24m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c61160d46a44c5eb696d9eb77ca9983 : SUCCESS in 1h 04m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e2edb4b9273a43fb8e9a756dfb6f83f1 : SUCCESS in 1h 03m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fdfcb39abd6545bf911112a09bae14bc : SUCCESS in 30m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e4e51d08e2784c919c8fea3ebe980e63 : SUCCESS in 1h 15m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/415f900a553144b6b3a1123071431f7c : SUCCESS in 1h 06m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9fc16f8301c920f759bc92fff567ca4c1fe540db","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-30 09:44:22.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/794130/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/794130/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"0f15f8fbf95ece51193b068ce5d1f6b15229b744","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-30 10:52:09.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"deb08644f32855035e749d504d1c7204beb3a987","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-07 14:02:31.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e5609d015ec5238203e44da176434fbe568681bc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-07 14:02:56.000000000","message":"Patch Set 5: Code-Review-1\n\n-1 just to draw your attention to the questions","accounts_in_message":[],"_revision_number":5},{"id":"d47f2bd315a82c3caa1594d5e80505ea8364acb9","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-07-08 01:42:09.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ef4947b77272dd9af73d6c3f568e21369eaadf55","author":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"date":"2021-07-09 03:19:04.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e16bf7dd59198d676d5d7033a63a59588355b076","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-07-12 07:02:09.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d7107cda09fe5db5589269a9222ce01d75316ec3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-07-12 08:07:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b9bc7dcb256d3c62fc490f0bef41f7419aff20e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 08:48:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b3384a472c645cfa16fdf499f3f1ce0 : FAILURE in 39m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38211a85f8c548ccb3b4fe0f67d3a689 : FAILURE in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b47ab4ef89de4978c21152606027a146301edf2d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-12 09:14:25.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/794130/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/794130/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"2f82a7a2791a73f671563bb5011ffe182562f3e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-07-12 09:23:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0dc98060e300f20450bf5a5a7e8251502f32ca99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 10:03:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bdfd07cf90a4f52bb510facaf906436 : SUCCESS in 38m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de6a9c055e17412c9cbc8b1721978569 : SUCCESS in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f270d6191e2d907f2e38b994134fc53ae7669367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:53:55.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fb4942536f854ff1b49be7e144ed00bf : SUCCESS in 45m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/63672a84c95a40cfafad4ad912cfae5d : SUCCESS in 2h 28m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6ec8fc0b5b1246528045ad825d96ccab : SUCCESS in 1h 41m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8c33303fa06344bfb27b37b2a952f7e0 : SUCCESS in 1h 52m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c2b1b7ab744e4f7fb2dbf55f9d1c8715 : SUCCESS in 2h 11m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a2c4c0cf26141a0aeaac116592551c4 : SUCCESS in 39m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2b5666708e64777a2bc2ca1c2ad4b99 : SUCCESS in 28m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16cc550421554f14989fe1d526607ff9 : SUCCESS in 40m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb89ad37a1d4c3b8015f0e7a23d9b16 : FAILURE in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/547e65852ef64b0986554ab836e1b887 : SUCCESS in 21m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c638a94b3fc44988b9521f9d507e95dd : SUCCESS in 26m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e805dcc866b400a9c39825f85f154fb : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d59e8a1c40440dc8e8c9c218cada443 : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7577993aec864ab7bb87f4e972483293 : SUCCESS in 55m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/768316b80d524203a0a37916231cf91b : SUCCESS in 1h 21m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3457872c13ab478183767ea0e47989a8 : SUCCESS in 1h 04m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/91ac278370184731926a85f90dccf24f : SUCCESS in 1h 02m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/547f58842f2043cf8447f7693ed84553 : SUCCESS in 55m 35s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/383bc751c794430c9f7b2774d7ca1992 : SUCCESS in 1h 06m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/85bffe699ede490da9df311c515d2fb3 : FAILURE in 1h 17m 10s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/128d0d8015f148bcabda4213299a69f5 : SUCCESS in 1h 48m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29ecc01661c748b8b99860b4dc7ce702 : SUCCESS in 45m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/112ddd91d78d4fb7a2b81118efed41c5 : SUCCESS in 46m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f35cab2b9e634d5390f1f7d3522fab2e : SUCCESS in 31m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/266fcd828fac41d59e1e17894e3fe8bf : SUCCESS in 1h 16m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b9b42072bfb94e48a725ccfc2d210e0c : SUCCESS in 1h 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ae0fd96b234951cea4a1d6f195f17c2c6e0eb475","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-07-13 00:23:53.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ac035ed4f4e4cec372d2baa4ee8044be2e342f79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 01:01:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e400351cae4409b988743ccaa8d9759 : SUCCESS in 33m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/934d394b8f4c4f0582d0d08b637f0197 : SUCCESS in 33m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6ee09706ed8b9c4f469f435ae835cb119b25c14c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-13 01:53:32.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794130/8 : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":8},{"id":"52490bd1a245fee5ac133cada9080f122abbefd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 02:41:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e054df46c9174cfaa3285c9792e7e0b4 : SUCCESS in 36m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/84796a3e39ad4a429cd75115af62418a : SUCCESS in 2h 06m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e04ce13622194720b07d47a430c06122 : SUCCESS in 1h 45m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a4c600ef23124ec2a875dd5db331ad6d : SUCCESS in 1h 24m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/316f82b4894a4df59b4ac9759c25d5c8 : SUCCESS in 2h 13m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac47137e7c344bde9e70777c7a15f5b6 : SUCCESS in 35m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5cddaa52a1f4d399878ce0e4e4a8e58 : SUCCESS in 30m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24582cd167de4f7391a98a419265f070 : SUCCESS in 43m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a41b87f1085041acb7ff80e126c83e4b : SUCCESS in 13m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eca2e5290a244ee85bbcb7f4b00757d : SUCCESS in 27m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce96c37d5c48416ca8c3439a4e2ea7b9 : SUCCESS in 20m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17f4ad722a144edfb9a2d28ac007e079 : SUCCESS in 24m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a12d1eafaa4a4fcc86ab0957ad22dc86 : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee2e78ab2b2b408db14726b6f5c27d7d : SUCCESS in 47m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfaf860e3d65427ca6c5c9a53db8103a : SUCCESS in 1h 19m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4be0ede637a641268d30f713dcf42c18 : SUCCESS in 1h 21m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb5baac1398140e0a14be31326e148ec : SUCCESS in 58m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/41e661eca357422e9120c4da30b30510 : SUCCESS in 1h 16m 42s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eec0bb2cb5694a9eab317f6fef0b0b88 : SUCCESS in 1h 07m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fab2e52332504fdf96e31dfbf9efddd4 : FAILURE in 1h 07m 18s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/2ffe4f80ce8b4f9f96606bcfdd5165f4 : SUCCESS in 1h 23m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5240de00394d45839ac891349eab95f4 : SUCCESS in 49m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/730782acd6604c158d51b02c79e1b367 : SUCCESS in 56m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b57e272be7f9479994622985d45db6b7 : SUCCESS in 30m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a2c59a0b92e49258f059b48eb174513 : SUCCESS in 1h 13m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0f15a5f6c9dc48aabebabd62226ec14b : SUCCESS in 1h 07m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7b84f9f913b35a4ceda15f89fc674ec680e186e2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-16 15:33:30.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/30/794130/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"4363a7123b3f90bf95b2fa0322f667bc51bf36d8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-16 16:21:09.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/30/794130/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"0199e266d2e6e6f1265c9ba3fc7d3ec23bf7baeb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-17 02:02:55.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/30/794130/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"eeaa6a3ed870c1d4f7fff6a4533da8400219bc02","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 11:05:11.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"59567dd3200cf20a4d97a4fac1b26b66df0194aa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-06 12:01:26.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"032184e1578c58faa08d8eb785cfc3cb649d9797","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-14 16:22:15.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1d6454a11679b70c5dff318356ba781e69f08ae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 16:23:26.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"ab018d943361beb0a74e61a30b23082b13fc0b24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 18:39:18.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/39357e8d912246ea9ae2d1f7d388406e : SUCCESS in 54m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5a08f393a0e64fa9acba9ded9bb02664 : SUCCESS in 1h 48m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/464b391fd616477a831f67761bcd7925 : SUCCESS in 1h 24m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f28e18dc148d439ea82dfd72c3287f29 : SUCCESS in 1h 30m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8ca18f7910d4a0a94489ce0bd960dad : SUCCESS in 1h 13m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27eaf96066e14f87a13c021995f918b2 : SUCCESS in 46m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdff20acc4624adfa6a8173a4d2fdf50 : SUCCESS in 15m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77f197d600fd4ef79792b6e58f52177e : SUCCESS in 23m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a240ed38744a42ae98fd0be9fef94663 : SUCCESS in 29m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7376448fb0476e98493f5bd89392a5 : SUCCESS in 11m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee47300a591e40efa3c0fba1792cc5b4 : SUCCESS in 50m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ef9fc8cb7f34a5aac8fa69697184e5d : SUCCESS in 1h 33m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c04810831be488e96f8bbc6ae57c298 : SUCCESS in 1h 07m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/46fe70b5f231411487dffb7486ecb7c8 : SUCCESS in 1h 00m 13s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/0aac8e04bbdb4e8cab71883eb302c9a2 : SUCCESS in 1h 29m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d8fba99bd6a4968a149110c5d727548 : SUCCESS in 55m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/34f0750d310c479ca9d70c068ab4a400 : SUCCESS in 57m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/96d1cf6c32e84b16a53136fe3d36aa1c : SUCCESS in 34m 30s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/14f97a981ae64c86812cc97776cc1acb : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":8},{"id":"4b80093fe9fccf3ff4e2c162c8cadfae0d5090da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 18:39:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"ae13814ebe87c519b7db469097056580f0a7a900","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 18:41:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb2a969886949d4ac0cba9a7000b3fe : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e0ea4a51ba36e227130c3b2a29d4eadc84615a6b","revisions":{"766288d591d162b7d0f61793437bfcb9301cc2e6":{"kind":"REWORK","_number":1,"created":"2021-06-02 05:30:00.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/30/794130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/794130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/794130/1"}}},"commit":{"parents":[{"commit":"c8a4beb95b8885c7c9c110ef40241c25bb2bceac","subject":"Merge \"[Doc] Update Freenode to OFTC as our IRC server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8a4beb95b8885c7c9c110ef40241c25bb2bceac"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 05:23:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 05:26:13.000000000","tz":480},"subject":"[L2] no provisioning block for internal service port","message":"[L2] no provisioning block for internal service port\n\nProvisioning Blocks [1] was introduced to manage composite object\nstatus. Port object was the one Neutron set provisioning block\nduring the port processing life cycle. Here is the compute\nport (VM\u0027s NIC port) processing procedure:\n1. nova creates port\n2. the \u0027openvswitch\u0027 mechinism driver inserts provisioning_block for this port\n3. nova calls related interface to plug the device\n4. ovs-agent sets the flows for the port and call update_device_list to neutron-server\n5. neutron-server try to set port status to ACTIVE\n6. neutron-server notify nova that \"vif-plugged\" success\n\nThis works fine for VM with its ports. But for neutron service port,\nlike router_gateway, router_interface and dhcp, it is unnecessary.\nBecause there is no dependency among neutron resources. Neutron\njust knows that the ports had been set properly. And another thing\nis, for most of these internal service port, there is no\nsecurity group/port security.\n\nSo for neutron internal service ports, the procesure can be:\n\n1. Neutron L3/DHCP/X related service plugin creates port\n2. No provisioning_block\n3. L3/DHCP/X related agent plug the port\n4. ovs-agent sets the flows for the port and call update_device_list to neutron-server\n5. neutron-server set port status to ACTIVE, then done!\n\n[1] https://docs.openstack.org/neutron/latest/contributor/internals/provisioning_blocks.html\n\nCloses-Bug: #1930432\nChange-Id: Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/766288d591d162b7d0f61793437bfcb9301cc2e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/766288d591d162b7d0f61793437bfcb9301cc2e6"}]},"branch":"refs/heads/master"},"b339ad5b2b0f56bbd54c9160cbcc74b0ee31a7d3":{"kind":"REWORK","_number":2,"created":"2021-06-03 07:50:35.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/30/794130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/794130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/794130/2"}}},"commit":{"parents":[{"commit":"c8a4beb95b8885c7c9c110ef40241c25bb2bceac","subject":"Merge \"[Doc] Update Freenode to OFTC as our IRC server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8a4beb95b8885c7c9c110ef40241c25bb2bceac"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 05:23:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 11:18:10.000000000","tz":480},"subject":"[L2] no provisioning block for internal service port","message":"[L2] no provisioning block for internal service port\n\nProvisioning Blocks [1] was introduced to manage composite object\nstatus. Port object was the one Neutron set provisioning block\nduring the port processing life cycle. Here is the compute\nport (VM\u0027s NIC port) processing procedure:\n1. nova creates port\n2. the \u0027openvswitch\u0027 mechinism driver inserts provisioning_block for this port\n3. nova calls related interface to plug the device\n4. ovs-agent sets the flows for the port and call update_device_list to neutron-server\n5. neutron-server try to set port status to ACTIVE\n6. neutron-server notify nova that \"vif-plugged\" success\n\nThis works fine for VM with its ports. But for neutron service port,\nlike router_gateway, router_interface and dhcp, it is unnecessary.\nBecause there is no dependency among neutron resources. Neutron\njust knows that the ports had been set properly. And another thing\nis, for most of these internal service port, there is no\nsecurity group/port security.\n\nSo for neutron internal service ports, the procesure can be:\n\n1. Neutron L3/DHCP/X related service plugin creates port\n2. No provisioning_block\n3. L3/DHCP/X related agent plug the port\n4. ovs-agent sets the flows for the port and call update_device_list to neutron-server\n5. neutron-server set port status to ACTIVE, then done!\n\n[1] https://docs.openstack.org/neutron/latest/contributor/internals/provisioning_blocks.html\n\nCloses-Bug: #1930432\nChange-Id: Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b339ad5b2b0f56bbd54c9160cbcc74b0ee31a7d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b339ad5b2b0f56bbd54c9160cbcc74b0ee31a7d3"}]},"branch":"refs/heads/master"},"0af0b87acc5ea9c27f16a85349102b4ab6a52a99":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-15 01:08:01.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/30/794130/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/794130/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/794130/3"}}},"commit":{"parents":[{"commit":"df147a112e495a22964563f53e87676ac798e1aa","subject":"Merge \"Load Linux Bridge Trunk extension if service plugin configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df147a112e495a22964563f53e87676ac798e1aa"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 05:23:44.000000000","tz":480},"committer":{"name":"liuyulong","email":"i@liuyulong.me","date":"2021-06-15 01:08:01.000000000","tz":0},"subject":"[L2] no provisioning block for internal service port","message":"[L2] no provisioning block for internal service port\n\nProvisioning Blocks [1] was introduced to manage composite object\nstatus. Port object was the one Neutron set provisioning block\nduring the port processing life cycle. Here is the compute\nport (VM\u0027s NIC port) processing procedure:\n1. nova creates port\n2. the \u0027openvswitch\u0027 mechinism driver inserts provisioning_block for this port\n3. nova calls related interface to plug the device\n4. ovs-agent sets the flows for the port and call update_device_list to neutron-server\n5. neutron-server try to set port status to ACTIVE\n6. neutron-server notify nova that \"vif-plugged\" success\n\nThis works fine for VM with its ports. But for neutron service port,\nlike router_gateway, router_interface and dhcp, it is unnecessary.\nBecause there is no dependency among neutron resources. Neutron\njust knows that the ports had been set properly. And another thing\nis, for most of these internal service port, there is no\nsecurity group/port security.\n\nSo for neutron internal service ports, the procesure can be:\n\n1. Neutron L3/DHCP/X related service plugin creates port\n2. No provisioning_block\n3. L3/DHCP/X related agent plug the port\n4. ovs-agent sets the flows for the port and call update_device_list to neutron-server\n5. neutron-server set port status to ACTIVE, then done!\n\n[1] https://docs.openstack.org/neutron/latest/contributor/internals/provisioning_blocks.html\n\nCloses-Bug: #1930432\nChange-Id: Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0af0b87acc5ea9c27f16a85349102b4ab6a52a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0af0b87acc5ea9c27f16a85349102b4ab6a52a99"}]},"branch":"refs/heads/master","description":"Rebase"},"68c188db207945a5ab37ea69ca94660469723411":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2021-06-22 01:53:00.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/30/794130/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/794130/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/794130/4"}}},"commit":{"parents":[{"commit":"2e403a3bfbf297d0955cf9e2f9ff21a4809337bb","subject":"Merge \"Report ExternalDNSOverQuota exception for recordset quota error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e403a3bfbf297d0955cf9e2f9ff21a4809337bb"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 05:23:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-22 01:48:05.000000000","tz":480},"subject":"[L2] no provisioning block for internal service port","message":"[L2] no provisioning block for internal service port\n\nProvisioning Blocks [1] was introduced to manage composite\nobject status. Port object was the one Neutron set provisioning\nblock during the port processing life cycle. Here is the compute\nport (VM\u0027s NIC port) processing procedure:\n1. nova creates port\n2. the \u0027openvswitch\u0027 mechinism driver inserts provisioning block\n   for this port\n3. nova calls related interface to plug the device\n4. L2-agent sets the flows for the port and call update_device_list\n   to neutron-server\n5. neutron-server try to set port status to ACTIVE\n6. neutron-server notify nova that \"vif-plugged\" success\n\nThis works fine for VM with its ports. But for neutron service port,\nlike router_gateway, router_interface and dhcp, it is unnecessary.\nBecause there is no dependency among neutron resources. Neutron\njust knows that the ports had been set properly. And another thing\nis, for most of these internal service port, there is no security\ngroup or port security.\nSo for neutron internal service ports, the procesure can be:\n1. neutron L3/DHCP/X related service plugin creates port\n2. no provisioning_block\n3. L3/DHCP/X related agent plug the port\n4. L2-agent sets the flows for the port and call update_device_list\n   to neutron-server\n5. neutron-server sets port status to ACTIVE directly, then done!\n\nThis patch will set the ml2 ovs to skip the provisioning_block.\n\n[1] https://docs.openstack.org/neutron/latest/contributor/internals/provisioning_blocks.html\n\nCloses-Bug: #1930432\nChange-Id: Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68c188db207945a5ab37ea69ca94660469723411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68c188db207945a5ab37ea69ca94660469723411"}]},"branch":"refs/heads/master"},"d382c88bfa6c6eac8b9c6a0226e584098fad96d7":{"kind":"REWORK","_number":5,"created":"2021-06-30 02:18:22.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/30/794130/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/794130/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/794130/5"}}},"commit":{"parents":[{"commit":"3127bd1d575594dc3882bb5deb52f0f9816744fa","subject":"Merge \"[ML2] Change way how list of supported API extensions is made\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3127bd1d575594dc3882bb5deb52f0f9816744fa"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 05:23:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-30 02:17:26.000000000","tz":480},"subject":"[L2] no provisioning block for internal service port","message":"[L2] no provisioning block for internal service port\n\nProvisioning Blocks [1] was introduced to manage composite\nobject status. Port object is the one Neutron sets provisioning\nblock during the port processing life cycle. Here is the compute\nport (VM\u0027s NIC port) processing procedure:\n1. nova creates port\n2. the \u0027openvswitch\u0027 mechinism driver inserts provisioning block\n   for this port\n3. nova calls related interface to plug the device\n4. L2-agent sets the flows (or rules/devices) for the port and\n   call update_device_list to neutron-server\n5. neutron-server try to set port status to ACTIVE\n6. neutron-server notify nova that \"vif-plugged\" success\n\nThis works fine for VM with its ports. But for neutron service port,\nlike router_gateway, router_interface and dhcp, it is unnecessary.\nBecause there is no dependency among neutron resources. Neutron\njust knows that the ports had been set properly. And another thing\nis, for most of these internal service port, there is no need of\nDHCP, security group or port security.\n\nSo for neutron internal service ports, the procesure can be:\n1. neutron L3/DHCP/X related service plugin creates port\n2. no provisioning_block\n3. L3/DHCP/X related agent plug the port\n4. L2-agent sets the flows (or rules/devices) for the port and\n   call update_device_list to neutron-server\n5. neutron-server sets port status to ACTIVE directly, then done!\n\nThis patch will set neutron *AgentMechanismDrver (including built-in\ndrivers: linuxbridge, macvtap, sriov, openvswitch) to skip inserting\nthe provisioning_block for Neutron internal service ports.\n\n[1] https://docs.openstack.org/neutron/latest/contributor/internals/provisioning_blocks.html\n\nCloses-Bug: #1930432\nChange-Id: Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d382c88bfa6c6eac8b9c6a0226e584098fad96d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d382c88bfa6c6eac8b9c6a0226e584098fad96d7"}]},"branch":"refs/heads/master"},"2b21bfa2e7b118238855880e6af8b952364f75f6":{"kind":"REWORK","_number":6,"created":"2021-07-12 08:07:00.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/30/794130/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/794130/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/794130/6"}}},"commit":{"parents":[{"commit":"3127bd1d575594dc3882bb5deb52f0f9816744fa","subject":"Merge \"[ML2] Change way how list of supported API extensions is made\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3127bd1d575594dc3882bb5deb52f0f9816744fa"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 05:23:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-07-12 07:00:51.000000000","tz":480},"subject":"[L2] no provisioning block for internal service port","message":"[L2] no provisioning block for internal service port\n\nProvisioning Blocks [1] was introduced to manage composite\nobject status. Port object is the one Neutron sets provisioning\nblock during the port processing life cycle. Here is the compute\nport (VM\u0027s NIC port) processing procedure:\n1. nova creates port\n2. the \u0027openvswitch\u0027 mechinism driver inserts provisioning block\n   for this port\n3. nova calls related interface to plug the device\n4. L2-agent sets the flows (or rules/devices) for the port and\n   call update_device_list to neutron-server\n5. neutron-server try to set port status to ACTIVE\n6. neutron-server notify nova that \"vif-plugged\" success\n\nThis works fine for VM with its ports. But for neutron service port,\nlike router_gateway, router_interface and dhcp, it is unnecessary.\nBecause there is no dependency among neutron resources. Neutron\njust knows that the ports had been set properly. And another thing\nis, for most of these internal service port, there is no need of\nDHCP, security group or port security.\n\nSo for neutron internal service ports, the procesure can be:\n1. neutron L3/DHCP/X related service plugin creates port\n2. no provisioning_block\n3. L3/DHCP/X related agent plug the port\n4. L2-agent sets the flows (or rules/devices) for the port and\n   call update_device_list to neutron-server\n5. neutron-server sets port status to ACTIVE directly, then done!\n\nThis patch will set neutron *AgentMechanismDrver (including built-in\ndrivers: linuxbridge, macvtap, sriov, openvswitch) to skip inserting\nthe provisioning_block for Neutron internal service ports.\n\n[1] https://docs.openstack.org/neutron/latest/contributor/internals/provisioning_blocks.html\n\nCloses-Bug: #1930432\nChange-Id: Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b21bfa2e7b118238855880e6af8b952364f75f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b21bfa2e7b118238855880e6af8b952364f75f6"}]},"branch":"refs/heads/master"},"22ff4d0ba2eb8dde156359813ee772b9b0cd7d7a":{"kind":"REWORK","_number":7,"created":"2021-07-12 09:23:48.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/30/794130/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/794130/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/794130/7"}}},"commit":{"parents":[{"commit":"3127bd1d575594dc3882bb5deb52f0f9816744fa","subject":"Merge \"[ML2] Change way how list of supported API extensions is made\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3127bd1d575594dc3882bb5deb52f0f9816744fa"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 05:23:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-07-12 09:15:52.000000000","tz":480},"subject":"[L2] no provisioning block for internal service port","message":"[L2] no provisioning block for internal service port\n\nProvisioning Blocks [1] was introduced to manage composite\nobject status. Port object is the one Neutron sets provisioning\nblock during the port processing life cycle. Here is the compute\nport (VM\u0027s NIC port) processing procedure:\n1. nova creates port\n2. the \u0027openvswitch\u0027 mechinism driver inserts provisioning block\n   for this port\n3. nova calls related interface to plug the device\n4. L2-agent sets the flows (or rules/devices) for the port and\n   call update_device_list to neutron-server\n5. neutron-server try to set port status to ACTIVE\n6. neutron-server notify nova that \"vif-plugged\" success\n\nThis works fine for VM with its ports. But for neutron service port,\nlike router_gateway, router_interface and dhcp, it is unnecessary.\nBecause there is no dependency among neutron resources. Neutron\njust knows that the ports had been set properly. And another thing\nis, for most of these internal service port, there is no need of\nDHCP, security group or port security.\n\nSo for neutron internal service ports, the procesure can be:\n1. neutron L3/DHCP/X related service plugin creates port\n2. no provisioning_block\n3. L3/DHCP/X related agent plug the port\n4. L2-agent sets the flows (or rules/devices) for the port and\n   call update_device_list to neutron-server\n5. neutron-server sets port status to ACTIVE directly, then done!\n\nThis patch will set neutron *AgentMechanismDrver (including built-in\ndrivers: linuxbridge, macvtap, sriov, openvswitch) to skip inserting\nthe provisioning_block for Neutron internal service ports.\n\n[1] https://docs.openstack.org/neutron/latest/contributor/internals/provisioning_blocks.html\n\nCloses-Bug: #1930432\nChange-Id: Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22ff4d0ba2eb8dde156359813ee772b9b0cd7d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22ff4d0ba2eb8dde156359813ee772b9b0cd7d7a"}]},"branch":"refs/heads/master"},"e0ea4a51ba36e227130c3b2a29d4eadc84615a6b":{"kind":"REWORK","_number":8,"created":"2021-07-13 00:23:53.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/30/794130/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/794130/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/794130/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/794130/8"}}},"commit":{"parents":[{"commit":"3127bd1d575594dc3882bb5deb52f0f9816744fa","subject":"Merge \"[ML2] Change way how list of supported API extensions is made\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3127bd1d575594dc3882bb5deb52f0f9816744fa"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-02 05:23:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-07-13 00:23:01.000000000","tz":480},"subject":"[L2] no provisioning block for internal service port","message":"[L2] no provisioning block for internal service port\n\nProvisioning Blocks [1] was introduced to manage composite\nobject status. Port object is the one Neutron sets provisioning\nblock during the port processing life cycle. Here is the compute\nport (VM\u0027s NIC port) processing procedure:\n1. nova creates port\n2. the \u0027openvswitch\u0027 mechinism driver inserts provisioning block\n   for this port\n3. nova calls related interface to plug the device\n4. L2-agent sets the flows (or rules/devices) for the port and\n   call update_device_list to neutron-server\n5. neutron-server try to set port status to ACTIVE\n6. neutron-server notify nova that \"vif-plugged\" success\n\nThis works fine for VM with its ports. But for neutron service port,\nlike router_gateway, router_interface and dhcp, it is unnecessary.\nBecause there is no dependency among neutron resources. Neutron\njust knows that the ports had been set properly. And another thing\nis, for most of these internal service port, there is no need of\nDHCP, security group or port security.\n\nSo for neutron internal service ports, the procesure can be:\n1. neutron L3/DHCP/X related service plugin creates port\n2. no provisioning_block\n3. L3/DHCP/X related agent plug the port\n4. L2-agent sets the flows (or rules/devices) for the port and\n   call update_device_list to neutron-server\n5. neutron-server sets port status to ACTIVE directly, then done!\n\nThis patch will set neutron *AgentMechanismDrver (including built-in\ndrivers: linuxbridge, macvtap, sriov, openvswitch) to skip inserting\nthe provisioning_block for Neutron internal service ports.\n\n[1] https://docs.openstack.org/neutron/latest/contributor/internals/provisioning_blocks.html\n\nCloses-Bug: #1930432\nChange-Id: Iaf7788bf0cba19a693cbf456f98e50d7b5de9e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ea4a51ba36e227130c3b2a29d4eadc84615a6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ea4a51ba36e227130c3b2a29d4eadc84615a6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
