)]}'
{"id":"openstack%2Fneutron~794892","triplet_id":"openstack%2Fneutron~master~Ie0205785cab307c132fbe409588739685cade7c0","project":"openstack/neutron","branch":"master","topic":"bug/1930926","hashtags":[],"change_id":"Ie0205785cab307c132fbe409588739685cade7c0","subject":"Use TCP keepalives for ovsdb connections","status":"MERGED","created":"2021-06-04 19:49:05.000000000","updated":"2021-06-09 14:28:04.000000000","submitted":"2021-06-09 14:25:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794892-bug/1930926","meta_rev_id":"4216b001b9f3c14ce19005f39173a42804e17a5d","_number":794892,"virtual_id_number":794892,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 14:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2021-06-08 14:14:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2021-06-09 10:58:38.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2021-06-08 06:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2021-06-09 10:58:38.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 20:27:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-04 20:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 20:52:20.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-04 21:42:09.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-06-07 17:25:05.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-06-07 17:25:05.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-06-07 17:25:05.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2021-06-07 17:25:05.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-06-08 05:49:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-08 06:57:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"e6047160efa80ae0934f8960615cd758d35b6233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-06-04 19:49:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a87dd317623999834c1dff67f17a51ee1e9db09b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 20:27:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d25f97b418aa4a958e38e1352597445d : SUCCESS in 33m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27c63b8ff9194ba8bf7c4cc1bd735b56 : SUCCESS in 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a81586f54624be0b7a6c29b9452705b9f45a9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 20:37:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c7eb4699d3f403fb0ae42357a27e392 : SUCCESS in 45m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecf0c6a29cc54de8be22ebd75f61aa47 : SUCCESS in 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c52cbec80c4ad89eb68c05bb30196a : FAILURE in 16m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba72793fe2b5470ca4074f6a5e96b85d : SUCCESS in 30m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d1ccab8339e45ccaddad49357eb7502 : SUCCESS in 31m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e38d56fcfe304e7794ddae049b05bc97 : SUCCESS in 28m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78fa120c6c44a08934513008db175e4 : SUCCESS in 9m 39s\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":1},{"id":"508f077bbd1aecb9bf264b88b785fb38aed84fd7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-04 20:52:20.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/92/794892/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/92/794892/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"39a2c70412b9f5882eea8f057d7a73d706fa8eee","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-04 21:42:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794892/1 : SUCCESS in 1h 52m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a6dcc3c67d1570eaab14186eb79522494a49a66a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-06-04 22:06:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ec6564de531b7c7ad8b88e9d8e88c1f0bf147dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:48:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06b7361ba27a4dc49ebe09291fdb1430 : SUCCESS in 42m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15825875e8c7410ab62a16b9ef3d79c6 : SUCCESS in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a236c67c96de929ce18e5b84e27a1286c7bba010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:55:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5139a9fdfda436283d6f19f22a372af : SUCCESS in 43m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/247edca796814e1096de9d93ab05eda9 : SUCCESS in 42m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bb0a0b520d84affbffad4b24094629c : SUCCESS in 14m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fb26a3fb21c4dfb92eff97b3c08607f : SUCCESS in 29m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f0d95a7c05446a9b1a54ca0f1ddd3a9 : SUCCESS in 22m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/041f808c330f439299d6a580bcb2169d : SUCCESS in 28m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e936b79381be4d8793a2b826d60ca5d4 : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":1},{"id":"424e49e926f10634a5c12f16e380af154a7d6d39","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-05 00:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794892/1 : SUCCESS in 1h 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b2b32dc7bfcf043c989d47566d306af45e2853a9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-06-05 02:02:04.000000000","message":"Patch Set 1:\n\nIt might be good to set TCP_KEEP(IDLE/CNT/INTVL) and make them configurable. KEEPIDLE could be the probe interval.","accounts_in_message":[],"_revision_number":1},{"id":"844ea034f5bd9fe721a3b251ad4506cdbd60729c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-05 06:53:17.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/92/794892/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/92/794892/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"744f39845275f1426ada1ce4e26f119477b17252","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-06-07 17:24:44.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e It might be good to set TCP_KEEP(IDLE/CNT/INTVL) and make them configurable. KEEPIDLE could be the probe interval.\n\nThough keepalives should be able to handle much shorter timeouts than echo requests since they shouldn\u0027t depend on ovsdb-server load. So tying them to the ovsdb_connection_timeout of probe_interval settings might not really makes sense.","accounts_in_message":[],"_revision_number":1},{"id":"d3c37bbbd5ab9cd084048bf0d7722c4b54c3a0f3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:49:28.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b3dc2865e9ea252e8641837a3ec92e9434f7f181","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-08 06:57:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"343d73efe3620a5b1157fc99bb02ed3b6a39e6d3","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-06-08 14:14:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"74e8bc44054b8335a77b5b071b0f2973de7f3484","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-06-08 14:18:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ada2aa12efaf023a6d464f3050d87fab1d70bc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 15:14:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1975f4330f2407aab7d15997a0e98ed : SUCCESS in 42m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fae8165c7b7464ab6d7ac9f9438a86a : SUCCESS in 40m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfe7933a971d1ecdb6905e5f555429bd7acb2504","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-08 16:17:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794892/1 : SUCCESS in 1h 58m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e5a74ddd2c9e3859554ccf531a56256c2ab43b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 16:55:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fe7c85cff05c4498a57421544207ebca : SUCCESS in 42m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c2cfacf4193b40acb0c503a73999a34b : SUCCESS in 2h 16m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f00cb56b0d3242488b94fcbfafbcfb19 : SUCCESS in 38m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e03b23426d734d68820fa39b46c0be57 : SUCCESS in 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa327b6f1f3418c942f6149be103638 : SUCCESS in 12m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11ec50959f6549e4a5558e4b8265fc64 : SUCCESS in 26m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beadaf7a6bae42089ca702a1559642a4 : SUCCESS in 33m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d62fd2495a044ebbd273d9e42bdd91b : POST_FAILURE in 24m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aee2dd44cd24223b526e418643307fe : POST_FAILURE in 11m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2431426686bb410c93c25229ed748193 : SUCCESS in 56m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6c3c636b764c48339e280865a2bb44f6 : SUCCESS in 1h 18m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4857b37b9e4f4eb98fde96e6de667345 : SUCCESS in 27m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f047a16584fa43cc8304ca80eae0b97f : SUCCESS in 1h 25m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/74764015d9cd4051a9eda73fadb85915 : SUCCESS in 1h 17m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"149c4f183dc8f5f3f6d54061b818ff89aaddd7f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-08 17:36:05.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/92/794892/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/92/794892/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"0bbc6a2c310b5ea6f9f5fb535fcf1bf4d914aa00","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-06-08 20:29:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ac8460303aaed5029d655748c919ccc417f516a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 21:05:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16365afc50ad4bdcbb9b9bd61ee97297 : SUCCESS in 32m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01413513eb9b45238bdcdadb236486ad : SUCCESS in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e5b800aa7e32aa448dd31434049b9868fa43936","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-08 22:02:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/794892/1 : SUCCESS in 1h 33m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bbd0eeb7252c2f13efc074668af1140fd1117032","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 22:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a621f816997e440892cac475f720fe5f : SUCCESS in 40m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c283daaab58442cc9940affd65044a9b : SUCCESS in 1h 58m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46aa273901444475af5b59e544e900da : SUCCESS in 33m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebd6fc5a32aa4a75925f49b3f9442f50 : SUCCESS in 28m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aabf107ea37942a89e115fd2699a2bbe : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8482571025247b8b3ef4cfab813459a : SUCCESS in 24m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/733818b1ea6f47e593592ffcbb79c494 : SUCCESS in 23m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/992160680c544aba95d83ad81fe548eb : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b35e4c5e97a45f9bb767791c257c958 : SUCCESS in 9m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ded906f37da5438e8614bc88806a3816 : SUCCESS in 43m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1780b4923135496282f590cecc1f3aa5 : SUCCESS in 59m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d5afc98d556c428cbdff4a28a5a93d8b : SUCCESS in 27m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a60e5b792e934ab386e529f3611f5bad : SUCCESS in 1h 13m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9efb960e3c9b44a38e2af836bbacf384 : FAILURE in 2h 07m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"380923bcae934d3d5fbfb41ef2b986574807b97e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-09 00:11:20.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/92/794892/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/92/794892/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f6b10bf2bbd04c7426c1fbd122dbe3d4ebff8779","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-06-09 10:58:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good! Thank u Terry!","accounts_in_message":[],"_revision_number":1},{"id":"fe67b7f610f3fb24a0794dbb665ad79935cfe55c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:59:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b0a7c4a48a48a6b68f12328d1cef84150140bbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:25:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9fff0ba7aeba408bb577c7471d97b2a4 : SUCCESS in 48m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6997a2b2845545ce83c967b760753274 : SUCCESS in 2h 02m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00175ac7763a4c69ae86b99d3589dcf3 : SUCCESS in 36m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40c219e883b4c6a9e210c3d6f991747 : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f018fca0a6714e859ea591c785ba6891 : SUCCESS in 31m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37cbccc4f54b4eb8b04e08d12fc0bd28 : SUCCESS in 33m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16bb943ab5c94b21879c1fba83d0f838 : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c119c5a413214b349a9ce48f32d85f21 : SUCCESS in 51m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d6a9c948b3614a5e9429d68d9b2cc6f5 : SUCCESS in 1h 11m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/386798c7ba8e44bbb46b885afe71e321 : SUCCESS in 35m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8b6b43a9a0a665fc5f3067ab93b7b3967a446684","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:25:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4216b001b9f3c14ce19005f39173a42804e17a5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:28:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfa09c0a09d4cf0aa3102fe03e27b4f : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65cce351d74a9a637fdb2a9d5e0e63445dda9ea9","revisions":{"65cce351d74a9a637fdb2a9d5e0e63445dda9ea9":{"kind":"REWORK","_number":1,"created":"2021-06-04 19:49:05.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/794892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/794892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/794892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/794892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/794892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/794892/1"}}},"commit":{"parents":[{"commit":"7e98d189270c408b4a5f5b69e9e479a458407b22","subject":"Merge \"Switch neutron-rally-task job to ML2/OVS and skip \"test_models_sync\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e98d189270c408b4a5f5b69e9e479a458407b22"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-06-04 19:47:36.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-06-04 19:49:02.000000000","tz":0},"subject":"Use TCP keepalives for ovsdb connections","message":"Use TCP keepalives for ovsdb connections\n\nWhen failing over OVN DB servers from one server to another, the\nserver which originally hosted the VIP doesn\u0027t notice the connection\nis gone and doesn\u0027t reconnect. Ultimately, this is something that\nneeds to be fixed in python-ovs, but setting the SO_KEEPALIVE socket\noption avoids the issue. This also has the benefit that the client\ndoesn\u0027t need to send \u0027echo\u0027 requests, which can time out on an\noverloaded ovsdb-server, which causes a disconnection which then\nadds even more load on the ovsdb-server as it has to send the entire\ndb contents over the wire after the connection.\n\nCloses-Bug: #1930926\nChange-Id: Ie0205785cab307c132fbe409588739685cade7c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65cce351d74a9a637fdb2a9d5e0e63445dda9ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65cce351d74a9a637fdb2a9d5e0e63445dda9ea9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
