)]}'
{"id":"openstack%2Fneutron~795300","triplet_id":"openstack%2Fneutron~master~I143cd55612118f243c0e502fb77a611d1ee48761","project":"openstack/neutron","branch":"master","topic":"bug/1930401","hashtags":[],"change_id":"I143cd55612118f243c0e502fb77a611d1ee48761","subject":"Use 2 dhcp agents in TestLegacyL3Agent","status":"MERGED","created":"2021-06-08 11:41:18.000000000","updated":"2021-07-06 09:12:00.000000000","submitted":"2021-06-08 22:04:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"795300-bug/1930401","meta_rev_id":"c3734af6dc56d34c105a540eba4488c89b8b02ea","_number":795300,"virtual_id_number":795300,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-08 22:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 12:16:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-06-08 12:29:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-06-08 15:17:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-08 11:42:07.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-08 11:42:07.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-06-08 11:42:07.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-06-08 12:55:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-08 13:52:50.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-06-08 14:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 01:54:05.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-09 02:02:59.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"341db3ba95c98f11dfdc6cea1891664f0b1c3b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-08 11:41:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7be2cd0c9c7f338f76bd16266318c8fc6371f713","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-08 12:11:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df7cd0642102d79d620df3864622ef5d27b33e61","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-08 12:12:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e4a219200062880931f6d9e00bf5a8708fbf070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-08 12:15:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e702d232797dbfb9636b98193a13b19b58377633","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-08 12:16:12.000000000","message":"Patch Set 2: Code-Review+2\n\nI hope that Zuul will be happy with that change","accounts_in_message":[],"_revision_number":2},{"id":"6482b23d284718922c8797c8d9b61be8f568538a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-08 12:29:18.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks, let\u0027s wait zuul","accounts_in_message":[],"_revision_number":2},{"id":"35ec26e3e038c922c2237dec832141bd43dc5691","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 12:55:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eaee92d7072484dacce364b7be0fdb4 : SUCCESS in 35m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d892515182cb4c4680c77a25386f0027 : SUCCESS in 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c43eb0f2722fa2c1caf103eaab71b6003002c2eb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-08 13:52:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/795300/2 : SUCCESS in 1h 36m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c2e40edd5a7db048e5a6175cbe9fbee9e006db35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:26:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ed7162e69074480a8a9cb1583ca16bc3 : SUCCESS in 40m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3447e07bc39f49f0862f2e9a60e77106 : SUCCESS in 1h 53m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2adc1e9bc5fe4d74b56ac611effb4239 : SUCCESS in 2h 04m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a1c4ca945eb42f98ac161fc87a970a1 : SUCCESS in 1h 34m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/67d39185c4c84b7b8c7e25de30e29aa9 : SUCCESS in 1h 59m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48944eac75af4ca98e02febc4ad676e7 : SUCCESS in 41m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/914d7d57a04d4cda841f5915f81a0dec : SUCCESS in 35m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f26c2ab5d394bce91e5d3a51e67cbbb : SUCCESS in 16m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eda4dcfddf95446aabc7a79e6124d891 : SUCCESS in 31m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fbac0cb87ce4f8da73cd05576887c17 : SUCCESS in 26m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7413dacfbc9b4313a1f13e4e3654b2b4 : SUCCESS in 33m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9143e9b1d2024a72968b1c0b743dfd4e : SUCCESS in 11m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c24f208de4554bddb28b4df53db76816 : SUCCESS in 1h 01m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c35842ba05641f8a593537ae38c18f3 : SUCCESS in 1h 28m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/29cb4e34ccc743c49ceaf2cc23cb3911 : SUCCESS in 1h 25m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a9146377272434dbbe29d2e6a703b6e : SUCCESS in 58m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/738ef84f820a43abbe141a6dce98ecb5 : SUCCESS in 1h 07m 50s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d6f22c7d8fde4099b83371c85733d8ad : SUCCESS in 1h 24m 47s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2fe969c911044b179e34fa10f0e810b1 : POST_FAILURE in 35m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b67871e8ffa04d9494a198e43786c710 : FAILURE in 1h 51m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/776b3cfeb5a5444881663f4115559ce2 : SUCCESS in 1h 09m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/df2813433b5d442fa5c14f8b1819b38b : SUCCESS in 1h 20m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c425d705a5da48f4a8de919ff425e55e : SUCCESS in 39m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/758b56449c3e4beb92d05ad014dcbb1e : SUCCESS in 1h 31m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/db5acecb8f3c41b6b958332b47cc6fcd : SUCCESS in 1h 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c018362897f922ac92e97b2c690b18febe3477e0","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-08 14:31:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f0bd41122c24a4bf39909e5319393077667407d9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-08 14:31:40.000000000","message":"Patch Set 2:\n\nneutron-fullstack-with-uwsgi is fine so far","accounts_in_message":[],"_revision_number":2},{"id":"4871e578f211512a714556ec00cb86c328731298","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 15:14:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfcf7aa1a4364e5fa39da8688d668b8a : SUCCESS in 38m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a81b14fadb84e5d97fcb658030333f4 : SUCCESS in 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34654890ab1dcd17f10a5510130d2b2532de93d6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-08 15:17:28.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ded4e95f3c5dd9a7d1c1704f3c4ea49612db87ff","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-08 16:09:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/795300/2 : SUCCESS in 1h 33m 45s","accounts_in_message":[],"_revision_number":2},{"id":"1fddfc373508dde6df3319c89a0607d85b7df0ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 18:16:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4a0f9d1bd1ad4bc3946f8571b97a2b3f : SUCCESS in 52m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/70d8afcb0c7d4bf6b534001a96104381 : SUCCESS in 2h 08m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a1d9f994f93847539115b76d3a47862d : SUCCESS in 2h 05m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/801c36b9e7804e7296ac67f0665a951b : SUCCESS in 1h 10m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/34cddfb5fb4246548b094a292ff8e1a7 : SUCCESS in 2h 40m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3ac64c3fde04187a6ac0d3f96f5eb3a : SUCCESS in 33m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3760d96b68d493cab29c3aef7664052 : SUCCESS in 27m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f723e53b63bc488caa1e2a9b3d19c7e2 : SUCCESS in 14m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/979f2b87627a413bb16aeb0f9d4dd64f : SUCCESS in 32m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e298c1624382407488a5c634e27ccdd1 : SUCCESS in 30m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c9092fc78f54733abe200ae783a701c : SUCCESS in 32m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05de77ea43f745379d3cd12839ab79f2 : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/267362e907dc4e2fb58ff7be803e417c : SUCCESS in 51m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d6968b4ec2045368c2743ac23dc0e72 : SUCCESS in 1h 31m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a897342800b1461dad3fce220e96ba62 : SUCCESS in 1h 40m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/64f9b1ab9f2c4308a1bf2f37ce091deb : SUCCESS in 1h 16m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eff701c0a500485ca7a5d22c766bf61c : SUCCESS in 1h 03m 34s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c3e93db8b5f742b9a1eddc38054e8f03 : SUCCESS in 1h 22m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8501a597e9b24459b43d9ff2dc4409f2 : FAILURE in 25m 45s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c1aa30e1a349489b87536bf29b9e102f : SUCCESS in 2h 08m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41c96a0916bf4b99b5f1b4e0ffe8ab70 : SUCCESS in 1h 23m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dd63a432499c484594e457daeb4e3009 : SUCCESS in 1h 16m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c6f96fa84a8b4053923bf9e7c3b69870 : SUCCESS in 32m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ffdbac94552e48f99b58e51ef8d1b4f7 : SUCCESS in 1h 37m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6760574546434f18b7d9146b5d6e2c81 : SUCCESS in 1h 27m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3237840d249e11e9458c734fa78a3c189d4e6035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 18:17:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80997bb6b39bf6e3bcc9d9f88bbdd05f515782e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 22:04:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2fafaf1a41cc4010a78abcafb45f2716 : SUCCESS in 49m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5b1ca80f6f6947f08533a825fcdfc097 : SUCCESS in 2h 20m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/11747c8975c94b8d9e615a1d51e370a3 : SUCCESS in 1h 57m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/55a43dc4338a4ccbbbb5683566aec7b3 : SUCCESS in 2h 30m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/acfddfc6cb08408d8acde3626506f293 : SUCCESS in 1h 30m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e2628368f04463896b4cf130f265f1d : SUCCESS in 34m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1103e645aaac4401935275f464eb3761 : SUCCESS in 14m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/593487634e604eecbea48e28c0f3718f : SUCCESS in 32m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c69f82b5cd604e749adb319644a8713d : SUCCESS in 28m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bdf8a8a44f548dcb1babcff33e416db : SUCCESS in 10m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/002717a462934ddf90d363e3d005936c : SUCCESS in 50m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/148ae9681f764a5d82d0c37d9cf10a89 : SUCCESS in 1h 38m 23s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/beb6300f70d64ccaa9319a207538e05b : SUCCESS in 1h 26m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4119913599f840ff87534e94eb9aa8fb : SUCCESS in 1h 03m 49s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b899cdeb49ec4b22ade02c61792c130e : SUCCESS in 1h 19m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/43f33f8e06f94405996f8273e6b1497f : SUCCESS in 1h 24m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d66682b6083343eeaf3266d1cfb59034 : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b9e0f01742918561b9ebafc99fba3e4fdc1b5704","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 22:04:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0122dc7fc62bbbe81a6672b0cb48c857861eb64a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 22:07:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4199aa6ba83740618259b5926361874a : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ffaafbd979a2b53e4e05af506e0471390524ec1f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-09 01:54:05.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/795300/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1251055cb283cdcd067ffcd1c98dc375fc7a1f5b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-09 02:02:59.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/00/795300/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c3734af6dc56d34c105a540eba4488c89b8b02ea","tag":"autogenerated:gerrit:revert","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-06 09:12:00.000000000","message":"Created a revert of this change as I1e7087ea109a24a846444772ca6deda73eb9790e","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"07337f9e99cdcfb1af3546a537d5595330e8bded","revisions":{"c55b222ec79cfa3443fc24c73430ff7e22f1b4ba":{"kind":"REWORK","_number":1,"created":"2021-06-08 11:41:18.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/00/795300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/795300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/795300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/795300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/795300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/795300/1"}}},"commit":{"parents":[{"commit":"df94641b43964834ba14c69eb4fb17cc45349117","subject":"Merge \"use payloads for PORT AFTER_CREATE events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df94641b43964834ba14c69eb4fb17cc45349117"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-06-08 11:39:43.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-06-08 11:39:43.000000000","tz":180},"subject":"Use 2 dhcp agents in TestLegacyL3Agent","message":"Use 2 dhcp agents in TestLegacyL3Agent\n\nThis is a workaround for privsep hanging issue described in bug 1930401.\nProper fix is developed in\nhttps://review.opendev.org/c/openstack/neutron/+/794994\n- this fix will revert current change to reproduce and verify\nprivsep issue is fixed.\n\nRelated-Bug: #1930401\nChange-Id: I143cd55612118f243c0e502fb77a611d1ee48761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c55b222ec79cfa3443fc24c73430ff7e22f1b4ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c55b222ec79cfa3443fc24c73430ff7e22f1b4ba"}]},"branch":"refs/heads/master"},"07337f9e99cdcfb1af3546a537d5595330e8bded":{"kind":"REWORK","_number":2,"created":"2021-06-08 12:15:24.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/00/795300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/795300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/795300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/795300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/795300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/795300/2"}}},"commit":{"parents":[{"commit":"df94641b43964834ba14c69eb4fb17cc45349117","subject":"Merge \"use payloads for PORT AFTER_CREATE events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df94641b43964834ba14c69eb4fb17cc45349117"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-06-08 11:39:43.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-06-08 12:15:08.000000000","tz":180},"subject":"Use 2 dhcp agents in TestLegacyL3Agent","message":"Use 2 dhcp agents in TestLegacyL3Agent\n\nThis is a workaround for privsep hanging issue described in bug 1930401.\nProper fix is developed in\nhttps://review.opendev.org/c/openstack/neutron/+/794994\n- this fix will revert current change to reproduce and verify\nprivsep issue is fixed.\n\nRelated-Bug: #1930401\nChange-Id: I143cd55612118f243c0e502fb77a611d1ee48761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07337f9e99cdcfb1af3546a537d5595330e8bded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07337f9e99cdcfb1af3546a537d5595330e8bded"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
