)]}'
{"id":"openstack%2Fneutron~796032","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I2a5e1fe3dbc6f2f342feaec92f4c122cfccce6d1","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1926170","hashtags":[],"change_id":"I2a5e1fe3dbc6f2f342feaec92f4c122cfccce6d1","subject":"Added common config and SR-IOV agent config to sanity check","status":"MERGED","created":"2021-06-11 14:53:49.000000000","updated":"2021-06-15 19:00:49.000000000","submitted":"2021-06-15 18:56:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796032-bug/1926170","meta_rev_id":"1e1a081ea9900bf6eaedc599838dcb203b23049a","_number":796032,"virtual_id_number":796032,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 18:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-06-15 07:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-06-14 08:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-06-15 07:42:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-11 17:29:59.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-06-11 21:59:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 08:02:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-15 07:42:43.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"ed8b38023c1ff2a70ad4192f7b745d25c3374af6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-11 14:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee00b325cbce0ae58f379ada987f7c92d01d5135","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-11 17:29:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/796032/1 : SUCCESS in 2h 25m 03s","accounts_in_message":[],"_revision_number":1},{"id":"62e66f057ce5775f474dc8a61adac2c188c6ec53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 21:59:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/08c4caace9eb44aba1af31f64e220950 : SUCCESS in 49m 23s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/7f3a409b875a4a2daccb99eee1fb5ebe : SUCCESS in 2h 06m 29s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/676b34b6240149b8b049947e0bcd9fc1 : SUCCESS in 2h 34m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/69c1ef30f3e24c1f81430961bbfc7662 : SUCCESS in 1h 29m 56s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/e1dbbdbe522b4274b0ebef51f2a28332 : SUCCESS in 1h 44m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38f078d3baea4e5699fa3e5db1b17ad9 : SUCCESS in 38m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26096e8e77614928b3f46b9b6cd0c9bf : SUCCESS in 15m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d03c37cba74c4f61a3146b2241c4aad2 : SUCCESS in 30m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d76fed783f1460683effb2101a20aae : SUCCESS in 32m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1425ffbfc7db4f5eaad83634258b44ed : SUCCESS in 29m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77995c85f619433298bc8dedfeedec3d : SUCCESS in 11m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e29a9069c0794109be65714fd1d10b75 : SUCCESS in 57m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/22026d3b5fa34bd89a0e298579d9835a : SUCCESS in 1h 28m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c441a43d74c9487c8f21a45136e364c0 : SUCCESS in 1h 22m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dc0075ad983040da86412ccfa95f9787 : SUCCESS in 1h 37m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/595816d50b044490bed1113fcaaf140e : SUCCESS in 1h 09m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/93d6c23b4b7a4cdd9f206c7e29b3c45c : SUCCESS in 1h 20m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a14c21b1c04940d5a44a188189d44d47 : SUCCESS in 1h 22m 40s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/421d5101c904452cbb2ef2ea3e055169 : SUCCESS in 1h 13m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c7aa27751f64421695206ad04d6c354d : SUCCESS in 1h 17m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b208e4131a8e4769b0a72687fa17806d : SUCCESS in 1h 52m 05s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3a32596831084f21ab5a19054f8a5a74 : SUCCESS in 1h 38m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06a989ef95944ae0bd70f7a129646a16 : SUCCESS in 1h 10m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c21e4c9a51ec4f0ca3c8dc68dedfe8cc : SUCCESS in 1h 07m 01s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ffebf7ee5f9d4c9b866e95a58485f6ca : SUCCESS in 1h 47m 02s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ccadc9f64cd45e98a6afdc534debcd3 : SUCCESS in 1h 00m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d0f35f5821d94ce1a150a7ecb30cc346 : SUCCESS in 1h 15m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d6a117de8b0a4adb925ee2c46502191e : SUCCESS in 40m 29s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ae35bc8c742c4c24b06545ea38ef3724 : SUCCESS in 1h 18m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/91cf899888ca42d4a31cbd1062579e01 : SUCCESS in 58m 26s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/63264e2a74f5450d9881da212d4f06a8 : SUCCESS in 3h 40m 56s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ac839eaa8e214521898b9e843131e3b0 : SUCCESS in 1h 29m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd5223ede7f147fa95098a7979410191 : SUCCESS in 1h 41m 14s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/85b7ed40103048e284e8e09312ca6436 : SUCCESS in 2h 13m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/227d79d4723b440199e99ff9275af4c3 : FAILURE in 1h 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65243fa45ecab1514ac5c6d29222811d39a9bb55","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-14 08:02:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee98472bb7086c80bd08be6c0bbce92b1ffdc718","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-15 07:42:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1da285677268f05b2c887a90fce5af60084787f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 07:43:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e47a9197d3fff222b0f94bc61940176b04d7c52c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 10:00:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/a1a8d9e6185f4293ade07f164e3677a3 : SUCCESS in 43m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a45066555d47d9ba64f47fac908d98 : SUCCESS in 13m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebb0364378fe4bc4a86ea0eba60d6e66 : SUCCESS in 23m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/465bae0ad92748da99f4342380ee31df : SUCCESS in 29m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d1a37d18b24fa2a6db287e2d343e12 : SUCCESS in 31m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/42d845b79cea4cadbbeed7cf61cdcd6a : SUCCESS in 1h 18m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0331aa24b6b14788957228f85a651829 : FAILURE in 24m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ec616555a6ca4107b6a62efaea1db395 : SUCCESS in 1h 13m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6d9f2168553f42518a08cc98aeca39ee : SUCCESS in 1h 05m 28s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d457701f4d564cf6a48eaac6b1b2497e : SUCCESS in 1h 19m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/572ce8daedb84465b0dc0ca3b31c7561 : SUCCESS in 1h 05m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1a4f9bf7254e4e5d9c20983de6941305 : SUCCESS in 1h 12m 28s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/253ab0b7fd224e4288c7696a17fee0b6 : SUCCESS in 1h 29m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/273bd43102d040a2b2f0574a8bd7b469 : SUCCESS in 1h 05m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b1afe550a6d74f03bace258d24348253 : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9240e110348ab4f0a4433a6b6f4e919c4a24e5eb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-15 11:47:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ab01844fa8ac6a3649708b88f6347f4aa10215c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-15 13:34:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/796032/1 : SUCCESS in 1h 45m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0fc7cfacc5c8ce34e7f5e376ef3c8c810a5acb78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:02:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/56a5815702714a3494b453c7cce7bbb2 : SUCCESS in 1h 09m 24s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/97cfaafc1a6c4b2c9a4941931d335bc4 : SUCCESS in 1h 54m 42s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/508a4cc49b0844978d107b74ed18982d : SUCCESS in 1h 45m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/09f4fa41a4ae4b7e8860c8891f5aa95f : SUCCESS in 1h 41m 45s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/3da7e3fba5b5437aa1f713f0d72ea970 : SUCCESS in 1h 59m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3143e6916a2d4ff8a79d9d0e21e9c81e : SUCCESS in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b048874d645840d5a9d67dbcb7ecaf17 : SUCCESS in 14m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f57e42850e6c425da592253cf51e88bb : SUCCESS in 22m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4136dc794802447ba0f1d496d5886ee5 : SUCCESS in 29m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/025c2728ee5a419c9a0751c673e0b5ab : SUCCESS in 26m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d4c88f10ac456cb6e9120013813c38 : SUCCESS in 12m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/10569d19c8774482aa629442d413e2ca : SUCCESS in 1h 02m 10s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e7944e3893974624864cad786cf42830 : SUCCESS in 1h 23m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/476b8b8f6c074d91b6d461ae3787106a : SUCCESS in 1h 26m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/346fb4a7a21542cc9fd80d056abe67ac : SUCCESS in 1h 29m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/46f7cc2ec34e46cd817ca567824b4ead : SUCCESS in 1h 07m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dedd275f94824556874a967d62549ff9 : SUCCESS in 1h 15m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c731209a33db477c80bf2f12e32cfdbc : SUCCESS in 1h 16m 19s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/07345bd3c8cc413fbf0f081d316511cf : SUCCESS in 1h 02m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9b656774694142f1a9f76eacf25ad96d : SUCCESS in 1h 07m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f3c9e31f0cd5482ab20b05d109d96798 : SUCCESS in 1h 34m 51s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c5ac7735fe2046e99a06c7c18f00272a : SUCCESS in 1h 27m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/527ad389ac3c41cc89b932365af39618 : SUCCESS in 1h 02m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5bbd430bfe7743f5ae7a4228008b3778 : SUCCESS in 1h 07m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8df6dc939456442794c96214feaadaca : SUCCESS in 1h 25m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e6a4aa44ccb4c8198af6b90a2637a1b : SUCCESS in 1h 13m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/241f7847300942d8bae138adc5208f45 : SUCCESS in 1h 14m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/aed441ad69aa40828e97cb743e614205 : SUCCESS in 40m 25s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/26116c78bb954b2da63dd82597f6ef16 : SUCCESS in 1h 21m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/620e0e5a0ffd4ebe85da8d7507d73c83 : SUCCESS in 53m 30s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a0c267c8b47f41e2836577605e148a27 : SUCCESS in 3h 54m 21s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7aee9d11c1d542bfb8c115b7b20b2887 : SUCCESS in 1h 20m 03s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/58ee4cf2a0784470a50bbae6cad39946 : SUCCESS in 1h 39m 52s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c976f778dd7945e087f99c3cc410e62f : SUCCESS in 2h 20m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/856ff61cb1c9460daa2d8fd2b41670b3 : FAILURE in 1h 40m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"931363303990fd8e0dabaa95abaab268dcb78306","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:03:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23ef90343b61fad50ca7c35b7851456ae2a15307","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 18:56:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/0f9542921e9c42d2b993ddf4d52de47a : SUCCESS in 1h 42m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35f2d8358d941e4986387ca77ca55b1 : SUCCESS in 15m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eecdc97d0ac34bbd9fdcc91677701eca : SUCCESS in 59m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83e8b23fe0d74306a5faf496a3a09130 : SUCCESS in 28m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc68259f20d344d894dad0176f9140d5 : SUCCESS in 10m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/81ce2262af19440f8ad212d8c64be017 : SUCCESS in 56m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9f6c596be9934080a6863c53b738eb0c : SUCCESS in 1h 17m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/644e868502a24bce8607edf0a60ef6e9 : SUCCESS in 1h 17m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/16c25e97318a4849bac3d73d933833dd : SUCCESS in 1h 08m 45s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5f80a229e2aa4b499d5d6d18bca9eb7b : SUCCESS in 1h 11m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1bb67ee5abe473b800415f034d8827e : SUCCESS in 57m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e8ea79cbcf4d47ce86be500d1c6b4094 : SUCCESS in 1h 15m 58s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e80756da31b04a4bbccdb8fbbd1d0cf7 : SUCCESS in 1h 23m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5aaca8fb3cdc496b8856909662ca8dac : SUCCESS in 1h 08m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8e106a0efc0e46f1bf7dd968b08d969f : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f3ac3926ef8ebb457473fb5e44b39c9839707038","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 18:56:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1e1a081ea9900bf6eaedc599838dcb203b23049a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 19:00:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde2a62528ff418bab1025fe7c5ec86a : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69a53de9bcd395276414db8710908c22b6dd32e4","revisions":{"69a53de9bcd395276414db8710908c22b6dd32e4":{"kind":"REWORK","_number":1,"created":"2021-06-11 14:53:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/32/796032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/796032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/796032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/796032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/796032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/796032/1"}}},"commit":{"parents":[{"commit":"a9a68c6686777d4398a5dd99a45224e446c4b671","subject":"Merge \"[DVR] Send allowed address pairs info to the L3 agents\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9a68c6686777d4398a5dd99a45224e446c4b671"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-26 15:51:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-11 14:44:08.000000000","tz":0},"subject":"Added common config and SR-IOV agent config to sanity check","message":"Added common config and SR-IOV agent config to sanity check\n\nAdded common config and SR-IOV agent config parameters to the sanity\ncheck script, to add the following missing configuration parameters:\n- default.notify_nova_on_port_status_changes\n- default.notify_nova_on_port_data_changes\n- sriov_nic.physical_device_mappings\n\nChange-Id: I2a5e1fe3dbc6f2f342feaec92f4c122cfccce6d1\nCloses-Bug: #1926170\n(cherry picked from commit 28cd6c82e9a0dbf908eb4ece54085a975c2fe5d4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69a53de9bcd395276414db8710908c22b6dd32e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69a53de9bcd395276414db8710908c22b6dd32e4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
