)]}'
{"id":"openstack%2Fneutron~796221","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ie6ae31e40488fd8f3d43e83b25e536a7dd9d938c","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1930996-stable/wallaby-stable/victoria-stable/ussuri","hashtags":[],"change_id":"Ie6ae31e40488fd8f3d43e83b25e536a7dd9d938c","subject":"Provide the rpc_response_max_timeout parameter to sriov-agent","status":"MERGED","created":"2021-06-14 14:02:28.000000000","updated":"2021-06-16 10:34:21.000000000","submitted":"2021-06-16 10:31:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796221-bug/1930996-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":796220,"cherry_pick_of_patch_set":1,"meta_rev_id":"bc52a58da465fa28afbc358ba212774dab4cb7d7","_number":796221,"virtual_id_number":796221,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 10:31:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-06-16 08:44:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-06-15 07:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-06-16 08:44:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 16:26:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-06-14 18:06:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 07:41:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-16 08:44:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"39836c1e5cce5082b6bac9f4cc0fada984f0fa34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-14 14:02:28.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"ab3122a43bbdea849315bb02ce0bbf6e434fd6f7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-14 16:26:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/796221/1 : SUCCESS in 2h 22m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9c69e304a0b31f2a58da75c08e330751c77deff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 18:06:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/31cb629ee8864d16863548a9718d66dd : SUCCESS in 50m 28s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/bc4a3b23c42d4b2796f2475833cc0d94 : SUCCESS in 2h 05m 44s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/69fdad48523a498baddab29329b65867 : SUCCESS in 2h 41m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/d60d3af2eed142f2aafa6233b228d53d : SUCCESS in 1h 35m 35s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/32814622bac840ff8b39848212c7b337 : SUCCESS in 1h 25m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f14ae27052254455b69b77bab40d9f32 : SUCCESS in 28m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5d332e38bb2456aa3a8c525a8a553f1 : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9614bba3af744a5498263c1d64e10b24 : SUCCESS in 28m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c47fab3dd9794e7597fab5b08a5486e4 : SUCCESS in 24m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95a279bf20f541b5a047e31b02665475 : SUCCESS in 23m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047f541e4bf043a3887ea0339c790b9c : SUCCESS in 10m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a80adfd4aa0b4e979d565b67f8b756d5 : SUCCESS in 58m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7bc8b29d500142cab7ee8c0c2d75fd92 : SUCCESS in 1h 16m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ba625eedc2a47848d6bf103a7195aa0 : SUCCESS in 1h 23m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2ee79baab714487bba455e771d25f349 : SUCCESS in 1h 36m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c2bd407757347fa8ceb22b876cd5ce0 : SUCCESS in 1h 12m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3b5056b2dd2945b7bad66146fad100a2 : SUCCESS in 1h 15m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/81bb2ba027f5406ea1a4bdb2a3d2b781 : SUCCESS in 1h 06m 58s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4d297e1554c4d41bf4766f65c59565d : SUCCESS in 1h 09m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6f1893b35a2f4447bad0175a1b3a2587 : SUCCESS in 1h 02m 30s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9526839b1f424a90bcdadf6f50828a6a : FAILURE in 1h 29m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/612d7f4966874fdc917795670d28b869 : SUCCESS in 1h 20m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3c94fb46eedf41308e95afa84a3c8efe : SUCCESS in 1h 01m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/62bc3163b6724ba4aa9226049a6cc638 : SUCCESS in 58m 01s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8017e7f9fa4a4ca2ad0142edc1c32c4f : SUCCESS in 1h 40m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24de1cbba8694ce28ab9a7c96f918239 : SUCCESS in 57m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1678e8a0b99a4bc7b916232ec6cd5a71 : SUCCESS in 1h 11m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b9036b05312a40da821af1c0a3a5ec2e : SUCCESS in 37m 49s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ce71193cafef4888a926177af8259a74 : SUCCESS in 1h 22m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b5906e21ae3468382740cf17b6b76b8 : SUCCESS in 49m 02s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7aea9726494340fe8162c934f78d1792 : SUCCESS in 3h 58m 45s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fe3f6cea5669412685a3c6fb56cd3c19 : SUCCESS in 1h 25m 38s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c3e297a97ce4b38b6d56910c5e547dd : SUCCESS in 2h 01m 47s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e19d35e32bbb44229c3af01142712bf7 : SUCCESS in 2h 18m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fa9fdd90623c4ef7ae2f52264caae6b9 : FAILURE in 1h 44m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1e172125238904cfc4e822b57f79b9c6177995a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-15 07:41:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03d36a27781184148d491bfdd773c1f7b1ae1f55","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-16 08:44:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cdb4290b61e49debe16a125c2851a502630ce61b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 08:45:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94b244fc0cc6dca3d41e4ab6673d1394d1b73f84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:31:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/6f174ee76f144d16ada5530a1c83d071 : SUCCESS in 51m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51cbdc40f3b645e7be04bc6c78797df7 : SUCCESS in 14m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6022a6f0224b459189087f4aea7cd61d : SUCCESS in 26m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc343c2e9d2e4f00a1fe4130ac7d1ff0 : SUCCESS in 27m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7605eda0254afa8478fa1dfd3a7c5f : SUCCESS in 9m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a92f2e9ebcab4f9dbd31c3f1f84f7aed : SUCCESS in 55m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/41bda50175ce4317ae7e10fcff04ba04 : SUCCESS in 1h 24m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6507a44a054247bda5f9a55eb9b9dfa3 : SUCCESS in 1h 04m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8cfa60eaa5804363a47ec1c4a156c1db : SUCCESS in 1h 06m 58s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00887f43ac4e469a8db775af25a80139 : SUCCESS in 1h 09m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40ec829408064c86b9abc7ecd96d2f9d : SUCCESS in 1h 11m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/30b60046a9a24509ba4a8c661baf8808 : SUCCESS in 1h 20m 40s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6fa7e0174a44de9b13f9af37f4669aa : SUCCESS in 1h 31m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffe15e7b414242c4b47cee0266259298 : SUCCESS in 1h 13m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f124dad1983b40fca9cda76c157a99f2 : SUCCESS in 1h 22m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2428ff5341f40473ab08b6a434f773294fc1d966","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:31:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bc52a58da465fa28afbc358ba212774dab4cb7d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 10:34:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd35c3f84814bec96d86032772acc24 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6badfcd6503fdc51f1206667e4e863521397503b","revisions":{"6badfcd6503fdc51f1206667e4e863521397503b":{"kind":"REWORK","_number":1,"created":"2021-06-14 14:02:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/796221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/21/796221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/21/796221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/21/796221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/21/796221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/21/796221/1"}}},"commit":{"parents":[{"commit":"d10ee544b2619c7b23b5fa8f2be35a82a41a3459","subject":"[DHCP] Fix cleanup_deleted_ports method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d10ee544b2619c7b23b5fa8f2be35a82a41a3459"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-06 09:20:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-14 14:02:28.000000000","tz":0},"subject":"Provide the rpc_response_max_timeout parameter to sriov-agent","message":"Provide the rpc_response_max_timeout parameter to sriov-agent\n\nThe rpc_response_max_timeout parameter is used in comminucation over\nmessaging queue, thus should be available for sriov-agent which\ncommunicate with neutron-server over messaging queue.\n\nChange-Id: Ie6ae31e40488fd8f3d43e83b25e536a7dd9d938c\nCloses-Bug: #1930996\n(cherry picked from commit be43141a5ff6875a146746cb75f927445328c871)\n(cherry picked from commit 04f4d9d406b186d1fc0bda4fe436a567dbc4b70a)\n(cherry picked from commit 41e603b0c3824e9fb62de925c94fe8d5401994b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6badfcd6503fdc51f1206667e4e863521397503b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6badfcd6503fdc51f1206667e4e863521397503b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
