)]}'
{"id":"openstack%2Fneutron~797410","triplet_id":"openstack%2Fneutron~stable%2Ftrain~Ic08ca5b167db4884efa07112cf9a6c3637e1b827","project":"openstack/neutron","branch":"stable/train","topic":"bug/1923633-stable/wallaby-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"Ic08ca5b167db4884efa07112cf9a6c3637e1b827","subject":"Force to close http connection after notify about HA router status","status":"MERGED","created":"2021-06-22 14:37:53.000000000","updated":"2021-06-23 19:04:45.000000000","submitted":"2021-06-23 19:00:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797410-bug/1923633-stable/wallaby-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":797409,"cherry_pick_of_patch_set":1,"meta_rev_id":"5d8018e2053a4d8f58afe08a5f33e4e4a34134ab","_number":797410,"virtual_id_number":797410,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-23 19:00:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-22 14:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-06-23 11:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-06-23 11:36:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-22 14:49:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-06-22 15:05:17.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-22 18:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-22 19:03:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-06-23 11:36:06.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"1d62cf3f6455dc43f9e3dfa97a35b1edf7e26b60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-22 14:37:53.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"9cb9e74e4f975734b319222b0acd6b28cfac8802","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-22 14:49:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"47f6fb0c092aef9f837cdaa1d015df8d3120f8a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-22 15:05:17.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/10/797410/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/10/797410/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0552899f40c99f8ace5d77874d92d91ed65f09aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 18:34:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/a41153fa1446450bbce75c16c16c4a40 : SUCCESS in 2h 41m 34s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/5023f3ab80684a66b10dee168b54d4a6 : SUCCESS in 1h 43m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/73d05a17bbe241e2b72e6c1f700291d3 : SUCCESS in 1h 47m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e445a0ca1804d74a1bf73a1af335903 : SUCCESS in 27m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f89a2f8abf4902a95fb6c67c8e55a8 : SUCCESS in 11m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc8885ea38f04331a603040a6345861e : SUCCESS in 24m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d9684385cf446adbe5a407bfe40c562 : SUCCESS in 26m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dea003479fb241f1a3f34a36d8b22d6c : SUCCESS in 24m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/764edb572bb64ba4bec4feabb016ba72 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/3e52e264515443bf8885235d5fef94c7 : SUCCESS in 56m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ca583e667a4849678da8d7fe48914fd8 : SUCCESS in 1h 16m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dce07d42e3b74c3995b50bdd174451ac : SUCCESS in 46m 07s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7d2411742248483f84d43662695bfa29 : SUCCESS in 42m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/23ccb52210bc48e9b6158d1ae392253b : SUCCESS in 1h 30m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5afe14594acd47c4be6ead1ac7d6997b : SUCCESS in 1h 36m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5764904fb0474a65956e9054a6cbe966 : SUCCESS in 1h 15m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/eb61cce4058644e7862bce63c206f6bd : SUCCESS in 1h 13m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a5580b7912d446d8c8ce2fec59a50e3 : SUCCESS in 1h 03m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a61d86d088d2450db501746f8619ccf6 : SUCCESS in 1h 00m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/08ee3b550ba3455b9ff94a499bba5f7c : SUCCESS in 1h 49m 18s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/192bde23da1a414aae644484403283b2 : FAILURE in 1h 37m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/50391b7046dd4599a7c994122ec69a2a : SUCCESS in 57m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5d917abc4ab94b19ade72954c0c0bb15 : SUCCESS in 56m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/913ac97c6d7e426baae56b51ef75eadd : SUCCESS in 1h 05m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85cd3bfd35c84a7592bebf56075a76f8 : SUCCESS in 1h 34m 19s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/8f0f355c442341339a8b9c3d8089a90f : SUCCESS in 2h 05m 27s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/9590c0f729f84c39b6537da6a30e201b : FAILURE in 20m 31s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6ebea5fb1f254357a57402d6a8a16fe3 : SUCCESS in 1h 03m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8db186855c82496da7d4397b37a65491 : SUCCESS in 1h 09m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/412f21c42916422598751a0180a181d3 : SUCCESS in 1h 27m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d309fff9167c4c95b5d243a4f8166e87 : SUCCESS in 48m 27s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/508124f4cc7e4150b793af9fa60da744 : SUCCESS in 1h 17m 41s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0927b74082e146579c90dad9f201f820 : SUCCESS in 1h 08m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f24f4c8c86690ddda8236e3881050f94a0e29242","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 19:03:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/797410/1 : SUCCESS in 2h 48m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9c9b7a4e89615efdbd831f127eebc602799c7e2f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-23 11:36:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee4b977a19f3e4cdd8a989d8ac7316d6bb4967b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 11:37:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15abc236803d387155101010afd7ae8aac6e16db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:00:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74841e796a8a4da3aea802fc092f9250 : SUCCESS in 15m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c8d6222a327438e9ba2e8fef5ad84b4 : SUCCESS in 26m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b64eaa17a194974a08891752bb85023 : SUCCESS in 23m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3aad316ade548a098698f85286443b3 : SUCCESS in 36m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c138bd21464cb4af76001101b0d435 : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/3bce9055edc14a48bc8ef5bc6868c2a5 : SUCCESS in 1h 21m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4cc0118218284a9081a19f0a845835f9 : SUCCESS in 1h 06m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0de4ff2947424a4e970bc98a55c367fa : SUCCESS in 52m 06s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/1278fb6e28114af89215901c06f0d348 : SUCCESS in 43m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0e0bfb991f504475a5bb48ce59e4ebd3 : SUCCESS in 1h 19m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5ba38ed3958d4f6698f1a2618cf5d625 : SUCCESS in 1h 29m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/25c4b4743f5941c5a71d32dc04f4a597 : SUCCESS in 1h 10m 47s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eef9622222044a6bb94b192589f42677 : SUCCESS in 1h 35m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e95cd1276b65437e86dccdb1695cc510 : SUCCESS in 1h 04m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7caf9ab311245f593835ce37bb75e68 : SUCCESS in 1h 08m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e0a35a72ffaf417ba0c837509e6c6eb9 : SUCCESS in 58m 02s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b9b30a6ce824f3f8b850e41fe4d9920 : SUCCESS in 1h 33m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bd2d3ceea49e4109ab41d40a5402ae2c : SUCCESS in 1h 02m 20s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6efffaf3d0446288eb64115fc6f642a : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e7c70f5f95e076953fa0854a3b1227aa050c01c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:00:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5d8018e2053a4d8f58afe08a5f33e4e4a34134ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:04:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760cf23eed1d4cc69b258935e0d03380 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"939ad7016f0920530cdd937cb79649d886b65ade","revisions":{"939ad7016f0920530cdd937cb79649d886b65ade":{"kind":"REWORK","_number":1,"created":"2021-06-22 14:37:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/10/797410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/797410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/797410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/797410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/797410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/797410/1"}}},"commit":{"parents":[{"commit":"6ca1be0c1bf1c5d0ff54a6ea8698c3b53c62cf0b","subject":"Merge \"Add a single option to override the default hypervisor name\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ca1be0c1bf1c5d0ff54a6ea8698c3b53c62cf0b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-17 13:40:20.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-22 14:37:53.000000000","tz":0},"subject":"Force to close http connection after notify about HA router status","message":"Force to close http connection after notify about HA router status\n\nNeutron-keepalived-state-change-monitor process is using HTTP to notify\nL3 agent that state of the HA router was changed. It is done through\nunix socket.\nWith HTTP 1.1 connection isn\u0027t closed properly and if\nrouter\u0027s state was changed more than once in short time, it could happen\nthat wsgi worker on the L3 agent\u0027s side didn\u0027t process second request at\nall. That caused problem with transitioning router to master state after\ncreation.\nRequest to close connection should be explicitly defined in the header\n[1] and this patch adds \"connection: close\" header to the headers send in\nsuch request to do exactly that.\n\n[1] https://www.geeksforgeeks.org/http-headers-connection/\n\nCloses-Bug: #1923633\nChange-Id: Ic08ca5b167db4884efa07112cf9a6c3637e1b827\n(cherry picked from commit 976cba61331702e1a8e4781af478fa4676afd13d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/939ad7016f0920530cdd937cb79649d886b65ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/939ad7016f0920530cdd937cb79649d886b65ade"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
