)]}'
{"id":"openstack%2Fneutron~797588","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Iaecfda2700c5316cb25a93496d24ece366e40a4a","project":"openstack/neutron","branch":"stable/wallaby","topic":"subnet_create_opt-stable/wallaby","hashtags":[],"change_id":"Iaecfda2700c5316cb25a93496d24ece366e40a4a","subject":"Improve Subnet create performance","status":"MERGED","created":"2021-06-23 15:35:16.000000000","updated":"2021-06-24 21:30:02.000000000","submitted":"2021-06-24 21:28:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797588-subnet_create_opt-stable/wallaby","cherry_pick_of_change":786399,"cherry_pick_of_patch_set":6,"meta_rev_id":"9c88f61807d9a6f2c9dc855fff9168f8683981a1","_number":797588,"virtual_id_number":797588,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-24 21:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-24 16:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-06-24 14:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-06-24 16:21:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 16:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-23 18:01:47.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-06-23 18:16:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-23 18:44:56.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-24 14:39:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-24 16:21:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b767d49443afe885328c21d18cff753a8d3b02e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-23 15:35:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"249c84e5cb0fd51b5c5427efb95df9437ab93967","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 16:21:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c495793326cd4f1bb51dd353aaec1f52 : SUCCESS in 34m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/683cb2f8021745c3900d6d26498d7e19 : SUCCESS in 43m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b73353179402121c8d1a140dfc9296277e098e0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-23 18:01:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/797588/1 : SUCCESS in 2h 24m 37s","accounts_in_message":[],"_revision_number":1},{"id":"672a77acb1de2b15a1bc4c07e99bf8804c3626e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 18:16:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/af14bf0534d04befaed3d5274fbadf7b : SUCCESS in 54m 58s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/afd68ed9cd3e49a887b4eca72457faca : SUCCESS in 2h 20m 39s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/e4d17a3e060749ff8d248e54e152044d : SUCCESS in 2h 07m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/2388751d7e88404dbc4e9c2f805eec1a : SUCCESS in 1h 48m 09s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/9f3a811e0e1442629a8b99aabc4b8059 : SUCCESS in 2h 34m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d85fddcee0de4f48ac2a55a33ac64ae3 : SUCCESS in 41m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c9cc935511488c8a0a82eeac44728a : SUCCESS in 13m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69cb0383faea4ba19f8e133ce2ab18f6 : SUCCESS in 28m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c38df440be44a1fb01a0d3d6d12887f : SUCCESS in 33m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f3d6a89e85c46de8b06c8646cebf0b9 : SUCCESS in 32m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dea906ca4714202824c61970a51b363 : SUCCESS in 11m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e989d0ed55742de922e1445f9cd3eac : SUCCESS in 50m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f262c90b76d40f59a3e5cf91b4aaabc : SUCCESS in 1h 20m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/80c857b8313844fa990bf20e033a9e3d : SUCCESS in 1h 20m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a16165f39fc4e35bdd7a28315a8d254 : SUCCESS in 1h 01m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/18a8e47ad11648bdade3a5e9f5ae5455 : FAILURE in 1h 06m 56s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f61711bf388b4698b30481dc3c668257 : SUCCESS in 1h 20m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a404c9aaa6a43548c24fd13758c93ec : SUCCESS in 1h 28m 47s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c4cdb2c01ce446109b584c9c493a27c6 : SUCCESS in 1h 49m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09667d52bced446db075480cd4d7e40f : SUCCESS in 1h 10m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6b68688f365249828f99a22f55bc3f04 : SUCCESS in 1h 20m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/00f061f3f4e74f4ea8fd26147ef4df07 : SUCCESS in 37m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/967716e759314d77b8bc0cecf4580382 : SUCCESS in 1h 21m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67c32b9c368547878575ab54f5b6fabb : SUCCESS in 1h 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"44a5327ad4243ba894197f607f91434684a20ee1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-23 18:44:56.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/88/797588/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/88/797588/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7b27f6171f2960c511f730191f7263a777487228","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-24 14:39:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"35a5f987d44130da560532dbac531321d338aac3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-24 16:21:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ba4aa5df2186a1cedb6fa9607791461ebdf29c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 16:23:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c11c76f5e2e658276ec3c7ae46525dbd4a3245b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 21:28:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/df5dd6485e6e4d1eadf763abe241df96 : SUCCESS in 1h 01m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07282efc42b496a944f332c86068d9e : SUCCESS in 15m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32cbe3dd29684f5987ca0766d77d0201 : SUCCESS in 31m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b8e14372369476eb45fb8a13d134465 : SUCCESS in 23m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/253c170307194ebfbf7b5c7f9ea626fb : SUCCESS in 12m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f3ae6ddd5da4c89882c347a33caa949 : SUCCESS in 52m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3a85f3f393f43cab6b34a64fb5cc70b : SUCCESS in 1h 24m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3510481b0eed4592afcf21ce41fcfdc7 : SUCCESS in 1h 14m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/179965df69c34ea586bdcd9f9fa5639b : SUCCESS in 1h 05m 21s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3fde3e36a61d452ca8f4f104508ee092 : SUCCESS in 1h 40m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2141f63083f94f8f9842f922faca9ca2 : SUCCESS in 1h 08m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0d52ebfd000b48b0adafe23220017fbc : SUCCESS in 1h 08m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/52941b67467d48d1b7feb6649d4431b7 : SUCCESS in 42m 10s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8c39a3c0594846f885dd6b49d1dd9468 : SUCCESS in 1h 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"79fe82fb6dac594b22abeaab62edf9435f1ba650","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 21:28:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9c88f61807d9a6f2c9dc855fff9168f8683981a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 21:30:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55df5453949d4b4ea953000d1a2e4bf8 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9afbe8b65f1d2b9dde60f7bc1cd2961c113c7bb3","revisions":{"9afbe8b65f1d2b9dde60f7bc1cd2961c113c7bb3":{"kind":"REWORK","_number":1,"created":"2021-06-23 15:35:16.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/88/797588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/797588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/797588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/797588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/797588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/797588/1"}}},"commit":{"parents":[{"commit":"d25dcf360ec5a1d1fb5fd8091bfccb3c91791067","subject":"Merge \"Add a single option to override the default hypervisor name\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d25dcf360ec5a1d1fb5fd8091bfccb3c91791067"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-04-15 12:45:16.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-06-23 15:35:16.000000000","tz":0},"subject":"Improve Subnet create performance","message":"Improve Subnet create performance\n\n- pass network dict from ml2 plugin to _create_subnet_postcommit\n- skip ipam subnet fetch for non ipv6 auto-address subnets\n- don\u0027t count subnets (DB request) if subnet has no segment\n\nChange-Id: Iaecfda2700c5316cb25a93496d24ece366e40a4a\n(cherry picked from commit f52280287f956f67c7b9c5c53ad945d96dd69103)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9afbe8b65f1d2b9dde60f7bc1cd2961c113c7bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9afbe8b65f1d2b9dde60f7bc1cd2961c113c7bb3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
