)]}'
{"id":"openstack%2Fneutron~798718","triplet_id":"openstack%2Fneutron~master~I58f6e5b309e089f6681d2c4bbff4ff7fda96435f","project":"openstack/neutron","branch":"master","topic":"bug/1933638","hashtags":[],"change_id":"I58f6e5b309e089f6681d2c4bbff4ff7fda96435f","subject":"[OVN] Do not fail when processing SG rule deletion","status":"MERGED","created":"2021-06-29 17:00:01.000000000","updated":"2021-06-30 15:26:29.000000000","submitted":"2021-06-30 05:03:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798718-bug/1933638","meta_rev_id":"fd819888d4898c00ff591999880d046ada3a5fa0","_number":798718,"virtual_id_number":798718,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 05:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-06-29 20:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-06-30 02:54:06.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-06-30 02:54:06.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 17:18:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-06-29 17:45:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-29 18:14:54.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-29 18:52:28.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-06-29 19:36:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 20:49:00.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-29 22:18:39.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-30 02:54:06.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-06-30 15:21:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"11e20c43425cbfbaf0f5cd394c38877ea622a845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-29 17:00:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5a4ee03e2b9630b723356b7b6b49bca7ca2386e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 17:45:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d382482b47a4dcb8669f2451b62c30e : SUCCESS in 42m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0df3722791df4b56840f5c072052922f : SUCCESS in 43m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9fd05ab046282448722d24541e227f5b6bd612a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-29 18:07:51.000000000","message":"Patch Set 1:\n\nI created a test patch for this in Octavia to make sure it solves the issue:\n\nhttps://review.opendev.org/c/openstack/octavia/+/798728\n\nWe should also add a unit test case that raises and exception in this case.","accounts_in_message":[],"_revision_number":1},{"id":"2bd57ea05edc57243b9e3dceff7b473ca145b1bf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-29 18:14:54.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/18/798718/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7e17f4dfd971b3864436523c7db267962a7848dd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-29 18:52:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/798718/1 : SUCCESS in 1h 51m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a84bad4b881321bfc80899f99328d9b2ddbeae26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 19:36:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8eb73b40140c442694fadf7337adf33c : SUCCESS in 45m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/70ee5769c4d440a9be1ff10c324f60f4 : SUCCESS in 1h 54m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a606a3bbc4824d86848d79a0fa1c0caa : SUCCESS in 42m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87c14da4cb0c4bcdb9530cd4f4bb8acb : SUCCESS in 39m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f1684967a44872a59904413d957421 : SUCCESS in 16m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb0f6eff642b466f89ba012694c3a9f6 : SUCCESS in 29m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aebb4560c78242529f0a28afc8df7a43 : SUCCESS in 27m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f89d00f7d3b242439c65d349385ea594 : SUCCESS in 28m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd92e8379db4fb7b1b795c392b05a8e : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06d72aa972a746daabf7fa32a3decd27 : SUCCESS in 51m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eb6b95888f56417fa3ce752f7379462e : SUCCESS in 59m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5d64cc8e36564e7f8c6d8bfed751c6a9 : SUCCESS in 1h 09m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/090dea1850dd427da08deecbd275da22 : SUCCESS in 1h 36m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/33b9d3ad79944739b05fe4ba1dc33bc8 : SUCCESS in 1h 20m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffbc329034df576e3c41d537013469cf9ce43cea","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-29 20:49:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fbeeef38750dce28aca8e8c6c0aa359c5a1e984d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-29 22:18:39.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/18/798718/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/18/798718/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"0d37eba221cae639f1b3bd1b5f81ceb4c73809a1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-30 02:54:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4690d2afc6fa8c37b7e600f270835d2460263ff2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 02:54:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9873881539494d5a18e9db0409543b0a63866622","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 05:03:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fb2603e1348e42e09adfe67689a3e068 : SUCCESS in 38m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1e0c80be01fb4a7c9f057c7e3dca1933 : SUCCESS in 2h 07m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fed28db02b30415c9e023eb7497b656d : SUCCESS in 37m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3acd9c21d8e8497d98abbcf0b5ad0192 : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf8ec0ea750c41419099eaa3e5ea9b1b : SUCCESS in 29m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72d89cd0ce74459895a5a100c954e644 : SUCCESS in 29m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecfb8f3f64ac44f58e7234fc405fed57 : SUCCESS in 12m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c86947a7bc1405389d017b03fdd133a : SUCCESS in 42m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cd99753804f84cd1b6064c9cf35b6938 : SUCCESS in 57m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2b65927fced3431aaa5e0368f46d497f : SUCCESS in 42m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1d84913162bc814d95060c15319f493fa80e2258","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 05:03:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"531b065d01a3c5586a8539c3486bc0fc20cc7335","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 05:04:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e20fcb4b83d4bf58a667555b70cf669 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fd80e92a1b474a6920eb0a6dca7c04fd36c64a51","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-30 15:21:56.000000000","message":"Patch Set 1:\n\nNo test coverage?\nThis broke the API contract by returning the wrong HTTP status code.","accounts_in_message":[],"_revision_number":1},{"id":"fd819888d4898c00ff591999880d046ada3a5fa0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-30 15:26:29.000000000","message":"Patch Set 1:\n\nI agree with Michael - there should have at least been a unit test, sorry for not putting a -1 with my comment to make that clear.  If there is a way to trigger this in tempest it should be done as well, I just wasn\u0027t able to do it via the client.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a74cd76fd006d7c88575fc7ae39a3b499ac7d54","revisions":{"6a74cd76fd006d7c88575fc7ae39a3b499ac7d54":{"kind":"REWORK","_number":1,"created":"2021-06-29 17:00:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/18/798718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/798718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/798718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/798718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/798718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/798718/1"}}},"commit":{"parents":[{"commit":"6196c0873b4af9abd9055bf92478ff05ae090104","subject":"Merge \"Refactor DHCP common config options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6196c0873b4af9abd9055bf92478ff05ae090104"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-29 16:48:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-29 16:48:26.000000000","tz":0},"subject":"[OVN] Do not fail when processing SG rule deletion","message":"[OVN] Do not fail when processing SG rule deletion\n\nWhen a security group rule deletion command is issued, before executing\nthe database operations, a \"BEFORE_DELETE\" event is raised.\n\nThe OVN handler attending to this event should not fail if the security\ngroup rule does not exist; the database transaction [1]  will in case of\nnot finding it, raising the correct exception and HTTP 404 error:\n\n  Jun 29 16:58:28 dev20 neutron-server[8820]: INFO neutron.wsgi [None \\\n    req-1821ec9f-2439-420b-80eb-1138896de865 demo admin] 192.168.10.70 \\\n    \"GET /v2.0/security-group-rules/missing_sg_rule_example HTTP/1.1\" \\\n    status: 404 len: 348 time: 0.0352871\n\n[1]https://github.com/openstack/neutron/blob/6196c0873b4af9abd9055bf92478ff05ae090104/neutron/db/securitygroups_db.py#L858-L868\n\nChange-Id: I58f6e5b309e089f6681d2c4bbff4ff7fda96435f\nCloses-Bug: #1933638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a74cd76fd006d7c88575fc7ae39a3b499ac7d54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a74cd76fd006d7c88575fc7ae39a3b499ac7d54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
