)]}'
{"id":"openstack%2Fneutron~799162","triplet_id":"openstack%2Fneutron~master~I5e1de571034be41f1147c130fce66e6cf70b1369","project":"openstack/neutron","branch":"master","topic":"keystoneless","hashtags":[],"change_id":"I5e1de571034be41f1147c130fce66e6cf70b1369","subject":"Add fake_project_id middleware for noauth","status":"MERGED","created":"2021-07-02 06:42:49.000000000","updated":"2021-07-06 08:24:15.000000000","submitted":"2021-07-06 08:21:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"799162-keystoneless","meta_rev_id":"9f1f0ba7a31fa54fa5a0ff507502a772acc4a528","_number":799162,"virtual_id_number":799162,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-06 08:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2021-07-06 06:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-07-05 15:57:41.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2021-07-06 06:05:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-02 06:52:09.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-02 06:56:08.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-02 06:56:08.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-07-02 07:01:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-02 07:37:19.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2021-07-02 07:50:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-02 09:02:32.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-02 10:01:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42092b047c44389e0854176afe1ea5af71badca9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-02 06:42:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8dacb041e0bbf3d40a1a8ee9d6abe6cb3145025","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-02 06:51:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9e54d1672c22a4096e275a0a5298d3698571904","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-02 06:52:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/799162/1/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/799162/1/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"f2436b46ef270bbee51c16f90e5e01fb960a5228","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-02 07:01:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c1b1dd4ff2aadde9fa65e7765f35f72728a0d078","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-02 07:05:11.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e890f646336094677dfec8544c7f32fb21287e17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-02 07:09:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c695e5c1830e159c527c5267a7dcd8ed48dc623e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-02 07:09:56.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/799162/2/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/799162/2/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"c123a71cf0283e7e0d1e98f8f71bc68e00c5db62","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-02 07:25:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa2fa7cf573897b3dbfbfbb68db6bd5626083365","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-07-02 07:37:19.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69699a6d0df2717205b7c679cdb2998105ed891c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 07:50:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03fb8bab576543198e580bbe30e37a8e : SUCCESS in 39m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50303d41367c4d9c8aef16506c38cedb : SUCCESS in 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6e97d6528f79ec6bfb7a5599f70ff34ed0c7847","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-02 08:12:11.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/799162/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/799162/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"a8b3c082fd999deecb3d10ca6b4ac0ab3e1532a2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-02 09:02:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799162/3 : SUCCESS in 1h 52m 34s","accounts_in_message":[],"_revision_number":3},{"id":"a7f3c93d52bc93021d0450f0998d230a0b62fd38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 10:01:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/906d14386d8741aab8deb49172fdfd81 : SUCCESS in 55m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e93af76ce40e413182714641625ab4c5 : SUCCESS in 2h 34m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8d6fbae0957f49ab8b28765194982134 : SUCCESS in 2h 19m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/94796e8013384c02afe4c3ff13dd24b6 : SUCCESS in 1h 35m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5de79a8ccb964a448206a2d810fb9b42 : SUCCESS in 1h 37m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/535942a45872423bb0140daeca2ed721 : SUCCESS in 27m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba4882cfe2454987b4a50577d9d2cbdf : SUCCESS in 37m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea1a3299882942c38e32c3a2829dbf93 : SUCCESS in 41m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b99a0339794da4a41984dd2097c713 : SUCCESS in 15m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94c04372df484e83bff931f59cfad073 : SUCCESS in 28m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c000cbe976d4da69c767ddd2cb44028 : SUCCESS in 25m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21d7f917a1b84b22a6cfdfec88bf83d0 : SUCCESS in 24m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc20a81b4dc40a9925456d485964580 : SUCCESS in 11m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a05633cdea14c51974e5e4609fde811 : SUCCESS in 58m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/00004af55a0a4439a5d919bf3627bd94 : FAILURE in 54m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/63d75b986ec6466184ab57259b560f9d : SUCCESS in 1h 31m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c4685f402e1494a8f70ae91caa0133d : SUCCESS in 1h 05m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/09bd3536781e4fd286071dc825669686 : SUCCESS in 1h 19m 06s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c365d3ff6e0b4ad580561b2621aee68f : SUCCESS in 1h 07m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7a4b1ec40a974ab383fcad5567c4d7ef : FAILURE in 1h 18m 52s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/ee34af5e5f7f4b979f854daa506297f6 : SUCCESS in 1h 29m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/508ef1f95c2d451a80bc0e4dbcd73f8d : SUCCESS in 58m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/095358987fac43238e0911dd8e7b3c6b : SUCCESS in 1h 08m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c0f3eecc6a2a4b4695b6f8466086464b : SUCCESS in 43m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa22629601d34284aac571c5f09cbe63 : SUCCESS in 1h 19m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b802e655c6474c618e682289413c8346 : SUCCESS in 1h 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ae36b4322b52466ba078332d7e0967016a2ece5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-02 10:46:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fcc24533b671271d4cfb40a028a52f9f980d9b94","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-02 10:50:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61319d299d1d6036b6338a6e0089f6ace3f1e94d","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-02 11:04:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"810b605e7cc185100e30f979376b5dd75797e8c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 11:29:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/782009b411734b4aadc136c4cba6e686 : SUCCESS in 41m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e1597eb81db41a58a11d6b96c485587 : SUCCESS in 40m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"801476f0ffd1d62d284a88535df508813e26c4d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-02 11:57:41.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/799162/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/799162/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"1fffc129aa98c896f8827fe58c8b8af3e7126980","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-02 12:37:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799162/4 : SUCCESS in 1h 49m 51s","accounts_in_message":[],"_revision_number":4},{"id":"09b77de80ee7f95cd89d0ee79565edb8d7e12a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 13:01:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5897081d784a44c2a5f987988f037449 : SUCCESS in 35m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2d9ca390efda4e7288dbec0cc606e9d0 : SUCCESS in 1h 40m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/33819475f52f4cc4852452725eeab43b : SUCCESS in 2h 12m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/31d8f47ab1a2400e89b5f260df04920f : SUCCESS in 1h 12m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bb0c4d9f81024c0f920818f2fb4c2139 : SUCCESS in 1h 58m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11ae6e36038f4d468bc957af1ed7e261 : SUCCESS in 27m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40d0cefdbfc14ae89538fdd8c338075f : SUCCESS in 37m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca752720b1ba4af3b3782d34583761a4 : SUCCESS in 36m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29ef910ab4548feba19dfb46d9e9ca0 : SUCCESS in 15m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb1d1fce57cd4b9b8b963d96c58e056b : SUCCESS in 25m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ec593bd0e9548bab01d3ba8fa36271d : SUCCESS in 23m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b618e89aaae3487cb87126f9a1d41899 : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0d4cedc08345dc868ce9917aa183d5 : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/92f889a7837b4fb8a88cf5a1f6f0c9b0 : SUCCESS in 44m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a10d77f2eab1489db9c6b341a02810eb : SUCCESS in 1h 25m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d55a18c01c044daab9195615d683ade2 : SUCCESS in 1h 08m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7ff1f178602474fa755552d2c8011bf : SUCCESS in 58m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/245f7578f1cf4f468566b23d33af0ead : SUCCESS in 1h 02m 54s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eb5aec70a2824561bd02485d4ec25569 : SUCCESS in 1h 06m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1d00ac10278c4cc2be9c9a253b8a6835 : SUCCESS in 1h 11m 23s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/0e6b838e18714990b369f6f348cb9797 : SUCCESS in 1h 26m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a06b9dd8b9ac43e1ac952e4b1e5e7bcb : SUCCESS in 49m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4739a58971ea449896296230be382176 : SUCCESS in 56m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/92ac92acbf994ea4bca8aa04ba281755 : SUCCESS in 31m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/47ec72aeb0274b8b9ba46ea0cd27a4cc : SUCCESS in 1h 08m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/36ad58906e2f4e32bea18c8278dc806d : SUCCESS in 1h 12m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"62a4b2d8a0f2c4ab33bb80ca886e585bc4f7c33f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-02 19:37:47.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nI think that this deserves release note too","accounts_in_message":[],"_revision_number":4},{"id":"92e73444f04bd5d7285405e404cd1ab9d5ea3b61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-03 02:47:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bd04135dc0c3f5b23821c5f58604529f88f100a6","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-03 02:50:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cab6da55aede4a2c07c3a5ec22622edf3fda9919","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-03 04:42:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799162/5 : SUCCESS in 1h 54m 51s","accounts_in_message":[],"_revision_number":5},{"id":"024368ff5acd301068c8aa02b3e416e2ce7a7276","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-03 04:48:43.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/799162/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/799162/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"166ce72305d84c9b02f4291afff6c43f659e1338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 04:52:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d977079118da4b7fa5d6e1854313120b : SUCCESS in 42m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3ab6eb1f57914f8a9a8921960e03b4c4 : SUCCESS in 1h 55m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/333ffe09aad945fdad0c7b86b0dde35b : SUCCESS in 1h 41m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a94948cd49814bdca659308320f77fe3 : SUCCESS in 1h 03m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3137ce58c6544d7aba728861acc70b2b : SUCCESS in 1h 59m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc58a7ac55bb4c16931fe844f47f9c1c : TIMED_OUT in 1h 04m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d636958cc9c494a91b5b5666cf3495e : SUCCESS in 40m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2161f47c21c643b2a07ef1b5d2eb0e44 : SUCCESS in 40m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca0f099e982348aa9759ac5b3afcfcd2 : SUCCESS in 11m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52856f155d564f278fa7f25646affb19 : SUCCESS in 22m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f18f601bd82642e5aeaaf9d1f593ce3b : SUCCESS in 20m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28023a56a5e149fabc74bec6e9ce7ab8 : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/187571f9ef284d06bf7ebfafa92278b9 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16342fbc16414de4972312d7a347da0f : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2edf5a2858ca4655a2a809be46aee478 : SUCCESS in 51m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/81b5401836794f3c8d492f918a7c82ac : SUCCESS in 1h 26m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e6cc5ee1f8f24c25a957c846ac09b2cc : SUCCESS in 1h 09m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1b19714338e419eb38da15a4e7bff74 : SUCCESS in 1h 05m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd42ee677e594f06a2a7ea4665e9b980 : SUCCESS in 1h 08m 41s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/03f4d2c5e44c42e5837a90accfd018a2 : SUCCESS in 1h 15m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e6779e2a35c344878e24310ed9a472a9 : SUCCESS in 1h 24m 14s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/df26787bedfd4b7ab6081ee2654c7cb5 : SUCCESS in 1h 33m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f644f6eafe924f448cdcf1419c69e45a : SUCCESS in 58m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2d1b76d3396d4d9caf3f944a0c80e694 : SUCCESS in 51m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c2e2ec8d24eb4498b9247457ebd12907 : SUCCESS in 32m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/12a9301974c84745b7bc1974a5dcca15 : SUCCESS in 1h 02m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f7269167c1474dbeada18e8290e01da8 : SUCCESS in 1h 09m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7ce9cfb761dba6e32121962373a2a8fc4223f13f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 05:05:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c43fd73970274f17b32260a35130dde8 : SUCCESS in 43m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6dc2f116dc514663a287768893bd1c8a : SUCCESS in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c78013e34d9d9c7b6110ea98125abec7594e086b","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-03 05:13:28.000000000","message":"Patch Set 5:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":5},{"id":"0e027050f614ba13d32fffe9eee63b137f2979aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 06:08:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c816a9a03ec44cde961203c45e6800aa : SUCCESS in 42m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b15274016ff45fd93bd77d2cf8ae721 : SUCCESS in 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"252dd46725e21d94f52cc86e520fea7eb6215e3a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-03 06:25:50.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/799162/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/799162/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"f779e146e80441f2cff94ff76099dfee0d26458c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-03 07:01:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799162/5 : SUCCESS in 1h 47m 16s","accounts_in_message":[],"_revision_number":5},{"id":"06cad9a2d0141e7f860b3ad6fae7183cd1d76de1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 07:39:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ebad38ab1c304390a045717d3643498a : SUCCESS in 47m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a6db0554bcb24125843d8cb226f2d462 : SUCCESS in 2h 19m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ced370d9b63542768123e7a41a686432 : SUCCESS in 1h 39m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3a5a72623eb74b55af56fa3590539bf7 : FAILURE in 1h 23m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a20f537651604452838890c3840d82ab : SUCCESS in 2h 06m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/968b7d18151a49eeba88af0ee807031b : SUCCESS in 33m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee378a7eed874c4daeac0a47c6e45e68 : SUCCESS in 41m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c5c9dfee75c4cb5861476c3ff657d5f : SUCCESS in 43m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851884ef8e75439eb79dae66425506c8 : SUCCESS in 12m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67cc2b4ef4534307ab51b66605dfc8ad : SUCCESS in 30m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea90ab737040495b8926845a09740c54 : SUCCESS in 26m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ef05470e488451b81aede73823cbb33 : SUCCESS in 29m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca8eec157bab45d8b12ba6ba8d1ad841 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90091efd99434a3fb53614c380300622 : SUCCESS in 10m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48a49a0491c2432f9648159be88a4666 : FAILURE in 54m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ae7f2b169c44134abe33bfceafe86f2 : SUCCESS in 1h 17m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ef8d8ee55b224d19ac084338656b4c8b : SUCCESS in 1h 17m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0316f73189ef41b08cfaab873221f2da : SUCCESS in 1h 06m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3d7ccd579b1549bfbefab8ae7a6bcb0e : SUCCESS in 58m 09s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3f94da4c42b14ca78eef80290f83daca : SUCCESS in 1h 13m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c345939cd6d64bf1b43df4cf23b9eb9f : FAILURE in 1h 03m 35s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/8cc1edea13ba4cf4a4ac4b279758193f : SUCCESS in 1h 35m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f6e618e0f0a45abb0a5b961742bfc36 : SUCCESS in 46m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dbe5133a7ec04c798d8ecbaf54b1c9d2 : SUCCESS in 53m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b3a4663d07f44a4b97b1b6947f6aa66c : SUCCESS in 33m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c930dc8c59f498b87ef4b1b7f49e5b0 : SUCCESS in 1h 07m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1cd475273d4143928a36415601e156a6 : SUCCESS in 1h 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"207f47f1629ea9b9e339c699c176d0e6ba70bb27","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-03 12:54:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e118d43b8b966f0c2ad8793ccfd7dbe37e683625","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 13:38:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c12a37a95d64aca93cbf5f921ceb50d : SUCCESS in 39m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c77b76b919b412b81de0b7f5deadc7d : SUCCESS in 41m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2e3f8a4879b5883538ce4ac195cf2440feae1635","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-03 14:06:50.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/799162/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/799162/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"86914db876c72ac71f6b4ec23b868452233b35bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 14:32:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0e41cad38db34196add6d749807839af : SUCCESS in 42m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6c1d1bdeb65546358042833c9b551a63 : SUCCESS in 1h 31m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5914356e9ec6403ebe70b75903eec4a6 : SUCCESS in 1h 27m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5052f49872c4a53a846dbfc9eb4413c : SUCCESS in 1h 20m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/77f76542eeed4264a410c2dab6f1275b : SUCCESS in 1h 22m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9148babdd1f4f91bad232026bfa577e : SUCCESS in 28m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ff85d9c0037472896fc29fde412be75 : SUCCESS in 37m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a9207a98e504295b894114e68559733 : SUCCESS in 35m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44db489965bf46d68a676b873ea38aee : SUCCESS in 13m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d19adad3f9cc47bba80a83817a2e4b01 : SUCCESS in 27m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfee93b7533d4e8285cb2a48424e1364 : SUCCESS in 21m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edd53bd9f532480cbcaf5a6cb1f13d88 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb6941792314506b79973e6df1ab8f6 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af912c8d3a6a441b8bb6a602f5a0865c : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc360dbbe9334c7fa5a7e9f8d8015f5f : SUCCESS in 39m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e11466856514d049a448d80402c43b6 : SUCCESS in 1h 16m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/da4e40410c894ff0afda9183289700cf : SUCCESS in 1h 02m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dc1c07da28040318733911b3b31bdb2 : SUCCESS in 55m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/64696030cc4c4a3488b9f2b00d89e047 : SUCCESS in 1h 01m 30s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bc5ff93a74734c2f83a3d6f07b278e2b : SUCCESS in 1h 09m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6691005f4e7140f0bc46635b7599d30b : SUCCESS in 1h 15m 43s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/4693b565b36544f086856d57ac4d7e7c : SUCCESS in 1h 24m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e6d94d9b0154cc4b155c5411fd9547c : SUCCESS in 50m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e9ee3a4323354088974631b9509a375c : SUCCESS in 53m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dbab889c979c40548fa4dffa0e193852 : SUCCESS in 25m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1a22ab504e0341fa83d3355936c8f437 : SUCCESS in 1h 15m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9028443414d946b391a532eab861b17a : SUCCESS in 1h 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"84a01af1e7f6345322c58d115e1e0e350fb985d8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-03 14:43:14.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799162/5 : SUCCESS in 1h 47m 45s","accounts_in_message":[],"_revision_number":5},{"id":"8c4554faa271a672c806098e95187075afe9d046","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-04 02:28:45.000000000","message":"Topic set to keystoneless","accounts_in_message":[],"_revision_number":5},{"id":"650a110dd88d2fda38a13526999cdcc7d5d08cf0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-05 10:40:39.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"08c25fa0acdd35a4b7cb9fadaaf1c578af069cba","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-05 10:58:16.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ab999fee023a1b29c399f557d162280434d40888","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-05 14:53:47.000000000","message":"Patch Set 5:\n\n(5 comments)\n\nThanks for the response. I checked both of the existing and new behaviors with noauth auth_strategy. It works well.\n\nI would like to have some more in the release note. I added a comment in the release note file based on my previous questions. I keep my (week) -1 due to this.\n\nIn addition, I have one thing on a context used for noauth. I would like to hear opinions from other reviewers. See my comment at L.48 in neutron/auth.py. (I don\u0027t think it is a blocker and we can improve this if needed.)","accounts_in_message":[],"_revision_number":5},{"id":"66fd814081cae56def3524a7bfbb39fd37093fb6","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-05 15:22:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"376f6376a4e47311f7f87460e2341581d9099063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-05 15:48:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ef22f9bcfb87ee7deaba7f501f7ec93e89c0e74f","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-05 15:57:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9c57095934ed17ed773bb4983db63c6a3c558ce9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-05 15:57:41.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks. I am fine with it.\n\nIf we hit any problem around \"POST\" method check in CI with the recent patch set, I will take care of it.","accounts_in_message":[],"_revision_number":6},{"id":"2005ed5bb9a693273b9a6159a443f5279ba8dfc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 16:30:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a755e24638d74cdfa3e7433786c9753b : SUCCESS in 39m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6e8afc7e6f24f77b2993e7f9b6a52f0 : SUCCESS in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1768b7773b7ec2b413d50ad2f2b3fba6ef400b75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-05 17:09:30.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/62/799162/6/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/62/799162/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"5e577de844d9a29ca0bb7988910999c31663d97e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-05 17:20:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799162/6 : SUCCESS in 1h 30m 48s","accounts_in_message":[],"_revision_number":6},{"id":"8fee63b0d85b58f98271fe6f63b32428724f024d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 18:18:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e980a1ed1fe646de8d5b741f77c93e8b : SUCCESS in 47m 55s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8cdafd6e068a49a9856876aed9de303a : SUCCESS in 2h 25m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64c8dd5f523945ac8a3d6eacc874ffd6 : SUCCESS in 2h 05m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6693fbba6774452a7f4aae4ceb18ac0 : SUCCESS in 1h 58m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/03144ec3af544086b5d2b5f429869c0c : SUCCESS in 2h 13m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/99e01f01bb3248bf8ba8bf073d84eaa9 : SUCCESS in 34m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a13430dda22f4a0ca0ccfb7f4e008039 : SUCCESS in 35m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3677a210a894bd8bf55804da28f8e90 : SUCCESS in 40m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4127f9cadf9243ef8d16f64e0f0c4091 : SUCCESS in 16m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5fb5266f95846328f7d58cc41950ad0 : SUCCESS in 28m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/010456e2ba09486a8c97bc35a75e0ee8 : SUCCESS in 28m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50f06a0015dc4e3ca07ca983a8c7cebe : SUCCESS in 28m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e465f5670584284b9c645abd05a684a : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d355ae604c4e418694de213ab24008c0 : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95b849cf4fbf4855adef8c086a9c12d1 : SUCCESS in 49m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6462b72a7f2442b9a82dfc95a50e6626 : SUCCESS in 1h 20m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f235810002d844d791412f4999a6aca2 : SUCCESS in 1h 03m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0a12828d03e48a4bf09aaa14f945b56 : SUCCESS in 58m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/589177f2841b4f11b365c0bbf596dbd5 : SUCCESS in 59m 39s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/50ccfeb8558c45e8a34c69c3c7c46e8c : SUCCESS in 1h 12m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2bb96d5c5cd04edea012235c0da5bc42 : SUCCESS in 1h 11m 40s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/46f8d2e4444849a5957532bcbbe7db19 : SUCCESS in 1h 24m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fdf016ff02747838b102ffe39a71207 : SUCCESS in 50m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4addc0fb9d9e4c79b2e42a2184f4baa9 : SUCCESS in 50m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/baead863ddb44694a7837a3928936fb3 : SUCCESS in 31m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1f7d47a5f5a4170a2c56688ca69f5ad : SUCCESS in 1h 08m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c4ddd0e7d1364b1f9ad6e9781618e685 : SUCCESS in 1h 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"984b17d3dd06d81fa3e95a1c083c0328e57d611e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-05 19:48:18.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d6ce4816b8ce1eb54e7f62e704d68b44b66c1376","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-07-06 02:02:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"82c1788accab50e50656e035b64a38d5a7529778","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-06 06:05:20.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"06245ce7b7cd79c6b52b1248f6f932433d5da127","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 06:05:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"57359f4f14cdd7d02878040b18fcdb15342ebdb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 08:21:31.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7dd830b99abd46cd9878ce761f2378ac : SUCCESS in 48m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cda3457ca03743c096286caa907e327d : SUCCESS in 2h 10m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7a45785ffb654b9f9c3bf532174238c0 : SUCCESS in 1h 55m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/96c7543c77924080811e97ea38b38868 : SUCCESS in 1h 20m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e29d6739355e4dba80cfdd4c8c92c02f : SUCCESS in 1h 13m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c3d2519174044608a2d2d745d640784 : SUCCESS in 40m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1686a1d9fb142899e7cbd22afa805bb : SUCCESS in 15m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13cec9e3d5564abf96686d7c49f60217 : SUCCESS in 28m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8206e6808aa34810a9220339b2797133 : SUCCESS in 29m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5ffd882ada440a8af1bfbb5c1eda7a2 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef62940ff9064747b8673634ae6b713e : SUCCESS in 12m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd5931ceaf8d464a9da95286157b73d7 : SUCCESS in 48m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7c09cd0452148e78fc661f64c263e9f : SUCCESS in 1h 26m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5beb23f8f800476781b3fb70cab609b6 : SUCCESS in 1h 19m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/53c293be0c5a459e913055d4a5de5e6c : SUCCESS in 1h 06m 42s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/3d6efb20b65e46839708c4542bfc59a7 : SUCCESS in 1h 19m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf0931f8f5ef4fd4b24b842c22346154 : SUCCESS in 58m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c8dcba85d8834ac38478cbf65a48f1ec : SUCCESS in 59m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f38ed5ce79db401c91b48e33705fc673 : SUCCESS in 34m 54s","accounts_in_message":[],"_revision_number":6},{"id":"9cd97ecf494662fbe3ee74d76740c93caea42051","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 08:21:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"9f1f0ba7a31fa54fa5a0ff507502a772acc4a528","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 08:24:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2219d8caba341e19a0faf12339a67b0 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8627799c037347b799da9b57ffc03883 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ec550f5f52c5377f57499eeadd7e0d665dcdb73d","revisions":{"26fc23ffb9bcab6b63b03d594f7f96afc8368ee7":{"kind":"REWORK","_number":1,"created":"2021-07-02 06:42:49.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/62/799162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/799162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/799162/1"}}},"commit":{"parents":[{"commit":"13e96bf6bd740a09283d9e2849fc2b6adbbf29e3","subject":"Merge \"Add extra logs to the network update callback in L3 agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13e96bf6bd740a09283d9e2849fc2b6adbbf29e3"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 05:37:22.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 06:42:44.000000000","tz":330},"subject":"Add fake_project_id middleware for noauth","message":"Add fake_project_id middleware for noauth\n\nThis adds a middleware for noauth that would inject a fake\nproject_id for create requests. This would ensure that api\nconsumers don\u0027t have to provide a fake project_id in requests.\n\nCloses-Bug: #1934039\nChange-Id: I5e1de571034be41f1147c130fce66e6cf70b1369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26fc23ffb9bcab6b63b03d594f7f96afc8368ee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26fc23ffb9bcab6b63b03d594f7f96afc8368ee7"}]},"branch":"refs/heads/master"},"7e24fdebe87fb82bfe253cecace9aa242828a2dd":{"kind":"REWORK","_number":2,"created":"2021-07-02 06:51:55.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/62/799162/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/799162/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/799162/2"}}},"commit":{"parents":[{"commit":"13e96bf6bd740a09283d9e2849fc2b6adbbf29e3","subject":"Merge \"Add extra logs to the network update callback in L3 agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13e96bf6bd740a09283d9e2849fc2b6adbbf29e3"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 05:37:22.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 06:47:59.000000000","tz":330},"subject":"Add fake_project_id middleware for noauth","message":"Add fake_project_id middleware for noauth\n\nThis adds a middleware for noauth that would inject a fake\nproject_id for create requests. This would ensure that api\nconsumers don\u0027t have to provide a fake project_id in requests.\n\nCloses-Bug: #1934039\nChange-Id: I5e1de571034be41f1147c130fce66e6cf70b1369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e24fdebe87fb82bfe253cecace9aa242828a2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e24fdebe87fb82bfe253cecace9aa242828a2dd"}]},"branch":"refs/heads/master"},"f44796656c244055983b3504d83939f94dd198e8":{"kind":"REWORK","_number":3,"created":"2021-07-02 07:09:17.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/62/799162/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/799162/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/799162/3"}}},"commit":{"parents":[{"commit":"13e96bf6bd740a09283d9e2849fc2b6adbbf29e3","subject":"Merge \"Add extra logs to the network update callback in L3 agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13e96bf6bd740a09283d9e2849fc2b6adbbf29e3"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 05:37:22.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 07:09:02.000000000","tz":330},"subject":"Add fake_project_id middleware for noauth","message":"Add fake_project_id middleware for noauth\n\nThis adds a middleware for noauth that would inject a fake\nproject_id for create requests. This would ensure that api\nconsumers don\u0027t have to provide a fake project_id in requests.\n\nCloses-Bug: #1934039\nChange-Id: I5e1de571034be41f1147c130fce66e6cf70b1369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f44796656c244055983b3504d83939f94dd198e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f44796656c244055983b3504d83939f94dd198e8"}]},"branch":"refs/heads/master"},"6d3594baa541a8891f45f1d1e6539a8fa82c234b":{"kind":"REWORK","_number":4,"created":"2021-07-02 10:46:31.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/62/799162/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/799162/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/799162/4"}}},"commit":{"parents":[{"commit":"13e96bf6bd740a09283d9e2849fc2b6adbbf29e3","subject":"Merge \"Add extra logs to the network update callback in L3 agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13e96bf6bd740a09283d9e2849fc2b6adbbf29e3"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 05:37:22.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 10:45:36.000000000","tz":330},"subject":"Add fake_project_id middleware for noauth","message":"Add fake_project_id middleware for noauth\n\nThis adds a middleware for noauth that would inject a fake\nproject_id for create requests. This would ensure that api\nconsumers don\u0027t have to provide a fake project_id in requests.\n\nCloses-Bug: #1934039\nChange-Id: I5e1de571034be41f1147c130fce66e6cf70b1369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d3594baa541a8891f45f1d1e6539a8fa82c234b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d3594baa541a8891f45f1d1e6539a8fa82c234b"}]},"branch":"refs/heads/master"},"70856e829b7ccdd823b4b28953832d30082a2516":{"kind":"REWORK","_number":5,"created":"2021-07-03 02:47:19.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/62/799162/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/799162/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/799162/5"}}},"commit":{"parents":[{"commit":"13e96bf6bd740a09283d9e2849fc2b6adbbf29e3","subject":"Merge \"Add extra logs to the network update callback in L3 agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13e96bf6bd740a09283d9e2849fc2b6adbbf29e3"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 05:37:22.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-03 02:46:46.000000000","tz":330},"subject":"Add fake_project_id middleware for noauth","message":"Add fake_project_id middleware for noauth\n\nThis adds a middleware for noauth that would inject a fake\nproject_id for create requests. This would ensure that api\nconsumers don\u0027t have to provide a fake project_id in requests.\n\nCloses-Bug: #1934039\nChange-Id: I5e1de571034be41f1147c130fce66e6cf70b1369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70856e829b7ccdd823b4b28953832d30082a2516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70856e829b7ccdd823b4b28953832d30082a2516"}]},"branch":"refs/heads/master"},"ec550f5f52c5377f57499eeadd7e0d665dcdb73d":{"kind":"REWORK","_number":6,"created":"2021-07-05 15:48:32.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/62/799162/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/799162/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/799162/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/799162/6"}}},"commit":{"parents":[{"commit":"13e96bf6bd740a09283d9e2849fc2b6adbbf29e3","subject":"Merge \"Add extra logs to the network update callback in L3 agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13e96bf6bd740a09283d9e2849fc2b6adbbf29e3"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-02 05:37:22.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-05 15:48:09.000000000","tz":330},"subject":"Add fake_project_id middleware for noauth","message":"Add fake_project_id middleware for noauth\n\nThis adds a middleware for noauth that would inject a fake\nproject_id for create requests. This would ensure that api\nconsumers don\u0027t have to provide a fake project_id in requests.\n\nCloses-Bug: #1934039\nChange-Id: I5e1de571034be41f1147c130fce66e6cf70b1369\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec550f5f52c5377f57499eeadd7e0d665dcdb73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec550f5f52c5377f57499eeadd7e0d665dcdb73d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
