)]}'
{"id":"openstack%2Fneutron~799210","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~I58f6e5b309e089f6681d2c4bbff4ff7fda96435f","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1933638","hashtags":[],"change_id":"I58f6e5b309e089f6681d2c4bbff4ff7fda96435f","subject":"[OVN] Do not fail when processing SG rule deletion","status":"MERGED","created":"2021-07-02 14:01:01.000000000","updated":"2021-07-05 14:19:11.000000000","submitted":"2021-07-05 14:16:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799210-bug/1933638","meta_rev_id":"5c94e1b8bf94665141b2da1075f1219b0ff4cee1","_number":799210,"virtual_id_number":799210,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-05 14:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-07-05 12:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-07-02 19:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-07-05 12:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-02 14:14:12.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-02 15:42:14.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-02 18:30:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-02 19:34:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-07-05 12:35:45.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"d94a639a0812f290fcc83d9a92f4e84e43125234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-02 14:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cd9532730132bf2a7b1e66a1316f909b2b4b3b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-02 14:14:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/10/799210/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/10/799210/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"91f4cfa080d2d8fbdb915dafadbab9aa44fd6999","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-02 15:42:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799210/1 : SUCCESS in 1h 40m 23s","accounts_in_message":[],"_revision_number":1},{"id":"706fc95497f59d3114d2c75cf7a9950c4aa69ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 18:30:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/a48231b2031249b59942d5541e804eec : SUCCESS in 44m 25s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/75b204efa55b4a0bb2b15cbeff723fea : SUCCESS in 2h 38m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06800c23773443dd92ad408551208a76 : SUCCESS in 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aba39f33c55415ebfdd454bcbfb591e : SUCCESS in 14m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee0a2f619047476dac55f85f8e919fad : SUCCESS in 23m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d93038ad22ac40189b6cb4f82b5aff3b : SUCCESS in 24m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea11c9242f14e268813d9ca3618f949 : SUCCESS in 12m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/12552d5e14d7476f8623844a99700537 : SUCCESS in 1h 09m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa01cbfa2d9745c7837d248008c29405 : SUCCESS in 1h 23m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8900e2dec2e943209b145c7e3a58919c : SUCCESS in 1h 24m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e415339123b468ea645d81390e8f6f0 : SUCCESS in 1h 04m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ca1a59f44ca2417194c5c85b0afcd6c3 : SUCCESS in 1h 03m 51s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/c36dc9a95b4d4edd9553e0bc9d3ad1a4 : SUCCESS in 1h 21m 25s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8830a87d3de245b6aaee943bd520b6b8 : SUCCESS in 1h 01m 17s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/758885a0e7d64c06a69e982b7a7a53e2 : SUCCESS in 1h 06m 03s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cd2b6ca3e6b7481a8afb8ac6ae70807c : SUCCESS in 1h 13m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/55839564bb644133a2045fdcf0c13e01 : SUCCESS in 1h 25m 21s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06c8cfbcc2b348fd94c879ae9168dbfb : SUCCESS in 58m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a4a8ee2949094263b9d7a973d30a53b8 : SUCCESS in 1h 27m 43s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66309e78ccdf4aa8b32f6588594fa280 : SUCCESS in 1h 33m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cf610431c58481c9d9e623f8f979023 : SUCCESS in 59m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d6259ef9d67d429a937999a7803d16b9 : SUCCESS in 1h 25m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4922677c1ebf40ffb131e47322c46b3a : FAILURE in 34m 54s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d1f0f0b0ba5f41bcb2229785e80078fc : SUCCESS in 1h 24m 41s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/33bdc243f5ba449aaa590faa7639e6da : SUCCESS in 3h 54m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4d622058cd7a4774aa8c6d8fb685f8af : SUCCESS in 1h 30m 55s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/96a4a7bcfc6545b58841e8c15fb6f774 : SUCCESS in 1h 34m 02s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e4f4d6e868f442d4a1470ffac0fcc20a : SUCCESS in 2h 20m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/496a2e64aea24e4aa31b2466868d9a00 : FAILURE in 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc8174b5627223d8971480a0d8bb3b9d4fb7ed95","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-02 19:34:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1849747045edcbb2329d3ce59e6082a938d8e8ff","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-07-05 12:35:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"570dc95e2b6df6d7c86352c71626e0fa878acc19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 12:36:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bf745b099d431fc0500ed2e0e49fe90722dddc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 14:16:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/2d3269085f3f4e23b066a1861497639c : SUCCESS in 44m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2e4d3282f84d578f861fcfb0f753ea : SUCCESS in 12m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0123a33e5f414ec0bd66fcae620be219 : SUCCESS in 22m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e617a422aba944e1ba82afb6d9473be9 : SUCCESS in 21m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a79cf1bffc04ffc8b1c9cb4c32f2e01 : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d1fbb72a87645a2881c94a5d18d68f4 : SUCCESS in 1h 05m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4eb8de2c320a49049f348599694913c7 : SUCCESS in 1h 19m 07s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a17b2bf93e44017b1e47df2c19cf9e5 : SUCCESS in 1h 06m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e4e2e2f67cce41d68324252c3afdd7db : SUCCESS in 57m 04s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9c7b4c4ed9cf43eeb8cda35df735064e : SUCCESS in 1h 01m 32s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/823bab0851824d05a393b51305a03e3f : SUCCESS in 1h 09m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40459fe87e5a4316a5680daf6d49117a : SUCCESS in 57m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9c50e8da81e4e3fa88d78c7e558f71d : SUCCESS in 51m 45s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aab311cef35744ee9b49d0359aaa17da : SUCCESS in 1h 29m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e37334a2c804fe190e15719b5b6dffd : SUCCESS in 54m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0152e8ce7a31411ab9cba013e61cb6af : SUCCESS in 1h 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ce9ff44c760f1ff24201381f65bd2bfda3628d10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 14:16:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5c94e1b8bf94665141b2da1075f1219b0ff4cee1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 14:19:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af0ac052d2ec46d49107bfaac6f3073c : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32c6a39a8cf35c75800c35ac5fe6b15ca99e3c34","revisions":{"32c6a39a8cf35c75800c35ac5fe6b15ca99e3c34":{"kind":"REWORK","_number":1,"created":"2021-07-02 14:01:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/799210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/799210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/799210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/799210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/799210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/799210/1"}}},"commit":{"parents":[{"commit":"3d9d30394736503efff265a84c4a555b44203300","subject":"Merge \"[Stable only] Set irrelevant-files in some additional jobs\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d9d30394736503efff265a84c4a555b44203300"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-29 16:48:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-02 14:00:44.000000000","tz":0},"subject":"[OVN] Do not fail when processing SG rule deletion","message":"[OVN] Do not fail when processing SG rule deletion\n\nWhen a security group rule deletion command is issued, before executing\nthe database operations, a \"BEFORE_DELETE\" event is raised.\n\nThe OVN handler attending to this event should not fail if the security\ngroup rule does not exist; the database transaction [1]  will in case of\nnot finding it, raising the correct exception and HTTP 404 error:\n\n  Jun 29 16:58:28 dev20 neutron-server[8820]: INFO neutron.wsgi [None \\\n    req-1821ec9f-2439-420b-80eb-1138896de865 demo admin] 192.168.10.70 \\\n    \"GET /v2.0/security-group-rules/missing_sg_rule_example HTTP/1.1\" \\\n    status: 404 len: 348 time: 0.0352871\n\n[1]https://github.com/openstack/neutron/blob/6196c0873b4af9abd9055bf92478ff05ae090104/neutron/db/securitygroups_db.py#L858-L868\n\nConflicts:\n      neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n\nChange-Id: I58f6e5b309e089f6681d2c4bbff4ff7fda96435f\nCloses-Bug: #1933638\n(cherry picked from commit 6a74cd76fd006d7c88575fc7ae39a3b499ac7d54)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32c6a39a8cf35c75800c35ac5fe6b15ca99e3c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32c6a39a8cf35c75800c35ac5fe6b15ca99e3c34"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
