)]}'
{"id":"openstack%2Fneutron~799439","triplet_id":"openstack%2Fneutron~master~I409bc674b65e4f495ebd42d03e97a09d51482339","project":"openstack/neutron","branch":"master","topic":"bug/1933273","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-04 10:25:51.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-04 14:20:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I409bc674b65e4f495ebd42d03e97a09d51482339","subject":"[DVR] Fix update of the MTU in the DVR HA routers","status":"MERGED","created":"2021-07-05 07:43:48.000000000","updated":"2021-11-04 14:32:31.000000000","submitted":"2021-11-04 14:20:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"799439-bug/1933273","meta_rev_id":"883d2fa8925311d46e116d2b61f942b9030fd4eb","_number":799439,"virtual_id_number":799439,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-10-13 13:36:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-04 14:20:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-11-04 10:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-19 07:10:36.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-11-04 10:25:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 08:26:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-05 09:15:09.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-05 09:54:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-05 11:49:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-07-22 23:22:37.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-09 07:21:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2021-08-09 07:21:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-09 07:21:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-10-25 12:39:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"ffdfdb16ff5986db6be4c9d6fd8f7979dab3647c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-05 07:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7bbb8c9378ea7109e266641ef19b92e8b58888f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 08:26:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41754193680f4b1f80629956040bb85d : SUCCESS in 40m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e525809430c4204996435d54b9a907e : SUCCESS in 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e6dd5df5ebe329075e72a325756c8371a8c95d2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-05 09:15:09.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/799439/1 : FAILURE in 1h 29m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6296c59d109e63e752ae1be4257aeaada90fc6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 09:54:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ebb3a070efc64c6a9ae35295d1d600fb : SUCCESS in 34m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c5022ab180be4d308d34d571abaf055e : SUCCESS in 2h 00m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/26ae894a0dee4fdfb0a7e73cd2998189 : SUCCESS in 1h 49m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f998cf6aab6a484a94a2e3378c46e829 : SUCCESS in 1h 44m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/329eae5f8ccd4afaaa10dbc1887959be : SUCCESS in 1h 29m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b062ea4496ec41d89a2597efa9d63962 : SUCCESS in 35m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e3648ddfd124de28adbf20408d139f5 : SUCCESS in 46m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1009e20d9ad943249c4272ad77eec6aa : SUCCESS in 43m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e4b8b94a94470b8260a22b94f4b74d : SUCCESS in 16m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dd1ae5f441e4a98acbe1c5daa0a75b9 : SUCCESS in 28m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23d44860f9d0473893ced441eef892d6 : SUCCESS in 29m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b0e01c2f6244f9a801642875c9488a2 : SUCCESS in 32m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7447775d9c944742824da6c0f09eaf42 : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c8f2bd2c4b94a318b65704b78820092 : SUCCESS in 55m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1604c7b424f94769a8d0f77ce77ee849 : SUCCESS in 1h 34m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a7c1208f8f39440dae2ab11c7ed5ea87 : SUCCESS in 1h 18m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/17d2757d63e44c9287c407d1c6dda55b : SUCCESS in 1h 04m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c38fd2ca2f36422fb25d668e5e7a058a : SUCCESS in 1h 01m 13s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/044591d551aa47358b13a1bdf34ca6a5 : SUCCESS in 1h 10m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ea5f00abe93c42d2af22422bcb82ddd0 : SUCCESS in 1h 09m 58s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/ee92d74d50584a55b685c271b71d3aba : SUCCESS in 1h 25m 34s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1dc55be14634f818b366a8cf9cf79d0 : SUCCESS in 54m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b96eb8e5c1ac429a8c2e4ddaa5cbed4c : SUCCESS in 52m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9086b7079e604672b7143df83f2a8c65 : SUCCESS in 32m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb3ed4341521457c8a45752abda22f01 : SUCCESS in 1h 19m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/79085b968f48423895dc596089ea4b13 : SUCCESS in 1h 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d0812038c640d72b35dc24143ec479046c49149","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-05 11:49:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d0555023d2cbc8fcee820f0e0925470afec3050","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-22 19:11:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d8ff8cdcf55f713ca58b619c7fd7e7ecad15c49a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-22 19:18:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bd03e990243f0b3009237329c2d684608b38a18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 19:58:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fabca6202ed4d09b99a0458fa361967 : SUCCESS in 45m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/153322e60d6049aeafeefb35d88a6163 : SUCCESS in 44m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d8839a75ca57888507ca0fa0772a1f47fac0378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 21:20:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/415a3c8a7c5449ed8b29c695407159fe : SUCCESS in 41m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/13af987dc0704ffe94206dfa0d5d735f : SUCCESS in 1h 49m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fbe03a8de8d94898918ea842eeaf4e92 : SUCCESS in 1h 32m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/253f27c180ab43b1b6ff86d0368b4996 : SUCCESS in 1h 50m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0cbeba3630eb4ffb83a949da198c21bf : SUCCESS in 2h 08m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/19dda89a14274cf49ebfdac340f17345 : SUCCESS in 35m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b5e45bd3b9745b5812f96af5a17c33e : SUCCESS in 41m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44676a8da5be45cc90204dacc98386f1 : SUCCESS in 40m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0051f7fd6806422cb1fd847987627cc7 : SUCCESS in 14m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a93c13f2dfa14a9682199e98bc413280 : SUCCESS in 31m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36b385b5171b40bea40d2b535e8eca74 : SUCCESS in 20m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31d4c740739e4bcb8fc379948261137d : SUCCESS in 22m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3312ad91ac426a9ab5886b2a5669f4 : SUCCESS in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca611f8db0634e0581663610651c7e9a : SUCCESS in 44m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fe723bfcd29450e849d89c6105afc2a : FAILURE in 1h 10m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c6562c88c1aa4f899c670063b6ab7234 : SUCCESS in 1h 16m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84c6809e4d584bbab6c4cc428de8e69a : SUCCESS in 1h 06m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2fdc2419cf0b4f2d97b19d6447fef8e8 : SUCCESS in 59m 44s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0d1441810e61445fa39b31f67e12b4f1 : SUCCESS in 1h 07m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2ba7391d2cce46d48ac0ee40229012c8 : FAILURE in 1h 30m 21s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/eb236ceaa7534f9db1ab1a88c100cbe8 : SUCCESS in 1h 33m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c05008b47bb463585b2adcafca4962c : SUCCESS in 1h 01m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d426f1e648214c21bafb3299ccfcda9e : SUCCESS in 1h 08m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0835eeba365c484ea7609b46770f45b8 : SUCCESS in 27m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d815be6e35024b6e9dc8c340aef6bd18 : SUCCESS in 1h 07m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6332e6280f6e4ac28207641471335b6e : SUCCESS in 1h 04m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"915280dd38a89d0c3417ec3866bedf596d805c77","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-22 22:35:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799439/2 : SUCCESS in 1h 39m 44s","accounts_in_message":[],"_revision_number":2},{"id":"29817ede4646301c8f30db07f38a252bb9a778f9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-22 23:22:37.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/39/799439/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7fc13650cb784e0cbc4540e48d2c8d53b570a6d5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-04 08:12:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"670714bbe0e872ba8ae926ebcd05d27265c6151c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 08:51:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/326f498ed25749f2b4d9d9ffe07ea2cd : SUCCESS in 34m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5afbebfa5eac48f6b5064a844d2bdf80 : SUCCESS in 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8939f5aa5d7625932934cc9933d548d8d073ca87","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-04 09:48:20.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/799439/2 : FAILURE in 1h 34m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b8af48a0f2b27cb7d243901f2ec0065e8b5f932a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 10:23:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e3af38d3a17947c3b48e16c41b359ec2 : SUCCESS in 37m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/62b8c2950a724cdf8ce38704ea9ac0fd : SUCCESS in 1h 38m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c74c04e9ae5f47439916032edf4b64b0 : SUCCESS in 1h 57m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/83072280770c49cc9bacea03af05a587 : SUCCESS in 1h 15m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dcc05c9d712b40c7a3cb2fea69d6f0fe : SUCCESS in 2h 01m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b8dd97c07a6c427ea2ef64f2e235940e : SUCCESS in 35m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98f77e4f743b4209a5ba96e70eebd473 : SUCCESS in 37m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60e1d0eeb3fd4067a75756f9a22e0e15 : SUCCESS in 40m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/885c29a5f53142cead2b1b75e9259e45 : SUCCESS in 13m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd0986952ff4494a886c04825acfd149 : SUCCESS in 26m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e70b11c84f3c414f97e200faf6ee0641 : SUCCESS in 27m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b529b3431184a95901afbb5fbf90bbd : SUCCESS in 27m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa40cf63059842bfb1499b375eecd6a9 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9456fae8163a41b7a539173d39e52bbd : SUCCESS in 48m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6481f40598843c29713a77dfdf5257c : SUCCESS in 1h 22m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cdbbfae685e84425bbfdb47f282b1a45 : SUCCESS in 1h 13m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/83e80d803175491cbb4a2d1c0d1e5e56 : SUCCESS in 1h 02m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ff2de58616da46cc8ca1e2cd6b5c77c8 : FAILURE in 1h 12m 27s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/98844d5a90fa4448995565199e9d0169 : SUCCESS in 1h 09m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/16fc3ace822a4a60ab8466b83d2f14db : FAILURE in 1h 04m 02s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/dbbef0fec1ea408185cdbb5b8644271b : SUCCESS in 1h 26m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3e61d42b05f4085a1536e45879aadb8 : FAILURE in 12m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5dccf4a92b6342b5a12a81e032a9bfe4 : SUCCESS in 1h 01m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/84f3611b530d484daa09acbad79cd2a1 : SUCCESS in 35m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b72c3289ae254ecaace1c2a637f9ab9e : SUCCESS in 1h 09m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e763b483a0fd4f2f84c65236034f128e : POST_FAILURE in 1h 20m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6fe221a9842272f3b79ecd085f5d34e13b7f7324","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-09 07:20:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"66bc1a1c9f1d6906a7fc32231f436756bf74810a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 08:08:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/080bc81af9c045c78f7ac4e6131924c9 : SUCCESS in 41m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9aca0665baee430f9a3bc7a67d6db2e5 : SUCCESS in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c77415fb97320e76dffe7801a4f740206ac5c2f","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-09 08:19:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ac6a12b60141f3c10ba24440e8c0768ea045bb6d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-09 09:17:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799439/2 : SUCCESS in 1h 55m 43s","accounts_in_message":[],"_revision_number":2},{"id":"7efa9df43b690fa31cf7ec4cff2984aacc47af67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 09:35:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a898b2659b564b419f4bcf46f325e00c : SUCCESS in 37m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ca5d8518fb0948a29814585156fa0ba0 : SUCCESS in 1h 57m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/72048421123b431daa3864b9fee0ea38 : SUCCESS in 1h 53m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ba5351095a849559cc2ddba9b78908f : FAILURE in 1h 22m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b082d91a16634750829fef80ced7d65a : SUCCESS in 2h 10m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d909776aed8d44d2b36b5b0a656038ba : SUCCESS in 31m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36dcac956c65465f948d561c45051fef : SUCCESS in 36m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87800d6fd6404f12a2f5de7e54f880f1 : SUCCESS in 29m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64c898a9324b4728a26d7ca2b1408d97 : SUCCESS in 14m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aa8b703e6464c3faf47f9753681c385 : SUCCESS in 29m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a189cc72ac264684ac4352f2fba5f0ac : SUCCESS in 26m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9236689871b4305b1880eeb0e7da3a4 : SUCCESS in 24m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d68707c9831544309a7bf6dcd5081576 : SUCCESS in 9m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad684bb0874f4a259a702134f08fbb6f : SUCCESS in 52m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e53de6208ab4a51b01e633a57390b42 : SUCCESS in 1h 24m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc1a57d367404f3dabd101c12d53ad9b : SUCCESS in 1h 09m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/48f80cd0da5d48ec90986afd1139e895 : SUCCESS in 1h 03m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/06b8dcb91be54c23982bc25dfd4ca150 : SUCCESS in 1h 04m 11s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aa84e50cd7a341319604c55a3fb00663 : SUCCESS in 1h 10m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0af37d7523d4408d8d752969fcae6d77 : SUCCESS in 1h 14m 38s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/0d599189f98049a8952fc86261c3ca4e : SUCCESS in 1h 29m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1616d75a5ba4c4199e7d7bc989b0d27 : SUCCESS in 57m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0cec4c9de48a40ada568615b4b715999 : SUCCESS in 1h 01m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/306adaf4ac0947028161e60fd1f7b3f9 : SUCCESS in 33m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3a8c6c4e9d5a43ce9d2e88be13c9da5b : SUCCESS in 1h 10m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4b0ecb9bb1ea4d31bd2e81633598460a : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":2},{"id":"a98e7d3895507a0a34670a6990b07152d979a3d6","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-09 10:43:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7db8c89460bb50c221e71577bb317815214410be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-09 11:21:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"865a6af1417e18a93e3e9b677e46ecf6e83a587f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-09 11:21:54.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5509f6d93a4cc2399ee3e2bad6a6da5bc3dbc89e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 12:04:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e60ef28a054f48828f7c43a6f009123a : SUCCESS in 34m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/961f7b16c48748a5a6aae3cebcdc81ad : SUCCESS in 35m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d18464d084e00fac50ab4175497807093bc0ee52","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-09 12:59:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799439/3 : SUCCESS in 1h 30m 32s","accounts_in_message":[],"_revision_number":3},{"id":"ad77d3bd49a68ec2bf1b07af1f90af72f8cf644d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 13:20:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ee218d3300e943a8912323df26958f7f : SUCCESS in 39m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e90c81b29a8e4f7fb440b2c69ebece35 : SUCCESS in 1h 51m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eaf315a1a6d04cd5b74b8465f8d10e31 : SUCCESS in 1h 50m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2e1ae2d52db447858b8a6edceae5f9d8 : SUCCESS in 1h 52m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/398917a103a24ad089662b7c93c1941c : SUCCESS in 1h 52m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a2a60dd46274a4c8161caeb58b5f1c9 : SUCCESS in 32m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/096ea70288ae410ba38249c4f4aa7379 : SUCCESS in 36m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8133ff0635542a7af5d043e94f821bb : SUCCESS in 37m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0edc54c29c2a4dec9e654cb92ecd753a : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/022e70f0464d4b89ae74a4f3a2075c5c : SUCCESS in 39m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bec5de2fbcb04df2b6f2e61be9f97219 : SUCCESS in 27m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0fa1b5e63b945a29e9e53b2d5d009ee : SUCCESS in 23m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0051cc4d8e60496da99082a1ad3d735a : SUCCESS in 11m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9015fd2e9da4744b4072641ffb0634e : SUCCESS in 47m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0a3eaef5b46480792d920fe9b75f31e : FAILURE in 1h 05m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/55f85f0c6c63411482aa93dcfc937b4a : SUCCESS in 1h 03m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6938384c2d9643b6a86f41dafe4ecc59 : SUCCESS in 1h 00m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fe46026541c447a59675ebd309eadb37 : SUCCESS in 57m 43s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/54f6f400912a483d80fe86c8cd036843 : SUCCESS in 1h 08m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6cebc4ab4d484722ae28c7f807d6f4f9 : SUCCESS in 1h 15m 38s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/6bd900095752468d9d6675528897a370 : SUCCESS in 1h 24m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d551b909c35741189be8c612c7ae0d20 : SUCCESS in 55m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1da40da94ae140738c496d3696387188 : SUCCESS in 57m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ecc1266a94c44dbc9b8de92a51769379 : SUCCESS in 25m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/94178104a32a452ea611a9a00834fee1 : SUCCESS in 1h 04m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3a35814d91c14b09893b4106643e1f18 : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":3},{"id":"357c37671916fcce034e82729e358d4f6e4b45d1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-10 07:53:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cb499f5188a5adb2616969bd6f5d55aeeadadaad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 08:39:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9ab44db19f64c17b4e19f1144dd0a5b : SUCCESS in 36m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20a36bb0561a44c49b2fbd8848ac625a : SUCCESS in 43m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"05622a0453129d07c010235e33f6b58d7e79953b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-10 09:23:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799439/3 : SUCCESS in 1h 29m 24s","accounts_in_message":[],"_revision_number":3},{"id":"fc9763b18148a06385880d48c83d15f922a29ea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 10:15:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/398af4aa8e1549a3ae882b0de9ee8ffa : SUCCESS in 42m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6c29e3a3e2674ee381f087d9f7d0fb5d : SUCCESS in 2h 10m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a7e1be91a696451ca246d61c6d4744d5 : SUCCESS in 1h 44m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5e9e0ace548c4d05aa72b99abe9fdab4 : SUCCESS in 1h 31m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/727736ee3ed8455e914d89de88ccc941 : SUCCESS in 2h 14m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/417880a10ed548318a5a78488aa8b8f5 : SUCCESS in 28m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d4bdb886ab549b5814a684818a9bcee : SUCCESS in 34m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebc6dbd819204d2cb616ff888e346ddc : SUCCESS in 36m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67b34b72bf374ce8a23e8ac0647aa720 : SUCCESS in 14m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/707c8e5dbc264b2d88a14b6ef70abd36 : SUCCESS in 26m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7c57ffc0a294c7e926a1e98937ec9ca : SUCCESS in 25m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64d7d67eca444ad6be3741ed47a7498c : SUCCESS in 25m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c9a5af7da7447a95d4064d01653085 : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e276f939f90545a9a666ebf01ba339b9 : SUCCESS in 46m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9debfb16237049aa935620b86219e868 : SUCCESS in 1h 44m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/720baebf96434a04997131f34dedadb1 : SUCCESS in 1h 05m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/568a15cebc5048dca643c7e50edff130 : SUCCESS in 1h 03m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf9d7c1cdcb4488fa3c9af4ac4501a9a : SUCCESS in 59m 43s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ae7380915a3c4d2680a4297c51824c76 : SUCCESS in 1h 10m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c98140b0741f4339b24e120119a76876 : SUCCESS in 1h 18m 52s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/7694c77223074645b3ef18ffe8fccfa9 : SUCCESS in 1h 30m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/951e903be92546cd88f3f29f1c7afe78 : SUCCESS in 56m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/77352140283a4fc89483654d7e4a2500 : SUCCESS in 1h 01m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ac7a298b6a88448b9607ee5ea86a213c : SUCCESS in 29m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5c61d0dcfe334984b61372fed2f4e855 : SUCCESS in 1h 15m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ec57fcdab0224fe1b6d54c9b2b1142f9 : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":3},{"id":"2bc99aec2782add3d21b33664878639544d1f29b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-17 16:26:53.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/39/799439/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"31da5790b8adff3f6a19a46c0091b27bd172b8e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-07 09:17:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e014e6c10799401db3edb53ff8756d9df9d8eb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 10:57:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed3ac09bbafc4b43aa36556fe09862d3 : SUCCESS in 44m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32d5cc1eaa5649f4bc41cf6456afaed5 : SUCCESS in 43m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6d432e977233cd6ba93f9396e0fe3c3172b0e785","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-07 12:47:54.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/39/799439/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"371ff8253fbd6d1f5385037306aa1220365f5260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 13:08:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f74fdc4ceb6f463485983f04efc8a2fa : FAILURE in 45m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e70df967ff0b45dca9e3dd1b3f7d1e12 : SUCCESS in 2h 02m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/951f0b66f4f8447890a41af786cd6487 : SUCCESS in 2h 00m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a3e9936817e944c5994d7c27e833f11a : SUCCESS in 1h 28m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/efc03de04da142cbacc225640a1028d5 : SUCCESS in 1h 55m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/75fbf69103394901bedbb5f0bdcfcfe3 : SUCCESS in 32m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0337ac71c2d4f2fa971d62396911562 : SUCCESS in 39m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81aae69dca864ee7aa3db6f343657a0e : SUCCESS in 42m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d179dc2660684ecc8b19a11220baed43 : SUCCESS in 14m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf5e0224cc4741f4bdbbc88905ad83c7 : SUCCESS in 23m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5b0006cd6b14f2f900db1e4fe057a10 : SUCCESS in 24m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1be83bfeda54e75934cfdcdbc2ff022 : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d899e66029340db988bc0c2634db6cc : SUCCESS in 11m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54067af865c0459db52188a12e5fc19f : FAILURE in 53m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa3cf104ca3149969c76a0751be4e927 : FAILURE in 1h 29m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a1c6c2027dc74680b6808f58f546692f : FAILURE in 1h 16m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ae438f3fddd4227ae34dbeb1505c62c : FAILURE in 18m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f5762fe53af44472878f79e42452c1f9 : SUCCESS in 1h 03m 30s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/de455969277a4400a7ad2b3b581ee1b2 : SUCCESS in 1h 12m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3644e475e32047bd994b68c7bda45b25 : SUCCESS in 1h 27m 01s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/41307ecdc20548e0965a82ef910eea4f : FAILURE in 2h 30m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/297f7a985528456a840a555e4040f248 : FAILURE in 12m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8f0883e5a701416993a4a545c93edd4f : SUCCESS in 52m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9396a163b45043b2aa8b311094f7ee02 : FAILURE in 13m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6913e425bd96496b8b6f059bf1afa1ba : FAILURE in 11m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7d300cf5c05e4573ae647333f4f097a3 : SUCCESS in 1h 39m 25s","accounts_in_message":[],"_revision_number":4},{"id":"3e36e0b233db541a01ddf5be65d1c03e744860b5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-12 07:53:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"08cc20dca724d86eee10ee0829f2c2b1b078a351","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-12 09:46:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799439/4 : SUCCESS in 1h 51m 57s","accounts_in_message":[],"_revision_number":4},{"id":"148f420f48bab1457fc2f7d3386c0875af3a4d53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 10:00:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/530df0379c4c4271b1ad91a55593d1d8 : SUCCESS in 52m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfa9b542ebf04f01a6bb82a9c60f7691 : SUCCESS in 46m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0a76c525463bf44d3f9efa472edec34a2b52b742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 10:13:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5405f2ec44b44a15bf1acf09fa7c52f2 : SUCCESS in 44m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/588b7a4124c14d7fb951fac38563aacb : SUCCESS in 2h 03m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/954f54e293834f01af5c6ea20a816735 : SUCCESS in 2h 13m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3074222747d545859a8efad381ede9b3 : SUCCESS in 1h 34m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9035e74f9ee24f17800b9d2a35d88bdd : SUCCESS in 1h 52m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/866336cdede142ecaa16b1b3c349ab64 : SUCCESS in 33m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08f84891c50a4e7481ed21d609065b96 : SUCCESS in 35m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed9eedc53b934c03b473730926a28c38 : SUCCESS in 38m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49da73cfee904949920348e43240fa04 : SUCCESS in 12m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7acd8c33392422497c934fde04e0410 : SUCCESS in 26m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d31dd23d60054b7caf5df89e0e0978c4 : SUCCESS in 26m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e98eaac2920f4cc2b07bd2e8319f07b9 : SUCCESS in 24m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b104c3d6a14b4e5988da269b11d6921d : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea1bf7469ca044d98a1649e6e928d77b : FAILURE in 50m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7183f066ef6c48c5a62fcffd9abd0bb3 : SUCCESS in 1h 41m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fd6922a72e4645f5bff32853de7f4d90 : SUCCESS in 1h 25m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/130f8e862d4f4e49b46f47f36d996917 : SUCCESS in 59m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/244b88056a774584911c631e172168f2 : SUCCESS in 1h 03m 16s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8e4b0c7d674c437c82d2a550d9b18af5 : SUCCESS in 1h 19m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cc0f88e094ae4989b17ac8f78df3a0ce : SUCCESS in 1h 39m 40s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/9045bb63717f4ea0bceef767c93d6373 : FAILURE in 1h 24m 43s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14c1747c7df24fe7a2f939f1236675a2 : SUCCESS in 59m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/14c8a2fe2c9243eeb05d064abffe7bc7 : SUCCESS in 55m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e490f7a3b3a94bfc984dfc32a4070b61 : SUCCESS in 34m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f6611de866a3465b8e836d6de2cbb9f6 : SUCCESS in 1h 09m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/08c92730b6ca4fa79dc666dca56aafdf : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":4},{"id":"4eccdda4ebf0face07c3a04d47476a432d24c3e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-13 12:33:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a417718daf74bcdff970cc4b86d99ffc046ec478","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-13 13:36:27.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/39/799439/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"16c87940371041d2915b59e4bb098c78a50879fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 13:37:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce154780063f48bcab966245b5fdb0a2 : SUCCESS in 41m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/382d4a2cfad34280ac9d218b5bd3d21d : SUCCESS in 49m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"24200bdea332c59df3816bcd6405f441c9dda521","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-13 14:07:43.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/799439/5 : FAILURE in 1h 33m 40s","accounts_in_message":[],"_revision_number":5},{"id":"99a0c58bcf824f965201ce6a91136d034e7d49a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:22:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d36285b0d12447e6a649ba3a6006b8a0 : SUCCESS in 53m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3b931da69d354d04b872ea0bb8564b2a : SUCCESS in 2h 41m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/82dd710963f249a180f4876b0c97aff5 : SUCCESS in 2h 15m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9dd8b1e0466c40a3bc9b83a9e339ba51 : SUCCESS in 1h 29m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/27b960378c5e43e7b07e31ecdd5398ca : SUCCESS in 2h 12m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65f42f016fc8410da964306175085bb6 : SUCCESS in 34m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acbadc15c1f74a6b9b0d26981c94b064 : SUCCESS in 37m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/155400601c7641daa34c26d813572fa6 : SUCCESS in 43m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b604ddfd404dc0ab5762cedae1ce15 : SUCCESS in 15m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c00dbc92a4c4e288252f3b8e5b1a4bc : SUCCESS in 27m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c325b3dfbe945498852e7e37c4d3f17 : SUCCESS in 28m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6346e5e33ae424f9f93136bd3c912f7 : SUCCESS in 25m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da99b9cb5c742dcbb0ba507a475dba5 : SUCCESS in 11m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db1f0ca6a3084d9ba9e7ffd354f61f10 : SUCCESS in 51m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e010f21e92aa4289941e3e151c852749 : SUCCESS in 1h 44m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/109c759f6b4b40c6801dcd643b3750d2 : SUCCESS in 1h 10m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ce2695c708242b0b567471b0c5ba8ca : SUCCESS in 1h 00m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/11598ce32e0642af9146af64804c0599 : SUCCESS in 1h 01m 03s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1a878767667f4b3c9c89c685cc3ebc43 : FAILURE in 58m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/58bccfb8f61a4dee85c34201b60e7599 : FAILURE in 1h 40m 39s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c22029bfb99740fc9ce39822700e5f3d : SUCCESS in 1h 36m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f5031e1f78e4392a6d87b176fba8bf8 : SUCCESS in 1h 00m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ecfcc6880c11484e82256384fc84fe53 : SUCCESS in 53m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fd24c2af7c1749258e4f2ea9384af9f4 : SUCCESS in 34m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e21dc30fda14b41a99fb65edabdeb72 : SUCCESS in 1h 12m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d6ac8e6d3e224750805f7c1d006b2625 : SUCCESS in 1h 21m 37s","accounts_in_message":[],"_revision_number":5},{"id":"5251b18d494244f16153d73ccd551beb153c18b9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-10-18 11:00:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"26a51104558548c5a3ca0b6013ac3b6bac9cdc95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 11:41:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/404080cbb9744e2cb0e471e866c752a3 : SUCCESS in 34m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d6ab55024e845fa92046fb69b016204 : SUCCESS in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"838696d2da7ed9c03a9567a3751c0b8d27e8660f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-18 12:49:26.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799439/5 : SUCCESS in 1h 48m 25s","accounts_in_message":[],"_revision_number":5},{"id":"395d951a7edcdc773e4098e41ac2995aa1b4e377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 13:32:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1ede23fba7684f9e8b3fb6a5ebed73f3 : SUCCESS in 42m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/67bcd50ce08c4b5b9e38385e658866bd : SUCCESS in 1h 52m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d721a84ecb5744fb8fa507037123e3c4 : SUCCESS in 2h 19m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d7a4109c1653412cb78d1e10980c4520 : SUCCESS in 1h 40m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4a3c283adc1f4399b3a60ddd8a8d3c87 : SUCCESS in 1h 39m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/60884766a0de449495d4dc8292c82c38 : SUCCESS in 40m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf6445f4ad5a4de7a432255ef523dc07 : SUCCESS in 39m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3eaff057592d424cb5e08e551836638f : SUCCESS in 30m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a3008958824a8fb6c1eb81852d8f51 : SUCCESS in 14m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa03d3c3c9c54fc99ab18821a42bc2be : SUCCESS in 23m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae51872ea6e94fcbac4fbc68a582586b : SUCCESS in 29m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e907309e2d4a483984b16412f1abf95e : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eef106152a24553b710d8d855b27e04 : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6980096a203f4a139efd0961048527a1 : SUCCESS in 50m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/72f50c3e98014f2d9ee65f234a49e1b6 : SUCCESS in 1h 46m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/411c4b792a324a1a8fe1fc38702c639b : SUCCESS in 1h 12m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4972629e075045f080d830d5b15a2781 : SUCCESS in 54m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1a971765543b4dc79b0a062c116813b9 : SUCCESS in 57m 15s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ce360df1cc3444cfad3092c4661f79de : SUCCESS in 1h 04m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/035a14508fb34552b86af45b7e88a460 : SUCCESS in 2h 06m 02s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/e6b65f00f0a941769ce0208bf40dbeb8 : SUCCESS in 1h 23m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35dd570ad3a44234a6b3d766229e79a5 : SUCCESS in 55m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0945016dcce04555b6b4c024ea8ee149 : SUCCESS in 57m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7efcaad991ba41a881482e7dc10c86b8 : SUCCESS in 34m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c55d65bb8dd74228960264007724b664 : SUCCESS in 1h 05m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2f78627c1f7146afbb26887ddd64a851 : SUCCESS in 1h 13m 20s","accounts_in_message":[],"_revision_number":5},{"id":"c27546db0c95ad926b9eff491e384cc7c2da301c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-10-19 07:10:36.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d65301fa0579b4282376a35b09752958a2308b45","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-04 10:25:51.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"25444fcf814c776d57810bf3cc90b5ba97d8ce54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 10:27:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"602523c5e2cf6b326781a7f0d6ee2d669ce950be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 14:20:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3da643f3d66248a1b3aa2e9dd7cce5b4 : SUCCESS in 42m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/76f2a36286da4a9fa40014821378d664 : SUCCESS in 2h 01m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9ba25ee2681c45cfb3ada7470f12269d : SUCCESS in 2h 09m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0fb88156acd44753b74ef8a1b0fb4810 : SUCCESS in 2h 08m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7c0ac9945cf40d490505bf2ad3b1d91 : SUCCESS in 1h 34m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/980bcd90aa3c4ad986a0f523b7857e0c : SUCCESS in 46m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c6eed0958e940cf8956b58274b81140 : SUCCESS in 13m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a63848510f1d4d82af2e836bee28cb7d : SUCCESS in 41m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4884ba4d156246a1ab1f539b45b4cbb7 : SUCCESS in 38m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24da1df8ba4e4b97afeb1f79913950e3 : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e03fa06c61ca4bb59f8df9ff275639a5 : SUCCESS in 50m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab1d38bbadcb45338a3a0c3d3463e729 : SUCCESS in 2h 06m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4d5941c5058d4a52ba4d11e2e55e3e61 : SUCCESS in 1h 08m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/853c20b8624b49fe8242637ff6a92d84 : SUCCESS in 1h 23m 44s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/db4a0401c75c48e2a81219abbc6800d0 : SUCCESS in 1h 23m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/095b663adf414fef9d9576cb813b42d8 : SUCCESS in 52m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b565f3fa9547472582994ebb512f64e8 : SUCCESS in 1h 04m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fb9b5faf9d084f089cbb48e54eaef604 : SUCCESS in 25m 57s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fb7c68f82b26416d9bcd8329e0338db4 : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":5},{"id":"20ec436aa157fb3cd9b51d237fc8fa06118de473","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 14:20:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"883d2fa8925311d46e116d2b61f942b9030fd4eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 14:32:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd7f36d48da40939c3b27ae72b3f874 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"21eabbcf03cb29be9899adbd6eaea207892cf07b","revisions":{"7c0fba450a4013581f38feeaec50a2639ba12c3f":{"kind":"REWORK","_number":1,"created":"2021-07-05 07:43:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/799439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/799439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/799439/1"}}},"commit":{"parents":[{"commit":"314a5963431d7912c6efeab86e1e0ccc29f6d15e","subject":"[DVR] Fix update of the MTU in the SNAT namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/314a5963431d7912c6efeab86e1e0ccc29f6d15e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 07:40:28.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 07:40:28.000000000","tz":120},"subject":"[DVR] Fix update of the MTU in the DVR HA routers","message":"[DVR] Fix update of the MTU in the DVR HA routers\n\nThis is follow up of the patch [1] which fixed updating MTU in the\nsnat namespace for the DVR routers.\nIn case of DVR-HA routers there was additional issue with that as\nL3 agent tried to update MTU for the qr- interface in the\nself.ha_namespace which, for DVR-HA routers is snat namespace.\n\nThis patch fixes that issue by setting MTU on the qr- interface in\nqrouter namespace and also setting MTU on the snat interface in snat\nnamespace.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/799226\n\nCloses-bug: #1933273\nChange-Id: I409bc674b65e4f495ebd42d03e97a09d51482339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c0fba450a4013581f38feeaec50a2639ba12c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c0fba450a4013581f38feeaec50a2639ba12c3f"}]},"branch":"refs/heads/master"},"860a9392d89965bb79eef93e0f38b2fa5ad8ce12":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-22 19:11:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/799439/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/799439/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/799439/2"}}},"commit":{"parents":[{"commit":"beaee27d30f640944d25148a1b7a8b015d7cbcf7","subject":"[DVR] Fix update of the MTU in the SNAT namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/beaee27d30f640944d25148a1b7a8b015d7cbcf7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 07:40:28.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-22 19:11:16.000000000","tz":0},"subject":"[DVR] Fix update of the MTU in the DVR HA routers","message":"[DVR] Fix update of the MTU in the DVR HA routers\n\nThis is follow up of the patch [1] which fixed updating MTU in the\nsnat namespace for the DVR routers.\nIn case of DVR-HA routers there was additional issue with that as\nL3 agent tried to update MTU for the qr- interface in the\nself.ha_namespace which, for DVR-HA routers is snat namespace.\n\nThis patch fixes that issue by setting MTU on the qr- interface in\nqrouter namespace and also setting MTU on the snat interface in snat\nnamespace.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/799226\n\nCloses-bug: #1933273\nChange-Id: I409bc674b65e4f495ebd42d03e97a09d51482339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/860a9392d89965bb79eef93e0f38b2fa5ad8ce12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/860a9392d89965bb79eef93e0f38b2fa5ad8ce12"}]},"branch":"refs/heads/master","description":"Rebase"},"15a7824a22c325a78707b90d634e81e6a12ffdd1":{"kind":"REWORK","_number":3,"created":"2021-08-09 11:21:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/799439/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/799439/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/799439/3"}}},"commit":{"parents":[{"commit":"b4b69718ee9c0c59a5adfb3cd6618e2b81d6ae28","subject":"[DVR] Fix update of the MTU in the SNAT namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4b69718ee9c0c59a5adfb3cd6618e2b81d6ae28"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 07:40:28.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-08-09 11:21:28.000000000","tz":120},"subject":"[DVR] Fix update of the MTU in the DVR HA routers","message":"[DVR] Fix update of the MTU in the DVR HA routers\n\nThis is follow up of the patch [1] which fixed updating MTU in the\nsnat namespace for the DVR routers.\nIn case of DVR-HA routers there was additional issue with that as\nL3 agent tried to update MTU for the qr- interface in the\nself.ha_namespace which, for DVR-HA routers is snat namespace.\n\nThis patch fixes that issue by setting MTU on the qr- interface in\nqrouter namespace and also setting MTU on the snat interface in snat\nnamespace.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/799226\n\nCloses-bug: #1933273\nChange-Id: I409bc674b65e4f495ebd42d03e97a09d51482339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15a7824a22c325a78707b90d634e81e6a12ffdd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15a7824a22c325a78707b90d634e81e6a12ffdd1"}]},"branch":"refs/heads/master"},"d7f93ab772616e0949279223604e4446c9151ec2":{"kind":"REWORK","_number":4,"created":"2021-10-07 09:17:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/799439/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/799439/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/799439/4"}}},"commit":{"parents":[{"commit":"dfcbb4cce0f206bbf824a1062de028223a27388f","subject":"Execute \"migrate_neutron_database_to_ovn\" inside the same DB ctx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dfcbb4cce0f206bbf824a1062de028223a27388f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 07:40:28.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-07 09:16:33.000000000","tz":120},"subject":"[DVR] Fix update of the MTU in the DVR HA routers","message":"[DVR] Fix update of the MTU in the DVR HA routers\n\nThis is follow up of the patch [1] which fixed updating MTU in the\nsnat namespace for the DVR routers.\nIn case of DVR-HA routers there was additional issue with that as\nL3 agent tried to update MTU for the qr- interface in the\nself.ha_namespace which, for DVR-HA routers is snat namespace.\n\nThis patch fixes that issue by setting MTU on the qr- interface in\nqrouter namespace and also setting MTU on the snat interface in snat\nnamespace.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/799226\n\nCloses-bug: #1933273\nChange-Id: I409bc674b65e4f495ebd42d03e97a09d51482339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7f93ab772616e0949279223604e4446c9151ec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7f93ab772616e0949279223604e4446c9151ec2"}]},"branch":"refs/heads/master"},"21eabbcf03cb29be9899adbd6eaea207892cf07b":{"kind":"REWORK","_number":5,"created":"2021-10-13 12:33:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/799439/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/799439/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/799439/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/799439/5"}}},"commit":{"parents":[{"commit":"f40aab3437770eb89fc2d10ca44e3179e2d5b88f","subject":"Merge \"[OVN] Update the DHCP options when the metadata port is modified\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f40aab3437770eb89fc2d10ca44e3179e2d5b88f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-05 07:40:28.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-13 12:33:19.000000000","tz":120},"subject":"[DVR] Fix update of the MTU in the DVR HA routers","message":"[DVR] Fix update of the MTU in the DVR HA routers\n\nThis is follow up of the patch [1] which fixed updating MTU in the\nsnat namespace for the DVR routers.\nIn case of DVR-HA routers there was additional issue with that as\nL3 agent tried to update MTU for the qr- interface in the\nself.ha_namespace which, for DVR-HA routers is snat namespace.\n\nThis patch fixes that issue by setting MTU on the qr- interface in\nqrouter namespace and also setting MTU on the snat interface in snat\nnamespace.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/799226\n\nCloses-bug: #1933273\nChange-Id: I409bc674b65e4f495ebd42d03e97a09d51482339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21eabbcf03cb29be9899adbd6eaea207892cf07b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21eabbcf03cb29be9899adbd6eaea207892cf07b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
