)]}'
{"id":"openstack%2Fneutron~799446","triplet_id":"openstack%2Fneutron~master~I80c3076b9545be55b11858c4422402dd5ae1a68e","project":"openstack/neutron","branch":"master","topic":"bug/1934512","hashtags":[],"change_id":"I80c3076b9545be55b11858c4422402dd5ae1a68e","subject":"Remove SG RPC \"use_enhanced_rpc\" check.","status":"MERGED","created":"2021-07-05 09:29:50.000000000","updated":"2021-07-07 23:07:53.000000000","submitted":"2021-07-07 23:05:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799446-bug/1934512","meta_rev_id":"3656a3b8f0c9cd8a66567bd7b4aca7477c5030f3","_number":799446,"virtual_id_number":799446,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-07 23:05:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-07-06 12:06:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-07-07 07:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-06 13:55:43.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-07-07 07:04:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 09:39:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-07-05 09:43:19.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-07-05 10:41:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-05 11:18:52.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-05 11:38:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-05 12:44:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"65f736eeaa392fcbad6cdf0375f67a48a2607a9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-05 09:29:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"177cbe52a1e05337bf5d85a24e1d3015d6cd0ec4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-05 09:39:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"125620be4feaf53d4ab8c9d3aeb068275c50afba","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-07-05 09:43:19.000000000","message":"Patch Set 1: Code-Review+1\n\nwaiting for CI","accounts_in_message":[],"_revision_number":1},{"id":"3d33a6b21250224e84bb14c1356912316e855ff6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 10:41:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2c5f3e5045c40e09536036538489e56 : FAILURE in 40m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99618b6acb6445a3877abf421416b12f : FAILURE in 39m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a6d51fc903b05df5874a6d7ceaa32f4eda8a17e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-05 11:18:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799446/1 : SUCCESS in 1h 48m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bcdd7548d9c10d808f429972a218189d9610ead1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 11:38:14.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f62bc3765be34e4bac8d8364557cfae2 : FAILURE in 2h 04m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/938688a6902742e3a6576d1f52edae49 : SUCCESS in 1h 29m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/23ad82e09f654106aa8fdcf9d553224f : SUCCESS in 1h 26m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4456932847c545caa70a3861a52a49b1 : SUCCESS in 1h 46m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee15329601384470ad438f3cde949a8e : FAILURE in 29m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0528adbd0dfc4fcaa0135d6c0dfcca2b : FAILURE in 33m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80233737901f4e0689a4b857fa65e30d : FAILURE in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11b373d5982b4f9faacbaffb66063d04 : FAILURE in 25m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9129a0c29ac142c0914cdde9d670f616 : FAILURE in 24m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee07237671ee4e63af938f5b42d66c2e : FAILURE in 23m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d1dd43325da4950945d626fa9601b63 : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/36d29ae2aec9486a93a7f27d53a986df : SUCCESS in 46m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/192b95882c46441c9e981dc2380c5aec : SUCCESS in 1h 03m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4e9096ccc55d4dc085259711f89aeb56 : SUCCESS in 1h 06m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6bdccd4bb964de587a49e17de42923f : SUCCESS in 59m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/936185d43dfb4133bee043ae6bf74da0 : SUCCESS in 1h 00m 31s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/714bcf67c2084140beed1a94e7c7be20 : SUCCESS in 1h 11m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ca3fdab027d24bbbb403f9819be7ff6b : SUCCESS in 1h 17m 21s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/06865fdc4970469ba2fe1d61e680aa56 : SUCCESS in 1h 24m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0af46be675e41d383d308db7eea20d8 : SUCCESS in 48m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/deee592176f04c308c0aedace9b73726 : SUCCESS in 52m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8b3fd4f339de4dc89bf2a0bd229598b6 : SUCCESS in 31m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/35564fc92f1b42dda7fe23ff2435dfcd : SUCCESS in 1h 07m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d14b19c4fa2a469f9c6cf69f1bb7d55e : SUCCESS in 1h 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfb1ab3f47393933a8b4c5d27b0324a7db0c38dc","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-07-05 11:47:34.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"eaf6473bd9466dbb79c057410863e2be3d438cc4","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-05 15:27:55.000000000","message":"Removed Code-Review+2 by Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"996a7684c755a7690c749e9c29fe075326b3b8fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-06 08:10:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"09ab724a0101ae20d2011baa3431c608a2fca304","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 08:51:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/538e98da8fe24b7b9e7d4c2b91e7a56d : SUCCESS in 39m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f270a22ab244dc2b3fe86b20bfcb1cb : SUCCESS in 39m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91fabdbcb9e0a25f802b564b9849fcecc42896f9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-06 09:41:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799446/2 : SUCCESS in 1h 29m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7b7536f528fdcbac0c8ff3bde5fd17cebfb92fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 10:22:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/58007d12367d4777aece877730e97efe : SUCCESS in 1h 57m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/85ab0e90b0764d0ba02eb11690738d2e : SUCCESS in 1h 40m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e47f2e545c2445b6968060f5037a7f09 : SUCCESS in 1h 22m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b860f34da481453bac2877074883bcb7 : SUCCESS in 1h 53m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69549e32706f4b278e08fbbfb640f197 : SUCCESS in 36m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3175155f0be44957abf3d177273c7a70 : SUCCESS in 34m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd2a140b7e74673ae4bf04dfcf9af12 : SUCCESS in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48742749dcb54013a6323ddde82382d9 : SUCCESS in 22m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aac4e52f75b44d0945ae373587e1b7a : SUCCESS in 23m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5ab72102e30414f8299d1e10ef3c75f : SUCCESS in 29m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986ea92a465c49c28c65de9601e0a174 : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/21afa983b8c640cfbb0516695b569df9 : SUCCESS in 49m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1c66b8940444cf1aec040c50196920d : SUCCESS in 1h 26m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ca465e02512e46458a603eb55f11c064 : SUCCESS in 1h 07m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5fc7c09411674514ad9a1c7581565891 : SUCCESS in 1h 09m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e0c3a5d3123f4db2927561af707174cb : SUCCESS in 1h 06m 59s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b5538aee95e4a71a2532e3019cb831d : SUCCESS in 1h 16m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3de1248668bb4ab7bccc5c849697e37d : SUCCESS in 1h 23m 39s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/43713060372b4447b6bc3e05b7cebd77 : FAILURE in 1h 32m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f9f0bff68c4470eaedf3651f1e03557 : SUCCESS in 1h 02m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/55ea4e8d40294388afe366bb9332c8cc : SUCCESS in 53m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/916a4c7d840445baa590308ad2f08c5e : SUCCESS in 32m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ba8c8c471f4c4bd09b8932801e68a49b : SUCCESS in 1h 08m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1718d493bbf4448bae286925b71b9433 : SUCCESS in 1h 17m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8e0f6d7c461988e6ed2088fd19d636b4798e147","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-07-06 12:06:23.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck unrelated test_subnet_details failure","accounts_in_message":[],"_revision_number":2},{"id":"577666af08065a942f8f145ee5d6396563d67dc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 13:02:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c2e64bf7a0c410b826c0f752c0d36ce : SUCCESS in 44m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f890ac45d2cb41fcba9f91217464e94b : SUCCESS in 45m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30197bfdb26429dc2d486cf37c59b08027c896d4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-06 13:41:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799446/2 : SUCCESS in 1h 33m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c9a3de2428ee34fd0062721660615c7624991d88","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-07-06 13:55:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e900fa52f231d99fe97b25300cedf7d6ee3b38dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 14:17:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/99cc69ff9be947d1878056fbbd31f130 : SUCCESS in 1h 39m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/20948a97ab1d4b9da2867cebd2f6d0fb : SUCCESS in 1h 43m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e19a1f95a5a45f28e8c8be3bf29f53f : SUCCESS in 1h 05m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7d1c4d1c9aad4c8191a860748ac282cc : SUCCESS in 2h 08m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/587c4cb78e9a486cbce918329eddba15 : SUCCESS in 32m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ece4a7dbe02d439f80626b7e96287751 : SUCCESS in 44m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16d50e1b63c1494cb06d5f4e23280b56 : SUCCESS in 15m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a670c98c45264e30ae3718b7d617adbb : SUCCESS in 25m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a286497f5b84bf9ab8422c21efb79d9 : SUCCESS in 24m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df80f8e605a841948982737eaf2ae612 : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5aae9fcc16547b4900ca11ee76eada6 : SUCCESS in 12m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0467000a29b64e5b8612eb73e5933b3b : SUCCESS in 48m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3edc4678a522416386f60b12aed018a1 : SUCCESS in 1h 51m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0c1a48db27064ad48d3475586f3fc164 : SUCCESS in 1h 15m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/09f8fa05870443ddbd47116f5909059b : SUCCESS in 58m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c1617b458e604ee6a2e3950f8096e6f4 : SUCCESS in 1h 04m 46s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d84031aba763447fa04306fa3ec126c9 : SUCCESS in 1h 07m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/223f1a0d2e64422daf37e1995dae2c8f : SUCCESS in 1h 35m 30s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/e42cd248e1284bf8b76a6b1d0cb17333 : SUCCESS in 1h 21m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea625aaeb74641ee954c2814f9856df8 : SUCCESS in 59m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/79c8000b1f5b4657876c44378c42c40e : SUCCESS in 52m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/31a5d61fe03a415c8be44e6a04da8407 : SUCCESS in 26m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d7cb763f41dc4bb9aab67dcebb77f572 : SUCCESS in 1h 10m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f684f7d83c1d41d8aed9a29b7b854aa9 : SUCCESS in 1h 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e28f33faa58544d37d7fa555e82ffa0587c856e1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-07 07:04:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b51baff8fce246055ad6d08a062409bb1cca80d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 07:04:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"125a318e56699739264739edd72c4e8d75818bc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 09:20:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/187f9f9913984cbfb53a0125702bde38 : SUCCESS in 1h 58m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8636171c37264df0b57226eada0f5b62 : SUCCESS in 1h 51m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/68a0f77678c84dbe8bbb025b233a3825 : SUCCESS in 2h 09m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1b3fd6ca3ab9498fbbd96ee2a9e9a2fd : SUCCESS in 1h 38m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e6c87a6a59e466ead70c76e33d0766a : SUCCESS in 41m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f621804b394b81aab1486bc826f0af : SUCCESS in 14m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6730a29f8f6c469d8ffc317d99474df2 : SUCCESS in 29m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da73b63fd3f840dabf0a77cac2be1166 : SUCCESS in 25m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43bd8d401a4f43c49097e8c916bf5276 : SUCCESS in 9m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7b936a6c91e4437933d50f13e6cf011 : SUCCESS in 53m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/101ce40a028b441faf32e9f702faceb3 : FAILURE in 49m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2255701e93624b9ba3fd18f465643d57 : SUCCESS in 1h 18m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a189845d439b4940ad207beeb121652e : SUCCESS in 1h 13m 21s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/ae4a84d5a25e45c6b4071ce884c3133a : SUCCESS in 1h 31m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5389dea7c644dab8442777c8b1d5fc2 : SUCCESS in 58m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a052393e466344e49b9516992c47934f : SUCCESS in 57m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/54f2f7634373496288a2853dfa77b4bf : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":2},{"id":"80ca3f32492d52be4e33506c21c97e17bc8b223e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-07-07 17:55:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b16f034087c5d9357800086d5f1c77c1410cb6a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 18:33:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc7f4921a1f94360a66188da29c2092a : SUCCESS in 36m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86853f584c1b43418fd073e31d3354ff : SUCCESS in 31m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"526acad256afd143e0d65c945686cfd854ff8933","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-07 20:08:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799446/2 : SUCCESS in 2h 13m 02s","accounts_in_message":[],"_revision_number":2},{"id":"bd5022ccafab6292878ba15c47abac2190a5e8a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:15:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/40b725c4fdc4410885b7eb475c6ef089 : SUCCESS in 1h 31m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b943d16f5ad745298e72d069d5fab1fe : SUCCESS in 1h 42m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d0239922fed44a2a8cf7ff01e540e72 : SUCCESS in 1h 49m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ff2be6a051d74ba2b995e5bd71e69983 : SUCCESS in 2h 16m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d7d538e53d84362b53433e5fa1e3909 : SUCCESS in 37m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/120d9378ff3648e7be92e45aca13ac3e : SUCCESS in 43m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/245dd78ce4d2470fb2df20c063114bee : SUCCESS in 15m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95054eb704b146ca9fadc5c75eb997e4 : SUCCESS in 28m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e088b491e7e42619e07b1282690943a : SUCCESS in 21m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e118f7333704136a7dee00ac3f14f26 : SUCCESS in 20m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12bec40a1efe41938bc72699e074f039 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d362603fc5146cd92bb9bbc0a0ecb33 : SUCCESS in 49m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/240a4a6b4b4a4ef0aaf8d028c009e1fa : SUCCESS in 1h 25m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/507344eb740a4bb5843968b73f57666b : SUCCESS in 1h 03m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b98c6d4e47504b91a1b728e55a4e96f5 : SUCCESS in 55m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0b76e4ecf43943eaa7e951e0e3ca16e0 : SUCCESS in 59m 40s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/94c4bf9ceb5944289d33c031e5b4a6ec : SUCCESS in 1h 03m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/700b74a39887470cb595535fd3d7590b : SUCCESS in 1h 13m 31s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/f4b80faf5f244bab847f780d201e8a3b : SUCCESS in 1h 20m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e689a809be24a55bc50ab1841c671fa : SUCCESS in 50m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2de6808750bf4d47bf943f70646d344e : SUCCESS in 53m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a3cc1aa994294521b931b04d3e692f8e : SUCCESS in 27m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8ab1e59c84e648e69fe97010340cfcb6 : SUCCESS in 1h 07m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cf476796af484795b6764c49785b9f15 : SUCCESS in 1h 08m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad61a20f2eb68ffd74ddb37db5f2b74e12818757","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:16:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb893d30d4402df47476012fdecdac44e32c15d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 23:05:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e9c28c96b7924c11a0ec4b648ef73edc : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d33dee7b5dad460ab10a2640a552ff17 : SUCCESS in 1h 36m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7404603878e045b29e5dce00ef3312f9 : SUCCESS in 1h 52m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9da32539c52c4870b4902eca2ffa3775 : SUCCESS in 1h 16m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9beb0f1a1024aa88ed7f89fe8d85ee1 : SUCCESS in 33m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ba53f58a2d4017a48bee5fde8667ff : SUCCESS in 12m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67bc5290f7a849268a6b1cbfe8e521c7 : SUCCESS in 22m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d642b99878c451789369d84a630490a : SUCCESS in 23m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460c15dd6b6243c692a68ef76f3f5d52 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/05b243bf6f244c7fafd1601c55fe482b : SUCCESS in 46m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd33ef4998ad446e84dbaf956d5e7751 : SUCCESS in 1h 22m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d2c83e942ff84feeaa7117209373c402 : SUCCESS in 1h 02m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b4bcda52d3844729aab7cc827cf853c : SUCCESS in 52m 58s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/8d4202c73fba40d49af6f8c799261b73 : SUCCESS in 1h 22m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c401e15494ce41f9b8f486296541292f : SUCCESS in 55m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/98ee5ffe29da4ed8bb6155af0d8e60ff : SUCCESS in 58m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d3de6710ad7640f78eb615730a58a36d : SUCCESS in 34m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b91577a900d78b96d067993e1c1c5c7a3efd2d6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 23:05:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3656a3b8f0c9cd8a66567bd7b4aca7477c5030f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 23:07:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77845f95138e493eaf3a611a659523da : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f637a1f60e4f38c519184cc9ce7a0dea857a95b0","revisions":{"216897f6fe20b05bb093c0dcb5ed084d4539354a":{"kind":"REWORK","_number":1,"created":"2021-07-05 09:29:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/799446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/799446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/799446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/799446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/799446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/799446/1"}}},"commit":{"parents":[{"commit":"97618b08769636ec07b994b26af5d73bae4e5a6c","subject":"Merge \"[OVN] neutron-ovn-metadat-agent add retry logic for sb_idl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97618b08769636ec07b994b26af5d73bae4e5a6c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-05 09:24:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-05 09:29:29.000000000","tz":0},"subject":"Remove SG RPC \"use_enhanced_rpc\" check.","message":"Remove SG RPC \"use_enhanced_rpc\" check.\n\nIt\u0027s been a long time since [1] was implemented. Enhanced RPC is now\nsupported by default.\n\nCloses-Bug: #1934512\n\n[1]https://review.opendev.org/c/openstack/neutron/+/111876\n\nChange-Id: I80c3076b9545be55b11858c4422402dd5ae1a68e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/216897f6fe20b05bb093c0dcb5ed084d4539354a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/216897f6fe20b05bb093c0dcb5ed084d4539354a"}]},"branch":"refs/heads/master"},"f637a1f60e4f38c519184cc9ce7a0dea857a95b0":{"kind":"REWORK","_number":2,"created":"2021-07-06 08:10:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/799446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/799446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/799446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/799446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/799446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/799446/2"}}},"commit":{"parents":[{"commit":"97618b08769636ec07b994b26af5d73bae4e5a6c","subject":"Merge \"[OVN] neutron-ovn-metadat-agent add retry logic for sb_idl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97618b08769636ec07b994b26af5d73bae4e5a6c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-05 09:24:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-06 08:10:09.000000000","tz":0},"subject":"Remove SG RPC \"use_enhanced_rpc\" check.","message":"Remove SG RPC \"use_enhanced_rpc\" check.\n\nIt\u0027s been a long time since [1] was implemented. Enhanced RPC is now\nsupported by default.\n\nCloses-Bug: #1934512\n\n[1]https://review.opendev.org/c/openstack/neutron/+/111876\n\nChange-Id: I80c3076b9545be55b11858c4422402dd5ae1a68e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f637a1f60e4f38c519184cc9ce7a0dea857a95b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f637a1f60e4f38c519184cc9ce7a0dea857a95b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
