)]}'
{"id":"openstack%2Fneutron~799650","triplet_id":"openstack%2Fneutron~master~I5721f0986c69adf92193cd75133b19f897a186b2","project":"openstack/neutron","branch":"master","topic":"ip_version_constants","hashtags":[],"change_id":"I5721f0986c69adf92193cd75133b19f897a186b2","subject":"Use IP_VERSION_{4,6} constants in ovn_client module","status":"MERGED","created":"2021-07-06 11:56:14.000000000","updated":"2021-07-09 12:45:28.000000000","submitted":"2021-07-09 12:41:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799650-ip_version_constants","meta_rev_id":"4c32c73f715ba6c72383f182420fccb857b01a3a","_number":799650,"virtual_id_number":799650,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-09 12:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-07-06 12:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-07-06 13:01:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":2,"date":"2021-07-06 19:17:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":1,"date":"2021-07-06 19:17:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 12:31:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-06 12:46:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 13:01:10.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-06 13:01:15.000000000","updated_by":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"reviewer":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"state":"REVIEWER"},{"updated":"2021-07-06 13:36:12.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-06 13:47:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 19:17:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"39e277ed9009a743dfed96157f0304ab96dc5a11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-06 11:56:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cab74f0467f364067ca84a403e016a2c64fdc594","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-06 12:31:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"99682c780b0b1584051138fb8856fba0ca8b4608","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 12:46:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/733f72a7173d49088c81cc777b18be42 : SUCCESS in 42m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d82acf42061e4ea3b0b0679ab6487475 : SUCCESS in 43m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e93523d3b485533f5287f1ec7eb9fadcf6125567","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-06 13:01:10.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/50/799650/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/50/799650/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a662a127febe87dc06addb7d53506ff6efd59134","author":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"date":"2021-07-06 13:01:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"613b2da283ba861ebd85c30499870ad03c34dea3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-06 13:36:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799650/1 : SUCCESS in 1h 39m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1da34c475c868f455059d389469807daf06b5193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 13:47:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5dc624b9fd0042a7811ef52ba7ea051c : SUCCESS in 35m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/86c0153141b248ad9cf9fdc563373301 : SUCCESS in 1h 50m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edd2c15ff18e4441b9d6c8c785d29c93 : SUCCESS in 31m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f24fbb4dab37414594dadf5e2c3c68c6 : SUCCESS in 34m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3faebbf0ac6a4cf7b3af37a700e052ba : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06c35f088b8f4d0284510548a529e259 : SUCCESS in 23m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a81823ec7ecd479da9b25cc9932e9e0a : SUCCESS in 26m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da79b7ef85d54fdca5c468330a99f101 : SUCCESS in 23m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d818348b4de44c0d8dec79a64527b7b4 : SUCCESS in 8m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f2b608f90db466da2a411f28f4c9b39 : SUCCESS in 45m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/06898560d9f541e6aa66a84bcced6d5f : SUCCESS in 1h 03m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c378f43745a3448e8719233785844211 : SUCCESS in 27m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/583e99a456044eea9d15953dd4663b4f : SUCCESS in 1h 07m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/484c6fe7e96149cea4c740b8afb22f95 : SUCCESS in 1h 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41544013ef103d5fd12f03d3be8b7a56f8b3a682","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-06 19:17:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7958c04c9e4fcb35f554eb5b021ef115e7ee297","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 19:18:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb6282a3a763d530f3ec122ddbd48ee24f2f920c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 23:54:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/799f6b57be7140dd9ec99b857ae2ff69 : SUCCESS in 47m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d6c6fab3500a481ea991a766a193bac6 : FAILURE in 1h 38m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46976c6325ac4f79b750d181c462674a : SUCCESS in 41m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d84727cc2474e378373a6028099a0b3 : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48580fe63ec84403a54dd26b1e940145 : SUCCESS in 22m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21813df0e28840dd8d56d9d6a87fa0f3 : SUCCESS in 22m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ece49f8ba244f749be11ed23dded2ed : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/65dc62e321d74ea6aab1a3e3e55ca8b7 : SUCCESS in 42m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4183481a80fb48c399885aa782884862 : SUCCESS in 49m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/01a107dac95b45d192c94c4f8c498b63 : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":1},{"id":"79d59c2ec4246e7d13e251367a7026f94d33584c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-09 07:54:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a638753fa55128a198698c57d5afae3c4ea3cfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 08:44:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00341b78670d418db6cc38218c628dea : SUCCESS in 43m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35bdf533b20746fead635361afbfad30 : SUCCESS in 32m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"231d1d7d4dfa899c0ae1a3862ca60dcf0f9addf5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-09 09:09:27.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/50/799650/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/50/799650/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"5afb8571e066d6eee877b822e7b094b91e306127","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-09 09:42:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/799650/1 : SUCCESS in 1h 47m 23s","accounts_in_message":[],"_revision_number":1},{"id":"27e4fb493b6a1b328512b8f734fc96eb3b43be03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 09:56:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ba87243fa66e4700a2765dd77ade525e : SUCCESS in 34m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f9148549747f458db50192ac65052373 : SUCCESS in 1h 58m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2d217bd1b2f42d4916ceea29ee5f0dc : SUCCESS in 31m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e99aaa78c8e5451697bf5876ef0d8265 : SUCCESS in 37m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf6e13d3b9e410694c63c5d18fa985e : SUCCESS in 13m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d658cf7bbdd741309d86f226b92929bb : SUCCESS in 25m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81d04483582249778b8914cfe9de9aea : SUCCESS in 23m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec0f00baeca741cb9e3e8779c2b776a7 : SUCCESS in 24m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9fc51fd08504a5996291247c8642bcb : SUCCESS in 9m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b0eaa2b49f443d498fcde815bd790e4 : SUCCESS in 43m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2c1e1e6ed8654d5a9d33a43157d1f458 : SUCCESS in 54m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6f9d8b06bb6649a5b3bf68bcadc658a3 : SUCCESS in 31m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/23b7d8e532544f92bab5ab2b06a8df1a : SUCCESS in 1h 04m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0c1dfbcc36a0451383640831c81b9987 : SUCCESS in 1h 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3277a910ff2c842e9903367fd430da590accbe1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 09:57:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb28b86dc388f766879339ab4e0043642466adcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 12:41:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5cc77c77f0dd41438712c2b62dc4a80c : SUCCESS in 35m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a0ddc8a3e7744f32a080c27bbf32d40d : SUCCESS in 1h 28m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f85e808419344d488d62780ae90c97c1 : SUCCESS in 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a5f39939c1406c8b843b9623c283af : SUCCESS in 12m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01806655b1a446f6a651e30485496dae : SUCCESS in 21m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a9f7ec662c64ba6a3eb65d8b8143e2a : SUCCESS in 23m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/905fad7637a641e1b508ce38f2193613 : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf5383515e8d4ea194188110cb07cbb2 : SUCCESS in 42m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/018949ee015e47e58123b49fc5cf55f4 : SUCCESS in 54m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/38d83cf901984a7da43b7ff102e0070c : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"61fa01bab0e6a3639ac79b863306adb673d2865e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 12:41:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4c32c73f715ba6c72383f182420fccb857b01a3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 12:45:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d97a6d2cac442d839a741fe0febb5b : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"838b1d29c97f5bc4e7ec56acb1ab1adc27a1f42b","revisions":{"838b1d29c97f5bc4e7ec56acb1ab1adc27a1f42b":{"kind":"REWORK","_number":1,"created":"2021-07-06 11:56:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/50/799650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/799650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/799650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/799650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/799650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/799650/1"}}},"commit":{"parents":[{"commit":"2acb96c37468120ba79de7d8de48ffb942e78a2b","subject":"Merge \"Change fullstack dhclient lease file to tmp folder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2acb96c37468120ba79de7d8de48ffb942e78a2b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-06 11:55:00.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-06 11:56:06.000000000","tz":120},"subject":"Use IP_VERSION_{4,6} constants in ovn_client module","message":"Use IP_VERSION_{4,6} constants in ovn_client module\n\nIt wasn\u0027t used everywhere in that module so this patch fixes that.\n\nTrivialFix\n\nChange-Id: I5721f0986c69adf92193cd75133b19f897a186b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/838b1d29c97f5bc4e7ec56acb1ab1adc27a1f42b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/838b1d29c97f5bc4e7ec56acb1ab1adc27a1f42b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
