)]}'
{"id":"openstack%2Fneutron~800679","triplet_id":"openstack%2Fneutron~master~I454f3b52376a02f6cc9ca9124083295631bd03d5","project":"openstack/neutron","branch":"master","topic":"bug/1935983-rpcworker-post-fork","hashtags":[],"change_id":"I454f3b52376a02f6cc9ca9124083295631bd03d5","subject":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case","status":"MERGED","created":"2021-07-13 16:03:54.000000000","updated":"2021-07-19 21:15:26.000000000","submitted":"2021-07-15 10:27:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"800679-bug/1935983-rpcworker-post-fork","meta_rev_id":"b6166ae3fc2b586efdf3a2350c52a57f1921e416","_number":800679,"virtual_id_number":800679,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 10:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-07-14 15:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-14 17:37:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":1,"date":"2021-07-19 19:23:16.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-14 17:50:46.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-14 17:50:46.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-13 16:10:17.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-13 16:10:17.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-07-13 16:10:17.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-07-13 16:43:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-13 17:33:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-13 17:49:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-14 11:10:21.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2021-07-14 11:10:21.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-07-14 11:10:21.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2021-07-14 17:37:40.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2021-07-14 17:50:46.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2021-07-14 18:21:59.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-17 10:59:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 19:23:16.000000000","updated_by":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"reviewer":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"state":"REVIEWER"}],"messages":[{"id":"5cc700c7a75ff10f0e8afbb92ac983416bf2a187","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-13 16:03:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d49b5c668b328373bdfe42b7c5b32aa9ae0fef95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 16:43:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3125f0875c3c476fb0369a3c516100b9 : SUCCESS in 37m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b04a9df6e0964c7cae03e9ad747926cf : SUCCESS in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cef1861479e458544c5ab55469a7aac82971839d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-13 17:33:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/800679/1 : SUCCESS in 1h 28m 48s","accounts_in_message":[],"_revision_number":1},{"id":"98e629a766d6d9ea7a320a00244e420b8305f736","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-13 17:35:07.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nI\u0027m ok with this patch although I would prefer to push it only in stable releases. In master, we can document the change introduced and how to configure it using [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/795781","accounts_in_message":[],"_revision_number":1},{"id":"19f935aad550ebf0e2889140428e41e48f184711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 17:49:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e3dce95d00cb4f1e9cdcabd2e418f81a : SUCCESS in 36m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c6fff5f7864a4c92be5b1ddc446e641b : SUCCESS in 1h 44m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75d5a27e400a4603b5ca890fe542df8b : SUCCESS in 29m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c690e6b84a884a5ebf17ab2c9fc267c1 : SUCCESS in 35m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf0fd2763c0c4d61953f26c7fde0e6aa : SUCCESS in 12m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b542f5ceccff4f4595e14ad6828a223c : SUCCESS in 18m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a05e41eff3c47adb9f7cf9e569fd9cd : SUCCESS in 22m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ac8b149545b40e9baf769647c1bf9ea : SUCCESS in 22m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d6dfa23166455a95c68262ec90e575 : SUCCESS in 9m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7555784ddcec42b683c38cbafaea989e : SUCCESS in 50m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9ee1296facf146d5b555f7734a2f7944 : SUCCESS in 49m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ee6b97b44bc64011857f4b89ae03b0eb : SUCCESS in 31m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/67bf1327fc814515b199dabf1079469e : SUCCESS in 1h 08m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a6751d7feb3d4df59022a68baaad266b : FAILURE in 1h 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6c6982602c3fa10fbfc897f10f2e9db97d3911d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-14 08:14:46.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6a374de74cccad3c721a33f3978d087ae271101","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-14 09:04:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e00976cb037d5a4e01cba9d0807477eed1b1ab98","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-14 09:07:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fdf386cccdfd855b4103adce61b47496c5252a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-14 10:06:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eb042d917702b24e55d1cd02412cd46b86077f3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 10:53:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0ca8de156e94758b4ac02f97a2b63f5 : SUCCESS in 44m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f4b2f5129b04becb9d2fab4bc8e3940 : SUCCESS in 45m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f790b4587242378d1a07240479efe249012265c9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-14 11:41:15.000000000","message":"Patch Set 2: Code-Review+2\n\nI\u0027m fine with this solution.","accounts_in_message":[],"_revision_number":2},{"id":"71ccf5df3f4f9d0812cb7531645b52fe0837d8b8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-14 11:41:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/800679/2 : SUCCESS in 1h 33m 55s","accounts_in_message":[],"_revision_number":2},{"id":"e6f9db1b02ea631ac792b1473ddfe9e7ffd8d063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 11:58:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6d33b23897ff4236a953827b65348284 : SUCCESS in 47m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/582cecfb60bb404995a22d76019cbd75 : SUCCESS in 1h 29m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a515275c0bac498a8187525c57dd4f98 : SUCCESS in 1h 50m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/79f720f19b1e4992b875edbca589f0da : SUCCESS in 1h 16m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d9a69bcd16ed4e85860b11015a44b52a : SUCCESS in 1h 29m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b5e32c8cba6b49238cde39a1f9b7d847 : SUCCESS in 28m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c6790a02f88414c8decf41b2674939f : SUCCESS in 30m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b5505200e724fcba47e269dde366ea6 : SUCCESS in 31m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6b01c702964b16a2b500b0f8802fce : SUCCESS in 11m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc0c9e468d5e428cb0bc6097262fd53f : SUCCESS in 22m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2396413475ae4249ae776bf997f6424d : SUCCESS in 27m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/336c2bc65a144cfdba81f0f20ee6a212 : SUCCESS in 29m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02ca40dd5544afeb1fe988bf8b9c169 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca9a307ef32c45ac99fdb8edc4674f19 : SUCCESS in 11m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0748ff5477924661a74538e06765b681 : SUCCESS in 53m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7270b017e6c2409ebd0b8693b50f7267 : SUCCESS in 1h 25m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d3f864520103487289649d2c781d38d2 : SUCCESS in 1h 07m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d7f95c6fc3f4d67938078d3b1e5bbbc : SUCCESS in 1h 10m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/46a1602040c84618b14de31ebeeb2c0c : SUCCESS in 1h 00m 25s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dac382666dc84507a81be4cad77d07d3 : SUCCESS in 1h 05m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b97849fac33c49f48da7e42d00e30737 : SUCCESS in 1h 09m 50s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/62532229a8524d6e99043a86e66b2ddb : SUCCESS in 1h 20m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b84c5ed65414a20bc749d24682fba00 : SUCCESS in 49m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d01da767945a416e9b262a51333c9301 : SUCCESS in 53m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8b00d331e3614c7fadcf1f75b648b31e : SUCCESS in 30m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/942c2b67c5724a058cf437dc1c340209 : SUCCESS in 1h 02m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f0b2b5daa2624c609579f0830492abea : SUCCESS in 1h 08m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96e12f1b660e5f19d3fd1affd0c59bbf78767458","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-07-14 15:07:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\njust some philosophical questions, I\u0027ll +2 if we don\u0027t want to just remove all of this. :)","accounts_in_message":[],"_revision_number":2},{"id":"84342937b2361c1b330ca6bfc7f1dc39444e01c8","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-14 15:34:38.000000000","message":"Removed Code-Review+2 by Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"47c7f3a220f58d91a1e091a66ee85e20db0955f4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-14 15:35:15.000000000","message":"Patch Set 2:\n\nOk, looks like this config option is not well received and the RPCworker is needed in several scenarios.","accounts_in_message":[],"_revision_number":2},{"id":"f0af25883ef400d5191dcdddd1890610c604c6d0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2021-07-14 15:35:39.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e just some philosophical questions, I\u0027ll +2 if we don\u0027t want to just remove all of this. :)\n\nWe\u0027re using such \"weird cases\" all the time ;) - either for bare metals, or for unbound ports for DHCP for external devices. Until there will be proper DHCP for unbound(external) ports in ml2/ovn (i.e. there won\u0027t be a need for Neutron DHCP agent to use here - I would say allowing it should be default behaviour","accounts_in_message":[],"_revision_number":2},{"id":"051850ecbbabc38c07e6880d64c1eb69e6482045","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-14 15:50:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"44adaa82100df62e5ba731900b32fe5a14d949ca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-14 15:53:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"52783f69a46d3d845d4e5e435a1ba1d17af2098d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 16:40:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3facbc9fc03649a9b97238369a61efcc : SUCCESS in 37m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d2a71fbc52b44cd8a559f2edbb2ae0f : SUCCESS in 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c2cb2d03000af1731b833536b4d113bdbb0fc358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 17:19:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f61f6dc7e7d542c4a9c370de27361871 : SUCCESS in 37m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/66813f3df0864a4f89ddfda74bcacef7 : SUCCESS in 1h 27m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beb4179ceb3c420e86874fcd25160835 : SUCCESS in 38m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f3a7e030b8c4d5db6c8dc9a2f962fa0 : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272f00c7138d4000aa07821b20674724 : SUCCESS in 13m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d2ad3ea9b5c4a3eb3d760c3236740da : SUCCESS in 27m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b57624466374f98b74b3ee402d0b3f2 : SUCCESS in 20m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/190aebc3fbd94f569c15c72c10421df6 : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40af59edacf14c2a898df517650fa52e : SUCCESS in 9m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdf32e77ed894b389bf4cc384fbe189c : SUCCESS in 44m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ef9c6e304f494687bcf7547d1dded257 : SUCCESS in 1h 10m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/aee44d4032dc45a98455b44bbe5ae128 : SUCCESS in 30m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1195dc611e6f4a4abc507b17d8644849 : SUCCESS in 1h 04m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9321c4a7bf5c48358f36668d4ed7bde7 : SUCCESS in 1h 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8abf3fe364c523bcb0748518ccc41cdc3c3f5e23","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-14 17:19:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/800679/3 : SUCCESS in 1h 28m 02s","accounts_in_message":[],"_revision_number":3},{"id":"46451b0a2aa048d33f4cfa7a5bc65f1aee4318e1","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2021-07-14 17:37:40.000000000","message":"Patch Set 3: Code-Review+1\n\nThis also fixes OVN deployments that make use of the sriovnicswitch mechanism driver. I would suggest we backport it to Victoria and Ussuri too as it fixes the already backported breaking change in Ia198d45f49bddda549a0e70a3374b8339f88887b.","accounts_in_message":[],"_revision_number":3},{"id":"713677b3eb90989e4bf9c23ebfb0fe72db0f5087","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-07-14 17:50:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"81fdcb1f3f98c42ae2bcc732ab0fda1f63103a6d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-14 18:21:59.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/79/800679/3/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/79/800679/3/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"53b93e93dd46a9dd15823463a6442cc279f613f2","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-14 18:28:59.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review+1\n\u003e \n\u003e This also fixes OVN deployments that make use of the sriovnicswitch mechanism driver. I would suggest we backport it to Victoria and Ussuri too as it fixes the already backported breaking change in Ia198d45f49bddda549a0e70a3374b8339f88887b.\n\nThanks Frode,\n\nI\u0027ve already proposed backports down to stable/train https://review.opendev.org/q/256ed3b72fa5b2a8030814e9e6cf299a887ea9b7","accounts_in_message":[],"_revision_number":3},{"id":"2ed7d72a97206deae6ba6fda35c025a4144d62e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 08:21:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"71f1f1d14e195b5af14d73bb59b18023963218e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 10:27:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fcc37ecdfe21422f900841167767623b : SUCCESS in 38m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/79d7cef9d0f84621a2d5eaf6d1260865 : SUCCESS in 1h 50m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c85321f28d30458bb84ad7d64debfa3f : SUCCESS in 39m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad882c923c014f3dbc89c83c2a6ec9fc : SUCCESS in 13m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b63145cdae2d4dc0a5953507c12c4f3e : SUCCESS in 24m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5703810f36348ceaf353e548196a8ea : SUCCESS in 23m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302e203d81054cee8034879263d57894 : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b679657c6d1d4d8db6252a6312b59db1 : SUCCESS in 49m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f228821d83b54128abedee717278c991 : SUCCESS in 55m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e59831afa574446ab82cd262c38a7be3 : SUCCESS in 28m 56s","accounts_in_message":[],"_revision_number":3},{"id":"230ccf3b7d8b2dbe648a496c323e63df98d94244","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 10:27:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"42e744d068b83518f977e7049a6b62da65f8ed52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 10:30:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4076517ff9fd44d798028601383ca515 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":3},{"id":"a80889b86375f01a1e00a6690e443cbfb8d1e08f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-17 10:59:52.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/79/800679/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"dc3eaa316b0cabf5975e6709d0ca82d09c30064d","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2021-07-19 19:23:16.000000000","message":"Patch Set 3: Code-Review+1\n\nlooks good to me!","accounts_in_message":[],"_revision_number":3},{"id":"09affaa8ded3c659236f223d135e67e0424cd38d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-19 21:09:53.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b6166ae3fc2b586efdf3a2350c52a57f1921e416","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-19 21:15:26.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e5f19a29dcfb70ea08fa39267bc4aecd44bec2eb","revisions":{"9ea24676c6017592843c840efd19a3d696788135":{"kind":"REWORK","_number":1,"created":"2021-07-13 16:03:54.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/79/800679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/800679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/800679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/800679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/800679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/800679/1"}}},"commit":{"parents":[{"commit":"3cae410b3094e74549a1941e0b7d833229ae51d5","subject":"use payloads for PORT AFTER_DELETE events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cae410b3094e74549a1941e0b7d833229ae51d5"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-07-13 15:56:37.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-07-13 15:56:37.000000000","tz":60},"subject":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case","message":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case\n\nIn an attempt to minimize the number of connections to the OVSDBs the\nML2/OVN driver stopped connecting to the OVN dbs if the Neutron worker\nwas a RpcWorker but, that introduced a regression for those using\nML2/OVN + Neutron DHCP agent (for baremetal provisioning).\n\nBecause the RpcWorker didn\u0027t have a connection to the OVN database the\nNeutron DHCP agent wasn\u0027t able to create the DHCP port and failed with:\n\n2021-07-13 13:36:18.116 17 ERROR neutron.plugins.ml2.managers if not\nself._sb_ovn.chassis_exists(host):\n2021-07-13 13:36:18.116 17 ERROR neutron.plugins.ml2.managers\nAttributeError: \u0027NoneType\u0027 object has no attribute \u0027chassis_exists\u0027\n\nI understand we now have a configuration option [0] which allows us to\nmake RpcWorker to connect to the database but:\n\n1) It still makes this a regression as the baremetal use case no longer\nworks by default, operators will need to changing their deployment\nscript/tool to accomodate to it.\n\n2) We can\u0027t backport [0] as it adds a configuration option.\n\nThis patch just adds the RpcWorker back to the list of Workers that will\nconnect to the database.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/795781\n\nChange-Id: I454f3b52376a02f6cc9ca9124083295631bd03d5\nCloses-Bug: #1935983\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ea24676c6017592843c840efd19a3d696788135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ea24676c6017592843c840efd19a3d696788135"}]},"branch":"refs/heads/master"},"ecbe2596657f0c7f88483a52959406342ad320ef":{"kind":"REWORK","_number":2,"created":"2021-07-14 10:06:42.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/79/800679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/800679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/800679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/800679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/800679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/800679/2"}}},"commit":{"parents":[{"commit":"3cae410b3094e74549a1941e0b7d833229ae51d5","subject":"use payloads for PORT AFTER_DELETE events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cae410b3094e74549a1941e0b7d833229ae51d5"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-07-13 15:56:37.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-07-14 10:04:01.000000000","tz":60},"subject":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case","message":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case\n\nIn an attempt to minimize the number of connections to the OVSDBs the\nML2/OVN driver stopped connecting to the OVN dbs if the Neutron worker\nwas a RpcWorker but, that introduced a regression for those using\nML2/OVN + Neutron DHCP agent (for baremetal provisioning).\n\nBecause the RpcWorker didn\u0027t have a connection to the OVN database the\nNeutron DHCP agent wasn\u0027t able to create the DHCP port and failed with:\n\n2021-07-13 13:36:18.116 17 ERROR neutron.plugins.ml2.managers if not\nself._sb_ovn.chassis_exists(host):\n2021-07-13 13:36:18.116 17 ERROR neutron.plugins.ml2.managers\nAttributeError: \u0027NoneType\u0027 object has no attribute \u0027chassis_exists\u0027\n\nThis patch changes the default of the configuration option [0] to\nmake RpcWorkers connect to the OVN databases by default. That way,\nwe do not break people that depends on use cases using Neutron DHCP\nagent by default and allow others to disable this connection if not\nneeded.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/795781\n\nChange-Id: I454f3b52376a02f6cc9ca9124083295631bd03d5\nCloses-Bug: #1935983\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecbe2596657f0c7f88483a52959406342ad320ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecbe2596657f0c7f88483a52959406342ad320ef"}]},"branch":"refs/heads/master"},"e5f19a29dcfb70ea08fa39267bc4aecd44bec2eb":{"kind":"REWORK","_number":3,"created":"2021-07-14 15:50:28.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/79/800679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/800679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/800679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/800679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/800679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/800679/3"}}},"commit":{"parents":[{"commit":"e71ba8d2d1eddd43c86c38d21cc94208303eea7c","subject":"Revert \"Config option to enable OVN IDL on other workers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e71ba8d2d1eddd43c86c38d21cc94208303eea7c"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-07-14 15:46:47.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-07-14 15:50:16.000000000","tz":60},"subject":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case","message":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case\n\nIn an attempt to minimize the number of connections to the OVSDBs the\nML2/OVN driver stopped connecting to the OVN dbs if the Neutron worker\nwas a RpcWorker but, that introduced a regression for those using\nML2/OVN + Neutron DHCP agent (for baremetal provisioning).\n\nBecause the RpcWorker didn\u0027t have a connection to the OVN database the\nNeutron DHCP agent wasn\u0027t able to create the DHCP port and failed with:\n\n2021-07-13 13:36:18.116 17 ERROR neutron.plugins.ml2.managers if not\nself._sb_ovn.chassis_exists(host):\n2021-07-13 13:36:18.116 17 ERROR neutron.plugins.ml2.managers\nAttributeError: \u0027NoneType\u0027 object has no attribute \u0027chassis_exists\u0027\n\nThis patch adds the RpcWorker back to the list of workers that should\nconnect to the OVN databases.\n\nChange-Id: I454f3b52376a02f6cc9ca9124083295631bd03d5\nCloses-Bug: #1929633\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5f19a29dcfb70ea08fa39267bc4aecd44bec2eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5f19a29dcfb70ea08fa39267bc4aecd44bec2eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
