)]}'
{"id":"openstack%2Fneutron~800757","triplet_id":"openstack%2Fneutron~master~I7221a6238170494109975377e190cb7f2d90ec95","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2021-10-14 16:15:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7221a6238170494109975377e190cb7f2d90ec95","subject":"Bump OVN and OVS versions","status":"ABANDONED","created":"2021-07-14 09:20:25.000000000","updated":"2021-10-14 16:15:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aecd8d04f75b5fadd0c77fcbb90c156d2739e068","_number":800757,"virtual_id_number":800757,"owner":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-14 11:31:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2021-07-14 16:07:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2021-07-15 10:44:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-14 09:46:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-14 09:59:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 10:50:32.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-14 11:31:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-14 14:19:42.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2021-07-14 16:19:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-07-31 02:34:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"},{"updated":"2021-08-04 08:36:57.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"CC"},{"updated":"2021-10-14 16:15:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"a2dde535c1ce797627b2e0dabe704bf5c5410338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-07-14 09:20:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35bfb0517f138d28be7f9bd57da0130ef6f82b5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 09:59:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4416328980c479ca5462a05aac6ed9d : SUCCESS in 32m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/733dcf9f1dbb4ff8bbd7309aba48e806 : SUCCESS in 36m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37ab70f1207361f6c975be11847ca9ea4a65d7e5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-14 10:50:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/800757/1 : SUCCESS in 1h 29m 22s","accounts_in_message":[],"_revision_number":1},{"id":"91e2567e3c58892fcbadde1c696edbf7c244d2f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 11:31:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/18f3ceabd5064fc1b1d1048a26daa733 : SUCCESS in 44m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ba1a35149fe442259d8090879fef0483 : SUCCESS in 2h 01m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6748d30957914f5da21fd39067f3d6a6 : SUCCESS in 1h 52m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e17d19144bc447898badbfec3c8df1c5 : SUCCESS in 1h 42m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/68d1bf66ade24f9292a15a5cbc1689b1 : SUCCESS in 2h 05m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d44e6e05977441cbeb108319e0ff3b4 : SUCCESS in 37m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d12cf2ea7ed48b49a07956674bcbf2c : SUCCESS in 37m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd17fbd32fde40018bc48eceba530bf1 : SUCCESS in 42m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d0b2f1e341b42b28583855de9e6de6c : SUCCESS in 15m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eb3e1df57bb44f696f919d4ad644b8c : SUCCESS in 32m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a64bd71bf3544d10b353e03491f948ce : SUCCESS in 25m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a5adff773464168bb5340fef32b8ca5 : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e58c501b43c24934b6d503c061ebfa20 : SUCCESS in 10m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d5ebdec9bcb4b61b1758aa49f08ea04 : FAILURE in 50m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4afc6c5c41ea4c05aeed4019db0e60a1 : SUCCESS in 1h 23m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/94e24e4e89c24f35909a911bbc746719 : SUCCESS in 1h 13m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a964b1403454db2a42b75f9beeda8ff : SUCCESS in 1h 07m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7108afdf4fbf4b18b9522c1eb5b5126d : SUCCESS in 1h 05m 46s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b2fcd8325ef24d6c8793287068565697 : SUCCESS in 1h 05m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e64e3493e74b42a5b848d40d2e8f00ad : SUCCESS in 1h 11m 14s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/522e3cd52706430e9ee656e453c0d82c : SUCCESS in 1h 19m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b19a892b932d4fe18bc2d78c4c61bf86 : SUCCESS in 52m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/38a3a371c34a4785a6dbd9bbffd3e176 : SUCCESS in 49m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0b6e37896167451ab8cc81ef4569e898 : SUCCESS in 27m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9471cc672c545aba4eeb92259616c59 : SUCCESS in 1h 03m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8c707074ba3f44d9a91e47d9fc6548c3 : SUCCESS in 1h 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6daa0f7c0cb5aa634496bb79389e1a9ebabe9a7c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-14 12:49:27.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eacfee0dfe13f0fb3acec0e3144d0712f78b7912","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 13:37:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdb5c24fb813482db784063e6ca197c6 : SUCCESS in 29m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21ee3ab6252d401d971b9c5ff0502b95 : SUCCESS in 46m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fab5f1e08584c93ca276cc10e20489218300b01","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-07-14 14:19:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9ffc9d407fb1814764ef18018dd515cb8d2ded4d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-14 14:30:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/800757/1 : SUCCESS in 1h 40m 35s","accounts_in_message":[],"_revision_number":1},{"id":"2264a7f3688ba2152a3e2e8e0632cc1e0150a48b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 15:31:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7a47000b140a42a289e91e69ee6b013f : SUCCESS in 44m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e7efeceaa0ec412a9b8914ef79162f2c : SUCCESS in 2h 03m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b1d0f70445e14d4f84f6676605e03962 : SUCCESS in 2h 14m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f3b6e2374c814382bca56ece7509de2b : SUCCESS in 1h 30m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3b19f87f66fe49f183fd037b5e935771 : SUCCESS in 2h 35m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f3929d0660734e6d86286f4b3f3cc756 : SUCCESS in 32m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e093bc206cda452e8e81b37e5b31249e : SUCCESS in 39m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cbd0ff278604832a3fa1539c9ce668e : SUCCESS in 40m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea94860c0e3d48158281098512d527c9 : SUCCESS in 14m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f4ac8e7cfb14fe7b1622e3de14a5f59 : SUCCESS in 27m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d9e74fb462241a3b7bd7760e6611b7b : SUCCESS in 25m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c24a0ada5b246f9b252e4dd338a82c3 : SUCCESS in 26m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb180a7e5d941bc8514211e30431244 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/df89a522628e4ae7bce5632ac0958085 : FAILURE in 51m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccdd98e2f1f34717abd9490c1bf44891 : SUCCESS in 1h 29m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dbb3d8556ffe4f668142f93ffacd5caa : SUCCESS in 1h 14m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/78c2b9b4bda942eb81e829f10e9de1ce : SUCCESS in 1h 06m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/df920a8876ad4105a37c5a3e497f2418 : SUCCESS in 1h 09m 12s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6631f4ebfea34a6480af0c49c1099b7d : SUCCESS in 1h 10m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1fb0f57ce41d48aa9478962c2288dc78 : SUCCESS in 1h 21m 05s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/d4bf7aa4c7064ca892f5b6a64c6612ac : SUCCESS in 1h 37m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/394f9eb7c8204741be7452ba4803c9f3 : SUCCESS in 53m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/01ce21e3861646a9b98d3b683ddab3cf : SUCCESS in 1h 00m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2fdc3fe4a9634dd78f5fb2844a112558 : SUCCESS in 37m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/10a9e86b3d2a49ffac3ec1919353a4e4 : SUCCESS in 1h 16m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b0068daaf78741b99d06938ca3114599 : SUCCESS in 1h 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c56f7b5188fea387a84dc683ccf2bf3d134e903","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-14 16:07:47.000000000","message":"Patch Set 1: Code-Review-1\n\nFTs are failing consistently: https://269871e24a6c1ee55d57-6d5075f46c71e56c664515fc83197169.ssl.cf1.rackcdn.com/800757/1/check/neutron-functional-with-uwsgi/df89a52/testr_results.html\n\nThis is because we are retrieving chassis_private instead of chassis with this new OVN version. You should implement something like [1]:\n- chassis_private --\u003e chassis.chassis[0]\n- chassis         --\u003e chassis\n\n[1]https://review.opendev.org/c/openstack/neutron/+/797796/4/neutron/plugins/ml2/drivers/ovn/agent/neutron_agent.py","accounts_in_message":[],"_revision_number":1},{"id":"4e1fa38263d24debf65fd189062c1828f1b543f7","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-07-15 10:44:05.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"0147e2dc3ec9b516bb7b88dfc88bff526a62adb7","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-31 02:34:27.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":1},{"id":"59ab630302e774f3964fc613822e5c5398b545a3","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-08-04 08:36:57.000000000","message":"Patch Set 1:\n\nrecheck\n\nThe patch https://review.opendev.org/c/openstack/neutron/+/797796/ is now merged, let\u0027s see if it fixes the problem with chassis_private vs chassis.","accounts_in_message":[],"_revision_number":1},{"id":"0037df34a6e46575c5d29e00d47e5ea0120ab6b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 09:15:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae9984a644ba43f38adc7be9658b94ad : SUCCESS in 33m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf0c7365b8ef427780a8ecca1407ccdf : SUCCESS in 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48f749ff4a596938c8f232da854520a86162ca1f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-04 10:44:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/800757/1 : SUCCESS in 2h 07m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d23c22475410559b3b0a467c009035c93473e45d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 10:50:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8627491d1e7e4c92821fe6b2161c4e6f : SUCCESS in 42m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/79ca0e63b8484f708a482d8f9a163d82 : SUCCESS in 2h 09m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/370971995d48458d806239d8a16f7435 : SUCCESS in 1h 57m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8cef8f0e8e2f47e3af68df11cff3ee4a : SUCCESS in 1h 22m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/13944190d99e49b199e7366065aa5c0b : SUCCESS in 1h 59m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/320733476e654532b4c804709f013e76 : SUCCESS in 35m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47321039dceb4328a1f009b38fdfff19 : SUCCESS in 39m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5c6bba4135f497393fd6e13a3e0cf88 : SUCCESS in 38m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40e83b6e55a4ec9a7b491c1b44b372b : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7974cc5917046f38a4520410357ff14 : SUCCESS in 25m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c726fa38478e4aeba550394c6e257848 : SUCCESS in 24m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/430ff632935849749a8662389ec7a697 : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1b6a3729af4dada966303591c16bae : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c210087402c243d08dd35d2ff85ad166 : FAILURE in 48m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af1569f35d784752b4c128ce6f1cd479 : SUCCESS in 1h 28m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0209ad07f44b4b03aef9339ac886df3c : SUCCESS in 1h 11m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2c5f7e258c24653acb0a7a130fe9fde : SUCCESS in 1h 00m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e924ab86c4324077bffd997859d612cc : FAILURE in 1h 18m 05s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4e4c0449c5f74c33ad30b925d8b744f2 : SUCCESS in 1h 12m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15b1e290f3724ee1906d64214de64a5a : FAILURE in 1h 29m 12s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/daec0fff9d8e49899e7a1e1f88a7b0fe : SUCCESS in 1h 34m 18s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ce1588b53914713883465c888d1f406 : SUCCESS in 56m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ebe39b7bc8ff454283cb11a29bd63836 : SUCCESS in 1h 12m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/176cc9dd2e2549dba366639510e6acce : SUCCESS in 35m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/158421c83a41446c9bc080e760871717 : SUCCESS in 1h 21m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c198700376914aaf8f002f6adb0e354e : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":1},{"id":"aecd8d04f75b5fadd0c77fcbb90c156d2739e068","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-14 16:15:01.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c452652c9ae7225a9274de71cd862a8e81a6c97","revisions":{"8c452652c9ae7225a9274de71cd862a8e81a6c97":{"kind":"REWORK","_number":1,"created":"2021-07-14 09:20:25.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/57/800757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/57/800757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/800757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/800757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/800757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/57/800757/1"}}},"commit":{"parents":[{"commit":"3cae410b3094e74549a1941e0b7d833229ae51d5","subject":"use payloads for PORT AFTER_DELETE events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cae410b3094e74549a1941e0b7d833229ae51d5"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-07-14 09:18:28.000000000","tz":120},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-07-14 09:18:28.000000000","tz":120},"subject":"Bump OVN and OVS versions","message":"Bump OVN and OVS versions\n\nThis patch bumps the OVN version to the latest release v21.06.0\nand OVS to v2.15.1.\n\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: I7221a6238170494109975377e190cb7f2d90ec95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c452652c9ae7225a9274de71cd862a8e81a6c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c452652c9ae7225a9274de71cd862a8e81a6c97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
