)]}'
{"id":"openstack%2Fneutron~800780","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I1a14f217d0b6bca84cd074ef5aa4e07797fdf7e0","project":"openstack/neutron","branch":"stable/wallaby","hashtags":[],"change_id":"I1a14f217d0b6bca84cd074ef5aa4e07797fdf7e0","subject":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case","status":"ABANDONED","created":"2021-07-14 12:52:55.000000000","updated":"2021-07-15 08:56:50.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"a7f59d1e5225d83407becd530d0db47398205c67","_number":800780,"virtual_id_number":800780,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-14 17:32:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"recommended":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":2,"date":"2021-07-14 13:03:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2021-07-15 08:40:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-07-15 07:00:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-14 13:32:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-14 13:03:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-14 13:32:52.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-07-14 13:42:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-14 14:40:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-14 15:16:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-14 17:31:32.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-15 07:00:55.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-07-15 08:40:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"efb16fe620e804e771f8075ec95b1b6dc5036dad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-14 12:52:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99327eeb9a4b8d44ee61ef0b8fb1a1b0c0f7cd14","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-14 13:03:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a84b1d6678081fd5fa2cdd0df43f20b31b155301","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-07-14 13:32:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"532b88e2c0c9d5f5f8a0cccda39409474005e28e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 13:42:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5cebf2fc47a4ad0948e142706a5fc67 : SUCCESS in 46m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23c8a3e06c3d434abaf07ab6e4e7ef96 : SUCCESS in 47m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac7847380c33b9038004675cde02bfca9c7e0245","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-14 14:40:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/800780/1 : SUCCESS in 1h 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"aad2a0b950ccbc7117f608d525ebf20593b55c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 15:16:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/af05a54f5c944b0abbea73c6bc959c90 : SUCCESS in 1h 10m 41s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/bde1912d7f56414b93f9aa106abc0609 : SUCCESS in 2h 20m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/437c352eda97426dbe1e5273637d1125 : SUCCESS in 30m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a8679846fb44ad781e3e3935c62149c : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91bc63b84ebe4b799245c42aac4ef0cd : SUCCESS in 25m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a1b38794858494983d8182ebe6ad924 : SUCCESS in 23m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2225308d1acd483c8124d25c13474c89 : SUCCESS in 22m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1292aaeb46ab4433aead7cbf2d9b843b : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d0505ca1f574633b6f9e75e8175f6bf : SUCCESS in 43m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5c7e6131020b4dc7a835c1dcdebf03ff : SUCCESS in 1h 02m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6e9ab176618a46edbb4310c523ab064a : SUCCESS in 33m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/41128372638348be8a1a6dd19c8a6bea : SUCCESS in 1h 16m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0aed70dc395440c2bc4f63097ca399ef : FAILURE in 1h 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"aa12d82bfcb98ade44284639b1cd0e46be05ec28","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-14 15:42:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c49492a8f43fe4fefe4bc93725e61f7f8c7444e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 16:20:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/124140467a7c4ea7ba340f763f03f05c : SUCCESS in 33m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf4dc28f9eb24330a228dfa1cc65e687 : SUCCESS in 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6fc8eee5f98c5bb679c8d834737f315b6f897b4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-14 17:20:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/800780/1 : SUCCESS in 1h 38m 09s","accounts_in_message":[],"_revision_number":1},{"id":"389aa66970899b567c19a4f6189e0d09bbaf8d88","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-14 17:31:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/800780/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/800780/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"984d8219b5ed61cbbfee832e6a9a1f0777097a92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 17:32:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/dad5ff211f624547999f83da46cf916c : SUCCESS in 40m 59s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/66dda36f845246509e44a3993bc77303 : SUCCESS in 1h 48m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/037254a44511486887e309855ac6db9d : SUCCESS in 38m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb07b852bafd40aa977774faca8de322 : SUCCESS in 12m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fd73aad5f804259921686cbab838ba7 : SUCCESS in 26m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ec74ee77ac04375b509bf730880318b : SUCCESS in 25m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6fc3df09ab549cfbbb1d7b9afc08798 : SUCCESS in 28m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7356725b1b2247cf8de170903aeb1215 : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf29a6b7b23147b3b3a837eba6f8b225 : SUCCESS in 50m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6e1dcccb1f9142108f151e009b42dfd7 : SUCCESS in 1h 08m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5147518c0a364aa1b6c0b8bb1e8841d0 : SUCCESS in 30m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f0d20de557440f7ac75fdcae28d6fcb : SUCCESS in 1h 19m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c79eb1df516946a2954565d026a291c4 : SUCCESS in 1h 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"62d58875fa20eabf540cf932434c58145103d32e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-07-15 07:00:55.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5df360e405a96e96b96cab0ce9b6b4a9b2bd72f1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-07-15 08:40:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nSee comment below, I think for backport clarity it should just be cherry-pick from I454f3b52376a02f6cc9ca9124083295631bd03d5","accounts_in_message":[],"_revision_number":1},{"id":"416aabbb5099b39aa245b3a21aae62e4c7fc03f2","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-15 08:54:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7f59d1e5225d83407becd530d0db47398205c67","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-15 08:56:50.000000000","message":"Abandoned\n\nWe\u0027ve changed the master branch approach and now the patch is backportable","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"256ed3b72fa5b2a8030814e9e6cf299a887ea9b7","revisions":{"256ed3b72fa5b2a8030814e9e6cf299a887ea9b7":{"kind":"REWORK","_number":1,"created":"2021-07-14 12:52:55.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/80/800780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/800780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/800780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/800780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/800780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/800780/1"}}},"commit":{"parents":[{"commit":"c1390d86a0e1748b739544a6dc9b6aff070eeff7","subject":"Add fake_project_id middleware for noauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1390d86a0e1748b739544a6dc9b6aff070eeff7"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-07-14 12:46:21.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-07-14 12:52:39.000000000","tz":60},"subject":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case","message":"[OVN] Fix ML2/OVN + Neutron DHCP agent use case\n\nIn an attempt to minimize the number of connections to the OVSDBs the\nML2/OVN driver stopped connecting to the OVN dbs if the Neutron worker\nwas a RpcWorker but, that introduced a regression for those using\nML2/OVN + Neutron DHCP agent (for baremetal provisioning).\n\nBecause the RpcWorker didn\u0027t have a connection to the OVN database the\nNeutron DHCP agent wasn\u0027t able to create the DHCP port and failed with:\n\n2021-07-13 13:36:18.116 17 ERROR neutron.plugins.ml2.managers if not\nself._sb_ovn.chassis_exists(host):\n2021-07-13 13:36:18.116 17 ERROR neutron.plugins.ml2.managers\nAttributeError: \u0027NoneType\u0027 object has no attribute \u0027chassis_exists\u0027\n\nNote this fix is different from the master branch. Reason being, the\nmaster branch included a new configuration option [0] that we could\nleverage to fix this problem but that is not backportable as our guideline\ndo not allow for new features to be backported (and config options).\n\n[0] https://review.opendev.org/c/openstack/neutron/+/795781\n\nCloses-Bug: #1929633\nCloses-Bug: #1935983\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nChange-Id: I1a14f217d0b6bca84cd074ef5aa4e07797fdf7e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/256ed3b72fa5b2a8030814e9e6cf299a887ea9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/256ed3b72fa5b2a8030814e9e6cf299a887ea9b7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
