)]}'
{"id":"openstack%2Fneutron~801076","triplet_id":"openstack%2Fneutron~master~I97e9ca244c0716510fcd4ec81d54046be9c5f8f8","project":"openstack/neutron","branch":"master","topic":"bug/1936675","hashtags":[],"change_id":"I97e9ca244c0716510fcd4ec81d54046be9c5f8f8","subject":"Use explicit select condition in SQL query in \"_port_filter_hook\"","status":"MERGED","created":"2021-07-16 14:57:14.000000000","updated":"2021-07-20 21:26:13.000000000","submitted":"2021-07-20 13:31:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801076-bug/1936675","meta_rev_id":"282044264b25f4426b3dc3d863108c9fcf78ac1d","_number":801076,"virtual_id_number":801076,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-07-20 06:35:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 13:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-20 10:55:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-07-20 07:36:45.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-20 10:55:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 15:40:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 16:31:17.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-16 16:31:59.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 17:47:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 22:39:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-20 06:35:08.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-20 07:36:45.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-07-20 10:55:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"74c30e0fd2f0f82f8c9493d77130e826fd1596b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-16 14:57:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce1b430ae4489e4e026b5e31ce7a5eb07024e709","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 15:40:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e186dda426df4c13aa1d91abae6ad4eb : FAILURE in 41m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cee2fd840623414ea6cb72cf7e0b4966 : FAILURE in 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80922598325ed7e9a4bafae09c258020346098b3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-16 16:31:17.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/801076/1 : FAILURE in 1h 33m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a54a4a1670ed7cff16b46ce873ee1c0025e3578e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 16:31:59.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/76/801076/1/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/76/801076/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4cf1ae6039ae3c57dabc17f1d2e23d8af569362a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:47:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cb00a2867efa488193a54ac4d4a0526f : FAILURE in 46m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/497b9b38e9dc49fa9dc766b251c2b95d : FAILURE in 2h 20m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/77fd07bcb2f24a249965994d7e51207a : FAILURE in 1h 51m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c66e8c77c1034c23b858cd28b17dc916 : FAILURE in 1h 39m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c9a17c756b6f473f9292bfd93219fc0b : FAILURE in 2h 46m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/659081ca80bd4b0c96aae0295a15e1bb : SUCCESS in 32m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba951c542fdc471697b537451e4c3584 : FAILURE in 34m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/056dbcb2d93a41a688d761040f4398c5 : FAILURE in 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7d023882ff2478f92114e0bc22a390b : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8298d74df7b945c29fdad049854dcbe7 : FAILURE in 26m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/567bbd02dcbc484da0070b23b1e8b572 : FAILURE in 26m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42eca85375ee445eaab0ee810d33609f : FAILURE in 25m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49acf2039304247aa1f648484454a3e : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/580a2d4cf50843598daaae5d24771e95 : SUCCESS in 49m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fad07a9b07a34d4b8f16d5b47ae558b7 : SUCCESS in 1h 24m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/42ab5e46adac48ba8200a4637d954c18 : SUCCESS in 1h 07m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dec550bba8ac45f69ec09b203d3c1524 : SUCCESS in 1h 06m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2d1402d2711b45abbdfcc6934f90649c : SUCCESS in 1h 02m 40s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0e13d89bc34c449cb706bcb027290642 : SUCCESS in 1h 08m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/374978ab70b040299b70d71a4d4dda91 : FAILURE in 1h 03m 37s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/863a73c1cee9404fbe2e4cc84438b4af : SUCCESS in 1h 31m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2d0b2e54ad545c5a7ee52ff58c064b5 : SUCCESS in 53m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f7bebc44a9824020822b274f4d0353a5 : SUCCESS in 1h 01m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3abaf0bbd7214811b22540321217eb13 : SUCCESS in 32m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3949e82de60f4837ac1319f68b68480c : SUCCESS in 1h 09m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/00113c467e074c41891ec9f112f4230a : SUCCESS in 1h 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc43ef34749b3afe7ce4cfceb72e56ca967b9f3c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-16 18:32:09.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/76/801076/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/76/801076/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f970fa83a25782419ce934d584d64705cb9ca648","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-19 13:28:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2f79ec54eb06610a164e246b3a987ca52b691b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 14:09:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e70f21827135411db2531a51919ba555 : SUCCESS in 40m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e60bc6b781d47bb8a958d00b17a4eab : SUCCESS in 32m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba5406aad3e08f523072cab9b3e6127064afab8b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-19 14:58:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/801076/2 : SUCCESS in 1h 30m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6c14dd6a744536b0ce6d7a0a9eac2ad167255625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 15:22:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c4f28d6658eb41e19b4d36b4c394602c : SUCCESS in 46m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4cda60274a9249af98aecd4d3f47691f : SUCCESS in 1h 52m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c5102488e6ae4da388e3c225a8e04fde : SUCCESS in 1h 34m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6e230c398bb848faa5b64c7b9f138474 : SUCCESS in 1h 12m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b81f5381569b46b8a53d979dc2fe0875 : SUCCESS in 1h 50m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ca2b7580b6644c138141080249ad3846 : SUCCESS in 27m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d427fa84d8742f5bd6dd4d9b63c821d : SUCCESS in 38m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd057ac4336942198e8d79833e428927 : SUCCESS in 41m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db7e4c349fcb4b47b50eb80f58df15e8 : SUCCESS in 18m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d453586f92a648fab701efa3835b3882 : SUCCESS in 30m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe8fe22c73084a768c1a4a9b02f57e3e : SUCCESS in 23m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9cf5376b4f04f5fabe82e7ad79fa473 : SUCCESS in 22m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4fd5b5f9a241b0a8f88793863c0400 : SUCCESS in 8m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2d97f1e31784f13a73fc5985e46a2fc : SUCCESS in 51m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/67d4a531a3e1495886b3ad8580ff2d12 : SUCCESS in 1h 37m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5367d161b4d44b958361eea33f9580cc : SUCCESS in 1h 10m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/25b281f904284fac86c996ad13277eb0 : SUCCESS in 53m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/721357e7d4e24d478bca8be7b0ea9910 : SUCCESS in 1h 00m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/230fa51a12074c9c982882b596a0bc83 : SUCCESS in 1h 02m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bdc64b5ecfea4b939abed24b991ec1ab : FAILURE in 1h 11m 26s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/e373ea4a1737418685dda49661f5d9da : SUCCESS in 1h 42m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/044b47117835415ba04d40e4bd5b973b : SUCCESS in 54m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d7ef0459275c46f3b73b34b0229b0b59 : SUCCESS in 48m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/108efe63676d454091607475983d07b5 : SUCCESS in 24m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/55767d3d33f741ebb02464539e513ee7 : SUCCESS in 1h 10m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5ced0531b8cd46c0be52fbfff50e35e1 : SUCCESS in 1h 04m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c11da3676b7740b3846e78871aa8b4380a063246","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-19 17:24:44.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/76/801076/2/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/76/801076/2/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"85202e9bd1584e8044920ca294cbfec72ebd3bfc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-19 22:39:53.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/76/801076/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4387740e9deea1cb90af0731dc099aac69004a4b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-20 06:35:08.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/76/801076/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"ad4ad7fd5dbf26ae916f45e7e9ce3b58299461fc","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-07-20 07:36:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d9c3df1d449264c451303a7f7bd01d794765608b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 10:55:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d3e5339754ef514ac4c5e1aaaacf0dab8612b06d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 10:56:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a378484fb5aeae72fa0472b17c4a790f627efab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 13:31:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3d79af353444415ea66c167a220b90ef : SUCCESS in 41m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e36d26782d91442d9bbe63cb352e9e1c : SUCCESS in 1h 59m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fa96d3c9ff044aebacc69d32a75c43c5 : SUCCESS in 1h 42m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/842bcee2cff04a40936d157f30ea8681 : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1b03037e82e4434a90ce4a36ce5d3b22 : SUCCESS in 1h 15m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d49fc324b1143f2a19b0eb9ddf8c066 : SUCCESS in 33m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c6973a1cc140818589aa22fa2fb033 : SUCCESS in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b496a6710b3d440087c46f551d5fefef : SUCCESS in 26m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7d96e377251440eaeed9d4ab6cdf1a7 : SUCCESS in 25m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/987b231382a34575b4107ce21b802625 : SUCCESS in 11m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca2c2b3a399c4ca786405a25cf6dcfed : SUCCESS in 48m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f1329755eb64b20845d56a5753411af : SUCCESS in 1h 17m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bd019f68540e495baa29d19b8f790594 : SUCCESS in 1h 09m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/489db2e3f3164554b2359b2f5309eb5a : SUCCESS in 1h 11m 29s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/0e290d7124a64b4b94081604b5669b15 : SUCCESS in 1h 25m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b59fa90557c249b180c58ed7a6cb0821 : SUCCESS in 56m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d7258e209eee4bfd9e9ce77f0096791a : SUCCESS in 54m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4a3dd30374a34ee492af07cc11b19491 : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b71b88497dcd8a6d319bf470945145c93a17bf28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 13:31:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"67f7e2f0417a11657a1287fd180f556394ced514","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 13:33:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1d5fb4ee3740e0998b87dc2cd03a22 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2},{"id":"282044264b25f4426b3dc3d863108c9fcf78ac1d","tag":"autogenerated:gerrit:revert","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 21:26:13.000000000","message":"Created a revert of this change as I26020bbb75a97604ce347349dc956ffc5da675b0","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"923284fc3791949d93d7ded9820a44f98fc734b8","revisions":{"6a46ea09be3dae8d9ecba0fe2c7c47f4b758ca07":{"kind":"REWORK","_number":1,"created":"2021-07-16 14:57:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/801076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/801076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/801076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/801076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/801076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/801076/1"}}},"commit":{"parents":[{"commit":"d20b8708bc03fa0dfe45656b48408b3f74a67709","subject":"Merge \"Add devstack local.conf sample for ML2 OVS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d20b8708bc03fa0dfe45656b48408b3f74a67709"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-16 14:55:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-16 14:55:55.000000000","tz":0},"subject":"Use explicit select condition in SQL query in \"_port_filter_hook\"","message":"Use explicit select condition in SQL query in \"_port_filter_hook\"\n\nInstead of executing a subquery inside a select, use a proper filter\ncondition instead.\n\nCloses-Bug: #1936675\nChange-Id: I97e9ca244c0716510fcd4ec81d54046be9c5f8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a46ea09be3dae8d9ecba0fe2c7c47f4b758ca07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a46ea09be3dae8d9ecba0fe2c7c47f4b758ca07"}]},"branch":"refs/heads/master"},"923284fc3791949d93d7ded9820a44f98fc734b8":{"kind":"REWORK","_number":2,"created":"2021-07-19 13:28:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/801076/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/801076/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/801076/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/801076/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/801076/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/801076/2"}}},"commit":{"parents":[{"commit":"d20b8708bc03fa0dfe45656b48408b3f74a67709","subject":"Merge \"Add devstack local.conf sample for ML2 OVS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d20b8708bc03fa0dfe45656b48408b3f74a67709"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-16 14:55:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-19 13:27:41.000000000","tz":0},"subject":"Use explicit select condition in SQL query in \"_port_filter_hook\"","message":"Use explicit select condition in SQL query in \"_port_filter_hook\"\n\nInstead of executing a subquery inside a select, use a proper filter\ncondition instead.\n\nCloses-Bug: #1936675\nChange-Id: I97e9ca244c0716510fcd4ec81d54046be9c5f8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/923284fc3791949d93d7ded9820a44f98fc734b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/923284fc3791949d93d7ded9820a44f98fc734b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
