)]}'
{"id":"openstack%2Fneutron~801478","triplet_id":"openstack%2Fneutron~master~Id8cdd9c69e4fef2d9c335447b498958add8b7816","project":"openstack/neutron","branch":"master","topic":"bug/1578989","hashtags":[],"change_id":"Id8cdd9c69e4fef2d9c335447b498958add8b7816","subject":"Revert \"[OVN][Placement] Add a SB Chassis event to track changes in BW config\"","status":"MERGED","created":"2021-07-20 21:25:54.000000000","updated":"2021-07-22 00:59:39.000000000","submitted":"2021-07-22 00:58:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":776701,"submission_id":"801478-bug/1578989","meta_rev_id":"39062b32f94411ac0f96fe1f1e59b025f98b6ae4","_number":801478,"virtual_id_number":801478,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 00:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2021-07-21 02:12:27.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2021-07-21 02:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2021-07-21 02:54:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2021-07-20 23:42:04.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-07-20 21:25:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-20 23:38:35.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-21 02:54:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"610cbb6c0c3a6662ac1a5ebd8101b1ade7fd4add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 21:25:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6897ab00c06127df38ae00855fcc51a73ce18e34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 22:10:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb3e563bcdf94818af9a21429a7c484e : SUCCESS in 37m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61b5195b562749a7a7b649cd38f7502c : SUCCESS in 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cf91e8bb50f381a7de59e62669cc1feb51cf5a9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-20 22:57:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/801478/1 : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0474dd51c14d752e18d50be41846e1960c983a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 23:26:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b14a95de81d4b479c2600f166eff0bc : SUCCESS in 42m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/930436ca3b084243b3dd0390ea09c0a4 : SUCCESS in 1h 54m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4dad6d89e0f64c95ab63e94f3c3dd7ee : SUCCESS in 1h 53m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ef2c9b3d07314f04930d1f8dfc747b80 : SUCCESS in 1h 30m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1093d7678d044e779c5bfb8eaf3cf0c1 : SUCCESS in 1h 43m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bef6a0c023744fdda998f5d6ed97e6b2 : SUCCESS in 34m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76d30d8250ab4b63828bec445b906cf5 : SUCCESS in 42m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81e84241081f41c98a030096aeaa06c3 : SUCCESS in 40m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa2fed1c99f949b1a85e18f1349b8061 : SUCCESS in 15m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4e048aae2ba4f308cda218bee636405 : SUCCESS in 32m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14477b8791cb4d9db1f1a315a48e072c : SUCCESS in 24m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3762a2c191b747a3a7c97c9a6e918b1d : SUCCESS in 29m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ee384da382482ca3fec28c7a3c681b : SUCCESS in 9m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/11ca5f78f17049b68d93998dc5e58ba9 : SUCCESS in 50m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a6321ad399e49a598aea26e1aaf5f15 : FAILURE in 1h 12m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/388b7833d5d54d57aa9ef11ef8b5628f : SUCCESS in 1h 20m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a89487b2b72c4861a7ca6d593d3d3365 : SUCCESS in 1h 08m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ccea473f7acf4a378cca118dc0682236 : SUCCESS in 1h 05m 51s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/62e5003f53c64ff8805119b0e080eef1 : SUCCESS in 1h 13m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/60719e0fd5f842018481dd9076e459fb : FAILURE in 1h 13m 41s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/20aa0b8fdd124b21901ec5aa7f3ecc01 : SUCCESS in 1h 35m 09s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5424591c8f2540ae866eb67b63b6e4bd : SUCCESS in 56m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8c535697ba7443b198cafb4640198abc : SUCCESS in 1h 05m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f2274674800942c692f9f03d8e1c286f : SUCCESS in 37m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/33be2fc15cee4fd3a766c010748ad876 : SUCCESS in 1h 14m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1e92d8f927cd46ee8c79e67a982a846c : SUCCESS in 1h 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83a2294dcc90ce1f386e0a8824e3ec3231b332c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-20 23:38:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1fa164ede52d8650f011868158e71ddd187fe9b6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-20 23:42:04.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"dc9e408fcf3ff2a708f765420e088805b5c1d602","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 00:15:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d55286333ccf4fbb9e16ca6ce9d4a00e : SUCCESS in 32m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06bf744aa19e4d1284db78afb2bb0f23 : SUCCESS in 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4d84337b628b1f8a4505c117dbed875a0449dfba","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-21 00:16:22.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9342b3a7fbed1c56a669f42c704f20086bfd6e95","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-21 00:22:04.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"4c8ec4ab50a3b2d625c92eb4692e89f51efb1bdb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-21 01:09:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/801478/2 : SUCCESS in 1h 30m 52s","accounts_in_message":[],"_revision_number":2},{"id":"953f5d469e5bd4494f3666b92ee53e14464bc7b3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-21 01:25:55.000000000","message":"Patch Set 2:\n\nThe CI is still running, but we got good results so far.\nThe status of the key jobs to be checked are as follows:\n- neutron-ovn-tempest-slow succeeded\n- Check of nova jobs https://review.opendev.org/801550\n-- nova-live-migration and nova-ceph-multistore succeeded\n-- nova-multi-cell is still running but devstack run (which previously failed) already succeeded\n- Check of tempest jobs https://review.opendev.org/801537\n-- tempest-full-py3 succeeded ()\n-- tempest-slow-py3 is still running but devstack run (which previously failed) already succeeded","accounts_in_message":[],"_revision_number":2},{"id":"e3276e204537c95abd4714239a2d8b26ca506069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 02:05:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8049927cb778499b8a271c3d964bc59b : SUCCESS in 40m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1a16ad86fb0543b29785e2b1131d2f55 : SUCCESS in 2h 11m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/33e3b8258be2426398378653727faf53 : SUCCESS in 2h 00m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ec914cd6cf04928908a83bf3ce1cd60 : SUCCESS in 1h 32m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/10ff68b9289f48a88a027b1223b8c576 : SUCCESS in 2h 21m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e0305ace20f4e0b8e47da2d01784e98 : SUCCESS in 31m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9a7ec3ddfba46269af0ac575219900b : SUCCESS in 36m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5482aa1e4edb4c3085dfd2e89ab42951 : SUCCESS in 42m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cecd367d207f42c7b3873251ec792d55 : SUCCESS in 13m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41f387e354cd465198b084b05f173ff9 : SUCCESS in 30m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5179e4e3a4345038ffa662107dda07e : SUCCESS in 30m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e906dbbc4dd84b69b80e1e6bff87ae2e : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4e99b4c5a04161900edc8fd2db2de8 : SUCCESS in 9m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e0cef02f3564403ac62a5746a70f127 : SUCCESS in 47m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7c3df8b0f4c422394381f3ae29c907f : SUCCESS in 1h 26m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b99a5adf28b74185b01f6bda2aef1f6c : SUCCESS in 1h 17m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b885730fb4846ada82e4bb3badd6bae : SUCCESS in 1h 01m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/511466822516455b904a2d380e8ff183 : SUCCESS in 58m 10s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bcda33b222e3417788d1ed6fb4a0c93f : SUCCESS in 1h 05m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/13a2f9018ff94f0da4cd1c4f844594c0 : FAILURE in 1h 04m 49s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/b4be11f891a347ccb00877e06b6639bf : SUCCESS in 1h 25m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8106e6f224544756b59c4d84f3407796 : SUCCESS in 55m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/542ec71922604776b7583dc6d65e1458 : SUCCESS in 57m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1fd9b773b36c4ff0a6111fa40da61b31 : SUCCESS in 31m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50e8f98552094ab4b510c909604568d9 : SUCCESS in 1h 06m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/78625ecd4d4f41ec888d54c56c6d9af0 : SUCCESS in 1h 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ef0b6c1de62a13877066884adee46a0e8008c72","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-21 02:12:27.000000000","message":"Patch Set 2: Code-Review+2\n\nI confirmed all related jobs succeeded.\nLet\u0027s wait for other eyes for a while. If not, I will put it into the gate.\n\n\u003e Patch Set 2:\n\u003e \n\u003e The CI is still running, but we got good results so far.\n\u003e The status of the key jobs to be checked are as follows:\n\u003e - neutron-ovn-tempest-slow succeeded\n\u003e - Check of nova jobs https://review.opendev.org/801550\n\u003e -- nova-live-migration and nova-ceph-multistore succeeded\n\u003e -- nova-multi-cell is still running but devstack run (which previously failed) already succeeded\n\nnova-multi-cell succeeded too.\nFailing test (nova-lvm) looks unrelated to this change.\n\n\u003e - Check of tempest jobs https://review.opendev.org/801537\n\u003e -- tempest-full-py3 succeeded ()\n\u003e -- tempest-slow-py3 is still running but devstack run (which previously failed) already succeeded\n\ntempest-slow-py3 succeeded too. Failing tests (openstack-tox-py36, neutron-tempest-dvr and tempest-full-test-account-py3) looks unrelated to this change.","accounts_in_message":[],"_revision_number":2},{"id":"2afce1403e9f0d5e643c87e2f7500f83b35e8b00","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-21 02:54:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMerging as this is breaking other gates.","accounts_in_message":[],"_revision_number":2},{"id":"e0f4a9f51a425cc029c6f6e59bd5118008621d80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 02:55:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03a90081ea1feae157d05248a7d2fa0a1e303027","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-21 06:56:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"50ac1a8e91d5faf26eea54d3fb7e0f62ba11590f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 07:36:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ae9beba38954ff7a7e72d6073b4dc46 : SUCCESS in 30m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/949335125df7408e927d08a34c119f1f : SUCCESS in 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9d723279eb5eb717bfe31005d0cb3f3d6ee8847","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-21 08:24:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/801478/2 : SUCCESS in 1h 27m 02s","accounts_in_message":[],"_revision_number":2},{"id":"91705ed76d0557c8bb266f832dbeea6a6b1fe512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 09:31:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0b9b2bd33ca04d6e94b6ab275230ba0b : SUCCESS in 38m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/04340478eb014bc2a231b206727be1db : SUCCESS in 2h 32m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4f1df5bbe46d41bfac4fdf99f33573e1 : SUCCESS in 1h 29m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dac945476fd5484d8a3001edfa0dec31 : SUCCESS in 1h 53m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e0e1e8d33be440d480b48b870071bf95 : SUCCESS in 2h 05m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6582008db3c247379e091f93d1ce93f3 : SUCCESS in 49m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce7711a858f747409c2fc884f5fe912b : SUCCESS in 29m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3238160ee4864d76a65cd253e84e0697 : SUCCESS in 50m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3df96e5dc3417ca7189ba5683cf12a : SUCCESS in 12m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cba7d83b3429443bbeb19b1cc4d21d5a : SUCCESS in 40m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a8216e4ba154c60b449b9ebc363cab7 : SUCCESS in 25m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77e7ffb46c4049c985ab4024e026e0a7 : SUCCESS in 37m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99461a01434c406592e3ad7495df7fc4 : SUCCESS in 12m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/58b3997d42c5460784b2142b0bb42599 : SUCCESS in 54m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/69599b41f1864115ae9b455d117ebba6 : SUCCESS in 1h 23m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7f32eb12f470473e857feb7759476ed7 : SUCCESS in 1h 34m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac28067b31b44b79a94acbbcdf2e7bfd : SUCCESS in 56m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/47b85943417c4ce98ad8784ccc85856e : SUCCESS in 55m 08s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/763ff53c378c471788777db9639fce22 : SUCCESS in 58m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/85e4f6cf1c964d52b1a35aacaf5a668e : FAILURE in 1h 34m 13s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/3ddc512f904248d28e5fed6c47faf273 : SUCCESS in 1h 16m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f4c8a25df144e12ab5cc9c849dc5083 : SUCCESS in 51m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c8df79827e424d8daad2cb86146a0409 : SUCCESS in 45m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7d2cdc6c07a742fe8b813f326b81e857 : FAILURE in 6m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/233b7b797b014cf58ca218e46ec7abf0 : FAILURE in 7m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5f9b46883c904a159094cc7fe25572ff : FAILURE in 1h 05m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89916132b3262be0713d924b1e69d607f14da598","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-21 09:46:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1bbd511952b656eefb932d0cbd40859e1e5ee6e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 10:28:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f89d2b5edbb54e87915be13c9e637dfa : SUCCESS in 37m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e2f4b93f231445f92a6a64186856b9f : SUCCESS in 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d34483e720f6a7a030d7a9f93008922aa8bc818","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-21 11:15:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/801478/2 : SUCCESS in 1h 28m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e12136e932ecb52c2753bef3c0d6c8f66917f0b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 11:59:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9625ad7ac25f48a485b852d9614493e0 : SUCCESS in 34m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/78460debc15947a99af76e0e83038def : SUCCESS in 1h 23m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5c7e290d44244ba9a959e0500247fa7b : SUCCESS in 1h 40m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/856ec477209c44c19c992c5108f53730 : SUCCESS in 1h 14m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bf431034b7ab4ca3b1f2c2b7e93bd4de : SUCCESS in 2h 08m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/58cb4e214348487084489f7b5911d726 : SUCCESS in 27m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3ba0a120d29497fa0ddfb48746fe8ea : SUCCESS in 31m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c136adb8dd6478da9bde18131d5a6c0 : SUCCESS in 34m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ce1c5d9f6944448f2e1e6d53c88e6b : SUCCESS in 15m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5054abd7400499d8766eb7a74dbfc73 : SUCCESS in 23m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46003492a16a43479989a692f80b4d59 : SUCCESS in 26m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80ec409bab424400a96d4416686d7b13 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56fa46f890f4d48b31ed4b0eec1094f : SUCCESS in 13m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cea6fbd1032a45f5be4cf65973e775c9 : SUCCESS in 49m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9daa594435da4861b809efcb8f1721af : FAILURE in 58m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c2851bbf3ead4e339fb1a1549c931ed9 : SUCCESS in 1h 06m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/99d71cbdcacd403d9318667433625e44 : SUCCESS in 1h 02m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd5fbdd4080045c9aad7916a6e52afa0 : SUCCESS in 1h 05m 59s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ae0795be5dc64974a68f6d26dd792523 : SUCCESS in 1h 03m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d76c04af882949c9952f521efd6cad90 : SUCCESS in 1h 13m 23s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/067caf050a4745da9d75da34e4d2c1a0 : SUCCESS in 1h 23m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e23932c18e6f4597b83e3440f9f4a85e : SUCCESS in 54m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2083d8677e194edb93ede6c547941306 : SUCCESS in 51m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/285dd5ac9f98476ab4a7ddf12f20ddd7 : SUCCESS in 28m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7d0c91d60a024aafbfc7ae9f5a8a082a : SUCCESS in 1h 04m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/62f9fa80c0a6420a9f04f662126f879f : SUCCESS in 1h 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4f70687ac6658402aa55d3566a4114bf86b5b33","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-21 12:17:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"39f72c1021e9f8e4f014b29bb57acf428be7994f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 12:59:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b9d51d342a2486984345541b3973736 : SUCCESS in 37m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ee25349bd8f426aa8adc35383af35e8 : SUCCESS in 35m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81e7e9326aa24581f8f237bd1958eb275bda84dd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-21 13:48:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/801478/2 : SUCCESS in 1h 30m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6649323912d3d9ffed6b6f1643e587b4a2b66efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 14:41:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c2a62e53e3dc401888a1d1725ce82f17 : SUCCESS in 35m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0e0e342c70374201a865bcf4cb16ef6a : SUCCESS in 1h 40m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/70539c8c7f6f4ff4be3894f3a889f4b5 : SUCCESS in 2h 17m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0041a57f708540f19172d81a4ea742f6 : SUCCESS in 1h 27m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e75db6491334403097f3c65f476da5df : SUCCESS in 2h 15m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a00b1fc74ac140c7ad8e7bafb7ef487c : TIMED_OUT in 1h 04m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5265eb248c84455bffa80a886f413f0 : SUCCESS in 32m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22390a8967a440e3a4f030f3308e1e03 : SUCCESS in 40m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e00290a44ed451a82ed9da9c69fd112 : SUCCESS in 19m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8a6e8b6143d40ecb9682d82e238d92a : SUCCESS in 35m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aceea3cdb86649a1811273d05bf40494 : SUCCESS in 24m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbeede07b4654f938fd39376f9a0b1d4 : SUCCESS in 29m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fdd88cc7e346a794825ad311c22b9c : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed398a8094224330b836eb546920fbad : SUCCESS in 47m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/20cf741b5b154ccf805a3bf191a6652d : SUCCESS in 1h 29m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/df2a66064bd747f2ae30e85de07aadee : SUCCESS in 1h 06m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a9d1350924e4e3cb97d78e07c884b88 : SUCCESS in 59m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/45ed5abd14a5400bb2f60ed9f8fa7158 : SUCCESS in 1h 00m 02s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0713f54e53694398abac16a0263464c9 : SUCCESS in 1h 06m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5490fbfb45d4c8d9b0e7b8dd735e4f1 : SUCCESS in 1h 59m 48s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/cecfa21490b243998995e292638afbe3 : SUCCESS in 1h 27m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4937f931e57a4814b5e16f3d5287d780 : SUCCESS in 1h 01m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ee874159e01d4b09ab2a76152708b22f : SUCCESS in 56m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0f03c5ff699244a1973e76c00a407999 : SUCCESS in 31m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8aa3dfc0f92545a38e946cf571287535 : SUCCESS in 1h 06m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/627e150416bb4b2092e700c106be2eff : SUCCESS in 1h 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4888d33a76836940dce12bba71a144b154622070","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-21 14:48:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2d4896481e3a63aa3f5a3a047b46241ef0b0a7ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 15:31:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d3b18994b17492586aa05ec98592487 : SUCCESS in 41m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e789c10907546e68148711fa76f9697 : SUCCESS in 40m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72b24a5ffd8cbbbe63ec70a17f2b1d85865d16ee","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-21 16:18:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/801478/2 : SUCCESS in 1h 29m 53s","accounts_in_message":[],"_revision_number":2},{"id":"310b66bbec873ed0b0a5f7e8c9646e2007df3a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 18:45:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dfeb1cafeed84175a118dd59f21e95c9 : SUCCESS in 36m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e0f6ae24fcf849d2a09e7d96fdff65e4 : SUCCESS in 1h 22m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ed053c9074ff469c96194ec9ac680e45 : SUCCESS in 1h 31m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98f92dcdf4944daa85c43124bd818f21 : SUCCESS in 1h 13m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d63caf1f203e46c28277e1c1bc3569e4 : SUCCESS in 1h 59m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/086dcbe10d6e40099855d8066279fa82 : SUCCESS in 25m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efdaa5a653064f398be41eef4b21f9fd : SUCCESS in 28m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8e086909e4b4104934e5f43f903fe9f : SUCCESS in 37m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da0d44dff2404bbfac2d8f10f13ef852 : SUCCESS in 13m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65d9383e53fa47e697f5343dc35c22e0 : SUCCESS in 21m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f7a3f5e514b4bb692eb923685ee78b4 : SUCCESS in 26m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0224efcd50a145c4ab548831907d3faf : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a954a21cfca40cb988c230cdb9c3952 : SUCCESS in 9m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da5c66e5029a42b1969e47ecabfe84d4 : FAILURE in 48m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b19f671073b4fe0a9f60960bcecb077 : SUCCESS in 1h 26m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2eac7899c45d4dad94bbc0370b140028 : SUCCESS in 1h 00m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bea69e900fd244bb90b112a705969278 : SUCCESS in 1h 04m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/07eb8a112dff49f79ab1ff695998a545 : SUCCESS in 1h 06m 11s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2272b0610a6346aea2b1782937f5f5db : SUCCESS in 1h 02m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d15e295148324f25a9cf13f8ce9843f4 : FAILURE in 1h 00m 26s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/66e4c819123e4851bcf396a67e14595e : SUCCESS in 1h 21m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80134d4b075b4ece972e2f77ae9c10c5 : SUCCESS in 51m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c742d204465343eda2fd2a8c434bfe57 : SUCCESS in 50m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ae1303cecfbf46ca902103e4b657e499 : SUCCESS in 30m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c05d1be36394656b84eb09049a46526 : SUCCESS in 1h 06m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5079f022558e40199affb59d863be8c7 : SUCCESS in 1h 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"117804c759cc336a886d0e83c5f91f3d0b4b3899","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-21 18:48:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7dd419ce48c9091dc243890c718762d0a48f7ddf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 19:33:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b4f9f3c5f774c42a6fe01e1927a1e77 : SUCCESS in 40m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8628adfa2884385a7201e7fb6665a32 : SUCCESS in 42m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"900f8cd5e3c327763a132dd7380dc1a811ca9ab1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-21 20:17:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/801478/2 : SUCCESS in 1h 27m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f94c6f4329d1e1426e589ded65b461446bfa5aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 20:44:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9f86b717679647e19aabf238d3f27c9f : SUCCESS in 34m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d91aadb3941a47f289d6d971b3056c50 : SUCCESS in 1h 34m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8b34441dd3cc4315bb3071137001abc1 : SUCCESS in 1h 31m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/24075c62e8f64438bf6550435b48cdbf : SUCCESS in 1h 19m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5d8257d456c641a79b7fb847cb1f7827 : SUCCESS in 1h 52m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d2a7940640bb4ae1a485b3d330a98622 : SUCCESS in 26m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/576dc653e1b348eeaa4a49ddf100cd55 : SUCCESS in 31m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd1a73eab8224262b3c718e65aed8dd1 : SUCCESS in 35m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9331a629300b485fae8627c43409c102 : SUCCESS in 15m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6ccaec170cd4a569448e04089e2b35a : SUCCESS in 21m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e11212b88a8445adbaf42c7ff44012d6 : SUCCESS in 29m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a664eddb8d6343e59e04134babfa4920 : SUCCESS in 29m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b03d49e876457899157dd4a5e8d265 : SUCCESS in 10m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/35bdf629ff4644119093f74208a47bc7 : SUCCESS in 45m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/88b5b0d2121742ccbead0b0fa0998861 : SUCCESS in 1h 19m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f84043e8ce74453094e8eeb77aab17b9 : SUCCESS in 1h 01m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd606d6445774f1eaf6561ee67a7497a : SUCCESS in 46m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/74eb907a72ed4f03988be4a65af363f3 : SUCCESS in 57m 23s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/037ce9125e3a4e59af2df3e8506c9f42 : SUCCESS in 1h 03m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/67ba4dadc2134e0faa63b5c59fdb3afb : SUCCESS in 1h 10m 34s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/8b1c80873601438ead86bb030b7975f2 : SUCCESS in 1h 19m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b22a81571eef42e180c28c35eff22de2 : SUCCESS in 42m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b3d8dbcf80804f3aadbbf92659ef899a : SUCCESS in 55m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bbee67acc1ad4a1fa4a8da68e829968a : SUCCESS in 26m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6e1cd69cb3cd43b59b510cef2efc9f2d : SUCCESS in 1h 08m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/68be2280c70b4872a9dfc2c23d6cbd1f : SUCCESS in 1h 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5189e9a45e0f8ba2a8ec4bbf6cf06220c533a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 20:45:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c02f7e5a2373cb80d61670e25579f90946e3174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 23:15:09.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d5a68262b5efe210db0a991602e4aa3c83d293b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 00:58:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ace2057164bf4d32bcf4a4c92d5d4f3e : SUCCESS in 50m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b44ab056a77a4227ad55412aa22957be : SUCCESS in 1h 30m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7706f444c11b4439a1a37a893beddfa6 : SUCCESS in 1h 38m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5f77f0c30be74a98a10ff9dc87029259 : SUCCESS in 1h 40m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b8de451e24934439a357fd40ce71b54f : SUCCESS in 1h 38m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/985139f899f444ec87c8ceb23adfe1d1 : SUCCESS in 40m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ac316f5fb7a45d5b0305715b50bb3c3 : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35578a1fcf0d4ac49a6fc710162d19db : SUCCESS in 22m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ad1ea78cc5545ed99c6609defd4736a : SUCCESS in 23m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270b81a854514a44aaed1547aed00d0c : SUCCESS in 8m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4e23a9d2ab743b78bf500f60a8e2494 : SUCCESS in 51m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74230d5619724503a5484e300e5f4149 : SUCCESS in 1h 22m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0a26f84cbb1042b69157e532359dc388 : SUCCESS in 1h 31m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/49acdbffe60f4b41a766339d4f710be7 : SUCCESS in 1h 01m 17s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/9f4a719c7cdf42e2b6f20c1b98c6d801 : SUCCESS in 1h 37m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a834175c9c04b56a1825b1119651b6b : SUCCESS in 50m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/87ea9d8f6db94dddb34c8e38fc2693cd : SUCCESS in 51m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/88d50b80c8724e3e8857c219fbec1f6d : SUCCESS in 29m 01s","accounts_in_message":[],"_revision_number":2},{"id":"79f2986ac56edeec1c40cb14a7580d6360dc5bee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 00:58:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"39062b32f94411ac0f96fe1f1e59b025f98b6ae4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 00:59:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6560365f1cc844f8ad7e6fef91df8272 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c3b99a64fc117a3dc8f6d0ff92272183753fcf3e","revisions":{"722a5af4a10a6afec36ee21a90bc6f1c8526d2d3":{"kind":"REWORK","_number":1,"created":"2021-07-20 21:25:54.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/78/801478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/801478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/801478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/801478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/801478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/801478/1"}}},"commit":{"parents":[{"commit":"df5cb28737453e2dfee17e177160fa6d7a59b7e0","subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df5cb28737453e2dfee17e177160fa6d7a59b7e0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-20 21:25:54.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-20 21:25:54.000000000","tz":0},"subject":"DNM Revert \"[OVN][Placement] Add a SB Chassis event to track changes in BW config\"","message":"DNM Revert \"[OVN][Placement] Add a SB Chassis event to track changes in BW config\"\n\nThis reverts commit df5cb28737453e2dfee17e177160fa6d7a59b7e0.\n\nReason for revert: Just test if that will fix issue from LP#1936983\n\nChange-Id: Id8cdd9c69e4fef2d9c335447b498958add8b7816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/722a5af4a10a6afec36ee21a90bc6f1c8526d2d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/722a5af4a10a6afec36ee21a90bc6f1c8526d2d3"}]},"branch":"refs/heads/master"},"c3b99a64fc117a3dc8f6d0ff92272183753fcf3e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-20 23:38:09.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/78/801478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/801478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/801478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/801478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/801478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/801478/2"}}},"commit":{"parents":[{"commit":"df5cb28737453e2dfee17e177160fa6d7a59b7e0","subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df5cb28737453e2dfee17e177160fa6d7a59b7e0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-20 21:25:54.000000000","tz":0},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-07-20 23:38:09.000000000","tz":0},"subject":"Revert \"[OVN][Placement] Add a SB Chassis event to track changes in BW config\"","message":"Revert \"[OVN][Placement] Add a SB Chassis event to track changes in BW config\"\n\nThis reverts commit df5cb28737453e2dfee17e177160fa6d7a59b7e0.\n\nThe reverted commit triggers the failure of tempest-slow-py3 job.\ntempest-slow-py3 is equal to non-voting neutron-ovn-tempest-slow job\nin the neutron CI. It is a non-voting job so the error was not detected\nbefore merging it. To recover the tempest job in OpenStack wide,\nthis commit reverts it.\n\nSee http://lists.openstack.org/pipermail/openstack-discuss/2021-July/023764.html\n\nCloses-Bug: #1936983\nChange-Id: Id8cdd9c69e4fef2d9c335447b498958add8b7816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3b99a64fc117a3dc8f6d0ff92272183753fcf3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3b99a64fc117a3dc8f6d0ff92272183753fcf3e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
