)]}'
{"id":"openstack%2Fneutron~801479","triplet_id":"openstack%2Fneutron~master~I62eeae78e1dffc7c6adeb91b34bd9d00be25d3d9","project":"openstack/neutron","branch":"master","topic":"bug/1578989","hashtags":[],"change_id":"I62eeae78e1dffc7c6adeb91b34bd9d00be25d3d9","subject":"DNM Revert \"Move mech driver VNIC validation to SimpleAgentMechanismDriverBase\"","status":"ABANDONED","created":"2021-07-20 21:26:03.000000000","updated":"2021-07-21 06:59:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":779310,"meta_rev_id":"d7b4c6e780c64849257e807fae9be8b53bb9909f","_number":801479,"virtual_id_number":801479,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-20 23:42:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-20 21:26:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-20 21:26:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-20 21:26:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-20 21:26:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-20 21:26:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-07-20 21:26:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e13400bd77337e059a9be96d9904ef97708195d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 21:26:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df3eed75d5c27e178f2fee26c5dd483b59289cb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 22:11:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a0cb18577ee40c7ab25465f05da3274 : SUCCESS in 32m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cbf84f46b5644a16a8457114b34c184c : SUCCESS in 37m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f610233d69053b78c389a89fa3a25efdd30fcd22","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-20 22:56:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/801479/1 : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":1},{"id":"661d988fbdfd23f61bc3fe1a8520036440c6d500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 23:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aa15782d3f2346e4be575fe6e5825071 : SUCCESS in 44m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b7caed45caa34d6cbe4ca26fa7a98fe7 : SUCCESS in 2h 06m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/674ca17926c447118bf649f308f8efc1 : SUCCESS in 2h 10m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7cd3c6c51ba4659a36072c7e8ca55c8 : SUCCESS in 1h 32m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9ab50c601fdb4a0fb24c853f6a97debf : SUCCESS in 2h 10m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ec82b750f134712bbf925579e74bf8a : SUCCESS in 35m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49ea018213c94c058c2ffa749e26abd1 : SUCCESS in 40m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ee46b94e7344899257f171ed0f2bfa : SUCCESS in 15m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb904f1e742945998cc5c8073c2ba06a : SUCCESS in 26m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbae625fd74146b0acbc0080717d1643 : SUCCESS in 24m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/085c1163ad65479dbbdd975c64bd925f : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b65e98724b24144994c7c019c1fbc63 : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/112c7c4555f445c2925262dccd9ad7e6 : SUCCESS in 47m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/96f9a266e787437e924ab3cdbcf52c8f : SUCCESS in 1h 26m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/696129c3c4994b24949fc78c40cf9cc9 : SUCCESS in 1h 22m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5992401f19614280ae5d5e2ede2bee21 : SUCCESS in 1h 06m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0447a4c85d1f4f4f9eab121b831678e4 : SUCCESS in 1h 14m 28s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f2e676d8c3124d378901d0e2acdaeff8 : SUCCESS in 1h 17m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/86334c4447464c2a9f6c3c6257bbf113 : SUCCESS in 1h 19m 03s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/595325b3ed2d4136b999ca112540a80f : SUCCESS in 1h 39m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c628bf687a9943c88f453047c5337022 : SUCCESS in 56m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e82ca27117a84b77bf80ff6ef19f276e : SUCCESS in 1h 07m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1f7a4b05a8d9440d9dc48a8073a69f52 : SUCCESS in 37m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/20e9a148e71f49399804e9e8b8280a3b : SUCCESS in 1h 11m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/43c9e816f3244c6d83e44038bc734043 : FAILURE in 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"324915b787ef32f07a72a39ec2191532a016f967","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-21 00:17:48.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d7b4c6e780c64849257e807fae9be8b53bb9909f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-21 06:59:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1406c1cdc5cf941a8002a017221a2f15e2258150","revisions":{"1406c1cdc5cf941a8002a017221a2f15e2258150":{"kind":"REWORK","_number":1,"created":"2021-07-20 21:26:03.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/79/801479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/801479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/801479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/801479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/801479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/801479/1"}}},"commit":{"parents":[{"commit":"f8ca735e5f3dfb1a6cff4f23de7a554447aa7f30","subject":"Move mech driver VNIC validation to SimpleAgentMechanismDriverBase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8ca735e5f3dfb1a6cff4f23de7a554447aa7f30"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-20 21:26:03.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-20 21:26:03.000000000","tz":0},"subject":"DNM Revert \"Move mech driver VNIC validation to SimpleAgentMechanismDriverBase\"","message":"DNM Revert \"Move mech driver VNIC validation to SimpleAgentMechanismDriverBase\"\n\nThis reverts commit f8ca735e5f3dfb1a6cff4f23de7a554447aa7f30.\n\nReason for revert: Just test if that will fix issue from LP#1936983\n\nChange-Id: I62eeae78e1dffc7c6adeb91b34bd9d00be25d3d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1406c1cdc5cf941a8002a017221a2f15e2258150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1406c1cdc5cf941a8002a017221a2f15e2258150"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
