)]}'
{"id":"openstack%2Fneutron~802834","triplet_id":"openstack%2Fneutron~master~Ia74a9404411862dc88b48c4a198d5c53f5f52704","project":"openstack/neutron","branch":"master","topic":"bug/1938478","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-05-06 07:43:33.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-05-06 07:01:01.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-05-06 14:58:47.000000000","reason":"Change was submitted"},"30380":{"account":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"last_update":"2022-05-06 14:58:47.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-05-06 14:58:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia74a9404411862dc88b48c4a198d5c53f5f52704","subject":"[ovn]support read chassis update time from nb_cfg_timestamp","status":"MERGED","created":"2021-07-29 07:52:43.000000000","updated":"2022-05-06 15:00:40.000000000","submitted":"2022-05-06 14:58:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802834-bug/1938478","meta_rev_id":"ba64bfe21ac99b0546e7a5cebf3e2e72b9e2f4f2","_number":802834,"virtual_id_number":802834,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-05-06 04:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-06 14:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-05-06 07:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-06 07:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-05-06 07:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 08:27:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-29 08:58:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-29 09:26:53.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-29 10:19:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 13:22:20.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-04 07:14:34.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-14 16:14:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2022-01-18 09:45:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-02-15 12:51:10.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-03-22 11:06:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2022-04-12 03:05:58.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"bdde19b3128ef45ce8246ffe9b8eb9320e6295b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2021-07-29 07:52:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ea75f23a65d896480df15c00c9f5ad2a553e7a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-29 08:27:09.000000000","message":"Patch Set 1: Code-Review-1\n\nWe need:\n- UTs to check the timestamp conversion.\n- FTs (creating a chassis, writing the nb_cfg and checking the updated timestamp)\n- A launchpad bug","accounts_in_message":[],"_revision_number":1},{"id":"8a7202263dda5c115fbf0f47f26fdd94c39eb2d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 08:58:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae7641f2b0064f32b3fc1f7b46411605 : SUCCESS in 41m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ee1268b266248b082e02c56c0015080 : SUCCESS in 1h 00m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b276238bfc77bd1621d09af8d28f4f3ec203761","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-29 09:26:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/1 : SUCCESS in 1h 33m 33s","accounts_in_message":[],"_revision_number":1},{"id":"df32a8e3a6dbaff65dcc7dd7340f2adc7aa4528c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:19:07.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d36a782d875e4f80810a44fdf154dce1 : SUCCESS in 31m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5401b66a85f04d6f9c7a9e69ba9b5bc5 : SUCCESS in 2h 24m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce9f26948e374fcda05b05573b3e33ab : SUCCESS in 29m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b95260f76b734a079b54ce91ce6a0c86 : SUCCESS in 38m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a751bc62a34ac09c7f1c45d96a8d73 : FAILURE in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fba97f411a3c44b898c4b060360f0b5b : SUCCESS in 25m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84c2e9d1893e48d0a31b2b10c45346c4 : SUCCESS in 24m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efa8ccd009994690b6d6dae73685ff6d : SUCCESS in 24m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c167a0f669451080a7f747756bab5f : SUCCESS in 9m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9782430921d84abe8dc963ed97451ad0 : SUCCESS in 49m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cd8aa1d2f1d34e63bda171008ac0df59 : SUCCESS in 56m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fa74b8d6d1a34e188f720b557306bae8 : SUCCESS in 32m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9e3e8b4a763948a58b619e2c347b3c80 : SUCCESS in 1h 06m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cb8c3906a5fc4e19812a23a287179bd2 : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ea839d1fc025c9ec63fac977ba7e7b94389790ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-29 13:22:20.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/34/802834/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/34/802834/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"aa8f261b99e90c8023b04f083cf5da727bef40a5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2021-07-30 01:13:09.000000000","message":"Topic set to bug/1938478","accounts_in_message":[],"_revision_number":1},{"id":"c92572938e287c1dec02e759ec706bde03fbbfb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2021-07-30 09:26:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cbd9277d731ede888ed56289be57937a66331a32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 10:09:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24de9c5766924938bc0fe2730c5b79ba : SUCCESS in 30m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/caca5bd372764b3ea18ab98aadedeb53 : FAILURE in 41m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c67271029c03e7a2885c86dd4ae6ab6f28526cd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-30 10:55:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/2 : SUCCESS in 1h 28m 35s","accounts_in_message":[],"_revision_number":2},{"id":"464a00c56857efa1c164f4fb9979211fce3234ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 11:25:51.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7d5010d22fcf499fa20e3ae157e3fda1 : SUCCESS in 44m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a2ae2598ed6845bf969414b4cdeee880 : SUCCESS in 1h 57m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f4cff390c4b416f87089594dd14a494 : SUCCESS in 48m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fba0073e2f947ebbef3a4908a299636 : SUCCESS in 49m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bea727bded2040ec993bd191e640f0f2 : FAILURE in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dcd2031f343426a898401577f8507f2 : FAILURE in 25m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c273d3871a1427b8fc82cc52146647f : SUCCESS in 22m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62ba0fa1891c4b82b92a72a5cb618f5a : SUCCESS in 22m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b71b533a048245a99c0e3528de647e66 : SUCCESS in 10m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b413b89e3cb47c9b7b482d992a99072 : FAILURE in 1h 01m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5341a073ce5b4ee4b0a431aeeafbaef0 : SUCCESS in 1h 13m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/056ac7d2504f40469f3d226607e7b9de : SUCCESS in 27m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd66f8003c134bf99a9dd10ebef4da05 : SUCCESS in 1h 26m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/44607e8b1bd146e2b0a85fa762fb7f77 : SUCCESS in 1h 13m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b460fd4660c82b56b38fdef7d5c29f2b917291a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2021-08-02 11:00:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73618774a4ccc68540b12c8f6373a2f1dfd01796","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:45:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/804a1f027c3e47e78c72053f384640cd : FAILURE in 37m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18ecb92c8ee544dd94cefad21f736239 : SUCCESS in 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c10d37bdfe3893c55ced56cb3d63e3ac6d5545c8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-02 12:31:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/3 : SUCCESS in 1h 31m 08s","accounts_in_message":[],"_revision_number":3},{"id":"af77d9aef09ccab833ab43d8bbfea0890aff125b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:47:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6d3bc768e7e545f19f7606423c3602ad : SUCCESS in 40m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bf24e90d494745aaad9c59843736d9cd : SUCCESS in 1h 42m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f09b93b892394c6c9426cb4795ff4bbe : SUCCESS in 36m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35d0c40e3c124ed0b35e931ee654259c : SUCCESS in 37m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91e76fbf95d24ee384ad4d557859ef34 : SUCCESS in 14m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b68536e4b1e04777909c77bc83554c66 : SUCCESS in 25m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b532cf389d1e4fa38bc111534467fec9 : SUCCESS in 24m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cae0592c976f4f0bb0bbfb8d96c5762b : SUCCESS in 25m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d03209ffd249cc8fb2a42c2049d4ab : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/83f2bf97a60e47e3bbf4c3b44c9d3e2f : SUCCESS in 48m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3114ccfb6eac438ebc30e3a530ba6df4 : SUCCESS in 53m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f53d2caaa423464f82944eba3925a65b : SUCCESS in 31m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/96f0f8fef15a492ea29d1e60070c6167 : SUCCESS in 1h 26m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d4e1dec1fdd34c6198e60e097c03b1d6 : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":3},{"id":"fce49853b68b4f44620cc4220e0d200c9eb3e978","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-04 07:14:34.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/34/802834/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"66e6b4ee2e0da86d1e0650d1999507071007b4c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-04 13:49:54.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/34/802834/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8365f7e8cc1b3e850b34071304e0ea9c90f06792","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-05 07:26:20.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/34/802834/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"594419b7000ff9ebfd5694d412f8d186e3ae4588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2021-08-17 11:22:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c3ac09bbae7b67f223ce643c34a63ebec3a20ce9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:04:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfcad68681b84c77ae2b920d1ae01a50 : SUCCESS in 40m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8308c0e14eb4da1b01b84df207324c1 : FAILURE in 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"66c62b44bee0ece69671dd4afb6efde99548e3b9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-17 13:12:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/4 : SUCCESS in 1h 49m 41s","accounts_in_message":[],"_revision_number":4},{"id":"f3a73abe8dd620ca582eacf372b4520508b3af6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 14:05:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/deff519f7a9b4865b1e91e70c6ada7c0 : FAILURE in 40m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4fef8db63a69488292f2cb5d26a265ac : SUCCESS in 2h 40m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59972bb349654753846dcd3256977bf5 : FAILURE in 41m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90c7b6bede314ddd8eb51e66fc9367a5 : FAILURE in 43m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6761b318df604ec9a814b9abfc4db01f : SUCCESS in 13m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7102e9fd4ffc40ff98bd6e07e27d0dc3 : SUCCESS in 28m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/479d55d2ce6148ba8af47e92796f9e4c : SUCCESS in 31m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7078e3d43c4e424d8b4d54e89c379d87 : SUCCESS in 25m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f1f78c3784c481b871d3ba8d4674054 : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/743c96965bd84bf1b669dc158cb95fd9 : FAILURE in 53m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8834e77455a74c6f8fbb00f5cd8963a9 : SUCCESS in 53m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/de38dcb255564e4092322be35e94b234 : SUCCESS in 42m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8242b53292e64dc89cc45dc2dc982671 : SUCCESS in 1h 01m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d6fe4d96be9a4b2ca6ac3f2d897ba0cf : FAILURE in 22m 55s","accounts_in_message":[],"_revision_number":4},{"id":"6a22b7aaf735b4ebf8db2db398c42d8716e92a0e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-27 01:09:08.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/34/802834/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"32eb901bbe7d68a99e6037cd000e02d7ed086054","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-14 16:14:02.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":4},{"id":"e6357caf9b50fa8758cadab12f635f55a6c6a261","tag":"autogenerated:gerrit:restore","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-18 06:47:20.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"055bec33c10a149cbf67a818f6e19aba711ee524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-18 06:47:32.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"468b848d1629898040499a47c3b02d3942853378","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 07:31:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e36bc3826cd461c85831a6a8a6b07a5 : SUCCESS in 41m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ab0a23091df40feb1df3046322135d2 : FAILURE in 32m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7c7dba9c4f9828f2e3107bc00fb5e57aae883785","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-18 07:38:51.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/34/802834/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"01ccfbc710cb5a6281323a36594d8334e79eec37","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-18 08:42:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/5 : SUCCESS in 1h 54m 35s","accounts_in_message":[],"_revision_number":5},{"id":"d29169b8b10e0552e85a3dec0560103d618a84f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 08:47:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8d8bb16c3c2b415fbe59b2ec1760a50c : SUCCESS in 45m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/00812dd421084196924a078373bcffe5 : SUCCESS in 1h 03m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6911e26052374fb5810ccdef6bbe8e7e : FAILURE in 34m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39052ae54190429b8f9c6f0f80ca64b1 : SUCCESS in 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eafcb6714094af58f7a696834fcce32 : SUCCESS in 12m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c5cbbb656bf47399ed32c16c35f2baa : SUCCESS in 22m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4602ecaac4488582f513153e6adee0 : SUCCESS in 19m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f787b4d9b914768b4156404a0e5c9c3 : SUCCESS in 13m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee3f397131cf4d19984c722d9586f155 : SUCCESS in 46m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0583f6ccb7e74dfc944baf71b16ed185 : SUCCESS in 1h 57m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/16a8a709b8564bcfae7b1519b570c310 : SUCCESS in 1h 21m 05s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b44c5fd5eae04cf28f3a0f72a59f853d : SUCCESS in 19m 24s","accounts_in_message":[],"_revision_number":5},{"id":"b1c964312b14575e3d33902940e880d728fc93cd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-18 09:45:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dc96dd2085672b4d357485f4c00ed2d0c76eefa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-18 10:37:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4264bf57c3d0ee62b54e01efa4bd57a1b565cb2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 11:47:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72ae1bf866a74a27981c2fdc1097c5dc : FAILURE in 42m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c4bb9c3b8df4b4cbef237e4fc36aa69 : FAILURE in 46m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6fb2dc51bd06c0e51bd4238c57d6e7681412467f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 11:55:28.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/98938cfdde2847f480483a80465a12c2 : SUCCESS in 41m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2100dcaf79b442f58748318f427fe9a7 : SUCCESS in 1h 13m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7384d470b09d4800967e8dea0d3bb44d : FAILURE in 32m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/593152a32ecb41229af642433528e62e : FAILURE in 35m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e6e5b4e10c41d08435e9f980f3dfbb : FAILURE in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85e6c7931df74b5f862539dfb3730a09 : FAILURE in 27m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5d0188fe8944d3597e0b32c440685d0 : FAILURE in 27m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a089e8d888640b7bc0e1c225c66fe20 : SUCCESS in 11m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/74bbfe9246c14439a077b91078ef2a7b : SUCCESS in 43m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d9614640f594ed5bc3976c6a07a0fe0 : SUCCESS in 53m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/93aa80a9887f43c58ec92fe620d70d95 : SUCCESS in 1h 03m 09s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9d7db39048f4361a4d7d4f96615526c : FAILURE in 27m 30s","accounts_in_message":[],"_revision_number":6},{"id":"190c140a388ccda2d4a25852ee78e24a0af29ee4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-18 13:09:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/6 : SUCCESS in 1h 32m 17s","accounts_in_message":[],"_revision_number":6},{"id":"4783a17f9b5a64d539a674e60aca2f0850e9c512","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-18 18:48:30.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/34/802834/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"98e8a138e103de0b8bf728f67b4f18bc3219e225","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-19 01:10:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"91b242c0c5c7d927e3b5eab492f753d21867f381","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 01:49:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4dc5aca4fa44831aec204d7fd9cb0d2 : FAILURE in 33m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cec8207597864963b340077f0de78700 : SUCCESS in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d6e70393a0294fc9d5b21d2d61e99629db313a46","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-19 02:07:14.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/34/802834/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d217ae0cdab62e5597b0acb7f29e71e96eb80ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 02:15:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eac40f107b2a4fc9a2f917a1b3f4bf2e : SUCCESS in 40m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d5a4e279d706457bbfaa9fbd6b710fe1 : SUCCESS in 49m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd56bda6457f4ea283dd7147aa0283ec : FAILURE in 26m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba42d69ebc87443ea8c8be5c8b02de99 : SUCCESS in 26m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704b5f7acf134ea29cdec2abbfe975cc : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/286e94503c7445d98665807187746da4 : FAILURE in 22m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0db6ede143f45d68cee5ab0c0f7fab7 : SUCCESS in 22m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75b6bb41c634e918d95f6d96f22aa2f : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eadfe9306e9346b7890c5002b014f9e7 : SUCCESS in 42m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d99120a1eda54242b06779894c0e7c87 : SUCCESS in 52m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cff42ac7649c4de59fb70886e6e19dab : SUCCESS in 1h 02m 16s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b99afec141c84b7f9a4d214f6193c21a : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":7},{"id":"ddb2843aded7257816b8ee6d8999f01d9f5b82f4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-19 03:02:07.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/802834/7 : FAILURE in 1h 50m 35s","accounts_in_message":[],"_revision_number":7},{"id":"9d21ff4acd8bb1ae6f7c733482a166c676f969ae","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-19 07:37:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6925b7fed35c5534b47395d6537e20f919970c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 08:57:20.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/79d4ea6ad5d042339ad6a76c4ff559e1 : SUCCESS in 37m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/38817e294712423c860d0834f23d1eb4 : SUCCESS in 50m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0591ecfb0c38462186f3e1e2bf22effb : SUCCESS in 40m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e96cebac9daa45f0ad9db9ee918885d9 : FAILURE in 35m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8dc5b5f3f64f1b8b0d9733f35a4b80 : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/065ac0362e054dc187a66949298f5837 : FAILURE in 24m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20f73b88a6454252a2639b2e0de14017 : FAILURE in 28m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1156895d4be94180ad1ea606a5ab1563 : SUCCESS in 10m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d1d77151c734b14a1ac619c41f25761 : SUCCESS in 49m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1e42f5747e141adad30ab81c1a9df7f : SUCCESS in 50m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b572639bab864975b59cbd3a6a71349c : SUCCESS in 1h 14m 10s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/041fbf19bad6481faf3e909d00b80a42 : FAILURE in 26m 25s","accounts_in_message":[],"_revision_number":7},{"id":"f834aaf99ea04db9d3871fe46c355cde00000b8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 09:00:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79f7acbcc0f944ebb2184ba4d6d408e1 : SUCCESS in 35m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8477d5d750d64afaa47b177b40b295e1 : FAILURE in 1h 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fc1140162e92ddd0d853dc4b9f5bad7d5bdae988","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-19 09:48:50.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/7 : SUCCESS in 2h 10m 03s","accounts_in_message":[],"_revision_number":7},{"id":"70a23a5001f8268be3d4514e09aaf42f4c4dcd7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-19 10:59:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3fdd1fa4751fcd5f1eb720df53c5128abac04cdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 11:50:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e96fb6e6c1514582b30528304566836a : FAILURE in 41m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c7e2fce31424bc6b2f112109589c625 : SUCCESS in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"804b68dac31e27651633e72e240fe84e153d94d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 12:14:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eb45471dd9ae4c3b8f4a2a3358f09e82 : SUCCESS in 46m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e691678d49da40f4895f57ca055c1571 : SUCCESS in 44m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4d4940ccaab4b3f998ba799b6f9a507 : SUCCESS in 36m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6a5b57a252f443eb08229d8f27c1a74 : SUCCESS in 36m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e33a6afa41741ed813e8068c107eaac : SUCCESS in 14m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d332d777171a4cc5a86d1fa20e82c552 : SUCCESS in 26m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6958d416e324441391aa1411d956a839 : SUCCESS in 29m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4e11894dad4fad8d210840468a3e06 : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fdee5d00d6841ed92ed483700cb5b42 : FAILURE in 55m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31e9492d4204493bb873f53924a65392 : SUCCESS in 58m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a876a4d40d1b4d1286e1654ecd695fbd : SUCCESS in 1h 09m 18s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6920e64a3c2048f69294d2959574a782 : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":8},{"id":"0b33d6ecab922ba55d4aac2e57a3e8f09d93848a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-19 12:47:42.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/8 : SUCCESS in 1h 47m 46s","accounts_in_message":[],"_revision_number":8},{"id":"273c5c7794a8a1eb0971bd43470b18aa5aff5bd1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-19 15:09:19.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/34/802834/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"5944f2439bd5e36355fb83ad5ccaa02a041520ad","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-19 23:57:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"964fe6e93a7fa9e998943e96754fa8f7689b482f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 00:38:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e8e55f2e5cc4a8c83f3c9e28f2615ce : FAILURE in 35m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0074f705b73e4088b0b417bda9d96ea9 : SUCCESS in 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9056d18b3d52a46935eaa94063f7e7d5426802ed","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-20 01:51:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/8 : SUCCESS in 1h 54m 03s","accounts_in_message":[],"_revision_number":8},{"id":"8c02b3c7a2fd26f9073b36bac40042d04aad0618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 02:14:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5802ee25b69d4171928cf8e0ae370e00 : SUCCESS in 49m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/af07166cfd9b442abb92b4dca70c6abb : SUCCESS in 55m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f79143e22bd942b5a5516a4bf5e9efa0 : SUCCESS in 34m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60efb32b7b474b0eb0eddc9dcb24efd6 : SUCCESS in 29m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a10765734d074499a3ec9c5d774cfbc6 : SUCCESS in 10m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a19f1949fc64324a4117a7bbf2b5cc9 : SUCCESS in 27m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a832f53bde8462397aabbc40e12d8bd : SUCCESS in 25m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7d35ad4059424b81583e143ff1f709 : SUCCESS in 9m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5c40a624adf470b9e69480d87bb9123 : SUCCESS in 49m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b11cbf54a9e84d6dbc5e17c21023675a : SUCCESS in 58m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4fb578ee30e4d52a17646a070ee54e5 : SUCCESS in 2h 15m 27s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7bb963b4b414660b1f8b8cfa2478c5a : SUCCESS in 26m 39s","accounts_in_message":[],"_revision_number":8},{"id":"75b7481c76aff30260618a5429886febef3bd1bc","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-20 10:52:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"35fb21bc0e5d5e9ace398ecacc5215cdd38a5c9c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-01 09:26:55.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"cd526457364823000bfcc71c9b33069637e8d959","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-15 12:51:10.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3c186cd40b7f39cda1ce716e347ec9550df6ddc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-02-16 01:00:29.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"f156cbcb36464410cf616d7cadc0ebafcfa4b189","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-02-16 01:03:51.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4e41dc9d7f69a18b00b88f693bd4a695e11b1105","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 01:55:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae3515a33b0445f58914bf0a43418756 : SUCCESS in 52m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3181af86efa043d5af31d0057c2287a4 : SUCCESS in 52m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f332ca221456b7ce3dbeaa8df14b6749a0cdb16c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-16 02:01:02.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/34/802834/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"94d3535f11c682e983d9f98e81864e7ab0daf2ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 02:28:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0ec9fd47e530474ca1edb28abb33e897 : SUCCESS in 42m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5443c00fdfd04d2da754a34c7ab32dff : SUCCESS in 1h 07m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57f3c5ff5f4347d6986b989956c96c8c : SUCCESS in 32m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3aaed3e68de64281aeb7eb06163959f2 : FAILURE in 41m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae5e21feb1b4aeeb5867f21a734dbda : SUCCESS in 12m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a32cafc7b7a349bd969b695c7bb56eea : SUCCESS in 28m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5370e0ad4aa451f9bc86aaa7ff721ec : SUCCESS in 27m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acabb3d965b14e5595336439839f800d : SUCCESS in 8m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b74e45dfdd449e0b15b0f52867de53d : SUCCESS in 49m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dd4618cbf454cdbaac03ee98ff2d715 : SUCCESS in 1h 03m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/559a8075afa54700ae389b80a9112a87 : SUCCESS in 1h 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4570524e0dea92fc68a8ee2cff0617a2dd16dbfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-02-16 02:42:33.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"14b338538e8aeef52f12c3ff577bd3bb96581670","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-16 03:30:44.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/34/802834/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"d3560ccd0875edd35a90b73e7267167b75e8e143","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 03:55:22.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5e1485aee564a98a7a7949cb9c1360e : SUCCESS in 46m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0d136872b8044aab461da43f5153f39 : SUCCESS in 54m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"57b579659c9289014d5a924ed54f9ad4297d1be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 04:09:57.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8818723f76e6493595f625448383103a : SUCCESS in 53m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2f7b902ab7a14ee8bee6369f7289e192 : SUCCESS in 53m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f974f312c14442949dc8668f8a10d966 : SUCCESS in 39m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/916d68d8b9ba40089b6a041fdca92815 : SUCCESS in 35m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65d17e209157418fb41e6961a6e7c02f : SUCCESS in 15m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b94521ed1eb4d168a8288d42600d958 : FAILURE in 27m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a307021bb864784bbbd2abcd9425cee : SUCCESS in 25m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d19a7af6d554321a331c90fdb2b15c8 : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b402a8aa213343f78f075225fbdc110a : SUCCESS in 52m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4277ca1c3b64d8f88f3dfc92a709cf5 : SUCCESS in 59m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b16a07271f744b69b3c7f63c87d4833 : SUCCESS in 1h 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ba23ccaeef7703e09f610bab7a9c6d0f11927ea3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-16 04:16:50.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/10 : SUCCESS in 1h 33m 36s","accounts_in_message":[],"_revision_number":10},{"id":"1054c3c5b855d456172835c587fb9c87f72babeb","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-02-16 07:03:33.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"819386d6257342ec63cdb5e76d7495e74fd96a7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 07:43:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14cef73b5a39401a9cd72d0ac8e927f7 : SUCCESS in 33m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a324647e5c5b4c5cab2a71f9b17fed5f : FAILURE in 31m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6c85572ac8113446840f87668b723f522d26c415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 08:31:47.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0de4a8eb0273440a87578958a8290c91 : SUCCESS in 45m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/00e3ec2b43494afca3be9f629f177171 : SUCCESS in 1h 10m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efc8e3e472f64b93921dddf4ba515143 : SUCCESS in 32m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36f7cefc5fdb472c978b088c9eb4035b : SUCCESS in 37m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dad7038f97d43bfa9d5d7fbfcc5d505 : SUCCESS in 13m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f847d8b173343c5b7178dbfe3451c0d : SUCCESS in 37m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/661393773f854f729bcc48df64aed08d : SUCCESS in 21m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/244b2b4d9c434265baa05a665c99f9a5 : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c890f8b1d254051a77bd97e37b0d4ce : SUCCESS in 38m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aa70a9d0c5345e28ff1bd28b83e65f5 : SUCCESS in 57m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2d7df3958b174d4f80d04f95516b4972 : SUCCESS in 1h 05m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"197a3371d020b47254d1555b718f5e83849d107b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-16 08:37:14.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/10 : SUCCESS in 1h 33m 08s","accounts_in_message":[],"_revision_number":10},{"id":"abfb3f95ac30f7b08369018ccb7615f0b39aa93c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-22 12:10:12.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a219c63817b503b0be5e35395d12bf0636d70261","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-01 14:20:56.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"45dd5a8d623cc71fb107f5ba7ba8febf3fb62b26","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-01 14:21:00.000000000","message":"Patch Set 10: Code-Review-1","accounts_in_message":[],"_revision_number":10},{"id":"0f642f68c3dcd2ad4b75f4e35885273279bc2359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-02 01:27:30.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"06d81d4215c9e36803bc170e97d97c6daecb0b9f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 02:01:33.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/34/802834/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"77f6aaa6aee198126021736a75c97d33b6dd5560","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 02:10:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1fd6641fd194415b41cd83ef604bae9 : SUCCESS in 40m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab781ddbc4774481bb87b10a73524d8d : FAILURE in 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a52a92c2016dbc46feae727772253d8c6dbc6d4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 02:40:32.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6ded457c01fa44cb9713f8adef3bc25e : SUCCESS in 35m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/64210f1e91b74b139f81f293253b88d2 : SUCCESS in 1h 05m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/403b807233004f4d944d9fcc6e49c12f : FAILURE in 32m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3666f68aea8c486bb19ed28344e5afd9 : SUCCESS in 31m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543d6ea89fd84d748561f7a9e708091f : SUCCESS in 15m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d20c1f6332940cdbee6a22b17691f97 : SUCCESS in 27m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a705ee1735b4de4ba2f5febf6fae19a : SUCCESS in 26m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34cf696f1dcc4626b568b71af390c274 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/36f806edaf514e6a8d554b6b70db9fe3 : SUCCESS in 45m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/449c9d049be4452da8fa72d4dd017aed : SUCCESS in 44m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/039e0616965844f4884d168f6f1ffed8 : SUCCESS in 1h 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"53e755c4fb06e172cf6f98cd0946c5b67479d178","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 03:32:29.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/11 : SUCCESS in 2h 04m 15s","accounts_in_message":[],"_revision_number":11},{"id":"74e9062667925d79e3e2535826b1cd30aaa8e743","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-02 05:39:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c47e404fe07ea6cde547ae31ad3c11356f1ca9cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 06:33:13.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74676965b25f4279affa97a3dbc01b6f : SUCCESS in 45m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d32c8083f6234c8e8a4475095dd3057a : SUCCESS in 44m 29s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8a8d1b52e35daebb07a5c4f1b8f4af13e5e1c08b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 06:56:21.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5c053465e70a4301bcbed710b0dd97e5 : SUCCESS in 46m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d96537d913e8422d818e0f398503dd83 : SUCCESS in 57m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0b0ebdb01ce4d3e91927f37a9e67692 : SUCCESS in 37m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/539774796bd44814ae4c8b151690ba07 : SUCCESS in 34m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a0ded6f4bb421ab428c7160ddf9b92 : SUCCESS in 15m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/852859cf9d9d4ee19090a42251ab7daf : SUCCESS in 28m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73c0e11e173840babfffe81af2760208 : FAILURE in 26m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d13000a832432b92fbd26fc3b2eba9 : SUCCESS in 10m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/92462c9552bb42c29f4db49171bc8686 : SUCCESS in 48m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89d7b11cb1d143ddaf5ba7bbc037a059 : SUCCESS in 56m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3b02af364f2146d6a18428f1627a2601 : SUCCESS in 1h 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8572a6d0b6310342af9674b7dc7751a5841683ee","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 07:12:25.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/11 : SUCCESS in 1h 32m 47s","accounts_in_message":[],"_revision_number":11},{"id":"3bb90b4b2d1f7c4b18b7c1b510e4ec5a2ee3bfca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-04 01:54:54.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"fc9ed50099640d2d75e1e15fa75aefd0921b9d35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 02:48:21.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62d3ffec60cd4e8ab0aa01792231582a : SUCCESS in 51m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/694da883053340e58df0231afd0c17ec : SUCCESS in 51m 14s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"42deb292bcfc91713660e3bca2456858f45958d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 03:09:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/32c26d194e68405ebd8e3a1bde92dd0a : SUCCESS in 59m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0324eae84f0342f6b4f57d240d760c83 : SUCCESS in 1h 03m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69c1f629de9d484fb16957b53a737b28 : SUCCESS in 39m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcf7dcda22ea4c00869de313d23efcf1 : SUCCESS in 33m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38327a76a1d44d81b21f5938c4190fd6 : SUCCESS in 14m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38757577262f417192f65c03f8d6b1c8 : SUCCESS in 34m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ade3f768a5344cc599539131da28e55a : SUCCESS in 27m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d152cd00b19f45ff86e256a697d85d98 : SUCCESS in 11m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b492225248e48c69884407b6bb882ee : SUCCESS in 45m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba36b37fbdc54bda8c942a3444e7a3e2 : SUCCESS in 1h 07m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/78bdbb01e7574760908ef7c895c23204 : FAILURE in 19m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"beb230e617cbdcbb9f38b3d976077b461085b3b0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-04 03:35:03.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/12 : SUCCESS in 1h 39m 27s","accounts_in_message":[],"_revision_number":12},{"id":"74634bb653965ef88c49785664659b02e0f62ac2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-08 00:16:21.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/34/802834/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"f4a3d2fe6c69125a448c24cf0c4a1147b96b2366","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-18 10:51:46.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a416b4507a724387f263b9836c7fec0a0a04a0e9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-22 11:06:53.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"929ac6fbda6f85cbba240972edcf7e0ed9184603","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-19 10:46:12.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c430153c6be02dde8dc1f83ff1c756e0966bac08","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-19 10:46:49.000000000","message":"Removed Workflow+1 by Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":12},{"id":"a8204f622a346571497c390b20aea5970028eab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 10:47:05.000000000","message":"Patch Set 12: Verified-2\n\nThis change is unable to merge due to a missing requirement.","accounts_in_message":[],"_revision_number":12},{"id":"4c506ec4440b9e75a0558081a9aaf2c7e9fea021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-04-19 10:49:35.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"2aec36ed3d74bad61c36138f05b6f0a658d2461a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-19 11:03:55.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/802834/13 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":13},{"id":"5d0f8dc96e44c63b3599923f87778dfcdf435bb0","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-04-19 11:21:50.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"febff2b8fcf2215d80ee488a24dc17010dadee0b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-19 11:32:52.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/34/802834/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"075fbb8c1d3181f21e93c454e513016ec133b836","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 11:38:17.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05d0f70f5b764a998a8144a43e251bf8 : SUCCESS in 44m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63c8a98adc184a8bbca330aa65733ca1 : SUCCESS in 43m 51s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"019973337608b1697d3edb27867dfa6ee7deec6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:06:40.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1de1c6e113664e9fa0ffd818f7e6bd72 : SUCCESS in 43m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1740a426facf44819b9430afd8adf937 : SUCCESS in 55m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebf9128fffa44f45ae450027e28dc490 : SUCCESS in 31m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c85cf596dbd2487ab9a19c56dc738a12 : SUCCESS in 29m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a386a880948947ca871e966db7fb8b6e : SUCCESS in 14m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83eddda9b9c347df938ffca6ace395ab : SUCCESS in 23m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/337a6c34aecc4ee5800af0b632947dc5 : SUCCESS in 26m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e411ef7af47489aa046fe0e50be8d06 : SUCCESS in 25m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90703edfcdba42fabd2eb80970b3d9cb : SUCCESS in 10m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/71e3dde5b0084c67a91206effcdd9335 : SUCCESS in 56m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79ab2bef05d34fc2a4f2d3497db15a20 : SUCCESS in 1h 14m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50bd4d7ce7454017bee22fdeedfabd18 : SUCCESS in 1h 08m 30s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4c3458ae7a55e7a20a634bf71a6d3dcdc0c7be47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-04-25 02:42:06.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"076ce4e0d0c4e463bed42d90c73e56322f96c1f8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-25 03:31:04.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/802834/14 : FAILURE in 48m 22s","accounts_in_message":[],"_revision_number":14},{"id":"8ab1461aa262892de6744f992782188857487bd7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-25 03:34:34.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/34/802834/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"924b8cc22041fc3de7771bd432cfd1b95a96edc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 03:40:54.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65651c16a13e4f9f9607bf8131e1769c : SUCCESS in 54m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e19006dbe0874e21a3aa8c6cafd70442 : SUCCESS in 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a60d61fa19272714a80ea1832fffb93e86740643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 04:16:55.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/67e186d99c97423fa281477a88abaf5f : SUCCESS in 35m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/becc0b3b45374a41b0f7af03cca58d7f : SUCCESS in 1h 00m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d209234b72446c5a75e9778f78f0ddc : SUCCESS in 36m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1ad63da834b4e3d915dd8ab78202774 : SUCCESS in 29m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a820fb00639045419c7635ddd1771bc3 : FAILURE in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/119796eb0af94f828abed7408d6ed8c8 : SUCCESS in 24m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5380b25c36b14c5491e12f3304c181bc : SUCCESS in 27m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c79860e36f5b4fb185b98ce279fae82e : RETRY_LIMIT in 2m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e4f83a916f49c98ad6ec88bebb5ea0 : SUCCESS in 11m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd22bad582f246cca02497995c502833 : SUCCESS in 56m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b95d72b8ad5430da255d708ddb29bdb : SUCCESS in 1h 05m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d237eb8e1ee947ec8a62e43a13293d59 : SUCCESS in 1h 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4cc68dc5d51e617b0a904fd8d7078f95b5ce6902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-04-25 05:41:44.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"f294cdd93e8d50cf081d2de2c1244ad90809ec6f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-25 06:14:58.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/34/802834/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"6bb9013203c70c18f9cdfbcc7350bcd58c28a203","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 06:24:18.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af53dbed179e43b39877de49564385a4 : SUCCESS in 39m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b43a9854acea4b1bbca957f55e1f356b : SUCCESS in 40m 18s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1573541d46c48d0dfc63ca438ff55fe436993d47","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-25 06:40:21.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/802834/15 : FAILURE in 58m 00s","accounts_in_message":[],"_revision_number":15},{"id":"094a5c9e197bd9ae9d5f1415c300a5635e2b4f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 06:56:59.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/076d9ff8450145ea8c58a764e4e7598a : SUCCESS in 51m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5aab41e8a605407caece67bfe9ec37ce : SUCCESS in 54m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73dac1213e2d419bba12cb0d362ca686 : SUCCESS in 1h 09m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cab7c05d0d847e6a8874bd41fecac58 : SUCCESS in 36m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f812934df60e46f1aebabb6668d78bf8 : SUCCESS in 13m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4838307bbfc349598a59993bcb4d66b0 : SUCCESS in 23m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd55ecc403a94c90aa4b11e368ffc479 : SUCCESS in 23m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da006f4449f4463a91a57b9db53247b6 : SUCCESS in 22m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/391d51b7aa96412c8d957a60a7a6e2d6 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3247776977c74821a55eaaf119aa2114 : SUCCESS in 50m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eae0d09b1f1e4a26b8a9b34095e80322 : SUCCESS in 51m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c78a977fcda464a963c41ccfd297e5f : SUCCESS in 1h 06m 28s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"54116e240b66b352ae8633a188b5fa9745a0c68b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-05-05 03:19:15.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"2320d4da7e27706c3254438b810b504d1e39d870","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 04:15:21.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/145e056f140849a3a6c26137c888a769 : SUCCESS in 46m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6e6a0d8bcbd421eb20d33000f4e08b7 : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f90c15cca60b478aeba5ada33bd9c7f5c3eff031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 04:45:15.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3e9ad5a48fab40b0a578844260ac8701 : SUCCESS in 43m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/afc0ff6cf1c24af9ab1659619915476a : SUCCESS in 55m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c2349eb608540aba39560158d91053c : SUCCESS in 39m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/427b37ba6b924c05a58bdcb087e98456 : SUCCESS in 15m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ad1c72ce7b040fea6737256ea30e92e : SUCCESS in 25m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18d0260ace7f47f6aeaae03eb7055104 : SUCCESS in 37m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f27381a813b43bda77c7c0a958b1ef3 : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b40d6ea307649e8acfacb164af48827 : SUCCESS in 10m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b65dbbc7ab24b65857f4dd0e0a2e04c : SUCCESS in 54m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9facaeb0de04887804f8ed4237d6475 : SUCCESS in 1h 09m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4a65d4212b754c36a67630d9000b19d3 : SUCCESS in 1h 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d630d8c6fc2102309cc355788759eccc957f66b5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-05 05:04:51.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/802834/16 : FAILURE in 1h 45m 00s","accounts_in_message":[],"_revision_number":16},{"id":"04f70233101f04cd95e30fff4f2c25f918cd4683","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-05 16:07:22.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/34/802834/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"957fd7e5b716a37cce104c2c40486b129e55fc2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-05-06 02:37:13.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"eb76ca8175f2f6e5a82b303b79586567651402a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:12:54.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11c3a1ccb0054f7d99f35d1429ace03f : SUCCESS in 41m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27cf3834a4d64abeb5c5bd5310b9dda0 : SUCCESS in 42m 45s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"acef916491149631a74358bf8abd3a9dcf493887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:30:53.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a56f833d253642d287ee10cd9e738a08 : SUCCESS in 49m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/acec43a56ffc49f2b00fc1e15a1c5ae3 : SUCCESS in 50m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92f5bca72ff44e9eabd9d49c30c13217 : SUCCESS in 39m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7338cd8b3c04001a7f0e09483688100 : SUCCESS in 13m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2852c9abcc1d4225a690390daf9a2159 : SUCCESS in 21m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9268665605a44bceaafe86ed89b157cb : SUCCESS in 20m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efe9745c8e7448cda6ea2d0b7f99021e : SUCCESS in 25m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb42731f873a46b4a8cc5d54450c4e16 : SUCCESS in 10m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85446fe606cf4e5791dad6f43cd0441d : SUCCESS in 57m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a2b05d947ed4021a473a72cfa63460a : SUCCESS in 1h 05m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd7e9bfb43f44439a2133735c3f505ff : SUCCESS in 1h 16m 00s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4781f5ccecd4f0cd34fc7406090955fbe3926c18","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-06 04:36:03.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/17 : SUCCESS in 1h 58m 14s","accounts_in_message":[],"_revision_number":17},{"id":"bfba2f926f00c9ff48bb6a063113faccccc55571","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-06 04:56:18.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/34/802834/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"602bffaa83b0a3210663c7f8798f6b4bff7a0e95","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-06 07:01:01.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"1fa08a83d1bd5725c8bc1be9873afd47e7ccec8f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-06 07:43:33.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"77a26dae7e59def68a92d09940ea0382a5a2ca04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 07:44:15.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"077e3faf379ad062c7674c70db56a2eb4a424105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 09:12:10.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0e41c255cda74e448a25b96c3c3b46f3 : SUCCESS in 45m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d0fd3c660282413c9332a56efee21211 : SUCCESS in 1h 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b1c9234cca4f07a147f7008efd60da : SUCCESS in 13m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d36bdeb7ce94aa887a3c157b8566728 : SUCCESS in 23m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49e07114dcc74e20b0868e2ac0f6d381 : SUCCESS in 23m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef96b0a36233416fb4bd7d59e84d9592 : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b8085a74f634d709f4720defefa5ca9 : FAILURE in 50m 13s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9ddf6bfa426429d830a280aee28e07a : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":17},{"id":"c4af2054ab9f4de521db4216c60941a7e090eabc","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-05-06 09:13:56.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e69462ad56943ae21cf2beaa5780d5182565a1d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 10:03:44.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c60344ba0ed448b1ab1521d5f9d6c284 : SUCCESS in 46m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e803d49b31ce4a6a974cfbb51610a07a : SUCCESS in 44m 49s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0ce326579d318b73720db44ae1df75f208b46b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 10:29:04.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6fc540fee09f4b6995bf0dd1f44ae65f : SUCCESS in 44m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0946fa3cc717497ba5fa507d814cca3b : SUCCESS in 1h 11m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ce44706dcb0456aba2fddc95c421c29 : SUCCESS in 36m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5887ce177b62435e804bbaadac3af89e : SUCCESS in 16m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dbb7a765c684c86b2be9cb0dda3172a : SUCCESS in 30m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c884b659812b4a38b3bb64b195edf8a7 : SUCCESS in 29m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18ce2658585743e4aa08156fe545c4d3 : SUCCESS in 27m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2de4f1500bce44d38d5f929e649b196c : SUCCESS in 10m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc03afff2482411395c68efa5c3f6fdd : SUCCESS in 50m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6889003e996b435a925973e268979459 : SUCCESS in 56m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4ea77898cb7c4635b0e34edc9fd6a59c : SUCCESS in 1h 05m 54s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7c571bc08f4d1b44bd8b73f2a6aee23312ecb9ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 10:30:00.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"3ef8de0eb8a875fd3ab66ec5fbd98823445a0189","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-06 11:02:53.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802834/17 : SUCCESS in 1h 43m 16s","accounts_in_message":[],"_revision_number":17},{"id":"cb1f1c245ffe5016ea10648549a4f22443d076c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 14:58:47.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/babdf0862f984383a929a239dded5ead : SUCCESS in 51m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8b917664014448b7a070af2a6d2f3ea4 : SUCCESS in 1h 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f242e43063e4a218f43fdaeb82ce7c5 : SUCCESS in 15m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e850424435d541c29800690e73ce8963 : SUCCESS in 27m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5a318cbf32543249f4a158f6d1f93ae : SUCCESS in 28m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a96b9ab509b4b45afe7530bc5383ff6 : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcedbbe013354c35ab9cb781db09d47a : SUCCESS in 57m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09cd896b253d4c8e8e13c3217ff4d4b7 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":17},{"id":"48995a728c424af8b759ce4b92ddc226c18b1494","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 14:58:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"ba64bfe21ac99b0546e7a5cebf3e2e72b9e2f4f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 15:00:40.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c70d891a2a8a40f0981e011b366048ed : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"9e263dcf00b7ddc102a54f5e681bcf09931f3a72","revisions":{"a69ebed777add36bb1810151073a18dfe29a9b15":{"kind":"REWORK","_number":1,"created":"2021-07-29 07:52:43.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/1"}}},"commit":{"parents":[{"commit":"0411743bb974e5a352dede939d3a824d09234ead","subject":"Merge \"Sanitize MAC addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0411743bb974e5a352dede939d3a824d09234ead"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:46.000000000","tz":480},"subject":"[ovn]support read chassis update time from nb_cfg_timetamp","message":"[ovn]support read chassis update time from nb_cfg_timetamp\n\nnb_cfg_timetamp can better reflect the status of chassis\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a69ebed777add36bb1810151073a18dfe29a9b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a69ebed777add36bb1810151073a18dfe29a9b15"}]},"branch":"refs/heads/master"},"a166173a9b762c9b16d536f857b95d5a2d17f9bd":{"kind":"REWORK","_number":2,"created":"2021-07-30 09:26:06.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/2"}}},"commit":{"parents":[{"commit":"0411743bb974e5a352dede939d3a824d09234ead","subject":"Merge \"Sanitize MAC addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0411743bb974e5a352dede939d3a824d09234ead"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-30 09:24:38.000000000","tz":480},"subject":"[WIP][ovn]support read chassis update time from nb_cfg_timetamp","message":"[WIP][ovn]support read chassis update time from nb_cfg_timetamp\n\nnb_cfg_timetamp can better reflect the status of chassis\n\nPartial-bug: #1938478\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a166173a9b762c9b16d536f857b95d5a2d17f9bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a166173a9b762c9b16d536f857b95d5a2d17f9bd"}]},"branch":"refs/heads/master"},"2f3dd776994c9887383517ebba08ecfabd214454":{"kind":"REWORK","_number":3,"created":"2021-08-02 11:00:03.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/3"}}},"commit":{"parents":[{"commit":"0411743bb974e5a352dede939d3a824d09234ead","subject":"Merge \"Sanitize MAC addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0411743bb974e5a352dede939d3a824d09234ead"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-08-02 10:59:39.000000000","tz":480},"subject":"[WIP][ovn]support read chassis update time from nb_cfg_timetamp","message":"[WIP][ovn]support read chassis update time from nb_cfg_timetamp\n\nnb_cfg_timetamp can better reflect the status of chassis\n\nPartial-bug: #1938478\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f3dd776994c9887383517ebba08ecfabd214454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f3dd776994c9887383517ebba08ecfabd214454"}]},"branch":"refs/heads/master"},"bbd7a5920ea34e9d742e48115187b4ad561f6840":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-17 11:22:25.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/4"}}},"commit":{"parents":[{"commit":"96f1ea140edb1386159ce8b103c48b4e6a816ca1","subject":"Merge \"Remove ``ConfDriver`` code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96f1ea140edb1386159ce8b103c48b4e6a816ca1"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2021-08-17 11:22:25.000000000","tz":0},"subject":"[WIP][ovn]support read chassis update time from nb_cfg_timetamp","message":"[WIP][ovn]support read chassis update time from nb_cfg_timetamp\n\nnb_cfg_timetamp can better reflect the status of chassis\n\nPartial-bug: #1938478\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbd7a5920ea34e9d742e48115187b4ad561f6840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbd7a5920ea34e9d742e48115187b4ad561f6840"}]},"branch":"refs/heads/master","description":"Rebase"},"384a428f9bdf9afc29a2f40a8bb56e27279a77c6":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-18 06:47:32.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/5"}}},"commit":{"parents":[{"commit":"a9730576b1d86e995ce4cf0db165a9830e2ba800","subject":"Merge \"Exclude router gw subnet ports from port list while port delete.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9730576b1d86e995ce4cf0db165a9830e2ba800"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2022-01-18 06:47:32.000000000","tz":0},"subject":"[WIP][ovn]support read chassis update time from nb_cfg_timetamp","message":"[WIP][ovn]support read chassis update time from nb_cfg_timetamp\n\nnb_cfg_timetamp can better reflect the status of chassis\n\nPartial-bug: #1938478\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/384a428f9bdf9afc29a2f40a8bb56e27279a77c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/384a428f9bdf9afc29a2f40a8bb56e27279a77c6"}]},"branch":"refs/heads/master","description":"Rebase"},"ddb76cce155b422ca75a5c0184c695aa6b03c8bb":{"kind":"REWORK","_number":6,"created":"2022-01-18 10:37:17.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/6"}}},"commit":{"parents":[{"commit":"a9730576b1d86e995ce4cf0db165a9830e2ba800","subject":"Merge \"Exclude router gw subnet ports from port list while port delete.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9730576b1d86e995ce4cf0db165a9830e2ba800"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 10:33:12.000000000","tz":480},"subject":"[ovn]support read chassis update time from nb_cfg_timetamp","message":"[ovn]support read chassis update time from nb_cfg_timetamp\n\nnb_cfg_timetamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddb76cce155b422ca75a5c0184c695aa6b03c8bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddb76cce155b422ca75a5c0184c695aa6b03c8bb"}]},"branch":"refs/heads/master"},"489c6878193a23dd785ffbce0ec492853078ff97":{"kind":"REWORK","_number":7,"created":"2022-01-19 01:10:49.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/7"}}},"commit":{"parents":[{"commit":"a9730576b1d86e995ce4cf0db165a9830e2ba800","subject":"Merge \"Exclude router gw subnet ports from port list while port delete.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9730576b1d86e995ce4cf0db165a9830e2ba800"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-19 01:10:17.000000000","tz":480},"subject":"[ovn]support read chassis update time from nb_cfg_timetamp","message":"[ovn]support read chassis update time from nb_cfg_timetamp\n\nnb_cfg_timetamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/489c6878193a23dd785ffbce0ec492853078ff97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/489c6878193a23dd785ffbce0ec492853078ff97"}]},"branch":"refs/heads/master"},"d82b38b5a3cca3a3b4dcfd68b89333f0eef06e4d":{"kind":"REWORK","_number":8,"created":"2022-01-19 10:59:05.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/8"}}},"commit":{"parents":[{"commit":"a9730576b1d86e995ce4cf0db165a9830e2ba800","subject":"Merge \"Exclude router gw subnet ports from port list while port delete.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9730576b1d86e995ce4cf0db165a9830e2ba800"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-19 10:58:44.000000000","tz":480},"subject":"[ovn]support read chassis update time from nb_cfg_timetamp","message":"[ovn]support read chassis update time from nb_cfg_timetamp\n\nnb_cfg_timetamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d82b38b5a3cca3a3b4dcfd68b89333f0eef06e4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d82b38b5a3cca3a3b4dcfd68b89333f0eef06e4d"}]},"branch":"refs/heads/master"},"59134c967ca7f3129b5a447d70c2f28d552968e6":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2022-02-16 01:00:29.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/9"}}},"commit":{"parents":[{"commit":"a9730576b1d86e995ce4cf0db165a9830e2ba800","subject":"Merge \"Exclude router gw subnet ports from port list while port delete.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9730576b1d86e995ce4cf0db165a9830e2ba800"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2022-02-16 01:00:29.000000000","tz":0},"subject":"[ovn]support read chassis update time from nb_cfg_timestamp","message":"[ovn]support read chassis update time from nb_cfg_timestamp\n\nnb_cfg_timestamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis.\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59134c967ca7f3129b5a447d70c2f28d552968e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59134c967ca7f3129b5a447d70c2f28d552968e6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fbe784eaa91ad3f69860df062e4986035e2aaa5a":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-02-16 02:42:33.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/10"}}},"commit":{"parents":[{"commit":"1df18a885e8af6a55e98d9f56d1c042dcda12db3","subject":"Merge \"Updating python testing as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1df18a885e8af6a55e98d9f56d1c042dcda12db3"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2022-02-16 02:42:33.000000000","tz":0},"subject":"[ovn]support read chassis update time from nb_cfg_timestamp","message":"[ovn]support read chassis update time from nb_cfg_timestamp\n\nnb_cfg_timestamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis.\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbe784eaa91ad3f69860df062e4986035e2aaa5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbe784eaa91ad3f69860df062e4986035e2aaa5a"}]},"branch":"refs/heads/master","description":"Rebase"},"cd04b110e3e38ddfab84762f17dc05132cc9187b":{"kind":"REWORK","_number":11,"created":"2022-03-02 01:27:30.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/11"}}},"commit":{"parents":[{"commit":"1df18a885e8af6a55e98d9f56d1c042dcda12db3","subject":"Merge \"Updating python testing as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1df18a885e8af6a55e98d9f56d1c042dcda12db3"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-03-02 01:26:40.000000000","tz":480},"subject":"[ovn]support read chassis update time from nb_cfg_timestamp","message":"[ovn]support read chassis update time from nb_cfg_timestamp\n\nnb_cfg_timestamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis.\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd04b110e3e38ddfab84762f17dc05132cc9187b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd04b110e3e38ddfab84762f17dc05132cc9187b"}]},"branch":"refs/heads/master"},"6b5b0166ba1022bb6be39a164fbc42aaf1534ce4":{"kind":"REWORK","_number":12,"created":"2022-03-04 01:54:54.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/12"}}},"commit":{"parents":[{"commit":"b072cbf05f079519e379b2bfebe27846ae612275","subject":"Merge \"[ OVN ][Migration] Reload systemctl daemon after removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b072cbf05f079519e379b2bfebe27846ae612275"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-03-04 01:54:48.000000000","tz":480},"subject":"[ovn]support read chassis update time from nb_cfg_timestamp","message":"[ovn]support read chassis update time from nb_cfg_timestamp\n\nnb_cfg_timestamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis.\n\nThis patch updated some unit tests. ensure mock \u0027time.time\u0027 is\nstopped after test. if not stop, may affect \"timeutils.utcnow_ts\"\nto obtain the real time, cause test case\n\u0027test_agent_with_nb_cfg_timestamp_not_timeout\u0027 failure.\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b5b0166ba1022bb6be39a164fbc42aaf1534ce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b5b0166ba1022bb6be39a164fbc42aaf1534ce4"}]},"branch":"refs/heads/master"},"53811ed163769ca57d2357bafba9b8dc7c481a70":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-04-19 10:49:35.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/13"}}},"commit":{"parents":[{"commit":"b5e10bf7277aa2cb74eb356b9dbf7f172218ee79","subject":"Merge \"Don\u0027t register config options on imports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5e10bf7277aa2cb74eb356b9dbf7f172218ee79"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2022-04-19 10:49:35.000000000","tz":0},"subject":"[ovn]support read chassis update time from nb_cfg_timestamp","message":"[ovn]support read chassis update time from nb_cfg_timestamp\n\nnb_cfg_timestamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis.\n\nThis patch updated some unit tests. ensure mock \u0027time.time\u0027 is\nstopped after test. if not stop, may affect \"timeutils.utcnow_ts\"\nto obtain the real time, cause test case\n\u0027test_agent_with_nb_cfg_timestamp_not_timeout\u0027 failure.\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53811ed163769ca57d2357bafba9b8dc7c481a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53811ed163769ca57d2357bafba9b8dc7c481a70"}]},"branch":"refs/heads/master","description":"Rebase"},"b00082eb827fb90e1e991ba043d2f4c8d697708e":{"kind":"REWORK","_number":14,"created":"2022-04-25 02:42:06.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/14"}}},"commit":{"parents":[{"commit":"b5e10bf7277aa2cb74eb356b9dbf7f172218ee79","subject":"Merge \"Don\u0027t register config options on imports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5e10bf7277aa2cb74eb356b9dbf7f172218ee79"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-04-25 02:40:37.000000000","tz":480},"subject":"[ovn]support read chassis update time from nb_cfg_timestamp","message":"[ovn]support read chassis update time from nb_cfg_timestamp\n\nnb_cfg_timestamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis.\n\nThis patch updated some unit tests. ensure mock \u0027time.time\u0027 is\nstopped after test. if not stop, may affect \"timeutils.utcnow_ts\"\nto obtain the real time, cause test case\n\u0027test_agent_with_nb_cfg_timestamp_not_timeout\u0027 failure.\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b00082eb827fb90e1e991ba043d2f4c8d697708e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b00082eb827fb90e1e991ba043d2f4c8d697708e"}]},"branch":"refs/heads/master"},"9937024b080c6d3b59f115ee15cdc6db93ed2344":{"kind":"REWORK","_number":15,"created":"2022-04-25 05:41:44.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/15"}}},"commit":{"parents":[{"commit":"b5e10bf7277aa2cb74eb356b9dbf7f172218ee79","subject":"Merge \"Don\u0027t register config options on imports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5e10bf7277aa2cb74eb356b9dbf7f172218ee79"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-04-25 05:35:13.000000000","tz":480},"subject":"[ovn]support read chassis update time from nb_cfg_timestamp","message":"[ovn]support read chassis update time from nb_cfg_timestamp\n\nnb_cfg_timestamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis.\n\nThis patch updated some unit tests. ensure mock \u0027time.time\u0027 is\nstopped after test. if not stop, may affect \"timeutils.utcnow_ts\"\nto obtain the real time, cause test case\n\u0027test_agent_with_nb_cfg_timestamp_not_timeout\u0027 failure.\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9937024b080c6d3b59f115ee15cdc6db93ed2344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9937024b080c6d3b59f115ee15cdc6db93ed2344"}]},"branch":"refs/heads/master"},"91183df201b0b8945177d50afc0e804c64117089":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2022-05-05 03:19:15.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/16"}}},"commit":{"parents":[{"commit":"aff0a75ecb76e12f240b4ab00a8ecfc75776216b","subject":"Merge \"Handle properly ObjectNotFound while deleting network from DHCP agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aff0a75ecb76e12f240b4ab00a8ecfc75776216b"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2022-05-05 03:19:15.000000000","tz":0},"subject":"[ovn]support read chassis update time from nb_cfg_timestamp","message":"[ovn]support read chassis update time from nb_cfg_timestamp\n\nnb_cfg_timestamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis.\n\nThis patch updated some unit tests. ensure mock \u0027time.time\u0027 is\nstopped after test. if not stop, may affect \"timeutils.utcnow_ts\"\nto obtain the real time, cause test case\n\u0027test_agent_with_nb_cfg_timestamp_not_timeout\u0027 failure.\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91183df201b0b8945177d50afc0e804c64117089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91183df201b0b8945177d50afc0e804c64117089"}]},"branch":"refs/heads/master","description":"Rebase"},"9e263dcf00b7ddc102a54f5e681bcf09931f3a72":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-05-06 02:37:13.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/34/802834/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/802834/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/802834/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/802834/17"}}},"commit":{"parents":[{"commit":"a4a2d91b0e08dfd91693e083a3fc1e89d431a9de","subject":"Merge \"[OVN] Add baremetal support with Neutron DHCP agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4a2d91b0e08dfd91693e083a3fc1e89d431a9de"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2021-07-29 07:49:39.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2022-05-06 02:37:13.000000000","tz":0},"subject":"[ovn]support read chassis update time from nb_cfg_timestamp","message":"[ovn]support read chassis update time from nb_cfg_timestamp\n\nnb_cfg_timestamp: The timestamp when ovn-controller finishes\nprocessing the change corresponding to nb_cfg[1]. it can better\nreflect the status of chassis.\n\nThis patch updated some unit tests. ensure mock \u0027time.time\u0027 is\nstopped after test. if not stop, may affect \"timeutils.utcnow_ts\"\nto obtain the real time, cause test case\n\u0027test_agent_with_nb_cfg_timestamp_not_timeout\u0027 failure.\n\nPartial-bug: #1938478\n[1] https://www.ovn.org/support/dist-docs/ovn-sb.5.html\n\nChange-Id: Ia74a9404411862dc88b48c4a198d5c53f5f52704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e263dcf00b7ddc102a54f5e681bcf09931f3a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e263dcf00b7ddc102a54f5e681bcf09931f3a72"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
