)]}'
{"id":"openstack%2Fneutron~802860","triplet_id":"openstack%2Fneutron~master~I1dc56092d099223accc3aefa8e303310c4f6787e","project":"openstack/neutron","branch":"master","topic":"bug/1938428","hashtags":[],"change_id":"I1dc56092d099223accc3aefa8e303310c4f6787e","subject":"Wait until workers have been launched","status":"MERGED","created":"2021-07-29 10:34:41.000000000","updated":"2021-08-10 09:38:18.000000000","submitted":"2021-08-10 09:36:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"802860-bug/1938428","meta_rev_id":"029e3229f80c1a7dd8b30f0e208434bad02ac483","_number":802860,"virtual_id_number":802860,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-08-04 11:30:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 09:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-04 14:38:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-08-03 08:39:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-08-10 07:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-04 14:57:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-08-10 07:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 11:26:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-29 12:08:06.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-29 12:22:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 13:47:01.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-08-03 08:39:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-04 08:56:38.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-04 14:57:37.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-08-10 07:18:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"14bd87aabb056c4b35b776e942233d159d491a93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-29 10:34:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5624166f2bfd65d0175deabd2d5396d242977f04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:26:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f10cd32e8c6741f1ad1afaa534c9e173 : SUCCESS in 45m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ce81e457bf84ef5b80a2b8b5979ea48 : SUCCESS in 36m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af5e8f218c009f538d4f8ffeff7699fbd2263c3c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-29 12:08:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802860/1 : SUCCESS in 1h 32m 39s","accounts_in_message":[],"_revision_number":1},{"id":"65b30a7204f9a5929e0dbd5345546cd0aa8f934f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:22:21.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6685f8ea29bf4a308e2c0602e660d2a5 : SUCCESS in 43m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/56d3e3605dc843dbb90f847076138a3c : SUCCESS in 1h 44m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/832735144ddb4c9a8e9508440563f6cb : SUCCESS in 1h 42m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/49d518f9449a4b74b47a944588267b4c : SUCCESS in 1h 20m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a2ef0e6816764f4195b62abf2ed3162e : SUCCESS in 1h 46m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/878a9a75c355400b9a0274447df11366 : SUCCESS in 30m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c87c559bae6420e90525422b2488057 : SUCCESS in 30m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0623724742314822bda470d6e70afe77 : SUCCESS in 35m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cff69c348214b5ea1a647695ae0c3a9 : FAILURE in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72ad25c698cf4c739902480afdf28ab9 : SUCCESS in 23m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/108a47dae2c343ee8ffde02b0eecff43 : SUCCESS in 23m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21da6739bac4476b8166843d5aca8bd4 : SUCCESS in 24m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a78103801e4d3484444b7c5a35a0fe : SUCCESS in 13m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7bad0673ab634213bc3b0e5b8cb83719 : SUCCESS in 50m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d145b92ae33f4864ad813424ceab6b5e : FAILURE in 49m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8804a29a29e94a6ab8156f1ff3d63d29 : SUCCESS in 1h 06m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/762f955bfcf6459c8ed07926c12f94d1 : SUCCESS in 1h 02m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8da1106b5bfc43a59c15268a44e5b165 : FAILURE in 1h 11m 47s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f6408d84268b4f8e82e90a5a741a6a93 : SUCCESS in 1h 10m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e542e7a59c9649bfb5989f5f040dac07 : FAILURE in 1h 21m 14s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/0328715327ba468caf1dd99bfd944480 : SUCCESS in 1h 27m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa5675ea6b51498f9a9d9164c813fe83 : SUCCESS in 1h 01m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d254fd59cb514fa099f9a1767ff968e4 : SUCCESS in 55m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6788d9197abe49d3b465171c5bf65d1f : SUCCESS in 32m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/05897a7f910e4262be7899fccae29ef8 : SUCCESS in 1h 11m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/91fe358183584ac2a6cc778c3b54f5b3 : SUCCESS in 1h 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c950ebdb469210708fdd026b55e81edb6fe572bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-29 14:10:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad1040c3df76452027b60340ad6953f7062af85e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 14:58:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbb08fc8183f4827b49cfe26aa729f57 : SUCCESS in 41m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f63f39c3c03d4c4bbd1c233275b68912 : SUCCESS in 39m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"433948ef384a1e0df70a268e22b1eeea75d2b7c8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-29 15:48:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802860/2 : SUCCESS in 1h 36m 11s","accounts_in_message":[],"_revision_number":2},{"id":"97ea0ed54d98bfb8ec32238e9f4bfa019467020b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-29 15:50:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a758ccdf155d54a5ca710f74c2bfe11a5cab5a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 16:37:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/274bd67b545f474da87ec49921131566 : SUCCESS in 44m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1835298a11af4e3fa10a8b5e59f14d68 : SUCCESS in 43m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c8fe91690a5befdfc74c387308de51a778518dc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-29 17:20:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802860/3 : SUCCESS in 1h 29m 33s","accounts_in_message":[],"_revision_number":3},{"id":"44abd393144e06dc503f898ffe4954618a7b4e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:50:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/359557790e284bd7a6138d36a3da574b : SUCCESS in 44m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/47f53ee8b7994546a502c5782ba17626 : SUCCESS in 1h 39m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/460f3c66e45049abb3c98e9ce483f781 : SUCCESS in 1h 33m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ff739f76a2d04e0c85f57eda58ec8733 : SUCCESS in 1h 13m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ac6f89bd9357439f8c697251fcdb165b : SUCCESS in 1h 56m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/98242d56a5d545a3b33fa2b704aa37a5 : SUCCESS in 29m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3322a81e1e0472aa27eed036fd324e7 : SUCCESS in 36m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efeb1ae086574399a824dade8e9e566c : SUCCESS in 38m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc987cb63a44f89b093fee7c48bce35 : SUCCESS in 14m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f37204631e1b49ce8891a4ff116b9531 : SUCCESS in 27m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94440c6baf374d9abac5d40cc78f55ee : SUCCESS in 26m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffbdd184517c47ca8b3c05913a146cc8 : SUCCESS in 26m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eec30e35c804a0fb406223cadca46e0 : SUCCESS in 11m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3c5d6bbb86a49d68f4c1c45ea51aaff : SUCCESS in 46m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/75fccdcc64c94aae97e4b7ea4d520958 : FAILURE in 1h 19m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/acd027aa7df94375b37fa0717cf12e3e : SUCCESS in 1h 06m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c901d8ee15a45f79517d947556e3e08 : SUCCESS in 57m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d29507c12100484cae393de565e33eae : SUCCESS in 1h 02m 42s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/52985ad8ec704f83b11126d844c50e5e : SUCCESS in 1h 05m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/309a095c40d94ea793a9473cf1bd1804 : FAILURE in 1h 13m 17s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c32943d2b08a44969210a42418069dd3 : SUCCESS in 1h 27m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85c88cd068ae42adadd47b59942deaf2 : SUCCESS in 57m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5398d46b486147f881fd33ee3ec9cbeb : SUCCESS in 1h 00m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/977f1ea67d5f4e449c77f4aed1b77029 : SUCCESS in 36m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/adbff11df8574b8696185d77c5f52254 : SUCCESS in 1h 15m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/311b94ea5f644646849134839f8c0cc0 : FAILURE in 1h 18m 09s","accounts_in_message":[],"_revision_number":3},{"id":"d9cc80218796661c0c14450c2ea4341007008fcb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-03 08:39:57.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a7841f5827fbe5c6cb8b4ad306251fab5fae585c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 09:24:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e4d825778ee413ba4d84c443ef25421 : SUCCESS in 42m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/edd031aab433440a85363cc786c055ae : SUCCESS in 42m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f86bc9a6d19cb359c9e1f65ca17b11fb22d0749d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-03 10:42:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/802860/3 : SUCCESS in 2h 01m 37s","accounts_in_message":[],"_revision_number":3},{"id":"38c4642bf0248f1acb9958b2b4c10b09792ba58c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 10:57:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0b99c3f52fe54b7880cb56c10c62e64f : SUCCESS in 46m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e7810efdbf8a4cf8879ccb6171c1f092 : SUCCESS in 2h 12m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/942b3ce86e214f3e9bd8fa96fbfe588b : SUCCESS in 2h 08m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5009b7326361404c97ec930583fa4b2a : SUCCESS in 1h 16m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/724db9f2f5ed40039e804638c3fa5c3f : SUCCESS in 1h 56m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1386848015ee43e3b5ad4b113b149013 : SUCCESS in 32m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0c4e8f86aab4deaaef9e34da6bedc4b : SUCCESS in 39m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b87e753b568c45cd88a82edb370219e0 : SUCCESS in 40m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18314093bf464aaebb9462a9f1e2dbe3 : SUCCESS in 16m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff14a8bdb45846b6a8999318c37cc959 : SUCCESS in 25m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dae7e9dbcc56406baa73670a1127552a : SUCCESS in 25m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55ca012380814cd3b0e75b21302f6b7b : SUCCESS in 25m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd4942d6d9049f29de5edcd9f7af502 : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c251e7af00c41d3b8b7fb6fff0990ec : SUCCESS in 50m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/75a88801847c474590322ba7be950f81 : SUCCESS in 1h 27m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/eb06ad088e2f43f99fca5edc6c13ef79 : SUCCESS in 1h 10m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/97503237a2ab4447bfae5c375a0481d2 : SUCCESS in 1h 03m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1c30ce038c3040908decc169c22e9982 : SUCCESS in 1h 02m 53s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/45398ec90ec64068be294ac5b0c3bb9c : SUCCESS in 1h 08m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0da4598862454cc2a24891d3da69f3d3 : FAILURE in 1h 12m 27s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/3b9ae731012e4b90afa9f2512518f1c1 : SUCCESS in 1h 26m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25a327032b384057b4db304b5334bf12 : SUCCESS in 54m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b329022c93074f49b0d3b47eb5b33cff : SUCCESS in 58m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/34d75701c1184ea6a8618c8a001641da : SUCCESS in 33m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf1a0beab60f4ba19c865e3a624cbf3c : SUCCESS in 1h 07m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5a0af5859f8f449f9288af54de1f0df7 : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":3},{"id":"dbb7c361678d075e28dad18509a3baca594b782a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-04 08:56:38.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/60/802860/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"aac36c77d2b0f7bbaafbf3773ccf68793db78a51","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-04 10:30:13.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/60/802860/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a330c15ca57cc9068042b9f5bd42bb97696a6ab3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-04 11:30:38.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/60/802860/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f7b02fb931d91e01c550465bbdb95360e720a051","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-08-04 14:38:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f4d432d796b61366ecf26f35af0692a8e2bd2caa","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-04 14:57:37.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)\n\nSome nits","accounts_in_message":[],"_revision_number":3},{"id":"093be707dda76c9ada3c99f3a127e89eba7cbd72","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-10 07:18:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nI think we can merge this now","accounts_in_message":[],"_revision_number":3},{"id":"262912c72c1ef6973c6f3e6d970390957f4916e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 07:19:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d7a54bf23d954570fbb5e986eb8046a0dbb5783","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-10 08:04:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb6daa6d55b095ee41cfaca5f8865e0d8dc50f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:36:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1d54d2df286a49c28fa6f44fc25b8a88 : SUCCESS in 40m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a9450d54bc304de29559050d5ee6e081 : SUCCESS in 1h 54m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/39e22475e50047fba28a4159a57d657c : SUCCESS in 2h 02m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5273857a5bee4704af299a72416d65e6 : SUCCESS in 2h 17m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6eaa41570a4642a59f5cd521611ddc85 : SUCCESS in 1h 18m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3731ecf37eef47e4994a94b46812ce87 : SUCCESS in 40m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99de5eb4b8ed44ed88a466e4243e9f2a : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3b98d89a67f4e05bf0e1823c0009f2c : SUCCESS in 26m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6714ee53f0f94120abd27b7e4fe9c8ce : SUCCESS in 24m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55cd6032b3c40f494ccf3f7dca81240 : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3afaf5b2f8c4586b5ca31fa92e8ecb6 : SUCCESS in 49m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2a3145b70324bafaa7c265d7a4d88fc : SUCCESS in 1h 42m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3fa6ab39c7b24c9093a3e48dc2466273 : SUCCESS in 1h 14m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/823d98637e564c36aa60bd9e8b0f0bbf : SUCCESS in 1h 03m 59s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/4c64615d7b1a43f6a3ca4d6005971320 : SUCCESS in 1h 24m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f1e821d9d1b4d928d367490b2a709ff : SUCCESS in 52m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ddf78b967dfb428f88f4034153f37da0 : SUCCESS in 52m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/027aa06a8bbc4c36976a7f2e08442e2b : SUCCESS in 31m 54s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/53e70b54ee5a47e990173c904bb9c5ef : SUCCESS in 1h 10m 07s","accounts_in_message":[],"_revision_number":3},{"id":"9c9702510286c39ebbc7bea09629589a6a4a5956","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:36:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"029e3229f80c1a7dd8b30f0e208434bad02ac483","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:38:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64de168dfd294f63bc05f0ea792be0f1 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fafcabdbe0dccf2747884a35f3a094486c806fa6","revisions":{"66fe0aa34a1b20f2d00cafa3dc1e2f8c9632f7a3":{"kind":"REWORK","_number":1,"created":"2021-07-29 10:34:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/802860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/802860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/802860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/802860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/802860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/802860/1"}}},"commit":{"parents":[{"commit":"0411743bb974e5a352dede939d3a824d09234ead","subject":"Merge \"Sanitize MAC addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0411743bb974e5a352dede939d3a824d09234ead"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-29 10:30:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-29 10:30:01.000000000","tz":0},"subject":"Wait until workers have been launched","message":"Wait until workers have been launched\n\nIn \"test_restart_rpc_on_sighup_multiple_workers\", the test needs to\nwait until the RPC workers have been properly launched by\n``oslo_service.service.ProcessLauncher.launch_service``. Once this\nmethod returns, it is guaranteed that the child worker processes\nare running and the signal process handlers are attending the\nSIGHUP signal that will reset them.\n\nCloses-Bug: #1938428\nChange-Id: I1dc56092d099223accc3aefa8e303310c4f6787e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66fe0aa34a1b20f2d00cafa3dc1e2f8c9632f7a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66fe0aa34a1b20f2d00cafa3dc1e2f8c9632f7a3"}]},"branch":"refs/heads/master"},"3e09a1b42698db57e8bbadd1725c404565e2c92b":{"kind":"REWORK","_number":2,"created":"2021-07-29 14:10:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/802860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/802860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/802860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/802860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/802860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/802860/2"}}},"commit":{"parents":[{"commit":"0411743bb974e5a352dede939d3a824d09234ead","subject":"Merge \"Sanitize MAC addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0411743bb974e5a352dede939d3a824d09234ead"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-29 10:30:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-29 14:10:36.000000000","tz":0},"subject":"Wait until workers have been launched","message":"Wait until workers have been launched\n\nIn \"test_restart_rpc_on_sighup_multiple_workers\", the test needs to\nwait until the RPC workers have been properly launched by\n``oslo_service.service.ProcessLauncher.launch_service``. Once this\nmethod returns, it is guaranteed that the child worker processes\nare running and the signal process handlers are attending the\nSIGHUP signal that will reset them.\n\nCloses-Bug: #1938428\nChange-Id: I1dc56092d099223accc3aefa8e303310c4f6787e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e09a1b42698db57e8bbadd1725c404565e2c92b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e09a1b42698db57e8bbadd1725c404565e2c92b"}]},"branch":"refs/heads/master"},"fafcabdbe0dccf2747884a35f3a094486c806fa6":{"kind":"REWORK","_number":3,"created":"2021-07-29 15:50:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/802860/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/802860/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/802860/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/802860/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/802860/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/802860/3"}}},"commit":{"parents":[{"commit":"0411743bb974e5a352dede939d3a824d09234ead","subject":"Merge \"Sanitize MAC addresses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0411743bb974e5a352dede939d3a824d09234ead"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-29 10:30:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-29 15:49:50.000000000","tz":0},"subject":"Wait until workers have been launched","message":"Wait until workers have been launched\n\nIn \"test_restart_rpc_on_sighup_multiple_workers\", the test needs to\nwait until the RPC workers have been properly launched by\n``oslo_service.service.ProcessLauncher.launch_service``. Once this\nmethod returns, it is guaranteed that the child worker processes\nare running and the signal process handlers are attending the\nSIGHUP signal that will reset them.\n\nCloses-Bug: #1938428\nChange-Id: I1dc56092d099223accc3aefa8e303310c4f6787e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fafcabdbe0dccf2747884a35f3a094486c806fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fafcabdbe0dccf2747884a35f3a094486c806fa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
