)]}'
{"id":"openstack%2Fneutron~803045","triplet_id":"openstack%2Fneutron~master~Ie994bdab62bab33737f25287e568519c782dea9a","project":"openstack/neutron","branch":"master","topic":"bp/qos-minimum-guaranteed-packet-rate","hashtags":[],"change_id":"Ie994bdab62bab33737f25287e568519c782dea9a","subject":"Add API extension for QoS minimum pps rule","status":"MERGED","created":"2021-07-30 14:38:02.000000000","updated":"2021-09-30 11:19:14.000000000","submitted":"2021-09-30 11:17:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"803045-bp/qos-minimum-guaranteed-packet-rate","meta_rev_id":"5f8f24477f717a7d7420201b0f3db397106f589f","_number":803045,"virtual_id_number":803045,"owner":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 11:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-09-29 12:57:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-30 08:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-30 08:59:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-30 14:40:13.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2021-07-30 14:40:13.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-07-30 14:40:13.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-07-30 15:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-30 16:26:21.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-07-30 17:16:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-03 09:57:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-08-03 13:17:00.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-03 13:17:00.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-03 13:17:00.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"5feb2e87545297c59a60bb8e3f9a4593c4de7234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-07-30 14:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee3c767f30ea262e0648013a348c4a889432b098","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-07-30 14:38:23.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"514b71ed9db1894bb27cc5ce48094be2851870bb","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-07-30 14:40:13.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"e04639ca0c3c609323ce887a8ab384c5cb2578f1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-07-30 14:40:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"b263c06f23c7221d71ba4e3a201a958fc3ed52e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 15:17:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05016987f623446f99e3588f80b7c933 : SUCCESS in 34m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0cb77c2ac994a82824f1f6eb92b6f42 : SUCCESS in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef660a7379d4774e467f4fa1cb31a43f1fd9ce5a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-30 16:26:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/1 : SUCCESS in 1h 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"027963196117b0d6941930a9a01a17732a67fdaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 17:16:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d3565da7c0984476af7a7b85e86c032f : SUCCESS in 6m 13s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/80c53a19ddcb4876892abba10750fd44 : SUCCESS in 45m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/292d02bbb9984014b7dd4df74a81b466 : SUCCESS in 2h 31m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/80e6ea7ca3cc4f5c9284037b52c35282 : SUCCESS in 1h 31m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1592d20ae70648a79efbbe7646100b5b : SUCCESS in 1h 16m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2e396e0b475742d3a60df0f7f67d8e2b : SUCCESS in 2h 10m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d35ebb8c4524646999bb2264bb0d334 : SUCCESS in 29m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b9ba3cffc4e4f21b31892c94cbc3be3 : SUCCESS in 39m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34a591b2be9341f68f896dd79e824aee : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be0614b408a443249d737af955db6300 : SUCCESS in 16m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad4bf726dfd641ba93cada532759fe2e : SUCCESS in 26m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5df01b8dc6de4f039f8c44063043a965 : SUCCESS in 25m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37b3fc527c2e418eac97c0f2a7623f23 : SUCCESS in 41m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/222b12fcb88b49e5aaa371b91e643b3b : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c9c5677f80c44bb9d991883f1a0293e : FAILURE in 41m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc63e856d7d0400787b37bedb6594bcc : FAILURE in 50m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/085a1f57cc64412194bbda2e43810491 : SUCCESS in 1h 15m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b06909aa9f8408bbd02da23ab037c1d : SUCCESS in 1h 01m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/074a47f663de4a6583c2582052b43dd0 : FAILURE in 1h 06m 40s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3df2f0a29af04f149aa70c1c9dc6bd43 : SUCCESS in 1h 07m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3f8c97fbdd93430cacf099fff930247d : FAILURE in 1h 10m 26s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/62b9d3c54d8c46c8b87e10d3242a39ff : SUCCESS in 1h 23m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18dcd50f279b45b09df16a1ddd3c7e3b : SUCCESS in 54m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/45b42d380c0a44ca8ecbb658ab309518 : SUCCESS in 53m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4fb71103e6f543b296265c75f01e0703 : SUCCESS in 27m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd735605a4d046c59533d91c78c1662e : SUCCESS in 1h 09m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/432a6694839b4e94a14f94300496b6f0 : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a0d329e28d927d8416e776a18618158c0cf9fe7d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-02 09:25:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"056a6a74be2a6528fdaabe248249319bc6983df9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 10:12:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b7cad9961a045c1bc49915d574d23ed : SUCCESS in 38m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1577253b8a54f21a6c4f6c9a737eb0b : SUCCESS in 39m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5cea4f376f9597364a1f87ee3d75d959bc4fa04d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-02 10:55:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/2 : SUCCESS in 1h 28m 10s","accounts_in_message":[],"_revision_number":2},{"id":"458d299bd2636725fa861a8de405b0e3e463d10e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:32:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4bffd6ff9c234b8fae60b1d3246d3bb6 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0b354f848640442bad2db528450849a9 : SUCCESS in 38m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8f1edda14d6f440295e6ffc37c963ca5 : SUCCESS in 1h 40m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ed906ee9ac2e4b448b203f1df36732e6 : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f832781ccf0e49d69f010620471b9439 : SUCCESS in 2h 00m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0a3387e90ab84d7ab31a3b52585617fe : SUCCESS in 1h 53m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/868ec3c29bcf4708ad97cf280f980b22 : SUCCESS in 39m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc06aba98c404d2cadbf94f3f5659703 : SUCCESS in 38m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7704042e91942d495b3ec0b2a474691 : SUCCESS in 40m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8015b47839a940acb3c6b314614d92bd : SUCCESS in 14m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6745ae6de6249ff97d10546313848d6 : SUCCESS in 27m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa9912e7fe5940b190e2d47650d35731 : SUCCESS in 24m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d595a488da20457b8c36fedc0788fe14 : SUCCESS in 29m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d40297a6d24bbaadff640799bb3bfc : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77211d568107404398670ad222ab1877 : SUCCESS in 54m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6779c5cf6088436d977890441cb081c0 : SUCCESS in 1h 26m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/41851ce9387549cd8dd6d92bba507bde : SUCCESS in 1h 08m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b65e9241c87145bb9b9d495ee483621a : SUCCESS in 59m 48s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e7b1e2710c13496ba49e7b33c781bbd0 : SUCCESS in 1h 02m 02s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cde6d48f9c6142b6b623c97ac4c9fdce : SUCCESS in 1h 06m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e904d8863bb64fe09e98e80035d8726a : FAILURE in 1h 06m 40s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/b26888d6290947c49b0b5f5455fb59b7 : SUCCESS in 1h 23m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b56896cfa7247359c8875cd1ea625a3 : SUCCESS in 57m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6d1c36e07cbc44fd9944871c98592f7b : SUCCESS in 56m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3263eb88414942999f35c14f528d815e : SUCCESS in 31m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f096772e120f45ef8edc3ced987d77ef : SUCCESS in 1h 10m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ee6fe7b0b6484d39b6bd0b684726c4fa : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":2},{"id":"ce1ca57469b146ab2e694b421507eccc15f240c1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-02 12:32:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5d0c2fd6c6cb3bdac8e7ff25bf001404895489b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-03 09:57:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"277c9981b62280fd3260960cebc3d47fd1664f56","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-03 10:07:37.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"5ff93fb664d7139d4f052e9dbe8f3a0317e68133","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-03 10:07:45.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"487e257751b7686326865ca4aa5ef0bc6cc37ff6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-03 12:15:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bc724fd975df1328acc4e267b33e54a71b95729","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-03 12:49:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"647961bbea184c4604767f1f5878e45731917478","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-03 13:17:00.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"e4e5b00e525c63123fb48c7c698c3e45f9fd9972","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 13:33:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce7b0a847f874f93b9c0e82b86650a4a : SUCCESS in 41m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/baebf99f08e4430ab5d50f503a963a4e : SUCCESS in 41m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2041aaeb4a9015ebb8b8e9040421a2cdd86f276c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 15:22:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8ec9a18d471b416eb504cbdfafdf4553 : SUCCESS in 3m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/95b8709d2ed745d8845fb78c3537ee49 : SUCCESS in 55m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/87f2c6812af040329995c5176e5c74d8 : SUCCESS in 1h 42m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bd95fe7635f74d12942122c566735a8a : SUCCESS in 2h 30m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/58ff4ae33c98489da11d229770cea4e5 : SUCCESS in 1h 53m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e50e60b1d0864101a4d2b4a5c5bbbf35 : SUCCESS in 1h 43m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac68a14a960d4026ad48d5167f9ccbae : SUCCESS in 24m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/822aebfc9d3d40409b7dc17277196dbb : SUCCESS in 33m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/745e4dafd1dc4fe9b8065706cf701c77 : SUCCESS in 31m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c21e1e3809040d685724daf66d511b7 : SUCCESS in 15m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e5cd90f856d4ab78dcdcc2b6b41d2d0 : SUCCESS in 24m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6731b519709f4b0caf998f22445b60f0 : SUCCESS in 21m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc03a658c77f47c4a7af7d04e622c1e5 : SUCCESS in 27m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24693e42264e4ff39ccb5e7c55049127 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4c5ea74581340a8b83de8af41973bf7 : SUCCESS in 48m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b1e32e8edac4d38ac9c83907efbdc9e : SUCCESS in 1h 02m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/eb1ae3459ccc42b68b59cef40c5615ea : SUCCESS in 1h 08m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/469a3bf818b54b2da4291a5edcba046f : SUCCESS in 56m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4e500699d90c4c46aa476c47c777b090 : FAILURE in 1h 03m 33s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/05ac341e339f4d578276b79234a270fe : SUCCESS in 1h 05m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5fba3077c16e408799372ff40de5efbd : FAILURE in 1h 06m 28s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/4f12d9080e82479da1918bfd65fa727f : SUCCESS in 1h 20m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caa85ab6344f4ba28c0e4ba94f62529a : SUCCESS in 53m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a95cbe4da58a4dbf8bdfc99e1e5fba02 : SUCCESS in 55m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b71fe0ae89e046d69526e6311f03208f : SUCCESS in 28m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9864658605a4c76a6cde7df97dd2c3f : SUCCESS in 1h 13m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67fd04d5d4714b7c9a25cb1d30a1d882 : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e84919d8fec77fc6dd568daa8fddd86c4aa8761e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-03 16:46:52.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/3 : SUCCESS in 2h 00m 13s","accounts_in_message":[],"_revision_number":3},{"id":"323e0731c9ed61025dde495516078f0631571a38","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-04 20:56:59.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b41ddb9f72502fe7c6ebd31c8c2aa4953a1d2500","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-05 03:54:56.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"500a46ae1a7d596b03af4456f61251d0040198ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-05 07:42:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"436db582dcd8281caf6e75385d0c09102fee7b89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 08:45:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aef69dc507ab4f7299fb09da34ea095d : SUCCESS in 1h 01m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f17c71d56b1643a78d6566ae745bafcf : SUCCESS in 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ef344f0fcd1a64a503cd8b08eb2d4da5b1e68728","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-05 09:46:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/4 : SUCCESS in 1h 56m 37s","accounts_in_message":[],"_revision_number":4},{"id":"2ffeb5d9b3e42b2d1127f55f8f16c5ea3198d148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 09:49:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ab911ef082c840e48cd457e6d5fa7a9e : SUCCESS in 3m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/68d2ff685f014f05988c59b3adf87afc : SUCCESS in 40m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a9e5de76416c4425b09c72ebd0e9ee09 : FAILURE in 2h 06m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4adabb66178f439c9490ccac514a8808 : SUCCESS in 1h 45m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0356a466a69c4f84938c5ceffeb14463 : SUCCESS in 1h 32m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f9206ab1c95d4b038a0fb082701bd943 : SUCCESS in 1h 55m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31d372fa6f814940b7b4e4b6b289c61c : SUCCESS in 31m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a958b15e0a1a48d9966613f99235442b : SUCCESS in 35m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b05030dbc3d64c10a8540b1e72166fbd : SUCCESS in 34m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb296caa889c4a14944193323f651681 : SUCCESS in 14m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8886f21c6d9474d87653c3de325c1e3 : SUCCESS in 26m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66234fe85eb946f684b7d630321eec42 : SUCCESS in 22m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9e35a3e6d5041e899e53dd5416417e2 : SUCCESS in 26m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0995f7327fe47399a1f2da5d77d8b3d : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dddd1dda7d7347cab1642c50218e2d62 : FAILURE in 46m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/154bf935149e40f48f70e0193fe4a6d1 : SUCCESS in 1h 29m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5c038f9b46934d8c889806d9b3f9f941 : SUCCESS in 1h 10m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d39adacf47f4987821adb5afea6ace3 : SUCCESS in 1h 02m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3b756f15dc4c4771ba51b6520b0507d7 : SUCCESS in 1h 04m 12s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e8d75690f51947ae8bfd245221fc9ba1 : SUCCESS in 1h 12m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b5ef5b43991c4d25b597162278737098 : SUCCESS in 1h 20m 52s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/179b5841de534039b2af14afd6db63e0 : SUCCESS in 1h 22m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d50b36a25f70473f9a1981d43a6491d4 : SUCCESS in 50m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e34c47f5bb004759b4299bd6996dfebe : SUCCESS in 58m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f5f2a1b74f0e479a801983073515052d : SUCCESS in 32m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/53e3beb57e184f349651becd19e63b85 : SUCCESS in 1h 03m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3e4b5a4f008f4377b21427878d64f211 : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":4},{"id":"05c49e3e9c9a5a068806cc0f947617bbfc860580","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-05 11:56:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7ea835c267402e92aa5ab7aa9983091a79526175","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 12:40:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f47b956a5d743d3bd8d2a446dbe3654 : SUCCESS in 42m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0a3448cbefb4cbfa061bcfd23ee0189 : SUCCESS in 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"df94c2e31ea42d13d058d03c8b88569b80ebc58b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-05 13:43:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/4 : SUCCESS in 1h 46m 39s","accounts_in_message":[],"_revision_number":4},{"id":"5a39b5c56256e63718b5dca642133227b74c93fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 14:26:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/43c5889c14904e148d2be3e0a643de63 : SUCCESS in 3m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6080808ec7734287abd776b32bc0ecf1 : SUCCESS in 50m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fd627509b4be42cd87c149c0cf7edd9a : SUCCESS in 2h 21m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eafa37a8772f4f7c9b1c887a6c7b0c8f : SUCCESS in 2h 07m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b80da16e934431f9904179bfcc9be74 : SUCCESS in 1h 57m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b1e52aa961da48249d695df76322eb1b : SUCCESS in 2h 15m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/502e8b06d272459f9fb10047438d7b54 : SUCCESS in 32m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94aa7fb0dddd4e9d9bb46fcdc2774611 : SUCCESS in 35m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/925616642af949d1acaa4f286dfbdb24 : SUCCESS in 40m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdae708948544ea9fa6fd12aa5fa35c : SUCCESS in 17m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe7f6a25e24f4865a8dbfc9ffd285cd0 : SUCCESS in 22m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a56d758a8ea4ab8993a4d907e24cc12 : SUCCESS in 24m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0db85dc897704e9f8e15f21c96d67672 : SUCCESS in 24m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6589e9ad4c774508ba92a1ee6748b49c : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/540124700d2a4d6c9dfa107fd0b9575f : SUCCESS in 47m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa3ba7b9ef5e4abe824fb31246d6ae15 : SUCCESS in 1h 00m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a6a7794f5fdc4ee7a135f51c33972b85 : SUCCESS in 1h 08m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/02e2141445de4d70a3991ede7135a59d : SUCCESS in 1h 11m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6ba0cb3c3b4b412f9a6cecd54506551c : SUCCESS in 1h 01m 07s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5529ff4228b04348a4de3a4f3ac94568 : SUCCESS in 1h 12m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e70ce6581c843489762aa4983b1b977 : SUCCESS in 1h 25m 25s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/a87fb5cbdf5344349c26e1bc168b03a9 : SUCCESS in 1h 26m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bf4c69ae3ac42999fd49c5619869da2 : SUCCESS in 57m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/10f36b4ab9d440d78b1335aa3351dee0 : SUCCESS in 1h 07m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/099b9b50c95747199e4e3cb39de47349 : SUCCESS in 31m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e41f5ba2f1c94211b9331e162eb3c09e : SUCCESS in 1h 10m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e4969eb4b47048f59e5298beaf6bc0ff : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":4},{"id":"a630c8337406bcd5c2d8b40f9ae24af480b499e5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-06 00:16:48.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"545b51d4bfb99047796f210e8827ad91f0bdf8e1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-06 17:16:21.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"086f27c8e00593b1890c84030cff2004dc058cbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-10 11:20:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"24e4467abbd8ebeef001c43ab2176becefa8dd87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 12:24:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b28607cf90ac42cf9f490dcafeae0dbd : TIMED_OUT in 1h 01m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffcea761d77c4d19b6d5d29aaf0834c6 : TIMED_OUT in 1h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"76bb66b02d308ebc5149e587a7960d71dd50b161","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-10 12:55:14.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/803045/5 : FAILURE in 1h 33m 55s","accounts_in_message":[],"_revision_number":5},{"id":"79601a0ca2cfa00e8e4893aa1432ea9d2203a0de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:17:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bfa62ce457ea4d76a53b21ef7ff01359 : SUCCESS in 2m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/754e11d02acb432daeaa152f3ed465e0 : SUCCESS in 43m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f6b01276065d4aacb8c26027f7dfb0ff : SUCCESS in 1h 55m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/97bdff7947ea4f71ad319f3e8cf60ca1 : SUCCESS in 1h 47m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/90f97674c28546e9ab123d35f54b63b5 : SUCCESS in 1h 12m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e76d84db1f1849278d94815467cbdf14 : SUCCESS in 1h 38m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/47a964a35dde4bccbcf8a29779a37a85 : SUCCESS in 30m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fc871f42cc8488797fe78b6d593bb09 : SUCCESS in 41m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/598b83c598cc43c0ab83572cda4451fb : SUCCESS in 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7a398c3ccb345f0a4b9bcebebc6abfd : SUCCESS in 13m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e973cde152b45859aa4788604ce6c8d : SUCCESS in 26m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1246f7130e8c44a481c1e3f02619a33a : SUCCESS in 22m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea7d6acd4e144bd0a02b1c5965b35a3b : SUCCESS in 23m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c65e19dff0478ba5d68bf45c793b1c : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/90c31280fa1f4dc3a681fa60f3ec1f41 : FAILURE in 50m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b6e4bb1017c4744a0eadfa734709cdc : SUCCESS in 1h 42m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d8839a4419294965b0556ad9f0b836aa : SUCCESS in 1h 08m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5cb2cf65324444099afde75644d9ec8 : SUCCESS in 59m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f3b57714b61a45a4bd3b720b6024afbc : SUCCESS in 1h 00m 51s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/65f825f0d20649f9ab803fef7459e6aa : SUCCESS in 55m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/abd212bea2ed46b48bdf8e7f507bd333 : SUCCESS in 1h 19m 30s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/7ab8140645124529b4377f24b2276e83 : SUCCESS in 1h 27m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76ca298bc165473c8d0e308bd4e53492 : SUCCESS in 51m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3742fd12df904471bee2994d75fa8787 : SUCCESS in 56m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/de3f391201194862aec1a1da790aa70f : SUCCESS in 33m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad12a0b260cd464e93568c98d5876759 : SUCCESS in 1h 05m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/de6161a7a7a443039138845d6601e57c : SUCCESS in 1h 12m 13s","accounts_in_message":[],"_revision_number":5},{"id":"8c40327d2662573ca7bc81181a6b6da298d60e8b","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-10 13:22:47.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e9dfbf6d4ca71ada9f9ce9041abb4c7401ed5f13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:12:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f52338df5b174fdc872a71a442307fd2 : SUCCESS in 39m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bcd1b52adfbd40f8b2dd2f84d9e41aaf : SUCCESS in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1ae5b07b74fefa14084dab1c0c7f567458efee10","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-10 14:55:26.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/5 : SUCCESS in 1h 31m 45s","accounts_in_message":[],"_revision_number":5},{"id":"19db30ae329e321d0981b9b4965a17f62ec097d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:35:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/77e0f509fa6c4df9abe1a8ea17cadec1 : SUCCESS in 3m 51s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8dbabf386e7a4c93b4b4b7350e28d514 : SUCCESS in 42m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1c5bb891b2564577bc77dddfce7d25c2 : SUCCESS in 2h 07m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/76710395bce94043aee7a6f799ee6654 : SUCCESS in 1h 46m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/59cbae62e2fa4f628b7231fc4d1cc8a5 : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cfe98dd3f1bb4c5bbe06294e0d59bf42 : SUCCESS in 1h 51m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/964fe6821ce74b96bc2dc1cac6e642c5 : SUCCESS in 32m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/931fc17c73d74d57b6dc2cab06d54adb : SUCCESS in 34m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bef3d855e2b479ea3c6c67461684043 : SUCCESS in 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83af8ef4f83c469b86fbc9ed32789d44 : SUCCESS in 14m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c6b1529fc0d4abfbbe4443fde224d8b : SUCCESS in 27m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01783478aec543fca4e87801efe38219 : SUCCESS in 26m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfcd66e1dc034e3c851fc7d61f7d13d1 : SUCCESS in 24m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8950694a5537458a8a5d0dc4f49a1ddd : SUCCESS in 10m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ae154a1faf34a4496f298be10c63bb7 : SUCCESS in 48m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf4c09b790d5463fa1ea763f1aa41a58 : SUCCESS in 1h 43m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7f61679402814d6281fab2450c5a02ce : SUCCESS in 1h 08m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b149d2c64c74f9f9a3814592c4ba5a1 : SUCCESS in 1h 12m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c601512ab1cb43d2af4f400205253be3 : SUCCESS in 1h 03m 18s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c675eddb90bb4eeebf9813f6d9087086 : SUCCESS in 1h 28m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/02302f341fd44f31bf98db765edb6817 : SUCCESS in 1h 16m 10s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c6f06abd3c06424988f85c86b19f6666 : SUCCESS in 1h 29m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c700c4171459411e894f4a3060f93a58 : SUCCESS in 1h 01m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a2c3e4b392544799916d9b23fe177393 : SUCCESS in 54m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c77484c8442946149e9007ba2b49b0a2 : SUCCESS in 33m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f8c860da31464457a8ec06bf0eb309f5 : SUCCESS in 1h 09m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4012da897c314de6bdc815f649319b74 : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":5},{"id":"2838824404ff95a2e16f711b7f4e9844f30f5ad4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-11 10:37:00.000000000","message":"Patch Set 5: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":5},{"id":"93c3cb038f527595c613239f84f6ae0ea16408c3","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-13 14:39:12.000000000","message":"Patch Set 5:\n\n(14 comments)","accounts_in_message":[],"_revision_number":5},{"id":"bc680e4ed5597f67dc0716f2cae72022085ed273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-13 14:39:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d4bfba069d8f6b21f3792a0b59012b968e42055c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 15:23:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b0a9ac196de43ada38cdaecc19214b0 : FAILURE in 42m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c8ccac054db40bdabd460ae37f0dccf : FAILURE in 33m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"708ba4901701c1f92a039406626e37e45bed3421","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-13 16:12:43.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/6 : SUCCESS in 1h 32m 06s","accounts_in_message":[],"_revision_number":6},{"id":"dd9210d4cf8af3fe2be47d3a2204b64a5d5d103c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 17:29:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0c0d74370f6a4075976fced28a2dc687 : SUCCESS in 3m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5f9933bb3ce344aaa8463470907f7b3b : SUCCESS in 45m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3de7247943ab4aa7a88c5080f0fd20a6 : FAILURE in 1h 52m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c650cb18a7e84d889befeff8e256dda2 : SUCCESS in 2h 06m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5d2821e8f5714d7f86584c43c1f049c9 : SUCCESS in 1h 34m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b84aea3699f44982a2f2238225d55978 : SUCCESS in 2h 45m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2ef077fa76a430780162074b8e6ec6e : SUCCESS in 31m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffd13d774b284cfe96c3b0c44dfc441a : FAILURE in 36m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ac948ade2d24fd39494d688fb04da05 : FAILURE in 48m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1f4254580b48fc8284bf448cd3d786 : SUCCESS in 14m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fede250526b415ab5698ef1c3f46e2f : FAILURE in 25m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2c775ef9532479b8aafbf6bab954342 : FAILURE in 25m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/381be52948264b8b8acd22faacbc1461 : FAILURE in 29m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a12cb095267e4710b1e6eb68b2f761b4 : SUCCESS in 14m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/453bb53d6fc64c50aa54301c10ce05d8 : FAILURE in 49m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc37345a71e044f685f3a7753c4af8ef : SUCCESS in 1h 42m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/378634dbb55e49abb3f3f979c7d763ab : SUCCESS in 1h 09m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb239091c0844357bae447d7ec3577b9 : SUCCESS in 1h 01m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0ab6bc390c73475cadd8a1d4fcc8c91f : SUCCESS in 57m 39s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4f8da49f1dc540399471757c9bc8fc6b : SUCCESS in 1h 16m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/887f91cd10194f619a4e9e1edd9dd878 : SUCCESS in 1h 26m 40s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/1f18789af12a4a5ea96a2fab22672092 : SUCCESS in 1h 25m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4dd0fd667f540129366a06fc0157776 : SUCCESS in 1h 03m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/961b4d0b54e24099843db262ba27b24e : SUCCESS in 57m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c1afc0c0494f4d91b63340c4f76ec245 : SUCCESS in 27m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b358da65ffb489aa1bf5a6dc30ca1db : SUCCESS in 1h 09m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7a8ca39ef7a549f5ac6f44b3b4798351 : SUCCESS in 1h 23m 57s","accounts_in_message":[],"_revision_number":6},{"id":"e36a9eef1188051dd9e6896d5c307b5a75500c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-16 07:05:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ca67c16893999c58fbb692eaec135386d72597cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 07:57:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/192f54d17f5c487fb1cab59a0c011c7f : SUCCESS in 43m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20c46ddcac044734b0f167696ecb05df : SUCCESS in 43m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5683a839c20b631f0fa38bad2ea3348bcb281cdd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-16 09:03:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/7 : SUCCESS in 1h 56m 38s","accounts_in_message":[],"_revision_number":7},{"id":"6c662359bdd85ef60d5d6cd7dca8391cad70e545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 09:32:24.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8fa5a15cbf41455d92925033a6f65703 : SUCCESS in 3m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/adf01ca3ae2847d482b9eb2f5dbbf8e0 : SUCCESS in 41m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b7487bc785bd4fcea7c178a68a983c42 : SUCCESS in 2h 18m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2bfa6e96c7f6482dacb433c44994a8e7 : SUCCESS in 2h 00m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a0e070541ca341ad848b7d17fdd45a0d : SUCCESS in 1h 37m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/31cbf3251a2747c0b438e5238168e89c : SUCCESS in 2h 03m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a74966e50d4499cb50558326cdd0c64 : SUCCESS in 31m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e5e9ef2b267492c97fc8b9343d9c0f8 : SUCCESS in 33m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b08f2dfe458437bb51825601c43607c : SUCCESS in 39m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95356c74f11045bb9b555ae4bd71ec94 : SUCCESS in 14m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da3e7b671185472a9f3de7c441311f2d : SUCCESS in 39m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b25d18d7b3a4d36bf2239008911e4f9 : SUCCESS in 23m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02bc42f98e9b4ba8a2f38c19fd4315d0 : SUCCESS in 36m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b45435078e493cada632073feb37f1 : SUCCESS in 12m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9df16292d7f84803937c2de983b03191 : FAILURE in 50m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e71396070c864747be5d9984c249fa32 : SUCCESS in 1h 42m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9776dc6b7f90451c86dd416dffc7d657 : SUCCESS in 1h 10m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/716e0ee4dea7456684e9c42f9bc3f20b : SUCCESS in 1h 01m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9271213273d40d5abf495858c6fa961 : SUCCESS in 1h 04m 41s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4d86a05ddb3948d197cc1fa67a25daef : SUCCESS in 1h 03m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/664e2c2331a946b2ae8f687263bf1b8a : SUCCESS in 1h 13m 34s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/fafa2858ab1d40a7a046facdebfe7d6c : SUCCESS in 1h 23m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e836008cbab34a9b9c74e60aa25e45f3 : SUCCESS in 1h 06m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9fe68e1bef98484f964c9ba11268ed4d : SUCCESS in 46m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/989e4cb198bf4a3d97810c84e49d7764 : SUCCESS in 29m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3ee70946bd948ccb3b2c3f4f7f9e468 : SUCCESS in 1h 06m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ff10f5a1b0464634941d84ff977b6be9 : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":7},{"id":"bb9c50bf6e1112e10ba21539d4e19e6d81425299","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-16 10:00:02.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c2e1321f5a7151a5f9d27d746f5e93582769a0d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 10:45:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7143e600bcc34e5d82efdc4f09fdc648 : SUCCESS in 43m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6223e85dc99d469b9741b97c1c102dc3 : SUCCESS in 41m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"783611c7bfc0485a06bc30ac1a56bd84e9dbd2a6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-16 11:47:00.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/7 : SUCCESS in 1h 46m 26s","accounts_in_message":[],"_revision_number":7},{"id":"0dd8bd11683513c203f415a07acfd196d0aa664c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 11:53:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3ba0289982104c119fe27d8a40061881 : SUCCESS in 2m 59s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/54b26e02bae649168a70a22e36407d6c : SUCCESS in 41m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/01aa19909f8346f490991d9d222c9b5f : SUCCESS in 1h 48m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3fa910a67e82449d82b669a61481ed38 : SUCCESS in 1h 48m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c015d48a18aa4a61a5148917644b6a5a : SUCCESS in 1h 33m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bd916ecebbf94a048b54bb80972f06d7 : SUCCESS in 1h 47m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6791878436d84f958981d9d1079815e0 : SUCCESS in 36m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e15b101abf424a26b283430169b3f51c : SUCCESS in 37m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7aeacd1b3b944e29d2c924d63931906 : SUCCESS in 38m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7696233f89b3486d94f6bb7bbc02dbf5 : SUCCESS in 15m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/205815a75df743e690c58b1d7570df40 : SUCCESS in 28m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b64f5ec9abe418a93e38fba6cbd7b98 : SUCCESS in 27m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/571a50575a304217aa1fef5f3380d18d : SUCCESS in 26m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05333e07b1664cc3ab37d08e11785d53 : SUCCESS in 12m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e533a45b9add4fdbaa4587913106e26f : SUCCESS in 49m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b82b00bde024cc3a8ef0687a01f2479 : SUCCESS in 1h 45m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/09c09e1349a541c4952a8d6290921987 : SUCCESS in 1h 08m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/da6e7953e8c041b0a9971531437b23d3 : SUCCESS in 1h 00m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/97838e9a09724609ba0def3059a4496e : SUCCESS in 1h 06m 23s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/194c2c4eb5ad4a39985d8a3a395251cf : SUCCESS in 1h 06m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ed93a5ababa94c01bd02331b71bd1b5a : SUCCESS in 1h 13m 52s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/e9d8918c10e24beb8df9ce9efab6cb92 : SUCCESS in 1h 29m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27f5a011359d4268af45bc52ff62e0ee : SUCCESS in 59m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b608a7d74a144e228e9fb0594a29ae0c : SUCCESS in 57m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4d9a232a5a564761b15439291b21492a : SUCCESS in 32m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/667e5d1fb25745db88718a65123e3568 : SUCCESS in 1h 08m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d0caef11056d499ab9754c9bbd1f72b8 : SUCCESS in 1h 15m 41s","accounts_in_message":[],"_revision_number":7},{"id":"7c62d05d33cfa5d3731814845fea59e02f4d3ba3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-18 21:25:50.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"0c0733c9ea3714464db8ce365f152c1374ae3760","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-20 14:17:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"eaf4fb6599def38458464694259639c2dbe1fbc4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-20 14:34:50.000000000","message":"Patch Set 8: Code-Review+1\n\n(3 comments)\n\nVery nice, waiting for the CI","accounts_in_message":[],"_revision_number":8},{"id":"9fd5fcd7c3bf82eff9f61b83f0d066778cca97ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 15:19:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67551f8d497b4a90ba9521c107483311 : SUCCESS in 36m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b7982921bfa4bc688da0ebba74fa665 : SUCCESS in 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e52a998dc2b90dd0c56e34e8d765f4f8fd5b2c1c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-20 15:51:31.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/8 : SUCCESS in 1h 32m 16s","accounts_in_message":[],"_revision_number":8},{"id":"f5ed342f7e9423f83efef673f535aec86ef02d3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 16:59:16.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/10f68eb8f4fe49ebba80ddb8fe42667b : SUCCESS in 3m 28s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/99dddb7a0b6c47d2bab4058e6a085b9d : SUCCESS in 50m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2beb083b8231488f8abd0862aa0ea108 : SUCCESS in 2h 10m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fd1c747f4a0042f4b71d05f85e5fb37f : FAILURE in 2h 31m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d0b098e7f5ee453686c3c2bbaca26c47 : FAILURE in 1h 38m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7777739aa0f94ae7865465470e6c0f80 : SUCCESS in 1h 45m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/13c1874d5d93432aba4cfb34fe70fe73 : SUCCESS in 33m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d12833bf0dc4581afc735bd951d803e : SUCCESS in 32m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71675366e0e0420ebaf7b42744b3d2fa : SUCCESS in 36m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa1b493abc6c42c6bf62f96eafe9a65c : SUCCESS in 14m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c79db0bd2d340949d2057ea0081aa42 : SUCCESS in 25m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5786c3f0a7246a6a0f72c7118dbd973 : SUCCESS in 26m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3e4b1cb7e21498b9438e111ba7bd1aa : SUCCESS in 26m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbed8c5cbb034d62a5cda7a534abc9b7 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a43d966d263540c2bef10112df2bc4a5 : SUCCESS in 13m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28801c0d6c6c4abd8ba317d545db30bb : SUCCESS in 54m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a06c7cf4ef1485a858cb0c13e4aaf20 : FAILURE in 1h 28m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/621ccd126462452b83571a365692cd1d : SUCCESS in 1h 10m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d00c25984b0741f99bafec9dedac3eb1 : SUCCESS in 1h 03m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/258446eb145c48109b6fd534e9c44a76 : SUCCESS in 1h 09m 43s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/59a1e7fdc58b4ae897db7bc99965ac3b : SUCCESS in 1h 12m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb102080dc5546e1b921290704afc473 : SUCCESS in 1h 18m 20s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/316154e9278143388d796bc475124bfb : SUCCESS in 1h 28m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf2a753cd15d43eca70964340f71b925 : SUCCESS in 1h 01m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/827b823e3dd746c89ffe4622182dbad9 : SUCCESS in 55m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4a94e6ec6d584f2caa3955477007d02b : SUCCESS in 34m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4e359c206a37407381461598da493e00 : SUCCESS in 1h 09m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cf8a9544915641d0b09af5cb53cf55dd : SUCCESS in 1h 15m 01s","accounts_in_message":[],"_revision_number":8},{"id":"6bd9d9dff5050bc610a636621fa57e3b0c0a68bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-23 13:04:18.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"166e6fbceccae48460820887c877d362a602df9f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-23 13:09:42.000000000","message":"Patch Set 9: Code-Review+1\n\nWaiting for CI","accounts_in_message":[],"_revision_number":9},{"id":"71601728c78226db50840723f77a8e3cbda3f2d5","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-23 13:19:35.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9: Code-Review+1\n\u003e \n\u003e Waiting for CI\n\nThis patch is going to fail check pipeline until upper-constrains.txt [1] are updated for neutron-lib. At the moment there\u0027s an issue that prevents upper-constraints.txt from being updated for neutron-lib v2.14. I submitted a fix [2] for it.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/805352\n[2] https://review.opendev.org/c/openstack/neutron/+/805627","accounts_in_message":[],"_revision_number":9},{"id":"92cde3ae72001d39680da176aae8412beea788ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 13:34:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f787771df25a4657b25eb94e7f53916b : FAILURE in 8m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05b54c726c5a472798b3875655c00a67 : FAILURE in 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7a385a2333f9c45e8b82f4907ca302092d0364c8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-23 14:16:17.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/803045/9 : FAILURE in 1h 03m 59s","accounts_in_message":[],"_revision_number":9},{"id":"a517a17667553dd8ed372c055fabd6e4da66a98b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 15:01:19.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1719d3f4fd94d89b46a772ee12d6472 : SUCCESS in 3m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e74a7a15ea7244ffb10595b7b8e30edd : FAILURE in 19m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/569d0542804f4f71994e582c2d04f332 : FAILURE in 18m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c47b50f9e59a4d68af8f57822fc884c4 : FAILURE in 18m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f513abe001844a10b633fa10b662c006 : FAILURE in 17m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3aafffdacd1b49b1bb4f0769a1cb5ca5 : FAILURE in 20m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8bf31affbad34fb28cd005417525f162 : FAILURE in 18m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0762724b9bc457cae925227ba6102c1 : FAILURE in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e37e6f954fd24d539fa1108433fa7555 : FAILURE in 35m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8771cbf22b4b458e894e23e2a8ca039b : FAILURE in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76487c4f33d7485f8159f6ca84a7eee3 : FAILURE in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f94ba492e4ff472c9a54a67f57f66d7b : FAILURE in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9f76d8f874147f98ec266c269781e1f : FAILURE in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5720e4ab36e4a7abf9808a0403df9be : SUCCESS in 12m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/711c31b61053444cb95832540eee5a15 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e99df6f57d184f0abb1a6fc381b27024 : SUCCESS in 11m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6808098ae833444784f6eee49bac39df : FAILURE in 22m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/66d940ad7f09410fbab430e93e3ecdc2 : FAILURE in 20m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b36777c70d5f44cc9906028905a8c459 : FAILURE in 15m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e915d38d14494699baef79a85b5c1f53 : FAILURE in 49m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f689d49a5fb240ac9a5d36b83c7aae71 : FAILURE in 1h 01m 17s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d51a43a9a69b47d9bb50d53530847a71 : FAILURE in 18m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1f5566a647a846fda66aa78136eaf4aa : FAILURE in 18m 58s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/35a11e43c78241c58d2b8ecf2ae617e3 : FAILURE in 18m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8973bf5772e0416a816987fc1f52ebef : FAILURE in 19m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a17e6fa982ec466aa964a0732cbf6b57 : FAILURE in 20m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/309f5410d5c64d6ca8882c0554ff9147 : FAILURE in 19m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4d2d40ce0c9944ef99f396df6b6178f0 : FAILURE in 18m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/71e3b802262b4d49aaf9502f6e1b5480 : FAILURE in 25m 08s","accounts_in_message":[],"_revision_number":9},{"id":"94f9495fd6bef0382facf8d5dcf5137e06b4aa7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-25 14:47:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d1be77018d44f246e4d806523ec1a36e027197a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 15:32:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50dafd2879f34e97b295eaee156d5a96 : SUCCESS in 43m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6fd1e86fd5044d728b1f1345f5edce0b : SUCCESS in 42m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9c0e827390c97bb496126236d8d8e4c8b7d7b1d7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-25 16:22:56.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/10 : SUCCESS in 1h 31m 48s","accounts_in_message":[],"_revision_number":10},{"id":"4e9324b1e30240ce022954ace0fdbba669a5b36e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 16:52:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/541d81a566884137acc7a3c23fd86495 : SUCCESS in 4m 40s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3c934a51815844acab9750649e0441d3 : SUCCESS in 43m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0703838bc12d4096a4dc56fd22d94a45 : SUCCESS in 1h 59m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fc498418f45a402c8279c392341a9186 : SUCCESS in 1h 59m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/504904ed2aaf4511b1d1cac3bd09d209 : SUCCESS in 1h 20m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c83cdebd420d499a9678d3e9d5760beb : SUCCESS in 1h 53m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c851e2024bce45a48341e3ddee1c51b0 : SUCCESS in 33m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f79a42847f84e6d85bb434c7ce05687 : SUCCESS in 42m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/384129f7b8584bb6be121d89b7b26679 : SUCCESS in 42m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/217c7f0f9f1440439d0867400e313cdd : SUCCESS in 15m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc7cba736cc04b3799dbef82e67b1367 : SUCCESS in 25m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a739b3c01ca4687b46d2e1cc8c9b38a : SUCCESS in 23m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c3e24b7b92642d0a24130ebbc649144 : SUCCESS in 29m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2b1ba528734b2b9aeb1eb2d116f202 : SUCCESS in 10m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3de583ac4a354950a800b7cc8ddf4331 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78b8dbd9f32f4118a2d4d0f9c627cb81 : SUCCESS in 10m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1d77035c82a425fabaf894aa61ea900 : FAILURE in 49m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/856395d92709480696a146a8fe10b78c : SUCCESS in 1h 53m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/27c2f714930141bda21a8a758e6f8880 : SUCCESS in 1h 06m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f43c7a266b7455bb0a7fa34bfa2ea47 : SUCCESS in 59m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a6eca16da82346d9858c67ff01441742 : SUCCESS in 1h 04m 13s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dec3ca42fa054e718a933c37275ea79d : SUCCESS in 1h 17m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/88c46944beae41c0909cea5bd7a4f1b1 : SUCCESS in 1h 21m 22s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/2383e6408b334a52891beccf52115425 : SUCCESS in 1h 35m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3226f1ec3f8e49afa8f0176b808dac20 : SUCCESS in 55m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f4973b507dc740ebb161de9956362a27 : SUCCESS in 49m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/edeb5c4b61dd4a10a02bb24a7e8037f4 : SUCCESS in 36m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/de1ce3e32b9d4f478047f0a171e5d9d1 : SUCCESS in 1h 09m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cf03524cd13141a083498f73e0ab45e5 : SUCCESS in 1h 10m 05s","accounts_in_message":[],"_revision_number":10},{"id":"5598d624e6c77104af58c5208ced054f26aec232","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-25 18:32:16.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"26c58faaea15ab2062047a12fc78f5b2eccb72ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-26 04:00:15.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"0194aeff8e28d3c35f3f416d86696c12357a2b8d","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-26 06:34:47.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"416a656bf896ae536c3cc2f4017ad0c1b374304a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 07:33:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39da05c80c4941a69efce6dfa39d9369 : SUCCESS in 43m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76715d4c927e4d30bf0efb11ae916054 : SUCCESS in 52m 38s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e3544d76feb546060b64ad5c6b41e3c927b9266e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-26 08:40:48.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/10 : SUCCESS in 2h 05m 28s","accounts_in_message":[],"_revision_number":10},{"id":"bef2937acd08d314d75b7d7f9ead05b9f89afca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 09:17:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0148537963694e1798d84755274454d5 : SUCCESS in 3m 45s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5adafe4823554bdd8d1e51f18f2a376b : SUCCESS in 48m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/02ea5ce579b248b9bf77b1e835cb1a4b : SUCCESS in 1h 48m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/54209da8346443fcbe65724245399940 : SUCCESS in 2h 11m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/008d7dd42e014399857100c1d696536b : SUCCESS in 1h 39m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/eb6df6f24e3d4052ac3802e48d691efc : SUCCESS in 1h 54m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b2d4906fe26b499b91294480e9537fde : SUCCESS in 36m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6135e75701d640e3944911533d037c38 : SUCCESS in 34m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dac33d54a5946b1901186fc697d89a8 : SUCCESS in 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92f564dd53394a27b412ea1e32a3b286 : SUCCESS in 17m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bce295d6d32041e38b743b0a34e851af : SUCCESS in 27m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21b92c9fea724c7f8103d01802a41e88 : SUCCESS in 26m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b9217d68b654ee0a740f23aba891b70 : SUCCESS in 25m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/339b45f4859342dc832a7239c048f10b : SUCCESS in 10m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b7ed408c193a4b8ea9b5b9d6746c32af : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d4823e004344145a0ac3cbf5031007b : SUCCESS in 12m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/336d18159e0a4bec9d95ec5e79f510e0 : SUCCESS in 52m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d660c005e8d143c8bd66455d6e7b3ca3 : SUCCESS in 1h 43m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2ba5eba818ee48f3a1daac70cede0043 : SUCCESS in 1h 18m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee5777a24bbe4814b122c1209ae3b6b6 : SUCCESS in 1h 04m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/68f2c8424ba649e4b519427126a9b4ee : SUCCESS in 1h 06m 51s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3c73665cd1d34891bcbf876bc18871ed : SUCCESS in 1h 12m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/930c1fc8a86046db95a0bd96fa5faa51 : SUCCESS in 1h 14m 30s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/7c607f4fe4154885a6fd96e041bec0ed : SUCCESS in 1h 30m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a0b00e95daf43d396865cfd0354045a : SUCCESS in 57m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ac4335b45adf4b388c16c6f59a008e3f : SUCCESS in 58m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e4412135d4604540a7bd1efe0308c9d0 : SUCCESS in 34m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0589745d681a40f4b004f3a46a7a53b5 : SUCCESS in 1h 10m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/22e5ab827ea940208066a08ae72d96ee : SUCCESS in 1h 13m 21s","accounts_in_message":[],"_revision_number":10},{"id":"24d707fb75f6ad5e8ecc6a5d88cd5109d8c6aaa9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-27 20:22:36.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"07e53b8868e558232b87c289de24d200c460b97d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-28 06:11:22.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"867a2fccf2c7ea5fd72b2d3fed06a4af9e297d06","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-28 23:27:09.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"864df32d65889588bbd42e859bda05cdd6ab620b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-30 16:22:06.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)\n\nGenerally I am ok with it I have few small things as comments, tomorrow I will check it in devstack","accounts_in_message":[],"_revision_number":10},{"id":"98df8385e0682dd04939ef40cdab34e26bdf5e57","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-31 08:48:22.000000000","message":"Patch Set 10:\n\nI checked it this morning, and added a few api tests: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/806689 , it\u0027s ok, and works","accounts_in_message":[],"_revision_number":10},{"id":"e74c6d595311c5da0943ff3571afe7ca996a1dbb","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-31 09:58:21.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"3d9b6f001e91bbeab82975bbf34af976f90f359a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-31 09:58:34.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e9ca2f0b1a0ad7b2df14d27f31a6c5a2d7b4cea3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 10:49:16.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/272ba3ebcb6a45248ad98806f18a27a8 : SUCCESS in 45m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/095d40a521df453d862114d092c12e31 : SUCCESS in 48m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"09d7e3e4570ec76455368568e779aeb18b023412","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-31 11:54:21.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/11 : SUCCESS in 1h 55m 07s","accounts_in_message":[],"_revision_number":11},{"id":"456dcb61dbe8cb113c4298c865a4a65f2edb145b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 12:09:33.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4fc029edca0c4f6cad8b659b753d2488 : SUCCESS in 4m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f1aab0c8ed4f4ad2b2756983fa6f7010 : SUCCESS in 51m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3f0de14685294e2bb0f3f1830cc16270 : FAILURE in 1h 52m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f70bdebd4d364283890aae5c09d51bf6 : SUCCESS in 1h 49m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5bdd6273b87d489ab5fc55de0a03d0b5 : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b96262412ea04c49b1d0b5227b2b8015 : SUCCESS in 1h 59m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d37343a6171e4275947ef19ca8431e71 : SUCCESS in 27m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1df9192c312b4ee8a130757bbdf6b6b4 : SUCCESS in 29m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f90ea8c8e474a95a71c443ff9dd72c1 : SUCCESS in 40m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3e2556d263b4162bd76bf1baab3e7e4 : SUCCESS in 14m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/577789de453e4ebf9d4df27082b4de8c : SUCCESS in 23m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97a274cff0674f68bb98f71bfc0c0609 : SUCCESS in 21m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f04788e6e56042ccad18fbe0c4cdde8f : SUCCESS in 24m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e678b913e0af43cdafea70e81dc8ec74 : SUCCESS in 11m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/75e75fcecf6f4b2f8c6a019daff627f5 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9bee40b30344fab9310b7ff2e6ce49d : SUCCESS in 12m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d85da4df59f34a8998cf5d15d046d538 : FAILURE in 54m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/59e918404e4d4453b455b94a4fc5964e : FAILURE in 1h 32m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/807be0cab39c4747baeaabe3384aad3c : SUCCESS in 1h 11m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/112f028e138b40b69b8c636dd1678e50 : SUCCESS in 1h 13m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a9187985eec540d699d13b480eb0c45b : SUCCESS in 1h 00m 25s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bd463d1842884be2a41154fcd5065f6d : SUCCESS in 1h 15m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7df0ca8197a947159100886bf3de19f6 : SUCCESS in 1h 38m 40s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/e12c2fe3362947b4a7d52b419240f973 : SUCCESS in 1h 29m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b71c6bf4ac4b43fc9f2a8782317c53e8 : FAILURE in 48m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a70d518b95144db8bb59e260a6ad8ff0 : SUCCESS in 1h 08m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/eb768f595f71490b92e6d21f36685c3c : SUCCESS in 39m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f92c124a35274a6c8dcea6bddb6b9e23 : SUCCESS in 1h 15m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/043e9b897e4249dab861a3a49491dba8 : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":11},{"id":"e3dd4080fe9f03960d52a5b1197fec7acbaf3f50","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-31 12:21:39.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/45/803045/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"22e1b4ddf548c100562f380500a775467dab2ec6","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-31 14:19:44.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"7e9e31326696d14bf885d9e684eded4ecd8ec79e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 15:15:18.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61e62fc0fa054d7c81425acaa2348971 : SUCCESS in 41m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/051d8306ad7d474aa7143937dc6e4714 : SUCCESS in 50m 05s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"64fcf11ec03c28defb6234841b42e6d0363f7a94","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-31 16:55:12.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/11 : SUCCESS in 1h 33m 55s","accounts_in_message":[],"_revision_number":11},{"id":"7b05b9a81dde5ae83fbc273bce1ce0ea419549f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 19:05:14.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7af72944c57b48d49a4feb94860f040f : SUCCESS in 3m 04s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/636031633d794a34a6574cbf0f390f94 : SUCCESS in 45m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/edc6a2c6acb541e7b49a0623c4def1d5 : SUCCESS in 2h 20m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4667853f185241e29266d6fa54a9257c : SUCCESS in 1h 57m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/53fc4a3dd4c24433927f0c18ce5bb941 : SUCCESS in 1h 29m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/86991685539447b293a5a453e80a7d23 : SUCCESS in 1h 57m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8f597d9c98894d9bbdd3a4ea1abd6377 : SUCCESS in 29m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0a75e6b62fd4ba6b0962982e02c9050 : SUCCESS in 38m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a26de20c8fa4dcc8c9684f09fc055bc : SUCCESS in 41m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a338e5de5d2546bd90814f0fb925f7e6 : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cb69581c4f04189b2c19dea7f8ab1ef : SUCCESS in 23m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bc4a0c0e1a14d249747f34bbb4a2237 : SUCCESS in 24m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c43b81cc93314aed84274dd81aaa10f4 : SUCCESS in 24m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa1404de8f24db9b45dc3270c3b0d21 : SUCCESS in 10m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac2c2b4609e845e5ba72b0998a097d2e : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9af6755152564305a2298474e8af448a : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1793fb29dd3b4a3697c420a223bb6a43 : FAILURE in 49m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/96d02c06ef0447608f85f75441ac949c : SUCCESS in 1h 50m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/402e9c8ac1f8408ea76648a960d0e819 : SUCCESS in 1h 10m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0383723002424fa2990251a2b3611f10 : SUCCESS in 1h 02m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/918d86d15a8c473f97260d1068abf3f4 : SUCCESS in 1h 00m 53s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/73b388235b024bd587269e9cd72dee63 : SUCCESS in 1h 08m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4daf4f967aa44a56b10420ed000257d5 : SUCCESS in 1h 26m 55s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/9545a707ef744cc6bbf87ccf2b52b645 : SUCCESS in 1h 23m 00s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/568f6d44c8d34679ba4d88a1c091f5e1 : SUCCESS in 59m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ed2ef3982c704adfa769a6561af9e96e : SUCCESS in 53m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/563da3da24cb46618327ccf1bff9e821 : SUCCESS in 28m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7727a953f7e74cc898b1c18fe1039608 : SUCCESS in 1h 14m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0ae91becd3494d7187c0949df7818ef6 : SUCCESS in 1h 13m 41s","accounts_in_message":[],"_revision_number":11},{"id":"27e47510d007b20394f12f86138c73612f86e4ae","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-01 06:27:46.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"aafe3705dcd7aafb24c22be810147a25c7a2bebb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 07:16:12.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc72490fbd6a403198af72d047104b44 : SUCCESS in 43m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1914b75500641b19af1abb8fafc78cf : SUCCESS in 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a342db6aed5f0af8371ef2c9e8a0d098d00228bc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-01 08:18:48.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/11 : SUCCESS in 1h 50m 26s","accounts_in_message":[],"_revision_number":11},{"id":"663d983274db9146fc057f1b2e8c25ade38279d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 08:51:18.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a9fb7ae447834e82b5c19637cfe33db6 : SUCCESS in 4m 02s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/79767515df1b493a990f92a49bdbff10 : SUCCESS in 47m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8263a7151890488e937c0bdc1b978d80 : SUCCESS in 1h 47m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7db69b13634f4985b74b2ac97c2a88bc : SUCCESS in 1h 49m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd9a98b1e5054ce8907f12e56ce32bd0 : SUCCESS in 1h 21m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1b939f2c244040d98cb35dc8c4fcc46c : SUCCESS in 1h 48m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f5905cd98dbd4eee88a8c2b65ae46da9 : SUCCESS in 31m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f8400400e044a1fa3e11df239266fe1 : SUCCESS in 31m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f0066f35abe461c9ab089f38adccad4 : SUCCESS in 35m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78391ed62bea4626a0e87b339eea8e30 : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f73e62c87f14e8e90a0e606a4e19204 : SUCCESS in 23m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37748d45c92d45688e884d96006e49a3 : SUCCESS in 22m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2711062177422f8ed01d71b8bf7029 : SUCCESS in 28m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/176fa6fe3a984ec29a18e0e5348acd08 : SUCCESS in 11m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9d32ed4adf194f8f9da70ee80c9fad2f : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93d60ef05a244fa09fcd8522ecaef669 : SUCCESS in 10m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a82c682f60047c38091191271eb2b0b : FAILURE in 45m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/02b06423d1d84c31bf1fa93edcad477a : SUCCESS in 1h 41m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9c17420525044bd79b28a8f6473cd049 : SUCCESS in 1h 12m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9dbbbfd3eec04754a346f0967b60f120 : SUCCESS in 59m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4ce61c293b554251b30de7381c1fddc1 : SUCCESS in 1h 00m 37s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5871999489304e7f96cc2062e43e4996 : SUCCESS in 1h 07m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/43cc9b8a335b44aaaa911b416a4acb15 : SUCCESS in 1h 19m 01s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/95da3fca71774e1aba10bfd4d1903bba : SUCCESS in 1h 29m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70698366640a419d9d8cfbfc308cfdf2 : SUCCESS in 1h 00m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/433447193afd4ebe8a54d229811f9c9d : SUCCESS in 52m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a8906e7a279740ec86043271fbad041b : SUCCESS in 35m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/16c911fc731c4f96945b141b57215d6d : SUCCESS in 1h 09m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7d6d010280924d469ff2af6e0794bac8 : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":11},{"id":"12d321da078a0abe3f124312336a9caa842360af","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-01 09:21:12.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"55e165b259c2a2737655f0c81ebda08e2429e503","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 10:09:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6232936cb15047519ca827e711b56309 : SUCCESS in 39m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c6c4a0d132949208310b2e523dc4cc8 : SUCCESS in 39m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"cd78a9d672b5149f11978e373e59d9fc67cf2f1a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-01 10:53:45.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/11 : SUCCESS in 1h 31m 52s","accounts_in_message":[],"_revision_number":11},{"id":"f78a316d6af0aa1cb4b8234e17cb0125af1bc971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 11:20:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cbc3530a6a874c9bab212da9c9560519 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/97a98ad82a13459ca165f02b4c773968 : SUCCESS in 45m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e9b5736af5b642ad9965cf70e6b514a7 : SUCCESS in 1h 57m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/12c6f508e6ca424385c9ecc20b958361 : SUCCESS in 1h 51m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8da9170aea5491eae4559d64e6db41b : SUCCESS in 1h 19m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c787d381275044e8978df2a8ef753458 : SUCCESS in 1h 55m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70536f4a52204d1d8cbe0b5eefb334ae : SUCCESS in 30m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90b22b62d5744e618e22e23da2ab3196 : SUCCESS in 42m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cca35e5d90474f24bb8601bb57a89bf1 : SUCCESS in 42m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2641ad70e84f3dafbffc879d6bc3bc : SUCCESS in 11m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd4fa5b31ea6477bbbd7f22226281282 : SUCCESS in 25m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c44fcb61b07d45a3af949db99e81f122 : SUCCESS in 25m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e9648d4780142ec8b8f637d248bfa91 : SUCCESS in 21m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/851e251d896f47719c844c04cfcb7651 : SUCCESS in 10m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3ce5a3b4ba34dc98db2638cdfe75bdb : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c387dbc464474099ac0d39a11370a8db : SUCCESS in 11m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/57801a76925946358ab0475ce0d11ed4 : SUCCESS in 44m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31f6b26fe0304184a1d53701e05fd4c2 : SUCCESS in 1h 47m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0fa439f6bba44119ad574ad8da9f11b8 : SUCCESS in 1h 06m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a7199a5ec5f4d2dac61fe7c2d154c07 : SUCCESS in 59m 04s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/efc698c96f6f440cb469ad599b5083f2 : SUCCESS in 1h 02m 12s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c9d9dbd969d64f438a9ce4d2088f3dc6 : SUCCESS in 1h 07m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bd2d51cdcbcd42b09fee66cfc32c88cc : SUCCESS in 1h 15m 26s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/be3c41f596ab4f8bad436f7d75629aa1 : SUCCESS in 1h 27m 08s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e3a34ea23534082b987bdd6cdb8c22c : SUCCESS in 56m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3d5329b4fd674c64a7124588bec7df77 : SUCCESS in 56m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3d117fdef15f482c891d09b4f818d13e : SUCCESS in 32m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/10f7aac0f2a7430cae7390a45cc2d8b2 : SUCCESS in 1h 09m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4b6a3762792f45de90998a009379a38e : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":11},{"id":"b9a0110c20199ffb26d75c4051d2e291e643369a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-16 13:58:29.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"ad27fbc88b76b9c534ae223629ed90a54c1731cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 15:08:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d2e89437a274ded848a2a9131e44a7a : SUCCESS in 1h 01m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f68ce2520034b9c88e54d9f8a8b09d1 : SUCCESS in 1h 01m 58s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"aa457db5f4220d35de46b36b21890bdccb6b5a0b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-16 15:16:08.000000000","message":"Patch Set 12:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/45/803045/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"0c581a716995c810c2080e2c69dd22926302e577","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-16 15:56:21.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/12 : SUCCESS in 1h 51m 45s","accounts_in_message":[],"_revision_number":12},{"id":"79034cf65a81a54224f43d6514ffb1364fd3a0be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 16:31:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/22da209152a341e980640117a67dfb7f : SUCCESS in 3m 27s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8cbbdc08ef0c46a6b690dd81705e9fce : SUCCESS in 42m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/62a07d32e9114c16af6949689dba79d4 : SUCCESS in 2h 22m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/171c2e2813034d9ab6a56bdd972a80e7 : SUCCESS in 1h 41m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f6dcdb0453f14f459cadd70468219d3c : SUCCESS in 1h 37m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7e2c65b2b7474f5ab71a8e6f777efe81 : SUCCESS in 1h 37m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d41c6af29ee42ac81bb13e9bfe963ce : SUCCESS in 48m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfd6a9c04e004065b51ed65ef4400498 : SUCCESS in 38m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fbe405f77614813991e32c61dc6eb3f : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0bcaf753fa84e9e8a2bf95828c91189 : SUCCESS in 15m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e41a2e74b5a4ec89c314c6fd79701c3 : SUCCESS in 27m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/271d30f1aaa54ba2935967558b4eaf07 : SUCCESS in 25m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65eeb261e03c413b9521b8c9bf1d340e : SUCCESS in 29m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/068c3753cebf4c1e87103f49ff7c22d9 : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7aab866186b461cb231808cf08daaa2 : SUCCESS in 11m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/27ed6eb858684fee8f3c58acdd8810dd : SUCCESS in 47m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/00842c39d8f14f249a1de3c3f6dd7935 : SUCCESS in 1h 45m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7e142bfe2c8e4de198e72147cef4b74c : SUCCESS in 1h 09m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e586037e4764020a15202b641597b91 : SUCCESS in 1h 02m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/44a7504d07f1418db2b7fab357f5063f : SUCCESS in 1h 04m 26s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e970c822d3b848d38155c16a6d5ccc1f : SUCCESS in 1h 12m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/358421e9622e4a129b05b980f48675c7 : SUCCESS in 1h 12m 26s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c6681852faa743708c0852ba83e39e1b : FAILURE in 25m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4486d091e6041ba8a6e1e7dbaaca72d : SUCCESS in 51m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0f92735cc21f4bbea8d1e9c6406a8483 : SUCCESS in 1h 02m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4ec5ac6867dc4668816bd845e9e0f5f6 : SUCCESS in 28m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/510e33daee684463ba69c2f0a694f09b : SUCCESS in 1h 10m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5bdb4facf77c4b129eb6439f0a6ff981 : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":12},{"id":"7fd5450055af55a82f1296188db81265d83f5675","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-17 06:41:29.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"bbf0c9545500fc8c62d821ff6dfc30fb21c928e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 07:37:22.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89de112e552d4e80a367b169a5b9ce33 : SUCCESS in 47m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01b92bf030994599bbee6cbe5e25fe91 : SUCCESS in 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a4e2584ed25e5fd85f3c51b6f4f2434c10032fda","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-17 08:33:00.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/12 : SUCCESS in 1h 50m 53s","accounts_in_message":[],"_revision_number":12},{"id":"f59c75f2469a9010f203d478fd75dbd5c2c3f671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 09:58:44.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e3b1710d495443fa97b1892ea08f753 : SUCCESS in 4m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5f0de37c5cf940b188f0ec9272c03c0d : SUCCESS in 41m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c7848cee31de42c0af50449882c59142 : SUCCESS in 2h 02m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a1fa22911bd54711bc7c085090be6927 : FAILURE in 2h 40m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3c188d2a4f2e4e02adb63f9ba3e0b390 : SUCCESS in 1h 32m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f1b10432064642559912b07cc308200e : SUCCESS in 2h 19m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68f2f05c4fc74090b76a9ef93e2ef3c7 : SUCCESS in 36m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6697ca5b4df8449cadaf5c466b23cc75 : SUCCESS in 36m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77e7daf70d1042cbbcfa9a239c734152 : FAILURE in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ad03cd9e17b4ebcbfcd7c474358a130 : SUCCESS in 15m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4957221362274a1e8f6c3a7d6a02f5c4 : SUCCESS in 25m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2ab3103d3c8492a9dfdb4f682f9179d : SUCCESS in 25m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/039530ddf9d941ca8f6ee4fd1dbbfa4a : SUCCESS in 24m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd609bd99d0451fa03cc965094ce60f : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66d1948d63f54e96ab23196598913975 : SUCCESS in 11m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/533de56f18804646b4233af8de4ea3ac : SUCCESS in 52m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3501b0c41664060b55c7b16ebe02343 : SUCCESS in 2h 09m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/81a64dd42e4f4d7dbec3bc4266f54077 : SUCCESS in 1h 13m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/37bd0bc63d344b84997196df3dd8dd22 : SUCCESS in 1h 00m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/88cea1baac1c4199a80acd2f483bdd42 : SUCCESS in 1h 01m 36s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/108575002e94451dbf232ea72336993f : SUCCESS in 1h 06m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4f330e4ba15042b19f3b52c2a0be751a : SUCCESS in 1h 16m 18s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/d9b0af5ae89143e6a0d53f558b73e6a7 : SUCCESS in 1h 27m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6abbe0672667423a9e4839e43ad89b8d : SUCCESS in 56m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/44889c19c5e84e4ba016bf407a262cc1 : SUCCESS in 51m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9de1e8d42aea4528b4e3c4325e05e1aa : SUCCESS in 28m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cba208be9deb4cc2b64b74a302e66788 : SUCCESS in 59m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f8a3eff59a014475ad8bafb145ad209f : SUCCESS in 1h 17m 05s","accounts_in_message":[],"_revision_number":12},{"id":"ce2499959c88efe19adfe6384a68280c76964852","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-20 06:57:51.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"2d25407cdfd7108b5e8e4acfb0fa001067266855","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 07:56:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98e020b73ca34e1a95d549da9df8c7bf : SUCCESS in 38m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b22fe9d862f24e90a71a39b1f74bd95a : SUCCESS in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3aae558b5bcbc30231fc25a9b624ab4a9bcbc700","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-20 08:27:44.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/12 : SUCCESS in 1h 29m 02s","accounts_in_message":[],"_revision_number":12},{"id":"1a79ab99b97075400adb99f1a1f6ae34223bbb89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 09:49:09.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b23e8b7792c84c25a36201b1104c0ac3 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5775c18eba114aef9ccf305413783c4b : SUCCESS in 45m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cb87747bb08b45bfb130e95911cc11d5 : SUCCESS in 1h 54m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d07a1399a4364f99a4b832bb0e0707c0 : SUCCESS in 1h 42m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe020cd71ff94ef79150cfcce569a856 : SUCCESS in 1h 28m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/36674181a94c41b7a17da34c21b5a44c : SUCCESS in 1h 49m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c53c4557bae4aa98d6c997b52c05dc7 : SUCCESS in 31m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4084e7cfc58e472ea193275b3bc6f21c : SUCCESS in 37m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfc027bd6c7b49bd8c95c33337b9ebec : FAILURE in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9f24f3f1334d4bb93a8afa69ddc708 : SUCCESS in 15m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b3cc2655c4d437a9ecbb4184aef6afd : SUCCESS in 28m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8adf4253c7cf4fd9a19a46169f986baa : SUCCESS in 25m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af5fe20684404f1280c5a3f94d07be8b : SUCCESS in 26m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/826819365a564fe18eaaf31dbedf4a81 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0af6b63eac6e4cc284b90708060a1211 : SUCCESS in 12m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9210e6e5c7de41e985ae9c52407e0cbe : SUCCESS in 45m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f4ad52cc97f4880a2cfbbad6663ac04 : SUCCESS in 1h 45m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5d92a2dbc0c24178a1ef613198401162 : FAILURE in 26m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/686dcad252cc40c2954a17b2f0988abb : SUCCESS in 1h 04m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1f93ce5fc623410aa2bf094d57905ae6 : SUCCESS in 1h 02m 20s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/16dc1d20a15a473f807c197b99e57a6d : SUCCESS in 1h 11m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/90a33880f26b42c5ba8c7fc8e0573652 : SUCCESS in 1h 15m 56s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/88f8bb755d4746838f111450a67aad02 : FAILURE in 30m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c4ca06dbaec4bc3803257db3341482c : SUCCESS in 56m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b083729cbe4a45a78fd144e0c1f17b6f : SUCCESS in 52m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b4bd52772b9143f0a24403344ce4e0d2 : SUCCESS in 35m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9edb7f4c1afc4cca919f93bff24ba08e : SUCCESS in 1h 09m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7bf783567b9446eea84b6b1826161a0d : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":12},{"id":"061afad7db879516d7c8136f5e38de19c84c629e","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-21 06:45:11.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"19c5e67cf444330d02c0d59749a20b575122d358","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 10:01:38.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2ff538c2ea44e869498c9b08fc2ef72 : SUCCESS in 38m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5dda8072e61c4ff8bcfad7ed09a6a8ab : SUCCESS in 36m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4cda41786d696af7d3e986b014ab727a4705dd7d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-21 10:05:05.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/12 : SUCCESS in 1h 51m 40s","accounts_in_message":[],"_revision_number":12},{"id":"6ab739d7166edb4c17ea1b92ab4799aae8eb3963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 10:57:09.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4c37a6b0cf643628dc5fe48c3130516 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a0b37b21aead408ea56240df34d2420a : SUCCESS in 44m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f13453cfbe9c443e922b659db3e1ee49 : FAILURE in 1h 42m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3c561f3ec40b43c987cea5d120899718 : SUCCESS in 1h 51m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cdcc0c48806b4d41b9b66ab2472f34f3 : SUCCESS in 1h 32m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/115502e415a24fe19d55f028711c4803 : SUCCESS in 1h 52m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c5422aa35014d708d9d1155e8653839 : SUCCESS in 29m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eef97cbda60e49ebae8827082dba7146 : SUCCESS in 35m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d5f6eca35df44a180a12647d95d6ed7 : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/587091e9d06a4f50a80bfb2aad4f1e57 : SUCCESS in 16m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59dd377c2cbf4272987d7f20c43a66af : SUCCESS in 28m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44bd5ec45ae04881980c02a98c4259d9 : SUCCESS in 26m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d65f98c068bc40f8bd8455dcbdd87396 : SUCCESS in 25m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ef43ec83124a3ea4bfd43145295d09 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6618d62d9b744f0bbd2bc24c78f56a96 : SUCCESS in 12m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3387f8c8f294d45b423de58dd7b5cb5 : SUCCESS in 51m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1284cd30c7c849e3bbe5e4ac84d104f2 : SUCCESS in 1h 38m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/65e8ac2750114b1986ef2442fdb5378e : SUCCESS in 1h 11m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/69c81026e29741a8b5c10f3812b4babc : SUCCESS in 58m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9a9d22cfca04bf3a2f3c05d3d16679a : SUCCESS in 1h 02m 58s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8d9b56c991c4407889d3ebe1462a6bfe : SUCCESS in 1h 10m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cf5666f5741a46079b68c37f9a03b0fe : FAILURE in 1h 01m 29s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/9ff275c2f03744c58427516428bca77d : FAILURE in 1h 27m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e389e8701b164e9baf1c23aaf80e2b53 : SUCCESS in 55m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/65b74954ebbc40599d24092f2b94864c : SUCCESS in 1h 08m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b5aaa610ad2d47909632461cbd015830 : SUCCESS in 31m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e7f474c2f6b241edb912b4cc5dadc48a : SUCCESS in 1h 14m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2f105061efa2442e987407cbb0c7fb40 : SUCCESS in 1h 31m 18s","accounts_in_message":[],"_revision_number":12},{"id":"1246204e75171a7d62656a7715cd5f87e37cad88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-22 06:56:07.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"5e205bd0e6cb14b9328b3b1d8818992b9ff5f8f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 07:45:23.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b279142d53e34ae5b7ad56308b8cbf2e : SUCCESS in 37m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aef020b6def743c9866beabdb20b85f3 : FAILURE in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2353d03a5892cc6f99e3965995806c24d03816c9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-22 07:51:41.000000000","message":"Patch Set 13:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/45/803045/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"31a01f5fbdcac43df564b31a5c5bd14864934437","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-22 08:27:19.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/803045/13 : SUCCESS in 1h 30m 35s","accounts_in_message":[],"_revision_number":13},{"id":"54ea790dbbc9dd8e37928e357c9d78ce9a93ddbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 10:06:25.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f93364486a984d4393c341da0845990f : SUCCESS in 3m 27s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4c9ecff9270d4ef9ae11feaa11e5539f : SUCCESS in 42m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ee7df9359a1e43a88678da8e937a5bf1 : SUCCESS in 2h 03m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/01acd9d777644cf5a0f16d254ce1289b : SUCCESS in 1h 45m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0923e1074558450f824f4f01d9fd2039 : SUCCESS in 1h 25m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8af1333774cb4cdab282cfd6bfc58bab : SUCCESS in 2h 09m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af09d31a5bee4428b364ac06381baf3c : SUCCESS in 32m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ad44ab0613142f4b359f6c2e7c88062 : SUCCESS in 37m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6118e0aae8d45faaf3034b47bbd2668 : SUCCESS in 46m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e59732822ce4c879a8721c49c2f004d : SUCCESS in 15m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ab9d65c6b924065af62e495ed4af40a : SUCCESS in 30m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9114f7aded7e4ae6bbdc48b302b89883 : SUCCESS in 25m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d44b6a1112a649bca3f9e870b88efc0a : SUCCESS in 23m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ebbbe58ebc4635bf6b5b0d368b2d6b : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd8364ab58d643c3950a8561899501f8 : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/413700cedd0d4432a6a1bacb7d010e24 : SUCCESS in 46m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/63f114e6cd1749b38274ed55595ab30a : SUCCESS in 1h 45m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0cf5b15e228746269b0ab93ad1a201bf : SUCCESS in 1h 09m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c62eb5696197496bad6f8d846d52e9a9 : SUCCESS in 1h 02m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d008bd770c4e42cd9399a9c3e24365fb : SUCCESS in 1h 05m 27s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1fc67c1fe9a444359995f62b62c901bb : SUCCESS in 1h 03m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/11ee6b7a62cb471097471082f3d3cf3f : SUCCESS in 1h 13m 34s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/4ec08cce567942dbb4228882203218f0 : SUCCESS in 1h 27m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71276be086824dd29516dc2835192e82 : SUCCESS in 58m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8a5a35cbb982432c85beea2cb8bcda00 : SUCCESS in 57m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/26e31130c4df4bedbc3fcb52be6df186 : SUCCESS in 36m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/15cc5447fa8d4579a1ceb937a6c4a478 : SUCCESS in 1h 11m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dfb95d4ef6c449e9a2bb061cdb4fba56 : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":13},{"id":"01023df207196edd9599b3db765874d0c2bfd3e4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-23 13:19:02.000000000","message":"Patch Set 13: Code-Review-1\n\nIMO, the code is OK.\n\nHowever, you need to rebase your code on top of https://review.opendev.org/c/openstack/neutron/+/810036 and change the reference from Xena to Yoga in the DB migration.","accounts_in_message":[],"_revision_number":13},{"id":"329007ab0cb72ec7f64935a4bba75be5adacfe5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-28 12:40:58.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d60daf4d42da3db6db785e8de49379cc77c07171","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-28 13:28:35.000000000","message":"Patch Set 14:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/45/803045/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"518bc4a1b430e08a6ccedb606e6cfeb268a45e02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 13:33:11.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76db8f7a0bba489183c92f30a23af2d5 : SUCCESS in 36m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23a01a05be4e40e4a01c499ccb6b3dc0 : SUCCESS in 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2743f8ad48bb2afcdab12f018489c7e548942bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 14:21:55.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a85f08b513874fd7b137768074afe7f8 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b78472c597fa40cab85c256ed019b9cc : FAILURE in 28m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/713bfb8b310f4b819d6afeb87e82f199 : FAILURE in 26m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3a17bdf973234dd4a3bc3e157d06a99a : FAILURE in 27m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7d8542b08b694b8eafb8c548af965469 : FAILURE in 25m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/24f80b4e7c8043338784ad7531d2ac7f : FAILURE in 30m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03b594829958460490633d8f01a4f4f3 : FAILURE in 26m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acf1770f7a794ed8afae3a43a0da278e : SUCCESS in 36m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e8207cdbf5542d09316bd62dd338356 : SUCCESS in 40m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc07ea3b9c9412884d7778b496632b5 : SUCCESS in 18m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ecbb64dce7c468b9d23a073a51ac95f : SUCCESS in 28m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2f9c7edbb5743ddbe3366d336f31850 : SUCCESS in 25m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2035d7e4d274a77b1815ffc56a2f74e : SUCCESS in 25m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ab0d68955541b196d0d33c60496c54 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77000843239648f1b9c5afb869265e74 : SUCCESS in 13m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73467873a1a2497683faaefc720817e2 : SUCCESS in 1h 00m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e99389da8354b92bdbd8b4f394e1946 : SUCCESS in 1h 35m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8c109b1c2fdc4ad2a683d72979da5ab6 : FAILURE in 27m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9428b4bcf1bd4de599d4698bf3715d3f : FAILURE in 34m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2d024ca7fc5442daa5f8d5ada0285edb : FAILURE in 34m 13s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b6c9d71616a440608550e53b6295ca66 : FAILURE in 30m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e15baba6f325477aaab2a1b60c7bea66 : FAILURE in 31m 59s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/95c87fd368cf4897b65bc08bdb1d463a : FAILURE in 28m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a92480b496434a0299c94ff945eb0ee1 : FAILURE in 27m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7b405830fd0c41ec8d1aefea786ab0cc : FAILURE in 27m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a88644334b8045c49a9d16a61ec66cbd : FAILURE in 26m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a502d9ecbb83432c8765b837e2a50fe9 : FAILURE in 29m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8a8691838bd948889325790c69666e43 : FAILURE in 30m 59s","accounts_in_message":[],"_revision_number":14},{"id":"58c042d328d0158b3e45b19a8da8a7dcc9e52aa7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-29 07:43:47.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"aa86acc45999131fccc32d21e429433d24eabbaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 08:51:22.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bb69f3480634fb9ae8a42d7f8656bc6 : SUCCESS in 49m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35c35a9b5ef345bda6435b78006eeaf6 : SUCCESS in 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"991a0dfc4798106149349d24ca92112b3486d9bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-29 10:28:42.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"c646cf888353e7038caef573965dd269fac9c8f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 11:32:37.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22d7937039264f9f878a27bdb8007dd8 : TIMED_OUT in 1h 01m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21899c581ecb4710a1c13924464d5668 : TIMED_OUT in 1h 01m 46s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d810e8b9f713cbc9ee6ad218695f86e29457632e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 12:53:24.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1863396634684c92a3021b299baa07f6 : SUCCESS in 3m 11s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a199cfd001aa49cfaa4aec7b35555d25 : SUCCESS in 48m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f095465ff95943d192cae26d5d7d9ddc : SUCCESS in 2h 17m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d3b555be1de549748b138e7241d8515f : SUCCESS in 2h 09m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/374429767fae45d280b3311a25de250e : SUCCESS in 1h 29m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/70583e8063f84be48b12722e95e1e789 : SUCCESS in 2h 10m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5f56840bcbd943d094bee63d654dce55 : SUCCESS in 26m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c78928f6e1c47c8af1d2e93c7f91e32 : SUCCESS in 44m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cb82d9504ee430aaa34d6892c4b6a70 : SUCCESS in 36m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe69a1c084284da3b5d3427da0efe914 : SUCCESS in 14m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6a463535679453795f7410a1a987334 : SUCCESS in 33m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25229cb075fb436fb5492e9f35961c45 : SUCCESS in 23m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/280b8c68e78b4c5dbaa18a847761a3e5 : SUCCESS in 27m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd69db8912454deb8e10c9e8e2877a99 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/737c4637ed3549009bfa5a658ae93774 : SUCCESS in 12m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d9f2676c3e244c38320e736449ba53e : SUCCESS in 48m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0da02d0f2764bb582f3b98e157c01de : SUCCESS in 1h 42m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1a6279ac2ffc47eda5ff6f3d8762455e : SUCCESS in 1h 08m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e19e2e46b6248fb82d73c5df31951ef : SUCCESS in 1h 06m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/40c5294e63174f988cc9487ed2b95bb3 : SUCCESS in 1h 03m 33s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/65d74ff762ae47fa8580f16f723c71bf : SUCCESS in 1h 09m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a723379c45224e4192a15cd1761fcda2 : SUCCESS in 1h 44m 15s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/2d4765781f614f22a7b5b02fac6cf340 : SUCCESS in 1h 28m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e60256465efe47c6b22399c8e9146dbc : SUCCESS in 58m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/10bf12eed8ff435b9243e29b66796018 : SUCCESS in 1h 04m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/569a770732fe428aa36d8e981c5b9f76 : SUCCESS in 31m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/638253f1df1847eb8a5ef71beb415be3 : SUCCESS in 1h 11m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cc727708d6714913af3e576394613c9b : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":15},{"id":"6bb1e82d77bce4a88b93e6f01c1a1f74c20ee73e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-29 12:57:59.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"4ccdcf4a77b3e8ff2c4b4368462f6d1bb83d8cea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-29 15:16:00.000000000","message":"Patch Set 15:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/45/803045/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"4409052fabb753cc671ef871962f302065f6bc8a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-30 08:59:45.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"9360a8934ff169faa17537a1920dd3b4c68a1e75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 09:00:31.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"c1743292d78c4b495331fc630457cc1a07b63980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 11:17:24.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1cdf66c781af40e4b7ee9133256d365f : SUCCESS in 3m 14s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/29b20d35095e439881b003b4cc074f3e : SUCCESS in 43m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/460d8cae69e2459a8cbe8e747a42e2c2 : SUCCESS in 2h 12m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ba2ea0758c7e4fb697ed749298fcb00d : SUCCESS in 2h 00m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/075afccfc5e54541832aa83820410788 : SUCCESS in 1h 58m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e7551f6a8ad146cf80c2334a8146ee61 : SUCCESS in 1h 44m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abb166393f4a45118ee0a3c450e7d11f : SUCCESS in 31m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f5c76d5920b4a52adb0a9249183b40d : SUCCESS in 14m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/124256c54f1042e6829e6059cfcffcce : SUCCESS in 31m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b17949e2a90946fda642bcce36af36dd : SUCCESS in 27m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0461a684767844c184176f8c0da852f5 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42e6db4aef4c4d80bcd2017d87d962db : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e57dd61efa854c9a91d38fae23a70f8e : SUCCESS in 48m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e514b560d0b4011b3440b082bf4e73d : SUCCESS in 1h 36m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32930f23a75f4dae8600fbac4ccb2428 : SUCCESS in 1h 23m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8676b055096f48ffa40df95a9ea7c3b1 : SUCCESS in 55m 57s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/31231aff4d594edb8d250ada97e228b5 : SUCCESS in 1h 33m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9790d64c7904e8d8c70cb17b543d51b : SUCCESS in 59m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/07a7f795ece3438192a1fcaa85efd05b : SUCCESS in 1h 02m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/07db3ab2ead94b99853cc89304a24f1a : SUCCESS in 33m 27s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2a9e2b82f53b4dd88f7f7f544b8a96fd : SUCCESS in 1h 14m 24s","accounts_in_message":[],"_revision_number":15},{"id":"8589a831a657f7cbd19ba8cc9d96d7be0f2c4864","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 11:17:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"5f8f24477f717a7d7420201b0f3db397106f589f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 11:19:14.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58a0165f594e4de8a3ac594a8f222327 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/378daa3c47d540d9a5d253357402a21f : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"56044db26d97c65c7baec1a16623eff3a1614949","revisions":{"55fcc24b1b832caa7232bf4e7110f4a32d1c5c56":{"kind":"REWORK","_number":1,"created":"2021-07-30 14:38:02.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/1"}}},"commit":{"parents":[{"commit":"8e30639452312019de17f3e1e8364f579752944e","subject":"[QoS] Add rule type packet per second (pps)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e30639452312019de17f3e1e8364f579752944e"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-30 14:29:57.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate. The new rule is accesible through old-style API where\npolicy_id has to be provided in the URI:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nor with new-style API:\n\nDELETE /qos/minimum_packet_rate_rules/$RULE_ID\n\nKeep in mind that new-style doesn\u0027t use \u0027alias_\u0027 prefix, as opposed to\nother QoS rule aliases.\n\nPlacement or dataplane enforcement is not implemented yet.\nNew-style API is not covered with UTs.\n\nPartial-Bug: #1922237\nDepends-On: https://review.opendev.org/802486\nDepends-On: https://review.opendev.org/802487\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55fcc24b1b832caa7232bf4e7110f4a32d1c5c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55fcc24b1b832caa7232bf4e7110f4a32d1c5c56"}]},"branch":"refs/heads/master"},"200c03aec2305ee1410c5aee11efaefb3f3b3d30":{"kind":"REWORK","_number":2,"created":"2021-08-02 09:25:58.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/2"}}},"commit":{"parents":[{"commit":"2356f0eb874b01d4bab696e49e0286de51554f35","subject":"Bump oslo.log to version 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2356f0eb874b01d4bab696e49e0286de51554f35"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-02 09:00:06.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate. The new rule is accesible through old-style API where\npolicy_id has to be provided in the URI:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nor with new-style API:\n\nDELETE /qos/minimum_packet_rate_rules/$RULE_ID\n\nKeep in mind that new-style doesn\u0027t use \u0027alias_\u0027 prefix, as opposed to\nother QoS rule aliases.\n\nPlacement or dataplane enforcement is not implemented yet.\nNew-style API is not covered with UTs.\n\nPartial-Bug: #1922237\nDepends-On: https://review.opendev.org/802486\nDepends-On: https://review.opendev.org/802487\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/200c03aec2305ee1410c5aee11efaefb3f3b3d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/200c03aec2305ee1410c5aee11efaefb3f3b3d30"}]},"branch":"refs/heads/master"},"435a40acdc1314f05203ac58c7fca8774263e1d7":{"kind":"REWORK","_number":3,"created":"2021-08-03 12:49:52.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/3"}}},"commit":{"parents":[{"commit":"2356f0eb874b01d4bab696e49e0286de51554f35","subject":"Bump oslo.log to version 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2356f0eb874b01d4bab696e49e0286de51554f35"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-03 10:47:24.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nDepends-On: https://review.opendev.org/802486\nDepends-On: https://review.opendev.org/802487\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/435a40acdc1314f05203ac58c7fca8774263e1d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/435a40acdc1314f05203ac58c7fca8774263e1d7"}]},"branch":"refs/heads/master"},"d1c4abc776dd17ac52694676a32a2f0c2d34a364":{"kind":"REWORK","_number":4,"created":"2021-08-05 07:42:07.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/4"}}},"commit":{"parents":[{"commit":"2356f0eb874b01d4bab696e49e0286de51554f35","subject":"Bump oslo.log to version 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2356f0eb874b01d4bab696e49e0286de51554f35"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-04 08:50:23.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nDepends-On: https://review.opendev.org/802486\nDepends-On: https://review.opendev.org/802487\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1c4abc776dd17ac52694676a32a2f0c2d34a364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1c4abc776dd17ac52694676a32a2f0c2d34a364"}]},"branch":"refs/heads/master"},"b8d6ea39591ee6adacdb6f5af46e1c400612de32":{"kind":"REWORK","_number":5,"created":"2021-08-10 11:20:32.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/5"}}},"commit":{"parents":[{"commit":"2356f0eb874b01d4bab696e49e0286de51554f35","subject":"Bump oslo.log to version 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2356f0eb874b01d4bab696e49e0286de51554f35"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-10 10:19:35.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nDepends-On: https://review.opendev.org/802486\nDepends-On: https://review.opendev.org/802487\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8d6ea39591ee6adacdb6f5af46e1c400612de32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8d6ea39591ee6adacdb6f5af46e1c400612de32"}]},"branch":"refs/heads/master"},"02c1cbb4db29efce844a209286e521dd0009d992":{"kind":"REWORK","_number":6,"created":"2021-08-13 14:39:36.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/6"}}},"commit":{"parents":[{"commit":"2356f0eb874b01d4bab696e49e0286de51554f35","subject":"Bump oslo.log to version 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2356f0eb874b01d4bab696e49e0286de51554f35"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-13 14:37:52.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nDepends-On: https://review.opendev.org/802486\nDepends-On: https://review.opendev.org/802487\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c1cbb4db29efce844a209286e521dd0009d992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02c1cbb4db29efce844a209286e521dd0009d992"}]},"branch":"refs/heads/master"},"2f00734a40b43f8bc90c68ca11d33771eaf0cfec":{"kind":"REWORK","_number":7,"created":"2021-08-16 07:05:51.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/7"}}},"commit":{"parents":[{"commit":"2356f0eb874b01d4bab696e49e0286de51554f35","subject":"Bump oslo.log to version 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2356f0eb874b01d4bab696e49e0286de51554f35"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-16 07:04:37.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nDepends-On: https://review.opendev.org/802486\nDepends-On: https://review.opendev.org/802487\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f00734a40b43f8bc90c68ca11d33771eaf0cfec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f00734a40b43f8bc90c68ca11d33771eaf0cfec"}]},"branch":"refs/heads/master"},"fa330aa01990a6739b09312260a26f9a9e9bf672":{"kind":"REWORK","_number":8,"created":"2021-08-20 14:17:42.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/8"}}},"commit":{"parents":[{"commit":"2356f0eb874b01d4bab696e49e0286de51554f35","subject":"Bump oslo.log to version 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2356f0eb874b01d4bab696e49e0286de51554f35"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-20 11:47:31.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nDepends-On: https://review.opendev.org/802486\nDepends-On: https://review.opendev.org/802487\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa330aa01990a6739b09312260a26f9a9e9bf672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa330aa01990a6739b09312260a26f9a9e9bf672"}]},"branch":"refs/heads/master"},"0b6d55b38f5f3a3a4c45ae693e37ef0365c7467e":{"kind":"REWORK","_number":9,"created":"2021-08-23 13:04:18.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/9"}}},"commit":{"parents":[{"commit":"84d9bb1e0e5cde34acd9f4ee7f54baf9c89c7d81","subject":"Merge \"functional: Add debug messages around OVN services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84d9bb1e0e5cde34acd9f4ee7f54baf9c89c7d81"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-23 10:39:27.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b6d55b38f5f3a3a4c45ae693e37ef0365c7467e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b6d55b38f5f3a3a4c45ae693e37ef0365c7467e"}]},"branch":"refs/heads/master"},"5807ddd7672e493abac399ebd7ebb321b72261a8":{"kind":"REWORK","_number":10,"created":"2021-08-25 14:47:12.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/10"}}},"commit":{"parents":[{"commit":"e4e206ed7bd29ffb7852e8ea637e4d985d1dce65","subject":"Merge \"Remove shim in trunk/rpc/backend.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4e206ed7bd29ffb7852e8ea637e4d985d1dce65"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-25 14:44:40.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5807ddd7672e493abac399ebd7ebb321b72261a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5807ddd7672e493abac399ebd7ebb321b72261a8"}]},"branch":"refs/heads/master"},"e76efb014ce74fd4e7c3836a136a9782df57d2e7":{"kind":"REWORK","_number":11,"created":"2021-08-31 09:58:34.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/11"}}},"commit":{"parents":[{"commit":"a2ffbfa5523dbe04f64d55a49e46f28897704fd2","subject":"Merge \"Handle all portbinding attrs in case of bulk port creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2ffbfa5523dbe04f64d55a49e46f28897704fd2"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-31 08:56:50.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e76efb014ce74fd4e7c3836a136a9782df57d2e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e76efb014ce74fd4e7c3836a136a9782df57d2e7"}]},"branch":"refs/heads/master"},"38125c81300fb95ad1b7d52e3bb1096ab93d46e9":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-09-16 13:58:29.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/12"}}},"commit":{"parents":[{"commit":"a76e1d642cfeef990364e95f55988b9416d500b7","subject":"Merge \"[OVS][FW] Initialize ConjIdMap._max_id depending on the current OFs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a76e1d642cfeef990364e95f55988b9416d500b7"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-16 08:42:47.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38125c81300fb95ad1b7d52e3bb1096ab93d46e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38125c81300fb95ad1b7d52e3bb1096ab93d46e9"}]},"branch":"refs/heads/master"},"abf8e7f171cd8195c44330364921b09368e4876c":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-09-22 06:56:07.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/13"}}},"commit":{"parents":[{"commit":"8791947a6ba581bb40de9e54d20e281ba3ae6019","subject":"Merge \"Rollback db session in case of error during releasing quota reservation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8791947a6ba581bb40de9e54d20e281ba3ae6019"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-22 06:54:58.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abf8e7f171cd8195c44330364921b09368e4876c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abf8e7f171cd8195c44330364921b09368e4876c"}]},"branch":"refs/heads/master"},"aaae8662c5971622f2980a8974f854a2407ca959":{"kind":"REWORK","_number":14,"created":"2021-09-28 12:40:58.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/14"}}},"commit":{"parents":[{"commit":"1d450dbddc8c3d34948ab3d9a8346dd491d9cc7c","subject":"Merge \"Revert \"update subport status when trunk/subport create/delete is triggerred\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d450dbddc8c3d34948ab3d9a8346dd491d9cc7c"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-27 12:50:00.000000000","tz":120},"subject":"Add API extenstion for QoS minimum pps rule","message":"Add API extenstion for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aaae8662c5971622f2980a8974f854a2407ca959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aaae8662c5971622f2980a8974f854a2407ca959"}]},"branch":"refs/heads/master"},"56044db26d97c65c7baec1a16623eff3a1614949":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2021-09-29 10:28:42.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/45/803045/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/803045/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/803045/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/803045/15"}}},"commit":{"parents":[{"commit":"1d450dbddc8c3d34948ab3d9a8346dd491d9cc7c","subject":"Merge \"Revert \"update subport status when trunk/subport create/delete is triggerred\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d450dbddc8c3d34948ab3d9a8346dd491d9cc7c"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-07-12 07:02:02.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-29 10:27:30.000000000","tz":120},"subject":"Add API extension for QoS minimum pps rule","message":"Add API extension for QoS minimum pps rule\n\nThis patch implements support for CRUD operations for QoS minimum\npacket rate, for example:\n\nDELETE /qos/policies/$POLICY_ID/minimum_packet_rate_rules/$RULE_ID\n\nPlacement or dataplane enforcement is not implemented yet.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie994bdab62bab33737f25287e568519c782dea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56044db26d97c65c7baec1a16623eff3a1614949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56044db26d97c65c7baec1a16623eff3a1614949"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
