)]}'
{"id":"openstack%2Fneutron~803579","triplet_id":"openstack%2Fneutron~stable%2Fstein~I4bcba0ae82d99fac962d758b48b1727f344ec7bb","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1933234-stable/stein","hashtags":[],"change_id":"I4bcba0ae82d99fac962d758b48b1727f344ec7bb","subject":"Add extra logs to the network update callback in L3 agent","status":"MERGED","created":"2021-08-05 10:41:24.000000000","updated":"2021-09-24 13:05:36.000000000","submitted":"2021-09-24 13:02:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803579-bug/1933234-stable/stein","meta_rev_id":"5cb0a63231e098c06e183df5539dbc4939eaf28e","_number":803579,"virtual_id_number":803579,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-24 13:02:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-09-24 11:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-08-23 10:45:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-09-24 11:10:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2021-08-31 14:16:18.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-05 12:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-05 19:29:28.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-23 10:45:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-09-24 11:10:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"3051eeb195b473be134f2cad2a45c78702790fa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-05 10:41:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abf40a20953263f51e85d7f76a18b8c28cffd5b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 12:50:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/88abeea0f3e84e89b2869f52d4cd1256 : SUCCESS in 1h 43m 37s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/4ddbdb816dfc48789cde8243a5778c6b : SUCCESS in 1h 36m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/6958520c83ff4519993fc4da259a9f21 : SUCCESS in 1h 59m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/879ffd85de814beeb5679a79c4791bf7 : SUCCESS in 32m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd084263f2634b65a9cb0b84e4bb806b : SUCCESS in 12m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62f4b303f61a4184bc1bf304e0323581 : SUCCESS in 22m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/57a82aab2474466e8b5caa9d1feeb0ab : SUCCESS in 24m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5edd2af2f1744870b18c691980f72eae : SUCCESS in 21m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c41956329b2441cb6556511e3803c3a : SUCCESS in 25m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417bfd645d2d47249b54817365ba34fe : SUCCESS in 6m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca40302a4dd04650a7db420f16ddbb01 : SUCCESS in 1h 29m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e1246acfab2e4f62a55bfb83dc83e4fa : SUCCESS in 59m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d10a1a609034fdcb27b177d137dd2dd : SUCCESS in 1h 20m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/592cf06871aa4ab7be8c127212413b93 : SUCCESS in 35m 00s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/6c8b4418cdaa49748cdf53f90336a1fb : SUCCESS in 36m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8af767803f8b4d76819df788b9e19bc7 : SUCCESS in 1h 27m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/26de4af68142425e9aa1dc650b218380 : SUCCESS in 1h 36m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bd3d3430c88c493f8dcd5e8d6832af13 : SUCCESS in 1h 19m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/15aae20f42894a4ab27b8101e7ba2b25 : SUCCESS in 1h 16m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/136bf81b85e049e592bfcaf66ee79f94 : SUCCESS in 1h 11m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/641a786e346f4b408ba7d0a8686a8f2a : SUCCESS in 1h 03m 24s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cac99c1ea4074c25a8637c5afed01a11 : SUCCESS in 1h 12m 40s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c6c0a0942b694040aeb4f8ac6ac02def : SUCCESS in 1h 42m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/af02034a684b4997a263a081af6412b5 : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":1},{"id":"678e28a452dbf8aa0f73960adaf400a53ed38393","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-05 19:29:28.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/79/803579/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/79/803579/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"85b8b5811a54b74daa7512b4cf7b4b505f8424c1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-08-23 10:45:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b95c4124cef309252edd12bf8a0790dfcd9f8279","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-08-23 10:45:47.000000000","message":"Removed Workflow+1 by Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"3f15194404224e7e3080c09d7d77f2f28af07b3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:46:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65cf36caa9d2739374f1cf8839d6713d6cf27b7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 13:47:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3f39c7ba4b94dbbb241fa2a8dc1d083 : SUCCESS in 13m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8416dbd980b44fadad25a006d2a83610 : SUCCESS in 28m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d36888d6d64a474398b618d654998918 : SUCCESS in 32m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/390ebe603f0c4b6d9cf23ef6b68553bd : SUCCESS in 26m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c22d512a28c41b88f7719401f0c2e1b : SUCCESS in 29m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81638545c155410db9bb497d6d8c1ae5 : SUCCESS in 7m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d023af1898f74eb891c951bd8bef85cd : SUCCESS in 1h 41m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/03a580b2677a46c29facbaa786f2ab94 : SUCCESS in 59m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec51375a9c4e4c8faebf5630a21832b3 : SUCCESS in 1h 27m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4ee49e9cbb0044ada1d3c92425bf2fb7 : SUCCESS in 45m 15s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/2e35a57dacb04f399867b1dc2862780d : SUCCESS in 46m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2f01a7b150894927a97488a11ceacebd : SUCCESS in 1h 30m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7b4a0ea87cde4271ba95f6549154262b : SUCCESS in 1h 22m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/46e54198e94e4bd7b24f056bfc855068 : SUCCESS in 1h 40m 40s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1f11092843a45178dae29f3e83fad36 : SUCCESS in 1h 18m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79e22f19e0704da78c04fe6235d5bca0 : FAILURE in 54m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fdaea6e8037a4510bd92accd9e0fd3c2 : FAILURE in 54m 10s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bba837865dfa4b2ba85e5168d1002472 : SUCCESS in 1h 54m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aceb260abf75444eac1e027273d2616e : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":1},{"id":"343fe5949c553069cf5f679ac74ac5ca42846a6d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-08-24 12:27:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c82ccd59adac61f756fb2c86944d53d795165d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 14:13:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/a6c5f3c7a20b47d7842560d86d3a08d2 : FAILURE in 35m 44s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/8eae452bb76d4d569e2d86782ee95f44 : FAILURE in 31m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/8618117497774d74b71018c94eef94a5 : FAILURE in 29m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06e8efa349de40e5a6634020fe8b0a39 : SUCCESS in 26m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fcaeb07bd0b4bc88936c33c6ade3b6e : SUCCESS in 13m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97325a0b34244e7698acaa31b7419a8e : SUCCESS in 21m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/149194bce99f497b988ff5cafd90cbd2 : SUCCESS in 28m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbd91e61a9464ec4938e0b92ca3a15d1 : SUCCESS in 21m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cca4fc8a4e44450a85abf8688ba5aacf : SUCCESS in 27m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93bf1a628db64388890bb55a5bf6d613 : SUCCESS in 6m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/64649e25dff943158df8d94bdf5593fc : SUCCESS in 1h 39m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a36952c9a16f48d899b9d4e0ca1cb6c8 : SUCCESS in 55m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2fe209796c44fb497a0cc120e7d8ba9 : SUCCESS in 1h 30m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a8a9f386a4da4918aa8069b8702ddd12 : SUCCESS in 43m 52s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/df38832c374145938fa15c3976b9d827 : SUCCESS in 45m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c16f48cac5974c7e8e8b23b1effadfb5 : SUCCESS in 1h 19m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c0a46e97ea3140bd98871169f3aa0f1f : SUCCESS in 1h 32m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/77de504cca9a4c7dbae1ca3f5cc426a1 : SUCCESS in 1h 21m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/61f705b58318459eb49fc14cf1fd4320 : SUCCESS in 1h 15m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f2c7732fda9485ba9a53cc1081aa9a1 : SUCCESS in 1h 03m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ca4bf524547a4b2a8815a81df5662cb2 : SUCCESS in 1h 11m 07s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/537e94f0b93c48b581b34cadfc12664f : SUCCESS in 1h 21m 55s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/114eaea89db244c1937a3931ef7e6ed7 : SUCCESS in 1h 38m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9291dce96e5449da93de5a3a596faaf0 : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3cb0ef61fb6bcab6b7f4ae7a3a64d3defe19b52a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-24 18:43:24.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/79/803579/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/79/803579/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"4b4dbd5c552b9187e316e19890a34ed3704572b2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-08-31 14:16:18.000000000","message":"Patch Set 1:\n\nrecheck https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/806119 merged","accounts_in_message":[],"_revision_number":1},{"id":"566878291d6f3e1f21b01c095a26b1fbc9477ff8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-31 17:24:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/79/803579/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/79/803579/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3591ef4bcc854c86dcb7488c557b3a898807e938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 18:25:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/7185ef7a7cae41cd9fbb8766d2013c18 : SUCCESS in 1h 14m 13s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/08af1073aeca4e11b5ed9ec489bf12ab : SUCCESS in 1h 35m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/d412a16d9f0c4b17a10ed8ee578abde5 : SUCCESS in 1h 43m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54aea59a261c40659c9a941a748b2ecc : SUCCESS in 28m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a12d4055364c779880490bbf5dd8b1 : SUCCESS in 16m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3891f56899d04c5cba4c6f8f352ceb39 : SUCCESS in 28m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7440a297718d4659bdfb898359f914ad : SUCCESS in 24m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96e772268936431b803437e3f6e4e83c : SUCCESS in 25m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8a43eea507f472d9ba163e04e770b4a : SUCCESS in 24m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1436785c6914bce857483cb4f42ee5a : SUCCESS in 7m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1f9f73b133674a4e94418c89b0e20c02 : SUCCESS in 1h 33m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d04186ec5e8b47e7a30cafe8b85db987 : SUCCESS in 54m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f4d0519b0c64dd1a5fb6907f41bc941 : SUCCESS in 1h 22m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6f6404a1bf884d0b872b884768bccc16 : SUCCESS in 43m 08s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/ed35a9f45dad4c30907a5a6208ba64a8 : SUCCESS in 48m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c17b8493605445bcb01e5af0f08d61bd : SUCCESS in 1h 22m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/68374fd402d54caeb5e82820df76570c : SUCCESS in 1h 45m 36s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/33e7312d9ec64a478ec2cfe586f38f89 : SUCCESS in 1h 27m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e96f57936dbc420881644be4bffebe00 : SUCCESS in 1h 22m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/10ca52cf4cf94dd5a23bdcfa880e75f8 : SUCCESS in 1h 03m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3e120e6559894220807108c5832aa04a : SUCCESS in 1h 10m 21s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fc24d6f8f66f4fa7aabf937ad23b880d : SUCCESS in 1h 18m 44s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85c1ec1c4abf4c198ee356bae75bb77a : SUCCESS in 1h 39m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/01ed26d57ee54010bf36220be393f0a6 : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0f0c60bf7b8e51e41e6ad83e3c1392a60f9c538f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-24 11:10:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b11c45ca1ec5fa26d98ae463efa8a8a38d91094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 11:11:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29228bcff22d323db5117ecb92bce11a7bfe8e5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 13:02:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/292f0fb9a1684295923d294feaabec86 : SUCCESS in 12m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af6e13cd93b74904bb0806611da026f2 : SUCCESS in 22m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/22d712b233ca47e29ecd28b7448ac146 : SUCCESS in 22m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7255d0b0698a4d7a8c796fddf8bd2c89 : SUCCESS in 23m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f90805eec16401e845b40d68802b99c : SUCCESS in 21m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abc0d9d60184779a4f757a32b77d4f6 : SUCCESS in 6m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/08fc9aaa1d8f42ffa8e6e2e9cd28d381 : SUCCESS in 1h 30m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/714e88eebba04e7e92d542b135c3f1ce : SUCCESS in 56m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21dddb43769e4d82b6c1168088f9daae : SUCCESS in 1h 19m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/69204602f08d4010852c976688640243 : SUCCESS in 37m 48s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/bd38e1c4da3e4691891de6f18baf7db4 : SUCCESS in 37m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b4999edc3df14d6fba633e4e85042e79 : SUCCESS in 1h 17m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/75c97a59ccf744ea99b485e5089be5a4 : SUCCESS in 1h 24m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/04c186397c454aa383420f46f8e77065 : SUCCESS in 1h 17m 21s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ab35d16d2404c7cadad45065ad7dbd7 : SUCCESS in 1h 29m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04bfa192cae74b97aa43a4cff85c012c : SUCCESS in 1h 34m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/853340afe9564d9fa4eadeab1622b2a1 : SUCCESS in 1h 14m 25s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b746a73fc06d45d4a649dae0e058a7ff : SUCCESS in 1h 44m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9d9456215e8146e6b16d8a1d7cd9e4b2 : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ecf1f43bd43484fbbb2b0d88ab52f0cf6509de7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 13:02:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5cb0a63231e098c06e183df5539dbc4939eaf28e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 13:05:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3809d5b2c2424923be45803912cbad55 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7aab77d2084533bac7db7bf27eec328192310b29","revisions":{"7aab77d2084533bac7db7bf27eec328192310b29":{"kind":"REWORK","_number":1,"created":"2021-08-05 10:41:24.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/79/803579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/803579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/803579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/803579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/803579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/803579/1"}}},"commit":{"parents":[{"commit":"7f63139424adcbbc18cdbf1b0fa3efa3b0fd3f45","subject":"Copy existing IPv6 leases to generated lease file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f63139424adcbbc18cdbf1b0fa3efa3b0fd3f45"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-29 11:39:22.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-08-05 10:41:01.000000000","tz":120},"subject":"Add extra logs to the network update callback in L3 agent","message":"Add extra logs to the network update callback in L3 agent\n\nIt may be useful during debugging some L3 and events related issues.\n\nRelated-bug: #1933234\nChange-Id: I4bcba0ae82d99fac962d758b48b1727f344ec7bb\n(cherry picked from commit 5c9a7fe1b400c3549ee1cdd8ffd53895d6db1e86)\n(cherry picked from commit 25f4864d1206bd087ac98b4806111701bc96cf77)\n(cherry picked from commit 7de9aa109a343c83fe6ae0dbfd9fb7e38a99cf31)\n(cherry picked from commit 5dfd2e561556fc187d28e65039898d472e50b61d)\n(cherry picked from commit 85c33d4ec9fa81df2d058eacbf6bb1a37039358f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7aab77d2084533bac7db7bf27eec328192310b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7aab77d2084533bac7db7bf27eec328192310b29"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
