)]}'
{"id":"openstack%2Fneutron~803584","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I2efe66a7415f7a18fb85bd2536a1901e751d6203","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1933234-stable/queens","hashtags":[],"change_id":"I2efe66a7415f7a18fb85bd2536a1901e751d6203","subject":"[L3] Use processing queue for network update events","status":"MERGED","created":"2021-08-05 10:51:34.000000000","updated":"2021-09-27 17:43:09.000000000","submitted":"2021-09-27 17:43:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803584-bug/1933234-stable/queens","meta_rev_id":"562e69097e1bd7b99b0b9903d74958c4bfdb3d15","_number":803584,"virtual_id_number":803584,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-27 17:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-09-27 14:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-09-25 21:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-09-27 14:34:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-05 13:48:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 11:00:48.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-09-24 11:12:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"849c37b7e31aa81d858f57934bb7038db9af8b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-05 10:51:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb6cb7987b7bc66c42e1740750f65f415572c073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 13:48:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/963cca40fcea485aab6f7eb5c8fa90f3 : FAILURE in 48m 07s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/3685b4567b4a4f8595836e8644b6fe82 : TIMED_OUT in 2h 51m 27s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/adcfd1163a55426f8e3dfe8c0d2296d9 : TIMED_OUT in 2h 51m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5459cc73cec54139bc8b6202464d8cd3 : SUCCESS in 12m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/515998b10dfc4b268ae3f0926ddf0aab : FAILURE in 22m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eeb3b6bcbbdb4a5e9a26720344fa54b1 : FAILURE in 33m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f81aaa0f0ef24ad699acfb655c6d692a : SUCCESS in 7m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/920535cccd90419992050a3ab71cb535 : FAILURE in 1h 23m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/747de7765ee445e9ad7ef51a8d5d04fc : FAILURE in 56m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8e7253891de4aabb0f416c03a4274dc : FAILURE in 30m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cb3494995acc46feafb8fa18cd38209e : FAILURE in 43m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/816dd0a6cf254229b6b93d57b0c2c9af : FAILURE in 1h 26m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2242ff64873c40fbb9d6d3d1306eccbd : FAILURE in 1h 18m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaee271ed66a4c15a12814aec392d65e : FAILURE in 33m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3c538fa118c67db55050cc9c4afe6131a8f4d723","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-08-23 11:00:48.000000000","message":"Patch Set 1: Code-Review-1\n\nhttps://zuul.opendev.org/t/openstack/build/eeb3b6bcbbdb4a5e9a26720344fa54b1\nAttributeError: \u0027ResourceUpdate\u0027 object has no attribute \u0027update_id\u0027","accounts_in_message":[],"_revision_number":1},{"id":"ac3081404027ab16d18d45ca86cfdde1c4162f19","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-08-31 14:17:35.000000000","message":"Patch Set 1:\n\nrecheck https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/806119 merged","accounts_in_message":[],"_revision_number":1},{"id":"c8b5259f578206e566fd03eba12fab5ea8d88af0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 19:38:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/9117908ba4d547c5953c80200a1275aa : FAILURE in 49m 07s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/e2c7f45b15ac47f5b3a3ad70f7deb334 : TIMED_OUT in 2h 52m 09s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/940bfeb68ac24230b39aa0aaf4da07cf : TIMED_OUT in 2h 51m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96cad257d75f4bc1bc41f4b3e4f8624e : SUCCESS in 12m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/673e8f111c5a4fc78b8069a474183356 : FAILURE in 24m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/39241be9a9b046a1a94fc8025efed716 : FAILURE in 29m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/025c839969dd4f0db79ae3b563ef86b2 : SUCCESS in 8m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/806b7e20c7934d4f8d4dca68782a0c68 : FAILURE in 1h 48m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/03e0ce98d7d04fcbab5327d665b10ce2 : FAILURE in 1h 08m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bfa1ccc14144183bfe3b89094b92ef8 : FAILURE in 1h 28m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ec0d66e50586479498aa14c021293c12 : FAILURE in 40m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/899c05f5a9b0492daeb274e82bc049d4 : FAILURE in 1h 36m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1075948a85db474985b98d76b76e4317 : FAILURE in 1h 26m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69a6366e3667437ea60cde4dae0d1132 : FAILURE in 29m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c5fc26e775fec67f085cf3c2b297f85f8292f7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-07 14:00:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"493edf62fafba70c0d6bb329e590c3c52fc676b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 15:55:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/8b4007733ee64f42a66face9bf28ea34 : SUCCESS in 45m 12s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/6c17a683eedd4be8af2ad5a5dd0e9c2f : FAILURE in 59m 51s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/2c94a6e6df02416998a0e4ec29b77acd : SUCCESS in 58m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8c9a111462d4172bb3b3ea9f2af9884 : SUCCESS in 14m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1efddcd2f6254f55978bcab45a1888a9 : FAILURE in 21m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d4be4c14275147f09de0ea9251647656 : FAILURE in 24m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a2f2ea2b73854b0888383d8a3284c889 : SUCCESS in 7m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/54178de0c4a844f99a64a608af021379 : SUCCESS in 1h 49m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/44d3944734af4039bf7d08b91fc09cdb : SUCCESS in 52m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6528119266f7446ca95fa0a51f5c897e : SUCCESS in 1h 15m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e387b7def0c44b6addb68a35ce9b74c : FAILURE in 39m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a15e4c50295844bebb0f62fb44abad07 : FAILURE in 1h 30m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b4df56dd174449eab6b90e79733c619c : SUCCESS in 1h 15m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d47413be84e45979d33cd502a45ea83 : FAILURE in 32m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e1b155d18239208fe87e3946b42e8fbad9f8aa10","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-23 15:18:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bc2a8c6e1bbed53f90ba4c0c40b20ca4aa7dedca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 16:56:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/8772961ba3a44cd6b4ca687f3e28a852 : SUCCESS in 58m 30s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/5bd4a64c08a84fdf9678d966664bd342 : SUCCESS in 1h 00m 04s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/ef1f79dc6d64402ba31e5d58ae00b21f : FAILURE in 1h 08m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aed323b57444064838e1c8034952da7 : SUCCESS in 12m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c0e6011f3654edfbd95c0dbf40a63e0 : FAILURE in 24m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f4e794506fb4453fa2fccd3efcfe3f80 : FAILURE in 26m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7cce5dc7161f4347bad853fa98ef3036 : SUCCESS in 9m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2114325ae68d459691506b6e575d5eef : SUCCESS in 1h 21m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e0fdc5e827ea47f0ae45e8c61be450dc : SUCCESS in 58m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa98f860b3384bd49077826ebed82dfe : SUCCESS in 1h 17m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/099338ba294c4c37a8289cd8de561dae : FAILURE in 48m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4355f60f2fa442bc911900e7963def45 : FAILURE in 1h 17m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a2fe09aedef14b58bb325de6b053e3cc : SUCCESS in 1h 13m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a20fe5717c65446b98ab628682d3bf40 : FAILURE in 27m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d204f2e4a7ce5b031e289ceea23b1d6497e70c2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-24 10:07:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"04a1621afdb6c95f8f1a277b70e3926f53ac6b00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 12:25:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/d57036e581bc47849fa931afdd86566d : SUCCESS in 49m 45s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/6163a52067e34a8f883a2c853383fd45 : SUCCESS in 58m 02s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/6186c28751814764a942e594d1b999e9 : FAILURE in 1h 07m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35453609a6e4ec7baf8f07db7c775f7 : SUCCESS in 13m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/453689d6e92e41a8b497f98b0ecf3280 : SUCCESS in 28m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ba6fd70d162d45cd8c5eeccd1115d06d : SUCCESS in 27m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/be882c6a1a5d4474a1f1ad0bdb90328e : SUCCESS in 8m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d7114676a31841e1a26297b3adc65395 : SUCCESS in 1h 22m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/41bab92ca17a4208b563d82360be0f21 : SUCCESS in 54m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf468dabbc8f47b7b5d9be393bf9a20b : SUCCESS in 1h 13m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/adef2203bc4444f5a1926ae193cb52e0 : SUCCESS in 44m 45s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1fa021fc674542359a44be24aea9e8f4 : SUCCESS in 1h 19m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1f0f832000a44aa7b2d26d734368082a : SUCCESS in 2h 06m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/148f1d7fbf534dde82c33f7d80e98796 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":3},{"id":"b30aabcdb4648a090631c8615a7ddf64b6d9232a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-25 21:45:52.000000000","message":"Patch Set 3:\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":3},{"id":"6481f02446e899c30a58164b090748f93e2db43d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-25 21:46:01.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good, UPDATE_NETWORK in other place in queens","accounts_in_message":[],"_revision_number":3},{"id":"7b3089463c83d5617dc44c0b82466565bd84233e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-25 23:23:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/28399f53335048dca41fc247eca02e62 : SUCCESS in 41m 27s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/13e0e8ea63aa40128d22313cc59aea5d : SUCCESS in 54m 54s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/70be5562e59247d089bcfbd52495c974 : FAILURE in 1h 07m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeb341cc14a04087905ebe4f858e0728 : SUCCESS in 10m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c5d2c757013e4d8ea964532fdadf3dff : SUCCESS in 19m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/66b9b1e5643a4e07b8d344fd302a2663 : SUCCESS in 21m 05s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7ed973ce05b646e192ed6f4566a881a0 : SUCCESS in 6m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/41fc495f37ac4e989b4382c94b12582b : SUCCESS in 1h 15m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0fd49c6d70f64df8a782f5d8e87812a7 : SUCCESS in 56m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4863670f271498c9976bb57696db935 : SUCCESS in 1h 09m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0567de096df14974b1039d7f47302dd3 : SUCCESS in 42m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c701213924d8423ba8d47513dbf0416d : SUCCESS in 1h 31m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/79b05b1db0274b569a6bc884e9fca7d0 : SUCCESS in 1h 19m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc4b926e75e448409f2215dc205a4f52 : TIMED_OUT in 51m 14s","accounts_in_message":[],"_revision_number":3},{"id":"35ff90ff4b00da84ca2a49c179996fa8718d01da","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-26 20:15:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0366d88b511e1f57039d7cb4b80a13da937c12da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 21:49:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/f7823c69d647419e875b591a931c72ca : SUCCESS in 52m 26s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/eb1f946056dd44fc85a35e61c059f187 : SUCCESS in 58m 39s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/6d816340c2f74759a23dca2eeb8fe4c9 : FAILURE in 1h 04m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d54e81fb092443abefda83bebaf7843 : SUCCESS in 13m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3ee80715a1d4340b89c33139f58a73a : SUCCESS in 25m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/72b4bbde42894aff9c4a4ac5d6fb3e74 : SUCCESS in 25m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/62350a9c9f02483e92b1c12610e5bfc3 : SUCCESS in 6m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d373c7fb420e4f24befae6a7681451be : SUCCESS in 1h 15m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6b0cb574c9f549139c48045c6898f4a0 : SUCCESS in 44m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ccfc7fa3c364597a9eb2939baca79ac : SUCCESS in 1h 27m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c1a5704c6507480087b900dfcb7e969f : SUCCESS in 38m 16s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2dab7d33119e4eb5b5ebf96917b92754 : SUCCESS in 1h 28m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f506698f0f424b73bde5f106c30c074d : SUCCESS in 1h 20m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d68554fc9b854c59a45fa0d5861dbee7 : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":3},{"id":"a2c93f30cc327a3f7c5ea7fb8a2eaeb63b908427","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-26 22:19:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7df55a3291b483e881a076a845d781b2d56c01e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 23:54:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/5daf2a6508ff452e9f47e92f89144597 : SUCCESS in 52m 36s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/6efb79b7aaf94a6f9cb85eaeac2a8da1 : SUCCESS in 59m 47s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/6f52b59f6eee4446800797f268292ff7 : FAILURE in 56m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38caa22b6c904d6599f02916896c1b3a : SUCCESS in 11m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19ee24410cb2414bb44fb823e9f18567 : SUCCESS in 24m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ebdc65a8f8f44d7fbe422ced816f5eb4 : SUCCESS in 27m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2fec2d4dded44171a89169d451668504 : SUCCESS in 6m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7d54bac17a29452aa8f5c92da130d425 : SUCCESS in 1h 23m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3e90479cedfa4201928f53cc1a865b22 : SUCCESS in 58m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91cf9ac305964d5ea5c3de320a8917f1 : SUCCESS in 1h 16m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3808b6c8ce654c5fb8962213fd47fa42 : SUCCESS in 45m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ae91c3bdc2c34a909f24d78152ea740a : SUCCESS in 1h 29m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/aa2541037fc84c00a0db8dfe10c17b2d : SUCCESS in 1h 15m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87edf913b09749aaba5968eefaacd7fe : SUCCESS in 31m 55s","accounts_in_message":[],"_revision_number":3},{"id":"4741d24fc37e66cc4334b08905b95670ce071241","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-27 06:22:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f5148aaa41c5529529bb8c2b28109fb974566053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 08:15:02.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/a7d37ea6249b4f6d9579a40815f37a25 : SUCCESS in 46m 47s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/8b28472df556414daa7e8b37b74bd93b : SUCCESS in 1h 05m 15s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/731f73738326469c824d495cbd56acfd : FAILURE in 1h 05m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa404168ed9f4319a51acc671e1882f3 : SUCCESS in 12m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64265d327c8d4a2e93e83dd92597eadf : SUCCESS in 26m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f268c7fedc2348ceb9d645393b5a84e3 : SUCCESS in 25m 34s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a190e06609644f89a354091a8a76e84b : SUCCESS in 8m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a82bcd952ef547339f69e895e9ef93b3 : SUCCESS in 1h 13m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/75e7c41e38b441c0ada2cd0c04ff51e3 : SUCCESS in 49m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97e1717850394aaebfcdd2a9629f6b53 : SUCCESS in 1h 23m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e56f2c21788341b899601be57080bcd7 : SUCCESS in 42m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0196bc4c0dfe4a219ad47ced16508fb0 : SUCCESS in 1h 20m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4a0f1cf18cff4fffa2f85b2d49d28d26 : SUCCESS in 1h 13m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4ecaebbe9ee41ad856b12c1b34f266c : SUCCESS in 30m 02s","accounts_in_message":[],"_revision_number":3},{"id":"cd198487b7d2a415b7cde2d0d958042058237a06","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-27 11:04:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"21e5c9a10121df4d3258143939ae43643eaefcd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 12:43:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/07fb5c86fe2a45ec8167b268db95ae73 : SUCCESS in 45m 26s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/e3d62c646fe04f87963ed247bb3b10df : SUCCESS in 58m 03s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/3ed880b905ec474886f4466974399743 : SUCCESS in 47m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0b1359c94bd430ab92332048916ec96 : SUCCESS in 14m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87b226a03b01427ebf28e3c996542930 : SUCCESS in 21m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a77bf2477cd04f1682cb51dca786a8c2 : SUCCESS in 27m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ce1ce77d1af44a5bad0ceb06276d43eb : SUCCESS in 8m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e37dea5e939e4a47ad91618f7965a15a : SUCCESS in 1h 16m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5ea480e2390540e49078148f54cd0254 : SUCCESS in 47m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d7286d8e1964d4b85d413b68bbd8bf3 : SUCCESS in 1h 22m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c7b3030711524f3ea503958ac81c34a0 : SUCCESS in 39m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/69ea7522a5ea4e69914bb4d4b75178eb : FAILURE in 1h 23m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/04a267f86b6344ca8e5fbc358ec273bc : SUCCESS in 1h 30m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1787b7f4e0e4682a67f2163d3001c3f : SUCCESS in 27m 34s","accounts_in_message":[],"_revision_number":3},{"id":"4e7fbd6fd7a30d50a1d4c4045d94ded6df7e157b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-27 13:24:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b81b3968b9433658c9b3cf687659785ec22005fa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-27 14:34:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"325b57889c9088e62a17574322fc4e5be22bf39e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 14:53:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/63b0b2dbd1c64e9394bd75f9ff6519c7 : SUCCESS in 43m 20s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/87cb019a31db4697bcefeafa6befcb6d : SUCCESS in 1h 01m 26s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/d5e460b016624c9bbe285f7b3be1b722 : SUCCESS in 59m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f05f8d4dc03745c4ac1c20c31d027dc1 : SUCCESS in 10m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fa2778c6e45486491b2544592eefcac : SUCCESS in 25m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/13b6b657bed242c391fa2edc7efae927 : SUCCESS in 23m 22s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5aba9b3296fc4d0a8bd16d62130e9a59 : SUCCESS in 9m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0ed4aa625028432ebfecbb6ab7e3bdad : SUCCESS in 1h 10m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/668fecbd88e44f91a03aee87e69901e6 : SUCCESS in 49m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0f34e32bdb44864af70f492d615a3b7 : SUCCESS in 1h 16m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/254c124b89da42978dadcb7469b62667 : SUCCESS in 34m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0be65085eb324b89ac48ccebde2c9114 : SUCCESS in 1h 21m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a1ce8283d03747c9bf211ffd38d84c31 : SUCCESS in 1h 19m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a1111d7a0664a8a9170dbb5ec796e31 : SUCCESS in 29m 50s","accounts_in_message":[],"_revision_number":3},{"id":"fd78a759fff9526180ec35325f44e7cff33e6d9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 15:01:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"329585311f200f382a047dfb6a313416d5382749","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 17:43:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/0800ba7704b04b41be53328f83e3d6d2 : SUCCESS in 1h 09m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc006d350a84a01aeeb951b3385b7d5 : SUCCESS in 11m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e262fe1c586477184a49e22adb2f332 : SUCCESS in 24m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a4197347188a4aa89b04265397803644 : SUCCESS in 26m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/889147f51e47408cb07bffd2ec2f1a64 : SUCCESS in 8m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/91c3ba45581040b9ab22e528ce46aa75 : SUCCESS in 1h 16m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1c78b10058914fa8a018a4077ec0a874 : SUCCESS in 50m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/228810eb1e5e43fb9cec1ea6d30992c0 : SUCCESS in 1h 20m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1800f6ff79824565b1181e75ef2aa885 : SUCCESS in 37m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ddcd579cfb5d4d4383a60bd21e91586c : SUCCESS in 1h 19m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/88a874d3eca7479b975f169f3b6e473d : SUCCESS in 1h 13m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79fe328184ff489cb2754213af4efaae : SUCCESS in 26m 04s","accounts_in_message":[],"_revision_number":3},{"id":"562e69097e1bd7b99b0b9903d74958c4bfdb3d15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 17:43:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e950a57c1123ddecc91cc8ecb000d7b9eaaa515e","revisions":{"9498e6c8579b9184092a8141303af048674400aa":{"kind":"REWORK","_number":1,"created":"2021-08-05 10:51:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/803584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/803584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/803584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/803584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/803584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/803584/1"}}},"commit":{"parents":[{"commit":"951ba752256802eaed2780aec1b5a42dcf23a417","subject":"Add extra logs to the network update callback in L3 agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/951ba752256802eaed2780aec1b5a42dcf23a417"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-07 10:00:14.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-08-05 10:51:15.000000000","tz":120},"subject":"[L3] Use processing queue for network update events","message":"[L3] Use processing queue for network update events\n\nRouter_info\u0027s _process_internal_ports() method is the one which is\nmanipulating router_info.internal_ports cache and network_update()\nmethod from the L3 agent is relying on that Router_info\u0027s cache to\ncheck if updated network is connected to the router or not.\nSo they shouldn\u0027t be run together as that may cause some race conditions\nand unexpected issues, like e.g. described in the related bug.\n\nUntil now, network_update event was the only one which was processed\nwithout using queue of events. And because of that such race condition\nas described above were possible.\nTo fix that, this patch changes network_update method in the way that it\nnow adds update events for each router hosted by agent to the queue.\nThose events for single routers are then processed, checks if network is\nactually connected to the router and if yes, schedules router update to\nbe processed.\n\nConflicts:\n    neutron/agent/l3/agent.py\n    neutron/tests/unit/agent/l3/test_agent.py\n    neutron/agent/common/resource_processing_queue.py\n\nCloses-Bug: #1933234\nChange-Id: I2efe66a7415f7a18fb85bd2536a1901e751d6203\n(cherry picked from commit 6ce48c30bdc153e75c4b9b409203f472b065c11b)\n(cherry picked from commit 604b055c71c374bcf517359c118ace267e3a4ab1)\n(cherry picked from commit 05d0bc6d500aaafe26fd667a761dae408a8c8ef8)\n(cherry picked from commit 463083c71387b249b9aa834e126a17a1a0f3a189)\n(cherry picked from commit 6bd15da7ea77ec1d36df99a45eb26a57578f3d15)\n(cherry picked from commit c88793e6d8c59b67bad646d12af2a4fdad85baa0)\n(cherry picked from commit e2d5d65a41209423e2d980db407b0567f97b903d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9498e6c8579b9184092a8141303af048674400aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9498e6c8579b9184092a8141303af048674400aa"}]},"branch":"refs/heads/stable/queens"},"e37f912a7520936ed05902ade3ab0ef036805272":{"kind":"REWORK","_number":2,"created":"2021-09-07 14:00:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/803584/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/803584/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/803584/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/803584/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/803584/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/803584/2"}}},"commit":{"parents":[{"commit":"951ba752256802eaed2780aec1b5a42dcf23a417","subject":"Add extra logs to the network update callback in L3 agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/951ba752256802eaed2780aec1b5a42dcf23a417"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-07 10:00:14.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-09-07 13:57:53.000000000","tz":120},"subject":"[L3] Use processing queue for network update events","message":"[L3] Use processing queue for network update events\n\nRouter_info\u0027s _process_internal_ports() method is the one which is\nmanipulating router_info.internal_ports cache and network_update()\nmethod from the L3 agent is relying on that Router_info\u0027s cache to\ncheck if updated network is connected to the router or not.\nSo they shouldn\u0027t be run together as that may cause some race conditions\nand unexpected issues, like e.g. described in the related bug.\n\nUntil now, network_update event was the only one which was processed\nwithout using queue of events. And because of that such race condition\nas described above were possible.\nTo fix that, this patch changes network_update method in the way that it\nnow adds update events for each router hosted by agent to the queue.\nThose events for single routers are then processed, checks if network is\nactually connected to the router and if yes, schedules router update to\nbe processed.\n\nConflicts:\n    neutron/agent/l3/agent.py\n    neutron/tests/unit/agent/l3/test_agent.py\n    neutron/agent/common/resource_processing_queue.py\n\nCloses-Bug: #1933234\nChange-Id: I2efe66a7415f7a18fb85bd2536a1901e751d6203\n(cherry picked from commit 6ce48c30bdc153e75c4b9b409203f472b065c11b)\n(cherry picked from commit 604b055c71c374bcf517359c118ace267e3a4ab1)\n(cherry picked from commit 05d0bc6d500aaafe26fd667a761dae408a8c8ef8)\n(cherry picked from commit 463083c71387b249b9aa834e126a17a1a0f3a189)\n(cherry picked from commit 6bd15da7ea77ec1d36df99a45eb26a57578f3d15)\n(cherry picked from commit c88793e6d8c59b67bad646d12af2a4fdad85baa0)\n(cherry picked from commit e2d5d65a41209423e2d980db407b0567f97b903d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e37f912a7520936ed05902ade3ab0ef036805272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e37f912a7520936ed05902ade3ab0ef036805272"}]},"branch":"refs/heads/stable/queens"},"e950a57c1123ddecc91cc8ecb000d7b9eaaa515e":{"kind":"REWORK","_number":3,"created":"2021-09-24 10:07:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/803584/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/803584/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/803584/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/803584/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/803584/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/803584/3"}}},"commit":{"parents":[{"commit":"951ba752256802eaed2780aec1b5a42dcf23a417","subject":"Add extra logs to the network update callback in L3 agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/951ba752256802eaed2780aec1b5a42dcf23a417"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-07-07 10:00:14.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-09-24 10:07:17.000000000","tz":120},"subject":"[L3] Use processing queue for network update events","message":"[L3] Use processing queue for network update events\n\nRouter_info\u0027s _process_internal_ports() method is the one which is\nmanipulating router_info.internal_ports cache and network_update()\nmethod from the L3 agent is relying on that Router_info\u0027s cache to\ncheck if updated network is connected to the router or not.\nSo they shouldn\u0027t be run together as that may cause some race conditions\nand unexpected issues, like e.g. described in the related bug.\n\nUntil now, network_update event was the only one which was processed\nwithout using queue of events. And because of that such race condition\nas described above were possible.\nTo fix that, this patch changes network_update method in the way that it\nnow adds update events for each router hosted by agent to the queue.\nThose events for single routers are then processed, checks if network is\nactually connected to the router and if yes, schedules router update to\nbe processed.\n\nConflicts:\n    neutron/agent/l3/agent.py\n    neutron/tests/unit/agent/l3/test_agent.py\n    neutron/agent/common/resource_processing_queue.py\n\nCloses-Bug: #1933234\nChange-Id: I2efe66a7415f7a18fb85bd2536a1901e751d6203\n(cherry picked from commit 6ce48c30bdc153e75c4b9b409203f472b065c11b)\n(cherry picked from commit 604b055c71c374bcf517359c118ace267e3a4ab1)\n(cherry picked from commit 05d0bc6d500aaafe26fd667a761dae408a8c8ef8)\n(cherry picked from commit 463083c71387b249b9aa834e126a17a1a0f3a189)\n(cherry picked from commit 6bd15da7ea77ec1d36df99a45eb26a57578f3d15)\n(cherry picked from commit c88793e6d8c59b67bad646d12af2a4fdad85baa0)\n(cherry picked from commit e2d5d65a41209423e2d980db407b0567f97b903d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e950a57c1123ddecc91cc8ecb000d7b9eaaa515e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e950a57c1123ddecc91cc8ecb000d7b9eaaa515e"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
