)]}'
{"id":"openstack%2Fneutron~804213","triplet_id":"openstack%2Fneutron~master~I943f610c3b6bcf05e2e752ca3b57981f523f88a8","project":"openstack/neutron","branch":"master","topic":"packet_rate_limit","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-02-17 18:49:21.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-02-17 18:49:21.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-02-17 17:12:18.000000000","reason":"Lajos Katona replied on the change"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2022-02-17 18:49:21.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-02-15 07:44:36.000000000","reason":"Oleg Bondarev replied on the change"}},"hashtags":[],"change_id":"I943f610c3b6bcf05e2e752ca3b57981f523f88a8","subject":"Add table for pps limitaion","status":"MERGED","created":"2021-08-11 09:03:14.000000000","updated":"2022-02-17 18:51:55.000000000","submitted":"2022-02-17 18:49:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"804213-packet_rate_limit","meta_rev_id":"ae6f6f925118571ea1381d9160d6e317cf1563aa","_number":804213,"virtual_id_number":804213,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-02-08 22:45:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 18:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-11 10:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-15 07:44:36.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-17 17:12:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 09:51:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-11 10:32:48.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-11 11:10:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-12 11:29:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-19 03:03:47.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-22 16:27:43.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-10-22 16:27:43.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-10-22 16:27:43.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"2e668b9b370b7257f2afe77b56fccf921e725f6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-11 09:03:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"344a407bccaa33163d0a17f2bcc72f66ca8bb2aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 09:51:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/800b0a0bb33d4012885a8f5a05742e44 : FAILURE in 41m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c567145826794bcd9e432b4c8aeb50ec : FAILURE in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1ae0b7d998c561168798a3109438ab4bbb0c374","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-11 10:32:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/1 : SUCCESS in 1h 28m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1b322be53f2498ae25ca5940ed6b024e28e63a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 11:10:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5f5b9c7a49bc4a809653e4b6ac6432e6 : SUCCESS in 46m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8c8ceff5bbb348c49e04c16edc1f93ff : SUCCESS in 2h 03m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/19053f2960574086823f8f20245fc439 : SUCCESS in 1h 18m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feb0caead87e4898bff548fabad3e675 : FAILURE in 31m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01cc9c4521d2402d963effb3ee3f2710 : FAILURE in 40m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99eabbd8514f4ca6a854ce074f97e852 : SUCCESS in 15m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bc6162048be4c4c987cfd880952b662 : FAILURE in 25m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0ca6b30e14a4416a659d60219530750 : FAILURE in 30m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aec3449c4f7a404ca3a519c8942f014f : FAILURE in 24m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791c2cea15a549fe91ff1268212425d4 : SUCCESS in 9m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e9c057c0d3e4f868730e50d5319bf60 : SUCCESS in 47m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/630e7fceac934e3a98f60cdde7213de1 : SUCCESS in 1h 51m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/37857bd434184d60b455627391205f07 : SUCCESS in 1h 07m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef6fb5d79b1744d38a99dd4bcf15a01e : SUCCESS in 1h 00m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ac5086608ad6481688185523dcf0fd30 : SUCCESS in 1h 02m 38s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e33697e6175042bf8c09a6bbfe7552d4 : SUCCESS in 1h 08m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/21e0bc61ab4446cc83df908c0377dfdb : SUCCESS in 1h 12m 16s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/97b1b24887eb421ab95fe05247bec7fa : SUCCESS in 1h 26m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac735345908a4181a13d1dc98bfa150e : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":1},{"id":"438fc8f483cb844cb151368b3af5dbe8692e9e41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-12 08:19:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f2c6e4e422c1bd5d996bcb5cdf8adafc9df0edf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:06:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4760ddff755b4a24a17b988314a9d4fc : SUCCESS in 45m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37598944fa59439f84bfc970d86317fa : SUCCESS in 45m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e27858350886d34bf5750c24250a9fc1b5cf9ec9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-12 09:49:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/2 : SUCCESS in 1h 29m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a0ca2e3329b81fb9b6bb49b810d789aa6536c188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 10:32:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7ba12db5a84d49bb8f196ef0ef3a84a0 : SUCCESS in 41m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8ed8d3910d47450bb373aafadb29b6c9 : SUCCESS in 2h 12m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a9489686abd465582af8e19147b6f2d : SUCCESS in 1h 38m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ea53315b8284b86a98a8931389c9597 : SUCCESS in 28m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d85fb344a11745ce8c0ea176e67720d6 : SUCCESS in 34m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3266ee93964eac999fb17de3218e61 : SUCCESS in 15m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38d898b079574e6f8f7418dd80d3b593 : SUCCESS in 34m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/651a64080d4d4cf0b2737bec5edd7014 : SUCCESS in 21m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42acec70d43345d6a5ada3a205700253 : SUCCESS in 29m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea780e8bf3a4fdaa3367f77f2c26a02 : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3092ecb994814fdfa40fb196e36ce350 : SUCCESS in 53m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3331bb0355eb420bb607f3d63e2230f6 : FAILURE in 1h 30m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7955f9be7abf49f5a7d81a82bbec159d : SUCCESS in 1h 12m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a2ae24c8efe40c2ae931cd37080d0a2 : SUCCESS in 58m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/35f249374272444c9cca6a90dbfd21a5 : SUCCESS in 1h 06m 01s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e07d84f06b2840bdba441f5879962ab1 : SUCCESS in 1h 12m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3ac59c77f6cc4c368b11e7dbc7368542 : SUCCESS in 1h 10m 06s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/828929a8c39e4a979138ba058de3b3e5 : SUCCESS in 1h 26m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0931ee26a62e4be2a2bfca8dd9d2f7ae : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":2},{"id":"47e5045868bded6db7ad3a03c3ee81801ce7d1a5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-12 11:29:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nYou should update doc/source/contributor/internals/openvswitch_firewall.rst","accounts_in_message":[],"_revision_number":2},{"id":"b51648c0848389fa3bafedf361a33857aaac290b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-12 11:46:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd8b74bc651142049f10675c75f5229202b702da","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-12 11:48:04.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"913ad8622491f7bbdd16db9212b78552490394c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-13 08:15:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa62b17e7ae3ff0393b498294f86c57407cadc5b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-13 08:16:01.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e You should update doc/source/contributor/internals/openvswitch_firewall.rst\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"45c7eec3bafe51888f4e657089778078c1c6068d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 08:59:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/448c0e0920aa4cc0a1899680098d6ae2 : SUCCESS in 41m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6818965a990485ca76e2415f83af9b7 : SUCCESS in 43m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4717d60068b321f23aa3a60860a8011d1bf453ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 09:56:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4357907979fd4261b8a72359c2c6b8bf : SUCCESS in 49m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cbcaf24238bb4efb86f730435bf82d6e : FAILURE in 1h 30m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a8f2b9a3918e4b268696341673ac3290 : SUCCESS in 1h 37m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aceb7bd8a224f6dad262e4e3438adf3 : SUCCESS in 43m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9a6f2db4aeb4eabacb63ff1042d01f8 : SUCCESS in 40m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b5753e0fe94889ad565bf968195c99 : SUCCESS in 15m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00db9a0586f648f596c1e41c0b172140 : SUCCESS in 29m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e5b68e810154c5bb9affd2942f45086 : SUCCESS in 32m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d67d803b3d48439898d36f13a4bee7ce : SUCCESS in 27m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf655b2e47c482fab61eb82037194cb : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/773516280be941baa0ced1a2fb98e5e5 : SUCCESS in 50m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8559336727747c6bd36c6d27fc6fbe7 : SUCCESS in 1h 39m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a27f550a010f40a2a61040d3acd8e3ca : SUCCESS in 1h 08m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6bd226c145442f7810104c264f6bfe1 : POST_FAILURE in 57m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f728110f2779442e96a77b92b80fbc84 : SUCCESS in 1h 03m 52s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d739ebe922d3455faa5fdae218718609 : SUCCESS in 1h 18m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/469367f1807f46838a9002746c24700b : SUCCESS in 1h 12m 14s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/7a16f300b42d4efbbd7c4855c12e378f : SUCCESS in 1h 22m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1166c6ee1ab4ba185780baba5fc538e : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":3},{"id":"e60c5262cd68d77ed46d64ec40abadc0373c77db","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-13 10:05:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/3 : SUCCESS in 1h 50m 00s","accounts_in_message":[],"_revision_number":3},{"id":"7fc87178ca783d476d254005f409ec4b2f879b9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-17 04:36:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"f5bd6d6ec1c4188b72419758bda8a0eb87062618","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 05:14:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cac5ddf12eb4ff79bd8cd9876ab00fb : SUCCESS in 32m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fed498aa85ea41cb9477bef22bdc0ae3 : SUCCESS in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"467ffc6fd60c481e8cfaecc67d9a51ab859a8219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 06:22:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9c4c54f8c69d4659860ca4604e7c84d3 : SUCCESS in 43m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bb4c07de9e664eacb9739121b54fb4e7 : SUCCESS in 1h 34m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6f2890cc208646179f7c1665a61c35cb : SUCCESS in 1h 17m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b05155d9c1c540d7a8092b95f15dc75e : SUCCESS in 30m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69ff8af03ddc404e8100afcd4750405b : SUCCESS in 37m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/341cc1e9b7ba49ae8ebbacee9f83cf7d : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48d892d472e247c4866d96c728382e29 : SUCCESS in 26m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44a61c19383b494dad7e0ce336b759cf : SUCCESS in 24m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b2374cc6f744ce3b804ad2534cb89a3 : SUCCESS in 26m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/896f45f865af4051a78c13bb457e1a8e : SUCCESS in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b8e4e8a7b5b438392411134ce37dff7 : FAILURE in 40m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7351a9738e94ea097e6e0a0384d7329 : SUCCESS in 1h 43m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d4d83f5fe9564b949046c4078fb08128 : SUCCESS in 1h 06m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/42a5a0722f7848218233c03b12ea040e : SUCCESS in 1h 00m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7c37d9a4f5bb43029e78754eace866c2 : SUCCESS in 1h 00m 19s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bf8d8b0f2c5a4bd7a55146dc8312f893 : SUCCESS in 1h 09m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2ee23f6339754080aa70686e61abcd43 : SUCCESS in 1h 09m 31s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/a4f0349847374a3f99909036b9fbbb2c : SUCCESS in 1h 25m 08s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f1c484a1a2a43709c81f562843e57c6 : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":4},{"id":"a3751c35485c86259019c4b426356eb4d283cc4d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-17 06:24:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/4 : SUCCESS in 1h 46m 19s","accounts_in_message":[],"_revision_number":4},{"id":"abba40be48dcc50bbceb798ad4a89ef511e334cb","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-19 00:59:45.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"30d5e925b2b14278bcfeb10ebda30cc559070b86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 01:45:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00c026d14d51468b846b894bc022e823 : SUCCESS in 43m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9846b9fb048046c7b861b7adea690b09 : SUCCESS in 43m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ccf54340ebffa1c91f03be1a2ae90919b1994c9c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-19 02:46:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/4 : SUCCESS in 1h 45m 35s","accounts_in_message":[],"_revision_number":4},{"id":"edffb2e27ca2c844f403d7af21ee75e33bb53b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 02:47:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/027cce153d474eee8aff092c48b9d4ee : SUCCESS in 35m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8da3eaf1a80e4d958d216c74c0b380bb : FAILURE in 1h 45m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dae3543eb09d430083f4d0a2dfe2ef7f : SUCCESS in 1h 19m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32f953c3620b469c8e8c806eaa8a7b30 : SUCCESS in 36m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51e3e8a26d524762b4b98e60ae1998b6 : SUCCESS in 42m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43314509eb4c464d836df72508c766be : SUCCESS in 12m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b35374812d64d71866b8fbc36cda06e : SUCCESS in 28m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/707779bc86dc47d09b9a172be27dc8ee : SUCCESS in 25m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ad1e4e7c2f644b6917724ef1dae02b9 : SUCCESS in 24m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39b4142ec1045aa9cc62231c2ff2e1f : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8433f2eb3a074ec88ceb4b99b1f11e68 : SUCCESS in 43m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2032c87a4f1416284a06fe06847b2be : SUCCESS in 1h 43m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a5ecd64591a94d13b531502180010072 : SUCCESS in 1h 00m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/23f5e22c71034c82a1c999f3123ac099 : SUCCESS in 1h 01m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c755d90b8aec49f4a5a44bb15d530a53 : SUCCESS in 54m 19s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/91a8dc7477c74c96b2e47f290569f614 : SUCCESS in 1h 07m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b87851e1343b4951917926760c69f94c : SUCCESS in 1h 18m 19s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/7a074a34bccc4f579f700fa99b398fed : SUCCESS in 1h 31m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f9564be515d4613a52321408c07604f : SUCCESS in 47m 35s","accounts_in_message":[],"_revision_number":4},{"id":"b9e395e5e0ff6b7c529f223645e03367f86824ff","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-19 03:03:47.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/13/804213/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"94170a2fc7787b191c5c13b493545c95d0f93a76","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-20 00:52:40.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/13/804213/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a7b848affbe88e48863f72f7440299aeb455d0e6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-22 10:00:59.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/13/804213/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"2a6631de2767bc6825218d1ef81c04ae5295727e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-26 22:02:27.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/13/804213/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"5718bf6aaf75a1f276a920bf13f1cbb55eaf196a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-31 02:31:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6667fc0994d658b9a04f88c1a31e808d430477c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 03:06:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e3a5a7336ac46e488253cab15bebc2f : SUCCESS in 34m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4848a3fd7f2844e69888b42c2a91ec09 : SUCCESS in 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3393da16f4f51e3c4afc2a9394fa4aa7499b8307","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-31 03:25:26.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/13/804213/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"fbf9bc666510ba46d8ad5fffb2289226c0e1f5ce","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-31 04:03:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/5 : SUCCESS in 1h 31m 11s","accounts_in_message":[],"_revision_number":5},{"id":"5bf3e5d725f8fecc1075a9c7f3ae751884b29c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 04:34:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9ed72344f1d44aed92dbad965090d083 : SUCCESS in 46m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cb58e245ff1346b8a7a6368ef17eb081 : SUCCESS in 1h 43m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/895e8bbf76a34e37a1cbc961e0b92d69 : SUCCESS in 1h 55m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/484e7ff54c5340e9a19a983ec38aeb35 : SUCCESS in 33m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d1b092b418d4c6392f54096e592cc3e : SUCCESS in 33m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9816cfb8597e46d4a2d84069fe85c75a : SUCCESS in 13m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42cbc183341a4caf867125ce41a51756 : SUCCESS in 26m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7fb3a3177244cfa84f38c56f3aa27d7 : SUCCESS in 23m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/914e0774043f4cfebb62413fce3997a0 : SUCCESS in 23m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49602dbbce3b4002aa365c45082fc703 : SUCCESS in 10m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6748cc8bca9341bbaca57ee882d44c3e : SUCCESS in 47m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/59501bd6d53a4697806748c3395eff47 : SUCCESS in 1h 51m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c894fd3880e4cbfaa951e03a61eb233 : SUCCESS in 1h 21m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/89fdf062f9014a1888e921aa1ec0301c : SUCCESS in 1h 04m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f112d3dca9264e819b5623382fd82ae1 : SUCCESS in 1h 03m 16s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cf356110ae3d4fe69dc5f69756b57753 : SUCCESS in 1h 13m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0f485904ca304a5ba9772508bf745f5d : SUCCESS in 1h 12m 53s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/063ae94acef64cb2962ba25ff784bc83 : SUCCESS in 1h 30m 12s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/982b4409928c4e2cb6cbb45964456113 : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":5},{"id":"2d430e91ec1dad52f2c3f75d4ee0a2696868c7c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-09-29 03:50:54.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"0ce4d393c3d794f7b01e0bc2c0cdbdd4a36532f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 04:38:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51262b48c0294883a8d256ca10ea8f07 : SUCCESS in 45m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5109b53442704ac6a562a509e26b749b : SUCCESS in 40m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3c321e5139b52726a9b20661aacdaf7d664aa918","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-29 04:41:54.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/13/804213/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"e302cc08429d6ee7df7476a3fa6ea6c9233d7d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 06:24:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6f122cfd6c0f49a3a3ee9d9eb3fdffae : SUCCESS in 1h 26m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e75d2096c044f83b998ad9b4e7f07cb : SUCCESS in 2h 32m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/71adbac96432404d841ec3b6970abfe0 : SUCCESS in 1h 39m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6662b34e124845459f16dacdcebdce2c : SUCCESS in 37m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bed69fd1fcaf4439983c328cdaa7163a : SUCCESS in 43m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7041705920f9439c82fd9768d8907563 : SUCCESS in 16m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/981d03b8d7d74f13adf58c1eebc0f7ef : SUCCESS in 27m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/755ff1b2545d4f16b93c74963710af98 : SUCCESS in 27m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/919d85b949c24b0eb5e3bf75725c9b99 : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdcb93012455427b830963c08b1413db : SUCCESS in 10m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc7723169e0f4de1b29690ef3203bb73 : SUCCESS in 45m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b47e394f5d1c44cabcc3fcf765d49a34 : SUCCESS in 1h 43m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b0534421bb10481cb9ad71971362a7af : SUCCESS in 1h 09m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7aca6604c76457c8c5d3904528c83a8 : SUCCESS in 1h 00m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/67477dddb0c04d45bf4a04c3da75e30e : SUCCESS in 53m 32s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d351ff00df2248d2a32009ce17f7384c : SUCCESS in 1h 09m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/65e9635b7dc94f8ba0fb9e51f5402b43 : SUCCESS in 1h 06m 21s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/b3d1cc922fcd4dfa873ffbab6c612a37 : SUCCESS in 1h 27m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a96987eb04e44b9b877b5b161d969cd1 : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":6},{"id":"6c8129fc34df05be4bd5e47c5828379cd6f062d3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-10-22 16:29:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3f13375793bc3cac0d9781e28d47609eafffe3f9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-02 16:52:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"5aded5897d0e43d4100675e1e7bcdfac0e8e4a03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-11-05 09:33:14.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"55cd08264f4e10b632337ced946fec390e8d279c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 10:32:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2c212e8773343d2a6c2722db5732ee4 : SUCCESS in 57m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f92df3328e5a47ac9016325ace41aa47 : SUCCESS in 57m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"747d6fba87512a4a4f41206712bcc77f86455178","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-05 11:42:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/7 : SUCCESS in 1h 50m 54s","accounts_in_message":[],"_revision_number":7},{"id":"f20e83346da42ab3522fbed6144ab62f7839d2cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 12:29:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/026d94b1f69a4a999aa8a24cd60f0edc : SUCCESS in 52m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9df8739b2f3a48f69de3e6ac13373445 : TIMED_OUT in 2h 50m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c3eb17a35454341aee5b2d66679a639 : SUCCESS in 1h 53m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1edec6835b64ecca31b3ff82d536708 : SUCCESS in 1h 02m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aeb997c14a140e8abba2824956e8e61 : SUCCESS in 55m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab499b8490854903a2ceb06474cbdb87 : SUCCESS in 17m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf41806e4cde42a7a5aabfe85c2e518c : SUCCESS in 51m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f365652c28de4ad7b7e39cbf719eafcd : SUCCESS in 43m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94b963510f3e47aa83bea351f9c15264 : TIMED_OUT in 41m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dd4f31cb0e548e098fd3e90af2a7952 : SUCCESS in 12m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3c58b166bb049cf921d2c2745bca4ab : SUCCESS in 55m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/402d20d438554cdeb2529e25d08fea4e : SUCCESS in 1h 28m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9daf211057e94e34b477103b688a61d2 : FAILURE in 29m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa87915145b448c280e23539112a1132 : SUCCESS in 1h 09m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f72e6e5c78be45d4aa436fbcc6b51dba : SUCCESS in 1h 00m 35s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f211e47c5e4a4652b1c4dbc6dd7c9ef7 : SUCCESS in 1h 29m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/86852b6f81df494196c0a01fe6fd62f7 : SUCCESS in 1h 40m 06s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/43dd0a854d7e4040ab1c910c1edb6299 : SUCCESS in 1h 50m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a390bd849aa04c76bf0e575b45aa5e9c : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":7},{"id":"e431fdb4f648bdad22762d1d3305678e55bb9275","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-11-09 00:24:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1e16be43e353e00f9879f2b6ebe17a5feed9efb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 01:26:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c1d178443584e8ca463f9599df9a096 : SUCCESS in 59m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe99b644e7af4dc1bc692948d158ab99 : SUCCESS in 55m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e680c3a191a0630ca2646eb569b53f2bc48a4a43","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-09 02:14:24.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/7 : SUCCESS in 1h 49m 10s","accounts_in_message":[],"_revision_number":7},{"id":"4e83180fc43fd874b117d8dea449a7cb4b653077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 04:54:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fe06889b320243c68bd63b3377d28675 : SUCCESS in 46m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8fc6fcfe65344865be8d1f877e92b5c6 : SUCCESS in 2h 17m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/75cf215d29964e518d829fd3fca9aa19 : SUCCESS in 1h 36m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb8cde0a6cd641f594859957428a975c : SUCCESS in 48m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc694234ca3c414cad108cf67b7cd92f : SUCCESS in 48m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5497840f54674eb1b992250bc6a3b625 : SUCCESS in 12m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4136a293da549399481d5fd84dd0dc4 : SUCCESS in 45m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/747be1de5d9f4a46b55818e9c893d547 : SUCCESS in 41m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d0f4f3db01a4202ae2e6faa968d9ba9 : TIMED_OUT in 41m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd2f91d2cda4c95a78d9629685f34c0 : SUCCESS in 11m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ba34a2e771d41a59ceceffa425b32e6 : SUCCESS in 49m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/36f50697ea864aeaa3e42b4d413c1adf : SUCCESS in 1h 29m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ec28bdf49b974b1ab9fea2adf86f0bee : SUCCESS in 1h 16m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/50be277425ca482ea5f2b44af3f9c08c : SUCCESS in 54m 48s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7e8c55a4bd98418cbea879f9fa676fe5 : SUCCESS in 47m 59s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5ce8d8b979db4e8a8f54c4da5484930f : SUCCESS in 56m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/169dbb8d071245a6bbfab0094c50c44f : SUCCESS in 1h 14m 37s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/0185095636aa415bb73c9a03cba052c1 : SUCCESS in 1h 30m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ac35176a6c5436195d6824f026585d6 : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":7},{"id":"6df8a8e3cfec4aa769460b88912896800d993a25","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-09 16:33:37.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/13/804213/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"032e93b88ae82f4a501cc8d6481144c6066a8505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-08 09:24:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b6a1b6a47bce04a6e5a556ab4e0f1c84a595784e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 10:44:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f406a37c9a6446a989534fc95a0be440 : FAILURE in 1h 14m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/008ab155f4e84d3696d40ac23d5ac7d7 : FAILURE in 1h 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a6e9e02fc5c7007a07558722defdfa0005c0f42f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-08 10:57:55.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/8 : SUCCESS in 1h 32m 32s","accounts_in_message":[],"_revision_number":8},{"id":"e24b9a8c2f09ee6e7266e2091d11ca3f37e16afb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:12:44.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dfaba269c4be4049b0d2ab34e4a447d5 : SUCCESS in 39m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cb7fbb5508fd45059dbe9f2b891c5419 : SUCCESS in 1h 23m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/947fd5c0839641c2bc22a735f3d00554 : SUCCESS in 49m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a889c54c6d1c477fb73ddec930f20c1a : SUCCESS in 28m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1c6439d8643461587fef60e2fa45249 : SUCCESS in 34m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a07fac4b584fc4a2f6cfd27d20e11b : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddd8212773884c748fab5e15b974c6fd : SUCCESS in 30m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66e6f07699e4465fb113181641e59c1f : SUCCESS in 27m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d6171f177d4d51a41bb861ac01cb82 : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b6930549c7a46fc963fad79c7485da0 : SUCCESS in 49m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d32f9ca80b8e4e9ba214b6ba57038972 : SUCCESS in 1h 36m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/51fed6152598458eb9f365b947d4cf69 : SUCCESS in 1h 20m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2321cb7d45e45e48e723aea57c4e1f1 : SUCCESS in 56m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/70cd3ed2efd849ac9ae33fb89837ef41 : SUCCESS in 1h 05m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4cc3cff28687491b87caf251dd402003 : FAILURE in 1h 04m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b94a8e5e923425d8b9149dafa41f571 : SUCCESS in 1h 18m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52580ab2c8b74864bfcbe52e9e0beca1 : SUCCESS in 27m 04s","accounts_in_message":[],"_revision_number":8},{"id":"46239b55f0feedcfb8b1955892be8d714b264a43","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-08 14:01:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0a03018cd2580712ac93f4c0b698f9ee99755981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:49:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b53fac459bd343af88596364206566fd : SUCCESS in 47m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b73014f0ec924313a25ac877a01da245 : SUCCESS in 56m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/492d0b690d054bc0affc7cd866e50960 : SUCCESS in 48m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86f82c792ee94f9a8dc0ddef52e373d0 : SUCCESS in 36m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4888a6b381164fb38d972c0cb3098d40 : SUCCESS in 34m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06772f0ba7dd463dba752f2ba7f0b6e5 : SUCCESS in 14m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41d446dd136e43a89e54aba6bba221a6 : SUCCESS in 25m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8ef6cd3b42145e0b2c92567b8236af7 : SUCCESS in 40m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68c44d271d3443aa591b619d795ef46 : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38d989e28cf2479fa1059b2fac6579af : SUCCESS in 52m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ef83c7c8d854ef1aa8ef07a85071965 : SUCCESS in 1h 26m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e36773ef93ec4d10896849fb7e0867a1 : SUCCESS in 1h 18m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9cf30c77af244e89eeb7a7db7c53c26 : SUCCESS in 1h 06m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eea7f00c09194953a7edd8cc0f47e009 : SUCCESS in 1h 09m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8d7ae622838a48708e16124fd980f750 : SUCCESS in 1h 05m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/88a3af3933e24323913c1f736f38f4b8 : SUCCESS in 1h 05m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18e689b9d67847ed847e7b4e434a7aed : SUCCESS in 22m 42s","accounts_in_message":[],"_revision_number":8},{"id":"1b6b4682da8c6aebf21fabb3e256f688f1e3f584","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-08 15:52:04.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804213/8 : SUCCESS in 1h 49m 42s","accounts_in_message":[],"_revision_number":8},{"id":"faf4bab6fdb34ac05e80617d0d7abb0afef70fb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 16:13:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/441bbebea8d4442e8e1658e1e68fb8d7 : SUCCESS in 43m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/313a94c4708145a8bf308e901985c376 : TIMED_OUT in 1h 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"be690ab6d9a2f760a0f46cc6b0de23821bf89e72","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-08 22:45:08.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/13/804213/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"d4a12607e0215c321ebc78b926cc02e6233ca79a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-09 10:59:01.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f5fdfa506594a39b0cf3f48d98ee7bbf2535535a","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-09 13:34:58.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b3ee97435f5beb4624f73c97cbf5e70f09b12f44","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-09 14:32:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d5f5fa641f3d4a371f846635d35a0bf7f6e81a8c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-09 14:33:03.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3624a2e3095252a49d822cf966d41a4137abd7b8","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-09 14:33:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"aa400321987554201ea2abd3f416dfcf340b4b70","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-09 14:33:33.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4b457ac9450e666d3fb474a0952c67c04f523dae","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-11 10:33:06.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"97f03a9f49bf4b3f73ed5491e7c7306c8cc9bbae","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-02-14 11:37:54.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"23cecdc85e5ed99bdc9796a47eb4477ca3134e04","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-15 00:53:29.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a569697e3505bb25b2fa8708d53e82d73730f71d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-02-15 07:44:36.000000000","message":"Patch Set 8: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"337d83b3faf4ba3e6eb44aa4cd4242eb7cb82642","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-17 17:12:18.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e506d4406b32cd54469c0f5b8839aa53b3563b32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 17:13:17.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"fb2e7cdb0d1fd9fe1e4ebf2ed85c7473578ae649","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:49:21.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ba6bbe1b5fc64a79a055f7b037440a72 : SUCCESS in 57m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ba6bc30fed9748c3b7e3d6fbfb728208 : SUCCESS in 1h 23m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/470d9c97e63b4400887011867359790f : SUCCESS in 1h 16m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17c85c131fef42bbac34e928a768f89f : SUCCESS in 33m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bef0d0a1ab8844da99ae8296a676a003 : SUCCESS in 15m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1c4917b72284ec191a9470547fed99a : SUCCESS in 39m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a28b7abcb1944ea29b84c7d8bd931477 : SUCCESS in 32m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0645af0668d44952bfc786620b28ee2c : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6494c5af9504316b177f70d6d017eb4 : SUCCESS in 50m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/079ddc5f78b54310a10a091085ea2c83 : SUCCESS in 1h 34m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/467f22d8870244e5ba668386e8ee11a8 : SUCCESS in 1h 15m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c3b492e6b6e48dbb2423df98f209d31 : SUCCESS in 57m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e4b7ec8179d441a8b6b98eca8ae8310b : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":8},{"id":"2faf55b45888baa2c45779013368fdc1c542e5eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:49:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"ae6f6f925118571ea1381d9160d6e317cf1563aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:51:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aeb75b006bc4c59a39438fbb9eb654c : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"053a9d24eca2c9372bd2b767504dd6b6db30643f","revisions":{"1e5c4bfc0770e1277eb077c707ff34dcd349a2ce":{"kind":"REWORK","_number":1,"created":"2021-08-11 09:03:14.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/13/804213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/804213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/804213/1"}}},"commit":{"parents":[{"commit":"578f74ca8782badfa6251f438c3cfd3e8fd10009","subject":"Merge \"Add \"network_id\" to \"_after_router_interface_deleted\" payload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/578f74ca8782badfa6251f438c3cfd3e8fd10009"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-06 06:11:00.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-11 08:51:26.000000000","tz":480},"subject":"Add table for pps limitaion","message":"Add table for pps limitaion\n\nTable 59 will be used for pps limitation, the pipeline change is:\nall original flows with ``goto table 60`` will be changed to\n``goto table 59``, while table 59 has a default rule is goto\ntable 60. Then we can add pps flows to table 59 for all ports.\n\nBasic limit pipeline is:\nIngress: packets get into br-int table 0, before send to table 60,\nin table 59, check the destanation MAC and local_vlan ID, if the\ndest is resident in this host, do the meter pps action and send\nto table 60.\nEgress: match src MAC and in_port, before send to table 60,\nin table 59, do the meter pps action and send to table 60.\n\nWhy table 59? Because for ovs-agent flow structure, all packets\nwill be send to table 60 to do next actions such as security group.\nBefore table 0 and table 60, there are tables for ARP poison/spoofing\nprevention rules and MAC spoof filtering. We want similar security\nchecks to take effect first, so it can drop packets before filling\nour limit queues (pps limitation based on data forwarding queue).\nAnd we do not want packets go through the long march of security group\nflows, in case of performance side effect when there are large amount\nof packets try to send, so limit it before goto security flows.\n\nPartially-Implements: bp/packet-rate-limit\nCloses-Bug: #1938966\nRelated-Bug: #1912460\nChange-Id: I943f610c3b6bcf05e2e752ca3b57981f523f88a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e5c4bfc0770e1277eb077c707ff34dcd349a2ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e5c4bfc0770e1277eb077c707ff34dcd349a2ce"}]},"branch":"refs/heads/master"},"b76ec9f7511e92498941448b0831060e28b571db":{"kind":"REWORK","_number":2,"created":"2021-08-12 08:19:06.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/13/804213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/804213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/804213/2"}}},"commit":{"parents":[{"commit":"578f74ca8782badfa6251f438c3cfd3e8fd10009","subject":"Merge \"Add \"network_id\" to \"_after_router_interface_deleted\" payload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/578f74ca8782badfa6251f438c3cfd3e8fd10009"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-06 06:11:00.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 08:12:43.000000000","tz":480},"subject":"Add table for pps limitaion","message":"Add table for pps limitaion\n\nTable 59 will be used for pps limitation, the pipeline change is:\nall original flows with ``goto table 60`` will be changed to\n``goto table 59``, while table 59 has a default rule is goto\ntable 60. Then we can add pps flows to table 59 for all ports.\n\nBasic limit pipeline is:\nIngress: packets get into br-int table 0, before send to table 60,\nin table 59, check the destanation MAC and local_vlan ID, if the\ndest is resident in this host, do the meter pps action and send\nto table 60.\nEgress: match src MAC and in_port, before send to table 60,\nin table 59, do the meter pps action and send to table 60.\n\nWhy table 59? Because for ovs-agent flow structure, all packets\nwill be send to table 60 to do next actions such as security group.\nBefore table 0 and table 60, there are tables for ARP poison/spoofing\nprevention rules and MAC spoof filtering. We want similar security\nchecks to take effect first, so it can drop packets before filling\nour limit queues (pps limitation based on data forwarding queue).\nAnd we do not want packets go through the long march of security group\nflows, in case of performance side effect when there are large amount\nof packets try to send, so limit it before goto security flows.\n\nPartially-Implements: bp/packet-rate-limit\nCloses-Bug: #1938966\nRelated-Bug: #1912460\nChange-Id: I943f610c3b6bcf05e2e752ca3b57981f523f88a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b76ec9f7511e92498941448b0831060e28b571db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b76ec9f7511e92498941448b0831060e28b571db"}]},"branch":"refs/heads/master"},"41531672e7b13794541dbaa7336393aaf5cf20ce":{"kind":"REWORK","_number":3,"created":"2021-08-13 08:15:06.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/13/804213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/804213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/804213/3"}}},"commit":{"parents":[{"commit":"578f74ca8782badfa6251f438c3cfd3e8fd10009","subject":"Merge \"Add \"network_id\" to \"_after_router_interface_deleted\" payload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/578f74ca8782badfa6251f438c3cfd3e8fd10009"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-06 06:11:00.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-13 01:40:30.000000000","tz":480},"subject":"Add table for pps limitaion","message":"Add table for pps limitaion\n\nTable 59 will be used for pps limitation, the pipeline change is:\nall original flows with ``goto table 60`` will be changed to\n``goto table 59``, while table 59 has a default rule is goto\ntable 60. Then we can add pps flows to table 59 for all ports.\n\nBasic limit pipeline is:\nIngress: packets get into br-int table 0, before send to table 60,\nin table 59, check the destanation MAC and local_vlan ID, if the\ndest is resident in this host, do the meter pps action and send\nto table 60.\nEgress: match src MAC and in_port, before send to table 60,\nin table 59, do the meter pps action and send to table 60.\n\nWhy table 59? Because for ovs-agent flow structure, all packets\nwill be send to table 60 to do next actions such as security group.\nBetween table 0 and table 60, there are tables for ARP poison/spoofing\nprevention rules and MAC spoof filtering. We want similar security\nchecks to take effect first, so it can drop packets before filling\nour limit queues (pps limitation based on data forwarding queue).\nAnd we do not want packets go through the long march of security group\nflows, in case of performance side effect when there are large amount\nof packets try to send, so limit it before goto security group flows.\n\nPartially-Implements: bp/packet-rate-limit\nRelated-Bug: #1938966\nRelated-Bug: #1912460\nChange-Id: I943f610c3b6bcf05e2e752ca3b57981f523f88a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41531672e7b13794541dbaa7336393aaf5cf20ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41531672e7b13794541dbaa7336393aaf5cf20ce"}]},"branch":"refs/heads/master"},"c88a74dd3ba7015e389451739b26be1e5a1f6e80":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-17 04:36:54.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/13/804213/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/804213/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/804213/4"}}},"commit":{"parents":[{"commit":"96f1ea140edb1386159ce8b103c48b4e6a816ca1","subject":"Merge \"Remove ``ConfDriver`` code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96f1ea140edb1386159ce8b103c48b4e6a816ca1"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-06 06:11:00.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-17 04:35:54.000000000","tz":480},"subject":"Add table for pps limitaion","message":"Add table for pps limitaion\n\nTable 59 will be used for pps limitation, the pipeline change is:\nall original flows with ``goto table 60`` will be changed to\n``goto table 59``, while table 59 has a default rule is goto\ntable 60. Then we can add pps flows to table 59 for all ports.\n\nBasic limit pipeline is:\nIngress: packets get into br-int table 0, before send to table 60,\nin table 59, check the destanation MAC and local_vlan ID, if the\ndest is resident in this host, do the meter pps action and send\nto table 60.\nEgress: match src MAC and in_port, before send to table 60,\nin table 59, do the meter pps action and send to table 60.\n\nWhy table 59? Because for ovs-agent flow structure, all packets\nwill be send to table 60 to do next actions such as security group.\nBetween table 0 and table 60, there are tables for ARP poison/spoofing\nprevention rules and MAC spoof filtering. We want similar security\nchecks to take effect first, so it can drop packets before filling\nour limit queues (pps limitation based on data forwarding queue).\nAnd we do not want packets go through the long march of security group\nflows, in case of performance side effect when there are large amount\nof packets try to send, so limit it before goto security group flows.\n\nPartially-Implements: bp/packet-rate-limit\nRelated-Bug: #1938966\nRelated-Bug: #1912460\nChange-Id: I943f610c3b6bcf05e2e752ca3b57981f523f88a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c88a74dd3ba7015e389451739b26be1e5a1f6e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c88a74dd3ba7015e389451739b26be1e5a1f6e80"}]},"branch":"refs/heads/master"},"e028d45c1974caf807394a266cf5fb306efec228":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-31 02:31:28.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/13/804213/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/804213/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/804213/5"}}},"commit":{"parents":[{"commit":"701214481ff1df13b188d7826b44a1c8b51227ea","subject":"Use more pythonic approach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/701214481ff1df13b188d7826b44a1c8b51227ea"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-06 06:11:00.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-31 02:28:11.000000000","tz":480},"subject":"Add table for pps limitaion","message":"Add table for pps limitaion\n\nTable 59 will be used for pps limitation, the pipeline change is:\nall original flows with ``goto table 60`` will be changed to\n``goto table 59``, while table 59 has a default rule is goto\ntable 60. Then we can add pps flows to table 59 for all ports.\n\nBasic limit pipeline is:\nIngress: packets get into br-int table 0, before send to table 60,\nin table 59, check the destanation MAC and local_vlan ID, if the\ndest is resident in this host, do the meter pps action and send\nto table 60.\nEgress: match src MAC and in_port, before send to table 60,\nin table 59, do the meter pps action and send to table 60.\n\nWhy table 59? Because for ovs-agent flow structure, all packets\nwill be send to table 60 to do next actions such as security group.\nBetween table 0 and table 60, there are tables for ARP poison/spoofing\nprevention rules and MAC spoof filtering. We want similar security\nchecks to take effect first, so it can drop packets before filling\nour limit queues (pps limitation based on data forwarding queue).\nAnd we do not want packets go through the long march of security group\nflows, in case of performance side effect when there are large amount\nof packets try to send, so limit it before goto security group flows.\n\nPartially-Implements: bp/packet-rate-limit\nRelated-Bug: #1938966\nRelated-Bug: #1912460\nChange-Id: I943f610c3b6bcf05e2e752ca3b57981f523f88a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e028d45c1974caf807394a266cf5fb306efec228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e028d45c1974caf807394a266cf5fb306efec228"}]},"branch":"refs/heads/master"},"79683f7a34f9a88114f3833dd4b0ab2b0871d71f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-09-29 03:50:54.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/13/804213/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/804213/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/804213/6"}}},"commit":{"parents":[{"commit":"7c87ecb679928efffe4985b46c2305e638afcbe3","subject":"Merge \"Replace cirros 0.4.0 by 0.5.2 in ovn migration create-resources.sh.j2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c87ecb679928efffe4985b46c2305e638afcbe3"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-06 06:11:00.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-09-29 03:46:12.000000000","tz":480},"subject":"Add table for pps limitaion","message":"Add table for pps limitaion\n\nTable 59 will be used for pps limitation, the pipeline change is:\nall original flows with ``goto table 60`` will be changed to\n``goto table 59``, while table 59 has a default rule is goto\ntable 60. Then we can add pps flows to table 59 for all ports.\n\nBasic limit pipeline is:\nIngress: packets get into br-int table 0, before send to table 60,\nin table 59, check the destanation MAC and local_vlan ID, if the\ndest is resident in this host, do the meter pps action and send\nto table 60.\nEgress: match src MAC and in_port, before send to table 60,\nin table 59, do the meter pps action and send to table 60.\n\nWhy table 59? Because for ovs-agent flow structure, all packets\nwill be send to table 60 to do next actions such as security group.\nBetween table 0 and table 60, there are tables for ARP poison/spoofing\nprevention rules and MAC spoof filtering. We want similar security\nchecks to take effect first, so it can drop packets before filling\nour limit queues (pps limitation based on data forwarding queue).\nAnd we do not want packets go through the long march of security group\nflows, in case of performance side effect when there are large amount\nof packets try to send, so limit it before goto security group flows.\n\nPartially-Implements: bp/packet-rate-limit\nRelated-Bug: #1938966\nRelated-Bug: #1912460\nChange-Id: I943f610c3b6bcf05e2e752ca3b57981f523f88a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79683f7a34f9a88114f3833dd4b0ab2b0871d71f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79683f7a34f9a88114f3833dd4b0ab2b0871d71f"}]},"branch":"refs/heads/master"},"8eec718aac41aa5a9b5ef189d40aa89fc329516f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-11-05 09:33:14.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/13/804213/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/804213/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/804213/7"}}},"commit":{"parents":[{"commit":"899f254a091062d533beae303daabd6539954213","subject":"Merge \"Use the DB object when listing the SG rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/899f254a091062d533beae303daabd6539954213"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-06 06:11:00.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-11-05 03:06:34.000000000","tz":480},"subject":"Add table for pps limitaion","message":"Add table for pps limitaion\n\nTable 59 will be used for pps limitation, the pipeline change is:\nall original flows with ``goto table 60`` will be changed to\n``goto table 59``, while table 59 has a default rule is goto\ntable 60. Then we can add pps flows to table 59 for all ports.\n\nBasic limit pipeline is:\nIngress: packets get into br-int table 0, before send to table 60,\nin table 59, check the destanation MAC and local_vlan ID, if the\ndest is resident in this host, do the meter pps action and send\nto table 60.\nEgress: match src MAC and in_port, before send to table 60,\nin table 59, do the meter pps action and send to table 60.\n\nWhy table 59? Because for ovs-agent flow structure, all packets\nwill be send to table 60 to do next actions such as security group.\nBetween table 0 and table 60, there are tables for ARP poison/spoofing\nprevention rules and MAC spoof filtering. We want similar security\nchecks to take effect first, so it can drop packets before filling\nour limit queues (pps limitation based on data forwarding queue).\nAnd we do not want packets go through the long march of security group\nflows, in case of performance side effect when there are large amount\nof packets try to send, so limit it before goto security group flows.\n\nPartially-Implements: bp/packet-rate-limit\nRelated-Bug: #1938966\nRelated-Bug: #1912460\nChange-Id: I943f610c3b6bcf05e2e752ca3b57981f523f88a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8eec718aac41aa5a9b5ef189d40aa89fc329516f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8eec718aac41aa5a9b5ef189d40aa89fc329516f"}]},"branch":"refs/heads/master"},"053a9d24eca2c9372bd2b767504dd6b6db30643f":{"kind":"REWORK","_number":8,"created":"2022-02-08 09:24:37.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/13/804213/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/804213/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/804213/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/804213/8"}}},"commit":{"parents":[{"commit":"f324ddf769b9d9b68530a191dcf2e419ff2c25d4","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f324ddf769b9d9b68530a191dcf2e419ff2c25d4"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-06 06:11:00.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-02-08 09:13:13.000000000","tz":480},"subject":"Add table for pps limitaion","message":"Add table for pps limitaion\n\nTable 59 will be used for pps limitation, the pipeline change is:\nall original flows with ``goto table 60`` will be changed to\n``goto table 59``, while table 59 has a default rule is goto\ntable 60. Then we can add pps flows to table 59 for all ports.\n\nBasic limit pipeline is:\nIngress: packets get into br-int table 0, before send to table 60,\nin table 59, check the destanation MAC and local_vlan ID, if the\ndest is resident in this host, do the meter pps action and send\nto table 60.\nEgress: match src MAC and in_port, before send to table 60,\nin table 59, do the meter pps action and send to table 60.\n\nWhy table 59? Because for ovs-agent flow structure, all packets\nwill be send to table 60 to do next actions such as security group.\nBetween table 0 and table 60, there are tables for ARP poison/spoofing\nprevention rules and MAC spoof filtering. We want similar security\nchecks to take effect first, so it can drop packets before filling\nour limit queues (pps limitation based on data forwarding queue).\nAnd we do not want packets go through the long march of security group\nflows, in case of performance side effect when there are large amount\nof packets try to send, so limit it before goto security group flows.\n\nPartially-Implements: bp/packet-rate-limit\nRelated-Bug: #1938966\nRelated-Bug: #1912460\nChange-Id: I943f610c3b6bcf05e2e752ca3b57981f523f88a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/053a9d24eca2c9372bd2b767504dd6b6db30643f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/053a9d24eca2c9372bd2b767504dd6b6db30643f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
