)]}'
{"id":"openstack%2Fneutron~804380","triplet_id":"openstack%2Fneutron~master~Idcac4a29a4526b2e1041e8a41926ab2980f30d81","project":"openstack/neutron","branch":"master","topic":"packet_rate_limit","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-10-13 07:17:32.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-02-08 09:25:54.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-11-19 14:34:01.000000000","reason":"Removed by Lajos Katona by clicking the attention icon"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2022-02-08 09:25:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idcac4a29a4526b2e1041e8a41926ab2980f30d81","subject":"Reuse the qos constants from neuron-lib","status":"ABANDONED","created":"2021-08-12 10:55:26.000000000","updated":"2022-02-08 09:25:54.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0727d9de330e38d9d6cc762caa687047c1093a2","_number":804380,"virtual_id_number":804380,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-13 01:02:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-12 11:08:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-12 11:57:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-12 13:21:23.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-23 23:40:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-26 22:53:06.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 01:04:00.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-13 01:08:16.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-10-13 01:08:16.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-10-13 01:08:16.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"660f5d73e11b4f6fe896ef438b809e7a68df286a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-12 10:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b195642ab68278eda65bbc7579e799e9465d34cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 11:08:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3c9b392b7d64480935f03e01c610f7b : FAILURE in 8m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30fbdd16b61e415397ccf07d56939517 : FAILURE in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecee0d1b2c493281c34074833a43fa6779117f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 11:57:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d3de0357aa684ee18b1dc5a0e3fc4341 : FAILURE in 27m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b947e32c55cc492cbd8d7b14b675c8a4 : FAILURE in 25m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/721a8970cc9e4a89b0d67b0853452ddc : FAILURE in 24m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6fbe6f72e45e45bdbdac19af2951eebe : FAILURE in 26m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/042269005e5149b9b666c53099f06151 : FAILURE in 35m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf2b21b8fb664a94a972f0c3b2724179 : FAILURE in 26m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68ca6097f72a42139a85765f15f6ccc3 : FAILURE in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9590e289794849de91a1d49e2b826521 : FAILURE in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b67e221ab91c4a059657c3e45559cf5f : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31cb9e6bc0b1435a85362bdf88efee7d : FAILURE in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24582d392e9f47728f7c1f48a11814bc : FAILURE in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d0371ce4f0847aa86f7fa233d8284da : FAILURE in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fa9cf091bb43979e6bd2569a0b3a10 : FAILURE in 6m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/998d0654fc33415aaed0e17f3caee454 : FAILURE in 20m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc61dce3675644efa8355bf767b3e73b : FAILURE in 20m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6e9e5a4c157847bba523f9a5849aba7e : FAILURE in 26m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/227913b40c4242fe94444b918dc58ea1 : FAILURE in 54m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9a5f6b07229044689c31ee7e64259943 : FAILURE in 55m 53s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/64c51038d45c4750b8991b040f238638 : FAILURE in 26m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e685d6620be456dbc6b012f7866d0aa : FAILURE in 27m 56s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/29dae7b999c340d0b03287aa5fba8504 : FAILURE in 27m 34s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c41f3c713d774fc9a046a50ac1ed9f11 : FAILURE in 26m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6b2b8db131f148138f5ee8ddd39a205d : FAILURE in 24m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/325cd7338bd94a4f8fbb53d40b8f673c : FAILURE in 24m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa5c65f8c55c44d28ce42ee046168eab : FAILURE in 26m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d26d9b75ab454248abd6a1ff9c40884b : FAILURE in 30m 56s","accounts_in_message":[],"_revision_number":1},{"id":"065d9bb53b46e3b9bcd45a293b16b965c1c44f8c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-12 13:21:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/804380/1 : FAILURE in 2h 24m 38s","accounts_in_message":[],"_revision_number":1},{"id":"97c156a6ea00d520b969e6117b987f83aae66b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-13 08:15:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b4a306952a484ef2ded77d1810ebdf1b1a1ebaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 08:27:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f13d3db3a2194aaba245cdcc47d03fec : FAILURE in 9m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/056f5409d1e54e6faa6df3dab796168e : FAILURE in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd468a2ca20235b64764e5a3eba2ac17dede0ed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 09:16:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b6b6852299714018b213e25e4219a010 : FAILURE in 26m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9fc181fa9c0f42c58e762ff7576c8cba : FAILURE in 23m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6a44ac0ebde24ec0b88dbd1bb91cfd4f : FAILURE in 24m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/09ad0119f7bf462b9b175410e72f8263 : FAILURE in 25m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b9c711aa04d843e087b1d628e9f43b13 : FAILURE in 25m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31294925e6884f5f9c8b5c48437f71b2 : FAILURE in 24m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5f12cc0253846b39e5a0f4dd07161f6 : FAILURE in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c85cdf38635f44f49dc50497f376098d : FAILURE in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80ca39f37fdc46e2943eaf891065ccdc : SUCCESS in 14m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b5a7d3b7b6b40ca91d505cbb90b00ba : FAILURE in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d726144284254dc1a7c810125b05090c : FAILURE in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/191e578e31bb46e1aecdf8464538986b : FAILURE in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11064fad19af4f0aafaca931a96f8502 : FAILURE in 6m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/90b10503208642399962c5bbb8f605d2 : FAILURE in 20m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9b70fcd72304de18cf4e95c4c0f5461 : FAILURE in 19m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2cdc474af6a24a79bfb0f7650e9347ef : FAILURE in 24m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b09bdd85c604c6186f7d5ccf8154bc1 : FAILURE in 54m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7312900067584a679e343036eb99ac32 : FAILURE in 55m 35s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/64900b32f80e44cb8bbc68c06bdd3157 : FAILURE in 28m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3ab1bed9eb4c4279adcbb69e9a88997c : FAILURE in 26m 48s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/52a97350b4934b759c67568352886e08 : FAILURE in 30m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/885094a65ecf4c109bb3efd31aa6d3b7 : FAILURE in 24m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6439d84bd5064f9aa2a3e044a323e6e7 : FAILURE in 24m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0e4959ffa0374d04808e46f82707fb06 : FAILURE in 24m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fbcbdbabaeb648d09441031ac2193357 : FAILURE in 24m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2e2ae0a5ecbf46c39576df214544b1bb : FAILURE in 31m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ba0cc17070a31c0e105ba205cafac77287061ba0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-13 09:46:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804380/2 : SUCCESS in 1h 30m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f6fe5d69ec7de0a4c266d719e52ff4bbf22b1e67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-17 04:36:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0573ed1fa0dc4515aedf96e520b90310656cc477","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 04:48:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75ec95ed97c14bfaab6dcbea608974b3 : FAILURE in 8m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/354a354f4de1496eb836785f792aa56a : FAILURE in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e46f77b47213300a2ee6786da149b3a2f7711efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 05:38:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f2694591dc9548f4a4c066a3a619d216 : FAILURE in 27m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6130b30bb9b049948b359ebcb7dfdbf5 : FAILURE in 24m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/74cca1f8494340379e1f6fe0df8e1efe : FAILURE in 26m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3bc3acab0bab4c79a091e8827ba0f2ed : FAILURE in 25m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9a6778b95b8a4cd9a815f8ca89b8394c : FAILURE in 27m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4754babe44d64451baa5cd739b5185b0 : FAILURE in 25m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc92468a38624281a415082c653157f2 : FAILURE in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/109c06b8ec884f8984ef473d9d27aa3d : FAILURE in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b761d3dd9d6f436ebc397886d142c663 : SUCCESS in 16m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32747580f1d94f8cb3df112d6fca3ee0 : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a67ed1806fc42a5a641fae33c9d4035 : FAILURE in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cf03d13a0b34235b0ddec0e9cc7851e : FAILURE in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e9cfb635764fa1a85745773ee2d9dd : FAILURE in 6m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49cb1dd31ae847a5939eab3fe775724f : FAILURE in 19m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2ca8b954a9b41c1a1f15d1909b1b96c : FAILURE in 21m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/03562a4041704970af530f0b7fd1ce3a : FAILURE in 19m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/28fe0b37b15a492ba654615d5310bd94 : FAILURE in 53m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e031c0acb8524619b2ab43a9b5eb1937 : FAILURE in 56m 10s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/10fb16efe5ca439ba9549541303a7fa5 : FAILURE in 27m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2d6866dbe041401099e4ed01d06dc21d : FAILURE in 27m 22s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/f7dbd0551905418a9771b12b8dfb6edd : FAILURE in 27m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40c79df4a4d949a19295f2328f4890c4 : FAILURE in 23m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9b347e518e1b471eaa2a4bfa1619915d : FAILURE in 24m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/710fd4521fb64c1694d06cde9dd92c92 : FAILURE in 25m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a639e0fa8ad9440a9d5bc0d522574d8a : FAILURE in 25m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/32b924d533474a66b588a408a43d5d66 : FAILURE in 32m 25s","accounts_in_message":[],"_revision_number":3},{"id":"e889543f4052bff40e654782dfc89c66a1bd14c2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-17 06:28:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804380/3 : SUCCESS in 1h 49m 54s","accounts_in_message":[],"_revision_number":3},{"id":"f9d68ce29dae0184385e805164a76111087a6a05","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-23 23:40:23.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/80/804380/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"929c2f26539be198094dbe29dde84e7759652dd4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-23 23:40:45.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/80/804380/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8b44592f9e8823a6a6831f19a7d1ee1cb9b38608","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-26 22:53:06.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/80/804380/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b4b675f514b92cc42b9e890c8e7e893c0ba786ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-31 02:31:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3714ae816c5369eca81d3e06ff1a826d3fdb91dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 03:05:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b372a3fa1294577a94d6db03541f2e2 : SUCCESS in 33m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cafce458196e4755845610a8e9dddb6b : SUCCESS in 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9f0e1e259b0a406450d9031a8c1bcee6d5e65f40","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-31 04:05:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804380/4 : SUCCESS in 1h 32m 48s","accounts_in_message":[],"_revision_number":4},{"id":"00dcf005968c68ced4f85b3c64b993a101102dee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-31 04:22:46.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/80/804380/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ba17f75b96bdf2482bc246f3e62d8c0bf62cf8f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 04:55:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e5259730d354427c9a840c297e788769 : SUCCESS in 46m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2b8b75143cc7474997b594bceca289f2 : SUCCESS in 2h 18m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2d15949481c941dbb444eadee59e0d4a : SUCCESS in 1h 54m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de333b65712f4523a5470d70225ad741 : SUCCESS in 1h 27m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/57029735724046c2a06ba0e8a02d6722 : SUCCESS in 2h 16m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad6a9d5491044d6f8d19f834f3209593 : SUCCESS in 35m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e1c01328f214fb68851d4336e358945 : SUCCESS in 34m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb599b8b6c584dafb188a6f39c156f0e : FAILURE in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c50ff63ced4eccb6986e0cbaf4f096 : SUCCESS in 13m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b05652746de4df2bbbc972b58fff1be : SUCCESS in 28m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e98c11795ab5416ca3e39d618c040c9c : SUCCESS in 20m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6adda748933249a7ad7fe5398e0dab40 : SUCCESS in 22m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24011388081a4785967041cb3993e26e : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c94a8de25a04b7b9ad31a34c436018b : SUCCESS in 49m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/56ae894b2dfa474b9e087747553decdb : SUCCESS in 1h 44m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/80a9738678904fb68ac1173c852e28d2 : SUCCESS in 1h 11m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1026c51ea4c74e8189bdc8e5c282bb22 : SUCCESS in 1h 12m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ea9bbc728dd542e296c91c5966a073cb : SUCCESS in 56m 36s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/de7deaf661fa4f9da219197dbe99110b : SUCCESS in 1h 07m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/89e033fdf62847f08ee872a7a6d2a7c6 : SUCCESS in 59m 04s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/406074c7e65b44bdbea90a38d1a58549 : SUCCESS in 1h 28m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07f074b3809640d7bdd4f1bfe98b88af : SUCCESS in 47m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9179d3af5bbd4e4eb30cf6b5457772c8 : SUCCESS in 54m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5eb0a3c5d8da49efba74b81a701bda90 : SUCCESS in 32m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6961e594bc474727a4abb5b908c451dc : SUCCESS in 1h 06m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/19c5a6016c1147458a5352cfa1e31d65 : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c00db32cadf6ddd2d8063c814d476d35ea126b7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-08-31 05:27:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b372fee2155236548d990cd5b8223a48089ee96f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 06:19:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ff62d83980c4d2a80713acf0adaf724 : SUCCESS in 42m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0978c9b8413b4773b09fd6669108769e : SUCCESS in 46m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ed1d274034cadd80f78bc3e991d4fcbf81aa481c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-31 06:21:34.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/80/804380/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"abd682c8ccd7da743dfca29a7983031b316a6ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 07:39:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8a11e75d3763464e88f165d9101d65d5 : SUCCESS in 48m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a7a672641fd34b659b342881c5659d37 : SUCCESS in 2h 10m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3a85869208d24b309ff8fa711135d1d6 : FAILURE in 16m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c4b0650020554293aaa65d9f5b017994 : SUCCESS in 1h 25m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b5ef89b6221448e09bcdf45fdf61e0cf : SUCCESS in 1h 49m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/196d4b0356f8403e87419ef94b24c2c5 : SUCCESS in 28m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/888a83b89c554afba4d01b64019abb70 : SUCCESS in 37m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8738d17f212c40a0bed8ca88b7c83d4e : SUCCESS in 39m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08246994e5f940439093789940c60004 : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54ce548f8ae94881a0be3ca80096dfbe : SUCCESS in 25m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85a3b12345a74342a0d973cdf291f723 : SUCCESS in 31m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f67207378664a71850d06d7f7a84648 : SUCCESS in 31m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48014865eb44cc4851cc07d06fc0111 : SUCCESS in 11m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/84214a3314b54d6f8a6fa977be1e1415 : FAILURE in 3m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/012ec2df270b4e688f75b28d442fbfcb : FAILURE in 57m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0d9205224ca48a38b68d3592e8461df : SUCCESS in 1h 45m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1e3252994e42492a870ccb80439698de : SUCCESS in 1h 09m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cabece16d54e4dfbb8d6fa5f9d0a569a : SUCCESS in 58m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d51c062984bd47c4b566ad9c90461332 : SUCCESS in 1h 10m 45s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7eb1ae5c5b864c4f9507a3096bf3b5a9 : SUCCESS in 1h 07m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c5eadf4985e045fea260b60cf87a511b : SUCCESS in 1h 22m 03s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/de98a6c8004444e68add0f0649813369 : SUCCESS in 1h 25m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f9b0b8be0a14ec7bb24f555d495c957 : FAILURE in 50m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ef7aa5935fb049b6a6908538691c8260 : SUCCESS in 52m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b569b648c02c498fb564602906e1bfc4 : SUCCESS in 33m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27e2a7a0eb444207994573eda1c19bf1 : SUCCESS in 1h 28m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a4c201d08d7e45ff83d262a3adc477ff : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":5},{"id":"95e673b59223522b39f05e86417f01d8bd7a3d47","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-31 07:42:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804380/5 : SUCCESS in 2h 13m 33s","accounts_in_message":[],"_revision_number":5},{"id":"25db92ced87828adf1bc728c922805144ddc935d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-09-29 03:50:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cc995ad7204d2902c79b2a37e1bf1514c9ce4342","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 04:43:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74f3704f50894981a86be9fc1393bff5 : SUCCESS in 46m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/034c848a3e594cda9b029cb653065399 : SUCCESS in 45m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eabaeb0ebf1402b0be1207282136e17cef1dc3dc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-29 05:40:53.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/80/804380/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"79426b9c87a7e0615e972798ba96c22f37970134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 05:43:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/013934d09a754546b9a2466dd0b73e2f : SUCCESS in 41m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9ad86d80f9344f62aaad2b12c89c949e : SUCCESS in 1h 47m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2e9112355b97484891d4a64f2d168cb5 : SUCCESS in 1h 25m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/262adfe2e8b040218b8145d05c7ccac3 : SUCCESS in 1h 21m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/52855498f81a4facad2ecfaec4434cc0 : SUCCESS in 1h 40m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5eefb5cc49a44f8abe410168de547aed : SUCCESS in 27m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fdbf79b050b4f7d867205390bab290b : SUCCESS in 32m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5067235d265f483e8e97ff03cdf492d8 : SUCCESS in 41m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc86b93db9aa4d719604ecc86dd51004 : SUCCESS in 13m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7826f4a17131438f93d87cbad7143d81 : SUCCESS in 28m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb94fc3def6c419d903eab818bfea4f4 : SUCCESS in 22m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa3e1aa9145d411bb338e62c9b853efd : SUCCESS in 28m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aba3e7eb0bb341148fa56adc6c5ee76e : SUCCESS in 10m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96c9c2c768534cbe929f13624f816cf9 : SUCCESS in 49m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/44c054506bdc4872a6b75ab20f5a8748 : SUCCESS in 1h 49m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d824f4868c5c47a2927657b653caa1bc : SUCCESS in 1h 02m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c37d4a3177d44fb398edd5b11272f574 : FAILURE in 43m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bddff2af00f14701bfdc8c50816af9da : SUCCESS in 53m 12s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cf101a5dfe7449b6ba80859cd56fdeb1 : FAILURE in 53m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/06d1561f797c449d9806ebe806d6fa4d : SUCCESS in 1h 10m 26s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/260b36c44bc843a99ec77347f0bfc1d6 : SUCCESS in 1h 21m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b5fb277a27d4237aba103285ded4895 : SUCCESS in 51m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d5b6b212d1a645f7bf2770957e669cba : SUCCESS in 58m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ece4118aa4e84cf6b9add965471825b7 : SUCCESS in 28m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a763b5490b904a8a8951b9497b7df79f : SUCCESS in 1h 09m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f137782c6afe47089b75bee9208d7b47 : SUCCESS in 1h 12m 41s","accounts_in_message":[],"_revision_number":6},{"id":"0b57a95856a625802eff19833df0266f5b73dd02","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-09-29 14:52:34.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"64b9716365dfab3269014f367df67ed28a5ace6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 15:35:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5f0c486412a42aca4a386eaf6332b7a : SUCCESS in 40m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a93817f82694693aa34c066eb153214 : SUCCESS in 38m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e3746986494c9f6cfbfc9e8ad4fc286787b081d5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-29 16:21:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804380/6 : SUCCESS in 1h 27m 39s","accounts_in_message":[],"_revision_number":6},{"id":"effe31bb6a6a5248232500ec1b3c2236b1f6d057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 17:25:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2a5cc0f4e6484d9cb3eaeacae67aedd7 : SUCCESS in 51m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/07c1098a3efc476abcb71d4efa650662 : SUCCESS in 2h 27m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d4148bdaa8f94c4696f6eef22415744c : SUCCESS in 1h 51m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d709e65cef04b089e82e39c09b2d59e : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c22a8c16e6e240b893031957ded9e3d9 : SUCCESS in 2h 08m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70dd8a855a8d41c681a275d5d799b6dc : SUCCESS in 48m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3614baf8acad472887862a095561ea3c : SUCCESS in 34m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fcea53d80f646dc90b5292af0c2a2dc : SUCCESS in 45m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60958ff9e3e74e34a9bf8d9e180ae77d : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d9e1f81a59d4bc88caa77fd7339a3e4 : SUCCESS in 28m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/604c3b545d1948fb8f6a811c441f6377 : SUCCESS in 24m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/814a2d7843204f89a22422309cad964f : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8dcba7b19d4c4583f9646a056e3d87 : SUCCESS in 11m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/52f1e551808d4f1ea771e5a6b8f60dd9 : SUCCESS in 58m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6470533358f4a6fa573251987103d8b : SUCCESS in 1h 36m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/858dbed7aad04acb8b7fd21a1eb2f9f6 : SUCCESS in 1h 15m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/98f59bfca8e54b22ac66e7393bf999f4 : SUCCESS in 51m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a431d4e84f22485ca46179e4dc9b5d4a : SUCCESS in 1h 02m 27s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d5aa1803f02e46f28f1b4ba2b94c0648 : SUCCESS in 1h 19m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aa82b4aa3d074717823cdf946832d564 : SUCCESS in 1h 14m 08s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/6d0146d764ee4fd38269ceed1003d324 : SUCCESS in 1h 22m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9de6534f19914d2d9b32b0b80b59f18d : SUCCESS in 55m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/67ba9b94bab148b68ff6ca592b5f4c9e : SUCCESS in 55m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/14f5679d7e6b4542968be046725fe83f : SUCCESS in 49m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/79ab6a6301024a07b7a5a16a0a1602d8 : SUCCESS in 1h 10m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/26775c1b37674a8aa5e023823856e59a : SUCCESS in 1h 31m 57s","accounts_in_message":[],"_revision_number":6},{"id":"1b4474673b144391207d0c4ba0695371b4dcc09b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-10-13 00:53:58.000000000","message":"Abandoned\n\nThere are continuous developments for these constants[1]. Seems neutron developers are happy to add new constants in neutron. Looks soomthly, not involved the neutron-lib. So left this as it is.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/803045/15/neutron/services/qos/constants.py","accounts_in_message":[],"_revision_number":6},{"id":"42fabf20eb45a7445ffb047036b76898a842881c","tag":"autogenerated:gerrit:restore","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-10-13 01:02:05.000000000","message":"Restored","accounts_in_message":[],"_revision_number":6},{"id":"f3d748741d3de80fe95c67972743333d19e753c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 01:02:41.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"e2f698bcf0a73c8b6a5716b9a00c3920cd1e39c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 01:02:51.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"fb6906d8f27f722c8c60bde15681e473d33b3474","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-13 01:04:00.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"de3a96ae1a53b5efc95a477a0026cd4e0264185b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-10-13 01:08:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"06125e277ff05f209c272394a030f2f5dde36d18","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-13 07:17:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e0727d9de330e38d9d6cc762caa687047c1093a2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-02-08 09:25:54.000000000","message":"Abandoned\n\nRestore if someday we want this.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a1a664a5cc782bd60c53a184c5ac1ce13a4e0ee9","revisions":{"793cf3a2cacd9dc9f41dded8f4915de3f159a9d4":{"kind":"REWORK","_number":1,"created":"2021-08-12 10:55:26.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/80/804380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/804380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/804380/1"}}},"commit":{"parents":[{"commit":"b76ec9f7511e92498941448b0831060e28b571db","subject":"Add table for pps limitaion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b76ec9f7511e92498941448b0831060e28b571db"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:44:51.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:50:35.000000000","tz":480},"subject":"Reuse the qos constants from neuron-lib","message":"Reuse the qos constants from neuron-lib\n\nBecause neutron must implement the new QoS rule first,\nthen neutron-lib could add the type. After [1] was get\nmerged, we can now use qos related constants from neutron-lib.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/796363\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/804378\n\nRelated-bug: #1912460\nChange-Id: Idcac4a29a4526b2e1041e8a41926ab2980f30d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/793cf3a2cacd9dc9f41dded8f4915de3f159a9d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/793cf3a2cacd9dc9f41dded8f4915de3f159a9d4"}]},"branch":"refs/heads/master"},"b7e31fe4df7d76c24d85d82f537ffa50371f69a0":{"kind":"REWORK","_number":2,"created":"2021-08-13 08:15:06.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/80/804380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/804380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/804380/2"}}},"commit":{"parents":[{"commit":"41531672e7b13794541dbaa7336393aaf5cf20ce","subject":"Add table for pps limitaion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41531672e7b13794541dbaa7336393aaf5cf20ce"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:44:51.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-13 01:40:38.000000000","tz":480},"subject":"Reuse the qos constants from neuron-lib","message":"Reuse the qos constants from neuron-lib\n\nBecause neutron must implement the new QoS rule first,\nthen neutron-lib could add the type. After [1] was get\nmerged, we can now use qos related constants from neutron-lib.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/796363\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/804378\n\nRelated-bug: #1912460\nChange-Id: Idcac4a29a4526b2e1041e8a41926ab2980f30d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7e31fe4df7d76c24d85d82f537ffa50371f69a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7e31fe4df7d76c24d85d82f537ffa50371f69a0"}]},"branch":"refs/heads/master"},"09ef6dc894a2aec7a4ad4caa2a27f86713649c27":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-17 04:36:54.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/80/804380/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/804380/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/804380/3"}}},"commit":{"parents":[{"commit":"c88a74dd3ba7015e389451739b26be1e5a1f6e80","subject":"Add table for pps limitaion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c88a74dd3ba7015e389451739b26be1e5a1f6e80"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:44:51.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-17 04:35:54.000000000","tz":480},"subject":"Reuse the qos constants from neuron-lib","message":"Reuse the qos constants from neuron-lib\n\nBecause neutron must implement the new QoS rule first,\nthen neutron-lib could add the type. After [1] was get\nmerged, we can now use qos related constants from neutron-lib.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/796363\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/804378\n\nRelated-bug: #1912460\nChange-Id: Idcac4a29a4526b2e1041e8a41926ab2980f30d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09ef6dc894a2aec7a4ad4caa2a27f86713649c27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09ef6dc894a2aec7a4ad4caa2a27f86713649c27"}]},"branch":"refs/heads/master"},"984d44f5dd9f1b39907ce5980b3faedaa0acc7e2":{"kind":"REWORK","_number":4,"created":"2021-08-31 02:31:28.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/80/804380/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/804380/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/804380/4"}}},"commit":{"parents":[{"commit":"e028d45c1974caf807394a266cf5fb306efec228","subject":"Add table for pps limitaion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e028d45c1974caf807394a266cf5fb306efec228"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:44:51.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-31 02:29:02.000000000","tz":480},"subject":"Reuse the qos constants from neuron-lib","message":"Reuse the qos constants from neuron-lib\n\nBecause neutron must implement the new QoS rule first,\nthen neutron-lib could add the type. After [1] was get\nmerged, we can now use qos related constants from neutron-lib.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/796363\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/804378\n\nRelated-bug: #1912460\nChange-Id: Idcac4a29a4526b2e1041e8a41926ab2980f30d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/984d44f5dd9f1b39907ce5980b3faedaa0acc7e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/984d44f5dd9f1b39907ce5980b3faedaa0acc7e2"}]},"branch":"refs/heads/master"},"5e17e1bfb1716d298c86ba9a9083bf16a2cfabb8":{"kind":"REWORK","_number":5,"created":"2021-08-31 05:27:22.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/80/804380/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/804380/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/804380/5"}}},"commit":{"parents":[{"commit":"e028d45c1974caf807394a266cf5fb306efec228","subject":"Add table for pps limitaion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e028d45c1974caf807394a266cf5fb306efec228"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:44:51.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-31 05:25:18.000000000","tz":480},"subject":"Reuse the qos constants from neuron-lib","message":"Reuse the qos constants from neuron-lib\n\nBecause neutron must implement the new QoS rule first,\nthen neutron-lib could add the type. After [1] was get\nmerged, we can now use qos related constants from neutron-lib.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/796363\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/804378\n\nRelated-bug: #1912460\nChange-Id: Idcac4a29a4526b2e1041e8a41926ab2980f30d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e17e1bfb1716d298c86ba9a9083bf16a2cfabb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e17e1bfb1716d298c86ba9a9083bf16a2cfabb8"}]},"branch":"refs/heads/master"},"a1a664a5cc782bd60c53a184c5ac1ce13a4e0ee9":{"kind":"REWORK","_number":6,"created":"2021-09-29 03:50:54.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/80/804380/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/804380/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/804380/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/804380/6"}}},"commit":{"parents":[{"commit":"79683f7a34f9a88114f3833dd4b0ab2b0871d71f","subject":"Add table for pps limitaion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79683f7a34f9a88114f3833dd4b0ab2b0871d71f"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-08-12 10:44:51.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-09-29 03:47:11.000000000","tz":480},"subject":"Reuse the qos constants from neuron-lib","message":"Reuse the qos constants from neuron-lib\n\nBecause neutron must implement the new QoS rule first,\nthen neutron-lib could add the type. After [1] was get\nmerged, we can now use qos related constants from neutron-lib.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/796363\n\nRelated-bug: #1912460\nChange-Id: Idcac4a29a4526b2e1041e8a41926ab2980f30d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1a664a5cc782bd60c53a184c5ac1ce13a4e0ee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1a664a5cc782bd60c53a184c5ac1ce13a4e0ee9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
