)]}'
{"id":"openstack%2Fneutron~804571","triplet_id":"openstack%2Fneutron~master~I9f4763efc02e5cacca3a00e9f101f83d3f3fb1e7","project":"openstack/neutron","branch":"master","topic":"bug/1940009","hashtags":[],"change_id":"I9f4763efc02e5cacca3a00e9f101f83d3f3fb1e7","subject":"Add missing options to generated neutron.conf","status":"MERGED","created":"2021-08-14 14:31:37.000000000","updated":"2021-08-18 10:20:42.000000000","submitted":"2021-08-18 10:18:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"804571-bug/1940009","meta_rev_id":"978a4202ae14423ae3a041f31544a352d52b2565","_number":804571,"virtual_id_number":804571,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-18 10:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-08-18 07:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-08-16 09:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-16 08:49:57.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-08-18 07:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-14 15:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-14 16:19:06.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-14 17:01:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 07:24:35.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2021-08-16 08:49:57.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-08-16 09:35:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-18 07:39:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"2da27a746491b9e19547c04bd583c20f63420b41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-14 14:31:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b56d083be19fbc6d827b430ab8229fa1df1e92a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 15:12:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11466296c4a44416b1e3b83f14d52fd6 : SUCCESS in 38m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/125ccf15ae714d648f902d5992c40a72 : SUCCESS in 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b14e264f62145ffa7daed94f82e036771aa0c554","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-14 16:19:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804571/1 : SUCCESS in 1h 46m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c975fec24afca8e5fbea2ad48536b04e045ac155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 17:01:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/767f5763f0d7485b8a147c43871304d4 : SUCCESS in 49m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f690ca0cb5f14d5b8b1c8eb569c0dd81 : FAILURE in 1h 40m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5885a11911bf4c25b1c9b1b06ca891ab : SUCCESS in 2h 28m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/741851f883364740849ecfb69a7ddc90 : SUCCESS in 1h 33m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/db0edb4329364cf0a8ff138f4eed5e36 : SUCCESS in 2h 02m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3740a1125f2a46589be4fc5f2b45498f : SUCCESS in 31m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cacf77b6154476bb51f471cbb893e06 : SUCCESS in 38m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bbb0e9309e54faa83a1ad4c3c34d69c : SUCCESS in 42m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead356b5d2cc4c7db51852ab8f1cd3b0 : SUCCESS in 16m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/486036e9bebd4a16ab709b2530caa8d3 : SUCCESS in 29m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c2dfcf346cc4f6a98dea6118b86d72d : SUCCESS in 27m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/764057d43cad4018a309b0c9f210ade4 : SUCCESS in 26m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4addf2da65c24baf8f68b85ef4925a08 : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebc023083f29411b86d68fa2f2a9e5af : SUCCESS in 50m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f281b067cd0c4648bc78bb4176fcd85f : SUCCESS in 1h 41m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1c7d5e74234944d0a6702c0e4db98b51 : SUCCESS in 1h 05m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3b4bf9bc9164844b75a7bbeec876084 : SUCCESS in 1h 03m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e7fe5cda92fb41d58c02c3f92e413cd6 : SUCCESS in 1h 11m 18s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eab3f52976de465395c9ddf2f1ebf070 : SUCCESS in 1h 21m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/420ee08b4ae04f4cab33e55c69d658cb : SUCCESS in 1h 12m 24s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/825ed23fb55d407f8ab9154c73f5873f : SUCCESS in 1h 26m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/755b7a2d711647b6b36d3ee8f25d6e2b : SUCCESS in 53m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/16eda5259b11412896efde2924382b75 : SUCCESS in 53m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/deb963d079f84b2d9abe36028e609f8c : SUCCESS in 31m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa96b5fba19f46fb8ad3bf176c441592 : SUCCESS in 1h 07m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4640e10afcdf46de978b22c3bb2cd113 : SUCCESS in 1h 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b7d60c76651a0c449bb0f9d904f9f6b72202ecdb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-14 17:39:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3ecc32b46de5152c3866079ac1c5823fec64f88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 18:23:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2db47ff6e9a4987aafa0a01555c7e98 : SUCCESS in 37m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5acb56a789a4aef8fc10c313b5daf31 : SUCCESS in 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54a50992b4a46cddf9bac33e89704d439534084b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 19:27:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2e2196465e8c47239d9de04915ad0c1f : SUCCESS in 38m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3f537f09b6b7455d92c4ea6dedb10a8b : FAILURE in 1h 45m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/406f5855b06e4be385ee020f1e7ceb84 : SUCCESS in 1h 42m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5d57de32546d4f3cb3b89d2f6ad25ff0 : SUCCESS in 1h 21m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/db963dced2964b08b8fd1ecf88a186b1 : SUCCESS in 1h 42m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6adaafa948e4485fabc467ffcc2530fc : SUCCESS in 26m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed7086f5f9b74146b6a37700c24f1622 : SUCCESS in 35m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f290ed04ee5546cabddf81734bae695c : SUCCESS in 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db1a3d949a046f89ffddac1ab07d7fb : SUCCESS in 11m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86a36c3c881442c98f06bbc1916202ee : SUCCESS in 24m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/520933d647e3422ca019d46a62bf3256 : SUCCESS in 21m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe189e3e22d54c0693a57a35aaba9501 : SUCCESS in 28m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b73def41e5474e0f831e736bce2a092a : SUCCESS in 9m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9f98bec207942fa9f10829dd7428eb2 : SUCCESS in 48m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/12f64c0951d74414a3752d35878d7b01 : SUCCESS in 1h 37m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2a1ba3500a4249328eeedb7b4df0e20e : SUCCESS in 1h 07m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a25cb5ab4cc41d99e7c5a69d5f8aed8 : SUCCESS in 56m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/179bf11215ac4fd481341e3bec2d3f0c : SUCCESS in 55m 43s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4e91effc08b74f57a24b654e737ac5e7 : SUCCESS in 1h 11m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/41a3073fe9254bb08aaec5b1442f5d21 : SUCCESS in 1h 03m 09s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/f1176d49ed244c2c821c8facfb9ac0e7 : SUCCESS in 1h 15m 31s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aaba637fdab445e887a2a2e3f8f6eb06 : SUCCESS in 55m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/39c910303a7a4802bd0fc7ee138e63ef : SUCCESS in 1h 01m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/66c6f92c8e864175a590cfd38c57ae0e : SUCCESS in 30m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c70216533e6f4158a157db9713c063f4 : SUCCESS in 1h 14m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9acf37ebd6cb41d3a90948e4ada8ddcc : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f5572732ce7b5a5ce8eb2a8b799035dbc7f954f8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-14 19:29:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804571/1 : SUCCESS in 1h 48m 36s","accounts_in_message":[],"_revision_number":1},{"id":"525309e9a2fdf29b33f603937073e92b386b1ddd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-15 07:09:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a721809a03341ba4b280a7b99126e521d645e85d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-15 07:49:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c8562d02c4e4c03bd7a682c422be49f : SUCCESS in 33m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef7d35efb4a4489194ed9fc810882a28 : SUCCESS in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"421af625aacf305a83f5f91e2076941f31c54eba","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-15 09:04:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804571/1 : SUCCESS in 1h 54m 18s","accounts_in_message":[],"_revision_number":1},{"id":"49864b641f9dc86e50ee749e2ca6d719f9f2b295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-15 09:58:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/097616cfff1c4f71a66e7f3e18f9cdea : SUCCESS in 44m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0b1e03da0497408cb72c869cf58fa395 : SUCCESS in 2h 43m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dec22ca6f39048a0ab4115a307ce84df : SUCCESS in 2h 07m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/065ae094b48848f0bd704199c6f3c04a : SUCCESS in 1h 41m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6c906b2664ef4ad1894af44deb3b01c6 : SUCCESS in 2h 03m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3fa25e81534646f5b0fb3eab00ce8ca3 : SUCCESS in 32m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33d413a0651247e9a6f32d18efbce45a : SUCCESS in 36m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14cc05dbdb114dd781a4df7cd6bbb617 : SUCCESS in 40m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5b9b4f9d514a23977f16a124789043 : SUCCESS in 14m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a58df9339374b24964e67031603c5f2 : SUCCESS in 26m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb8e4d59dfa4486aac6859777a736cbb : SUCCESS in 23m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfd28451ff954066b91da9b3dfd9bb6e : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8576713a83af4ab1b3bf58b2e1452ac9 : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db834f2f694646a38e9f902a8e2677d9 : SUCCESS in 46m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9e52d1969d3485fb41ac662fccd84fb : SUCCESS in 1h 41m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6d3bd7fb1c1940dcb227d5852770c0de : SUCCESS in 1h 07m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9799e91419e2403eaa33cb693a53fd5a : SUCCESS in 1h 04m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/467050df98734784ae1bff6bacf36ca5 : SUCCESS in 1h 07m 28s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0adce895827a41ed8252e065a2737585 : SUCCESS in 1h 09m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/674dd2789858429b94dee074b8f9e3e6 : SUCCESS in 1h 11m 51s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c65ef48e355440c3b5aabb0f0f6a8302 : SUCCESS in 1h 28m 04s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/645a7826885c4e7d8bcbc10a9e691da6 : SUCCESS in 53m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6782e596ae8c4b7e8be0175c1c55893b : SUCCESS in 55m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/183a9192ac744ffdb57107d62cc684e5 : SUCCESS in 32m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9785653646c440c89aa2cc15eb558c86 : SUCCESS in 1h 07m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4b008b255ca94a06baf7d827bd1346d9 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fa9be1140a37e3ab986e6821c5492273af47c485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 00:17:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a4a3cd60d51f7cb544c39f907b576df573045458","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 00:59:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c94d797e00ce42229c8ac23a43cd51dd : SUCCESS in 35m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8220b6a2cf8547cc8ca70dc7ce15103d : SUCCESS in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b8f91e0eaea936bbbe192fd70c9cc51a6f46a1e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-16 02:10:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804571/2 : SUCCESS in 1h 51m 55s","accounts_in_message":[],"_revision_number":2},{"id":"38f53dc64ffc45fdd198c58619691f093717a643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 02:40:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/17779186b5af40c9873a811efabbb35d : SUCCESS in 42m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ae753111502545c2b49e02124e1332ef : SUCCESS in 2h 13m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ce4b0c9676ba4a70a18d333a36a05b97 : SUCCESS in 1h 58m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/582cecb3dff34b3caeab9a505ab8916f : FAILURE in 1h 45m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dae82348b08f4e89801f26f3194c7caa : SUCCESS in 1h 45m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cc57114a5bc04871b9b43c36f34053be : SUCCESS in 36m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcbebe6ed4284b47a6ba9462d655447a : SUCCESS in 29m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a17a4101d1544b6582ce31685e86259a : SUCCESS in 41m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbdef7d30aa47e696306235fb5715d8 : SUCCESS in 15m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6553783f8bc44775ab29e5d40e1f0450 : SUCCESS in 31m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c8e0d5a333246c4abd4b3e20894876a : SUCCESS in 25m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/114420e93ee0493196ca0db42ed95406 : SUCCESS in 25m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd62d588bd95412c9364521171cbe088 : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dcebdfa7e37f4c21a525fa0db79d31f4 : FAILURE in 44m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7dfe9a21aa645e28e1c2c6929d8b65d : SUCCESS in 1h 43m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e869523305b244b8bc1426d0f9003cb3 : SUCCESS in 1h 05m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc70a9dd5fd343db96f2654b9ffccd6d : SUCCESS in 1h 00m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9558529157184c37b97e46c4734d391e : SUCCESS in 57m 49s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/15d0af7164a74372bcb6941ac8fa1fc2 : SUCCESS in 1h 10m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/06dfbeb9489d4f228ad53fae9fac6a7a : SUCCESS in 1h 14m 08s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/517abe419ea44c17888fd12f122b9765 : SUCCESS in 1h 19m 43s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9608730b5bad46fcb3a0e3cf1dfc9318 : SUCCESS in 52m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/760b3cbfc9ff4651b97bb6fcdef9fccd : SUCCESS in 53m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d2a7393c062349889b48aba663b14a61 : SUCCESS in 29m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37fdcf4378a24c9283cff911c5d55fc6 : SUCCESS in 1h 06m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ce0abfdba08244e790c45bbdf78488b5 : SUCCESS in 1h 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f40b2249d7e8cc8d2d11d47b1b62424ae9cc7d42","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 03:52:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c478927c43d55907991d360ac7e61a882bc80b4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 04:34:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8ee2c0db6dd4b1c8aa68f52a7eb27aa : SUCCESS in 36m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6198b1317a724701a5fe1fb35b43f336 : SUCCESS in 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"960737f819350bd3c7cd1db86bec42099d11f858","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-16 05:41:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804571/2 : SUCCESS in 1h 48m 08s","accounts_in_message":[],"_revision_number":2},{"id":"3ac976f496121310b6d1297a096c6e3a826c7fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 06:02:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8ba040e6aec24b73ad126f7e0e22477a : SUCCESS in 44m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e4ca0987822f4a80bbe39dc24871b9e7 : FAILURE in 1h 51m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ec5a3d9564b4544b21d6e2bb4fe17c1 : SUCCESS in 2h 06m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/67e5864d12664a659de4eaad7335e98d : SUCCESS in 1h 20m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2f37801ed9584b5abe68980f93521048 : SUCCESS in 1h 46m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8daa3249ea6640369ee777c6b310f4ea : SUCCESS in 26m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35a2486dcd7d45468064e7e1fc0009f7 : SUCCESS in 40m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c1b60ab24bb4fcb840da6ae6be43385 : SUCCESS in 39m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f92cd37c5b4295a51689696ed4c989 : SUCCESS in 11m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96381c64d0aa43378d13e5d1f142e33f : SUCCESS in 23m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f12fc62eff664724a0045e7f20dad04f : SUCCESS in 30m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fcf93f916b3434dbe462a176b6fa659 : SUCCESS in 25m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae9115bbcd1408fa506ad7f366c895a : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f25de7406cf46b58bb402f599ee20b8 : FAILURE in 41m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0d99080314f4f63bb534bb2ebaf7ab0 : SUCCESS in 1h 34m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/68d35f9825844d7288bc518467ad1345 : SUCCESS in 1h 00m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/042b18027e564d7aad984712030e002f : SUCCESS in 50m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e57bed42d26b419e808ca4c1312728bd : SUCCESS in 59m 17s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2f3158edc2c4474b9058da4778bf7ddc : SUCCESS in 57m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a631786168fe44e3b6cd49ea4cc0156f : SUCCESS in 1h 11m 50s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/9409680373eb4f1992cc0072135f64b3 : SUCCESS in 1h 14m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5c67bdf56644016b64337a5658c96be : SUCCESS in 50m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/092203a67f5d49f39b6e65fa2d3832c7 : SUCCESS in 53m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/927505b1d2e3490588cd6852a972de23 : SUCCESS in 31m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9428ae3b187d47ad943bf82f839099cd : SUCCESS in 1h 04m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f72e9aae31e04f4dbde1404ace15979e : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":2},{"id":"53ffff9fbb9bf583b2b2cba0c7b2a356fb69644d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-16 07:24:35.000000000","message":"Patch Set 2:\n\nWhy not merge all together? https://review.opendev.org/c/openstack/neutron/+/804572, https://review.opendev.org/c/openstack/neutron/+/804573 and this one","accounts_in_message":[],"_revision_number":2},{"id":"479b4875c6513230f1888487bcaa42804e56ac47","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 08:37:16.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Why not merge all together? https://review.opendev.org/c/openstack/neutron/+/804572, https://review.opendev.org/c/openstack/neutron/+/804573 and this one\n\nI created separate commits considering we\u0027d want to backport these changes differently.\nFor example this change is applicable to Victoria and later, while the other two can be backported to further older release.\nHowever it might be just redundant to create separate commit for each small change.\n\nI\u0027ll update the bug report and merge these three into one.","accounts_in_message":[],"_revision_number":2},{"id":"31cc15a25461ca2d2ba70b75331497bef32b5ec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 08:45:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ecbd7c932124ee21f3fc9055484011904a4c62b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 08:46:18.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9e97f174489bc6b99f217be1b4ed97860a0c5dfb","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-16 08:49:57.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"3f6eb7d296e2ac54a3a48faa4ad1e4c341024b22","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-16 09:35:28.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":4},{"id":"d198585a22fcfca624695d2f857b848176896f45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 09:40:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/041af3832a38459b91c6d7619445cf2e : SUCCESS in 37m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8357cbde79348a295341ad02113e698 : SUCCESS in 37m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"38f4667611a6b38d69552d2752a1e7c013e42455","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-16 10:34:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804571/4 : SUCCESS in 1h 47m 51s","accounts_in_message":[],"_revision_number":4},{"id":"23bd6c41c87c7639982280c9bfdecfafddc3e166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 10:47:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/745d40931d26430baa452c244b08aa41 : SUCCESS in 41m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/100f60c2578b45a0b2aec08edc4a9ca0 : SUCCESS in 1h 42m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b17d3b21e9784cd9ae896a94a390da22 : SUCCESS in 1h 59m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/82acfc9a72fe4076b3402476b94bff7c : SUCCESS in 1h 50m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ed34eecd3f5d4ee795e7d27100480531 : SUCCESS in 1h 53m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5753414ed31a4cf6b890889bbc1a7061 : SUCCESS in 33m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0dc3b555fd54c2d8f25e3cf9590bf4d : SUCCESS in 31m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfba234896ee4ed98fd816d0373e90be : SUCCESS in 57m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82877c9217fe40edb2612e41c6f43dbb : SUCCESS in 13m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b99ff534d2444deaa14fb9dea11760fa : SUCCESS in 26m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/939baa8c16e44540ae515b3428ccdc0d : SUCCESS in 22m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc07e2141db24b47b8671ba06f9fc0e7 : SUCCESS in 25m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ffeb868ba04c57860901320502e670 : SUCCESS in 12m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd97b3ec3f46481781908746f3cfdfe8 : SUCCESS in 1h 02m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/93567387bbc84b4daa0ab8a5ac42673c : SUCCESS in 1h 46m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f5fafbe2c1554c96a2211e2baa8d6d06 : SUCCESS in 1h 06m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c420dbd84684572a8a168b25d1e94e1 : SUCCESS in 1h 04m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/88e25c648dfd4e37bde2da0f9ba365af : SUCCESS in 1h 01m 59s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1fd66b3dff28492d9b340ee7d66413cb : SUCCESS in 1h 14m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/97e37acf77e2482ba66f551d4da2fa2b : SUCCESS in 1h 18m 36s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/bc6c7fa57cc946ddbcb05e994e4ce547 : SUCCESS in 1h 28m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fa26123f5064d7f8525b9f4224e972d : SUCCESS in 57m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ba8961d7713e4787823317c4281baf0d : SUCCESS in 56m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/33071cce4fc342bca3edc321afcb67cf : SUCCESS in 35m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bffb2b1896ef4c9dbc3dcc665037df28 : SUCCESS in 1h 16m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/be03a93f8eca4699b7f22144777c1947 : SUCCESS in 1h 15m 42s","accounts_in_message":[],"_revision_number":4},{"id":"5287bd367534e1bcad22aca21a7f00527e7e0454","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 10:58:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e535e7c9977588e52c879ff8de3f2bcd16af4cee","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-18 07:39:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"81cb93e9eff8bd197181af271fd21d6a64fa10ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 07:40:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"15f3bc0cca4368109c6506213e949c14b1110b71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 10:18:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8373d30ea0154169ad3ed390c390a3c5 : SUCCESS in 43m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/10e41340a4ec492b88420dd2bb037eb9 : SUCCESS in 1h 45m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8991c75e09a9450388b315f434cf753c : SUCCESS in 2h 29m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ffcf69a5df6d4da2bd8ec1a071daa6b0 : SUCCESS in 1h 53m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/32e89d0cae8940f5ab896b56764a3da3 : SUCCESS in 1h 38m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95b5c6d5d034435891d8557831143e79 : SUCCESS in 48m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0310a1c767c346e5bfe17ed8a08c5b88 : SUCCESS in 15m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c23f1c86314174a405f8746960263d : SUCCESS in 22m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7e4fa40ccc04333875d9cf6e6a30bec : SUCCESS in 28m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5d606c767f4cafb8965765a9c2348d : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ecd2923392943b488815be271caa03a : SUCCESS in 46m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/658bb2759a7c4c9095b172a7a8e9483c : SUCCESS in 1h 40m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d1dea77f09ea42968ac76c01c4cf0833 : SUCCESS in 1h 11m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbe8bc37414d47bebe9c1609c0b365ab : SUCCESS in 58m 19s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/a4d0d1884a264167afc7509b57566d7b : SUCCESS in 1h 22m 43s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7293f61f098b4d2e86b778150f2c70a4 : SUCCESS in 49m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e1e0f017bce34d3da37d94a12dc2ae1d : SUCCESS in 55m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ce9446ec169442aaa9f64b1dd9f1205f : SUCCESS in 29m 14s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5a5d7e31ec514726a4a4676350561d79 : SUCCESS in 1h 12m 12s","accounts_in_message":[],"_revision_number":4},{"id":"e43daf17d7eecffdbe713c21439cc6a252595394","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 10:18:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"978a4202ae14423ae3a041f31544a352d52b2565","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 10:20:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8835923dd0a44ea48636cbebe10487eb : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c75df16e5021e28aaa9961ed0977f30fa278a1dc","revisions":{"96c57e795391b66e29e67e4c299398c3a0fc6c64":{"kind":"REWORK","_number":1,"created":"2021-08-14 14:31:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/804571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/804571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/804571/1"}}},"commit":{"parents":[{"commit":"d77cecd6c6bc7d6bc4cb99861403ae1c1751803b","subject":"Merge \"Skip DVR binding for ports with invalid OFPORT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77cecd6c6bc7d6bc4cb99861403ae1c1751803b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:29:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:29:49.000000000","tz":540},"subject":"Add options for the healthcheck middleware","message":"Add options for the healthcheck middleware\n\nChange 133200014ba4aabfda6c516dda2e42e7638c3e4d enabled the healthcheck\nmiddleware in api pipeline but parameters of the middleware are not\nincluded in auto-generated neutron.conf .\n\nThis change adds the missing oslo.config.opts endpoint so that\nthe available parameters are picked up by oslo-config-generator.\n\nChange-Id: I9f4763efc02e5cacca3a00e9f101f83d3f3fb1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96c57e795391b66e29e67e4c299398c3a0fc6c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96c57e795391b66e29e67e4c299398c3a0fc6c64"}]},"branch":"refs/heads/master"},"b6474a4eb76de5fc35924d3b9ae756d4f3a0ec4a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-16 00:17:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/804571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/804571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/804571/2"}}},"commit":{"parents":[{"commit":"d77cecd6c6bc7d6bc4cb99861403ae1c1751803b","subject":"Merge \"Skip DVR binding for ports with invalid OFPORT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77cecd6c6bc7d6bc4cb99861403ae1c1751803b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:29:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 00:17:25.000000000","tz":540},"subject":"Add options for the healthcheck middleware","message":"Add options for the healthcheck middleware\n\nChange 133200014ba4aabfda6c516dda2e42e7638c3e4d enabled the healthcheck\nmiddleware in api pipeline but parameters of the middleware are not\nincluded in auto-generated neutron.conf .\n\nThis change adds the missing oslo.config.opts endpoint so that\nthe available parameters are picked up by oslo-config-generator.\n\nCloses-Bug: #1940009\nChange-Id: I9f4763efc02e5cacca3a00e9f101f83d3f3fb1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6474a4eb76de5fc35924d3b9ae756d4f3a0ec4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6474a4eb76de5fc35924d3b9ae756d4f3a0ec4a"}]},"branch":"refs/heads/master"},"21349bcf69e8d5f3b0c2f87df12c6b804c54987a":{"kind":"REWORK","_number":3,"created":"2021-08-16 08:45:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/804571/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/804571/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/804571/3"}}},"commit":{"parents":[{"commit":"d77cecd6c6bc7d6bc4cb99861403ae1c1751803b","subject":"Merge \"Skip DVR binding for ports with invalid OFPORT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77cecd6c6bc7d6bc4cb99861403ae1c1751803b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:29:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 08:42:55.000000000","tz":540},"subject":"Add missing options to neutron.conf auto-generated","message":"Add missing options to neutron.conf auto-generated\n\nThis change adds missing oslo.config.opts endpoints to the config file\nto generated neutron.conf, so that the following options are included\nin the neutron.conf generated by the oslo-config-generator command.\n\n1) Some options of the oslo.service library, related to periodic tasks\n   and base service implementation\n\n2) Options of the oslo.reports library\n\n3) Options of the healthcheck middleware from oslo.middleware\n\nCloses-Bug: #1940009\nChange-Id: I9f4763efc02e5cacca3a00e9f101f83d3f3fb1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21349bcf69e8d5f3b0c2f87df12c6b804c54987a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21349bcf69e8d5f3b0c2f87df12c6b804c54987a"}]},"branch":"refs/heads/master"},"c75df16e5021e28aaa9961ed0977f30fa278a1dc":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-08-16 08:46:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/804571/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/804571/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/804571/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/804571/4"}}},"commit":{"parents":[{"commit":"d77cecd6c6bc7d6bc4cb99861403ae1c1751803b","subject":"Merge \"Skip DVR binding for ports with invalid OFPORT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77cecd6c6bc7d6bc4cb99861403ae1c1751803b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:29:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 08:46:01.000000000","tz":540},"subject":"Add missing options to generated neutron.conf","message":"Add missing options to generated neutron.conf\n\nThis change adds missing oslo.config.opts endpoints to the config file\nto generated neutron.conf, so that the following options are included\nin the neutron.conf generated by the oslo-config-generator command.\n\n1) Some options of the oslo.service library, related to periodic tasks\n   and base service implementation\n\n2) Options of the oslo.reports library\n\n3) Options of the healthcheck middleware from oslo.middleware\n\nCloses-Bug: #1940009\nChange-Id: I9f4763efc02e5cacca3a00e9f101f83d3f3fb1e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c75df16e5021e28aaa9961ed0977f30fa278a1dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c75df16e5021e28aaa9961ed0977f30fa278a1dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
