)]}'
{"id":"openstack%2Fneutron~804572","triplet_id":"openstack%2Fneutron~master~I851b856e622fdc7aada49b27ef01a1010a6afd5b","project":"openstack/neutron","branch":"master","topic":"oslo-reports-opts","hashtags":[],"change_id":"I851b856e622fdc7aada49b27ef01a1010a6afd5b","subject":"Add missing [oslo_reports] options","status":"ABANDONED","created":"2021-08-14 14:32:44.000000000","updated":"2021-08-16 08:46:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ab3c1695b3af7f323f789f9eaa879336ae0a6c9","_number":804572,"virtual_id_number":804572,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-14 16:47:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":2,"date":"2021-08-16 07:20:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-14 15:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-14 16:18:50.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-14 16:47:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 07:20:58.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"dd16685495a208fe8f6e7fbee6786fd4da84bd2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-14 14:32:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24084e69325ae303aab08d825cab4b6565cb4e05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-14 14:33:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"35b99dfba96ab63f09f7e9b7e91df5b4994bbd3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-14 14:46:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"14dfbbea5234830cda3d5c32f5a5189268e1f744","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 15:23:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e3f4d8292c54e489a9d9c22d92f0536 : SUCCESS in 33m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/afce8dcb23e8480980f8468586b00737 : SUCCESS in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf3280c77201967dc3aa9a027783117e2f236d36","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-14 16:18:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804572/3 : SUCCESS in 1h 31m 46s","accounts_in_message":[],"_revision_number":3},{"id":"64eeda72f2e9eff7c5097f235e2bbe380961bda9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 16:47:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a0ab6d8b25424c22b4888eda0719d6db : SUCCESS in 49m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/76406c8a5b914566a331f505261659fc : SUCCESS in 1h 56m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/38c707079fb6448a8d2d634007c0ebee : SUCCESS in 1h 45m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/82820dd9f4204c449b54ba5079156778 : SUCCESS in 1h 39m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dd1045407d7249ba869c5ecf20300563 : SUCCESS in 1h 52m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2da71977ad1240da9418a3cff363af57 : SUCCESS in 31m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c7cba36c7d7464fb24540bac06bf27e : SUCCESS in 36m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a077aeb1e3244a71a20266e385c8a069 : SUCCESS in 40m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07e9d4d2448486482410a57b53f6a49 : SUCCESS in 13m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7befbf27a3c40d9b82ceb5ff136dfd8 : SUCCESS in 27m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c13c56766514a80b1300481d95e8f5e : SUCCESS in 23m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fc1efb2a0144a1e90be87c8cb93aedb : SUCCESS in 22m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926a3fd852984129a4bf98baf1d58364 : SUCCESS in 8m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/419aabc7f5ce49809a7036294a103b9a : SUCCESS in 53m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7670574d3304c6ab58e629ca7a16590 : SUCCESS in 1h 48m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0806212b9ccc480c982fee49bddc91a7 : SUCCESS in 1h 11m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5fd4444306a40fab74ff4725071b3cb : SUCCESS in 57m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0717ae21afcf4f11aac6641b0b66b9e8 : SUCCESS in 1h 02m 46s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d7ed910803234aaab12a5c2c8e136a3c : SUCCESS in 1h 07m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/17c0d164d43c4756a8f906a2157151c4 : SUCCESS in 1h 10m 56s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/cf44a053e82e4baf8310518768acbf97 : SUCCESS in 1h 22m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca8b9f04edb44f4ebd0a466256341345 : SUCCESS in 53m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1ee041461777470a90f388199c7a5435 : SUCCESS in 43m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b2579c28efa2410bb39bc0772bad7e6d : SUCCESS in 28m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b5d3d1ce1474448996e912db31c110f9 : SUCCESS in 1h 07m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e49d6befed7543df8ab5cd7684575b9e : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":3},{"id":"1ff3bfb2dc523f7bd13415ee0bbb1a26d45449aa","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-16 07:20:58.000000000","message":"Patch Set 3: Code-Review+2\n\nDoes neutron use","accounts_in_message":[],"_revision_number":3},{"id":"9ab3c1695b3af7f323f789f9eaa879336ae0a6c9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 08:46:33.000000000","message":"Abandoned\n\nmerged into https://review.opendev.org/c/openstack/neutron/+/804571","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"86fbc448010ea4289837c0ba3f120c5ce0f46c2a","revisions":{"661944426e10bf7264b7cae1482838e80ee9aae6":{"kind":"REWORK","_number":1,"created":"2021-08-14 14:32:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/804572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/804572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/804572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/804572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/804572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/804572/1"}}},"commit":{"parents":[{"commit":"d77cecd6c6bc7d6bc4cb99861403ae1c1751803b","subject":"Merge \"Skip DVR binding for ports with invalid OFPORT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77cecd6c6bc7d6bc4cb99861403ae1c1751803b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:32:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:32:22.000000000","tz":540},"subject":"Add missing [oslo_reports] options","message":"Add missing [oslo_reports] options\n\nThe oslo.reports library provides some options under the [oslo_reports]\nsection. This change ensures these parameters are rendered by\nthe oslo-config-generator command.\n\nChange-Id: I851b856e622fdc7aada49b27ef01a1010a6afd5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/661944426e10bf7264b7cae1482838e80ee9aae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/661944426e10bf7264b7cae1482838e80ee9aae6"}]},"branch":"refs/heads/master"},"6b6137322e4123d72ddebabe3f381e78447235d5":{"kind":"REWORK","_number":2,"created":"2021-08-14 14:33:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/804572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/804572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/804572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/804572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/804572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/804572/2"}}},"commit":{"parents":[{"commit":"d77cecd6c6bc7d6bc4cb99861403ae1c1751803b","subject":"Merge \"Skip DVR binding for ports with invalid OFPORT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77cecd6c6bc7d6bc4cb99861403ae1c1751803b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:32:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:33:38.000000000","tz":540},"subject":"Add missing [oslo_reports] options","message":"Add missing [oslo_reports] options\n\nThe oslo.reports library provides some options under the [oslo_reports]\nsection. This change ensures these parameters are rendered by\nthe oslo-config-generator command.\n\nChange-Id: I851b856e622fdc7aada49b27ef01a1010a6afd5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b6137322e4123d72ddebabe3f381e78447235d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b6137322e4123d72ddebabe3f381e78447235d5"}]},"branch":"refs/heads/master"},"86fbc448010ea4289837c0ba3f120c5ce0f46c2a":{"kind":"REWORK","_number":3,"created":"2021-08-14 14:46:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/804572/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/804572/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/804572/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/804572/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/804572/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/804572/3"}}},"commit":{"parents":[{"commit":"0dc57958311846f672dbe414bc2f8dcf1a6dd5a6","subject":"Add missing oslo.service options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0dc57958311846f672dbe414bc2f8dcf1a6dd5a6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:32:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-14 14:45:15.000000000","tz":540},"subject":"Add missing [oslo_reports] options","message":"Add missing [oslo_reports] options\n\nThe oslo.reports library provides some options under the [oslo_reports]\nsection. This change ensures these parameters are rendered by\nthe oslo-config-generator command.\n\nChange-Id: I851b856e622fdc7aada49b27ef01a1010a6afd5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86fbc448010ea4289837c0ba3f120c5ce0f46c2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86fbc448010ea4289837c0ba3f120c5ce0f46c2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
