)]}'
{"id":"openstack%2Fneutron~804596","triplet_id":"openstack%2Fneutron~master~Icf7822bf83f5496f871f40a2fd08d9d01c2551b2","project":"openstack/neutron","branch":"master","topic":"bug/1937904","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-15 18:12:56.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-15 18:12:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icf7822bf83f5496f871f40a2fd08d9d01c2551b2","subject":"Use importlib instead of imp","status":"ABANDONED","created":"2021-08-16 00:24:36.000000000","updated":"2021-12-15 18:12:56.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"660099cb76f9430bd28d49ee22089c63630b5571","_number":804596,"virtual_id_number":804596,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-10-14 10:17:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-14 12:46:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 01:08:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-16 01:54:51.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-16 02:47:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 04:28:26.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-26 00:11:11.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-27 13:08:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-31 13:58:58.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-12-15 18:12:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"58dd4d8edaaae648d9bb1ae743edc3a26ca1b7bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 00:24:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b480c1383a269d439327d294551821d6a2350a37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 01:08:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/462877df43b0485b841554d7f4c3b71c : FAILURE in 38m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/094e6022dd2d40868d32773823a4cfea : FAILURE in 37m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32020286728ac92ee948450bc95ae43c7ab6ad29","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-16 01:54:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804596/1 : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5699998443af3baabacbc279aaec31e8810e4a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 02:47:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/68844bbfc26d4ba29eeaf94b0fc07091 : SUCCESS in 45m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7ec425fbb8a3487e8f556d540879dd7d : SUCCESS in 2h 08m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/28115f1522b64aeea2a65a8faee43a77 : SUCCESS in 1h 46m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c723ee66cb34db3be481bab462b1798 : SUCCESS in 1h 30m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/48efae3232144c089d44996d5d495bba : SUCCESS in 2h 17m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10bcdf4d3c444836944b880b35618d2b : SUCCESS in 31m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14e14512492243c797aa820dd043d436 : FAILURE in 35m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5cdbaf3c910479f85f23588057ffd0f : FAILURE in 36m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a01c6c234fc548bfad381ab7874d7408 : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4c0af0daf33499ab3e6961e5dd2f3e1 : FAILURE in 21m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/553ebb0bab4a49df81813fe49a7f6a88 : FAILURE in 27m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40203c6dfefa4fb7b095b2cc29b8dd31 : FAILURE in 25m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e46fcf7e74d44c681968a54cafc4848 : SUCCESS in 9m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/75feeb291ca44934bf6ca3dda9edf547 : FAILURE in 43m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/57c714e2e07d4f0f9cc5bea6093d4499 : SUCCESS in 1h 49m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2f061e295aca4f0e916b3a5e61e724a3 : SUCCESS in 1h 08m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/53ec27e8ba384e0aab5acccf3a14231c : SUCCESS in 1h 03m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ec5504e2f81244c59d26228ad03d3107 : SUCCESS in 1h 04m 18s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bb43307f1c3a451c9134a463652a1ad2 : SUCCESS in 1h 05m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5f797b41ef814d4c91661af790826d0d : SUCCESS in 1h 12m 53s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/91d5f755aa4b4c3890a60397e9dc5355 : SUCCESS in 1h 24m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4468ae3009f24c708633f429ed01ef11 : SUCCESS in 55m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3e8359ca62c24224b09fb2d90be2e98d : SUCCESS in 54m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e29a7e2eb2aa42b58f627dfb8bcf9a23 : SUCCESS in 28m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d70c09c67c704177a380f6491b44a28d : SUCCESS in 1h 02m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b6e5472a29874566b4f5c48d3904de41 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5ccb54b895622dc7349f5125b683e7926d6176d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-16 04:28:26.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/96/804596/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/96/804596/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"918ad14d3b5b339b69877c430478bdbcd3b4e8b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-26 00:11:11.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/96/804596/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"98537ebe6e836285315cbd32deaa059217f1e096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 13:07:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac2cdbff2fd4385ad7ed5c1a6b54793a1f9dc918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 13:09:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7850e3acf335a8fe77d20af0a7e3ced68f3a445b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-27 13:17:20.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"93ba0c13c01395786c449d83b6e0809198ee732b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 13:32:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0228fc93b5537d31d6972652477552b6abe18cb0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 13:39:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bbb74fd0d3a70b694f7584396a7dcea24357aad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 13:52:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f195594e01db48fda94a4e5e879f16bf : FAILURE in 41m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b22ed07fc64e43dd8ecd5c47b599419e : FAILURE in 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78d99ce21156f5b6b5c632408df80cf7de56b18a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-27 14:16:15.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/804596/3 : FAILURE in 1h 04m 25s","accounts_in_message":[],"_revision_number":3},{"id":"4ec3e249288055634c8054d452c0a07b0612baa8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-27 14:16:52.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/96/804596/3/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/96/804596/3/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"fff1857ff32f0d2bdfbd841afd51a59095d9569b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 16:19:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5fd0cadcdf6444d9a1393e5621d9c198 : SUCCESS in 48m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a051fa9d2f6b4df49fe862a9e94ada49 : SUCCESS in 1h 36m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6e6a7740d4bd49e68100c27312cb3a6f : SUCCESS in 2h 00m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ae17269bd6a94993815b994f5dc66f3e : SUCCESS in 1h 48m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/573314b77ad24bf89cfe2ad814f20b8c : SUCCESS in 1h 37m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df8f77e269fa464dad0e758e0a0f4ea5 : SUCCESS in 36m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5939332aab7c496c87e68e10c3297f71 : FAILURE in 28m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbdd212bb86b4ef7999e88a706258f7c : FAILURE in 52m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5229885fae143cdbcc8cf024d353e67 : SUCCESS in 11m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91b3a5146e3b4a1c9b8055cbbf96af1e : FAILURE in 41m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4eb810197f394d8aa9219fb5dd64ece0 : FAILURE in 24m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f63cf1f346be49b99a9d26d6b580c335 : FAILURE in 40m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8ef58ada484d4cbc0aa0ea2a440c3a : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa22cc46f0c943069b5a2e4fe251326b : FAILURE in 47m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/45d0480121a4465da267c83cb9c2c299 : FAILURE in 1h 31m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b3fc4c0f9b7c4050b90a31b6926c518d : TIMED_OUT in 3h 04m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/762fd84f4a2c432caf5b1e9504988938 : FAILURE in 1h 18m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cbe4c3690b664e34847ac8cbd20ac9de : FAILURE in 1h 12m 01s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/46a0d3523d40484f8f5b865f74491956 : TIMED_OUT in 3h 05m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/62fb92a06aa246419d57214c6c182788 : FAILURE in 3h 05m 36s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/d254dbf3b09c4b10a14817d5a227b659 : TIMED_OUT in 3h 03m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a9705babec24fa183ca6d859ce4fd0e : FAILURE in 2h 53m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/137f98dc60f44388b8e5229da71bf644 : FAILURE in 2h 40m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/60720b08687b47ac90394fc09ab3a39b : FAILURE in 37m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa025911babd455c88dffd124b8887eb : TIMED_OUT in 2h 13m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8fba4b7cc6954ffdade8a771bb7388e2 : SUCCESS in 1h 16m 02s","accounts_in_message":[],"_revision_number":3},{"id":"9f8f74ef664f53b1bf4aab2aa0ee5a8b19aabb27","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-27 19:14:10.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/96/804596/3/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/96/804596/3/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"4f0e9f8563eaf737cbc11a2cebeff2f4fc10324c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 22:19:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f95f6bd4810dc8de13f44636422a8d7318351543","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 23:07:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5421bdce4ae74c9ca3351bfa8e6961eb : FAILURE in 41m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b2aecc0dff84f6494a5d6c238fc116a : FAILURE in 42m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4d66d88de68d31a820aab2b5fc85aed560400193","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-28 00:09:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804596/4 : SUCCESS in 1h 48m 46s","accounts_in_message":[],"_revision_number":4},{"id":"91bb2bad024829e096d10f8e523e1ffe9373a770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 00:43:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6e19564931d141c88396834c095fc2b0 : SUCCESS in 43m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/79d339f09a0d4329a4028d92e7b2cdb1 : SUCCESS in 1h 48m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d088b56947cf44e2966a6c5c2eb4d8dd : FAILURE in 1h 32m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d1991e5975ed4622bdf588167a26a4a2 : FAILURE in 1h 31m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a4506d96ec8b465cbbfd2094d3e52d31 : SUCCESS in 2h 18m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0cc070f3fae0463caa484ba73368ae51 : SUCCESS in 28m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b082a179cd984e9f85a4c1d7abb6692b : FAILURE in 28m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73b2f472b4ee4bb6ab9cc3d8fe980746 : FAILURE in 32m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4baca31908f74147a4c0917d62ff37a4 : SUCCESS in 15m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07f885ca04cf401f85556fa3ca286c09 : FAILURE in 22m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e6de24244514d5eb1bbd9081f274df0 : FAILURE in 27m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aec2fa1205fc483095dd065b766bc3ce : FAILURE in 25m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b4b186e458495bb81ff59af1bb7e7b : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/631d4def3e7a42f5a2693c6bfc495894 : TIMED_OUT in 2h 18m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f34a3d227b4940f2a72a234383de460a : SUCCESS in 2h 09m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1b7905fdb46a4601b2d213c26223a56d : SUCCESS in 1h 27m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c1407e00dbd4c098edd04c31700466a : SUCCESS in 54m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/be370202ec054de9a6569b94ae62bd38 : SUCCESS in 56m 20s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/25d63dec3ec84c9c8f61e2d2937fea6e : SUCCESS in 1h 11m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b68d68e68c5e4bae9641da62cab0c174 : SUCCESS in 1h 13m 57s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/efbd4b9c74d740e38a584835bf4cab66 : SUCCESS in 1h 27m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e480e5bf96dc42e59f7a73442a5710e0 : SUCCESS in 45m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f4dcda1ba2734a729f12e0b9a6b9c9f2 : SUCCESS in 58m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/16bff3f9c8e141e78e8778bf30bb6e83 : SUCCESS in 26m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/57b214a7b1eb4bf1bde17ef4db30cfd4 : SUCCESS in 1h 14m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/58836ef6dcec4efabb51429ffc564cf3 : SUCCESS in 1h 15m 08s","accounts_in_message":[],"_revision_number":4},{"id":"4d55279911c11d8eac3d7be22dd5bab729f08a6c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-28 03:25:06.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/96/804596/4/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/96/804596/4/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"f583e70ffeeb7d8af1e7659eede5798830046066","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-28 03:28:33.000000000","message":"Patch Set 4:\n\nHmmm.... some unit tests for ovn fail because of failure during creating a security group.\n\n{\"NeutronError\": {\"type\": \"HTTPInternalServerError\", \"message\": \"Request Failed: internal server error while processing your request.\", \"detail\": \"\"}}\n\nIt might be relate to the override of extention manager but I\u0027m not very familiar with the actual reason ...\n\nhttps://github.com/openstack/neutron/blob/701214481ff1df13b188d7826b44a1c8b51227ea/neutron/tests/unit/db/test_ovn_revision_numbers_db.py#L144-L148","accounts_in_message":[],"_revision_number":4},{"id":"1f041113a9052ebd1c7936cd1391edd3df6ad562","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-29 14:49:59.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/96/804596/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8c9e0848f492eec4d64e00becceb40fd67d03982","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-29 15:52:10.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/96/804596/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"66aac405f9b6cc3f0e2697891bacd49a7a032a36","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-29 20:47:35.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/96/804596/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"6541e8ab2574ba431c5b98e607fe839b5331e40a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-30 09:26:39.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"9d184e95c96e01412f69b133e31e074b638fddd1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-30 10:37:53.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/96/804596/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"8314c65c9ea88506ff2d4b36be92452f05f95d11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:38:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7fa4c2161cee4a75b8c007238c0b25c3 : FAILURE in 34m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c88de4a39ea64d08809ce0944294d7a3 : FAILURE in 50m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b22d04d6ee4595976e91e9a0d22fd731acd01259","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-30 11:17:28.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/804596/5 : FAILURE in 1h 50m 09s","accounts_in_message":[],"_revision_number":5},{"id":"7246ef8747544013bcb9e82e627c7ff0083cbdfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 12:10:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b86673bbaaf54eaeb3936e3eb6a5de31 : SUCCESS in 46m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3c17216cd2cf47efa783a3422e1d0d9e : SUCCESS in 2h 32m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dea94bb40dd241cf841fc52e1e64bd11 : SUCCESS in 2h 01m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bf14770bffbd409789b414ad9d7d75be : SUCCESS in 1h 23m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6e075dc42b814efcb8eb9ee83e1408b7 : SUCCESS in 1h 53m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/910e0877835f499894efc70f9450adca : SUCCESS in 33m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/455755b0e6104b5b893c2f62cb23e8fb : FAILURE in 25m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f016ee415364d7b9c61b128fbd27fe0 : FAILURE in 38m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/545a5aa22556488489c1265a98ff54f5 : SUCCESS in 15m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98a2982696744d2ca32849e64fa07f5d : FAILURE in 26m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b058d1efc0ed4426a1ac2a8f1e7b7ae4 : FAILURE in 26m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3c8283658dc4fbfbc823121d613c321 : FAILURE in 27m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33941be4deed4872ba16abb80584bee5 : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ace4d3ac43c54317aa289e5ead8122c3 : SUCCESS in 53m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9166f1b769204981af59e8c7b7fea61b : SUCCESS in 2h 10m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e377572179e04f1589b78af66544ef38 : SUCCESS in 1h 20m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/63b82d8445c048599b86ae8ffac8e705 : SUCCESS in 58m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ba3e2f00315a4e0dbee81304ee3d3362 : SUCCESS in 58m 46s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/856b88ade6d048d797c37e575c62ece8 : SUCCESS in 1h 25m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/38fdf448603141438dacc0b29a8a9b4e : SUCCESS in 1h 11m 21s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/579c3d3f814146b1a8d226e396152424 : SUCCESS in 1h 26m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74890a336af649a385806f76353347ca : SUCCESS in 54m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/409369c38f3348d78c69841fad758bc9 : SUCCESS in 57m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d38ce0832143450b999ba1473b490e86 : SUCCESS in 35m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2b34a3c4ce954e71b0679f17f8dd8fbc : SUCCESS in 1h 07m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/de5fb50a2cda468fb3cbf8b39fbb01b6 : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":5},{"id":"722b7c83ef960638daf87e3ecda2f09446ab6824","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-30 14:52:43.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/96/804596/5/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/96/804596/5/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"6379b012ac7a4e06aa8922fc27a346aa1e719384","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 00:43:07.000000000","message":"Topic bug/1937904-patch1 removed","accounts_in_message":[],"_revision_number":5},{"id":"e1faba56bc4a067e9f5f45a4f096b42700c3c090","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 00:43:09.000000000","message":"Topic set to bug/1937904","accounts_in_message":[],"_revision_number":5},{"id":"dd1cf66f0edc85953f88be54433b189a2a532c80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-14 09:00:20.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"83ac8fba3a9d7200dd39271e1ef95d6e77a98ab4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-14 09:13:42.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/96/804596/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/96/804596/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"cda219f09b3bcd5fe20f6c67f7e81c1c146a9b41","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-14 10:17:01.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/96/804596/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"4ffd1a54564ea95231f7bc10b5c1a9e6a49d7189","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-14 10:30:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/804596/6 : SUCCESS in 1h 29m 58s","accounts_in_message":[],"_revision_number":6},{"id":"128a26ebb8896683321b209ea315b1274aa9b125","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 11:06:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8966071521845ffbe45ef48f5e2b3a5 : FAILURE in 37m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd1d940f0c1c4cec9cb7eb20b30a688f : FAILURE in 36m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6ee7868e2b60232ac2cfdb2370dfa8a434f37929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 12:46:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fd0703ca64d043ecb95d14a8d12067bb : SUCCESS in 40m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e5b1e2e1b46f40ab9c487ca59dc32904 : SUCCESS in 1h 58m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7c5b0c72b12240888ec5fc3e20a46f35 : SUCCESS in 1h 53m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36ac8c53d51c4c55adfbb3fda99e9901 : SUCCESS in 1h 35m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4b7b4547834f468eb2956fe867d219c1 : SUCCESS in 2h 23m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eee9c775c23f4799a43927c12239e2bf : SUCCESS in 33m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94d5accea37e40fda2c5ba7888e516c8 : FAILURE in 32m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c81aa94c6bf4974811dac024782877b : FAILURE in 32m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3338df6051ee43c081a3f5424ab258ea : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15f2a82879894a80bbd79c4fda515fff : FAILURE in 25m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c388d844b3214791b1676dc7440b2936 : FAILURE in 25m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dc0c07a958b4d4996ac5001cb9d059e : FAILURE in 26m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8550af13b6a24418afd1a5b7fd5c775a : SUCCESS in 14m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c545f3703c854726a05fbfc4b8b6c780 : SUCCESS in 49m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9dd3217fdc7f404eb100f7f74920012b : SUCCESS in 1h 42m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/49c29b50ad17471abd1e8a4bfcc09100 : SUCCESS in 1h 17m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/01c11eb5ea9c401fb974e3e407f230a5 : SUCCESS in 57m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d66ca51793d0467cb835a5dbd4e8c314 : SUCCESS in 1h 01m 47s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b71a2e317d974738886aac0f8516c59e : SUCCESS in 1h 09m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8272b2e975c948a0972d441a8a9bb1c1 : SUCCESS in 1h 10m 09s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/a1ce3ce8af8d4685a6568c8dfc256698 : SUCCESS in 1h 29m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f1cfd9abfd6439d9da1bb38fe796167 : SUCCESS in 1h 02m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7883033ee2d646b7a0fd72df87122d4e : SUCCESS in 43m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8582d3396f954bafa9534a487bb377be : SUCCESS in 29m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/72494aeb68cb40da83651c1a42a8fdf7 : SUCCESS in 53m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/673af80d51144fc7bd048fbe31ef994c : SUCCESS in 1h 13m 49s","accounts_in_message":[],"_revision_number":6},{"id":"660099cb76f9430bd28d49ee22089c63630b5571","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-15 18:12:56.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b74801f332568d7c6c4ed5b36ef9239b70f81072","revisions":{"85f0e9cdbaf916d4b4228042beaba900a73531f6":{"kind":"REWORK","_number":1,"created":"2021-08-16 00:24:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/804596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/804596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/804596/1"}}},"commit":{"parents":[{"commit":"b6bb4c0941800e29c07952e4d0b8deccff98e38f","subject":"Merge \"Make neutron-ovs-tempest-dvr-ha-multinode-full voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6bb4c0941800e29c07952e4d0b8deccff98e38f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 00:23:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 00:24:21.000000000","tz":540},"subject":"Use importlib instead of imp","message":"Use importlib instead of imp\n\n... because the imp module is deprecated since Python 3.4 .\n\nCloses-Bug: #1937904\nChange-Id: Icf7822bf83f5496f871f40a2fd08d9d01c2551b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85f0e9cdbaf916d4b4228042beaba900a73531f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85f0e9cdbaf916d4b4228042beaba900a73531f6"}]},"branch":"refs/heads/master"},"0119ff8eb0828a95e4b3e4d48f55380e5dd1b6db":{"kind":"REWORK","_number":2,"created":"2021-08-27 13:07:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/804596/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/804596/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/804596/2"}}},"commit":{"parents":[{"commit":"b6bb4c0941800e29c07952e4d0b8deccff98e38f","subject":"Merge \"Make neutron-ovs-tempest-dvr-ha-multinode-full voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6bb4c0941800e29c07952e4d0b8deccff98e38f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 00:23:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 13:07:04.000000000","tz":540},"subject":"Use importlib instead of imp","message":"Use importlib instead of imp\n\n... because the imp module is deprecated since Python 3.4 .\n\nCloses-Bug: #1937904\nChange-Id: Icf7822bf83f5496f871f40a2fd08d9d01c2551b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0119ff8eb0828a95e4b3e4d48f55380e5dd1b6db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0119ff8eb0828a95e4b3e4d48f55380e5dd1b6db"}]},"branch":"refs/heads/master"},"be62aa05b36a565a4a883e0690e0a58116806982":{"kind":"REWORK","_number":3,"created":"2021-08-27 13:09:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/804596/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/804596/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/804596/3"}}},"commit":{"parents":[{"commit":"b6bb4c0941800e29c07952e4d0b8deccff98e38f","subject":"Merge \"Make neutron-ovs-tempest-dvr-ha-multinode-full voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6bb4c0941800e29c07952e4d0b8deccff98e38f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 00:23:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 13:08:52.000000000","tz":540},"subject":"Use importlib instead of imp","message":"Use importlib instead of imp\n\n... because the imp module is deprecated since Python 3.4 .\n\nCloses-Bug: #1937904\nChange-Id: Icf7822bf83f5496f871f40a2fd08d9d01c2551b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be62aa05b36a565a4a883e0690e0a58116806982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be62aa05b36a565a4a883e0690e0a58116806982"}]},"branch":"refs/heads/master"},"fcbb40f8ffd30c12bf73948b0544aca4b4a7dd41":{"kind":"REWORK","_number":4,"created":"2021-08-27 22:19:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/804596/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/804596/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/804596/4"}}},"commit":{"parents":[{"commit":"b6bb4c0941800e29c07952e4d0b8deccff98e38f","subject":"Merge \"Make neutron-ovs-tempest-dvr-ha-multinode-full voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6bb4c0941800e29c07952e4d0b8deccff98e38f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 00:23:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-27 22:19:14.000000000","tz":540},"subject":"Use importlib instead of imp","message":"Use importlib instead of imp\n\n... because the imp module is deprecated since Python 3.4 .\n\nCloses-Bug: #1937904\nChange-Id: Icf7822bf83f5496f871f40a2fd08d9d01c2551b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fcbb40f8ffd30c12bf73948b0544aca4b4a7dd41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fcbb40f8ffd30c12bf73948b0544aca4b4a7dd41"}]},"branch":"refs/heads/master"},"10be983c181b5e712bdf2e6c4650922d3b9dd328":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-09-30 09:26:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/804596/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/804596/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/804596/5"}}},"commit":{"parents":[{"commit":"d3fa1ac3bfcdd7c78681184838863226cc0ad808","subject":"Merge \"Replace \"tenant_id\" with \"project_id\" in Quota engine\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3fa1ac3bfcdd7c78681184838863226cc0ad808"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 00:23:57.000000000","tz":540},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-09-30 09:26:39.000000000","tz":0},"subject":"Use importlib instead of imp","message":"Use importlib instead of imp\n\n... because the imp module is deprecated since Python 3.4 .\n\nCloses-Bug: #1937904\nChange-Id: Icf7822bf83f5496f871f40a2fd08d9d01c2551b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10be983c181b5e712bdf2e6c4650922d3b9dd328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10be983c181b5e712bdf2e6c4650922d3b9dd328"}]},"branch":"refs/heads/master","description":"Rebase"},"b74801f332568d7c6c4ed5b36ef9239b70f81072":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-10-14 09:00:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/804596/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/804596/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/804596/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/804596/6"}}},"commit":{"parents":[{"commit":"2b30fa7bedf58e19e150a1f40f82d83d48e84049","subject":"Merge \"Bump os-ken to 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b30fa7bedf58e19e150a1f40f82d83d48e84049"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-16 00:23:57.000000000","tz":540},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-10-14 09:00:20.000000000","tz":0},"subject":"Use importlib instead of imp","message":"Use importlib instead of imp\n\n... because the imp module is deprecated since Python 3.4 .\n\nCloses-Bug: #1937904\nChange-Id: Icf7822bf83f5496f871f40a2fd08d9d01c2551b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b74801f332568d7c6c4ed5b36ef9239b70f81072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b74801f332568d7c6c4ed5b36ef9239b70f81072"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
