)]}'
{"id":"openstack%2Fneutron~805000","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Id3f71611a00e69c4f22340ca4d05d95e4373cf69","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1920923-stable/victoria","hashtags":[],"change_id":"Id3f71611a00e69c4f22340ca4d05d95e4373cf69","subject":"Randomize segmentation ID assignation","status":"MERGED","created":"2021-08-18 13:05:43.000000000","updated":"2021-09-15 11:48:39.000000000","submitted":"2021-09-15 11:47:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805000-bug/1920923-stable/victoria","cherry_pick_of_change":782587,"cherry_pick_of_patch_set":5,"meta_rev_id":"129964c33dba67912a24a284005907df8bc9b6f6","_number":805000,"virtual_id_number":805000,"owner":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 11:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-09-10 13:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-09-15 09:54:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-09-15 09:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-09-15 09:54:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-09-15 09:47:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 13:05:43.000000000","updated_by":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-18 14:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 22:09:40.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-09-10 09:23:16.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-09-15 09:47:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-09-15 09:54:42.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"355b01ba9b43491a6d7c92f79d5ac60c211b01b3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"date":"2021-08-18 13:05:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* neutron/common/utils.py\n* neutron/tests/unit/plugins/ml2/drivers/test_type_vlan.py","accounts_in_message":[],"_revision_number":1},{"id":"26f5d96efde868826d1048f5abe88fccce620dfe","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-18 13:08:31.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease, add conflicting files in the commit message, same as in https://review.opendev.org/c/openstack/neutron/+/804058/1//COMMIT_MSG","accounts_in_message":[],"_revision_number":1},{"id":"f92c5dc003710100dd926d17890beec8f14688d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 14:17:09.000000000","message":"Patch Set 1: Verified-1\n\n(28 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/49f26853b5904fbb91592eee8579c55c : FAILURE in 21m 18s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/9fd5a7d55bd6432c915e4ed6351b4edb : FAILURE in 25m 33s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/b5bba2b36cb143d4a1ee7cf5ff5ec7df : FAILURE in 19m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/ad13d6615cd7473e9130c1b4dc8b4ef1 : FAILURE in 25m 01s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/178d052ad7854d9b9e8bc39c57226e00 : FAILURE in 23m 50s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/dd467582b70849208449d575eda438c8 : FAILURE in 22m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4152b33e2f154c6cafeddf6dfeb1cf42 : FAILURE in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67ab642974e411597ff6e441dc31e3a : FAILURE in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be8ae34e597e4bc6a4e0357adbef9eed : FAILURE in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/532abdaa99d64500b91c8129f206db15 : FAILURE in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d375a8f6cf4ccc85ec4afaa560fa38 : FAILURE in 7m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/55f0a4ee72df4c4b8d458f356cadbb9d : FAILURE in 17m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa96217aa4264413a39dedc60268e6e2 : FAILURE in 16m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/53a8ca969e0249bb934de2592db05211 : FAILURE in 24m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4e04a9284254d41996ed16ad7dc4622 : FAILURE in 57m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f519a7c5a6fa402c867acf475972d1ca : FAILURE in 1h 02m 35s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/e9a45f9335b640879cedb4d88e6650f8 : FAILURE in 58m 20s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/38d3ef50aad24e0faa9719c45ead56a9 : FAILURE in 16m 54s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e46e50743f1642caad08d0d3e7ab5369 : FAILURE in 17m 24s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/30f364d0338240ac817c8b36ca09343a : FAILURE in 21m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39b6cbdc3aa64cd58da6d9432bafa5c1 : FAILURE in 21m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/034dede663304a37b524e2e43a604e27 : FAILURE in 16m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/479a45d6ff6c446ea3c1b27901a69e7d : FAILURE in 21m 26s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6d7faf9e263a4dcdbe5d1e3cc6cba4a0 : FAILURE in 19m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/551277e7ce4442adb8c45711f3b909ea : FAILURE in 19m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/055dc2239ced48d2acdde60e4702c92f : FAILURE in 24m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4098628e1b534704a0c1283797ec8fdf : FAILURE in 26m 14s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d684493c013b4787a284865abd4e4d6f : POST_FAILURE in 26m 05s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/25f80b2f02c84bffb72a6b91b0f76c1a : FAILURE in 30m 08s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/443a90e1793f41b7917c19ab96049e1e : FAILURE in 26m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9838af28264c42af8a03f4f8854076d0 : FAILURE in 32m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0913bca1cbed11744a6ccf064a41253ff76a339","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"date":"2021-08-18 19:54:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e937450c53d807da7c5d945b358eb987f8d3f7fb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-18 22:09:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805000/2 : SUCCESS in 2h 13m 59s","accounts_in_message":[],"_revision_number":2},{"id":"db0ca4c028bd7d89722eb0eaf1a9f389d65af718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 03:28:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/acf257bc4d3c4a2093cc033a23f4b33b : SUCCESS in 49m 19s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/6c70c5b5cc554b7989645fb6e4851a04 : SUCCESS in 1h 56m 09s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/096ed9ffa1274927bcd0cda5ffa29435 : SUCCESS in 1h 40m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/ea28340dbcdd4c8eb80abf751f0f5352 : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/b9ce555a7e1249c8aa2c93f9bba00bfb : SUCCESS in 1h 54m 25s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/b9d5345419894b2693dca5e966d419b7 : SUCCESS in 38m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3553e35511b423ebdb25ea84fb7cd34 : FAILURE in 29m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd80f2234e8414fb207f2c685ede198 : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e9a378b38ef437586ae1760bc733353 : FAILURE in 24m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff88bba364c14568b95bb06963d36577 : FAILURE in 27m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5596cdcd3d94446d94d529aae6f34db3 : SUCCESS in 10m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2782c2dd64624cbf948ad297fc3fa576 : SUCCESS in 1h 14m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/817a049cd5e342ec8e23ac0841f823da : SUCCESS in 1h 21m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d5ebd54b87234297897d5fc046d681dd : SUCCESS in 1h 23m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/224324117e874d2092eb56637fbcce93 : SUCCESS in 54m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1e2749284aa478fae3dfb2a94f7722a : SUCCESS in 53m 49s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/69797a9dcfbb4d14aeeec78505d0511e : SUCCESS in 1h 01m 27s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5eef30c917d94eea806dc74575fc8c18 : SUCCESS in 59m 32s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4967452dec04b8fa53a68565515d643 : SUCCESS in 57m 11s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f666cdfb33b24696aa6a7f3b2cf526bc : SUCCESS in 1h 12m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e44da08b02c4fd8aed53733983fb988 : FAILURE in 1h 13m 27s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e00ebf12b902497da796b4f111243e28 : SUCCESS in 58m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5396c22d15f4430f8b557f049264e402 : SUCCESS in 1h 05m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/33716fa9db344bf7b211e8c6ae55e250 : SUCCESS in 1h 36m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/086495c9deec40178e28f88c6cf895a6 : SUCCESS in 55m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/82be29d957f0460f9e2eee7911d0571b : SUCCESS in 1h 13m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6462481a65c64caca1404ce6c1908b5e : FAILURE in 30m 01s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ac00d9ab91be408a8a28a54f129597f1 : SUCCESS in 1h 15m 08s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9598cf59c0894fc19684a95fec707c3e : SUCCESS in 3h 49m 49s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d153c0efc1d0486185d0ae3c3c91ea74 : SUCCESS in 1h 21m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bac9eaee0cfb42ae8440c30d77ab13ae : SUCCESS in 1h 32m 34s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3f44302e0011472aaf52605b0b7c346d : SUCCESS in 2h 19m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5e934f8fc97a45f18f31656febfb6e3d : FAILURE in 2h 22m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2912f643806d47733cc660422bf2233a21da686a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-10 09:23:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"813f9779c217c3e3a6d170e6834ecbddff1fa768","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-10 11:29:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805000/3 : SUCCESS in 1h 45m 20s","accounts_in_message":[],"_revision_number":3},{"id":"aa3998de13b8bd0ada1c2eef1d385d61eb7850f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 13:07:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/febc5b01494b46d7943cd9e3c4025ae6 : SUCCESS in 54m 52s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/a617a06e272d4f4fa4f92dd60366e08f : SUCCESS in 2h 13m 16s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/4dc6de53ff424e92aa86cecd7ce23283 : SUCCESS in 1h 30m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/ff6dff05911f4203947e34e148a7f857 : SUCCESS in 1h 30m 42s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/35429cfffb324d78adf80021f03695b7 : SUCCESS in 1h 57m 55s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/4ea988154d4f44668c9d16ba8b747bd3 : SUCCESS in 29m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/446be5e0c0aa4788a6b10f6360b3a153 : SUCCESS in 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09f656d743f94654b9528349a3fef9eb : SUCCESS in 14m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f06522aa648840b69c7dafae00f62204 : SUCCESS in 23m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/968a77b000204c4c8b78f03701d3e66c : SUCCESS in 25m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11633492c76455aa72fcd514299aa40 : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/63ba3cc0f49b4860bc9b33a917cffdb8 : SUCCESS in 1h 17m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9db98622b8584028a26699234be2ecc6 : SUCCESS in 1h 27m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8236c157469b416884d529ed6b68e23e : SUCCESS in 1h 40m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ffb0fb4ce10d411ebdae42ba5d2275bf : SUCCESS in 59m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/641d95f4f3534d20a5bd075adfb3dd3b : SUCCESS in 1h 07m 38s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/558113b61906485ea2dc0df88ab84399 : SUCCESS in 1h 10m 09s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e884c15361dc41ee9436d1c821bd47e0 : SUCCESS in 1h 04m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/32e1702083b947e48f5e831d79e5f8fe : SUCCESS in 1h 05m 02s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a146d53c630c4113a19bed79be030060 : SUCCESS in 1h 22m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e4f1e049cf984c9a8e3f11f5dce7fe75 : SUCCESS in 1h 31m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18ca718b70da4241bb162b7a6b0245b2 : SUCCESS in 1h 18m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3e1fba71c1784101ae3a28c5f96a7e8b : SUCCESS in 1h 25m 22s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/751b1b5e8cdc41ab89fd40fd49e4f9d8 : SUCCESS in 2h 01m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f2545df860049b9918bc01cab7d6a1d : SUCCESS in 1h 00m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/91888ce1c8aa4c4ca2808603ef826a64 : SUCCESS in 1h 10m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5101cc9d58574e9c83d45301c504ad5f : FAILURE in 36m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c798c003cdfa414f98b064c956b10b4f : SUCCESS in 1h 21m 34s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3bb604780f7043e4a6fd91a2ce7c636b : SUCCESS in 3h 39m 32s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5b35845485824143af70422f26b8e373 : SUCCESS in 1h 15m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5168808399e64bee9f1f5ebf679c3643 : SUCCESS in 1h 31m 14s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e93e55de24d1420282f4bfc6dbec57fb : SUCCESS in 2h 10m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b4f64e20a9154afb9d21ee46093499a9 : FAILURE in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"31df6fef88067ef107dcab6daa28159acefb538f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-10 13:43:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f1e9e1bbf4fb252270f3665dd9be94d96230e09b","author":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"date":"2021-09-10 13:51:50.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"80cc94b89361ee2576342df4c262f8e718f2ad0d","author":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"date":"2021-09-10 13:55:05.000000000","message":"Patch Set 3:\n\nThanks Slawek for taking this over!","accounts_in_message":[],"_revision_number":3},{"id":"e73a8e16cb04364e0cab6a8afe7203ce112d0646","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-15 09:47:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4ab780c438966a16c72fa19c37ba7d4595f37dc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:48:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"09968090f3d7038c2dee6e62358df414bfc94474","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-15 09:54:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks good and wallaby is merged","accounts_in_message":[],"_revision_number":3},{"id":"a3f335e1f0426e72fc412b996ab7cc2016193f6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 11:47:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/6696dca98129436999f3da3326e8e3f3 : SUCCESS in 44m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2d1edc3be84a3faf8434d429304f2d : SUCCESS in 11m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97e331d7e20c40d8b8d2aaa67a909f6d : SUCCESS in 24m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a513f00262d643eeb0e13a7afe1307fe : SUCCESS in 24m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/674f17a08a8f41d0bf5fa46deba3bca7 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7a00ed31c3c429cadcc4c4a2d9943fe : SUCCESS in 1h 24m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8147dc64af8846f19ea9ca6585fbfa3c : SUCCESS in 1h 16m 03s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/931d658a770042fdaba1f5f9052c40db : SUCCESS in 1h 25m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6d6b751e12bd4c34898078289c12489c : SUCCESS in 1h 00m 55s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b01a16f116a54d5ab2edd127e5b8d159 : SUCCESS in 56m 59s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1eb58205745646c48e8eacd541dc59db : SUCCESS in 1h 22m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/adc033173e0245d891c89aa41080b57c : SUCCESS in 1h 00m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1792ae061f244d12a19b87cca4d701aa : SUCCESS in 1h 12m 16s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55ce7e1f938b4bd282671ed90da2bace : SUCCESS in 1h 54m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ec2f53b99b8445ab53b4530f6f5ff3c : SUCCESS in 1h 05m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6c8f2c78209e4e5c9b16a9b9c4c39d84 : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":3},{"id":"e87cc97f650c851d2c82a68f88e916749d0a944a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 11:47:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"129964c33dba67912a24a284005907df8bc9b6f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 11:48:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea348d349254eb6bae8857370d87086 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ab56a5cd652f57890723d4df5ba6ed22845070fa","revisions":{"e0bebb3945ac8781381020aba6632b494f972bf4":{"kind":"REWORK","_number":1,"created":"2021-08-18 13:05:43.000000000","uploader":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"ref":"refs/changes/00/805000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/805000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/805000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/805000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/805000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/805000/1"}}},"commit":{"parents":[{"commit":"b98a64e68dfd190ed00f082eb1413b0c88978302","subject":"Merge \"Make ARP protection commands compatible with \"ebtables-nft\"\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b98a64e68dfd190ed00f082eb1413b0c88978302"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 18:57:31.000000000","tz":0},"committer":{"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","date":"2021-08-18 13:05:43.000000000","tz":0},"subject":"Randomize segmentation ID assignation","message":"Randomize segmentation ID assignation\n\nIf plugin \"network_segment_range\" is not enabled and a new segment\nis required, if no segmentation ID is provided in the request, the\nsegmentation ID assigned is randomly retrieved from the non\nallocated segmentation IDs.\n\nThe goal is to improve the concurrent network (and segment) creation.\nIf several segments are created in parallel, this random query\nwill return a different segmentation ID to each one, avoiding the\ndatabase retry request.\n\nCloses-Bug: #1920923\n\nChange-Id: Id3f71611a00e69c4f22340ca4d05d95e4373cf69\n(cherry picked from commit 6eaa6d83d7c7f07fd4bf04879c91582de504eff4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0bebb3945ac8781381020aba6632b494f972bf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0bebb3945ac8781381020aba6632b494f972bf4"}]},"branch":"refs/heads/stable/victoria"},"0eba9efd2021fd04e2678714577582bfc6f1e932":{"kind":"REWORK","_number":2,"created":"2021-08-18 19:54:41.000000000","uploader":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"ref":"refs/changes/00/805000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/805000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/805000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/805000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/805000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/805000/2"}}},"commit":{"parents":[{"commit":"b98a64e68dfd190ed00f082eb1413b0c88978302","subject":"Merge \"Make ARP protection commands compatible with \"ebtables-nft\"\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b98a64e68dfd190ed00f082eb1413b0c88978302"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 18:57:31.000000000","tz":0},"committer":{"name":"Thiago Brito","email":"thiago.brito@windriver.com","date":"2021-08-18 19:54:18.000000000","tz":-180},"subject":"Randomize segmentation ID assignation","message":"Randomize segmentation ID assignation\n\nIf plugin \"network_segment_range\" is not enabled and a new segment\nis required, if no segmentation ID is provided in the request, the\nsegmentation ID assigned is randomly retrieved from the non\nallocated segmentation IDs.\n\nThe goal is to improve the concurrent network (and segment) creation.\nIf several segments are created in parallel, this random query\nwill return a different segmentation ID to each one, avoiding the\ndatabase retry request.\n\nCloses-Bug: #1920923\n\nChange-Id: Id3f71611a00e69c4f22340ca4d05d95e4373cf69\n(cherry picked from commit 6eaa6d83d7c7f07fd4bf04879c91582de504eff4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0eba9efd2021fd04e2678714577582bfc6f1e932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0eba9efd2021fd04e2678714577582bfc6f1e932"}]},"branch":"refs/heads/stable/victoria"},"ab56a5cd652f57890723d4df5ba6ed22845070fa":{"kind":"REWORK","_number":3,"created":"2021-09-10 09:23:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/00/805000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/805000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/805000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/805000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/805000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/805000/3"}}},"commit":{"parents":[{"commit":"d961731a73128125ebf03fae52b4b34f3e7abf27","subject":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d961731a73128125ebf03fae52b4b34f3e7abf27"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-23 18:57:31.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-09-10 09:22:53.000000000","tz":120},"subject":"Randomize segmentation ID assignation","message":"Randomize segmentation ID assignation\n\nIf plugin \"network_segment_range\" is not enabled and a new segment\nis required, if no segmentation ID is provided in the request, the\nsegmentation ID assigned is randomly retrieved from the non\nallocated segmentation IDs.\n\nThe goal is to improve the concurrent network (and segment) creation.\nIf several segments are created in parallel, this random query\nwill return a different segmentation ID to each one, avoiding the\ndatabase retry request.\n\nCloses-Bug: #1920923\n\nConflicts:\n    neutron/common/utils.py\n    neutron/tests/unit/plugins/ml2/drivers/test_type_vlan.py\n\nChange-Id: Id3f71611a00e69c4f22340ca4d05d95e4373cf69\n(cherry picked from commit 6eaa6d83d7c7f07fd4bf04879c91582de504eff4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab56a5cd652f57890723d4df5ba6ed22845070fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab56a5cd652f57890723d4df5ba6ed22845070fa"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
