)]}'
{"id":"openstack%2Fneutron~805052","triplet_id":"openstack%2Fneutron~master~I78b8ea85548e11074191f46ef560728c4dc89405","project":"openstack/neutron","branch":"master","topic":"bug/1880532","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-17 18:43:51.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-11-17 18:43:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78b8ea85548e11074191f46ef560728c4dc89405","subject":"Add functional tests for ECMP routes","status":"MERGED","created":"2021-08-18 16:12:37.000000000","updated":"2021-11-17 18:47:51.000000000","submitted":"2021-11-17 18:43:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"805052-bug/1880532","meta_rev_id":"ccc6efc8772eef5bbc1d89fb800ef5430d816462","_number":805052,"virtual_id_number":805052,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-11-09 06:47:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-17 18:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-11-10 09:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"value":2,"date":"2021-11-10 08:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"value":1,"date":"2021-11-16 13:05:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 16:54:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-18 18:17:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 19:01:55.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-24 15:01:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-24 15:01:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-24 15:02:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"state":"REVIEWER"},{"updated":"2021-08-27 10:15:51.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 02:35:18.000000000","updated_by":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"reviewer":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"state":"REVIEWER"}],"messages":[{"id":"41d053ab6d4f18ee9efbd52414eacb5597d4c295","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-18 16:12:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c40ee5d237fcd4ae308c7768c601d06fcb791016","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 16:54:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e7046639cbf46e0ade90d2eb5542a89 : SUCCESS in 39m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/223d0c0fdb1c4079b249ecf8f7c218d3 : SUCCESS in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63ad4c827a787e77ae7819a774a66594149829ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 18:17:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2871a487594e42e89039edf065d8fdd2 : SUCCESS in 44m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d537c38f3f1d41788a812c1653c855ae : FAILURE in 2h 03m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3f2c501e8ae04c4595dc8c32bf35cd05 : SUCCESS in 1h 52m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f9b18fb1dbcd4994941ace446afb7bac : SUCCESS in 1h 27m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/93b5be897e1d42faa7d02785177ceceb : SUCCESS in 1h 46m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26e15d354e874835bb304f5f3b48c125 : SUCCESS in 31m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9424cf40b3d94f5ead1e076b7d68dba8 : SUCCESS in 34m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/977b9ea9aee74d518992b090a23b67ae : SUCCESS in 42m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f24f253276348cf83f2184c43e6bd78 : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c23cd533c7940219e5069ae29bdd380 : SUCCESS in 30m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ce78661f3ed4e40b0c8b53c3f86bb48 : SUCCESS in 32m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b285fc9f648148a4a9f7e6f1597a24c4 : SUCCESS in 32m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c82ec2a2b294ecbb26a40741d8df98c : SUCCESS in 11m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d7d35390f2b41249d139904d86070d0 : FAILURE in 50m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/590a0e4dc70b4afab0af0e0992e7f902 : FAILURE in 1h 33m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fb65da3046684426bcb4e9c75b4e880b : SUCCESS in 1h 06m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8f879fad10b42ec8060555198bd284f : SUCCESS in 1h 00m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/11e3f3a4347a412c8cb774384ddb2da3 : SUCCESS in 1h 02m 37s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1b5c308de49646b2a8b41113e105f4de : SUCCESS in 1h 13m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/581e6adef5754c538da864e1ce4a2956 : SUCCESS in 1h 16m 09s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/cd213ad1f6f344dc8cc436447e366fc1 : SUCCESS in 1h 37m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb61ac4c65774c70988f964b73c5186b : SUCCESS in 59m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e37db7746a2d48b8a6b581ab1def154b : SUCCESS in 1h 01m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9666633b9900429da8eadc83acf2b15a : SUCCESS in 32m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac321c1bf058405db106d45caee269e9 : SUCCESS in 1h 12m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/843b08b437b949cf9999c1708c7eae01 : SUCCESS in 1h 09m 25s","accounts_in_message":[],"_revision_number":1},{"id":"dfcfc4502e93e69217a2f484ca7b826437cb113a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-18 19:01:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/1 : SUCCESS in 1h 35m 24s","accounts_in_message":[],"_revision_number":1},{"id":"32ed2d0404c32fc912a48f4e672d8ed9e65f889e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-19 11:44:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d37f4a27742765639d740b6f6c4cd2b1112346cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-19 11:46:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50f30fc7b26ed36d05bdb9d99c0c0fc4480d89b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 12:30:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd9f0c9ba69e4ffc980334dc0fb3febe : SUCCESS in 42m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17f05c6092034c1fb23a9467842a9025 : SUCCESS in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a28c88d0b624dc08ce74409801854505300caff","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-19 13:38:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/3 : SUCCESS in 1h 50m 39s","accounts_in_message":[],"_revision_number":3},{"id":"f71827856a434bc650a3baf0667f4665fcfd0f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:57:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fc39e722dd2547909c018433383a8011 : SUCCESS in 52m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/88256cca092e4b538097d9def1c3ee80 : FAILURE in 2h 00m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6ac421185aab461f950041877a0f3724 : SUCCESS in 1h 47m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e5d7a1fadf34899872f9337de5222a8 : FAILURE in 1h 20m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7721052c5c4e48ec963b1783c211db10 : SUCCESS in 2h 03m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/14335b1ae3404beea88864c8ad880167 : SUCCESS in 28m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/100f09d6d68441819cde80e2e797830a : SUCCESS in 34m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/149c55840f42403590ba7d82321906f0 : SUCCESS in 36m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61081c5d6e7548f6a04020b85f5c623b : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68e3c4f496174f4da36acadb07348c1a : SUCCESS in 33m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cfc2d640df844bc9f57a2b4a39ae738 : SUCCESS in 32m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ea3b6c32dea4d74b08f1166890ee43e : SUCCESS in 33m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7086e29f200a4a57b7086e1f19f29c24 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/898d41f3a586405298d286e71ce44843 : FAILURE in 53m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/853b49b66b744e8f9dc6ca333f673444 : SUCCESS in 1h 45m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/93eee8c240bc495f85d74faab0f8d010 : SUCCESS in 1h 13m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3faa8f58bf3482fa17e2efab171a3ee : SUCCESS in 1h 00m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1ed6f84676164397830073a00cfce98c : SUCCESS in 1h 11m 57s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f1ab672105a14042b7be96f77f9e850e : SUCCESS in 1h 26m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c768eb6a124745b6901719639c73b355 : SUCCESS in 1h 17m 10s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/71167985155f4381885b8bae466e4163 : SUCCESS in 1h 34m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9b46827427549f9a335668f624ba37e : SUCCESS in 49m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/43576c504cee453ba18e8f54f44d2fa5 : SUCCESS in 56m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4d1e2d8245ca47cca3dcd42d8f76ef3a : SUCCESS in 27m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc5b120d5cea4e74bd73a5e68aa55bf7 : SUCCESS in 1h 03m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/37bd4799608043c593c3dfa1fdbebdcc : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":3},{"id":"0ca465d63d7f3a13d98c211fa6dbe8f98bca2b83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-23 12:41:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7e8298e1961c6d90ad8d81d85e77df278c376d39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 13:42:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f40cfed37a574c8790345c7fecee3302 : SUCCESS in 43m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d27ad0fea004740bc0e729500b5f102 : SUCCESS in 44m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3fd9aafaaf1055950c44cbb9460b34297975c36b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-23 14:15:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/4 : SUCCESS in 1h 30m 49s","accounts_in_message":[],"_revision_number":4},{"id":"6fc58a007b9bed0e960727ad0d0efd90aedcd3e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 15:04:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8a2c31bc2ac949868808efd1a6204c7c : SUCCESS in 48m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2439a7237270453fbfa0609098f706c1 : SUCCESS in 1h 52m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b0d4a9c671434cb9b2f579b18933c25a : SUCCESS in 1h 59m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d255c868e214c3dbfa8be30a55b9446 : SUCCESS in 1h 35m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/179276534e2946149c344784ed631df7 : SUCCESS in 2h 15m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c096b5997b5d459ca5ad75f2c0734069 : SUCCESS in 33m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab5b59e5472c43e6af11f24994f94383 : SUCCESS in 37m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11d6173af7e640b0a7092d9ea040d63e : SUCCESS in 49m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d048562d76d94962a761dba472d7784f : SUCCESS in 16m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e43f6eedcd2461791917cc75bcde812 : SUCCESS in 26m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/518ec5eb6c4d426cb1940e4dfa26b3d2 : SUCCESS in 27m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e3f16b27bce43739dfe08a028656e01 : SUCCESS in 25m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c3c68154174683a4bf2c9362207f08 : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d6c75f115264483b3ff2df718431603 : SUCCESS in 58m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3dcdc1289974b04a2761ed8b34ea5b2 : SUCCESS in 1h 52m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/30bc1f54c59347dc9b2442a7472afda7 : SUCCESS in 1h 11m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/db74a2d3e43048cb97ba8408ba7d7de4 : SUCCESS in 1h 03m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8cceef15e92d4c1baf18ca24cb79d5cf : SUCCESS in 1h 05m 05s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/69f4af67911f43c8856bf3453ab68851 : SUCCESS in 1h 09m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5d04d4e1854f465186c1ea20eab01bab : SUCCESS in 1h 17m 37s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/311b9f82e567482991e13bcd2adecb16 : SUCCESS in 1h 26m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de573d3fc696497ca39066c75e34f898 : SUCCESS in 1h 08m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bf41c6c1fd7945408f49faed186f2ff5 : SUCCESS in 57m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/424df8a5a09c4c6495097012f225bb5d : SUCCESS in 33m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5006fbaa08634a3ea8ee0022ba28d1eb : SUCCESS in 1h 05m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/18327a4573cb404691d4b4ffb1b583de : SUCCESS in 1h 16m 27s","accounts_in_message":[],"_revision_number":4},{"id":"85d59663295807d56b9867bbef7a95013dc3e084","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-26 10:59:20.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"67918565d484e1b3441daf87a467b7b551423bce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-27 10:15:51.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/52/805052/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a3d47c954bdf5cd69e96e3d77f0fd1e183e869b3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-27 11:19:19.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/52/805052/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"aab3ab277b219be3820c9abd4a912ac0bca93b77","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-27 12:16:51.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/52/805052/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"e69ceb9dff96493d055a6d9c6a2d6d86783d61d2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-28 05:44:33.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/52/805052/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"1df2c8f54984ca386fa45582cd20951681cf54b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-30 05:09:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b9756f7eb264350619cea8beb599288036a94a3e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-30 05:10:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e2a4b35f3dda578ef569b225548bafb03c144486","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-30 05:53:27.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/52/805052/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"fe4fe72a153b415ae603691adcb04fe643a964ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 05:53:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d9871a5aec04881a3f6977ffe7aeccb : SUCCESS in 37m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c1162a025aa4766988f1863a3557e99 : SUCCESS in 38m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cc466b75aef538a84d499157eba02dd4076c0334","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-30 06:56:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/5 : SUCCESS in 1h 46m 08s","accounts_in_message":[],"_revision_number":5},{"id":"01d70d747266c0eb5dc8d617fb22d5b76cdd0fba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 07:24:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2dc8d82cc70f439c8a12d499f097d2eb : SUCCESS in 39m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4faa779981fc4e46b7144462665785a9 : SUCCESS in 1h 33m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ee82038756d54bd49aad238210be645b : SUCCESS in 1h 35m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7557c89a1154aedb05832364a59d962 : SUCCESS in 1h 22m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/14a90027f8d7490e9fcc8bfc9b5d8bfe : SUCCESS in 2h 13m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1e71a36bb6ce4fd3b26df985c9058954 : SUCCESS in 26m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/292de26e57e0411ebbfdab8ccd03bc35 : SUCCESS in 32m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/147d56496ae641ef98e9206e5738cf6d : SUCCESS in 36m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513799f4d5bc404caaf9e9989ab7fc42 : SUCCESS in 17m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64bbde5bdc3b4472a61814f5e28db2a5 : SUCCESS in 29m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa9c5116c715465fbfdf31a615802484 : SUCCESS in 23m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f209d69f31243b19d9940d25044ba22 : SUCCESS in 30m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0154736f6f884c0ba577e4fb4bf1d494 : SUCCESS in 9m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eded7f91e2ac493cb58ac198dc10b46d : SUCCESS in 44m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed8eee409a6e43528a175828800fab69 : SUCCESS in 1h 47m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/65a7635d203d4864b9443c94ebd3d106 : SUCCESS in 1h 05m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2fa82f164df4b65b843e089fee7d8db : SUCCESS in 58m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/623600a6c83f4f349a4e17b8c72a59ba : SUCCESS in 1h 02m 49s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/be9044ab1bc645f8a8d047fa38e1c149 : SUCCESS in 1h 03m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/55d289f0f9d34b1c897ff1a232cb57e2 : SUCCESS in 1h 10m 55s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/2cf656834d9443678718ca68f029d59e : SUCCESS in 1h 26m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/437ad3e34b994d23951ebb05479f5594 : SUCCESS in 50m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/27b4382c030a4c60b3efe3561e159708 : SUCCESS in 57m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b408fa982a12485c845df882e3c0046c : SUCCESS in 26m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c26ddc37822b4a5593feed2eb37cecd4 : SUCCESS in 1h 03m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67f33ff3d2114eef843bd8a7d2f9257b : SUCCESS in 1h 24m 59s","accounts_in_message":[],"_revision_number":5},{"id":"c67cea1a6e7b2f9c2666d5b257502739d37738d5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-30 08:11:27.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1b389d01292d6a08804ffc0b4272e94a153c99c9","author":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"date":"2021-08-31 09:28:09.000000000","message":"Patch Set 5: Code-Review+1\n\nI\u0027m ok with it, thanks for adding these functional tests","accounts_in_message":[],"_revision_number":5},{"id":"b2dadfec26ddfe64cbd5b05912ae6945ccf915aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-10 12:56:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8595a5acb7a32fa29c9df227e241e110ead4c074","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 13:40:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eba28474af7540ff821379f2ba09e121 : SUCCESS in 41m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae13ceaf748447748c740ae9c4192b99 : SUCCESS in 41m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5a7c86ab33ed55ff85c989b09b8171bc288fc89e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-10 14:26:05.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/6 : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":6},{"id":"8bac2d5aa49555107fa178ceb6d6a5fc6b6acf93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:00:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5fba362fb35b483f83497b95b2349734 : SUCCESS in 39m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f1d2fdce749d4bc59b0580b676b545e5 : SUCCESS in 1h 53m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c5037c6ac25e490f91004568b29d1ae9 : SUCCESS in 1h 48m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/14811352d08448db9de9fc9a9d11e01e : SUCCESS in 1h 31m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4197a6c445d8483e95a8104ef009380f : SUCCESS in 1h 55m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ae3196bf2bf4434b4cee635f79cf9ea : SUCCESS in 32m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48c60c0601d8453aaec08fc807cef6a2 : SUCCESS in 34m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09fe3df28499432eb912e8ca98ba6fc2 : SUCCESS in 52m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ba31233e844453b6f65f6832475c80 : SUCCESS in 14m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20cfe553f4284db38ccde7effef76180 : SUCCESS in 37m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b531fd51027416fa18d95fa020d71f3 : SUCCESS in 24m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fe7606b43bf4755aa7733ef6c2c6031 : SUCCESS in 28m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e699645d95d4db1adef053e5118a61f : SUCCESS in 14m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48a64aee5c7e4b35a1a90bf6c8d9ed90 : SUCCESS in 47m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/94f3b579697d42458be21a12a63c6194 : SUCCESS in 1h 42m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8e590a201ae04666a4389048fd7a7af2 : SUCCESS in 1h 08m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/08fe5fa3421e4555bc44798d8270640c : SUCCESS in 1h 00m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6109066b98ee4a93b7a71c32fde320fb : SUCCESS in 1h 01m 02s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/538de68f2a6e443eb1fe977d9b77329b : SUCCESS in 1h 06m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5fa9f12ae7af4aa79e893dc32695b680 : SUCCESS in 1h 15m 44s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/8515fe0dbc0049a7bebe96279d617a26 : SUCCESS in 1h 29m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d811f6544bbc426abac5990273a698b7 : SUCCESS in 58m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7c6d74ea4a334bfc887cdd513675d2cd : SUCCESS in 1h 00m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/67762338ed9b4faebca6f9a13162128c : SUCCESS in 35m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/503ed017fa3c4a29856bd042aa5d9959 : SUCCESS in 1h 09m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1855e320ca53415b8e2526f5dc628266 : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":6},{"id":"68970edb0b1766e828254ab9bfa64efafc32928b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-14 19:12:17.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/52/805052/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"8577a5fb757abd0eeb336f990f52a1b7a0bacbbc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-16 13:24:05.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a6776969bb65d3d23f5097b6d548eb872718640a","author":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"date":"2021-09-17 02:35:18.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nNames are suggested to be in same format as test_dvr_router_update_ecmp_routes","accounts_in_message":[],"_revision_number":6},{"id":"12d42f1d8bc15010b15053a14eafddaddbcaefe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-17 09:07:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"539d974fdb5da95e4acfebdf997f0903a5f8e21f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-17 09:17:10.000000000","message":"Patch Set 7:\n\n(2 comments)\n\nthanks, good catch","accounts_in_message":[],"_revision_number":7},{"id":"d8bd8be6c65d7eb5411c47ea1fa4266d40dd81b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 09:56:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/091abf8be04f4444a580e7658ce53db3 : SUCCESS in 45m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf58c19bce9c4b9487d90f64b95d815d : SUCCESS in 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5b5b9c865fcf18a3d63d61e2c858d9ea50fd6006","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-17 10:37:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/7 : SUCCESS in 1h 28m 37s","accounts_in_message":[],"_revision_number":7},{"id":"37e6af21dbc9974fe2f1a77047165ad0b0acbb58","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-17 10:59:04.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/52/805052/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"41bcdb6a08b8453c24b68ffd30c35982297e556d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 11:04:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/317d02b930ee4d8e8bd6310d9e21c491 : SUCCESS in 39m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ebff433d1bb3431bb32558c662a7ecce : SUCCESS in 1h 51m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c1498aad51f14ba29d76381fa6a64cab : SUCCESS in 1h 53m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5686520675d4225a4a091f0d12929f7 : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/35396b1b706d4b4b8f985282800a6b07 : SUCCESS in 1h 45m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49c1c8edd4c340a8bb95938210d6872a : SUCCESS in 28m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f41bcbd2e9c4431ca3f00f95836e71df : SUCCESS in 31m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d509e0972484fa185fec671e39621d8 : FAILURE in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba67cd93e8b4b2898fcac8a5960fdfd : SUCCESS in 14m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cf4e07ed2184c999263298841cbd991 : SUCCESS in 24m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2300ae09d1df460fbf87ca025e37828b : SUCCESS in 22m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/124e9996f2364ebaac218a834f8000a5 : SUCCESS in 24m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0128d2ce2ae64cd985c95d62d38a5446 : SUCCESS in 11m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2ec36e9a02946749e7697628f0123e6 : SUCCESS in 44m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8cc6b1fe09b74238843a5fc71e9012e3 : FAILURE in 1h 30m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d09f638169b24041b48ce771de4df3ba : SUCCESS in 1h 05m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7326a3fa0ab74c5cae2a5ba387bb3a0d : SUCCESS in 59m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0058841ba5174501be1ec977c1b5800d : SUCCESS in 59m 27s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/807514552a8f46508c97aa5891610910 : SUCCESS in 1h 04m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3c34d663d68c446b9f0dc965fb651585 : SUCCESS in 1h 11m 39s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c63ff4c76bdd4132be2eba2aee8e4f70 : SUCCESS in 1h 14m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76a7a9e69cdc49ab98c0a7882a80c031 : SUCCESS in 48m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9bef91b3716a4779ba4b76b9f52113fd : SUCCESS in 48m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7a43f9aae4a24a4c89374eec32802af4 : SUCCESS in 24m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b061af3ac2af45d388f070ea50a2850a : SUCCESS in 1h 08m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a3ad42436265434ea53622262e288bc3 : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":7},{"id":"ce60b1f15819a511053aca1ccc5fb853b3875177","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-11 15:49:40.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"f70752fa632035a090f0e3813cde986852aabb94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:52:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44030e76dff845499addc22cbd36e188 : SUCCESS in 41m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f3ae58abc39487f8cd4fb72e713d9fe : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"026a37a25e3b1a959a583606ad1854393c88658f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-11 18:00:26.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/8 : SUCCESS in 2h 10m 05s","accounts_in_message":[],"_revision_number":8},{"id":"85320e0c76f7683102e69116310dd91ebaa3a18a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 18:02:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7cfeca0be5ec4addb7ab6936c6973b60 : SUCCESS in 45m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b21b5aa0bf1e4db99a027698d30daa15 : SUCCESS in 2h 10m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d445d3ea1a704b119d2ecf90766fc2b7 : SUCCESS in 2h 03m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ff224776585a4fd4a1eb1fd4b24ed6a6 : SUCCESS in 1h 20m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e81cb1d23fe1408d8a40785903e26594 : SUCCESS in 1h 51m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc0bb5b462a64ceca1f45fb9a6cc71c7 : SUCCESS in 29m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ba94a5b2de747f5921336ffa52f9e4a : SUCCESS in 42m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5112d69b96354248bfec2eb980668ec7 : SUCCESS in 39m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/014bab34f5724657ab3b8d3d47a0831d : SUCCESS in 13m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1909f1843d754cf1b3f8b716ffd4502b : SUCCESS in 31m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffb6ef83d6264a2b98ed70a0e5b34226 : SUCCESS in 25m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8894b4e220e74a9a849d723800d82fc9 : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59916450e08840beaecea152b388f7bd : SUCCESS in 9m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3cacdce3a8824bae98f4d6623af78ce0 : SUCCESS in 51m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aad517cce192490db3dc45477ee6bb04 : SUCCESS in 1h 33m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1183f0bacc50446c8c7eace2952bbcfb : FAILURE in 1h 06m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9bb665834b0447b8b135ea80cb71f45c : SUCCESS in 53m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0bd37c33421244d79880ee4fb809349d : SUCCESS in 58m 36s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6b2e34fc3a874c9bb3c124c2a1156b0c : FAILURE in 1h 21m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d21f08ed12e046569f057a88b8d66001 : SUCCESS in 1h 21m 30s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/dd35d2c6af774be69ecc48d1aa16371e : SUCCESS in 1h 36m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53e65aef99104ae0a24ce13ec005230b : SUCCESS in 59m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3132be650ebe4fdb9f7be15e1824cb8d : SUCCESS in 1h 05m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/724ee1aba5354b8e87d12e7dc0540a74 : SUCCESS in 33m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f273fcbe207945e69c9bf994e0e76b88 : SUCCESS in 1h 08m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/930bc2fd440e429e822134f89fa026cd : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":8},{"id":"38a5add7b6df3e24330fa841b8403c97a7be01b2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-11 18:50:52.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/52/805052/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"6efda2fde18ec285327ad02fab805bee18aec9fc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-13 09:17:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4b5a7860edf598dcdc221f5801b0a1082fbda085","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-13 10:47:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/8 : SUCCESS in 1h 29m 54s","accounts_in_message":[],"_revision_number":8},{"id":"d1b3e15f6a7029d1cc03c8845c90312e8781b106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 10:55:39.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/db916f1a2b6d454099cec1ecc4eb97ed : SUCCESS in 30m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2498319fae324f678619a1d706abb854 : FAILURE in 1h 28m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1bb2fb9630654f0faf69fe1a060ef6cf : SUCCESS in 1h 36m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4d78359b27f3455fadb7dbe4320e1aeb : FAILURE in 1h 08m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9c487ce91406478a8ec647ccba570954 : SUCCESS in 1h 30m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bff5cde7ded44b34b12a59eb921c488b : SUCCESS in 24m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9383fb1a525445d982a2556f723a060 : SUCCESS in 32m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e611fb1994fd4334a9766a0c78c44f73 : SUCCESS in 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4bc8a304524afcbd8ee75e48654393 : SUCCESS in 11m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c67b4271a63246c59c48257444bffe34 : SUCCESS in 23m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26192010dccb4df5912ee3a538e35bbc : SUCCESS in 21m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb15ec3b699143228d5003030e9a8163 : SUCCESS in 28m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90531187bc849aa924fea1d7de2ae02 : SUCCESS in 13m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c747b1b53de4b60b2d9b17d805af747 : SUCCESS in 41m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4306ebeca44648dcb04a66cc46789b37 : SUCCESS in 1h 34m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/475506f01dd943a2b5c3f5cff491eab2 : SUCCESS in 1h 02m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fb3bec696b641fc840273178c1ac303 : SUCCESS in 50m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/de5d60706ebe4e94bf3e518b2eb30761 : SUCCESS in 52m 55s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/99a4ff77f28146a5bbcbc016f400e0ec : SUCCESS in 1h 11m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/365d1b6f8ccf41ceb5c46c9a53f10ead : SUCCESS in 1h 26m 23s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/ab162b1ca770451086951af66d9c25a1 : SUCCESS in 1h 27m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8e3ed5db163467c9a3ddad59786d530 : SUCCESS in 50m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/263d89ff08134bf8a948a37c64e305cc : SUCCESS in 50m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0b644303704047c6a929f2baf365ce9b : SUCCESS in 30m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/42cabc5463c840218dfab91f4aaf9061 : SUCCESS in 1h 12m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fcc4b68275ed4270a020f26a264cc359 : SUCCESS in 1h 12m 46s","accounts_in_message":[],"_revision_number":8},{"id":"8c9f5ffba7e41cb1f767e0a592c5c4c0267f8f44","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-13 11:25:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a94c73629851d9fd316372089af09756dc21758d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 12:11:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04da9172b5fa466e8a6f82a3d4a0486c : SUCCESS in 47m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc4d31d573b6460eb70a4e2b0d072acb : SUCCESS in 43m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"daf54c8e21c304910f79db8266020dcaf1d2aa2f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-13 13:22:27.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/8 : SUCCESS in 1h 55m 48s","accounts_in_message":[],"_revision_number":8},{"id":"8ff14e1117038d527d09c678c5cd312d0e9f1232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 13:38:18.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dfb09d3d771d4dde9280aae746562b9c : SUCCESS in 35m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/20db1703ef014848ac5668ad6a73b8bb : SUCCESS in 1h 41m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2c32c23c2c874cebadcce4761bdfc969 : SUCCESS in 1h 45m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/118cbbd3a7844e8f9b2f95b551ed8a92 : FAILURE in 1h 20m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/74d1fdeec75e470b998f0bf2103fe48a : SUCCESS in 2h 10m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/15dcc30ad80b4c81b3ffecf768e38200 : SUCCESS in 24m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f3fb20b6d9249f5a81aa55b01378dc2 : SUCCESS in 30m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70777232eb3540709e7be3df67ed39c0 : SUCCESS in 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd89c78b14ca4602875e5a01cb3d9ac9 : SUCCESS in 12m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00e4205c7c18484f87fe835ffcfb87e7 : SUCCESS in 26m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/452bf4a8d17940bcb675acd906ce47e9 : SUCCESS in 22m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88e5a1374a3544ddbb7865ef15f9bca4 : SUCCESS in 30m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95cbcf8d286a4101bb9be943ff028a9b : SUCCESS in 8m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/93942222db0549709722e6df0cab746c : SUCCESS in 53m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a75fb4ddd2c147eeb80c154f956c5b39 : SUCCESS in 1h 43m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/94a41dda44024836b686963c2ee0a29e : SUCCESS in 59m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/795116afecce4747933da38e8d2d48f4 : SUCCESS in 44m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/79ce93f684f24728939a85bfaf11e407 : SUCCESS in 47m 46s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8d387f41141744299aa3b98260fdc06d : SUCCESS in 1h 00m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/022d2facf27b48fc9dad2abe9e0208b4 : SUCCESS in 1h 04m 57s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/a4a75b93f29948e5a6f61e9ab4087f87 : SUCCESS in 1h 12m 34s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21c31043dd584da0bf4d49914f335c9d : SUCCESS in 1h 00m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4fdcb10ffcbe4f9c8e9fdcb2d2f72264 : SUCCESS in 59m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/194b72c9228648a98dcd5c584b8341e2 : SUCCESS in 34m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/92985dcd1f7d4799aa6e3f76f95f1e6d : SUCCESS in 1h 26m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/00caae7446da45159295e03ccb3c9626 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":8},{"id":"af29d58096a539efba575810c9cc14abc35a5bf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-15 10:36:07.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"b08682c155127a55d18f6b3edb35ef39b50608cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 11:18:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e19d2afe51474df09e109ed0a15e0fe0 : SUCCESS in 37m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/45c0ebe8e8724c839e3ac19944a99203 : SUCCESS in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b555d0c73452c8963266aa951623d4222385fb0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 12:20:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ef24fa2bbb4b4d8b9a1179be961a6991 : SUCCESS in 40m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0bbd74a79b7f4d9f983e5c0a6906188c : SUCCESS in 1h 39m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e14b4ee044094504b657c2708bdaa8de : SUCCESS in 1h 27m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/19dcc3e3c128436190e37711988ffb01 : SUCCESS in 1h 20m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ce42a41d91ae46b8a1b45e697636c65a : SUCCESS in 1h 23m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/95cca809f4bf4fb1bbd81325a750fa67 : SUCCESS in 31m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c762095669274b82bf85e5323eeb2a93 : SUCCESS in 29m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fd42c3c29574f45b5a820a7150af492 : SUCCESS in 34m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4cee73d1e254eef99926e3842e7f9de : SUCCESS in 12m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0b25e76131545d9988ec0b9f3122237 : SUCCESS in 25m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ae0b7e54c69476eb10ddf2c7229e562 : SUCCESS in 25m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d4c17628b994b2ea4b1efc464d95e73 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31cf8c7db5c54506b350b845ecc4cd42 : SUCCESS in 11m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec642db0f9284575a8d7cc2f0e117470 : SUCCESS in 49m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/87eb8ce488e4491d99949ce1db4fa6cf : SUCCESS in 1h 38m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a4aa22d297f2415ea34d6760c7142064 : SUCCESS in 1h 07m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/658179422f7b4bf997f464cdc171172a : SUCCESS in 53m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d3ce485f8a564f2c8a2fb19fa6e3170e : SUCCESS in 55m 46s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/edea9ae3b3ad44c59104801a28257873 : POST_FAILURE in 1h 09m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/51ea2cc7fc324c34ae21a2912b809064 : SUCCESS in 1h 10m 26s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/aaaec6c77a1c437f9db3f0dd5d81c320 : SUCCESS in 1h 25m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28c7c05f421b456ba57ce48ed960ad8a : SUCCESS in 1h 00m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ef753df0505b45ebba72e0443022644d : SUCCESS in 53m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4856aa82bd4f4f52bbcaab4f80f184d0 : SUCCESS in 34m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5f6bfd05d9c048bc8802c389ac185271 : SUCCESS in 1h 05m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f2e519f3057944eea49c7bb5fd783a3c : SUCCESS in 1h 19m 07s","accounts_in_message":[],"_revision_number":9},{"id":"84cf09d747a523146ad76f31efceae3591e2ddaa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-15 12:30:30.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/9 : SUCCESS in 1h 53m 32s","accounts_in_message":[],"_revision_number":9},{"id":"80a636ec19aea6af8953e392f79d768093784561","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-15 14:02:09.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/52/805052/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"e238cc50765aa19a7afb448e6b6aee8911b31014","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-20 09:04:48.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"44434bb0d51121fe8a9fc068656f6cfa7686a594","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 09:49:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a049df0bafc248db94b3056a155e10f3 : SUCCESS in 42m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e682a5bcbeb4c55b63c148a22acde5c : SUCCESS in 42m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fa6b7edaff3eb65579038719ae548e9d1eed16ef","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-20 10:40:00.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/9 : SUCCESS in 1h 34m 37s","accounts_in_message":[],"_revision_number":9},{"id":"c5e36ce18fc0df7a7f2af63da05fa516e56351ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 11:24:58.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1f8524d1f627471abe0e6d53a2e7dd20 : SUCCESS in 48m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/58f48d14e54741108dfde3760ae624ae : SUCCESS in 2h 17m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b25fdd45d09a4ee699ea4418ec27d90a : SUCCESS in 2h 02m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6ed0047a6ad9430bbe6da66bcf5e560c : SUCCESS in 1h 48m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9e8553ef256a4fbcaaf38088fe895f7d : SUCCESS in 2h 00m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b484e216b1cf49129f924db31f836c7f : SUCCESS in 49m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b6b3573e6b94f23bf3560eca1d38138 : SUCCESS in 33m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19192ed3d719489d8a0e3a267540c12e : SUCCESS in 39m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de568dafc8844e359a0dd48962c19b2f : SUCCESS in 13m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a2ce0690e224fbf9018ab91a0346c43 : SUCCESS in 31m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/092247ecdac341cc86319aa3f862db03 : SUCCESS in 30m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77654fa1be5440f1bdf4d135d902ee24 : SUCCESS in 31m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a899f6865f4f5f8acec19b0a9b6537 : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d148c45daba4d18820e4e6d45a41cca : SUCCESS in 43m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/133c4d0fd8ca4196967530637c44dc71 : SUCCESS in 1h 42m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/af80361007c24f909905a248fbf495b4 : SUCCESS in 1h 11m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/81c273584e194cd286d22eadcd82fdbb : SUCCESS in 1h 02m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aa774ee974224b00a6dbed0dcc49fbf1 : SUCCESS in 1h 02m 23s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5205c6d674e44331bcccd78586bce112 : SUCCESS in 1h 06m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7851d7b873e74bafa7fc4c580ccde9cf : FAILURE in 1h 12m 55s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/aa6b792f58b34d889d88bd3c4e5ad98e : SUCCESS in 1h 58m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab6299c103d2486584e15287591dc4a7 : SUCCESS in 1h 05m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e4981d25564d4f009e7e4060e7439494 : SUCCESS in 57m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/75a6d74298594d72830d33125fce8e42 : SUCCESS in 27m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6bcbe2297e2a48c989c198370181f709 : SUCCESS in 1h 26m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5644cc36b8324fa9be84b55d2eeb9455 : SUCCESS in 1h 39m 56s","accounts_in_message":[],"_revision_number":9},{"id":"de3014620af4d726aa69d7ff8cea6be136fc12fe","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-20 12:10:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2ca8fbb98788bd728fc630e83400cc7be9405303","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 12:55:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87366e9b6b4548c2b8aca035d7bac680 : SUCCESS in 36m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb2c1b9d477f4673a93b00ef9c2ef3ab : SUCCESS in 41m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"491969c2b7a59518e257cd8dfe8957cfaf4e231a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-20 13:42:03.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/9 : SUCCESS in 1h 31m 00s","accounts_in_message":[],"_revision_number":9},{"id":"934bcbfd8c31f0fd718a06bec4a2859599d88b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 14:05:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3d6c7aad099341e7b8d612db074ce299 : SUCCESS in 41m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8b98d35177174351acb3c4c2eb1071ed : SUCCESS in 1h 42m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/08e468a322ee4d8aa8cea83d3c22cf93 : SUCCESS in 1h 31m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/093cd65b9c8a4589a005e5d6c979f7f9 : SUCCESS in 1h 23m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5f22c866f7e84c6898165e087b692951 : SUCCESS in 1h 47m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/db031b7d5090496299cf7bd8f2199dea : SUCCESS in 37m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07e49b2c14a34b46a056e00fafc3d336 : SUCCESS in 37m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83f72dbfab0e45789e554d98d62cad0b : SUCCESS in 41m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7d8bbfa13ab4c8abf2f6f5e4f536302 : SUCCESS in 15m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53918ccbacfd46f79077c1bc4780252e : SUCCESS in 30m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd4cc6d1bc9c4011addd326783a1604c : SUCCESS in 28m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9901b77541b644c39cee3a5a3f0f13ee : SUCCESS in 22m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79700e3f47494b4295633b3272d518e8 : SUCCESS in 9m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b61fbe6a049c46769579335e53865536 : SUCCESS in 39m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/47f8d5adfa0e430487498f166ebe69c1 : SUCCESS in 1h 29m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a876ff3635b64315bd3732c9e79662bc : SUCCESS in 1h 06m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ae392281593406ea18d5ab0f6d5230d : SUCCESS in 43m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/511535e48c454deda7307ad2e8ec2046 : SUCCESS in 51m 52s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/54a0dbf51354404e9ca673e1966d1f50 : SUCCESS in 1h 05m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ba6a6767b1cc4c9d8d55a8958834944b : SUCCESS in 1h 10m 14s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/ed90b7d148454e07ae02c52df7a423e1 : SUCCESS in 1h 27m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6213620272f482cae27d8a7dc05d529 : SUCCESS in 58m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fb1d5c7eac654f1caf03478f2ffd9ab1 : SUCCESS in 1h 12m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8c08a3a70bf9412e9c1ca33db8f2391c : SUCCESS in 32m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/265ca351566d41109cfbb2ceb8b729eb : SUCCESS in 1h 10m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/93e1d5a0f5024948a04603854d00ade8 : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":9},{"id":"dd0032bb375f9c9e4eb6067b2f78e00005ed9742","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-26 13:28:42.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"338418cd4fe06444fdf4eba1c26e1a4adc3c45ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-08 12:48:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6293f80663a8ad4ab4356ed07383ca0ec25746d0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-08 14:25:13.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/10 : SUCCESS in 1h 35m 24s","accounts_in_message":[],"_revision_number":10},{"id":"45b608c52ad9c5e235382d7b8346d63e3ad62dd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 20:46:37.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57e1121432864a7fbbf5be938e48b889 : SUCCESS in 56m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03b74df1e11144e18c89ffe6d9aee3af : SUCCESS in 56m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c7de913aa5904a848a31545f66ab028f6c2784a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 22:24:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cff3171a813e47c48fef5bb3c81f715f : SUCCESS in 1h 01m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ef48a27f7c304e40acd9fed61b0fd0c2 : SUCCESS in 2h 39m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9976aa28389a477f9d63094553d1806e : SUCCESS in 1h 54m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9169d5a5e42d461d86d8c2e7f6d6d5c6 : SUCCESS in 1h 39m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fc5ad26907c24d3aad9017c62b7f5b3d : SUCCESS in 2h 18m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5386814c41fc42a39cf943f9077b90e6 : SUCCESS in 36m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14cbf505eb5f4d6699aab8b35b98d08a : SUCCESS in 52m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e92d9d042b7643ecbec1594e7ac53ef5 : SUCCESS in 51m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9572fa83bcd74254a36cb6d9e06771a4 : SUCCESS in 16m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c0e9fc065474154848844d89631971d : SUCCESS in 39m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee47192a91249419ed3c65d509a1dd9 : SUCCESS in 43m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2a0b35a21c7431fb784d1564dbcb568 : SUCCESS in 41m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c6ef88e4dd4ae1b4031ed8b6d69b45 : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce320352654a462a97eac118086503f6 : SUCCESS in 1h 01m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/634825dc1f2b486a88dfd34cd6820327 : SUCCESS in 2h 14m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8c94095edafb4ee780a3117522e3448a : SUCCESS in 1h 10m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/920db6a3911c41369367f27adcdfabdf : SUCCESS in 56m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9030665b12b448d9bb757784cb52acc3 : SUCCESS in 59m 31s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f337ff16f24048f8a8af4d8187e29d5f : SUCCESS in 1h 06m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fe668f4823d54e1a9b95206b0a6bb0bb : SUCCESS in 1h 20m 08s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/d35d394d6b244b48b95a67f0e4c77a90 : SUCCESS in 1h 43m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2b5caec2e17454984d2d632247910b4 : SUCCESS in 1h 00m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/12afb4aa323840b8868a599074b1e9fd : SUCCESS in 57m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c0b220d345c44ee09a079a560f3117b6 : SUCCESS in 32m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/35ca248b263b4a759affd214c0a3cfbb : SUCCESS in 1h 08m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8b59821e7210497590ad4de238673ac5 : SUCCESS in 1h 32m 18s","accounts_in_message":[],"_revision_number":10},{"id":"76fd0ff9fb76529d14c57c2b4fa06a1f344cc07f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-09 06:47:09.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/52/805052/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"7317982cc6f3eb21028f04ca199495ffead1726d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-10 08:34:28.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"3bd193d4a088ebee42654af4f3990381259c756c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-10 09:35:47.000000000","message":"Patch Set 10: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"efa85c232c07b58654e2db762aba0b6c11f15d11","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-15 10:33:21.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"925f89ffc5bcf9a96a0b9ceb536bd4ef0893ab48","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-16 13:05:21.000000000","message":"Patch Set 10: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8fbdaa10cdd0e8cec534806dfc10b04ae1286ec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 13:06:29.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"fc801c42c14c3d7492018c08d7c42f47093e9f3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 23:38:52.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/80444b4a0c1040e592f3a15643d483ec : SUCCESS in 39m 55s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/192734fd364140ca992e70782e1e916b : SUCCESS in 1h 46m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0dac5cdb16a54042b04c561827aeecc3 : SUCCESS in 1h 49m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fa640d56b1e5450ea93e194c08c84e66 : SUCCESS in 1h 50m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/541eed01cbef4cb3b681f637f782eb66 : SUCCESS in 1h 34m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/668eb9ce80344e7aba3fb44e27cee4eb : SUCCESS in 55m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d0480b3eaef4a83820268468c6bccdb : SUCCESS in 13m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de8a4dcbd6734978817cc318e402ba53 : SUCCESS in 44m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c257e89401cf453b8adcfeb9c31374e9 : SUCCESS in 38m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d29259723f824c00a71c5802ceaebc49 : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ecee06fca404295a876895fdfaf601f : FAILURE in 48m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd77c6881ea2460fb91d2daa309bdb71 : SUCCESS in 1h 27m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bd89f324681140999ef16ff3df08a0ee : SUCCESS in 1h 20m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/135e229da716402187b6bab528fd3253 : FAILURE in 1h 27m 23s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/cd7b999e80814b009073c6e01bdb7243 : SUCCESS in 1h 28m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cb91ef6d3b0409a8b9b0b4544d164ed : SUCCESS in 59m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2fb89eb2a7a44b68bfe45724658eff45 : SUCCESS in 56m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/983c5da0d00445b6ae3ef96009396eef : SUCCESS in 34m 53s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fe7ce81baef7447f98803c5933eebac8 : SUCCESS in 1h 13m 50s","accounts_in_message":[],"_revision_number":10},{"id":"100918e90f13a55975aab17704330bf665c29d71","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-17 10:58:20.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e3a5dc6803b2275442d4d935cac57b99fdfa6036","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 11:57:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a19ad16384148a2a82af8c56a822c86 : SUCCESS in 55m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74a29db485124b5a89dce075b40a65bc : SUCCESS in 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a3a3c2265f1e489a8abc86e9247574797f229a56","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-17 12:51:39.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805052/10 : SUCCESS in 1h 52m 39s","accounts_in_message":[],"_revision_number":10},{"id":"78596a33593bc738e3cb05bf168f550132dc07dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 13:26:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/690a863e4e73486e887428a4c2bcfdd1 : SUCCESS in 43m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/280dba9bed024c3e9ac14f62950558a8 : SUCCESS in 1h 57m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8279926d7e3541ee93b009acd239fdb2 : SUCCESS in 2h 02m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/71760a495fd9476a86861bae42fb80d0 : SUCCESS in 1h 43m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/80109d32c6bd478e89005b28209c0d19 : SUCCESS in 2h 24m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51a14298489c447982cfaf73d1c23143 : SUCCESS in 40m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33efb51e47364b209e20cff8bb198bd3 : SUCCESS in 45m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/465df58f664b4db7a959d01d9a7198f4 : SUCCESS in 49m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497b3c502fc743a682baaf1987062d6b : SUCCESS in 13m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/168074d40b204df8af7cc871439af428 : SUCCESS in 45m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38548e56c3944700b7fb94327ec341b2 : SUCCESS in 40m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c56fb2dce464e46994b84d4700a180d : SUCCESS in 39m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50449f5cb9a045a08790d589bb818cf7 : SUCCESS in 11m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ea84881296c43fbb1cd394f2cd82d48 : SUCCESS in 46m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f1d7a7116f14228a39a4a428d3b48d4 : SUCCESS in 1h 32m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5d589ca98de14990ac5d148fe6a71839 : SUCCESS in 1h 24m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f04d13ad2364d07a3665ac04290cab0 : SUCCESS in 1h 01m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8a33325c69dc49189d40afe070c255a6 : SUCCESS in 58m 46s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1b1659be1c0b451398a12ef365008999 : SUCCESS in 1h 15m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bc36a38cdc9f4cb69b50b76021b4ecba : SUCCESS in 1h 14m 27s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/47f6f0ed1f8c460a8a7e2513547a0ba1 : SUCCESS in 1h 28m 39s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/611ed161f0ca4b67b8ac0428be1010c3 : SUCCESS in 59m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ef7b498b225e43eb923b7954bd9f73a4 : SUCCESS in 58m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d494b7ec82ec47a487d4934ae4e28bda : SUCCESS in 34m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cbce8916090f48a7858aef45c1b3ad8e : SUCCESS in 1h 05m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4e6346fa5c7e4eceac4f18100813895c : SUCCESS in 1h 18m 44s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba41682d999c42268421444e8e2e5425 : SUCCESS in 51m 55s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f13f9290862947aebb72440878e76686 : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":10},{"id":"c6a54e8ce26508f761004a10c515cb2032eb0614","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 13:27:38.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"74a5e22ce514f1967d7e2c6e46ddad4950d3c5e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 18:43:51.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a85ece1c46d44909b01b1f304d055d4b : SUCCESS in 56m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e1db0e3ab50f41ef8df66960bb332bbe : SUCCESS in 2h 10m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/10478fc245a64e3b9e6b6dcc3e9e483a : SUCCESS in 1h 50m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6d20fe9180b44e81859492e43b4b0902 : SUCCESS in 2h 19m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de8a00386ebb4038915007e8aec69548 : SUCCESS in 1h 32m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4da195d617a4cde9f90e92690837aac : SUCCESS in 1h 00m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72a9ab9004c46f3af49543e16b2d074 : SUCCESS in 14m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffa0c39c313643f0abd76920a8060934 : SUCCESS in 41m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07a1d6e218b04aa49ff73c674d3bb8d2 : SUCCESS in 40m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ffdf194c05c4ffea0233d76c9b0103a : SUCCESS in 12m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/067f72f04ec1452bbebc3c470069e93a : SUCCESS in 48m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/db58ea058a1449a19fb4b49930fdeafb : SUCCESS in 59m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3b4a4aeaff134435a4e7ed1df3833ef2 : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":10},{"id":"2c2b542e90bf59a18cdd9a3a74a4c7280bd7b9e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 18:43:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"ccc6efc8772eef5bbc1d89fb800ef5430d816462","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 18:47:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5535db5095a9439f8c882010e4005c94 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"3a9a17ad8216d039ca3dadce0e8fc160f3ec18ba","revisions":{"f09f7548480f0ce0d497cde9374f474bb3981b51":{"kind":"REWORK","_number":1,"created":"2021-08-18 16:12:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/1"}}},"commit":{"parents":[{"commit":"5dee9bbe75cfa79961681677718c509549a584ab","subject":"Merge \"Add missing options to generated neutron.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5dee9bbe75cfa79961681677718c509549a584ab"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 16:09:16.000000000","tz":120},"subject":"WIP: Add functional tests for ECMP routes","message":"WIP: Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f09f7548480f0ce0d497cde9374f474bb3981b51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f09f7548480f0ce0d497cde9374f474bb3981b51"}]},"branch":"refs/heads/master"},"61e852be92d8348d45c5acfacdf3d9b4b0cfc85b":{"kind":"REWORK","_number":2,"created":"2021-08-19 11:44:15.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/2"}}},"commit":{"parents":[{"commit":"5dee9bbe75cfa79961681677718c509549a584ab","subject":"Merge \"Add missing options to generated neutron.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5dee9bbe75cfa79961681677718c509549a584ab"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-19 11:42:48.000000000","tz":120},"subject":"WIP: Add functional tests for ECMP routes","message":"WIP: Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\nTODO: try to move common parts of the tests to some place to avoid code\nduplication.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61e852be92d8348d45c5acfacdf3d9b4b0cfc85b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61e852be92d8348d45c5acfacdf3d9b4b0cfc85b"}]},"branch":"refs/heads/master"},"a94abbeda97ad48f3c435c47687c9a4698a85a90":{"kind":"REWORK","_number":3,"created":"2021-08-19 11:46:17.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/3"}}},"commit":{"parents":[{"commit":"5dee9bbe75cfa79961681677718c509549a584ab","subject":"Merge \"Add missing options to generated neutron.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5dee9bbe75cfa79961681677718c509549a584ab"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-19 11:45:26.000000000","tz":120},"subject":"WIP: Add functional tests for ECMP routes","message":"WIP: Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\nTODO: try to move common parts of the tests to some place to avoid code\nduplication.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a94abbeda97ad48f3c435c47687c9a4698a85a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a94abbeda97ad48f3c435c47687c9a4698a85a90"}]},"branch":"refs/heads/master"},"e7b15d8398a85a00d287f3f1401408d210c17546":{"kind":"REWORK","_number":4,"created":"2021-08-23 12:41:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/4"}}},"commit":{"parents":[{"commit":"5dee9bbe75cfa79961681677718c509549a584ab","subject":"Merge \"Add missing options to generated neutron.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5dee9bbe75cfa79961681677718c509549a584ab"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-23 12:32:09.000000000","tz":120},"subject":"Add functional tests for ECMP routes","message":"Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7b15d8398a85a00d287f3f1401408d210c17546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7b15d8398a85a00d287f3f1401408d210c17546"}]},"branch":"refs/heads/master"},"f90726bd3afc055a506204acbeb18bc3808d92c3":{"kind":"REWORK","_number":5,"created":"2021-08-30 05:09:27.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/5"}}},"commit":{"parents":[{"commit":"5dee9bbe75cfa79961681677718c509549a584ab","subject":"Merge \"Add missing options to generated neutron.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5dee9bbe75cfa79961681677718c509549a584ab"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-30 05:08:25.000000000","tz":120},"subject":"Add functional tests for ECMP routes","message":"Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f90726bd3afc055a506204acbeb18bc3808d92c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f90726bd3afc055a506204acbeb18bc3808d92c3"}]},"branch":"refs/heads/master"},"abede21691d99071f7bc6c208e04b60fb9718b4f":{"kind":"REWORK","_number":6,"created":"2021-09-10 12:56:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/6"}}},"commit":{"parents":[{"commit":"c39896bd0775de032c471012deaff0edd82153c4","subject":"Merge \"[OVN] Change ControllerAgent type dinamically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c39896bd0775de032c471012deaff0edd82153c4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-10 12:41:18.000000000","tz":120},"subject":"Add functional tests for ECMP routes","message":"Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abede21691d99071f7bc6c208e04b60fb9718b4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abede21691d99071f7bc6c208e04b60fb9718b4f"}]},"branch":"refs/heads/master"},"ef230c63a9c1435d59789120fc91adc1de7c807e":{"kind":"REWORK","_number":7,"created":"2021-09-17 09:07:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/7"}}},"commit":{"parents":[{"commit":"c39896bd0775de032c471012deaff0edd82153c4","subject":"Merge \"[OVN] Change ControllerAgent type dinamically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c39896bd0775de032c471012deaff0edd82153c4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-09-17 09:06:20.000000000","tz":120},"subject":"Add functional tests for ECMP routes","message":"Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef230c63a9c1435d59789120fc91adc1de7c807e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef230c63a9c1435d59789120fc91adc1de7c807e"}]},"branch":"refs/heads/master"},"7e509cf411b63c483f0388326e5500add7c28bc9":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-10-11 15:49:40.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/8"}}},"commit":{"parents":[{"commit":"065e4b1705616198dee648892d730ba9652980a5","subject":"Merge \"Fix OVN driver validating Geneve max_header_size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/065e4b1705616198dee648892d730ba9652980a5"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-10-11 15:49:40.000000000","tz":0},"subject":"Add functional tests for ECMP routes","message":"Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e509cf411b63c483f0388326e5500add7c28bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e509cf411b63c483f0388326e5500add7c28bc9"}]},"branch":"refs/heads/master","description":"Rebase"},"a9cee5646ca3710c34a8e54fd29b4c227eb27f03":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-10-15 10:36:07.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/9"}}},"commit":{"parents":[{"commit":"2b30fa7bedf58e19e150a1f40f82d83d48e84049","subject":"Merge \"Bump os-ken to 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b30fa7bedf58e19e150a1f40f82d83d48e84049"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-10-15 10:36:07.000000000","tz":0},"subject":"Add functional tests for ECMP routes","message":"Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9cee5646ca3710c34a8e54fd29b4c227eb27f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9cee5646ca3710c34a8e54fd29b4c227eb27f03"}]},"branch":"refs/heads/master","description":"Rebase"},"3a9a17ad8216d039ca3dadce0e8fc160f3ec18ba":{"kind":"REWORK","_number":10,"created":"2021-11-08 12:48:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/805052/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/805052/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/805052/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/805052/10"}}},"commit":{"parents":[{"commit":"32c1762c363d2b55b76135a7d962f03dfe2bbea4","subject":"Create Keystone admin endpoint for Rally job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32c1762c363d2b55b76135a7d962f03dfe2bbea4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-08-18 13:19:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-11-08 12:47:51.000000000","tz":60},"subject":"Add functional tests for ECMP routes","message":"Add functional tests for ECMP routes\n\nAdding ECMP routes, where the destination is the same for multiple\nroutes available since [0].\nOne DVR test was merged with it, and this patch aims to cover other\nl3 setups: legacy and HA.\n\n[0]: https://review.opendev.org/c/openstack/neutron/+/743661\n\nChange-Id: I78b8ea85548e11074191f46ef560728c4dc89405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a9a17ad8216d039ca3dadce0e8fc160f3ec18ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a9a17ad8216d039ca3dadce0e8fc160f3ec18ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
