)]}'
{"id":"openstack%2Fneutron~805391","triplet_id":"openstack%2Fneutron~master~Ie283ad3a4ec433c88ac23f798908cd143159394b","project":"openstack/neutron","branch":"master","topic":"bp/qos-minimum-guaranteed-packet-rate","attention_set":{},"removed_from_attention_set":{"33652":{"account":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"last_update":"2021-11-14 02:05:50.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-02 21:18:44.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-11-04 15:06:53.000000000","reason":"Removed by Przemyslaw Szczerbik by clicking the attention icon"}},"hashtags":[],"change_id":"Ie283ad3a4ec433c88ac23f798908cd143159394b","subject":"Enable QoS minimum packet rate rule for OVS backend","status":"MERGED","created":"2021-08-20 14:18:48.000000000","updated":"2021-11-14 02:17:09.000000000","submitted":"2021-11-14 02:05:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"805391-bp/qos-minimum-guaranteed-packet-rate","meta_rev_id":"2fb324e52aebe5febd4c0d6ec6d14d5a25fb0794","_number":805391,"virtual_id_number":805391,"owner":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-10-22 22:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-14 02:05:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-11-14 02:17:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-02 21:18:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-02 21:18:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 14:44:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-20 16:41:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-20 17:03:47.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-21 07:51:28.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2021-08-25 14:51:34.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-25 14:51:34.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-25 14:51:34.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-27 23:34:55.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-31 02:24:44.000000000","updated_by":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"reviewer":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"state":"REVIEWER"}],"messages":[{"id":"85fd0abdfb8fb07d23e88cbc4b939793987a5231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-20 14:18:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cff13041d132cf4a6e107e38f852f1696abad51e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:44:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6a9de6777774964bf3215cfd46675d4 : FAILURE in 10m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67c6886cc50d4eb6a2de12d0991adf1d : FAILURE in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7421220e3d2452c78626296d6b7e475711145f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 16:41:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0acab612c008411897c888833375f3d3 : SUCCESS in 42m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c093367e05854c39b5509631b2323eec : FAILURE in 2h 03m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/22933981f2294526b7c707c10767088a : FAILURE in 21m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/795db7678e604266a85fc0dbeb37548a : FAILURE in 22m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c06f6133af71497c923ae56a95af4ba7 : SUCCESS in 2h 10m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d424131daa0e4768bb8bf35f9eb6295f : FAILURE in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c20b4aeb5c85418d9570776032011a21 : FAILURE in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdaaf774331c45f8a7f206efd7a0727e : SUCCESS in 13m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd3233575b38452f8f41473863ed8afe : FAILURE in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f295659cabf4369a2bf47344910f3ce : FAILURE in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2de791bf712f4ebabda0b8270e5cff2e : FAILURE in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db28245c04a4c6489cbbb0cdfa59704 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ee40cd1bfcd429c9ceb156aec5ec657 : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d5cca87b36f484e8c4f3062028f948b : POST_FAILURE in 29m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d115d4cef97345659b3525d340f018ac : FAILURE in 20m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/73498e1e3f094de6ad75887890b94e80 : FAILURE in 28m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4552cc92311142dd96d888bd01f32332 : FAILURE in 1h 00m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/23bf89af6f14473dabc7a9c23592e7ff : FAILURE in 1h 00m 42s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad456bd942c447efa5dc940df8afce43 : FAILURE in 34m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a00248885c1444d2a1587144a1b0d539 : FAILURE in 26m 32s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/aea574f439ec4c7496573213ae83f63c : FAILURE in 26m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1997b79deff43748255e7c7208688f3 : SUCCESS in 58m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/819e428276f6427098b336c34c9a788a : SUCCESS in 55m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/351a811f8e714e3caace8df1c8e714c3 : SUCCESS in 33m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4bea6119f8394b288140529df2e020bc : SUCCESS in 1h 06m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6fc270efef0e46d79f2b80c96ff83b09 : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7b569b3d81427400fab145a35fd4bbd8328c66e5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-20 17:03:47.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805391/1 : FAILURE in 2h 31m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dc48b825c615311c93ca711f8abf6ee73030ff95","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-21 07:51:28.000000000","message":"Patch Set 1:\n\ncool. I can create a port now with pps QoS policy","accounts_in_message":[],"_revision_number":1},{"id":"dd2aead7f7e6e806116e57e66342380a5ee9c92a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-21 08:04:01.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e cool. I can create a port now with pps QoS policy\n\nsuch port is now reported without resource request. But I guess that will be a separate patch...","accounts_in_message":[],"_revision_number":1},{"id":"4223149a5ff4a7fefef293ba856d29fe8259233b","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-23 06:56:06.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e cool. I can create a port now with pps QoS policy\n\u003e \n\u003e such port is now reported without resource request. But I guess that will be a separate patch...\n\nYes, exactly. Support for the new format of resource request is going to be introduced in port-resource-request-groups API extension patch.","accounts_in_message":[],"_revision_number":1},{"id":"ba94e2bb10164b86318c3488b917062d9835fe6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-23 13:06:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"27334b48d6018b70f7ed0ca424503710f2b95ab1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 13:42:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d41c3ff476be433baf5555c0e6d8b525 : FAILURE in 8m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80a6dc5ea28f400cb8d41d900d2a3baf : FAILURE in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a06fb08e9f3706592417ec394467f2de94c582c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 14:35:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0363193b2d3747adb80fd5bc31e6c537 : FAILURE in 16m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/04f60d5a138b4eae8caac852fe42de07 : FAILURE in 21m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3e860c3c751a4b0b8e50d1bd0332ed7d : FAILURE in 26m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7a23f50ef3f3403fb40609ede33e2f0b : FAILURE in 19m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/37c7bdfc80ed4576a79300408e0b396a : FAILURE in 20m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c7165b8b1db4272ad3c37ce7a1988fa : FAILURE in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cd8e001cce94da2a293baec812228f6 : FAILURE in 35m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd4540c325164c028367c0ed53419061 : FAILURE in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9ec5afba9064ae2bfa38389577a0240 : FAILURE in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3209f64a08444b6ba0b9767479b3778 : FAILURE in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7242436e21ac46ebb1edb2372daaa11b : FAILURE in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2243ccbe18342d588b1a296ac67caa3 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5c186cec73244a4b356847e0d51ddc7 : SUCCESS in 13m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3aaad449dd194c509c96092ee7457063 : FAILURE in 18m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4bb00a170d564a43b6bd9dfcd658c9e9 : FAILURE in 23m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c60071dfe05e448c89a0cb4a55d3b438 : FAILURE in 23m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36eb5f88f09248f58930573b6d23459a : FAILURE in 57m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4344eeb33f364497951bcfcb88adff3e : FAILURE in 1h 05m 05s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2acd68b2100b4848a3071aedaa161ca2 : FAILURE in 23m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e84fb9ec0b314658876295ba9d4e5643 : FAILURE in 24m 27s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c1b57fde9185403997744be7aae5f398 : FAILURE in 17m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba69a3dfc3614897ae920ab02fb7bcd1 : FAILURE in 18m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e8a0b6df929e49268c86bc5f0d65290c : FAILURE in 16m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/45525031979a4591a82b6c7f154d1ff9 : FAILURE in 15m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/be9672a3bd43467dad16290bc47a5367 : FAILURE in 22m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2f63fbc878564247a0b12d751f483913 : FAILURE in 22m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0877911f2763c7bc561ce334c173669c72c60901","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-23 14:39:49.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805391/2 : FAILURE in 1h 01m 36s","accounts_in_message":[],"_revision_number":2},{"id":"621421c18beae94f25388f81b68a77742e704723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-25 14:47:12.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c3565377a38708860520ac05e7991ffeb0370f99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 15:34:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51a7a42162bd4f31a42e2d6d0577b18c : SUCCESS in 39m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2e8ae7884f14bfaa96092dae82b7b88 : SUCCESS in 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aa05bef27409f43fa9c5fa15e1e43f732d4ff83e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:12:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/32cb356b172043f58a0f07ab969baf67 : SUCCESS in 45m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b00e3cc8df794a14ba7e71d0fe2dc8b5 : FAILURE in 2h 08m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/afe0b6aa7dda45aba36f47b7c3d8ec37 : POST_FAILURE in 1h 57m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c0e4a5ae519c418aadec4f79eb77fff6 : SUCCESS in 1h 42m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c900b8f080574a3f91ec8772876e69d5 : SUCCESS in 1h 59m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d59338a07b484346a17a714be0afaa40 : SUCCESS in 29m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7e0ee7808534c44811f8fb66fc75f25 : SUCCESS in 39m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5016c6e46b644909a0a14fd003e5446d : SUCCESS in 15m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5db9445405924537944e3d850d2f4dcc : SUCCESS in 29m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0370fbd96ad640349d9594ce57bb3f60 : SUCCESS in 24m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccf4256520864e0ca813ca3432013a99 : SUCCESS in 32m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0869b4682b994818b05847d405cb7657 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b9d3526ba6448df8bc5ec6b46cc2340 : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/58887131339c4a86a2d9611030c145c3 : SUCCESS in 51m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4598ac3270f4e0cb8cf875785dc4a72 : SUCCESS in 1h 47m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/71243e402b084fba9f9ca1f63f457d91 : SUCCESS in 1h 09m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7f2635a81214f21aac81d792a639488 : SUCCESS in 1h 00m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fbd18d8529e442f2a2817b7eb1c9298d : SUCCESS in 58m 50s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c8829844e4a643ed96c33e9dd1db32b0 : SUCCESS in 1h 05m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f0b64edbd5e846688c2cdd8134e60e05 : SUCCESS in 1h 24m 19s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/59cd6270570748cd9b9cdf320bc506c0 : SUCCESS in 1h 36m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e3a6c906ca6435bb7a6d0b5d4d52040 : SUCCESS in 58m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/afd30fe7e0aa465eb74fb9a4fb39788d : SUCCESS in 56m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d2fadde5bfe3445abd8f8b8c334e1e66 : SUCCESS in 29m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/705aed13234e455d8237970197435cf1 : SUCCESS in 1h 11m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/57e4cf4ebc894dc5bf382525c5b5dfdb : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":3},{"id":"6b5a069459d546f323761d239cea986e501275ba","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-25 17:30:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/3 : SUCCESS in 2h 14m 04s","accounts_in_message":[],"_revision_number":3},{"id":"3272db55b0b6d08dbdb87d46d2c596ec95a71c9c","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-27 11:02:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"27b8665da36b84ca50b49280155ebd52ccc88eb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 11:51:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/adaa542f313c424c85168491f3b458e1 : SUCCESS in 46m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04b3a41f5a29436dbe8a990ca5964027 : SUCCESS in 46m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e4c5146a1a535069285e31ffefb9212287722b5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-27 12:36:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/3 : SUCCESS in 1h 30m 39s","accounts_in_message":[],"_revision_number":3},{"id":"dbddf5516fae35f377a01cf2c87e0ac23b8b5755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 13:38:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/502926c2188f473bb7e525e2dbcadf49 : FAILURE in 13m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d87cd0327f154adda350ede88b9f8616 : SUCCESS in 2h 19m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/39378a94b5bf4db79a37e9a7711d37bd : SUCCESS in 1h 57m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e523fef43a8741bd8d8d40ee95ecd92b : SUCCESS in 1h 35m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/846914be39754ec4a3126e92ab0e5734 : SUCCESS in 2h 30m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5ec28eb75c44b9380ffcfcfdbe88593 : SUCCESS in 42m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d3fbc74e389489bac166c7c2e737cb6 : SUCCESS in 41m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5deb160c103247d88221cbf7b5396773 : SUCCESS in 15m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad828598b07b4919bd347db3b5e3cb3e : SUCCESS in 27m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f963b3493f9c4aac8dad009e338e8293 : SUCCESS in 28m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c4bea0751fe4b3d9cdfdcffa51bd266 : SUCCESS in 35m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f61de8bf9934b3998f29e6682193452 : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4a100d2da964ca2a62c1d8f52866220 : SUCCESS in 10m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a112ebf914024361abb4899033c77d33 : SUCCESS in 50m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb00853984c74e338cbd376cafea4ecc : SUCCESS in 1h 41m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e57daece4ea241e69800fe58e6f4d468 : SUCCESS in 1h 08m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/daf71033132d45b7a476509e6b1430c8 : SUCCESS in 1h 02m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/38b668a46e094ee9a7c357f74c835372 : SUCCESS in 1h 03m 06s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f8a23b42499d4df68f0c6398ec9a933e : SUCCESS in 1h 06m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8650b8b22d564e16bfbde0eac7ecf149 : SUCCESS in 1h 16m 36s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/d8ba41cf01884d76b210c1e2b5c4ed0c : SUCCESS in 1h 28m 43s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bbb4937652f47ee96ff21d7a5c3c991 : SUCCESS in 54m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fc04632f292b48f39210814490f541ee : SUCCESS in 54m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/95a9218719894ad797ac0c40b9dc1948 : SUCCESS in 37m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/414964c96e7a4ca696d4980064dd43b9 : SUCCESS in 1h 16m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/adfc648eab8f4146a5ac08b8f96c0438 : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":3},{"id":"7be75071179972def546644c3278cbee281be387","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-27 23:34:55.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/91/805391/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"25cc237c2c43bf506a400982c5d279f048b8eef7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-28 10:01:48.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/91/805391/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6c53c8b36fc357fd3f753c517a2960a7b9fa37ba","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-29 02:19:22.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/91/805391/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"c69d5e328f9f2caab68754f80cd102e987f5ae88","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-30 06:42:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8f122f2396edfc42ee194e1f19b350e06fa2e37f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 07:30:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/136a692457eb4945a227ca6f9ecd19fc : SUCCESS in 42m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d92aa81a03c4808a9791d3ac30d55be : SUCCESS in 40m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d2a0250f33da751bb819ad3e194c9cb4ad3b12a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-30 08:16:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/3 : SUCCESS in 1h 32m 27s","accounts_in_message":[],"_revision_number":3},{"id":"fcdb3e4c881a199cd6e6e58011c2ad40813667b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 09:14:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/60b394490aec4d408b36a5f14e28e5b2 : SUCCESS in 43m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/81370653235041848e7a946b061911b2 : SUCCESS in 1h 54m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c2a720d70046487a87ab7c81effeb2d4 : SUCCESS in 2h 06m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e1ecb4ab6aab4ad28891dc06c1b0a125 : SUCCESS in 1h 33m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e68ee920d898414cb0ca01285659de59 : SUCCESS in 1h 59m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8042cec09bf24f918e3b3cd8753a822f : SUCCESS in 28m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efbf723c86c0415783ed0532eb92472f : SUCCESS in 38m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94486122be994adda974baf4445a5a5d : SUCCESS in 14m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d17abc075152476c9ea5b48a6523e254 : SUCCESS in 26m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa47fc52f8344607b4a961d5ad4e6f63 : SUCCESS in 25m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b99befd37a9944d49e392900caec401f : SUCCESS in 25m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93e44c8378d2443b9bcc11395dd26a06 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b19b7e47d22842c3a4ba8c94f40e70cd : SUCCESS in 12m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcf97358d7c74609a421322b6e0bc4bc : SUCCESS in 48m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/81d057f04c6a453f89c2b3cf03015a63 : SUCCESS in 1h 49m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/419a7699e41d4e49a4c2d97c61416ba2 : SUCCESS in 1h 12m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84106a3db0c7433b8c058d106e8dc9a2 : SUCCESS in 1h 04m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/956b73b64eaf42f4ba39d983deee2a16 : POST_FAILURE in 1h 06m 16s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1860e66107694b5eb3da741539b06dda : SUCCESS in 1h 07m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bf507ff75178411ab8a0da8406a4ff7d : SUCCESS in 1h 26m 49s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/2f26ebeab10341859a2008eae63cab79 : SUCCESS in 1h 34m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fead6959eca48fbb2ef0183d9177759 : SUCCESS in 55m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0c78b6e016cb4f2f8c2d346908132f7a : SUCCESS in 57m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dfd7982140f44beaac41633c86fb30f0 : SUCCESS in 36m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/762ced86bdb94c1388f47d3d2ca60737 : SUCCESS in 1h 12m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/961f82702485445ca4b523c31a0fabaa : SUCCESS in 1h 19m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0049629d5b67e15bc94fe2a8c8357a91342d5b27","author":{"_account_id":33145,"name":"Kevin Li","email":"liwenhai3@huawei.com","username":"liwenhai"},"date":"2021-08-31 02:24:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4d75b7c0adec2722f8cbbe4e87f3c480f2dcbd80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-08-31 09:59:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"82a177c6048f8b2500bb904d6266900808900ea1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 10:55:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec9dc7bd358f4867a728c09f4f7f9280 : SUCCESS in 44m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8788ae83be3d42fd9344e4a6b77440f5 : SUCCESS in 48m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99ade281fc1ec54c64786201135616d33d8d15db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 12:44:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1adc0fc9513a4f928d6c2f29309db581 : SUCCESS in 38m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9c13a1658e3f4892b40afcee5597db19 : SUCCESS in 2h 32m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4f71c7ac733745d59c48d432ff7d2044 : SUCCESS in 2h 28m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b3a45ae8db89457086e0525e3f24a733 : SUCCESS in 1h 28m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/79a1d9fa52d64ce09c5e2a33f9fb0519 : SUCCESS in 1h 56m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f20bd7e927364406bf9d5c4f84b81ed8 : SUCCESS in 36m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ab60412a85945d08c6ae8eb4f3158a1 : SUCCESS in 40m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4aace465f441ca978c93064f6d6592 : SUCCESS in 19m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c698d20083244418b473d84f74202eb : SUCCESS in 33m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89ffda80148c46dc844fb554211e3ad5 : SUCCESS in 21m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f835b7faf5094d89ae1c33bdd746b068 : SUCCESS in 22m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff89b63b8edf4438a7c5b18d5270f471 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba2e6c9fe2d444b287e324f1e9c43566 : SUCCESS in 12m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/83c9a887593041299e1fa2d45a1032c0 : SUCCESS in 56m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/54e84da04b3b453f902fb904ee01c6bf : SUCCESS in 2h 03m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/16e2459db7bc44adb78191e55ee38cb9 : SUCCESS in 1h 23m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36810e1d850e4225993a8a4117bbca27 : SUCCESS in 1h 09m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/729f3c3d7f0c4a18bc37279c3ef66a42 : SUCCESS in 1h 03m 18s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4dc1553d9c57401f98ae90af74895611 : SUCCESS in 1h 15m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e4e2b919078b4c69b6d2d23b67b4b427 : SUCCESS in 1h 13m 32s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/4c09ff0436aa453caf0755b96639d03b : SUCCESS in 1h 28m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09ffd6e85848499b90150226c638647a : SUCCESS in 54m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/767b1558a03144a4baafeee0ade77441 : SUCCESS in 1h 02m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/afb9ad9fd51c4391978dc58b44d60023 : SUCCESS in 41m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/733e7437c0bc47bfb01267f54df5788b : SUCCESS in 1h 14m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a53216697f0a4b96a6a2b03c9fdddb6d : SUCCESS in 1h 15m 28s","accounts_in_message":[],"_revision_number":4},{"id":"38971c9392a80c8b331cb88f7fad43cc503b60fb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-31 12:55:36.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/4 : SUCCESS in 2h 05m 45s","accounts_in_message":[],"_revision_number":4},{"id":"9d32c6d91e3bed1db0d7f9950d181df3272d7083","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-31 18:31:53.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/91/805391/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"b63edd22bbdbc2ed17ccc0422e6efeaa82946163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-16 14:01:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"cacc4fa8d25c391b7278fa50c8202016e705a970","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 14:59:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe0a920ea29a408c84688f02114f997d : SUCCESS in 43m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02be2f796dfa4fdc85c43c0e8a47c0a8 : SUCCESS in 43m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6157a3494d57f5e28e4c8cb857001ba9c1606075","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-16 15:49:07.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/5 : SUCCESS in 1h 35m 29s","accounts_in_message":[],"_revision_number":5},{"id":"8c4434b155c62e9dfa79febff2512e0381eb5d4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 15:59:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/988bebd6bfea4f6f8aeb2e9d821c4854 : SUCCESS in 52m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/be9c68db7646485c96f253d84c8e4c38 : SUCCESS in 1h 46m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/36e2263e55d9462a912e5b0432500681 : SUCCESS in 1h 47m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4aaf52a660be42bba3124fc46af3e6a3 : SUCCESS in 1h 38m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1a6cdaa19e8940d1a661824b562c3896 : SUCCESS in 1h 45m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c910c9b2f0d041528a3b8d163024aea8 : SUCCESS in 35m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0384114d1b61406392d07e8f1fc52462 : FAILURE in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9900f5b149e4bb2890d5504d7430a3d : SUCCESS in 17m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a76dfd0213a24ab5baaaac5265931e05 : SUCCESS in 27m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3114ea90fca4550af0a45e7ec5bdf20 : SUCCESS in 22m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ca1ddb1df8f426798bb929e635e6a13 : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876415c0f3f346d689dcb1083f73c505 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f25066dda0a4e59af91cb954c0c0255 : SUCCESS in 12m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/077197baec9641768aa58541a26234e5 : SUCCESS in 49m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ceeac872c0b74ee78821944da0527812 : SUCCESS in 1h 40m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dc15192e855a42e9b9af35c6a42c61de : SUCCESS in 1h 08m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c28bca8f5e70434daa269c46e739b841 : SUCCESS in 1h 01m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ece21d01c6841a68219ba85521be12b : SUCCESS in 1h 00m 54s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e9220616d0cd4d64afe6108dcb3cf4cf : SUCCESS in 1h 07m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a5732514c9e24578ab0072be4fb1569b : SUCCESS in 1h 13m 24s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/4a15540ea579416691e9a6e3879a6e72 : SUCCESS in 1h 30m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fe297b5a7254f87af081e0ab229cc6a : SUCCESS in 54m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0f82fc2a76944a3c8ba5c75182fa245e : SUCCESS in 1h 01m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fe53e14d860941d0ac29ca0a428d67d1 : SUCCESS in 32m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/49c38e03ce104580a931e8d296ae53c2 : SUCCESS in 1h 11m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/eae30acf60554609be7012b1b4158236 : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":5},{"id":"ef46a5d1a19512ab9817909f5666fe645a43ff6d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-16 20:40:59.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/91/805391/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"26c3d9fb024b8b168faf551dfa04a4a1825daaa6","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-21 06:45:19.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5345cd21e4ac91acec3c19a98af26ec832ff2d54","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-21 09:59:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/5 : SUCCESS in 1h 40m 58s","accounts_in_message":[],"_revision_number":5},{"id":"08bed3ba617a26116fc03041753048fc7a8579d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 10:15:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aaa885fadc97478c8209e3fe7897ac3b : SUCCESS in 37m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e2f47df3a2745c8ac9c961280e7f105 : SUCCESS in 46m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e19009cbb42e59ae5af675bf7c06bd1ae2bd5619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 11:19:33.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0a6f0bfd85944eb4ae053e55f5207358 : SUCCESS in 43m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b9715bf43e334432bc62b97840b0d022 : SUCCESS in 2h 09m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9ea2ef16ad3d445388fd43158b31b18c : FAILURE in 1h 36m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d91d038c43d4d07881741775ca04077 : SUCCESS in 1h 23m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b53ca5715a8b42dfa459cd8e1a5c7637 : SUCCESS in 1h 55m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a6e864dec9c4dbfa14ae60f8964df40 : SUCCESS in 33m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b040c89b0e8d44c9a85da135ae6f98df : FAILURE in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8845db26c4594ab78fc5d8b4242c2d3c : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33cf3d39afd04a66b9c79e163b565462 : SUCCESS in 32m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffd560a0063749fa949a3c602c0efb24 : SUCCESS in 27m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d43c563db8974cedac250d22a552bb26 : SUCCESS in 27m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212ce67940d0464e87379b0044b9910a : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/202d38a30a614cf4a30a6b2ce86b39b5 : SUCCESS in 14m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/55b92335d07346e6b6be155d1750fdd6 : SUCCESS in 51m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82eaf134393c4ca38f68c245441d5e8c : SUCCESS in 1h 41m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/83784972229042c09308abbe483f38b2 : SUCCESS in 1h 10m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2e477258f434f3883b2d388c5933ff1 : SUCCESS in 1h 02m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6e29885d9bff4258800c800543e8c06b : SUCCESS in 1h 07m 35s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ccbea1d4d48b47528d812e421540e738 : SUCCESS in 1h 08m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5466fcc5f29c4ee1b6a0eab560c882f8 : SUCCESS in 1h 15m 38s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/95a96d454bbd4023acf7cc69e81bb1a2 : SUCCESS in 1h 36m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ae291593e3e4cb4b02992f31fd1f9ac : SUCCESS in 58m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1ab596f67e4841c1b45bfa4280cad8ce : SUCCESS in 54m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5b13b18d94d845d3a1305d499c91560b : SUCCESS in 34m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/90adc40c7cab4a26be59cc9a6d287e1f : SUCCESS in 1h 14m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f65d5a4583e1474689e4729645dedab1 : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":5},{"id":"5063808bb62b217e20a0368e343d0321a4ffbd48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-22 06:57:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"b13fc35e09a0902a68aa45c0d3a16843449ae047","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-22 08:54:03.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/6 : SUCCESS in 1h 53m 30s","accounts_in_message":[],"_revision_number":6},{"id":"2be5c00a95221b581c69c90b33d91cb06ff0654e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 08:59:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a6c7223b528436b94bb2a2df32b8e7c : SUCCESS in 39m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5f69d0492cc41b09a4ed8765d2ef7de : SUCCESS in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b9e316b5e9e638674bd60d7c0d5d76fe321db46e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 11:17:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d7ecf8d9832b467181353f5f3aa5aabf : SUCCESS in 57m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0ed2afc5e8d44613955c33f7ad9b548f : FAILURE in 2h 06m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0fc65b2df1b94d15ac94c7b381affd76 : FAILURE in 1h 42m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cbe07e0fd2af44cea25b1cb630c86228 : SUCCESS in 1h 19m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7b6c5fe0d9d84dcbb4cf2b6504410210 : SUCCESS in 2h 01m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e02f85c0d23c4673a17db97e091a1f4e : SUCCESS in 31m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a23f04775dc24917bc616a4033c4de38 : SUCCESS in 35m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3bdae82be141b0bec461b6c54f5c59 : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35df94046fd84b7b878b9b735d4fb2fe : SUCCESS in 27m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2f8317ca3114f59b3dcea33c1d6a215 : SUCCESS in 25m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5432148c160943eeb143f89a4e5d884c : SUCCESS in 24m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b1e43c54cc4d7a8c17b9682b2962f8 : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92b3226232ad4480ae7c3af764fd974c : SUCCESS in 10m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/981e9d33275f4c6fbd93590c6c936633 : SUCCESS in 50m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dccf773079874da3b7191d20cc2e0da5 : SUCCESS in 1h 37m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/833f2b6e40434104a6f988d0ed9d7b81 : SUCCESS in 1h 08m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee8ed482c98e40bd84d692eea7178ad9 : SUCCESS in 1h 08m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/90a8a140c74c4ae58c49341d551492ed : SUCCESS in 1h 05m 34s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/64c3eb7569154cfb9aef1bd413fd963c : SUCCESS in 1h 07m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/72fee9c14a6b4cae83dbbd0e75e5beb8 : SUCCESS in 1h 17m 23s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/2192e621574b4bcf8f6e44dd18f25b9d : SUCCESS in 1h 23m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70c1c0209d35401e966bb868e0895ab8 : SUCCESS in 59m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c3f687957e1a4032b952104fb5471392 : SUCCESS in 1h 06m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4da5dc2833aa4b7cb1f13cd3bb1e0269 : SUCCESS in 39m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7bcd4a806efd40f68c8a47ee23eacf44 : SUCCESS in 1h 18m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e2848e3c177c49db8ab4e3b9cc219e78 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":6},{"id":"5211d75e1ef61bfc15e5a80040d5af09011388eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-22 11:41:38.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/91/805391/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"572194d989cdaf18fbd29bee6c4c8d78993b4c74","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-23 06:57:06.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a1d2ff78ece744f8eb8701b7f5fa81ae1c4930d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 07:49:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f6f6cf3cec14755bb0d48eed33f0df2 : SUCCESS in 50m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/944e0968d2924d4193e18274644b2b91 : SUCCESS in 48m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"708db376120801f1c496346a129857b3b6862945","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-23 09:14:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/6 : SUCCESS in 2h 15m 49s","accounts_in_message":[],"_revision_number":6},{"id":"76f5e0b3512160cde640ac9906fcdb4cc9d74d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 09:25:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/026803c7923e4e568ddec927c5ac46c0 : SUCCESS in 45m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/17fdee351d5c45f7afc5dd2fcbe3660d : SUCCESS in 1h 54m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/016100bfe3614f3798721b6f1ad59c06 : SUCCESS in 1h 36m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e5d49604f38048c0a43947b68597e595 : SUCCESS in 1h 40m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/55a4292bf86c468fa2812a339bcd1fb8 : SUCCESS in 2h 05m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24f79d0d471549cd8920f7077ccf517e : SUCCESS in 39m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/779c46f9b1884ef8a6ec152ddf9dff07 : SUCCESS in 40m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f941f2ba7c8b467494708b5e6ac2aebf : SUCCESS in 14m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cea98a061d84d23894e2cbdbf90b602 : SUCCESS in 23m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b19d10dcbe7e49369233824e6bcc29ca : SUCCESS in 29m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/928123ec2dfe429dacd750e3cfbc560f : SUCCESS in 28m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3aa2a7c68494b45adccb695ac6fc9f2 : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53c314244a9940b7bd98ab10bf955a50 : SUCCESS in 12m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4489e8930def4e959ea80bc6ff36d620 : SUCCESS in 50m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3728a2d702214c12bcf02b7e230f1ab0 : SUCCESS in 1h 40m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/973fe57b7b324e1c8d92231b8eeaa1e2 : FAILURE in 26m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/89fe0a2d75424a849b9e882115707bd1 : SUCCESS in 1h 08m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/643d73ecb0944b52a564151f3769ca23 : SUCCESS in 1h 05m 52s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/636012663c264586a6aab4ccaeb06b5a : SUCCESS in 1h 08m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/18e5afafa0c245fcb2f9870f5b929414 : SUCCESS in 1h 18m 58s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/62231087149e47f1a55dbdce24b9e881 : SUCCESS in 1h 28m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88c441123cb44a45a8f533ab095a9921 : SUCCESS in 48m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/616225dfe64247a794ad27eacf2416df : SUCCESS in 58m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/97eebbd201094f7f81d7145be00c1752 : SUCCESS in 34m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ea39ad6b61b64cf6ac8ef39884ea4c74 : SUCCESS in 1h 16m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4b5f2cd0b9de4f0aab5aaefec2508f40 : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":6},{"id":"096df64f9cf44a421f02855b9c7bf134466de1cc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-23 13:29:04.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nminor nits in the docs","accounts_in_message":[],"_revision_number":6},{"id":"66bf5c160d3bb1b074a12f55fed68817da6841b5","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-23 14:54:23.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8b982b8a5a3dd49757b62a40429cc443a80b22df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-28 12:43:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f88385e7ec4ce51af1ea6ec0e3442e0a28faec8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 14:37:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1f49eb5d5558468f925576eafb47b679 : FAILURE in 27m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/be497a8a12084751afa470e28ab27878 : FAILURE in 26m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2f2e6c640d1a4341938a340333d8c1ec : FAILURE in 27m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/41f654c93b6b4c43bf1384ac1eebb0e8 : FAILURE in 27m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ccae8a4ed6ba4edeba7ff15524b182cc : FAILURE in 30m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed86cb1a653243d987154e710df79e6e : SUCCESS in 32m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c09c9ed8c8e140239286dd35e73ed5e8 : SUCCESS in 45m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c1c105ed20f44c49758feda255caf19 : SUCCESS in 15m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c99549d08c9e4844a4967e8977ed17f5 : SUCCESS in 27m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6362b12703754e18a1efe26bf5efcee8 : SUCCESS in 25m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c90e21583767412790f2a24831bbe0df : SUCCESS in 28m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb407e48ae344dcbf103d8328269027 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d5f68cbbb0c43058f72fcd7d6401ef8 : SUCCESS in 16m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f14b169a42c420ea12a434549b582bb : SUCCESS in 51m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9673e4846e894c229b8c766c55900f49 : SUCCESS in 1h 48m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9ed0f88432954100b263c95272a98fa3 : FAILURE in 22m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ff7f083f0784a05ab78cd59efb325e3 : FAILURE in 32m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/72286ca7377445ed8b3b84f468f18909 : FAILURE in 32m 10s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c4467a005bcc4aa7ba61bb73e0a60e59 : FAILURE in 24m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/23c1497eaa0e4908a8be83076ad5f51b : FAILURE in 26m 14s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/8a4679754ec6408687d33bc5c44ae5bc : FAILURE in 26m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4cb1a223afa430a9f5e7425e7517d01 : FAILURE in 22m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2d2625e80b8041a486394a0773a9cd5f : FAILURE in 30m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/aab84aed3e8f4092a438dd1fb41062c5 : FAILURE in 21m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9575ff7611104a1f83811a26fdb944d7 : FAILURE in 21m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/95434911ae704b84bcc24c0b954c85a2 : FAILURE in 31m 21s","accounts_in_message":[],"_revision_number":7},{"id":"8eecfd751a5d0997eae885ca63c7e2b34b9f76e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 14:38:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c5affc1f2184406b108f9e5139829ab : SUCCESS in 34m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7de9c6b32eab42c1b5c300e7c82a7cd3 : SUCCESS in 38m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7377d80fad7d09fec97bb315a336439d601478ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-28 16:58:49.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/91/805391/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"12eaadbaf30106a8bd60a4630ee4372b58fa79d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-29 07:43:12.000000000","message":"Patch Set 7: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7e19982f245dd465a7a7bec5819852d23e304517","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 08:49:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/162392122940408689a44deae478381b : SUCCESS in 50m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d41e90725b8843158d8f0ff61bd240e9 : SUCCESS in 47m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a1364ab57a7dabdddeb3222a13f94947c918b4eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-29 09:02:30.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"4a842a82615edf174f6266301ba48ddc6c50de9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 09:46:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/839497035abf4e2783c1b2aac47e4969 : FAILURE in 36m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c39a4f194de3440b94e8f99ab138ba8c : FAILURE in 40m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bba20edd0dd45136b8e037e611a9eaab9d7737e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-29 09:54:51.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"4cf0cdd993ad42ec13ce78cc38186d91971a2c1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-09-29 10:30:56.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent, and commit message as Patch Set 9.","accounts_in_message":[],"_revision_number":10},{"id":"206b293f1fa411d9f13df4ad72377832b2e1ca02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 11:37:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ceedbf45c2749fca2edd530f483d011 : SUCCESS in 1h 02m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae72bdcd3b494c37bdc6be4a7964c390 : SUCCESS in 1h 02m 38s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"da043f334e405914d33058b56c117be36e89bf2e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-29 11:39:47.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/91/805391/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"f8bc4394363bfc687292bc7b5b2933339d7c61de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 13:30:43.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4bf57df039bd411986b7d1b636655d52 : SUCCESS in 43m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/792f962a74b44aa291c47af94a4f5e6f : SUCCESS in 1h 48m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0461cc075931423f96758d70fb95aab8 : SUCCESS in 1h 59m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/10a70126c34b4e80948d905997f93d37 : SUCCESS in 1h 36m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9c7eb7411e1c4154907e9a5632023a9f : TIMED_OUT in 2h 51m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/404216c66dae443da0a41ed05c1a8802 : SUCCESS in 41m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db41cc9ec00e477985e672467b23bd57 : SUCCESS in 39m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe191ac043449619892f2b72e8cdf6e : SUCCESS in 15m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af52372197bf440287c14c7ec467a410 : SUCCESS in 26m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e08732b99c843ffaae3edbc266205fd : SUCCESS in 25m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5860cb87e05436ea2706240c8670132 : SUCCESS in 41m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3146f8a1d44fe4a1465bb078c9339d : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c5d4c5e01fa45fda9bfcaa193567fbc : SUCCESS in 13m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce4211b22c18459da74346f5439b5017 : SUCCESS in 53m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e030862f37446d9b943bce013061d23 : FAILURE in 1h 27m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/80290a3e460148b0acdc07e861df252d : SUCCESS in 1h 08m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/66ea0fa1c78d4159a492a5a061dcbdd5 : SUCCESS in 54m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7a6fe2e971a649a0b9e9cd75bd880355 : SUCCESS in 59m 26s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0c229533941248c68b76638e4325fbc1 : SUCCESS in 1h 07m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/95c9aa4cc49b40c98510ab42f465fef6 : SUCCESS in 1h 16m 04s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/cd391db8e0224b0ba178d91a7f84e64d : FAILURE in 1h 24m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20eae4ffca954d2e9d3defb7e6c66e04 : SUCCESS in 53m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e771c34e7181492d81346fb9611a3292 : SUCCESS in 53m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ec80acd2ea764750aded83c9ad251075 : SUCCESS in 28m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a5b78447d3b2400ca999edc63389f34e : SUCCESS in 1h 08m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/258a0737b4c24ba8a813001a311a69bb : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":10},{"id":"82f49f2cc80da7e13e70f8678410df0b5d01e912","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-29 14:18:25.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/91/805391/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"8a147bb2bac1915b751c585237bae01730835226","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-29 21:44:10.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/91/805391/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"b02be397b9a99af28a2782d8b142459cba802a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-11 13:32:53.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"1c6b2b24bb59a5177d2f96773668e98a627e0c7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:00:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fab6f86f26074d938b6b094c9df3ebc4 : SUCCESS in 45m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f5190b847274408b541bf26619b15f8 : SUCCESS in 48m 40s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e0821f7270ec54b55bb6150181c21d37ea85efed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:12:56.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/952eb8430c45468596b646286da6dd05 : SUCCESS in 44m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c315a6c462724012b057981b7eb5a8c2 : SUCCESS in 2h 27m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/89220028b4e2449eab97a8252f0a5324 : SUCCESS in 2h 15m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f645408b7712494e96cc9f8ab9f7e540 : SUCCESS in 1h 43m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/85c85893bd6440319d8b31f28b5eb8cf : SUCCESS in 1h 46m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/686cea5293924138bacd7933163eca88 : SUCCESS in 45m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4611afa065d24242a038618812a88d63 : SUCCESS in 47m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aff335b67884483afef755d36beda5c : SUCCESS in 14m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a061937f67d4bf386ea60d45d16e7c1 : SUCCESS in 28m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c55973f8667d4a8f8d1f8f883a1f8e71 : SUCCESS in 25m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bf6d1957ddd44cdb45c0b18a06571bf : SUCCESS in 34m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7692b42a2808455bbe3880e2944030ed : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44cea34816dd4be8b84b0569feb3bfbc : SUCCESS in 12m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/559491ca9a7f49c8aa9dcf7ff31752c4 : SUCCESS in 50m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c41a5278cee44a39b8999309123f05e1 : SUCCESS in 1h 38m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/24a6e34accda4c05bb0abf606cd6ba38 : SUCCESS in 1h 14m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/68babcce898e4b9bb10e664f35f8dcc0 : SUCCESS in 59m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9e49568ca9774dad8b5cbc459fff3101 : SUCCESS in 50m 11s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c15bbd7ff9c24492a866bdf8f23a29d9 : SUCCESS in 56m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fce2a41849a24504b54e53ccb3850bab : FAILURE in 1h 22m 05s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/9b6251f4dd2e45d2aa28204b64b4d24a : SUCCESS in 1h 13m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4d13e33e56846348daf3d5b592cc049 : SUCCESS in 48m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6f6563537f344e39ae9c777335113c3e : SUCCESS in 58m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8eca272903054483b6b175520d81b048 : SUCCESS in 24m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7b0ec867e72746c2aa873ea8f893953e : SUCCESS in 1h 10m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2bee1ef395f14d10b6fa83fc8e4856d4 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":11},{"id":"997b55a9e501bf969bb97b5a5dd06614f616d586","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-11 17:53:42.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/91/805391/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"93471bbe07d977514949c62bafb3a8c7738dd2d5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-12 08:00:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"51318f09ef0542a22e3d91a27cf0c2563e2c97d3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-12 09:35:38.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/11 : SUCCESS in 1h 33m 47s","accounts_in_message":[],"_revision_number":11},{"id":"c1530370bf100828b55acfcf09e3f5c3126ba8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 10:20:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0799b160a085414db76ad2f5dd3525db : SUCCESS in 45m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/242ab5a90e634ad3b6d3d6c186dce5fc : SUCCESS in 2h 14m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ea0636c257174feaa057da7929ce8b61 : SUCCESS in 1h 37m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/59fc4826b4804a67af56fcd0e4ae4036 : SUCCESS in 1h 30m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/26b3a1a42e234e4d81566b8de0707ed4 : SUCCESS in 2h 02m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c031eadbb1b433ba43adc6b0a4f0fc7 : SUCCESS in 44m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46d975b816fd4bad8e3d5096b8e7243f : SUCCESS in 42m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c92838a2c85f437cb7b396578e48c663 : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/728ef8373f1c489791765275283a6ff9 : SUCCESS in 27m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c550081d5bfa4306afb75a405cae8b38 : SUCCESS in 29m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db71be68f3f24e779cef5c0c0af25bbb : SUCCESS in 25m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0065c95140d4e1bbc0d308ab442f3a3 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d1409903fec41acbe5ca3dd50c7b08b : SUCCESS in 13m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42e9b1aad59940c7ae11ff91289e9033 : SUCCESS in 52m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/839250cf31cc4ceb988bda74cc43de41 : SUCCESS in 1h 42m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/56722733acc847c397a0221f5fc19426 : SUCCESS in 1h 03m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef79a60bfe8a4f6dbdba6ce2d09914fb : SUCCESS in 44m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/671e445614334f2a91050a316d338d13 : SUCCESS in 1h 05m 25s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/39147874b48b46fdb27928c5c0454461 : SUCCESS in 1h 09m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3c2216ea662e46189c28e5ac30e2b831 : SUCCESS in 1h 14m 31s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/f21e08ee68734926b587d7a2200b6a67 : SUCCESS in 1h 36m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55d1355579444f22bfe092d5d3d1dc35 : SUCCESS in 1h 02m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/493493ffab4149088d5117f85f938bf6 : SUCCESS in 55m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dc22365739f64cc0aaf7d640ff17ed84 : SUCCESS in 26m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2e558ca3007141199de7697159d5ec6e : SUCCESS in 1h 09m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a88619be86be48888db7b6706df1fb9e : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":11},{"id":"0fad6e175d7d6b46087aca787b201146650acb40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 10:57:05.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4c42899797a449ca10c77fced35d403 : SUCCESS in 49m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77e9a5fe98b9452ebf0273a6d37a75ba : SUCCESS in 46m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"55033fcc2f4e86fe55e097b8abe17c4717f91f96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-12 13:49:09.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"dbbbe5993b2fbda5d8becae560c7f1b3dd5732ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 14:36:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70ff8ac30c184052b2ac944ee49e05ae : SUCCESS in 43m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b8ae0a2fb7241afb3bb67be1a10ebe7 : SUCCESS in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"395927daebff548ef29d36047914ef251c9884ab","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-12 15:28:57.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/12 : SUCCESS in 1h 37m 51s","accounts_in_message":[],"_revision_number":12},{"id":"be31a3422e50e029d7e47007dfc876275b5e3493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 15:47:29.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/42bd8bdbe2bc4cd1a429e8d3480a994c : SUCCESS in 1h 08m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/78728ccf69f047d9a5fd19d9cf999d3b : SUCCESS in 1h 54m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2eac2d2b5b074a839bc6a2c5abe5ffd5 : SUCCESS in 1h 38m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6fe5edd7a134171a17689b420a268e6 : FAILURE in 14m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e1673f0851b445ceb763030b3085e672 : SUCCESS in 1h 31m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e211b0a414c4205a259db4e1d5f5337 : SUCCESS in 38m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d316b5c383445ddad1c871548867afc : SUCCESS in 38m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05087de2790b4d79b5f2b3ff94affb9c : SUCCESS in 14m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2ef387f9c434297a0174d63d5bc657e : SUCCESS in 27m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09af9c194dee4d1e98a69d80478a322a : SUCCESS in 21m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8582c72a7f744b5d8985a25bd79d4dc0 : SUCCESS in 21m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3718e018565440fe94d2bf5573686237 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c58205808ebc49b9a49f9fe444920e9d : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dbf7191eed5d4314886ac5dda4de359b : SUCCESS in 40m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef7568b8e1b948f5b4da602cc053bb45 : SUCCESS in 1h 25m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/62c8d735aadc4cf38b17bedf31cce8b5 : SUCCESS in 1h 07m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4a773bb70fd4a9991054f9707682428 : SUCCESS in 57m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0fb26885ee4e40d588ea7d78a6f29924 : SUCCESS in 56m 32s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5978fc9022bb4baf919cd9db31b325b5 : SUCCESS in 1h 01m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/060a9a52db7d43edb6c220f4c7ee1d58 : SUCCESS in 1h 13m 19s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/5001a36285484ebd8ce4b4ca43a6584f : SUCCESS in 1h 26m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30aa40770ba6474daeee1cc5a482528b : SUCCESS in 47m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dc285dd1c39b4fe9814f421919f250ec : SUCCESS in 58m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b80ec4c97b24458aa7e7b497c58e18b7 : SUCCESS in 33m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d79564fbd7cd45ba85f26786c51ae4ad : SUCCESS in 1h 35m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9fbdb691b47d4f16888549cdfb2b740e : SUCCESS in 1h 12m 33s","accounts_in_message":[],"_revision_number":12},{"id":"4d7a6516a0fb452e95f62e34ab77fe5c0406d7d4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-12 18:31:47.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/91/805391/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"9254f9ee7d17b0a4f0a35834b87d1c30ba412bb0","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-13 06:13:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c551a3871c182279021afc7a661944d68495b3c0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-13 08:13:14.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/12 : SUCCESS in 1h 58m 12s","accounts_in_message":[],"_revision_number":12},{"id":"dad6ad1cf898a148bf73306d9d2dd694dfb37d74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 08:55:58.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4760ecded5784acea0a17012e9845744 : SUCCESS in 42m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9949419c4664a9993168afad7d1f367 : SUCCESS in 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5c53716efa691ec378c59d1f24e79f2682d222a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 09:53:09.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a76ebdc6f65e45fd8f8f2fd999f3db07 : SUCCESS in 44m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/32144b87f7a343bfba2c7d442e66746d : SUCCESS in 1h 49m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/701ca6a9064b43b2969ae8fee657e0ae : POST_FAILURE in 1h 53m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26f07cc843384525a57da6fecf4a7679 : SUCCESS in 1h 12m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b358229ba88340d2a45574882a8e38e2 : SUCCESS in 2h 08m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a4b86a0cafa4942865dd81a8b4287a1 : SUCCESS in 32m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9079040a658949f8bfbd55ea4771f4d4 : SUCCESS in 35m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81556fcfe4cc4bcba9e64d7a93b7cca5 : SUCCESS in 16m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0251985df7a4d92b81dd434ca471768 : SUCCESS in 25m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a60c632d65e54d2da567662bfb13ebb0 : SUCCESS in 24m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef99f419ac8844acab80ff322c52dc64 : SUCCESS in 25m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33211c569ed74325ba7696b38c4f1151 : SUCCESS in 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4e6d8af97ec4a6f9bbd3db796339c1d : SUCCESS in 13m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/13b1558a698d4fbaa46e52bd7c4d3a5b : SUCCESS in 52m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0353aafe1f44d9aa261acafa002a597 : SUCCESS in 1h 44m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/302489cf15f04876960e552d550796b3 : SUCCESS in 1h 25m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c0d84b90a5a4d24990e48ed84d3412a : SUCCESS in 58m 42s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f87e7be749cf4cf18ee1d68e4ca6821b : SUCCESS in 56m 49s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/917ab0690bb34df197bbf51e508e34cf : FAILURE in 1h 25m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3e217a9fe4cb428cafc27a545ff9aa10 : SUCCESS in 1h 19m 07s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/3530df7c8e9b4ce4ac4df19e0f0a5d51 : POST_FAILURE in 1h 28m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1dd36ab70fd4ff989d78a383fa63c47 : SUCCESS in 58m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c7e274e979e7455bb739d4a534f0ca0a : SUCCESS in 1h 07m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/38143dc447134c6eb488da61d1591358 : SUCCESS in 40m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5255cdc8b77e42dab66babd8d024da6a : SUCCESS in 1h 13m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7b9dade85d3e4f0aaeaeeff167a881f5 : SUCCESS in 1h 15m 26s","accounts_in_message":[],"_revision_number":12},{"id":"ae0ac5dafba8d6b92328f0b0440b3ed6199e080f","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-14 13:56:39.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"21460496d398932da4a077f1af86af45e8789caf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 14:41:12.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff9ecdb5d721450d8840294d97548542 : SUCCESS in 31m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fdb88ee4c5942399a8852a0e8c20cc5 : SUCCESS in 40m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5ca1215245ca16998cafdc6a3469ea51fd240d35","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-14 15:50:24.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805391/12 : FAILURE in 1h 52m 29s","accounts_in_message":[],"_revision_number":12},{"id":"772f409e714eb91d21f34e26b78789faa52176fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 16:08:35.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a8f218ce7cc74d2d87926af0e2ce74ce : SUCCESS in 34m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/aa41bf5f85da4bffa5708b375efb0c9a : SUCCESS in 1h 59m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c1850e20325047d8962253a268917016 : FAILURE in 2h 08m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6e7319e18cf74ddcb62d44e90f714745 : SUCCESS in 1h 13m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/39adbe7c398947fc900cb34f3b56c3ef : SUCCESS in 1h 35m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ce95ea167fd41919b26a571c796aea5 : SUCCESS in 37m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/094e612aa9f0477a8a3a944f229fea33 : SUCCESS in 34m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/190ef3fbbdb14a6095bdd8d2e7ae2edf : SUCCESS in 12m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fffe256df0d74c82b2642a571051af0d : SUCCESS in 23m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/328b8f95a92d417596ba00aa5ebab27b : SUCCESS in 24m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc1c57fedc7044ae93a1582078bdce4f : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998e75c1e8144beabf4c9c20df117231 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd2624d099ab4d6bb04e97eb12f4c9a2 : SUCCESS in 14m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f838c10b16c4fb9b4e0282a58a59729 : SUCCESS in 52m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3664a2b6d3554bf79930cf5120031355 : SUCCESS in 1h 47m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cee99cba3b4046e5888ecda1078ddcd3 : SUCCESS in 1h 04m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/42ceca8a2f8a4ecfa651613af6b82a80 : SUCCESS in 1h 00m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3f2932bc5604410a979110f5ea2765f6 : SUCCESS in 1h 02m 59s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b97e415fad3c4b1588ce70c6397d9b14 : SUCCESS in 1h 22m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7934881570b54b24bbdb074511f97ac7 : SUCCESS in 1h 06m 32s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/16194ab387ce490cabe5610eb6ca2c45 : SUCCESS in 1h 25m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1acff0dd1c1a4d648b1f536c34ada170 : SUCCESS in 52m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/79615336d3bb47baacea10798d97a16c : SUCCESS in 53m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1a3c970c2534459da9d2acb1684dd7df : SUCCESS in 24m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fae5b7ad27574f728f6510320884023e : SUCCESS in 1h 07m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/df6c5d218bfa447daced7ab22d96824b : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":12},{"id":"3783c2d2c55541bb20b5c60b9d42edfbb3079e6d","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-15 05:19:30.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e0f5003b99150505e2915b5c99b90c8bee1fc802","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 06:15:29.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a623248fdc2744a28026a766d017bda9 : SUCCESS in 44m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79b5e8eec20549209b350d9e004c52f6 : SUCCESS in 42m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1d50b9f045f253bb40f132380a5cd0df54e618ac","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-15 07:11:35.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/12 : SUCCESS in 1h 49m 38s","accounts_in_message":[],"_revision_number":12},{"id":"4361ad244238902f3c76bbcce91c5bf551af89ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 07:50:06.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fdd538bab1d9415a81ac3bb668aec6f2 : FAILURE in 45m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a15c291154814d39aa9e2292d590dcda : SUCCESS in 1h 55m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f7d6c60d4bfc4473b0917037b2dd1e33 : SUCCESS in 2h 24m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1154102e687644109caa7d7de82752b0 : SUCCESS in 1h 31m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/398f08f314fa4b009439c4f515a19d88 : SUCCESS in 2h 08m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdf7f6700a0143e5af84ee4fd1f02cda : SUCCESS in 28m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e32d6ff4e99e47e495ea23a2868a090e : SUCCESS in 34m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c6662c543c46d9b595c13ffce80c3f : SUCCESS in 12m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef817c74d9f4459db8b277d36492f9ac : SUCCESS in 24m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39a5cfe90c4346d8883cfe8e608ae918 : SUCCESS in 21m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb630f1d761d405792004eec666b6cdb : SUCCESS in 21m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a57c3284e1724f9c9635972bde56f3e2 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e42e439071043229896767c22ce4844 : SUCCESS in 10m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e5ec7071d7c48e7867aac0c3ef24fed : SUCCESS in 52m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/63c9747ff7cf47feba7c4b1bdee1298d : SUCCESS in 1h 52m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9641f7ab8ce04c3faff8e59a426f3645 : SUCCESS in 1h 51m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/77b8381d05a540bcb481d3d66ef8e6e4 : SUCCESS in 52m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0344dde0a2914718884f62174b1789d3 : SUCCESS in 1h 11m 39s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/673f0f8c6df64d7bb60cd54ae39eff64 : SUCCESS in 1h 17m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9320450350f249428c068da8ca4502fc : SUCCESS in 1h 06m 29s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/e5a1d048596047c280763e1c0b545a58 : SUCCESS in 1h 23m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0375c55efd8246cf9dd905e797c64f37 : SUCCESS in 42m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/82ea1ed355234bb9ac823896441d6e96 : SUCCESS in 1h 00m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1a69876caba04cb39178eb5b71a82d45 : SUCCESS in 28m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a66dc42fe5d34870b42234a6cdecdad4 : SUCCESS in 1h 12m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fcccebf7eed847e283ebe5af24d902f9 : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":12},{"id":"adb52b021a184b8ee02f3bd8dbd64a3bcde2960e","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-15 11:33:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"970edef4a3ab2163adea4765a1d5b4b20b6d9b18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 12:11:19.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/84503d37cbb04cbc92487befdf2a71d4 : SUCCESS in 31m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/313f4ce722ba4857ba0710d989b270bb : SUCCESS in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9565a656714d2090feae54d17d820b795feee05e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-15 13:06:34.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/12 : SUCCESS in 1h 31m 56s","accounts_in_message":[],"_revision_number":12},{"id":"9ae0423f85ad6ea4291922f166bee7a6823377a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 13:44:02.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a12f0fa7992a469ca801585b80a11c6a : FAILURE in 48m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2711a19e03a842f0bd5cd516b3b3e9e8 : SUCCESS in 2h 05m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/76200154dcba4a2f81cc863bb38b80bc : SUCCESS in 1h 51m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/065908c5e623479c812cc12c8e345736 : SUCCESS in 1h 29m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c5bb73d61e5f48f498a8d9a782c3c235 : SUCCESS in 1h 53m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bd7521eec354af49da6940428185162 : SUCCESS in 32m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/490c4d56e0aa4fc7adc2e38f4ee88951 : SUCCESS in 39m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/564fbe051a7a4948bd2aa15ca1e75626 : SUCCESS in 14m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d455767b8a24cd394ecc5fa172cbf1b : SUCCESS in 24m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59d6f986e097410bad1a7b84c50ccb1b : SUCCESS in 28m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eebcd8b04d84b4795c08d0adfcd563e : SUCCESS in 25m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea62d856d92440d780a8ff54bccb68a5 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faa0b67a484c4ad49078f457fbca2467 : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4eaa91114c794afca39b375adb36545e : SUCCESS in 46m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3c72d38f92342748d81489b8f6983ed : SUCCESS in 1h 38m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6e34e6a4feb54a16b9e97590fea36d34 : SUCCESS in 1h 11m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/311e8849b7a344dab1061f95a55d887e : SUCCESS in 1h 00m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c66229e19af749cab8c9dd9b95040f4b : SUCCESS in 54m 32s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7631340c859a41aa90dca3ed41591b23 : SUCCESS in 1h 06m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/76f8544fefde44b88804f95cf9973ab7 : SUCCESS in 1h 12m 48s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/836e4cefd1434b34b8da6c59bbeb0ca5 : SUCCESS in 1h 26m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0eb7d7221734990866932cf11dc3f1e : SUCCESS in 53m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/312ea86bc2d244ed9eb78e05001827e7 : SUCCESS in 49m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/87a8eaf05bf642ca90e7b5ef729b9c18 : SUCCESS in 33m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5700c11f228a4922b99abd17da37bf31 : SUCCESS in 1h 05m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/86b4cc296d3a48ebbdab314444bf24b0 : SUCCESS in 1h 17m 09s","accounts_in_message":[],"_revision_number":12},{"id":"32027e97dc3528cab2bff903a7f177863b825f7c","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-18 05:59:42.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c12bf35b0647d748041631d31eda35fdc65c8c0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 07:13:08.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0aa915a29e974e1ba79862848c7020a4 : SUCCESS in 39m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/117898314fc04cdebca0450e59f8f4da : SUCCESS in 43m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"dd50a41aedee766d20b7269a078080e1650cd43c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-18 07:47:28.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/12 : SUCCESS in 1h 46m 32s","accounts_in_message":[],"_revision_number":12},{"id":"28a4d705707c4e9d6788e3cbcd088a9a8dca10f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 08:43:48.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6751c1abe2a34415add6d1521c8bfdf9 : SUCCESS in 44m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c6bc53a0bf904616bb4cbcf775b9dfbe : SUCCESS in 2h 23m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a9df8ee81d934c1bbdd6bbd64ec0119b : FAILURE in 27m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2e3116b004d04cfc95a63485b5568530 : SUCCESS in 1h 46m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/295da75fe39c4984bdb84aef9b2d6160 : SUCCESS in 1h 53m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d7cb3d128147b1ab7928e42d216c89 : SUCCESS in 33m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a548e1db0ff438997c1b683585566ea : SUCCESS in 32m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c9d32c89e74b828ea182f244096e28 : SUCCESS in 14m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ec698b8dc1444529e6e8a0e7369ac3f : SUCCESS in 23m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40aa4cf60e0742ada05ea4732527252a : SUCCESS in 30m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a84bcc947cc453c9d404c147b7454d4 : SUCCESS in 28m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9804426aaa74c53a08274b59a0fe8fd : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e216cb7a368240f994f2058c531f775e : SUCCESS in 10m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0531d11eb62f4a71b6b9e8060aa8fdf5 : SUCCESS in 54m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e592d6edd1644b48630ce7ba6484612 : SUCCESS in 1h 49m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/190ef2029e964647b1e9243f5c6f6b6e : SUCCESS in 1h 14m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/17bb2c511eee40c9bc0be6a38fcea859 : SUCCESS in 55m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6c20ed75262d4140b58bb9ab6df820e7 : SUCCESS in 58m 28s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3555969a43b044ccb9cd4453195a1274 : SUCCESS in 1h 20m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f9d071f7668e406e9481cfb861a537c5 : SUCCESS in 1h 17m 10s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/43a3931ffcc94d93836548cb09702e94 : SUCCESS in 1h 25m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/061fa6c3f42f4b418dba7eb99a0bf59b : SUCCESS in 1h 18m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c9dcf691ef314648b9e2e6f68b9688cf : SUCCESS in 1h 12m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e1c97f7f957b4337a225097e2b10dc40 : SUCCESS in 31m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb615b02538f48229e878eac4b9c87b1 : SUCCESS in 1h 10m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/87eb3c51f33e4dd2a7006c24700400fd : SUCCESS in 1h 23m 26s","accounts_in_message":[],"_revision_number":12},{"id":"553570e5fb730d87e85e90e88e634daaa0284450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-22 13:59:44.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"10edb87b5d0931419f385c3a71e7940c2653aa39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 14:49:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bd5e6fb0df14b94870f497d5aac6b9e : SUCCESS in 37m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a98e25c5198944e1833cad4f3fb13e2e : SUCCESS in 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"94f7dbfffa5d5139b33cce575a8ac7a2cd9a71e2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-22 15:49:31.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 42m 07s","accounts_in_message":[],"_revision_number":13},{"id":"8b7f497eee677360e5e695dd7225a1fa9fda51dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 16:29:29.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/46f2af442e144eaea35b6c41ec4054d0 : SUCCESS in 43m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dc0f5e3d74f444ce9b3a00ddb8d8ca16 : SUCCESS in 2h 08m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7e5a7117ab334f86807d8f3317263fe5 : SUCCESS in 2h 26m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aa1f6fb124844f6d86a9e1d96840b76d : SUCCESS in 1h 28m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/43917844d9394639af87a17088a636c6 : SUCCESS in 2h 08m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22506805a3cb439db65ea304688b09de : SUCCESS in 42m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0472262624be45f19d04d423c0e7eb60 : SUCCESS in 36m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e09e6112e3408ea9b13685a919e451 : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc6a3c3601b242a8abaa4f66098c0aaf : SUCCESS in 25m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d53e91b40bb442cd9a3c1b15aa7bc46c : SUCCESS in 26m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/399d11ac729f4fbab926a6cd5df4f6bd : SUCCESS in 24m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6819051aebfd42978c44a8519e13c7c2 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3798e4169846485b93617bf3cfcb3e84 : SUCCESS in 13m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6194446d42aa4c6ca1609c74f93d0296 : SUCCESS in 54m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/93731e520fe54b58a3792129c666ce1e : SUCCESS in 1h 24m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ea3725daf2aa4256b465ef6d2bba03e9 : SUCCESS in 1h 10m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b9c7cabaeeb4fe7894bb5e93d0aa920 : SUCCESS in 49m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/89c88abcad894b07a7f0203d4b9beb7d : SUCCESS in 56m 47s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/05187d7546be4c24b8900d2f4e3ce915 : FAILURE in 1h 00m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15b57f67b45946e8bcdc193b2f05fcc1 : SUCCESS in 1h 20m 30s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/eee6dbb4cfa14f47a8cb2c3eb644cbd4 : SUCCESS in 1h 33m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/455e7c119ac346b997fdb57f5ff70653 : SUCCESS in 58m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fa6325e009aa4ed7934c955677806476 : SUCCESS in 1h 02m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a70ee5923559485c83faed6ae7d26a70 : SUCCESS in 36m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3732a6a0e9f04c3eb4358443a8ef7261 : SUCCESS in 1h 04m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2703846fe5af462c950216b6d13c81d2 : SUCCESS in 1h 14m 07s","accounts_in_message":[],"_revision_number":13},{"id":"d9b17356dffa8f4e3204e924e7b488eeaa018f22","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-22 22:53:09.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/91/805391/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"ab48d96dfa577072a7db852e15e72de4d4d00cce","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-25 07:49:20.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1146fb946d51edd3c4ac29e8869018851c779935","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 08:31:50.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1302e3dc72a486c8f50f256ea2194d7 : SUCCESS in 35m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c16ef975fd504498996efce267a2a3d4 : SUCCESS in 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0cc7d9e6c3fcd0485d840d3316643ce92d0520c8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-25 09:38:52.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 48m 26s","accounts_in_message":[],"_revision_number":13},{"id":"02af1733c2b9c6e3b9b75b13180ad7aa8ddd66ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 10:31:54.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/81ee750ceb104a259f755cd11ff749e9 : SUCCESS in 44m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/513bca2f1a1e4ee9b336faca6a0556f0 : SUCCESS in 2h 41m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d08329689c824955a4bb7af20060fc0a : SUCCESS in 1h 53m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/07a144fc72874f9ca52582f13504b662 : SUCCESS in 1h 43m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7232a9b5fb5945dca32fd9b0c5c8b87d : SUCCESS in 1h 56m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/723dbeacf18040a29715a312411861a4 : SUCCESS in 1h 01m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82294c3daf834620aa3d6e1979d1cd93 : SUCCESS in 33m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfcc4df0d0d41dc88dc704ec488c670 : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d3a7e3f46d24d12b0db744219b9c333 : SUCCESS in 52m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aadc4e579ab44cf2807a42fb5702b8b3 : SUCCESS in 50m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8117e746a694087b963781f63d1f0bd : TIMED_OUT in 41m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa531e7f32cc47a5a86d5c35b17ca6b3 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/597d42f69208424eb694caa1a342fe8c : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f6894575934496c9a7ef66c9e60f2fe : SUCCESS in 45m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8005e63532a54192822249b014e6d123 : SUCCESS in 1h 46m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d60cf1695d83456f82aae3d253775802 : SUCCESS in 1h 03m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/15ec63398c48421a9726df441c1ccba8 : SUCCESS in 1h 00m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a1df51fc3fe46dd8bda35f274091205 : SUCCESS in 53m 47s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5296c363bdf24b9fbc38d2a16e7de13a : SUCCESS in 1h 08m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/297ae10ef88343619812968f3e328c1c : SUCCESS in 1h 00m 43s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/1632256261684531ada0cdb6f2828ebc : FAILURE in 1h 24m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcef0560a2af46e38d34900270fefab6 : SUCCESS in 45m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5bbc532c92294508b77c91e8cdf0a5c8 : SUCCESS in 45m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e8b6d86f42f143a78ea12198d78eedcc : SUCCESS in 28m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9e87134b7484fecbd03ddc1884305c4 : SUCCESS in 1h 06m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/58220750c81c4093be03a629b3d430d1 : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":13},{"id":"f9a0d77a9ba3acda801adf8fce1f76a2f5e29b40","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-27 10:29:46.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"38ac8208ab9c665fc92a2e37af8f5be33a847ae9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 11:15:38.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a086f06c4e1e45e5a5548b3eaf45d281 : SUCCESS in 37m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f142fe4619b542c6848d9d7556a55239 : SUCCESS in 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0d5415abe4489d622b53f7f1ad583cb5db143d21","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-27 12:29:40.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 58m 51s","accounts_in_message":[],"_revision_number":13},{"id":"c0d9e45d330d1f1af541a9613395de2049edd4d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 13:28:44.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0eb6c42b0a074d169906fe7eafa7ae28 : SUCCESS in 46m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9da94adc8bde46e9869d54ec198d2727 : SUCCESS in 2h 15m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e4e3192c850c435dbe61ee2fd2488bac : SUCCESS in 2h 33m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7f8804e6a37f41199128ece87c086912 : SUCCESS in 1h 38m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2d8daf18dfcc4049aad103a51a162026 : TIMED_OUT in 2h 53m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a560cd3a78140768489c350e6164687 : SUCCESS in 35m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95135c9b9f8f43af956b5e9c5192ad8a : SUCCESS in 35m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db406aa320d44e2bfa714b89f7c7335 : SUCCESS in 13m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d618b028bb5a4450b1b8ecca2db08a8d : SUCCESS in 25m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/813cede2c5694c87a6d66b547b834f6f : SUCCESS in 23m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb6b16546d7742ac8f385ea4535e2be5 : SUCCESS in 22m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea1256238a849699ddd9b88ecc84c8b : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e179c824d2146ff9abd7e633773643c : SUCCESS in 12m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/68d31949d77945779f53cef0e6969b46 : SUCCESS in 48m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6dd9dccf6d448539619024c2d85ec98 : SUCCESS in 1h 24m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b711d43ab8204740a7da31cf68fc4b49 : SUCCESS in 1h 16m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/102a1478ca0a4946954b448468f2c7ef : SUCCESS in 55m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/109509df012e45f3a61f1166cde432cb : SUCCESS in 1h 06m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/790214b5f88346a1a60561b9b9dcc8c5 : SUCCESS in 1h 12m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c207f2f71a543939c2fb8b73a9bcad7 : SUCCESS in 1h 11m 23s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/d76073eeb0e64afe929f723fba25cde3 : SUCCESS in 1h 24m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcea09549da14790a58254432a772b58 : SUCCESS in 1h 00m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4c1e1779d9d64504917d8360654b9666 : SUCCESS in 54m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dc97d05ec0ff4f178053e963cad9cb32 : SUCCESS in 30m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c73e0cd40a44f328474b6eff176281b : SUCCESS in 1h 09m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e2f8b4fb6a654243a610efe28354e67d : SUCCESS in 1h 14m 24s","accounts_in_message":[],"_revision_number":13},{"id":"b6e8760fb1f758758bfe2f9217997b94e3b808e4","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-27 13:47:44.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ebc69e4deca86334103a4cffe746c63369258e7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 14:30:38.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7040a755ddae4740aff1935b587d5d3b : SUCCESS in 36m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/577897118e0a47809ea4bc879a8266cb : SUCCESS in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d1f19ebcd6105e8c64caef09c831ba788c35d097","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-27 15:39:26.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 50m 15s","accounts_in_message":[],"_revision_number":13},{"id":"92644f61800548df8a413c71149cc3308bf9cdf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 16:33:12.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"6895aef3f534ac0ca74c4aef22094c29103e0827","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-10-29 13:08:12.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6eb7d7e9ab48108b581d078c2fb808ade7b318f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 14:09:19.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f1038b3ea09401c959fd53c82090f20 : SUCCESS in 59m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c18043ef0aa41ad89cf0593284cf3f7 : SUCCESS in 57m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"66786c18818f1314b09d902f452a791c4be1252d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-29 15:28:15.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 2h 14m 16s","accounts_in_message":[],"_revision_number":13},{"id":"6b57f04289898bea9921addfd7e0e19407618ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 15:34:30.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c7fd10d1fcca4b0c81e4c0255be2fca8 : SUCCESS in 42m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b204532fb35e453483e8cf8b142207e8 : SUCCESS in 2h 23m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/777aee3de20e41c49961381e75c96fb8 : SUCCESS in 2h 07m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a8776f921b2427ba3d62467fae1813f : SUCCESS in 1h 20m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/84283cb2896045ef9b08e8e7dc420760 : FAILURE in 2h 01m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b0081d7ba54350bb4008e06e6c7268 : SUCCESS in 50m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cc7bac7689745308da4f8b30e468935 : SUCCESS in 49m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f92d461aae47afb5b7b763215e42d4 : SUCCESS in 13m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d770e58ca1894a2181db1f880981352f : SUCCESS in 36m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0faf0bf0bddc4ecc936d0a68c7030284 : SUCCESS in 38m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/180891684cb6442fa0eae0ccb1fd44cc : TIMED_OUT in 41m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a70026cb91e4a3aa4a7247cca9f65ba : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce40e5d4feb9403d95dab8a9670de849 : SUCCESS in 11m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c91fc874a3d4f41919519671c06233b : FAILURE in 43m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5b041871d6346b79e81b6f50bc16a0e : SUCCESS in 1h 29m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cfe55bf48ce047a49f2565ea3acd14b2 : SUCCESS in 1h 17m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ae925a560694e88a12d866cbf0d9794 : SUCCESS in 56m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/43f823e500e943959fd197d191fd4ed5 : FAILURE in 26m 35s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8c685873571b440780bae2df74874e5f : FAILURE in 48m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/33afb9e5a7b54a6c83d879a2e127dc23 : SUCCESS in 1h 09m 41s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/265e8d5d311946cda4233526b0c558f2 : SUCCESS in 1h 27m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/376bb1a098bb4e52919baa359f86553b : SUCCESS in 50m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/35067c53ae4b4ce8a5bad094fec202d7 : SUCCESS in 56m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d7bb68cb42ca4f1e9e83bbf964b47171 : SUCCESS in 27m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/374eb415d4e34cabbd6051c540ee9808 : SUCCESS in 1h 05m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/35836850e4bf4cb6b2357e017f7f8746 : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":13},{"id":"62660791a65a9c4ff48d96725895ab1e392e3382","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-02 06:43:54.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ec6590ba8d3c67c0a2d492c3139e4977ef5bfc45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 07:39:00.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13ae18d14652407db8f2b3b8c8e8fcfb : SUCCESS in 45m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad070e3cc87f4173a672d618695bea2f : SUCCESS in 50m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"19596a493afd2980cb709ed36e8751aa8ea67348","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-02 08:35:42.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 51m 08s","accounts_in_message":[],"_revision_number":13},{"id":"62813e6ccca7064d2b2239a8e669c50258d5cb6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 09:19:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9bfdff2b3c794b2d8f28e322c7554a9b : SUCCESS in 44m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/45a7348efc63457d87cc07b7c81fdd18 : SUCCESS in 2h 33m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/774458a0b6484fd9aa38119ab0fa1005 : SUCCESS in 1h 35m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/96d78f66886e422d84c1c4adc6bd14a9 : SUCCESS in 1h 38m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/455adb53c3704a009ef0926a8159ee92 : SUCCESS in 1h 31m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81fe20818a884357a09fd95940c3b0dd : SUCCESS in 47m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/903436516a8a4950a42194c28cf2048b : SUCCESS in 58m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd00455496d24dd6ba5c7dfa5325e179 : SUCCESS in 13m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/badfeb651a694360b8d09b69df3db323 : SUCCESS in 39m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef9a5ec7d4c94c5ba77fa89e3c253d11 : SUCCESS in 39m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13ff783640864ec0b14992daa0a4b140 : TIMED_OUT in 41m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e345d12c643443b809a5d5f49cf7eb4 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69f557126b704a88836a19fc8edc78fd : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bae77caf7ed44c11bfd26a3f4f96ffda : SUCCESS in 46m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/58da1118ee374ebf83a5f44cfcde1659 : SUCCESS in 1h 59m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/aa02adf98e364f519cc5796b1a4c4675 : SUCCESS in 1h 01m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf188601feee4a13b9e3dbd9caf16aa0 : SUCCESS in 1h 09m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a3e5561be4a43e2a2ed974902ff4abf : FAILURE in 1h 06m 17s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4efdfc4ea5d94b7eb82fdadccd220403 : SUCCESS in 1h 14m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/287a9d50ed16497b8e5ddce436d893ee : SUCCESS in 1h 12m 56s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c70b0766f10c45ec9ba7c6a547569cf6 : SUCCESS in 1h 46m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/341a9e1cc20745f2816a869212914063 : SUCCESS in 41m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b1598a03606b46fa8d4e9e03d6add927 : SUCCESS in 1h 09m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fc7aa1a5a1a3459e875d80a7aea6c3fc : SUCCESS in 28m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5e54486244494b428600e780b205453b : SUCCESS in 58m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6c15ba540fd34a6b85485b13d14a74ef : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":13},{"id":"3f0fdafd0ac5206a8563ac193a6f1e5f06a39fe8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-02 21:18:44.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"871d565c1e40a06054b22c3a3a085dd9d2e3bd77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 21:19:43.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"6da1d412cecbe8e35f6788e6db22dacf3b9013e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 00:30:28.000000000","message":"Patch Set 13: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"1bbec26e831613d3dd81e73bab432d460bd8ed5a","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-03 07:34:01.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"45241e22e0906bdfe96ade7a61901c5e8eef5ef7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 08:39:15.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52772212e793465bb2c08c74298a8b11 : SUCCESS in 53m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/112ea216afb44cdc9d4350dfe3ec2ef8 : SUCCESS in 58m 58s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ad8df8ac9be38af2d0953bc5384c73828b280092","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-03 09:31:09.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 56m 26s","accounts_in_message":[],"_revision_number":13},{"id":"70f709ddb70a0e32682a89a55d66ca453e56a321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 10:09:47.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/65f65c717c6e4c47986efbf8625c915f : SUCCESS in 42m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/060b5572727a4eae96273488754fe0f5 : SUCCESS in 2h 15m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/93a83fdb114441318198a24f822ad5aa : SUCCESS in 2h 34m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/627be8714e0e4b42ab48f8f954f4ce37 : SUCCESS in 1h 34m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5ca3fcea85fa4daba6dc280e86c73c84 : SUCCESS in 2h 28m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f24fb4cd1544469cb939f5f08d4b0c1c : SUCCESS in 45m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44f7846188b949199514c5c6d26c3fc9 : SUCCESS in 49m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0eafb835f84c48ae7b280862e191a7 : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9897c87fea64eaeb2cee9631327e838 : SUCCESS in 39m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/268c31cf9d394dd18baae239873c546e : SUCCESS in 37m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00a34a3e1b0944d7aa60e3277c9d65e7 : SUCCESS in 36m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7994c6a535cc4fcfaf9e37e137d2a34d : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5483b2214b534dada1325711d3213fc2 : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/12fb84c487df49a8a75231926ee50ae9 : SUCCESS in 46m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c3883c256d54ef58d588148dfdab9d9 : SUCCESS in 1h 24m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f7b3536d67334be58d55bddcb8898e4d : SUCCESS in 55m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7284fb83d37e40128aa94e63d3fa171c : SUCCESS in 56m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/56e42a7a36ae409eb96772c193d72945 : SUCCESS in 1h 05m 12s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/53a13a73adf74922b9f2523733ed72a2 : SUCCESS in 1h 06m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6456f13845b2462b9a41d2cf5ad86bd4 : SUCCESS in 1h 08m 46s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/04801b4430a140ff809ae700489ae5ea : SUCCESS in 1h 28m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/252b99d6e6194c32af16d86d1fdc4e82 : SUCCESS in 55m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/31359bec3701457c991bc130fd53875a : SUCCESS in 56m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8643bba320164582a4b864365d6700c7 : SUCCESS in 28m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ed8734dfd3243569cf0367edee432e5 : SUCCESS in 1h 07m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1214cf53a13040e284c63dadce1a70a5 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":13},{"id":"a6645a5395c143e5402e791e790d562c73adf90e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 10:10:34.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"431040aa3079901a63843198106d7c1783da92f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:20:15.000000000","message":"Patch Set 13: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"c8fc5daa9cc579a2244dd4d8b30cd3ed6e0a0df8","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-04 15:06:49.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"015067da5aaf27f6ac63007c41dcd9d2b9a42f61","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-04 15:37:17.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : FAILURE in 13m 56s","accounts_in_message":[],"_revision_number":13},{"id":"3bd22688dad4827f00f24a7eeff47308cf5b5db4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:08:43.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/174be2a9d0c24a42a6ae5612026e1bed : SUCCESS in 58m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fadd771958b14f8e95f34f1aa4c3732f : SUCCESS in 59m 13s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a4886eab064a10d5b16d553b9952e04e448abcbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 18:05:46.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6e9e2f5aea6840b8a3b2cb11f64c811c : SUCCESS in 43m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/402b4e4862f0411fb21eb622b925b003 : SUCCESS in 2h 40m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ef57213659644f2b36749418aadfd39 : SUCCESS in 2h 19m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05a5a02d24864caf858a5a104ba5c007 : FAILURE in 1h 29m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f0baac110ec749cf950a6a1a2e793b5c : SUCCESS in 2h 21m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e945a54b77244118b226ee0a1ad536d3 : SUCCESS in 50m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd10613b2382460881913d355bc585c1 : SUCCESS in 51m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a242a81a6e514772ab6324aec258bfb5 : SUCCESS in 14m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd0652046aa54158bd242d797cd8cae0 : SUCCESS in 49m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7377dea9693b443e844cda049b64fc05 : SUCCESS in 44m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c70d94a4fa04e7f9bf7351200cc9696 : SUCCESS in 40m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be38d3dac9434125b5eabe330d5b3737 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2cb3fc2e5b34b23807d5a0a536a980f : SUCCESS in 12m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1426850dd4b4416980c50cc8c0da722 : SUCCESS in 49m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6a3982cbcaa459ba93b5f198241d814 : SUCCESS in 1h 31m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a1c05352f8424042aa7c1b598b528f79 : FAILURE in 29m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/81f91314967a4a928160bbeca895ed67 : SUCCESS in 59m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/09892436f02544beb518c3620be90599 : SUCCESS in 58m 39s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d0861dcbd7ee4b8ca50c6c6e0510d971 : SUCCESS in 1h 13m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f12aa6eada274a2aaf46c7003bc77444 : FAILURE in 1h 03m 10s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/53bc3b9d88a14722916153d2f19e93af : SUCCESS in 1h 47m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d7cd53baa544bc8ab3603fe2ccb1d8c : SUCCESS in 1h 29m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9626973086084e168e8cc5d35c1a4899 : SUCCESS in 55m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5962c1d3916e45a7bdd3b1c5ca39d17d : SUCCESS in 42m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1176ab95b6514771969e6dc775e07912 : FAILURE in 29m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6bbcd4042f024d16ac8323dd79f0e1f0 : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":13},{"id":"d727846fcfc037b30089a37701063cbb9ced9665","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-05 06:42:53.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4adb995141785993a18886089e36d462faef46a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 07:42:06.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f4e5ef0df824a34a02efbf87bbcaa98 : SUCCESS in 52m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ffdef6e2cc64dc4ac8822f6f7765ec5 : SUCCESS in 53m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9c2444922a2eddfc8b6930fa0736d0a27d613f55","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-05 08:31:57.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 48m 23s","accounts_in_message":[],"_revision_number":13},{"id":"08b8c45423982bf2d2f731316be953274b65adfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 09:42:48.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c3843caf5f5845e1a7c7cefadadd949b : SUCCESS in 40m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/23d76b86748a4ecab7329a214e5b2d67 : TIMED_OUT in 2h 52m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9d3e5d4e51c04017ac32fa91cba9daee : TIMED_OUT in 2h 52m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc4a6ed9ea7842ecae39c0a42ff7c923 : SUCCESS in 1h 23m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0cc907ec07864d249973fd5d2da5db27 : FAILURE in 2h 34m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/646794e04e534da187d024e8f71a753a : SUCCESS in 41m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e04158058d8348f6be9f15cad7b6260b : SUCCESS in 56m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6efb00cbcce34b5cb85199554d1bc49f : SUCCESS in 13m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dd2bb0a76ca44efbc7afc13fdf8eefe : SUCCESS in 38m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06b17ac063bd4ecfac76bd7144bcc2e8 : SUCCESS in 48m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5348c981ca3f4902a963ad4e013c5051 : TIMED_OUT in 41m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba46043047a414c96a88d542382d4b1 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a790a1a8fa3c4536882ed861d8f9b0c2 : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/87927537cd0c49ea8a15b27befcc1fe2 : SUCCESS in 1h 00m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/69a8043fee13480c82e3328db702664c : SUCCESS in 1h 22m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6e8930845bf443f38aa175859a6132e6 : FAILURE in 22m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4cf3a44901b54538a805f03fd9beff96 : SUCCESS in 56m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f542c9233852449c9783974c28c6d406 : SUCCESS in 1h 24m 34s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/59dd2ecd5be443039f737211f8406eef : SUCCESS in 1h 07m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6ad72fd4276741b0a1d5d033bb4ac78f : SUCCESS in 1h 17m 43s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/694c5a86082a45d48ac39a658f9ca7b9 : SUCCESS in 1h 21m 13s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cc4f5fc2ceb4031bd7a9953fa1c7a69 : SUCCESS in 56m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1051ad81d92c41599ebf2cf05f8d043b : SUCCESS in 58m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0cd6966272c542239ef3d6ff5c3bd4e0 : SUCCESS in 33m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9b7d5b4bb074484db36082e35a6629d2 : FAILURE in 24m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/64d68aeb76084c7aa5e5165f6bcc26bd : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":13},{"id":"8f76fe51c0d9b2b933710a0db0156be1c37fba41","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-11 06:50:23.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"cb0ad9e33d6014c558c681eb844c7448cc26ecb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 07:53:04.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7105091199624591b71ecc4efc017c49 : SUCCESS in 52m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e405ba9479741369dea7889a22fec6c : SUCCESS in 1h 00m 42s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"50b19f90fdda54b8043693856c6faa36c1e90f79","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-11 08:22:08.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 31m 04s","accounts_in_message":[],"_revision_number":13},{"id":"f250c4d37529eeed3ca1d9e9abff32f9dac7ac54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 09:49:39.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1e9a715f86754802bd38cfb9e5f253a8 : SUCCESS in 47m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/756aad12ba4e46e9939a7232f0b72c50 : TIMED_OUT in 2h 54m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/756a38241e634ad2ac9ef293caea7e3a : SUCCESS in 2h 22m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2be4c1fc2b0942458c28ad2542920b33 : FAILURE in 26m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4d4a2cd6df154f4190008a8b7e01f98b : SUCCESS in 2h 10m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39c022ae653f405b9e5fd48027baf8cf : SUCCESS in 49m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86b43d3172d44c43b2343791aa09196d : SUCCESS in 47m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3252e5c492964f508a0e95a6a48b9a02 : SUCCESS in 15m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6ec5c5cbe4a40c5bee2f5ea8de55ce0 : SUCCESS in 51m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e7edd89792416392af4833b14e1a45 : SUCCESS in 40m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb5ceba3bfc64257a04980bf979ac6b0 : TIMED_OUT in 41m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc3659bf23b4a8aa18c7d7ca9c455d8 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26798132688248b2abb8db55bcef5e7c : SUCCESS in 14m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e314832c68c24dbfb1e0bce62b960600 : SUCCESS in 49m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2d8b47f90c54fadb214841b5573aedc : SUCCESS in 1h 23m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/df71766a26ed4ef99d8db8b309eb660a : SUCCESS in 51m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc3cc2a873244137bc37f7f192e96c00 : SUCCESS in 1h 00m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2b016e3a70884131a244e7ea77446e5f : SUCCESS in 43m 34s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bc849fe836f343edabe333f159baec88 : SUCCESS in 55m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e39451e3d90049bea5d7e413fdac3302 : SUCCESS in 56m 44s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/36ce14fb94394436b1b6fcecbac243bd : SUCCESS in 1h 18m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddf9239dc91c4ddfae3e25e78e1f93e7 : SUCCESS in 53m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ac75a478fabf4f0f981672922e526d27 : SUCCESS in 45m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/44b985dc335c40168083454c31a5018d : SUCCESS in 32m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad66331b268f494b828da0bcd811625e : SUCCESS in 59m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6cf9270bc79e47be86610893d73f4b0f : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":13},{"id":"e57da1a22511b88482b70c518a1fbeafba87649b","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-11 09:50:28.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"274ae505cef853c090f1b5587191d456ef5fe356","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 10:49:40.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae1806592d7a428182f3ee80d97400d6 : SUCCESS in 56m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fb8df26eb9a4cb6a9d8e966ab3048fc : SUCCESS in 52m 43s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a28843ca63a4461ac9edb4eea94121a01c643731","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-11 11:40:08.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 48m 55s","accounts_in_message":[],"_revision_number":13},{"id":"da70aa83d99959d0ea9608c68a7d60cd4c87ec6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 12:28:21.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7960c8b97b47473381c3f6d87c7b2283 : SUCCESS in 42m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a116db9bacb34bb0ba9700588d611d5d : SUCCESS in 2h 03m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/62344cd82b1f44f1b2027fa3794ba45e : SUCCESS in 1h 51m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ab8dd14d69943a9a8886530be31b7ea : SUCCESS in 1h 23m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/df8118bf55a843c095ab295594f0c1bb : FAILURE in 2h 32m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0b96f80e7b34a098519b9d63cdd3da9 : SUCCESS in 51m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5fde75deaee4b1ca22475776349bdbe : SUCCESS in 56m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085c350a883546d7af77475d34269881 : SUCCESS in 12m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b08d7743b5b4e43b580b40551f39440 : SUCCESS in 41m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23dd8599ab104a81954aa626921b1df9 : SUCCESS in 44m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a456613098fa4c7cb4dab6c4c9dc8681 : SUCCESS in 41m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adbe127058724e8d97fc05d062a10b55 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1f8108acaa54213bcc9dd577b5c937b : SUCCESS in 11m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8eafebf212c4f2c843c03c848902226 : SUCCESS in 41m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cdcd8fdea1d479997ab075e09d20a63 : SUCCESS in 1h 26m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/933a23a07e134437a5ff50162acaa83e : SUCCESS in 1h 07m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b5c1c1919b542f789d1a7b90b68c90b : SUCCESS in 56m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/366a2ec779814bbfbab6f723447ab822 : SUCCESS in 57m 23s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5b10167395b04b139d7146ce91beb13f : SUCCESS in 1h 19m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8129cd094e6347ee9d0c0621ff2bd8d9 : FAILURE in 1h 02m 49s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/8e619c9773bb404bac146c78f17e08f2 : SUCCESS in 1h 33m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39e496343d6047e886c49932278be175 : SUCCESS in 1h 03m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cf07efd9881048e5b5875c857b79535e : SUCCESS in 55m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9bd8c54869db446795508e576ecb79ce : SUCCESS in 32m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7dbba3741278496fb0cc5a028cae86cd : SUCCESS in 1h 12m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/02371a63dc944e33a565b62be0211031 : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":13},{"id":"963f993d9399d271e8e0ef1846f412eb54498d82","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-11 12:36:39.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4665162282487b1527816a131dc31bc5311b8701","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 13:39:49.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed01ca9ce2a5438fa02038b98b446ef0 : SUCCESS in 57m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e79cab8d732f42e9874f313a029a8393 : SUCCESS in 1h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"551849e28483a81593bbedcfaf66cbc3956f2811","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-11 14:35:48.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 57m 47s","accounts_in_message":[],"_revision_number":13},{"id":"3c24ecc9677dad0d98470c3e387e740a5b23e61c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 18:55:29.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3f5382ac18bc42caa7ea53e494c6859e : SUCCESS in 51m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5ec60c1959af44508845880aafe6b32f : TIMED_OUT in 2h 51m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bd5aa571b9a840b0af13c212e406af18 : SUCCESS in 2h 24m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66229d1a8aac4101b316b58aa4a54da7 : SUCCESS in 1h 58m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ce607378ebcf4842b91224130bd1004a : SUCCESS in 2h 44m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d413f2437adc46a4bc9dfa318d4699a8 : SUCCESS in 54m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4a1136c75dc40208a46b30110bf2719 : SUCCESS in 50m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69cd44dcc82b4100b7e2cc874c2b4b71 : SUCCESS in 14m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4cf2f7334ae450991a31c8b38020bdf : SUCCESS in 43m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/802199289fc14866bfae84123f1c5e94 : SUCCESS in 38m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4828d9731772428a91beadda50a60aff : SUCCESS in 37m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c36db7cb4e4024aaa58e77430176c9 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb220f8e6af842fba887d6a7ffed5edd : SUCCESS in 14m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5be50cb6c9e4a4a93fd270a3051d6ae : SUCCESS in 1h 02m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/79d174ffdd0c461ab501f2d696d8b623 : SUCCESS in 1h 38m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9ff58fa1a17347d5b0bf1a44c2fce6d5 : SUCCESS in 1h 26m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7597484343cf48b7bda263bab2c307b0 : SUCCESS in 1h 11m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/99447c78aecb4a86939a35b8883fc304 : SUCCESS in 1h 11m 07s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/39de9a90085b413cbaa084ca73d4761c : SUCCESS in 1h 21m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/61ab8b0aa254436196875e06cd0d27d9 : SUCCESS in 1h 15m 24s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/6ac9af933bb3470b8a6e732c0c080cdb : SUCCESS in 1h 49m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b06f4c2a1e0b437884b255ff045e40c9 : SUCCESS in 55m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c5f091ce930145b8b1b70a5b3316f9aa : SUCCESS in 1h 02m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4394730fdaa4444e97c1a7e5484e28a0 : SUCCESS in 40m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/088f796311864540bf12186374e6b303 : SUCCESS in 1h 11m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/74e4bb5a30a84241bcc48675a57b21bc : SUCCESS in 1h 20m 35s","accounts_in_message":[],"_revision_number":13},{"id":"8f2d64206c62d2acc97e0cad519e78357f11aa4f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-12 08:09:43.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"fba2c9e7759efa4e4dc740951cb5e41b15bfe647","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 09:14:07.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3316cbda1b114afabb32c43f2d027f22 : TIMED_OUT in 1h 01m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e731a858aa6340b4b340783fbf0e5d40 : TIMED_OUT in 1h 01m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5aa644727107c4d58e6a6df98ce8285454bb36ac","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-12 10:17:01.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 2h 06m 24s","accounts_in_message":[],"_revision_number":13},{"id":"e314ce6d28b7e966d2b49e9c08fd83af70728ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 10:39:40.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8df3a2263a0443edbf1dc03493550bf4 : FAILURE in 39m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/51002d3f7e6f4b6aaa8a432e2d673606 : SUCCESS in 1h 43m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/abc0571d6618431c9c90b958962ece70 : SUCCESS in 2h 27m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/838d4d13ffc14521a96a633b8f6ed217 : SUCCESS in 1h 47m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/114d5fd54b4d46539f6cbd1ab847685c : SUCCESS in 2h 05m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8ed786b33914e529f50e479ad0b0579 : SUCCESS in 48m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a85ff029a0fa4e3a8243899328ee1feb : SUCCESS in 49m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aacc9df06c1e4b61b949d8f511452551 : SUCCESS in 13m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9d1e82157924400ac0184ea3eaea680 : SUCCESS in 39m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe75e3687db418ab389e602029a64b0 : SUCCESS in 37m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/deabc9e52eb741c99cc65418c1e60b41 : SUCCESS in 36m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30cea9a6b17e4178a9c37b025707bc09 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50e7217ca51494dae93abbecd637283 : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc25240fcaf947628258b463e713d954 : SUCCESS in 44m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/88313866c6824645b1e178224fdd6294 : SUCCESS in 1h 10m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/438b923ecb544b3fb80152033069bab3 : SUCCESS in 1h 06m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6038b07a0cbd409daa05aafb0778919c : SUCCESS in 43m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1da49eda4e994ecba4d6d0aa91f0b4d8 : SUCCESS in 56m 57s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1dc2138993be434599dcf2598d8f8880 : SUCCESS in 1h 10m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2c355c07dfe14104a3df348e158d8bff : SUCCESS in 1h 09m 27s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/4b276f6470db4e329ee7a5c9aae779a1 : SUCCESS in 1h 35m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c1749a561034ab8acbc82cd174b8bf3 : SUCCESS in 1h 00m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/627af1bf09ed42378ab5a36f6d32b951 : SUCCESS in 1h 00m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0e888e211c2a49029a55da812c39f499 : SUCCESS in 29m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65bf31c5e415417b8502dabea05194d8 : SUCCESS in 1h 08m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a16d0c9ba7bf45bc9bb32b3033b7fabe : SUCCESS in 1h 19m 11s","accounts_in_message":[],"_revision_number":13},{"id":"105f368a3aaf626e3f9bf0ad42cd72829577c8c2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-12 10:40:17.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"e5c95241157060b5179f687007f579b059e98c10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 11:43:54.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b395dc3006942ff8c772c1c498d676f : SUCCESS in 44m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/daaa22a87884490289609e48dcb5297d : SUCCESS in 1h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"6b9f1c8a4dba6ea14f2d8511430e4a1d2486f5bf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-12 12:47:42.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 2h 06m 42s","accounts_in_message":[],"_revision_number":13},{"id":"f3e89396b40255c8cc1f5f57566276bfe05cf9a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 13:39:22.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cc1a45288b354d7b8ca1cb70d5771a9d : SUCCESS in 59m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c27c4a4e4bff46dd8a4d0a5e29640d12 : SUCCESS in 2h 50m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ba1a98224d9f4ffcae485b3360432c58 : SUCCESS in 2h 44m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/77e166a7e29e4d41bba1278ec9e20da6 : SUCCESS in 1h 46m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/137f77e220be4e659ae4643947d65aa2 : SUCCESS in 2h 00m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bea32aaca8f4d2193cf3fca315756fe : SUCCESS in 57m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0541962ebb2040ed8129dc534c6d1455 : SUCCESS in 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a77e413fe74c52a52c484e9656b4e3 : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48da7c44109a4b44980cd85f67a4653f : SUCCESS in 42m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c580ed3150641b3a1e79e2cf63d750b : SUCCESS in 43m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/056ef0ea0c1e4a6e9878d9c8139918f0 : SUCCESS in 37m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d8db6f46b946ddb049bdf7cea7ec08 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00a099632d7a43dea3c2a733f2f140e2 : SUCCESS in 12m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e14e3a9ef2fc4d1782ad70811aff019b : SUCCESS in 56m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a11b325863940e2b5e96ba0477e58cb : SUCCESS in 1h 34m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/04a030c13f3642a28c0009f7e737d627 : SUCCESS in 1h 09m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a06eb33baf964907a8364b3c46e793fa : SUCCESS in 1h 08m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/01804cb12c6548f28f099c0a29df003b : SUCCESS in 57m 22s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aed709253b4a4393b9bd74b98edb3119 : SUCCESS in 1h 18m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7485964d068242aaa6fb0502134cbcb8 : SUCCESS in 1h 14m 30s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/6e76fbc5ddf941ae824d3c78cbdb370b : SUCCESS in 1h 53m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11e90e870c5f4f95a6a0049240349287 : SUCCESS in 58m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c7982bf9e6c24e57a068f6b9bd403634 : SUCCESS in 1h 04m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1e00d6c1d5154dd4a7e3b1338983e914 : SUCCESS in 32m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f448bd6305cb44c1bd251f94c469c03d : SUCCESS in 1h 11m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e2052c9613c84338b2827a68d9ba92d9 : SUCCESS in 1h 31m 12s","accounts_in_message":[],"_revision_number":13},{"id":"03bea549fe46852f8729e1af29084753ccb594f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 15:57:33.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"fd31b0f01741dbae9f2885759b8c5f0926f63d61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 23:09:19.000000000","message":"Patch Set 13: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"4c5516153e1486e15112e46c0370266b1f45b8b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-12 23:13:06.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"9fbeca07951a7c8ccc7b4b34103aaba9dadc7712","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 00:11:34.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7a8fa0ac97c4db5b01cd191f2299058 : SUCCESS in 55m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b053ea5698014bb2844bd0e7de5667cd : SUCCESS in 55m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a346b35aa7c3bca0edd65255b8988ccbb24ddde7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-13 00:46:55.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 33m 07s","accounts_in_message":[],"_revision_number":13},{"id":"35453c4146592d65fca53c8c352735e8abc1dd02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 02:08:18.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d82d09855b354d72801196d3b9fdbd8c : SUCCESS in 44m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e16f1a19c34a4b20b628f42fdb901983 : TIMED_OUT in 2h 51m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/461adb67d75245e993e3531c75dd0464 : SUCCESS in 2h 20m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/31a3c12e5d504df59db8263099c05324 : SUCCESS in 1h 37m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/624fb32455db48a58838c5a993526746 : SUCCESS in 2h 35m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e0e7be9433e4ad0972baa03d1207c24 : SUCCESS in 47m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4da3b4b5ee647b299bd3674d975340f : SUCCESS in 56m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb94d25c5d24361b4cb1e0dff1e9434 : SUCCESS in 13m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85e65293118e4bd8806ec119ee05cf23 : SUCCESS in 46m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c11f29459c0499bb608c9405cb8e2a5 : SUCCESS in 43m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9682732c227d4514858445316b9901cb : SUCCESS in 40m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29fb783cf1ca494b8e043cf6386d663e : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed30f648593040ad9dd2c4d0ec1e552d : SUCCESS in 13m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ce3723530134b4cb59f168b79793648 : SUCCESS in 52m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/254ac4ff4db64899826273e7f4400447 : SUCCESS in 1h 28m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7dfe76b48cc349fdb6d4fb84a8084ec0 : SUCCESS in 1h 23m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4da4fdfd8bc44981a8a5eba9d24c2b16 : SUCCESS in 1h 01m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fb79cdafbaca44fb91e226959a5aaa40 : SUCCESS in 58m 11s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/46b85ac112e14a4d9ce2642caf9f91b8 : SUCCESS in 1h 13m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f0ba0102c09c467cba49d66dcdc622a7 : SUCCESS in 1h 20m 46s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/93380598cef7416ba5d3d74841094424 : SUCCESS in 1h 35m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f725989f711246cb8ddc59486cb253ed : SUCCESS in 1h 00m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d1126aad505843dca7006b90cbeddedc : SUCCESS in 57m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/88f6c0e815e643498f22a70b44e593c1 : SUCCESS in 36m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b7a26218cf604c70abda08c51f0bf80e : SUCCESS in 1h 10m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9805d4a966614f399ede70d9c9fe87d6 : SUCCESS in 1h 21m 19s","accounts_in_message":[],"_revision_number":13},{"id":"973e87a41fbf8a013a986082e6a72a82ab250fc3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-13 02:13:47.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"4f305ee93ad142bb7ce7c285a5d7991b10a9d683","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 03:26:20.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cdc53bdf2a24fb6ae2fadb08c4fcb7c : SUCCESS in 59m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9abc4c57e1c4daba07bc9389733b0c0 : FAILURE in 1h 00m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c68ae2653c8cff8ccf66a88ae7c59f8d4158ff9c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-13 03:45:44.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 31m 15s","accounts_in_message":[],"_revision_number":13},{"id":"0ebc445ca0ae1ebe2da5afd038714d8892ea1a3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 05:30:06.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/595d605f045e45b9b65ac07a686a3de7 : SUCCESS in 46m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/95b86656837146f1bde3d03de28b6110 : SUCCESS in 1h 59m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9de31e55ccd34ae1bfce8f8029e4971d : SUCCESS in 2h 32m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ee5e51ecb434902a35d9c93760876b1 : SUCCESS in 1h 53m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6d7906b9c9c7497daa88b2f6b915d414 : SUCCESS in 1h 55m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e664f777d47642959a21723f297e8f23 : SUCCESS in 46m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/747988947222488a87b7238e59f4b9a0 : SUCCESS in 49m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1992c2b484c9443aa1c4bb60fbb35252 : SUCCESS in 15m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5717c5011494b5ca9de644dfa3867f0 : SUCCESS in 43m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9364099be2384cf8bb1a24bc590ba902 : SUCCESS in 40m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba79d72d57d84bf2b0093046c5589571 : SUCCESS in 39m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11ccc6e378d43509d9790f195a88780 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0fabfa2ef6e4ccabd6ec6d5fbfefae8 : SUCCESS in 13m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/156ec9524e0a42a18f4079552dc848fb : SUCCESS in 53m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/79c2bd1defc2465aab616547968dadd2 : SUCCESS in 1h 39m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ffe0267cb5ac4ba487312572c36f6762 : SUCCESS in 1h 13m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/70f74c662ebd40e5bcf55e3517c8d712 : SUCCESS in 58m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6ee0f44423d74d09a27c5d3b862e7c78 : SUCCESS in 59m 37s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9a7baa4ae3364816ad43c28115548d4d : SUCCESS in 1h 09m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e49ad0e0db2419c8c26dbc2b30af921 : FAILURE in 1h 02m 07s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/98b2cfffe585426b86c9462ba4715d22 : SUCCESS in 1h 41m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d4ce15796b14d8e8d43b2af12797c8b : SUCCESS in 1h 05m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aef115eceaad4a35bc8ac6aece5f5ba0 : SUCCESS in 1h 09m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/49b2b61215be4036b2fa632d436f9ed6 : SUCCESS in 39m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e971cc52bba44ef9225cbaa25545482 : SUCCESS in 1h 06m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5fb0cd9209b14e96945829cae4e20319 : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":13},{"id":"66dc72f664acbb9ea3cc6a0525718eed791165d5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-13 05:44:33.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"1ebf76b0c0d19271560a4ca4e4a00a60fdaaaf17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 06:40:49.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/578013a2d73c48d39b4cf9b21f0bd65a : SUCCESS in 51m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86be084932744ccf8765fd0ecbfafcc1 : SUCCESS in 44m 03s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"71800a87f0d86f60b22ad5c6290446b636f047c3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-13 07:15:06.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 29m 51s","accounts_in_message":[],"_revision_number":13},{"id":"381e84d8a0499482131f6cdbba6e1f7db9d213cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 08:50:17.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d718acf9313f46169ffb2e4ec7fffbee : SUCCESS in 59m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/475132beafa04f669e0d93f87052eb6c : FAILURE in 2h 45m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6a503a91c83d44f29d922c6eaa8fbdfd : TIMED_OUT in 2h 50m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9cc648a2624a4a2198fc68a4b941ea8e : SUCCESS in 1h 49m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/13f928be31434b9186bf292b099d9467 : TIMED_OUT in 3h 04m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4eb7715e566d4488873c5d74b82f52d5 : SUCCESS in 53m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5eb254104a6a44399e0e15d9204c8322 : SUCCESS in 56m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0838729db6444642998032551a319934 : SUCCESS in 14m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95c72355be4f4547a2c6365145ddc2b4 : SUCCESS in 37m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d0f9501d22941f895108a2c72b4a200 : SUCCESS in 41m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fef76ef3cf824985a6b9ab4b69d33501 : TIMED_OUT in 41m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7decac9a57544747954f702608c3fd07 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaa0593d1cc54862a49e9d8542a5b8b9 : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/957d511f19794924bcc3f69a579e69bc : SUCCESS in 1h 03m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1775aeebb8e448eaea256cc4745bcec : SUCCESS in 1h 31m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c2ea77f950f740cbb2c79216a82e469d : SUCCESS in 1h 13m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d253db0f9a8c4380928f5b4bb4526ac9 : SUCCESS in 59m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2694b6c0fc9e4a4f9c98a59a85fc7f92 : SUCCESS in 1h 14m 13s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d85c775f920c4d9196dc87bb9736a997 : SUCCESS in 1h 14m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d05b79e9b4854bbbac19acc78ed0b34f : SUCCESS in 1h 14m 31s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/fdba4a4ed0904e56ba71a274d1358cf4 : SUCCESS in 1h 28m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01cc91c744374c9e8e4e82a099d1c5a4 : SUCCESS in 1h 09m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c5480c5c262b46c4a8493ab367022b17 : SUCCESS in 51m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/437388e223ee47d3857a51166bda6c72 : SUCCESS in 37m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70fd2c91a9624729a9331002023e6bd2 : SUCCESS in 1h 18m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/672bbc884b524ea8aa0546182eb46d91 : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":13},{"id":"80ad7fab51670b9811b453b8909d7ab7a47a25f6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-13 09:15:19.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"e3f80fe76e366a534a1b5e011753acd72f4c0d85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 10:12:54.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fe22d02c6b54f4d9f86a9d52508edcf : SUCCESS in 49m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88dd78e9c341472c8260c93c4479090e : SUCCESS in 44m 25s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c2712a80e1f35fa315a5f23df2babfc84dd0b8d3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-13 10:56:50.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 40m 50s","accounts_in_message":[],"_revision_number":13},{"id":"f615eff5b7ce23c06e1e7ab1f96e69b9d861ac68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 11:53:39.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a68dc4984ffc49d2b0b919d819753b39 : SUCCESS in 49m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/95d72c3f92cd4328863dd6634faa0210 : SUCCESS in 2h 35m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6ba6bead9a3542c8b0670323a280d785 : SUCCESS in 1h 48m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8093e65f58745b6b8968e23dbb289af : SUCCESS in 1h 35m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/af44738a0425488c82d815c3f660ee25 : SUCCESS in 2h 16m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5a231d6107c4d44b295100455f43be1 : SUCCESS in 53m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74a32adba6ec486698300851bf58ff97 : SUCCESS in 57m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d476458c64e94c63b893a29a1b8a62d2 : SUCCESS in 14m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c73bdc473ec486b908a1ca251a79b3f : SUCCESS in 46m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0776780d3f1e4a9bb3024906036a540f : SUCCESS in 45m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce72c71ad98b4412a1e1a73d359f080c : TIMED_OUT in 41m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71daad30160f448aafae530676fefb82 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f415a481b745467b80cf4a434f32c6c0 : SUCCESS in 12m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eadaa5fa1fe549aaa4fa262443f51779 : FAILURE in 46m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e75c8bc01424e8f83049aa3f515cf18 : FAILURE in 17m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5b8cdd06d00a494290be2865f85744ae : SUCCESS in 56m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa4bc618ba174322a9babe0869de3769 : SUCCESS in 54m 42s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/826e4c1976ca48dc8dbd05757973e9b3 : SUCCESS in 52m 57s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/de93b3c8dfdb4d5aaec5df0dc8d3f420 : SUCCESS in 1h 09m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a067ba0cf0154b3ba2c1f17f7ba5bfb9 : SUCCESS in 1h 04m 57s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/56afcad754bb4631be159c271291e92c : SUCCESS in 1h 20m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e117ae5d8a7f48398c77d7352f8b576d : SUCCESS in 54m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9b05576f75d840c6be7002f34467728f : SUCCESS in 1h 05m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c004de3378314bf09269f77611560c59 : SUCCESS in 35m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a6ff7692050d44f49362eca3aaa0d138 : SUCCESS in 1h 04m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2dd5bb3fd65f45a58409287963cba23c : SUCCESS in 1h 21m 26s","accounts_in_message":[],"_revision_number":13},{"id":"57f7173b835987e6dce193507780bedd09f68738","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-13 12:16:00.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"3fe7cf32b7cbe765862ed0845e726602f05870f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 13:14:12.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c148babcba3a4e479850b3e1189ca496 : SUCCESS in 46m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ec78ba717764357a2754150189a13cb : SUCCESS in 56m 03s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"52aadb737b70e19fd6a7f6d7fd1f976841d322dd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-13 14:09:20.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 52m 39s","accounts_in_message":[],"_revision_number":13},{"id":"f12bdba9566118d1da284ddb8832435daabb0b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 14:30:45.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f25d8c6452a2403e90f1a81a9d2625ab : SUCCESS in 41m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c2a73654ec754e95b86fce1dc4ada797 : SUCCESS in 1h 43m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bdc15d6e2eb44549a24a602610af945c : SUCCESS in 1h 50m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42b5257b9dd24e09bc4f14c538729f7a : SUCCESS in 1h 17m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c858cc92ce3a4ea18bd4f39482d7ef41 : SUCCESS in 2h 13m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18ca96f591d541f48eeb80cbff0a4b0c : SUCCESS in 44m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fa03370c13e46ca85661db2898d0842 : SUCCESS in 49m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7324a21f139547749521ca696b6ed2fd : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c71664171e6f4593a921529411ae1c2e : SUCCESS in 35m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8afdc874fdb454a87c4f2e932d652d4 : SUCCESS in 37m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d35a503dad947dd8c20b6ae4042d70a : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe3938e4f184f9d9e1da68478af9a85 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dace795f88349a9bb5b5931c46415a1 : SUCCESS in 10m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ad6d0513c3f40838a603f7a54870771 : SUCCESS in 45m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e21328c3f58643949ca116218d4f152b : SUCCESS in 1h 32m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0ac3a0e3b36c476496581d4545ffe9c7 : SUCCESS in 1h 01m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/95f67b5dccfd43efb4e29a82c2725f76 : SUCCESS in 53m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e7f2e928acb1471c9f4326defc0cfc75 : SUCCESS in 55m 55s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/acde30b49edb4955b0a4517be4bbd671 : SUCCESS in 1h 08m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f84328e5f774487089a5e031e4855856 : SUCCESS in 1h 06m 37s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c7abc85b3e67455fa8c54f2233c5c33e : SUCCESS in 1h 25m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6178788ffaa540f38237dee4b04a1c29 : SUCCESS in 55m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f2b0ecf124744215ab02bbd06ebf35c3 : SUCCESS in 52m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/34563d2960564b55bb26cc55a19c1fdb : SUCCESS in 31m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4be21e1a85d940218473b567888815b7 : SUCCESS in 50m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/becbe03ba4854fd7b3886485ebfc7f18 : SUCCESS in 1h 10m 47s","accounts_in_message":[],"_revision_number":13},{"id":"38326072baeb781ea68f13c0189c995e9862ad13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 14:31:28.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"c7e2d43b8f9747156597dbb581ff5a9bfb4b2364","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 15:03:01.000000000","message":"Patch Set 13:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"42d532dfdc4aea2a546cc5e6c106c5e54f408406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 20:31:20.000000000","message":"Patch Set 13: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d2a81558087d4736a4b300d56f6fc220 : SUCCESS in 44m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/194a935386224c0b8c51a4178ad191cf : SUCCESS in 2h 06m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/33bcbc282881464a93e09c514d6c7922 : SUCCESS in 1h 45m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/133a90a838f944e5aa7310ee21021b16 : SUCCESS in 2h 40m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3a5c4a90b23f408c8f99a5db649da051 : SUCCESS in 1h 37m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a4d679935e346c2bb4d9a895551b094 : SUCCESS in 48m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebdec91671844ca681b1324e685c7781 : SUCCESS in 12m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/541e8ee56e08438fbdb05fc86718b28c : SUCCESS in 41m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a727629b17e487fa0958914c0d82c2d : SUCCESS in 34m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/156f24022ee843419fe5f634494b2169 : SUCCESS in 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/067156dcf4a441ed8300b9cabcc565bc : SUCCESS in 10m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/df2d460354194bea88e217acd31ad780 : FAILURE in 49m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5501fd8b47045f28c5910f60bd0fb37 : SUCCESS in 1h 28m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/09a51a1b0386493ba62ad743ea116738 : SUCCESS in 1h 08m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/156ebe4749ae42daa18a7669f228c89d : SUCCESS in 54m 25s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/1c30d30299164a0aa11c5ff9119eae2c : SUCCESS in 1h 25m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a317e5a3510a4a03963a1f29ead9624c : SUCCESS in 51m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f9aef4658aaf4ac7bf9f75d9b3ae3f73 : SUCCESS in 55m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7c541876c68d4b44b1251c5eb1b315e5 : SUCCESS in 29m 01s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/22d8175e2ae5492a927a1ec020a0fa4a : SUCCESS in 1h 18m 55s","accounts_in_message":[],"_revision_number":13},{"id":"9ea7e1fea16d42c83ca9e073bbc0ce340ce2213d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-13 20:47:56.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"abfeeb9ba03bb9b00eb5821553e2e154342f4061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 21:45:21.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/842b535461d94b0e9ed5a5eb8cbffde4 : SUCCESS in 55m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7a914bd9cb14fc6a75aa7f175c3148d : SUCCESS in 53m 52s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4da1443196a3af0a6b8963aa40870b2e3032b65c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-13 22:35:27.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/805391/13 : SUCCESS in 1h 46m 58s","accounts_in_message":[],"_revision_number":13},{"id":"03ada930142c4a266fb416f6680ec3e993b10c83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 22:52:38.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/791732a491b7473082552a464fb212bd : SUCCESS in 42m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d12e5a8ad8a5431fb934f206977dacd5 : SUCCESS in 2h 02m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4288b722a6e94ea1b9bc448c8b3a6442 : SUCCESS in 1h 36m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78a074bd0f6741baa645ee291e37519d : SUCCESS in 1h 24m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dc6d36c2b7004aee88829343e8ecda84 : SUCCESS in 1h 53m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/313f90e70afb47939e55813dbcd4c9ef : SUCCESS in 43m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab01414c7b2240aeba4b5fbaad0f8a38 : SUCCESS in 47m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e44a770a0964ff0a7ec7846f28a2d41 : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0bdac9274ac4c96969d4ceb6725382b : SUCCESS in 50m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60895f12abf44f8abab909bf57aec051 : SUCCESS in 36m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7071b2c66964ad1acf44fe0ffc5b9a1 : SUCCESS in 39m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/349680a9e4ac4cb4aab0befd3aaafecd : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c2ef8dd63c4bd99facc581db9ebf88 : SUCCESS in 10m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/be8a201c3e724b1196dba150bbcc8bcb : SUCCESS in 53m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1122f46d4ae54a4f9de93b94867ee75c : SUCCESS in 1h 42m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bd2ce73b5160476d92cab062ba729dc5 : SUCCESS in 57m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1b401173931440ab473bc87319f81d3 : SUCCESS in 45m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/476575b5cfad449f876d57a31b32fd95 : SUCCESS in 49m 50s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/54abdf43645e45739f93d507ff53facb : SUCCESS in 1h 01m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/93d1b65ed8494db98ddebf23f1b61eb7 : SUCCESS in 1h 18m 42s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/5486bf40e3a747c6ad96d9742a2a98bc : SUCCESS in 1h 24m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1a8977112804735b8a0e8d780d53121 : SUCCESS in 55m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/60a9f27f46da49b8845f82de20d1f53f : SUCCESS in 54m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f81a52f2d9d7425284f30a158a782190 : SUCCESS in 28m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d90ff82374cf4862a53ad6576eeca5b6 : SUCCESS in 1h 00m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5f122e4c750a4b72bdadfea812523b45 : SUCCESS in 1h 10m 03s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/167844b149254ccab47ac60141e93323 : SUCCESS in 59m 02s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d867bcba1f4445a3bcd4f5c4147dc02d : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":13},{"id":"67c621b8ecd4d905e1d4c22b650eb25cbb9006d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 22:53:24.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"0b707c99784381992033bd67b093df9c6dc7b1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 02:05:50.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bbe3a767c72b4bc6a569c64c292acc9b : SUCCESS in 31m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cc8296d9747447b89477fc85656645d6 : SUCCESS in 2h 15m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b4d51155756d400aaa3da9c80d1ebd38 : SUCCESS in 1h 54m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bc3b2fed1b09433fa2988e23ee848220 : SUCCESS in 2h 04m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/889d8f7095c04cf78039a27f6968d3fd : SUCCESS in 1h 28m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e7cefbce529421790541b0c202acf6c : SUCCESS in 52m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737df1325ac44c7f806b39ddb89308fc : SUCCESS in 12m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1a5741ebcfc40df9de6c6a085758fd4 : SUCCESS in 35m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/093a92d28c604da19cca9b4857fbd82b : SUCCESS in 39m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643aee3fc300477aa46fb21cffa99162 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/492b745b0b754d93be64d2faced50828 : SUCCESS in 11m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/08d1c2526de14953b3757e2b7588522c : SUCCESS in 46m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcb1c77a4fa348e0aaa63b185ab2eaf6 : SUCCESS in 1h 31m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d07f2307ff4a46628fb98e2ab66a9058 : SUCCESS in 1h 03m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/30bde45071ad4b7293930e8f46373851 : SUCCESS in 1h 02m 51s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/da4ba487c6c9495e85ccbb90e130d8c4 : SUCCESS in 1h 18m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f30f48347404ce5af8f3f3903c84e2f : SUCCESS in 54m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2875946307ff4ef7a332f5c2a4718485 : SUCCESS in 52m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9589c6e198394c1580ffac1027c333c3 : SUCCESS in 20m 49s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3830ba8806004e188b75158db01b80ae : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":13},{"id":"8a80202679b7f68bf72308653789396c854b717f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 02:05:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"ce64d98f92c1c82ddf20b253bd2258f02c10b9f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 02:17:09.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a489279001bd46c092deb781acc2be0b : SUCCESS in 2m 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1731111a18040a0a5894ee1973d1ff6 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"aada855f6dc4e0e8a6d087b47382fc4e30ac5dc5","revisions":{"e618fcce5b8ae66a938ca62fa7455eac835830fe":{"kind":"REWORK","_number":1,"created":"2021-08-20 14:18:48.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/1"}}},"commit":{"parents":[{"commit":"95fc8953de04337a14e07ddb2a4a4b2e83c4aab7","subject":"Report pkt processing capacity on Neutron agent RP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95fc8953de04337a14e07ddb2a4a4b2e83c4aab7"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-20 14:15:14.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e618fcce5b8ae66a938ca62fa7455eac835830fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e618fcce5b8ae66a938ca62fa7455eac835830fe"}]},"branch":"refs/heads/master"},"bae436655814edb478b44aced928fdd69ba84d13":{"kind":"REWORK","_number":2,"created":"2021-08-23 13:06:31.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/2"}}},"commit":{"parents":[{"commit":"0cef772fd15d50a9daaf312f30d41d5d530aeea1","subject":"Report pkt processing capacity on Neutron agent RP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0cef772fd15d50a9daaf312f30d41d5d530aeea1"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-23 11:17:17.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bae436655814edb478b44aced928fdd69ba84d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bae436655814edb478b44aced928fdd69ba84d13"}]},"branch":"refs/heads/master"},"76551f825fd357bae85a246e8b3a145efd2f1c10":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-25 14:47:12.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/3"}}},"commit":{"parents":[{"commit":"cb7f9de61ed3e3ce80d60c19a303ff8eeb609f90","subject":"Report pkt processing capacity on Neutron agent RP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb7f9de61ed3e3ce80d60c19a303ff8eeb609f90"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-25 14:44:41.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76551f825fd357bae85a246e8b3a145efd2f1c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76551f825fd357bae85a246e8b3a145efd2f1c10"}]},"branch":"refs/heads/master"},"8f89b9566bfb0e45428930f1b93bd675e84f1c6c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-31 09:59:42.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/4"}}},"commit":{"parents":[{"commit":"426df904ecf921f27d7d7fb55ead24b85f4d4d51","subject":"Report pkt processing capacity on Neutron agent RP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/426df904ecf921f27d7d7fb55ead24b85f4d4d51"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-31 08:56:59.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f89b9566bfb0e45428930f1b93bd675e84f1c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f89b9566bfb0e45428930f1b93bd675e84f1c6c"}]},"branch":"refs/heads/master"},"067782ba9ba71e46981422d9755f4a36f14340e5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-09-16 14:01:13.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/5"}}},"commit":{"parents":[{"commit":"bb8537b4f71ba5c919dbc371e528ebbe3e3f359b","subject":"Report pkt processing capacity on Neutron agent RP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb8537b4f71ba5c919dbc371e528ebbe3e3f359b"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-16 08:42:47.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/067782ba9ba71e46981422d9755f4a36f14340e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/067782ba9ba71e46981422d9755f4a36f14340e5"}]},"branch":"refs/heads/master"},"96566c2513bc042c0512b0be375e22779f0efa53":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-09-22 06:57:47.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/6"}}},"commit":{"parents":[{"commit":"ffa160b0e74d64362d3f19edc7858e8b349cbc72","subject":"Report pkt processing capacity on Neutron agent RP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffa160b0e74d64362d3f19edc7858e8b349cbc72"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-22 06:54:58.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96566c2513bc042c0512b0be375e22779f0efa53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96566c2513bc042c0512b0be375e22779f0efa53"}]},"branch":"refs/heads/master"},"277c88cb61b933e82b2816c55ba95d3e45d671ce":{"kind":"REWORK","_number":7,"created":"2021-09-28 12:43:13.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/7"}}},"commit":{"parents":[{"commit":"ea4de88f6484d3613b1c8ab01052f162a981e6a4","subject":"Sanitize profile column of ml2_port_bindings table in the DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea4de88f6484d3613b1c8ab01052f162a981e6a4"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-27 13:04:20.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/277c88cb61b933e82b2816c55ba95d3e45d671ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/277c88cb61b933e82b2816c55ba95d3e45d671ce"}]},"branch":"refs/heads/master"},"8d6d5ed420dfec4f1ee634f9d0a173d505c35c54":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-09-29 09:02:30.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/8"}}},"commit":{"parents":[{"commit":"305b4398c359d6b426187a3a383e50b85ca6db30","subject":"Sanitize profile column of ml2_port_bindings table in the DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/305b4398c359d6b426187a3a383e50b85ca6db30"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-29 08:55:45.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d6d5ed420dfec4f1ee634f9d0a173d505c35c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d6d5ed420dfec4f1ee634f9d0a173d505c35c54"}]},"branch":"refs/heads/master"},"1d052b7f59dcc821b4f8d5a071af9cd8a48bbb82":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-09-29 09:54:51.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/9"}}},"commit":{"parents":[{"commit":"d1e4e1492621e4cb4f5c9e000e93106d769b419a","subject":"Sanitize profile column of ml2_port_bindings table in the DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1e4e1492621e4cb4f5c9e000e93106d769b419a"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-29 09:53:54.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d052b7f59dcc821b4f8d5a071af9cd8a48bbb82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d052b7f59dcc821b4f8d5a071af9cd8a48bbb82"}]},"branch":"refs/heads/master"},"8d58b10e0414ea2e57f7de2d0d9c2449adaebcc0":{"kind":"NO_CHANGE","_number":10,"created":"2021-09-29 10:30:56.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/10"}}},"commit":{"parents":[{"commit":"6c94d82b1ee9c7eb3536ddab5f5ab49ce158e4d4","subject":"Sanitize profile column of ml2_port_bindings table in the DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c94d82b1ee9c7eb3536ddab5f5ab49ce158e4d4"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-09-29 10:27:45.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d58b10e0414ea2e57f7de2d0d9c2449adaebcc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d58b10e0414ea2e57f7de2d0d9c2449adaebcc0"}]},"branch":"refs/heads/master"},"9f602aeefef8d501ee1d6fc96315a7ab3314b16d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-10-11 13:32:53.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/11"}}},"commit":{"parents":[{"commit":"053cf8559b0f5f2a24a1d2a8af6c0f29f75cdcad","subject":"Sanitize profile column of ml2_port_bindings table in the DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/053cf8559b0f5f2a24a1d2a8af6c0f29f75cdcad"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-10-11 13:17:51.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f602aeefef8d501ee1d6fc96315a7ab3314b16d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f602aeefef8d501ee1d6fc96315a7ab3314b16d"}]},"branch":"refs/heads/master"},"be888443f591729f311f29a4e65bdb5f98ada3e5":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-10-12 13:49:09.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/12"}}},"commit":{"parents":[{"commit":"82e74240da4b43cb6febeeb4365ffb791d53d7a1","subject":"Sanitize profile column of ml2_port_bindings table in the DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82e74240da4b43cb6febeeb4365ffb791d53d7a1"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-10-12 13:25:52.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be888443f591729f311f29a4e65bdb5f98ada3e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be888443f591729f311f29a4e65bdb5f98ada3e5"}]},"branch":"refs/heads/master"},"aada855f6dc4e0e8a6d087b47382fc4e30ac5dc5":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-10-22 13:59:44.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/91/805391/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/805391/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/805391/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/805391/13"}}},"commit":{"parents":[{"commit":"d699a955cd55a6401a787630d37f04c5691eea78","subject":"Sanitize profile column of ml2_port_bindings table in the DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d699a955cd55a6401a787630d37f04c5691eea78"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-08-18 07:12:05.000000000","tz":120},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-10-21 13:17:24.000000000","tz":120},"subject":"Enable QoS minimum packet rate rule for OVS backend","message":"Enable QoS minimum packet rate rule for OVS backend\n\nThis patch does *not* implement dataplane enforcement.\n\nQoS minimum packet rate rule is enabled in OVS backend driver and\ncreate/delete/update empty methods are added to enable placement\nenforcement.\n\nPartial-Bug: #1922237\nSee-Also: https://review.opendev.org/785236\nChange-Id: Ie283ad3a4ec433c88ac23f798908cd143159394b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aada855f6dc4e0e8a6d087b47382fc4e30ac5dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aada855f6dc4e0e8a6d087b47382fc4e30ac5dc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
