)]}'
{"id":"openstack%2Fneutron~805930","triplet_id":"openstack%2Fneutron~master~Id6b1654768b20080cc5894cc27eaccdf2fa25fa2","project":"openstack/neutron","branch":"master","topic":"bug/1931953","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-04-14 19:38:14.000000000","reason":"Change was abandoned"},"30454":{"account":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"last_update":"2022-12-16 08:20:53.000000000","reason":"Change was abandoned"},"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2022-04-14 19:38:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id6b1654768b20080cc5894cc27eaccdf2fa25fa2","subject":"Use Openflow to implement DVR L3","status":"ABANDONED","created":"2021-08-25 02:55:22.000000000","updated":"2022-12-16 08:20:53.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0e7c58c72fa9f427a3c599e0d427cce0487c4106","_number":805930,"virtual_id_number":805930,"owner":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-29 11:25:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 02:58:09.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2021-08-25 02:58:09.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-25 02:58:09.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-25 02:58:09.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 03:00:40.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-25 03:04:45.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-08-25 05:01:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-25 09:17:54.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-28 16:58:27.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-29 08:31:59.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":31814,"name":"XiaoYu Zhu","email":"z503755743@gmail.com","username":"z503755743"},"state":"REVIEWER"},{"updated":"2021-09-30 07:15:50.000000000","updated_by":{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},"reviewer":{"_account_id":30744,"name":"uchenily","email":"uchenily@gmail.com","username":"uchenily"},"state":"REVIEWER"},{"updated":"2021-11-25 01:24:39.000000000","updated_by":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-11-26 08:09:35.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"CC"}],"messages":[{"id":"5aef467b3f6efc2ead9107b40e6c8d1e969cfa50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-08-25 02:55:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36873ced34225ac0b8b34789bfc57cf3a7f35d09","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-08-25 02:57:15.000000000","message":"Topic set to bug/1931953","accounts_in_message":[],"_revision_number":1},{"id":"0ad17048ec148f8d0b37c50b6f62298dcb13fff0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 03:37:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b28e0430a37d4133923c9459b03189d9 : FAILURE in 39m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0203d3e3425a4f7fa58ca062a8ca8329 : FAILURE in 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc683bc6d9fbf54178bb23bf56d35dd6a484a87c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-25 05:01:20.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805930/1 : FAILURE in 2h 05m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1013d4a8408832d0b7a6b6be429c6571ee15339c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 06:04:36.000000000","message":"Patch Set 1: Verified-1\n\n(26 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/040a5b05429b4d8d8606dadb290a04dd : SUCCESS in 45m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9503a421ad4349e0866423a1370d0261 : FAILURE in 28m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1964293805414052bdaf70dcecadba91 : TIMED_OUT in 2h 50m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e4f547db6f754682b992060381b7f625 : FAILURE in 2h 29m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0e25b021e5684a24b109d0df0beaded2 : SUCCESS in 1h 43m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e9e6bbd16d842ac8abc19bb019abe66 : FAILURE in 38m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2577efc4a0174ef6986ef1963d422dfc : FAILURE in 32m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/145e774e5b4c4ccbb9cfea479650c329 : FAILURE in 39m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe3eaac34e0492ebba920a1812cee09 : FAILURE in 9m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d694e391fe8547df94412fb1bc8bd1ac : FAILURE in 29m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f45efa62347046ba9c8a494a27e6fc81 : FAILURE in 23m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/879cd6b61e9e4d49a3ecda73bd376b22 : FAILURE in 24m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414e79d64a95406181c9217d6d25f918 : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c67f08098f047a49b78b6b6ff45da2b : FAILURE in 43m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca4ea17ea1d54e0e853b71121a2ee748 : FAILURE in 1h 52m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/adbc5f7bf9b94b34975681727a0ba6e6 : SUCCESS in 1h 10m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ad36389f3a440afa517820cb3cca613 : FAILURE in 59m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9099a4f5f23c4351abd450e5e0f6daa4 : FAILURE in 1h 20m 50s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6ef30d81e9fb4d2787bed9d77bfe3e67 : TIMED_OUT in 3h 05m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b8eedd170c148d9ba75880259e150bb : TIMED_OUT in 3h 05m 05s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/ea492b0d70034fe88298ba85ef3f8fb8 : TIMED_OUT in 3h 04m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a08ce27849f41f093c332b1ee1ec7e8 : SUCCESS in 51m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c7f506a7d4db4a11b48ebfbca234e590 : SUCCESS in 50m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4acf7e5ade89466c873c8239ebe0633a : SUCCESS in 31m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f61d426e41ac4f39b294d11a84c8db66 : SUCCESS in 1h 12m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cd09689f6b7f4e3ab383cc345cd5754d : SUCCESS in 1h 09m 23s","accounts_in_message":[],"_revision_number":1},{"id":"832b5341438e494f1ff8fc5148dfe25feee9fec0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-25 09:17:54.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/805930/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/805930/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"524a054a197d64807379cf0e6720770b4cc62bb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-08-25 09:26:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81f193cb2f1737fa8bcdd227e8df73a10841b384","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 10:13:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed06e207fa76433b9aa5b199a1dc8f8a : FAILURE in 40m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8542dfdeea35451ba29438b3cb99f1f8 : FAILURE in 41m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d533bc09fb4c9fe1b8bf268589e91c417d0cda5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-25 11:43:06.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805930/2 : FAILURE in 2h 15m 01s","accounts_in_message":[],"_revision_number":2},{"id":"706c91b27cdf6c47f75ca651ee11514ef4fbdfed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 12:34:43.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/db8523bb79934d9db3edf10096f36831 : SUCCESS in 44m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/61c2ce3e318c45ddb31cfdd53141c9f4 : FAILURE in 20m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3f48b40e3faa4ee2a9e616ed98cdf5b3 : TIMED_OUT in 2h 52m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3efa86c11974630abbab545f117372f : FAILURE in 27m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cb1867d2f7a54a80805223a159449d32 : SUCCESS in 1h 48m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/66769f0c98bb4cb38334d6a93b3ed219 : FAILURE in 32m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b32f260bd11845b292da0f11eb5c2996 : FAILURE in 30m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46c3205cc00f42498951b82dcd0be40e : FAILURE in 44m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be2aad0f88a849569fe154b6350b5f0f : FAILURE in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ad7921d3f6b49c4927ca7c761da53b4 : FAILURE in 23m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c211f7a93eb240b78f5dce2ba83f96c5 : FAILURE in 29m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de97890a807c4a1086fd4363e2aaf906 : FAILURE in 29m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d69963c7cb4f6b962ce2795948b839 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/185b90ddb4ad49d4b4a0b69fca171c03 : FAILURE in 44m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/13970becbbf84a7bb7b6f6ffdbe95c66 : FAILURE in 1h 37m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d6e4aee034014a7db7dbdfa48786cee7 : SUCCESS in 1h 11m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8acbec331ce7412a8a40be62f274aa40 : FAILURE in 56m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5f3f2828426a4c3b9cd51b882b83ec04 : FAILURE in 1h 09m 41s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/87a421a40690494e8b72dc801dc0efd4 : FAILURE in 2h 49m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/995e5f8cac5e46d9b8e757ae1d964c10 : FAILURE in 2h 55m 03s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/49549a32e17d4cd08260237508c5a602 : TIMED_OUT in 3h 03m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b22de68091514e5b836b2cb0e5dbd42f : SUCCESS in 56m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c7481e52e4044f978f77fe78c1fde931 : SUCCESS in 52m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/98156f616cbb42a0a360640c773fe838 : SUCCESS in 33m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d964644db91c4e91b6ee29a25bde82f4 : SUCCESS in 1h 08m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0a55c4cd366c49a2a669dd6d5f6cc72c : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b3b312f8b232e1b2246ae03ca37ad9ecc101b3ff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-25 15:02:09.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/805930/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/805930/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d8c5c7416ef9578671fd2e081dcf4da3b0293790","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-28 16:58:27.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/30/805930/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a0b4b31b9a6fd417e2bec8d9013824642ba53bf1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-08-28 18:17:24.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2021/Arista_Tempest_Testrefs/changes/30/805930/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c3ad82619dd9c15e20d83b86b44608cdbb01e458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-09-06 01:01:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"caebd7a109ecfecb6929e1fb98c77702f1084204","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-06 01:35:17.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/30/805930/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"194ffcca2f7a4e25c64ed17ec9b7cc243aa5d011","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 01:44:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70c1db51456e4628970e42c47b19e470 : FAILURE in 30m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c84808403534bcda6a8ad3b184e4dd5 : FAILURE in 41m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1fe7d1ded97af557d58c18d10b66a2749bfc7050","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-06 03:26:20.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805930/3 : FAILURE in 2h 24m 11s","accounts_in_message":[],"_revision_number":3},{"id":"fd2676a7a9bf561098241479275dcd64a45559b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 04:07:16.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2ff24d3b227046f996b1666207454d1a : SUCCESS in 49m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3d998315b4ef4f0d80ccfa00dc59ceab : FAILURE in 25m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/98a79c396dbf48dc941e37a7e7cf1e06 : TIMED_OUT in 2h 50m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21222ef9e1b14781905c60adaa2b3290 : TIMED_OUT in 2h 50m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/83d2539797c74cc7bfd89b7d460f6964 : SUCCESS in 1h 40m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1807012679ff4d68b646f21dbea41d37 : FAILURE in 30m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65fbab7886cd4ea0bdb06eae0dc8f722 : FAILURE in 42m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bea6abf450e41699cf2e0bedd62626e : FAILURE in 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb10d1404ff647df998b8c94e45554f7 : FAILURE in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e942567be684945bda585c60cc8b328 : FAILURE in 36m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d10a1195362446d198261e3ad297bc9d : FAILURE in 23m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feb931a32a0d4607abdc7187519ed42a : FAILURE in 23m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7eb8b3954d4cb48abfb728f697c3b4 : SUCCESS in 11m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d3351a5779e41829e22af0f4e8ce270 : FAILURE in 43m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e2bdb10c9674d1293899efd904a01eb : FAILURE in 1h 49m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/eb8a2ebeb64c4cf7bdb0f5e18b09ab5b : SUCCESS in 1h 23m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfcd5e5cd17344948283ef3ed2f631ac : FAILURE in 1h 03m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/94b08249e63d4dffb289642ef03e4f36 : FAILURE in 1h 09m 53s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/89e1a6259b374b3da766ee589bb3112c : FAILURE in 2h 19m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/20c8e7f0ad8d4e018e8ae3ea418390c0 : FAILURE in 2h 42m 43s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/0a58ad29b4a84c1eb56f41c071b29368 : TIMED_OUT in 3h 03m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c0110151a6a4ab4b179fccb4bd30c63 : SUCCESS in 48m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ec7c172fd944452d9329c000626ea9f0 : SUCCESS in 48m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/659ad3b9214a4eea9d62e27141aa5d16 : SUCCESS in 23m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/916011930fea4b23aa40d5b65d4fa08d : SUCCESS in 1h 04m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/84461ad885114f928a59f72b198063b9 : SUCCESS in 1h 19m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b4971bbb5ac6c8c034d971282d79e04343502308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-09-06 04:14:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a60bd8f4ecf732406c3f74821565d4db511c71cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 05:04:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c2447eb473c4e46b61484921df42c71 : FAILURE in 47m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/924098009b73419ba6e35d7b87820f14 : FAILURE in 42m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c211f6eb87648ac7ce986c89db67c1268ddf66eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-06 05:09:52.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/30/805930/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"658a763c8494572ada0d61b2cc9a21ab47a5fc28","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-06 06:19:04.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805930/4 : FAILURE in 2h 03m 39s","accounts_in_message":[],"_revision_number":4},{"id":"c8e81f282cf8e26080f50d7428c2a68e5a1b5652","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-06 06:19:25.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/805930/4/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"7235a9890a736476e152fd4383fc5435d1bad838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 07:27:38.000000000","message":"Patch Set 4: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dab26fd61d05492d9128dc25b16f8c3b : SUCCESS in 37m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ed440e3a24264d8b871fcabd761ab46c : FAILURE in 16m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ddc0a9472e9142d6aab0f7c40455dc46 : FAILURE in 2h 37m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8133e2f9319d447c85d4926109288b7f : FAILURE in 2h 39m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a9a368e1ef9e421f9de8114cccad6ca2 : SUCCESS in 2h 25m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/194a62138b8a4a268b53e006003fef5f : FAILURE in 31m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9657ce4be9143ae96cf161841372551 : FAILURE in 30m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd8ad9b1c1d7454aba99a2003add4b28 : FAILURE in 36m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b78030e3dd407a8eaf16f11787e91a : FAILURE in 10m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6df34ca196c4ffb924d83b22d2cc0bc : FAILURE in 22m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4f3e53701cf4c9786784a86d27311e5 : FAILURE in 19m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbe72493c1ef478db7dc013667db8369 : FAILURE in 22m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8b1370b7994dcb9ad2cdc115f97f4b : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95f1e3d2e7fc43f1a5a5727e4b075ee3 : FAILURE in 36m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/26708d23fd434e9c85975842ef38ba00 : FAILURE in 1h 40m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a14016fe17a240c79b440505a79013de : SUCCESS in 1h 08m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/264b77d9f40c404cb2bbb8059e525b1e : FAILURE in 52m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dafe0247c1e44005a315ec12496111be : FAILURE in 1h 11m 37s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dafeb6499ee348eeb40ccb1aba398fbc : TIMED_OUT in 3h 04m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9331068ccc764c9888f2237a98f7a284 : TIMED_OUT in 3h 06m 28s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/33dc3fb5b1634ec98b1ec2aca5e5ede8 : TIMED_OUT in 3h 04m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3d1ac5703c34208957252af2d49f7ca : SUCCESS in 58m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dd7b43dffd4441a88438cd1b5d67780c : SUCCESS in 50m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a8d9a51c269a41d8896ed52b00251c5c : SUCCESS in 24m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b364cc491b84ae6b93d7a6fc6cadf3a : SUCCESS in 1h 10m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d72f15ce8e5e4d0685e2ed55352762a1 : SUCCESS in 1h 03m 07s","accounts_in_message":[],"_revision_number":4},{"id":"0a7c1c5b7036b7e2be1aca2c3180e8289709fbd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-09-06 08:46:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bc514cb23b67a8a89ee99b1c162846f1d8058211","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-06 08:46:56.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/805930/4/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"648c2e202d1dfe7e2ad0012ae72d84233dd9ca4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 09:41:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e220497f17d4c999e90f8118b6c2e76 : FAILURE in 36m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc60df76f5684422b62c3892058133d9 : FAILURE in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9c72feab6e0bbb39fe150f646bb37cc1d0d8a50e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-06 09:43:58.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/30/805930/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"3f4ae581a8e9f1386ef2ccad297c77e7ab608064","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-06 11:06:56.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805930/5 : FAILURE in 2h 19m 57s","accounts_in_message":[],"_revision_number":5},{"id":"49245083a8741b75579f6e748f401f6979ce153b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-06 11:07:21.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/805930/5/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"8facc4eece2178865a7b170069c77600f4193e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 11:55:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/469c83120e7846e28767e1ac37e2aa37 : SUCCESS in 43m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/df3dae299c104f5b866e5740a691b689 : FAILURE in 25m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/113290e160ae41899b56a9af10336487 : TIMED_OUT in 2h 50m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ff3f17c31ed4e1c8b66259c0ecb3ae9 : FAILURE in 2h 37m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/029cd9ab17f646a5a01ad4d3cb8f8e39 : SUCCESS in 1h 50m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a42870441fd42208c9ed4b58bbd4adf : FAILURE in 38m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19715cef24c54478aee4959b794b002c : FAILURE in 33m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4794ed5d1f942a9a96ce33dd8a03891 : FAILURE in 47m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10279944d9d44822b2c8c85f8e8e77be : SUCCESS in 14m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/504a2e0db87246cf9597c45a9268f553 : FAILURE in 24m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d510f25963a4cad9df6b9cf5b0d365e : FAILURE in 24m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f3c39d37ddc45e398620abbcc07c7d1 : FAILURE in 24m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d22109499446ecbd3190b30d18ae47 : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c844e562d81342bf950e34f17370cc57 : FAILURE in 38m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/309686dcd5f8442386cef7449f9ed1cb : FAILURE in 1h 36m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9f87fa66fd13463499063d94c493d25f : SUCCESS in 1h 11m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/defd6432d5274e79b28614c1d08e74cf : FAILURE in 57m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/140e4128b9ac4527a4f64d90fa598f60 : FAILURE in 1h 14m 54s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c460929f5b3e483e9a1088e97543f151 : TIMED_OUT in 3h 04m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/01b7f15afd734de6ba34a3037e8117ba : FAILURE in 2h 50m 54s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/1da0c805cb594e6a814faab69efd4990 : TIMED_OUT in 3h 04m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b65d03545d443bca876f68a62ebfb01 : SUCCESS in 55m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b0fd68dca02a4a4d91ff488009ffc253 : FAILURE in 38m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6c0379fdd6f743cdb82a160228404451 : SUCCESS in 31m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/caa78446e2064c5485a36e31c1cd9907 : SUCCESS in 1h 03m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5983dddd1e0d453286fbd4cd42eb697c : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":5},{"id":"372192343ddd52933d4d259a1eb1203d861ed2c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-06 16:00:02.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/805930/5/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/805930/5/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"166196f25bcc4df1958b84c31aca96543be45500","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-09-17 09:19:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9572603eb4fca4ffad359307713f4b7cd280eceb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:05:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a80334ffabc44ac98d95eeee2fa2d2b : FAILURE in 36m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33816066db9d41309bef0d3c58c1b1c6 : FAILURE in 40m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f428b4ae3a14b5a6e1a0568d2eed1c4ba18d730a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-17 12:06:50.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/805930/6/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/805930/6/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"18139fc5de98a77e8b7ef9247547f66501644ae4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-17 12:09:58.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/30/805930/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"290e2313219c0e9ecc75887868f53bcc5b1c7c81","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-17 12:27:29.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/805930/6 : FAILURE in 3h 06m 58s","accounts_in_message":[],"_revision_number":6},{"id":"a2e2985c33415f49a0d4848bb8aedc65a984b177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 12:36:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5cfbc0ac217e4b4eb2fada9ea79ebe56 : SUCCESS in 49m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4a25bce7b0be4d09ab64c6f95beaa6da : FAILURE in 24m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5aac3ad3c9f24752badc22beecc049b9 : FAILURE in 31m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5f4e5dcffa274177bf80bd47c37ad624 : FAILURE in 26m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/804064895cab4ebc88fd434ef87a0827 : SUCCESS in 2h 41m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eee12560979d44f587792d2f21a6b2f2 : FAILURE in 26m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2fe6ad3680845c695e983124b5c636d : FAILURE in 30m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f485abdc374d4e5687b4692290c9b2a3 : FAILURE in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af55a8f11de4a1bb44d2ba0592a653e : SUCCESS in 14m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d94341f2d5e14e5fa807bdbb2133bd76 : FAILURE in 26m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e51c6dee96843f598c9065ac18508e8 : FAILURE in 24m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc160080cd504a2097e8dde3a7ee12ea : FAILURE in 24m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72082d9b06194c90890b217f2c5836f1 : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1a1801b65d145efb0aec2ab921ac855 : FAILURE in 38m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/61903f4bfe764e089d04baf02a1b7237 : FAILURE in 1h 48m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bc1cc4b3c2764da0ac8245284d524082 : FAILURE in 26m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/11036ec7c5074d86b6135924e59ebb5a : FAILURE in 57m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/35de474540da4936ab0b61a940a41de4 : FAILURE in 1h 04m 31s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fe488dced79b4c93b610688b24309a6b : FAILURE in 34m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a74e2e7ebebc46b6b69f746a0526ad66 : TIMED_OUT in 3h 06m 07s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/7f41621c924b40a19ec98f774bf59097 : FAILURE in 24m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cfa9a38623b4482996e680ae7194ad5 : SUCCESS in 55m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a3dc289c8fb7472e98f3043fb8e054c7 : SUCCESS in 52m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b292a4ae25e546409ccc8a64ad7645d8 : SUCCESS in 35m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a3ae7f8611404122bbab5fede798d90a : SUCCESS in 1h 08m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/134bb95326384d9b82207043aa9745b5 : SUCCESS in 1h 15m 36s","accounts_in_message":[],"_revision_number":6},{"id":"9e74bf295030aaca3706ffebd5a0949ba265597e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-17 15:14:13.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/30/805930/6/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/30/805930/6/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"290dd27a63a5f10d9b37c3943b28267790c6a188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-09-29 08:31:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0244e2eaae419e8e2a74864ae86a94ce512b9385","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 09:17:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/038558db72eb4b92bba6fdff8e7a0995 : FAILURE in 30m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a80ac55175694e65acd6cb5ca6a4b3b5 : FAILURE in 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7c921aa40b1ad34a1141de8fde47422496789afc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-29 10:10:03.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/30/805930/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"48ede81bdfa252b8578e71967c5f990a26a9a0d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 11:25:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/59a9a0bcd0be4b668c270aa4db23b264 : SUCCESS in 42m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c51bff73838141d3a50bdbc3ffa7a92b : FAILURE in 20m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/93794745569345289416bb045ba7d0c4 : FAILURE in 19m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/030733246925428d9b831ab8fde34942 : FAILURE in 20m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f10473bbe7e746a99240c7a886d6193d : SUCCESS in 1h 36m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eacaa7a0028548c68118c3c5229161e4 : FAILURE in 27m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/230d5566d29546a8ba0a04c91d41ca28 : FAILURE in 32m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af1aaf282dd54eec944de5ff6a34f339 : FAILURE in 45m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a13708b39d648f6b16f47741a57c145 : SUCCESS in 15m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa042872f8074d7b85ac22abbe337605 : FAILURE in 37m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9809e6c169540188151d9bce1380c80 : FAILURE in 27m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fea78f136ac74b84a0147e19eb5ad9e3 : FAILURE in 27m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/132a6c4788ab41c4bb7abd3dd3da54ca : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8376c81f3b39457cb38fdb9f369b73f6 : FAILURE in 43m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e1e3ff81e7994b2cba97f1ed01875a72 : FAILURE in 1h 48m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/22d7ac805f7c4f0789d707bb90187b5f : FAILURE in 23m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9364524968a425f8175bb2f6f09e89f : FAILURE in 52m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d213163a74fa4482b3c9a5566ae0b995 : SUCCESS in 1h 02m 01s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/806f22ea78a2472dbceb49f030b43312 : FAILURE in 32m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/915ceb6e4f2e441fb5d61eed011c10d9 : FAILURE in 1h 59m 46s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/ac213a2f19bb44ebbfffbd5dddf65067 : FAILURE in 25m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99a6fe15733a40a1b8ad654b520e4522 : SUCCESS in 54m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f1f2aec4358a4d56a8270b8301b16c57 : SUCCESS in 1h 04m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/10b830b5e1b04b26baefe02cb305171f : SUCCESS in 31m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f0983d25c9594aea9a177ed532f21f1e : SUCCESS in 1h 12m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6b179edfea7240b7b3a0a4383a404741 : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":7},{"id":"c38e22c67f49f9cdbdece3002fa728f1eeb5b362","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-11-25 01:24:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1316b4a428acb3b474ff2cb64c02869a5923b99b","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-11-26 08:09:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1ca5f8910fdfa1ff8e14675453fd97844f59d86b","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2021-11-26 08:41:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b0d083c7920ddeaaa477fc325667d19bcb2f799f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-14 19:38:14.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":7},{"id":"5911fd3d8aedbb470ec34288c8379ab9f3d6f020","tag":"autogenerated:gerrit:restore","author":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"date":"2022-10-18 03:45:45.000000000","message":"Restored","accounts_in_message":[],"_revision_number":7},{"id":"c2d4c302ed7a9c84147895150ab2b5864a6940c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 03:46:04.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 805930,7","accounts_in_message":[],"_revision_number":7},{"id":"3495b40b1910a8337d92dbaf79597d706fb99e0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 03:46:05.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 805930,7","accounts_in_message":[],"_revision_number":7},{"id":"0e7c58c72fa9f427a3c599e0d427cce0487c4106","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-12-16 08:20:53.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d93d0d1ea089e93ecb90222c28057d81e1a01b63","revisions":{"dd5d1ce324a4ef59ca05151266433a3003a3c83e":{"kind":"REWORK","_number":1,"created":"2021-08-25 02:55:22.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/30/805930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/805930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/805930/1"}}},"commit":{"parents":[{"commit":"3d0a4c26ac023879ca9a30f42ea3377cd2f92167","subject":"Merge \"Add limitation about QoS min bw for PF less SRIOV platforms\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d0a4c26ac023879ca9a30f42ea3377cd2f92167"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-08-25 02:36:23.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-08-25 02:43:06.000000000","tz":480},"subject":"Use Openflow to implement DVR L3","message":"Use Openflow to implement DVR L3\n\nThis commit is to implement the below spec:\n\nhttps://review.opendev.org/c/openstack/neutron-specs/+/796746\n\nIt implemented the below features:\n\n - Openflow-based ARP reply.\n - Openflow-based ICMPv4 reply for qrouter port.\n - Openflow-based routing between VMs.\n - ARP request for remote FIP or FIP gateway triggered by packet_in2.\n - Enhance ovs_neutron_agent to handle advanced packet_in2\n - Openflow-based FIP (without iptables/conntrack involved)\n - Openflow-based FIP QoS (used OVS meter action)\n - Openflow-based snat\n\nRelated-bug: #1931953\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\nChange-Id: Id6b1654768b20080cc5894cc27eaccdf2fa25fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd5d1ce324a4ef59ca05151266433a3003a3c83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd5d1ce324a4ef59ca05151266433a3003a3c83e"}]},"branch":"refs/heads/master"},"0c9010a1423e7832cb352e5693e8cf5cd9872803":{"kind":"REWORK","_number":2,"created":"2021-08-25 09:26:46.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/30/805930/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/805930/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/805930/2"}}},"commit":{"parents":[{"commit":"3d0a4c26ac023879ca9a30f42ea3377cd2f92167","subject":"Merge \"Add limitation about QoS min bw for PF less SRIOV platforms\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d0a4c26ac023879ca9a30f42ea3377cd2f92167"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-08-25 02:36:23.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-08-25 09:24:38.000000000","tz":480},"subject":"Use Openflow to implement DVR L3","message":"Use Openflow to implement DVR L3\n\nThis commit is to implement the below spec:\n\nhttps://review.opendev.org/c/openstack/neutron-specs/+/796746\n\nIt implemented the below features:\n\n - Openflow-based ARP reply.\n - Openflow-based ICMPv4 reply for qrouter port.\n - Openflow-based routing between VMs.\n - ARP request for remote FIP or FIP gateway triggered by packet_in2.\n - Enhance ovs_neutron_agent to handle advanced packet_in2\n - Openflow-based FIP (without iptables/conntrack involved)\n - Openflow-based FIP QoS (used OVS meter action)\n - Openflow-based snat\n\nRelated-bug: #1931953\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\nChange-Id: Id6b1654768b20080cc5894cc27eaccdf2fa25fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c9010a1423e7832cb352e5693e8cf5cd9872803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c9010a1423e7832cb352e5693e8cf5cd9872803"}]},"branch":"refs/heads/master"},"b1ef869f58f723762798129fe72a3fe566016675":{"kind":"REWORK","_number":3,"created":"2021-09-06 01:01:13.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/30/805930/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/805930/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/805930/3"}}},"commit":{"parents":[{"commit":"1d2a4cb35a6dc0a108325f4e5bee126d8e2c5b3d","subject":"Merge \"Use lookup() instead of iterating over table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d2a4cb35a6dc0a108325f4e5bee126d8e2c5b3d"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-08-25 02:36:23.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-09-06 01:01:24.000000000","tz":480},"subject":"Use Openflow to implement DVR L3","message":"Use Openflow to implement DVR L3\n\nThis commit is to implement the below spec:\n\nhttps://review.opendev.org/c/openstack/neutron-specs/+/796746\n\nIt implemented the below features:\n\n - Openflow-based ARP reply.\n - Openflow-based ICMPv4 reply for qrouter port.\n - Openflow-based routing between VMs.\n - ARP request for remote FIP or FIP gateway triggered by packet_in2.\n - Enhance ovs_neutron_agent to handle advanced packet_in2\n - Openflow-based FIP (without iptables/conntrack involved)\n - Openflow-based FIP QoS (used OVS meter action)\n - Openflow-based snat\n\nRelated-bug: #1931953\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\nChange-Id: Id6b1654768b20080cc5894cc27eaccdf2fa25fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1ef869f58f723762798129fe72a3fe566016675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1ef869f58f723762798129fe72a3fe566016675"}]},"branch":"refs/heads/master"},"115dec5b3c8952fdd4b054c4e5f6beb3111b2608":{"kind":"REWORK","_number":4,"created":"2021-09-06 04:14:48.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/30/805930/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/805930/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/805930/4"}}},"commit":{"parents":[{"commit":"1d2a4cb35a6dc0a108325f4e5bee126d8e2c5b3d","subject":"Merge \"Use lookup() instead of iterating over table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d2a4cb35a6dc0a108325f4e5bee126d8e2c5b3d"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-08-25 02:36:23.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-09-06 04:15:11.000000000","tz":480},"subject":"Use Openflow to implement DVR L3","message":"Use Openflow to implement DVR L3\n\nThis commit is to implement the below spec:\n\nhttps://review.opendev.org/c/openstack/neutron-specs/+/796746\n\nIt implemented the below features:\n\n - Openflow-based ARP reply.\n - Openflow-based ICMPv4 reply for qrouter port.\n - Openflow-based routing between VMs.\n - ARP request for remote FIP or FIP gateway triggered by packet_in2.\n - Enhance ovs_neutron_agent to handle advanced packet_in2\n - Openflow-based FIP (without iptables/conntrack involved)\n - Openflow-based FIP QoS (used OVS meter action)\n - Openflow-based snat\n\nRelated-bug: #1931953\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\nChange-Id: Id6b1654768b20080cc5894cc27eaccdf2fa25fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/115dec5b3c8952fdd4b054c4e5f6beb3111b2608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/115dec5b3c8952fdd4b054c4e5f6beb3111b2608"}]},"branch":"refs/heads/master"},"0aae68081a958cf4bfa8446b8e80b166ad5fd840":{"kind":"REWORK","_number":5,"created":"2021-09-06 08:46:20.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/30/805930/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/805930/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/805930/5"}}},"commit":{"parents":[{"commit":"1d2a4cb35a6dc0a108325f4e5bee126d8e2c5b3d","subject":"Merge \"Use lookup() instead of iterating over table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d2a4cb35a6dc0a108325f4e5bee126d8e2c5b3d"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-08-25 02:36:23.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-09-06 08:46:14.000000000","tz":480},"subject":"Use Openflow to implement DVR L3","message":"Use Openflow to implement DVR L3\n\nThis commit is to implement the below spec:\n\nhttps://review.opendev.org/c/openstack/neutron-specs/+/796746\n\nIt implemented the below features:\n\n - Openflow-based ARP reply.\n - Openflow-based ICMPv4 reply for qrouter port.\n - Openflow-based routing between VMs.\n - ARP request for remote FIP or FIP gateway triggered by packet_in2.\n - Enhance ovs_neutron_agent to handle advanced packet_in2\n - Openflow-based FIP (without iptables/conntrack involved)\n - Openflow-based FIP QoS (used OVS meter action)\n - Openflow-based snat\n\nRelated-bug: #1931953\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\nChange-Id: Id6b1654768b20080cc5894cc27eaccdf2fa25fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aae68081a958cf4bfa8446b8e80b166ad5fd840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aae68081a958cf4bfa8446b8e80b166ad5fd840"}]},"branch":"refs/heads/master"},"d77e58c36eb1f812a8938e32655b06fc04552d76":{"kind":"REWORK","_number":6,"created":"2021-09-17 09:19:52.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/30/805930/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/805930/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/805930/6"}}},"commit":{"parents":[{"commit":"b52e5ecae2556ae9c531d491a30539d7ad984aa6","subject":"Merge \"Update master for stable/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b52e5ecae2556ae9c531d491a30539d7ad984aa6"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-09-17 09:06:07.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-09-17 09:06:07.000000000","tz":480},"subject":"Use Openflow to implement DVR L3","message":"Use Openflow to implement DVR L3\n\nThis commit is to implement the below spec:\n\nhttps://review.opendev.org/c/openstack/neutron-specs/+/796746\n\nIt implemented the below features:\n\n - Openflow-based ARP reply.\n - Openflow-based ICMPv4 reply for qrouter port.\n - Openflow-based routing between VMs.\n - ARP request for remote FIP or FIP gateway triggered by packet_in2.\n - Enhance ovs_neutron_agent to handle advanced packet_in2\n - Openflow-based FIP (without iptables/conntrack involved)\n - Openflow-based FIP QoS (used OVS meter action)\n - Openflow-based snat\n\nRelated-bug: #1931953\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\nSigned-off-by: Yulin Chen \u003cchenyulin@inspur.com\u003e\nCo-authored-by: Yulin Chen \u003cchenyulin@inspur.com\u003e\nChange-Id: Id6b1654768b20080cc5894cc27eaccdf2fa25fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77e58c36eb1f812a8938e32655b06fc04552d76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d77e58c36eb1f812a8938e32655b06fc04552d76"}]},"branch":"refs/heads/master"},"d93d0d1ea089e93ecb90222c28057d81e1a01b63":{"kind":"REWORK","_number":7,"created":"2021-09-29 08:31:59.000000000","uploader":{"_account_id":30454,"name":"Yi Yang","email":"yangyi01@inspur.com","username":"yangyi01"},"ref":"refs/changes/30/805930/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/805930/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/805930/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/805930/7"}}},"commit":{"parents":[{"commit":"7c87ecb679928efffe4985b46c2305e638afcbe3","subject":"Merge \"Replace cirros 0.4.0 by 0.5.2 in ovn migration create-resources.sh.j2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c87ecb679928efffe4985b46c2305e638afcbe3"}]}],"author":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-09-29 08:31:23.000000000","tz":480},"committer":{"name":"Yi Yang","email":"yangyi01@inspur.com","date":"2021-09-29 08:31:23.000000000","tz":480},"subject":"Use Openflow to implement DVR L3","message":"Use Openflow to implement DVR L3\n\nThis commit is to implement the below spec:\n\nhttps://review.opendev.org/c/openstack/neutron-specs/+/796746\n\nIt implemented the below features:\n\n - Openflow-based ARP reply.\n - Openflow-based ICMPv4 reply for qrouter port.\n - Openflow-based routing between VMs.\n - ARP request for remote FIP or FIP gateway triggered by packet_in2.\n - Enhance ovs_neutron_agent to handle advanced packet_in2\n - Openflow-based FIP (without iptables/conntrack involved)\n - Openflow-based FIP QoS (used OVS meter action)\n - Openflow-based snat\n - Openflow-based ECMP routing\n - unit tests\n\nRelated-bug: #1931953\nSigned-off-by: Yi Yang \u003cyangyi01@inspur.com\u003e\nSigned-off-by: Yulin Chen \u003cchenyulin@inspur.com\u003e\nCo-authored-by: Yulin Chen \u003cchenyulin@inspur.com\u003e\nSigned-off-by: XiaoYu Zhu \u003cz503755743@gmail.com\u003e\nCo-authored-by: XiaoYu Zhu \u003cz503755743@gmail.com\u003e\nChange-Id: Id6b1654768b20080cc5894cc27eaccdf2fa25fa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d93d0d1ea089e93ecb90222c28057d81e1a01b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d93d0d1ea089e93ecb90222c28057d81e1a01b63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
