)]}'
{"id":"openstack%2Fneutron~806568","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Ief402048d99d40b64d81fcf58eb2e39b1ba7ebbb","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1939432","hashtags":[],"change_id":"Ief402048d99d40b64d81fcf58eb2e39b1ba7ebbb","subject":"Do not fail if the agent load is not bumped","status":"MERGED","created":"2021-08-30 13:05:12.000000000","updated":"2021-09-02 16:19:01.000000000","submitted":"2021-09-02 16:15:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806568-bug/1939432","meta_rev_id":"12572be0970d7f17d9adf99d24db765d529d7703","_number":806568,"virtual_id_number":806568,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-02 16:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-09-02 08:44:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-09-02 14:35:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-09-02 14:35:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 13:56:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-30 15:02:14.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-30 15:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 08:44:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-09-02 14:35:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a8fd26a6ba5af4a0a3c00b872c406b3f18b279cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-30 13:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a641a227ff8b18df424fe74ba0e87201fcc6e428","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 13:56:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46fc55d725d44818bc56079db1dbc315 : SUCCESS in 48m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c55005730ddb4d039853872646bebd47 : SUCCESS in 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"803190af1df99ffe142e9e6e11399fb0b69d0724","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-30 15:02:14.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/806568/1 : FAILURE in 1h 52m 38s","accounts_in_message":[],"_revision_number":1},{"id":"59ba00706981dd15613e040ce59c6690286868e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 15:13:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/ae678c1bcd99471e9270171a97f22a5d : SUCCESS in 49m 09s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/dd52b71542104550940fdd6e36c9b250 : SUCCESS in 1h 49m 49s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/c967c483e8a74ec68619d13909634504 : SUCCESS in 1h 47m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/a52c39dd9aee4dd79ce7dd103d80add2 : FAILURE in 2h 06m 45s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/20b86a79a48241d291f67455556e71d4 : SUCCESS in 1h 53m 59s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/7867c862d8e54599865c0aab82682d7e : SUCCESS in 34m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf86bf6597c94c249cc75f7aa12ffcc9 : SUCCESS in 43m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a365111402847fa88457647ac62f8f0 : SUCCESS in 12m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e317b3452cf4579848f5763f4a9e9b9 : SUCCESS in 37m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b017c1020f004f14960bd40d1c391600 : SUCCESS in 22m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68e01cd0e86444d2927fe393c2df56de : SUCCESS in 35m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4223711d8dfd4ffa8f753dfcacb91dca : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8c9fc7aac8543c9b080e99fc9c7efec : SUCCESS in 50m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b79f01c6f334a819b146bc75141d331 : SUCCESS in 1h 22m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7baef3d185c64e83828a03babee632d0 : SUCCESS in 1h 16m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b10ad0a204a64e8cba79f064d6921ad8 : SUCCESS in 59m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/90e243a6b61d4804ad4d3b4847dc7b09 : SUCCESS in 1h 02m 44s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/33852afc470946ebb25fb2e97e05be1b : SUCCESS in 1h 12m 49s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/54417f73cb614ec893cc46a74a2ed3fa : SUCCESS in 1h 32m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c12d892b20445e2af9b8c057729e819 : SUCCESS in 1h 32m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4379d58f92be4161af5e8e3edb9d0c00 : SUCCESS in 1h 07m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4e03221ff84f4aa59b602ea0d4acfec2 : SUCCESS in 1h 00m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c15ce90b00164ff49425d3cc6292645a : SUCCESS in 33m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3e0b7982a1e5481e999b6d167dd64b56 : SUCCESS in 1h 17m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fb3ffde1bedc42a8ad12b58eaf3e772a : SUCCESS in 1h 18m 19s","accounts_in_message":[],"_revision_number":1},{"id":"362d3759500a1635ab962394a9113b5ffe15fc81","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-31 12:34:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62b68cb118d08faf61c92e74e5ce5b6318c9233b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 13:26:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c56df697ad4f47f2b69893029e2d1e09 : SUCCESS in 41m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0c6ccd1d4aa4afaa36456747324e762 : SUCCESS in 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abb0f142a6118e27c73e8621771ccb8402fe41cc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-31 14:24:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/806568/1 : SUCCESS in 1h 47m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b31abe3fd74d4009795ebf4304948101ffc057fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 15:05:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/100e1365af334bd4a79f1287e8d805f2 : SUCCESS in 51m 00s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/f1a7fae228b646f6880a9787ed840a60 : SUCCESS in 1h 52m 05s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/dae44cdfafb44d49bca26b9bd424fef2 : SUCCESS in 1h 44m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/0bf5b0fa7234408f92df4457b7ca0fac : SUCCESS in 1h 47m 41s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/b8bafca7baa14d1199e213a06c635897 : SUCCESS in 2h 26m 31s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/d316aa8c41724d7b963d0293c3591439 : SUCCESS in 34m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ebd82b8213d4957bfe32505c454b120 : SUCCESS in 38m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d4b79a57ba4a27a75f35c88ce51d88 : SUCCESS in 12m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ee3857e3d7d4090a9a071a0a8830d83 : SUCCESS in 24m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b016e5cc70de4d8f8aaaec96b064026d : SUCCESS in 27m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb9aef9fa80f4ee183a58bc70318be18 : SUCCESS in 32m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af2d05343bb4caea31b3cf908a5678d : SUCCESS in 10m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49840e253078491db7b4566d4ccf174c : SUCCESS in 50m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f3136f3a5b34d3692fc5288d34d4a63 : SUCCESS in 1h 26m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ae68bbd55ac444f5820a02f8eb3052df : SUCCESS in 1h 19m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98d8087f690941fabc2d31cf1c298058 : SUCCESS in 1h 02m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e3f40cf44fa043748b7e512fa3451b75 : SUCCESS in 1h 02m 15s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6f1fb04424e14e089a93c97daa159e01 : SUCCESS in 1h 15m 47s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d532a1c830784141840490b0c0481a13 : SUCCESS in 1h 27m 22s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/96e82d322cac429190c38d94755179c9 : SUCCESS in 1h 46m 42s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/993bf58b150046f3a53b2ea760f96a14 : SUCCESS in 1h 02m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ac90e3ed5e774b5e95b546e24fa67247 : SUCCESS in 1h 02m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/38d48bce860140cbac6e052d8398539e : SUCCESS in 36m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/22da96a9526f441bbd12e9aee4523def : SUCCESS in 1h 21m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9e3dc260963b40428ef0c583dd4d4fae : SUCCESS in 1h 29m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d27ade5fd103b0eb2368469b1edf74a6e50d3ff3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-02 08:44:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"802f925b047ddc812ab9e25571760f025c9112f4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-02 14:35:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e968c98b389749f24c45ebaefa500524a2ba89b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 14:36:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d4532fb0b5f954969d0f540006ea16fc4813725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:15:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/4a5e04b920ce4a8faa5ae801cc3f4031 : SUCCESS in 52m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/028205bd3dab41029ca6b18f164f32da : SUCCESS in 17m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97b0a95f462b44bbadc29f3477d4fe69 : SUCCESS in 29m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4344bd04f8a74303be3a8afdde9888b0 : SUCCESS in 29m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f57952a16f444f9dc302fd486d2361 : SUCCESS in 11m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/24a8a10d76fb4d69827605478849f456 : SUCCESS in 46m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e34d632bb8a24ffd844c5948824511e4 : SUCCESS in 1h 21m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/622a6a4669564d5e86e3b57899e9af56 : SUCCESS in 1h 20m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba8f82fcaca74ad8b8c3eb704e103ea1 : SUCCESS in 1h 03m 27s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af67b9234e8142d9be8a53eb30accd9f : SUCCESS in 1h 34m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1128430ba7a747779246ed9b7f075ad5 : SUCCESS in 1h 07m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/42ffe1f1cb2c4656a36cd5817082e617 : SUCCESS in 1h 19m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dafaa890a82c4deb8fd4a418bd9209dc : SUCCESS in 37m 38s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0750cb28568245f083e300e59fd175e0 : SUCCESS in 1h 30m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ede2f445013a056d49b4ee76e5adfb9b3c76dfc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:15:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"12572be0970d7f17d9adf99d24db765d529d7703","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:19:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b10b2303a64fac881e0524eb83249b : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"816aca60b90b89038863d6974b5a9e0ee8983424","revisions":{"816aca60b90b89038863d6974b5a9e0ee8983424":{"kind":"REWORK","_number":1,"created":"2021-08-30 13:05:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/68/806568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/806568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/806568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/806568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/806568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/806568/1"}}},"commit":{"parents":[{"commit":"219ef6bc6b8f3b10d7d9428cb51717b44398aae7","subject":"Merge \"Follow up for replacing assertItemsEqual\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/219ef6bc6b8f3b10d7d9428cb51717b44398aae7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-11 09:13:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-30 13:04:14.000000000","tz":0},"subject":"Do not fail if the agent load is not bumped","message":"Do not fail if the agent load is not bumped\n\nWhen a new network and its first subnet is created, the DHCP agent\nbumps the \"load\" parameter to reflect the number of networks handled.\nThis \"load\" parameter is modified when:\n- As commented, when the first subnet of a network is created. The\n  \"load\" value is bumped.\n- When periodically the DHCP agent sends the status, informing about\n  the current number of networks handled.\n\nIf during the subnet creation this \"load\" value is not updated, it will\nbe in the next periodic update of the agent.\n\nThis \"load\" value is used by the scheduler to equally distribute the\nobjects to be managed by any agent type (DHCP agents manage networks).\n\nThe bug refers to DHCP but is valid for any other agent.\n\nConflicts:\n      neutron/common/utils.py\n\nChange-Id: Ief402048d99d40b64d81fcf58eb2e39b1ba7ebbb\nCloses-Bug: #1939432\n(cherry picked from commit 668b1cc652f076e555ef1fc1289684367159186a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/816aca60b90b89038863d6974b5a9e0ee8983424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/816aca60b90b89038863d6974b5a9e0ee8983424"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
