)]}'
{"id":"openstack%2Fneutron~806571","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ief402048d99d40b64d81fcf58eb2e39b1ba7ebbb","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1939432","hashtags":[],"change_id":"Ief402048d99d40b64d81fcf58eb2e39b1ba7ebbb","subject":"Do not fail if the agent load is not bumped","status":"MERGED","created":"2021-08-30 13:10:30.000000000","updated":"2021-09-02 16:30:03.000000000","submitted":"2021-09-02 16:27:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806571-bug/1939432","meta_rev_id":"3227a2815ed7b28b2c433475f1b1d3389f426f8e","_number":806571,"virtual_id_number":806571,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-02 16:27:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-09-02 08:44:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-09-02 14:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-09-02 14:36:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 15:18:32.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-08-30 16:55:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 08:44:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-09-02 14:36:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ca1a0450a1dbf8153178e81aaa33231f135e1621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-30 13:10:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f0a509860b1d227b93b066f78f98a685b42df48","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-30 15:18:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/806571/1 : SUCCESS in 2h 04m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f4435c6f8089cbb18e5340ed8ad3fb06be91a6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 16:55:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/f6932b1a561d4f4b993a0cbdd4596468 : SUCCESS in 51m 20s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/66fee7d1643846f399125a67e18ab147 : SUCCESS in 2h 08m 58s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/2ba40587e9ba4922b2f8df0ef00e1280 : SUCCESS in 2h 00m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/53beb013365f44729e15554eccb0984f : SUCCESS in 1h 50m 53s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/de4a696052d044c88b9c43e176dff5fe : TIMED_OUT in 2h 52m 01s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/63774f6eccb4460694468b349bc1bcd2 : SUCCESS in 34m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcf052ad67fb48cdb9230751691bfbd1 : SUCCESS in 43m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750affdb5fa84f82972481d042acc9f5 : SUCCESS in 14m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b43d55cabd464346b838076dfe8b0c3f : SUCCESS in 33m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58ac5732c46a4cea89ec3e52a3704915 : SUCCESS in 28m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cd5c0bce3574ca4855dc23dc344b43d : SUCCESS in 25m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/320ccf5c9c3e49eeaf1656c7c82009d0 : SUCCESS in 9m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a23ea5a557e74b05902e9b7971eff936 : SUCCESS in 51m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/10e005255e334d2a9f86d0a2544fae90 : SUCCESS in 1h 17m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c5a95cbca1e409b81af7cc2278ca0b1 : SUCCESS in 1h 19m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5738cbff897e45bc8ee20f217aac8b16 : SUCCESS in 1h 32m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/131bf189a276452daec10aa371f1c647 : SUCCESS in 1h 20m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e9d9fa73d250484f98fea9ec9e169509 : SUCCESS in 1h 14m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2b63519502f9423cae86c6c6ce226767 : SUCCESS in 1h 08m 18s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/94a6ccb7e29943e9af72cef889e71fa4 : SUCCESS in 1h 02m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/687315860f9f415baff5b9e91560731f : SUCCESS in 1h 15m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7886bd1deb8b4e6b9f170f44ba8491fd : SUCCESS in 1h 42m 44s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/24b394247c014b72853f918a6cbcc38b : SUCCESS in 1h 24m 57s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04bd587b51674c4a9ca73a7a8ef2f0be : SUCCESS in 1h 09m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/41c2d64e6a364cdeb0ba4f8d63189ea8 : SUCCESS in 1h 11m 32s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8d9cc53dc38d4b5ebac0a0084333158e : SUCCESS in 1h 22m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/745806ecf8f440e5b65e99fac29f5e3b : SUCCESS in 1h 20m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/82faa3d9fa4645da8a9b81fae96d2a02 : SUCCESS in 1h 24m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d80a69313d8e4efe9fcb742f96880a72 : SUCCESS in 41m 23s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e21b6f0b08d749a7af3a61b72c9d17ac : SUCCESS in 1h 18m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/123a14c2d2234b6c807a3a409c93f713 : SUCCESS in 55m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2aa70f3ce50441e89444510e0d149ac0 : SUCCESS in 3h 42m 28s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e2bf8f68b0c349639712953864f8d415 : SUCCESS in 1h 05m 33s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e4dae8ced92d4584b65be400ed909550 : SUCCESS in 1h 26m 42s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fcb6fed843714144b60142c45cc7624e : SUCCESS in 2h 22m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7889b5aa44574e33b1675470d491499e : FAILURE in 1h 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea13b2b98aa8c479906840501960c3f3fb1445f0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-31 12:33:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d27a71537cda683cbf02e2d7aa82f5ec856285ff","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-31 14:51:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/806571/1 : SUCCESS in 2h 17m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c03090f5261ecaa10c05f02de049bc666049aa47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 16:03:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/0ad5a6adf87d472ab060546f5df46148 : SUCCESS in 49m 54s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/6714c5182d2c4cdb9934942ee959ec15 : TIMED_OUT in 2h 53m 13s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/82c3adb8af744e3f8fac16524411c001 : SUCCESS in 2h 24m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/782334a9b5d5460eb58f9d3653a248bb : SUCCESS in 1h 20m 29s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/feff64c7908e4bb38a29ddb680f6fac1 : SUCCESS in 2h 11m 07s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/68ebab3a9aa842da8567314857135709 : SUCCESS in 34m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92f94556bd1742038d7b28f86993ac25 : SUCCESS in 28m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd9890bf63046aaaf584ae26b9014f3 : SUCCESS in 16m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ddd0aa5b2bd4a31a82ccd9b3e5f4c4d : SUCCESS in 28m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35ddbcbcdfee40c88c7a5af744871f7f : SUCCESS in 28m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/178ae5089e5d4f9fa13a8a7b9054a3df : SUCCESS in 26m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed043ee548fc43219eaed426b025e623 : SUCCESS in 11m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fad56627c8044d20b3fd87c4ef23cc36 : SUCCESS in 54m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/382c4bcc160f46b8874a08794462441c : SUCCESS in 1h 21m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e8cac51a5c14da59aaa64f1c9c4946e : SUCCESS in 1h 20m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ad482c95d11f498a96b38faa8c239feb : SUCCESS in 1h 51m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bdb72bfee3f49578cb1974aad62e64b : SUCCESS in 1h 13m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7e2ad01844704f87812620d0703a48a5 : SUCCESS in 1h 09m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d9287f9eca7244d9b0b81be7174398ea : SUCCESS in 1h 10m 30s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/58bc94f30726458caf845c5cfb5889b1 : SUCCESS in 57m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3145d700ff914e1fb589bd2178d30d7d : SUCCESS in 1h 03m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9678ba9a6d1f4fd4835db432b28216f0 : SUCCESS in 1h 34m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a989752058eb4237b16bd80ae6c9ce6d : SUCCESS in 1h 27m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6fef6e0df20f47e5af2335f16efb6d90 : SUCCESS in 1h 14m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/86608d3630bc4b09bd5ad6669781c020 : SUCCESS in 1h 10m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f0783f4a77114e99ad18f79afc257a36 : SUCCESS in 1h 39m 43s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15f817f5f8af48b6978cfe35fdef2c59 : SUCCESS in 1h 05m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2761e880c7ba4b6887665f49034fb14f : SUCCESS in 1h 19m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3238ad24d0754ca1ac73ab5e99756b5d : SUCCESS in 41m 24s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ae3e8f47215140c9a74a0527bf9d707b : SUCCESS in 1h 14m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc2270073a4b4d60aa22ef3f6d950144 : SUCCESS in 56m 43s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/23a6a2be82944d3f907bc44aec362113 : SUCCESS in 3h 28m 26s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/827abb9656a0442da612af3cec2bde76 : SUCCESS in 1h 18m 52s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8e70d343a2044f6f8a982abb6b759d53 : SUCCESS in 1h 40m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/aefd1ec63b8f43108821c4c07fa4d13d : SUCCESS in 2h 00m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a9e104217252490885384029e8b920c6 : FAILURE in 1h 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9af58c46284a186eda86aaa073ee1b526b64617","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-31 16:04:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"250a3590e0fe79b9af65873e6026615d0d0a949f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-08-31 18:26:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/806571/1 : SUCCESS in 2h 19m 27s","accounts_in_message":[],"_revision_number":1},{"id":"02bf80da7eefa761cc28cc66c3bbab75395f87be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 21:18:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/1269a24fe9d940708c1438d320d51481 : SUCCESS in 55m 42s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/a971c79abb3a4c03ba378023fe90f8b9 : SUCCESS in 2h 11m 55s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/53093ddabd884eccb5ceb5518efb68b0 : SUCCESS in 2h 14m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/2277d19d729e4a669c716a916e7b8ac3 : SUCCESS in 1h 48m 12s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/3e11267052ac45adb96959fd977dd2c3 : SUCCESS in 1h 56m 06s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/7dbf9e2a479a48aca5418061f0922ee4 : SUCCESS in 36m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce36c11738b94e318ca42330ee7d2b0e : SUCCESS in 30m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc318fec0f84a3da359a7ae8b567ee8 : SUCCESS in 17m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31b06cca00634fe3b0c00ef75e9d3dbb : SUCCESS in 31m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/979ac148f15b490e894c126451e821a1 : SUCCESS in 29m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31a8375694c34373a3643df44ecf58e7 : SUCCESS in 26m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26753e249b104844a01c09a08b5a0d80 : SUCCESS in 10m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/69d6caf91b634e1a8c5986d67f755236 : SUCCESS in 55m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3ce0d32f283c454f88d8812413bd05db : SUCCESS in 1h 23m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/05a869589a05402f903c16f07a273294 : SUCCESS in 1h 27m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3e9cf915b77944ceadb39e146049e130 : SUCCESS in 1h 27m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a69b4edfb474a24bb0d441f1d417aec : SUCCESS in 1h 15m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ba95a79a815426dbbefb7e67e176a6d : SUCCESS in 1h 17m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a8759f0f958c4275ae4f8acc85188307 : SUCCESS in 1h 12m 14s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d22c29c1af04486b3018f70c81da110 : SUCCESS in 1h 07m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f675895a10554aed856c0ecc51556a0f : SUCCESS in 1h 10m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c2bbcb7e177c42f4a4ccb44e74ea0cea : SUCCESS in 1h 36m 37s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1240686404b9433ab781a06ded36870d : SUCCESS in 1h 24m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2299c6275cd14e549073e21ebc045f0c : SUCCESS in 1h 10m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e9698d1d25b74e47812b203034d17f2e : SUCCESS in 1h 03m 48s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9164e512f0094659875f4a574ca08922 : SUCCESS in 1h 38m 20s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04a8ddabe2d544b188c78c6cce73d3b7 : SUCCESS in 1h 02m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de817b8ef2544a229d20a14c78f40dcd : SUCCESS in 1h 21m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ae6a0903bf9c4f18b5f7f0ae61c1d232 : SUCCESS in 47m 19s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/47f132b469e5400bb2beeaf64368254e : SUCCESS in 1h 19m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/51cd4a0bae234eb392d1fac690212af7 : SUCCESS in 56m 12s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/357567eb47bb4b13a87ad232db1abd07 : SUCCESS in 3h 41m 42s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/35bee69994514fa880f25d788ee476d6 : SUCCESS in 1h 25m 42s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a131ca1019f44bb98a2aab360cb9c2d4 : SUCCESS in 1h 40m 39s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f1291610510f439eb53bd9de29793c81 : SUCCESS in 2h 07m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1dbffd51895a48e1b408dace4f04d3ae : FAILURE in 1h 44m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1a09d6406f3dcd5b99cfaf70b8042eb8c83f8c6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-02 08:44:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a83b94ff381d1d262b4a2765689b888fa3c027ee","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-02 14:36:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"063026e4cefb1935ee5f38f01b02ca0849534baf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 14:36:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30f61f99ed6311984252b820a16f1db0fb411d0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:27:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/c79e76d0b7d84df0bf13fb8c55938a6d : SUCCESS in 48m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8d140d6c714d16b35f2d3601c26fa1 : SUCCESS in 14m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03a9848a2cf24aec830ba39338ddac3f : SUCCESS in 26m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/babe8962516b4d06a9d0ec3c437c1005 : SUCCESS in 31m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec292ebaef8e4ec9a105f6a47d26cf67 : SUCCESS in 10m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/18f4d68d18d6434ca402a00b842d99c8 : SUCCESS in 1h 00m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9c03a84656084030b2dd934f4f1e91f2 : SUCCESS in 1h 27m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3c134ae9c8454811ab54ceff7939074d : SUCCESS in 1h 14m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/30b347af35da4f079a6a39d00417e8b6 : SUCCESS in 1h 14m 45s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d17a01cf9ce240d8b3eb4831b77f2ce1 : SUCCESS in 1h 09m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/13bdb681f33648d0877acad76894f51d : SUCCESS in 1h 13m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/48f4e04c23764b6c87d5f570854022f0 : SUCCESS in 1h 29m 07s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88294a6e3f314fb283df765ca65bab14 : SUCCESS in 1h 42m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b37e10d0e75d43938a4007c8a63971df : SUCCESS in 1h 13m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c23dfcda302c445a99f02a05c0e41d5d : SUCCESS in 1h 19m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f66d4b6afa9af96d7e4ac279596564cbb29217c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:27:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3227a2815ed7b28b2c433475f1b1d3389f426f8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:30:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ecd71202e4142eb86d78b705477812a : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f315f85a7b0ead30877e19988db4e4f80fc960e9","revisions":{"f315f85a7b0ead30877e19988db4e4f80fc960e9":{"kind":"REWORK","_number":1,"created":"2021-08-30 13:10:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/71/806571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/806571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/806571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/806571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/806571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/806571/1"}}},"commit":{"parents":[{"commit":"8e581d31cd2863b821faef0f96344e1cb8bd5d38","subject":"Merge \"Trivial: check if ipv6 is available for IptablesManager\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e581d31cd2863b821faef0f96344e1cb8bd5d38"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-11 09:13:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-08-30 13:10:01.000000000","tz":0},"subject":"Do not fail if the agent load is not bumped","message":"Do not fail if the agent load is not bumped\n\nWhen a new network and its first subnet is created, the DHCP agent\nbumps the \"load\" parameter to reflect the number of networks handled.\nThis \"load\" parameter is modified when:\n- As commented, when the first subnet of a network is created. The\n  \"load\" value is bumped.\n- When periodically the DHCP agent sends the status, informing about\n  the current number of networks handled.\n\nIf during the subnet creation this \"load\" value is not updated, it will\nbe in the next periodic update of the agent.\n\nThis \"load\" value is used by the scheduler to equally distribute the\nobjects to be managed by any agent type (DHCP agents manage networks).\n\nThe bug refers to DHCP but is valid for any other agent.\n\nConflicts:\n      neutron/common/utils.py\n      neutron/scheduler/base_resource_filter.py\n\nChange-Id: Ief402048d99d40b64d81fcf58eb2e39b1ba7ebbb\nCloses-Bug: #1939432\n(cherry picked from commit 668b1cc652f076e555ef1fc1289684367159186a)\n(cherry picked from commit 816aca60b90b89038863d6974b5a9e0ee8983424)\n(cherry picked from commit 1eb6b8926a7a5ad442c5af6057c042999e2645f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f315f85a7b0ead30877e19988db4e4f80fc960e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f315f85a7b0ead30877e19988db4e4f80fc960e9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
