)]}'
{"id":"openstack%2Fneutron~806803","triplet_id":"openstack%2Fneutron~stable%2Ftrain~Icd1c9cb36536bfa5e86a2b31e03d9377e7b8896b","project":"openstack/neutron","branch":"stable/train","topic":"bug/1942251","attention_set":{},"removed_from_attention_set":{"32792":{"account":{"_account_id":32792,"name":"Alexey Pustovitin","display_name":"Alexey Pustovitin","email":"apustovitin@gmail.com","username":"apustovitin"},"last_update":"2021-10-14 16:13:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icd1c9cb36536bfa5e86a2b31e03d9377e7b8896b","subject":"patch l3 agent in Train bug 1942251","status":"ABANDONED","created":"2021-08-31 20:20:42.000000000","updated":"2021-10-14 16:13:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dd80bf4e8b2ed2fb18f7e95b09654f5480c1a8cb","_number":806803,"virtual_id_number":806803,"owner":{"_account_id":32792,"name":"Alexey Pustovitin","display_name":"Alexey Pustovitin","email":"apustovitin@gmail.com","username":"apustovitin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-31 22:17:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":-1,"date":"2021-09-02 11:10:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-31 22:00:23.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-31 22:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 11:10:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-10-14 16:13:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"4abb9a9b75e52e68a54ce20f9575c186d5f0faf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32792,"name":"Alexey Pustovitin","display_name":"Alexey Pustovitin","email":"apustovitin@gmail.com","username":"apustovitin"},"date":"2021-08-31 20:20:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b255a6fe0ef09906b19d7e42910fca2a30a810e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-31 22:00:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/03/806803/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/03/806803/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"4996e7e2d2226990c2721e9696c3d8c6307c0099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 22:17:57.000000000","message":"Patch Set 1: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/efd03806d4aa4db4a3d453654c52d92b : SUCCESS in 1h 50m 04s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/82f1d42de2324733ade9a44fb257b08a : SUCCESS in 1h 36m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/d2560e8aad0a40c88e999cb6bc219c1b : SUCCESS in 1h 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdf641cd11b7410a8a2a1885ef9bcdb4 : FAILURE in 23m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048083e639694056ad31adf865e5d616 : FAILURE in 10m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/405448b8daa34bbb8a9833bc71cc7627 : FAILURE in 23m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ed72c3bc264e25ac2c190ae4f41d9d : FAILURE in 21m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17bcb9556e864e4295c2973a6c2566bf : FAILURE in 22m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f11902c33a040cfa28413633dfe11f8 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/1175eed1c01948f0897bd80f5827c62c : SUCCESS in 58m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/36a55f7fb24a46019afd9ef9cd3915b2 : SUCCESS in 1h 10m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/114753e6047f4ba7bee2aae5ff1f0b40 : FAILURE in 36m 25s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b4f3b0fecf0a48328ca52557b3397231 : FAILURE in 43m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cb21594aae4a4676824b1e76c25f018b : SUCCESS in 1h 14m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fc9e3745e8d34180b5f1e4ba4b81be42 : SUCCESS in 1h 25m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8542614929e247d0a859e231345ae296 : SUCCESS in 1h 05m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/52cb46eca9cc4f99bc8a31f1a8a56073 : SUCCESS in 1h 07m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b02ca7cb5bea4ede8a9de1a7b168ec53 : SUCCESS in 58m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d0f1e8d5ad444692bc08007030500c84 : SUCCESS in 57m 24s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/646df700b1654f839728941e385e1c16 : SUCCESS in 1h 04m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7504da0e0d624bde820a06ac7eaa6f04 : SUCCESS in 55m 18s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/50a8a6c1d0804f4f9f94dcec89f1e898 : SUCCESS in 1h 27m 38s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/eb7c0374e20342c7a549ab2b47ea15aa : SUCCESS in 1h 54m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/67879ed282c243a08992d6cd2463027e : SUCCESS in 57m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e7b05703ce44298bbbdfba3ab2c5801 : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":1},{"id":"58cc10e0188c03cf44e510e4f5293c03d740cc96","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-02 11:10:30.000000000","message":"Patch Set 1: Code-Review-1\n\nbased on the lp discussion (https://bugs.launchpad.net/neutron/+bug/1942251/comments/4 ) this change is not necessary.\nFor next time please consider the stable policy on how to backport fixes:\nhttps://docs.openstack.org/project-team-guide/stable-branches.html#appropriate-fixes\n\nThe important part: always try to reproduce and fix on master, and backport it branch-by-branch (of course it is possible that it is not possible)","accounts_in_message":[],"_revision_number":1},{"id":"dd80bf4e8b2ed2fb18f7e95b09654f5480c1a8cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-14 16:13:37.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"595187baa751bb8405e701b421c10de0aa1bde10","revisions":{"595187baa751bb8405e701b421c10de0aa1bde10":{"kind":"REWORK","_number":1,"created":"2021-08-31 20:20:42.000000000","uploader":{"_account_id":32792,"name":"Alexey Pustovitin","display_name":"Alexey Pustovitin","email":"apustovitin@gmail.com","username":"apustovitin"},"ref":"refs/changes/03/806803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/806803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/806803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/806803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/806803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/806803/1"}}},"commit":{"parents":[{"commit":"9e3b4c6f1279f96604bbc708603e8a7386e6e705","subject":"Merge \"Add extra logs to the network update callback in L3 agent\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e3b4c6f1279f96604bbc708603e8a7386e6e705"}]}],"author":{"name":"apustovitin","email":"apustovitin@gmail.com","date":"2021-08-31 20:18:49.000000000","tz":180},"committer":{"name":"apustovitin","email":"apustovitin@gmail.com","date":"2021-08-31 20:18:49.000000000","tz":180},"subject":"patch l3 agent in Train bug 1942251","message":"patch l3 agent in Train bug 1942251\n\nChange-Id: Icd1c9cb36536bfa5e86a2b31e03d9377e7b8896b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/595187baa751bb8405e701b421c10de0aa1bde10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/595187baa751bb8405e701b421c10de0aa1bde10"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
