)]}'
{"id":"openstack%2Fneutron~807633","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Id3a529248d3984506f0166bdc32e334127a01b7b","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1942179-stable/wallaby-stable/victoria","hashtags":[],"change_id":"Id3a529248d3984506f0166bdc32e334127a01b7b","subject":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware","status":"MERGED","created":"2021-09-07 06:09:26.000000000","updated":"2021-09-07 14:37:20.000000000","submitted":"2021-09-07 14:33:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807633-bug/1942179-stable/wallaby-stable/victoria","cherry_pick_of_change":807632,"cherry_pick_of_patch_set":1,"meta_rev_id":"603ea21d1b25a4fbba62145e66b66eb65948d067","_number":807633,"virtual_id_number":807633,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-07 14:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-09-07 11:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-09-07 07:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-09-07 11:43:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 07:39:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-09-07 07:52:33.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-09-07 08:30:06.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-07 11:10:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 11:43:54.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"831cb937be0e8c0ae71eb8a24be50fb48ad35f90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-07 06:09:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"72a66799364974877de0817fd60291a57c0b9dae","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-07 07:39:00.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"9b2b3921b0ad9f13a8d7412819652e6bacead4cb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-07 07:52:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/807633/1 : SUCCESS in 1h 41m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d550601389a6b1a904783c1c9855b946fe209a81","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-07 08:30:06.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/33/807633/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/33/807633/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b27df5ba798e3f919e7019e5e5fd7d815536538b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 11:10:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/b1f61a22a1fc4cb0af41b31034abd23d : SUCCESS in 47m 34s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/e259cc729c07458daffa58733d734101 : SUCCESS in 2h 14m 05s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/1d9e9728944945fcb791a76b5c14290b : SUCCESS in 2h 02m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/0be64bcbca1a4ba0980a35a7c6d6238c : SUCCESS in 1h 37m 36s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/634fbd494b3e4f89a89f4954ded2e615 : SUCCESS in 1h 51m 08s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/7e7f0ac723a745ebbb78b584ee131d4a : SUCCESS in 35m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/477b7cafaad84b6c943ef817fa4bf43f : SUCCESS in 28m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb8cdbb90bc745f2b569ef6fe37ac1a6 : SUCCESS in 13m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8e3bb761aba48cdbe7dea01d1c215a2 : SUCCESS in 25m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63bce03b7bff468890d7d15eb69590bd : SUCCESS in 22m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8421784fdc7f48cd82f7c9bdcf65eee0 : SUCCESS in 11m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6501af60bceb4a049959913cf5a54e1d : SUCCESS in 1h 24m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ef460b1d0ee4015a8d0c1747f90df9c : SUCCESS in 1h 24m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bfb90ba2578646f1aeb8adb36157fe28 : SUCCESS in 1h 34m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c889bf838e64d25b0b9d656237e7258 : SUCCESS in 1h 04m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/85ed71d3c4894e63af511b0b5144e0d0 : SUCCESS in 1h 08m 21s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/ba294364269c4b30acb2256e966faf19 : SUCCESS in 1h 10m 00s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c8c8648535f741ccad91ac0176197193 : SUCCESS in 59m 43s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/eec7ad726352472eb057001eda39fba7 : SUCCESS in 57m 09s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/df2cdf9124a94c1da7ee84a7e55b6c8a : SUCCESS in 1h 16m 57s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/60b4b16d53b54abca23be787e70c83b2 : SUCCESS in 1h 30m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/629a14a26e1049cc92e9789691a00258 : SUCCESS in 1h 03m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/68dc3b70e1f74b04bb9ae6398c06ef2d : SUCCESS in 1h 13m 32s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75f2633a9ddc4d0b96e04f6a5b11567c : SUCCESS in 1h 42m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30ae98531e7d44f78ff18293f91ae8e5 : SUCCESS in 1h 10m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/13baa9289d664c84b37b329d4a205f2d : SUCCESS in 1h 33m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fae289bb8df844e69f278f4a37ba4ddb : FAILURE in 38m 13s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a5514284bd8c48f28c5501ae66310a8e : SUCCESS in 1h 29m 11s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1f58f2c179ce4b209bf1b74e2c621b8b : SUCCESS in 3h 53m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5b53a46a2062421a9c5118f2d2ece2f7 : SUCCESS in 1h 21m 37s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a5e0d7ecb9ff4cd0af3bfd11f69e0d1a : SUCCESS in 1h 32m 08s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0feec541331a4e9ebebd073668be6add : SUCCESS in 1h 53m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/007f2c7849254c1a938fb7c0c01bb5ef : FAILURE in 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b31e781865196ddb404cd5f8547c78baaa9a47b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-07 11:43:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9970186aa0a8da806f0847b44353869892d963e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 11:44:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ecdc323add281361fa5500fd9b745c4882e01be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 14:33:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/5d05646c4e38402c8cc2efa3cfe231fb : SUCCESS in 45m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f262a3111c44de9abc8416585ab9d0 : SUCCESS in 13m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99d29d1e1898415896b3e9cc6f70b3dd : SUCCESS in 22m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8f093067336478782d17fe56fccdbbf : SUCCESS in 23m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e74553aa9d548af9cfd5d35e623ea7d : SUCCESS in 11m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1f35edf3363484ca04fa91fb737cd34 : SUCCESS in 1h 18m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd98e0f73ded4d6c986af690573caf36 : SUCCESS in 1h 14m 51s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8b9c7f686134bf5a8e8be029ec28625 : SUCCESS in 1h 00m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dac09492edd74a80a0db56a730a5b040 : SUCCESS in 1h 03m 46s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/379525031b8c4e0486cd8b1586bfc819 : SUCCESS in 57m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ec7c30491ce246b3843411982433888b : SUCCESS in 1h 09m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/74047a0c85124f1e838d18d02e677ff1 : SUCCESS in 1h 07m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ea1eed8f67854890aef0daac66b6a039 : SUCCESS in 1h 01m 36s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3e178b4ea4a41288796a1bb0f95f4f1 : SUCCESS in 1h 37m 22s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28afe201e34240e6b303c166a8da55e7 : SUCCESS in 1h 06m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/21374d59ece046c0a0148565a4659186 : SUCCESS in 1h 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7b7eba1ac80726eca14e8e2cc546a9bc24f6975d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 14:33:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"603ea21d1b25a4fbba62145e66b66eb65948d067","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 14:37:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0675cce9c5da4b0aaa688857f39943f7 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d961731a73128125ebf03fae52b4b34f3e7abf27","revisions":{"d961731a73128125ebf03fae52b4b34f3e7abf27":{"kind":"REWORK","_number":1,"created":"2021-09-07 06:09:26.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/33/807633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/807633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/807633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/807633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/807633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/807633/1"}}},"commit":{"parents":[{"commit":"6c74196dd638e328a243e08dae61090b25fa0b57","subject":"Merge \"Remove dhcp_extra_opt value after first newline character\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c74196dd638e328a243e08dae61090b25fa0b57"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-09-03 14:04:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-09-07 06:09:26.000000000","tz":0},"subject":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware","message":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware\n\nIt seems that using default singleton\u003dTrue in the\nroutes.middleware.RoutesMiddleware which is leading to use thread-local\nRequestConfig singleton object is not working well with eventlet\nmonkeypatching of threading library which we are doing in Neutron.\nAs a result it leaks memory in neutron-api workers every time when API\nrequest to not existing API endpoint is made by user.\n\nTo avoid that memory leak, let\u0027s use singletone\u003dFalse in that\nRoutesMiddleware object, at least until problem with thread-local\nsingleton and eventlet monkey patching will be solved.\n\nCloses-Bug: #1942179\nChange-Id: Id3a529248d3984506f0166bdc32e334127a01b7b\n(cherry picked from commit e610a5eb9e71aa2549fb11e2139370d227787da2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d961731a73128125ebf03fae52b4b34f3e7abf27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d961731a73128125ebf03fae52b4b34f3e7abf27"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
