)]}'
{"id":"openstack%2Fneutron~807638","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Id3a529248d3984506f0166bdc32e334127a01b7b","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1942179-stable/wallaby-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"Id3a529248d3984506f0166bdc32e334127a01b7b","subject":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware","status":"MERGED","created":"2021-09-07 06:10:41.000000000","updated":"2021-09-07 20:55:45.000000000","submitted":"2021-09-07 19:56:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807638-bug/1942179-stable/wallaby-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky-stable/queens","cherry_pick_of_change":807637,"cherry_pick_of_patch_set":1,"meta_rev_id":"ca917b16977c9ffe32c4e1397ecda50da42f46d8","_number":807638,"virtual_id_number":807638,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-07 19:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-09-07 14:19:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-09-07 11:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-09-07 14:19:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 08:53:58.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-07 08:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 11:44:33.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-09-07 14:19:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"39e13b71ec8c36cf22075cb645b7cb0c4d31519b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-07 06:10:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"016da9c5998ce51a53804f0989f8f1c6c4596947","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-07 08:53:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/38/807638/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/38/807638/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5cae744945cf5f82396d9194644160444ad50571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 08:59:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/f1ede1c17b9d48ab8f3f2cfb33efb9a8 : SUCCESS in 45m 10s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/c26c764f54774e1c909d74a4dfaafbc4 : SUCCESS in 54m 30s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/f61ae16e92ed4a2199816fa2f418efd9 : FAILURE in 1h 09m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b45c06ef3ada40c2809478caf22c0409 : SUCCESS in 12m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8eed9a6c2918450d8fa9694058fc9755 : SUCCESS in 22m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ddc15c450ca0440a82c49ab5fd4a823d : SUCCESS in 24m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/010e9812a5de4f00922c9b4771a8cbf1 : SUCCESS in 7m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2c539d619b0c44bda2e1cbdd8d2627bb : FAILURE in 1h 10m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b4b3444ce7a946eb818e2636530b7fb3 : SUCCESS in 53m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c6ee8bee1534c0994e3f88164a6b599 : SUCCESS in 1h 20m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0f5aa3b055534a00a7577829d0d72544 : SUCCESS in 45m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4a57eca8840548e6b42acb76e2fc35be : SUCCESS in 1h 19m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3cf7dc0a693342d9afd301cde2fb31ec : SUCCESS in 1h 35m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31f1abe09b8e4c1295df1b4828d16420 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":1},{"id":"96429dc0b6df4ff5c9a0963537a946af14ab37df","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-07 11:44:33.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4355c21a3d96f5b8d292db6e647a875f4449f293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 13:14:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/5ed04d91d4c940529f8f81dba686cd64 : SUCCESS in 49m 44s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/2d5ffe05e2614d0684da2783de19244a : SUCCESS in 1h 01m 28s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/a0e99ee093c247d0877ea95a51a5cc5d : FAILURE in 1h 06m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca1575cc92b448738d344ce68f3abfdf : SUCCESS in 12m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ec8eedf43f443a493b022c32ae8c564 : SUCCESS in 25m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d311eaab6d34410391da3115365c5d87 : SUCCESS in 26m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5403d6d2d0204f5ba05bd3da2a3a12cd : SUCCESS in 9m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ad0e3587f40a45fcb4ea150161f5b33a : SUCCESS in 1h 16m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2e49127024904f14953ea9b55192234d : SUCCESS in 57m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b53bb34b43db42c0889c4449feca17e9 : SUCCESS in 1h 24m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/39f57c9657f2444bba9793ff27c808f5 : SUCCESS in 45m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/eade3fb7f58f4e44a3092d1122f7e3b7 : SUCCESS in 1h 24m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2e554ac1a14a49c08c91977617fc0d3c : SUCCESS in 1h 19m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75d747990bd44b9ab22bb949764ff72c : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":1},{"id":"13e058990cfbc71e2808ac0ae6d3a025958abd38","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-07 14:19:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"169f18f79f40ce09067aaa79f85062db42d82504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 15:50:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/aee5e3ca1bab44be832ae952252bb4f4 : SUCCESS in 59m 53s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/8f3447d5cc0d4258aa0ed146a3a5c68b : SUCCESS in 54m 32s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/6340bdc7566c4700b72c79ae19a3a175 : FAILURE in 54m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16573596cba34d90b1a304d190b48da6 : SUCCESS in 12m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f245fc236e0b48188dd4ef1d2f7468ec : SUCCESS in 25m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/690e3b7663e8437186c29734d293df0f : SUCCESS in 27m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1dceec90f075445ea8414a9ea5d3585c : SUCCESS in 7m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/178260732b9d45c3bdffd4ea974384e8 : SUCCESS in 1h 21m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2dc650681a224213bbff165ed67afb24 : SUCCESS in 53m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87f6ca93b3c147afa15349bbea7959f6 : SUCCESS in 1h 23m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d1e53d1ae43b48f7a1391d1a82fca754 : SUCCESS in 43m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7275f708845a406495358cdd8d157a09 : SUCCESS in 1h 17m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e1e6b1f0d1af4aa788e69c8b02dbf285 : SUCCESS in 1h 21m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5d9f3cac8e34aa7b24d22abd5f302dd : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ab014cede94b7024bad9de0ff0fe4597cd1c2435","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-07 16:48:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b22b1160992001c0dd014df0d11c8faf2781605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 18:23:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/d4359bc1a2ad43a2b07ace26105b7530 : SUCCESS in 43m 47s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/bef77e4a5ed046c6addd606047eaa501 : SUCCESS in 53m 48s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/084f511a7ac5447d89c5a596500dc853 : SUCCESS in 53m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f6e4271a3b45b8a2bcdb0aa59ac226 : SUCCESS in 12m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/472b6a3b109e4870881afb8c4e711b1c : SUCCESS in 29m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c7f46f1b9cb456991e9f3f685ba2bbb : SUCCESS in 28m 59s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6c5de2d669f8465bacdc27515e73cf03 : SUCCESS in 7m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1db9702bd7e34150bce7388018450368 : SUCCESS in 1h 19m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/69e28b689f36493c9cba55db7a3829a5 : SUCCESS in 1h 03m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/543f25eabca14b2f9a214da894b72712 : SUCCESS in 1h 30m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/76998f5488434d2db724f7583ab93422 : SUCCESS in 42m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a150b0972cbc42fcbff7b965efa2b767 : SUCCESS in 1h 24m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d6abe374741d45fe9e2ad3fadc21d937 : SUCCESS in 1h 27m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4239d91f2a534124b7e0187a248f79b8 : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":1},{"id":"00fc8718db14fafdc264e9e55940efcbe1a5b230","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 18:24:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0d014d344f529182aa6ee0a5017b84a3b0e9ec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:56:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/e4533c2be3844175a91e50c1479910ae : SUCCESS in 50m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca6523c3f51d4c288c31fdc4c198508f : SUCCESS in 13m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a16f295141b84f4a9df4519202741d06 : SUCCESS in 23m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/be36a95f08bf4f61bc7d1a4d4da7d014 : SUCCESS in 27m 53s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4af41e5888e74629a3550e604b53c468 : SUCCESS in 7m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5a62eb958c634f32a72406f24909f4e4 : SUCCESS in 1h 13m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6f1425108ba841549f588f95d439c370 : SUCCESS in 43m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ea1842d24e74d93b732a9f1ebeae0c7 : SUCCESS in 1h 20m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f376a62e15c34c58849f0ca26d5749b3 : SUCCESS in 37m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/093250683bdb44f7a07310923208c2ab : SUCCESS in 1h 27m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a59291b118e24719a2bcb9b1b8fcf8e7 : SUCCESS in 1h 20m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d6e2e9ca4b84e719a104252ebf6735d : SUCCESS in 28m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f610fc58476f4f022de385fb46619d745deba6ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:56:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ca917b16977c9ffe32c4e1397ecda50da42f46d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-07 20:55:45.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/38/807638/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/38/807638/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dfbb9bcb5dae282c575b33171d691a5fca946beb","revisions":{"dfbb9bcb5dae282c575b33171d691a5fca946beb":{"kind":"REWORK","_number":1,"created":"2021-09-07 06:10:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/38/807638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/38/807638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/807638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/807638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/807638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/38/807638/1"}}},"commit":{"parents":[{"commit":"b22122453b5ceb195812a3180e0819836c9e8920","subject":"Fix notify listener syntax for SEGMENT_HOST_MAPPING","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b22122453b5ceb195812a3180e0819836c9e8920"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-09-03 14:04:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-09-07 06:10:41.000000000","tz":0},"subject":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware","message":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware\n\nIt seems that using default singleton\u003dTrue in the\nroutes.middleware.RoutesMiddleware which is leading to use thread-local\nRequestConfig singleton object is not working well with eventlet\nmonkeypatching of threading library which we are doing in Neutron.\nAs a result it leaks memory in neutron-api workers every time when API\nrequest to not existing API endpoint is made by user.\n\nTo avoid that memory leak, let\u0027s use singletone\u003dFalse in that\nRoutesMiddleware object, at least until problem with thread-local\nsingleton and eventlet monkey patching will be solved.\n\nCloses-Bug: #1942179\nChange-Id: Id3a529248d3984506f0166bdc32e334127a01b7b\n(cherry picked from commit e610a5eb9e71aa2549fb11e2139370d227787da2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dfbb9bcb5dae282c575b33171d691a5fca946beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dfbb9bcb5dae282c575b33171d691a5fca946beb"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
