)]}'
{"id":"openstack%2Fneutron~807763","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I9f753be0ce5ae7870afb9b3cb74f89be8482356e","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1942469-stable/wallaby","hashtags":[],"change_id":"I9f753be0ce5ae7870afb9b3cb74f89be8482356e","subject":"Ensure net dict has provider info on precommit delete","status":"MERGED","created":"2021-09-08 11:35:00.000000000","updated":"2021-09-09 22:41:43.000000000","submitted":"2021-09-09 22:40:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807763-bug/1942469-stable/wallaby","cherry_pick_of_change":807243,"cherry_pick_of_patch_set":1,"meta_rev_id":"a5e6a43870ec23097a13781c383734cf0ad9cbac","_number":807763,"virtual_id_number":807763,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-09 22:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-09-09 19:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-09-08 14:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-09-09 19:41:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-08 11:36:45.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-09-08 11:36:45.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-09-08 11:36:45.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-09-08 12:04:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2021-09-08 12:26:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-08 12:51:53.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-08 14:00:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-09-08 14:22:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-09-08 17:01:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58eb162713ab0a88964f5171481ff3209b5affef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 11:35:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* neutron/plugins/ml2/plugin.py","accounts_in_message":[],"_revision_number":1},{"id":"952c6cb1c4012cd3fcbc8dcd6a3f27ff86cb516e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 11:36:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"70414b433e02a1b8b8d5f9b0ab28a1db178566af","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 11:36:45.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"09e3bf85679ded9e08d8e8966968100a1e0f7816","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 12:26:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c577d60f31840d587e813defd7538c3 : SUCCESS in 44m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c54d68f425641a4ab5abef6760e3219 : SUCCESS in 44m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e552c391f398c34a884564380253e1a6f34e80f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-08 12:51:53.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/63/807763/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/63/807763/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0047682cf555d8f0b5418650de7f2d1d6aaad972","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-08 14:00:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/807763/2 : SUCCESS in 2h 22m 26s","accounts_in_message":[],"_revision_number":2},{"id":"882cd812c1cf020f55e50cf28d5c947ee7c65fa2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-08 14:22:45.000000000","message":"Patch Set 2: Code-Review+1\n\nIt\u0027s not exactly a backport due to recent payload changes, not sure if there\u0027s a need for that info in commit msg.","accounts_in_message":[],"_revision_number":2},{"id":"04de3e417df077fed28b0e5b2b26dbae93c376d9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 14:25:16.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e It\u0027s not exactly a backport due to recent payload changes, not sure if there\u0027s a need for that info in commit msg.\n\nRight, I missed Conflicts section","accounts_in_message":[],"_revision_number":2},{"id":"27a23a5f45ddae384f813ebf75cc9e60a91b0449","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 14:25:47.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"035a173f32b7d17f57bd326a7200c55edf3bd189","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-08 14:48:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c72d1def5e63c45dc06187440c9036a48c2f9b32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 15:14:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9e151a176ac4912ac11ff1136c50c14 : SUCCESS in 38m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c8f7d8a2ca54c129b3640aa371e5afb : SUCCESS in 46m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d25dffd2fae7c2adc4cdc78d606c3c7369685bd4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-08 15:27:24.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/63/807763/3/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/63/807763/3/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2ec7522da42957515313f86e7001b46bcc9a02d8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-08 16:32:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/807763/3 : SUCCESS in 2h 03m 23s","accounts_in_message":[],"_revision_number":3},{"id":"58dff13c252c8ec531118496b996010814583e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 17:01:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/2ef15391400a463bb9d9316b1ec6e4ca : SUCCESS in 49m 50s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/f7b2c7b638b54e6d9a8307b55b2d1d47 : SUCCESS in 2h 26m 38s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/40d2bbd825b74805be94242ed241ea5d : SUCCESS in 2h 15m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/cb58543361fc41feac14a549aeb5cf60 : SUCCESS in 1h 31m 02s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/c03f2c001d9e4a4f8a2459b2863e8d6c : SUCCESS in 2h 05m 02s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/0115b037e69445989fbdd271f1efd790 : SUCCESS in 32m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a56f1b15e3934e98a83af06156f9fde5 : SUCCESS in 34m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8747f8e0e1154038b752b847f794cc98 : SUCCESS in 11m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/890df53c1d3f407aae1ba6297a664d2e : SUCCESS in 23m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e8c24ca511144fba24b425e1c6b7a40 : SUCCESS in 23m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9308021e0e0945a3ac4ac6de992c3963 : SUCCESS in 24m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24627921a3024d9e9215b3c97e406e9a : SUCCESS in 10m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/22d3a9bf912c40b8b623e3bcf7a64de4 : SUCCESS in 48m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/96aef782be934500956ee3613780e9e0 : SUCCESS in 1h 25m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2ef7e33e83104c6dbb2982b1940a9969 : SUCCESS in 1h 18m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/56398192e98b42d2b42b65fed7b6aaa0 : SUCCESS in 1h 03m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2cb0f08245b749d39cd6530a2a79254d : SUCCESS in 1h 00m 08s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/42d9637da5f54b6ba391b40976f474a8 : SUCCESS in 1h 11m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3d426771f2df47ef834a1eed3e9e1cc7 : SUCCESS in 1h 23m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8ed09dcabd946b09c4cb7ca4d324b62 : SUCCESS in 1h 37m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5dad9e075f847469427cb414eb55403 : SUCCESS in 1h 02m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dd68d071cec2491d81de3e8cb5210531 : SUCCESS in 1h 07m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9948d84c3b70486182129092f486b281 : SUCCESS in 45m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/36af5850ad10485db0c07c97740841b6 : SUCCESS in 1h 15m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d6f32a3768d842b582203ec9ecfa3534 : SUCCESS in 1h 22m 28s","accounts_in_message":[],"_revision_number":3},{"id":"1c200c825d68e771537c6c1570f3f703d06fc71a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-09 19:41:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ca28c46da55636f3919aba9966e0e76c495de26d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 19:42:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"203de9939346305b90de138b966c20d36e74d447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 22:40:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/927968cabaf246b885169317835959bd : SUCCESS in 45m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54fc8fece4cf4dba9bf6a5c7de87fb02 : SUCCESS in 12m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16a4e512cbde479291f0cd1d2810f6e7 : SUCCESS in 27m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d496d1bab1d49598f6d0602564cfbf2 : SUCCESS in 22m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba7a4ccc5174fd095aaa6e8e192c8f4 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4f231e9926b48f4baf029314c324529 : SUCCESS in 47m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8aa13683e1db4cad827038d8890adb57 : SUCCESS in 1h 23m 13s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e550c168e1c44a949f7a5ec86fb2a9bc : SUCCESS in 1h 21m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f858235212c44180a9c0e45009c1177c : SUCCESS in 1h 00m 22s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3a39c699f4b94ede8bf1919b90e03a34 : SUCCESS in 1h 21m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea25b89d0a244249ab7ade7d64d8ec1e : SUCCESS in 1h 03m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/df5a0d09e0d045dbbdb861c0ce5189a3 : SUCCESS in 1h 10m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/470c561d148240e4b33312ba0d0fa6d6 : SUCCESS in 28m 06s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7c14e11baeea482ea27d856d956f3b00 : SUCCESS in 1h 25m 14s","accounts_in_message":[],"_revision_number":3},{"id":"9fcce402da8cf95a181b92614bf233b52ad1387d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 22:40:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a5e6a43870ec23097a13781c383734cf0ad9cbac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 22:41:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14827d5703414047b11c120522b3d769 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a981849c815ec2a35fcfbcc4c22d4f1d92eadca2","revisions":{"2c19a4c2c1b2b1b61cea72087f4a73859d97eea3":{"kind":"REWORK","_number":1,"created":"2021-09-08 11:35:00.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/63/807763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/807763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/807763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/807763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/807763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/807763/1"}}},"commit":{"parents":[{"commit":"c5e86f4f8f7e36233f383323e24e72dbe28efc04","subject":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e86f4f8f7e36233f383323e24e72dbe28efc04"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-03 07:29:14.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-08 11:35:00.000000000","tz":0},"subject":"Ensure net dict has provider info on precommit delete","message":"Ensure net dict has provider info on precommit delete\n\nCommit 80eddc40390e63c9c1102b827997054708f2618b optimized\nnet delete by including net info into notification payload,\nhowever ML2 plugin needs provider info as well.\n\nCloses-Bug: #1942469\nChange-Id: I9f753be0ce5ae7870afb9b3cb74f89be8482356e\n(cherry picked from commit 27edf6b6d311d7d334090c89c4cee54d63162a55)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c19a4c2c1b2b1b61cea72087f4a73859d97eea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c19a4c2c1b2b1b61cea72087f4a73859d97eea3"}]},"branch":"refs/heads/stable/wallaby"},"10d3669beda25180044958a80e5e833266891714":{"kind":"REWORK","_number":2,"created":"2021-09-08 11:36:27.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/63/807763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/807763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/807763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/807763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/807763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/807763/2"}}},"commit":{"parents":[{"commit":"c5e86f4f8f7e36233f383323e24e72dbe28efc04","subject":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e86f4f8f7e36233f383323e24e72dbe28efc04"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-03 07:29:14.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-08 11:36:19.000000000","tz":0},"subject":"Ensure net dict has provider info on precommit delete","message":"Ensure net dict has provider info on precommit delete\n\nCommit 80eddc40390e63c9c1102b827997054708f2618b optimized\nnet delete by including net info into notification payload,\nhowever ML2 plugin needs provider info as well.\n\nCloses-Bug: #1942469\nChange-Id: I9f753be0ce5ae7870afb9b3cb74f89be8482356e\n(cherry picked from commit 27edf6b6d311d7d334090c89c4cee54d63162a55)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10d3669beda25180044958a80e5e833266891714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10d3669beda25180044958a80e5e833266891714"}]},"branch":"refs/heads/stable/wallaby"},"a981849c815ec2a35fcfbcc4c22d4f1d92eadca2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-08 14:25:47.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/63/807763/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/807763/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/807763/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/807763/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/807763/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/807763/3"}}},"commit":{"parents":[{"commit":"c5e86f4f8f7e36233f383323e24e72dbe28efc04","subject":"Don\u0027t use singleton in routes.middleware.RoutesMiddleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e86f4f8f7e36233f383323e24e72dbe28efc04"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-03 07:29:14.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-08 14:25:47.000000000","tz":0},"subject":"Ensure net dict has provider info on precommit delete","message":"Ensure net dict has provider info on precommit delete\n\nCommit 80eddc40390e63c9c1102b827997054708f2618b optimized\nnet delete by including net info into notification payload,\nhowever ML2 plugin needs provider info as well.\n\nConflicts:\n    neutron/plugins/ml2/plugin.py\n\nCloses-Bug: #1942469\nChange-Id: I9f753be0ce5ae7870afb9b3cb74f89be8482356e\n(cherry picked from commit 27edf6b6d311d7d334090c89c4cee54d63162a55)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a981849c815ec2a35fcfbcc4c22d4f1d92eadca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a981849c815ec2a35fcfbcc4c22d4f1d92eadca2"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
