)]}'
{"id":"openstack%2Fneutron~807765","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I9f753be0ce5ae7870afb9b3cb74f89be8482356e","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1942469-stable/ussury","hashtags":[],"change_id":"I9f753be0ce5ae7870afb9b3cb74f89be8482356e","subject":"Ensure net dict has provider info on precommit delete","status":"MERGED","created":"2021-09-08 11:39:06.000000000","updated":"2021-09-15 14:51:44.000000000","submitted":"2021-09-15 14:49:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807765-bug/1942469-stable/ussury","cherry_pick_of_change":807763,"cherry_pick_of_patch_set":2,"meta_rev_id":"f6a58081d5ca1307b82904b39474010a422de8f1","_number":807765,"virtual_id_number":807765,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 14:49:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-09-15 12:13:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-09-09 19:41:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-10 08:02:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-09-15 12:13:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-08 13:01:29.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-08 13:50:19.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-09-08 17:49:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 19:41:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-09-10 07:45:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2021-09-10 08:02:03.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2021-09-15 12:13:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"aa69286beeab6ba44f56f00091e22e6eca3ae8a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 11:39:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"df97950e3d9c1cf1d5f0a972022f00ea1dae67c3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 11:39:13.000000000","message":"Topic bug/1942469-stable/wallaby-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"f0f4ffc309955d23d699395a399b95b07f1f80d4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 11:39:19.000000000","message":"Topic set to bug/1942469-stable/ussury","accounts_in_message":[],"_revision_number":1},{"id":"7501df93d0ef5fbae7bf5b92c4b3cdf25957b4c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-08 13:01:29.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/65/807765/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/65/807765/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e8c11d5f9af05cf9100ce2ac39917b43948d0aae","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-08 13:50:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/807765/1 : SUCCESS in 2h 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"04b28e2aa3c45360331b6eb003ac6d7df6afbc30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-08 14:26:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a3381b060a99853ad822a611b4def14d49d3232e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-08 15:36:55.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/65/807765/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/65/807765/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e14d104d2cd3ef7f53024c9a6659164b08fbc33e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-08 16:19:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/807765/2 : SUCCESS in 1h 47m 30s","accounts_in_message":[],"_revision_number":2},{"id":"068a9cda83744c13d17651834bbde981a43c0633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 17:49:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/f00b98e46136443c937c2c632eb888e9 : SUCCESS in 49m 36s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/efed3a1ce1f542bf88405ca4f1ec285a : FAILURE in 2h 04m 01s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/73a850c6c19d40f09526285424e7eaf4 : SUCCESS in 1h 41m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/69b009d711604db2baed830448d0b514 : SUCCESS in 1h 34m 19s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/f81214629f7d40eba4c279e6d61eefba : SUCCESS in 1h 35m 39s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/83f262f5f86e435880b9655014a56645 : SUCCESS in 26m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87f5b7dca92c44b085c6bc49cfdac774 : SUCCESS in 27m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0553b9b2190446faac8ccb6c042a6c64 : SUCCESS in 13m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/316ccabfffbc44829538ceac469f1f5a : SUCCESS in 27m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2500fca2e464dec9b8007cdb2397f90 : SUCCESS in 23m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/919997aa751f4fa5a7098bcee6260dd1 : SUCCESS in 21m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe6db9f19864056840b9fe250def9d7 : SUCCESS in 9m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/58303c56b39441c7b88ab524ec3dab90 : SUCCESS in 46m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/06c2e29808e14531b80db15d6a1e78ea : FAILURE in 1h 03m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f413536352b04665a2cebe426cc51d06 : SUCCESS in 1h 17m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9eb35b8a859440d095ee5fb217eb8e66 : SUCCESS in 1h 25m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5766b6764bfd4ef281d3dfcc0a78945a : SUCCESS in 1h 05m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dd8459f4a0d24ff990c7d33762db503f : SUCCESS in 1h 08m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3f8ee4b16e604029acb75d777ea8df83 : SUCCESS in 1h 52m 34s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/24acdd03bca142efbc06f7246e73d30c : SUCCESS in 57m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a364934b2b4740f4aff1191825d8de7f : SUCCESS in 1h 10m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3c718e48626541a49995a9b20c7b47e7 : SUCCESS in 1h 29m 07s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d0ab9e9c59c44da3b7ed083a209beb8b : SUCCESS in 1h 20m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f33eb27b93824e80aa3c4f569cb1c48e : SUCCESS in 1h 07m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b582bbc4f5d1421bae59e72f1f630728 : SUCCESS in 53m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b5ea4462b84d430cb4654ba4261d0d20 : SUCCESS in 1h 27m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3e89a331bd1445ca621a655337fa1c2 : SUCCESS in 58m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5bc7b29ad3e34bc28a67f0357471baf7 : SUCCESS in 1h 08m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5e56b45e3297426bb6a42058136a960e : SUCCESS in 36m 19s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8c53268df4464f3ebef024021c85786f : SUCCESS in 1h 08m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad388713349b4168a48a8db24479f360 : SUCCESS in 46m 51s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3876ff0199434ffa81c57a38ba6e13ee : SUCCESS in 3h 20m 40s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/44149ed68e3d4572982ffa54a0c720eb : SUCCESS in 1h 04m 24s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6fd3bca8da6b42149099c2ac066ca20d : SUCCESS in 1h 31m 30s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8ff9208ed1be4a2699f13655b10a11d9 : SUCCESS in 1h 55m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/04f4ea7b1d5146ab97f45c2feb809a93 : FAILURE in 1h 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a60e4bbb47bdd829718af94a063b8f92f159f855","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-09 07:12:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"af89b45d0327297b9b4344561a32084755fb9a96","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-09 07:25:51.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/65/807765/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/65/807765/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"da53a85a5eb4d83915ab75730e05498e96b3d908","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-09 09:29:45.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/807765/2 : FAILURE in 2h 17m 08s","accounts_in_message":[],"_revision_number":2},{"id":"a99956f04bb26d04ad85752c18e8966482f473cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-09 09:35:03.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/65/807765/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/65/807765/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"204e8d99074f958e3f1a406e11f042556d350e9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 10:43:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/463e8a9bae544e32abad9df92902f6fe : SUCCESS in 45m 42s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/c4ca01180379455d96a7365bc49fac47 : SUCCESS in 2h 39m 38s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/8edcfdc0860543d2b76c31b774627dd9 : SUCCESS in 1h 39m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/39078309480a4bb591084b210312af25 : SUCCESS in 1h 37m 51s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/cbc5ee7983d04c33bf12b5eeb13c771e : SUCCESS in 1h 53m 47s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/eb9d8941fec043308f2a7814ac08ae32 : SUCCESS in 31m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b196b6dd75e49ca81dd8526aaa5545a : SUCCESS in 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/792e475c5eca4b8ba399a6b34e242a34 : SUCCESS in 14m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bddb44c7fa6b48f6a09bb57c19165d13 : SUCCESS in 24m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26d3b1ddcbb44dd1b92ceef1cd2e18e0 : SUCCESS in 29m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbd486061a36481db0f0332195e57c9d : SUCCESS in 23m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9696218fed45e58e36be354b5fba5e : SUCCESS in 10m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/41491979845f4c28bbc03c1b1aeb2c99 : SUCCESS in 53m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d513f06445164026ae562b930bbd80ee : SUCCESS in 1h 19m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/685dc70d58db47a98a027d31d1f8cb39 : SUCCESS in 1h 17m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5c356468187245888555d97f4fc93d78 : SUCCESS in 1h 31m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/48a9d7d8f0d949b6a16ce9079db02d87 : SUCCESS in 1h 05m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f8bf5736311e49a5b21148d8bb52d43f : SUCCESS in 1h 06m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2b1e9f47fe0d4cf4b1e26cb385dcc5c9 : SUCCESS in 1h 10m 53s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a38edea482940579a0bbbf5383c06f0 : SUCCESS in 57m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/748c4247d7264d56a42b05ed6ddba31a : SUCCESS in 1h 10m 32s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ff3cd9de3b844947afe54834ce632e70 : SUCCESS in 1h 37m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/01b498d13c4d49f8ac7a431ad6f90428 : SUCCESS in 1h 28m 31s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f90b3aefd6f940ca9ad32cff002eec5c : SUCCESS in 1h 03m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fe78bb938f724937a57e2919664141a6 : SUCCESS in 1h 03m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37ee2cae4b6f46a3aedcfa2e0ffc68cf : SUCCESS in 1h 24m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52c88addf41049d6af1d1a4c5c57302b : SUCCESS in 57m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2baee43f5f244323852574599fbaad0b : SUCCESS in 1h 06m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bc2d2878a7b247a9b6f73a8098a96da7 : SUCCESS in 40m 02s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dbc20f69874f49bd9e5e84075ae7ef93 : SUCCESS in 1h 12m 47s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d520e6532e56425eac86c74a6f2be0d0 : SUCCESS in 53m 37s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7029c93d5b5c4be6a95023ad3c5127ea : SUCCESS in 3h 26m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8a1b0037fe334ff0b3e36c9095df4fa0 : SUCCESS in 1h 14m 11s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5c9603c0432a4954b1ac3844c92317b7 : SUCCESS in 1h 38m 27s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/eea8923c74bd4538ae077b7d31e77efd : SUCCESS in 2h 00m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0a9e7b4ac1694e239d1349644f26721f : FAILURE in 1h 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e51346da2063b62d8d0079c6e757c46468dd1d3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-09 19:41:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"44e0cc564df40e6df21289ec989d327b8e44e981","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-09-10 08:02:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"33f321af8f780102da496788f5c3b119ba04b4e6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-09-15 12:13:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b824fc4f3d63eedf9412b501fced90e4a48bf6c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 12:14:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"42e38eb135b57dbcdee3d82308bf215e7b1d4880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:49:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/de067c93d7794d33bf774bf311e396e3 : SUCCESS in 46m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b148e974f8b44848eef64f6d0eaf77d : SUCCESS in 13m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42445b482d01412a8f7369c2eb2712b4 : SUCCESS in 24m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffb74d9d9bf24117bc2486d4f631df8a : SUCCESS in 26m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1193db9ff51c47b1a933d5639c275a2f : SUCCESS in 10m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b813154c54e548f3859fde9f0068d27d : SUCCESS in 49m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ca889c78677948c5afea94e1de06dd70 : SUCCESS in 1h 20m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/448783da13ad42258c60c4e0e0310ee9 : SUCCESS in 1h 14m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c124d35657514fe5abafd3655d5c7e72 : SUCCESS in 1h 03m 21s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b475ea49c60f4bd19f313b677b2c587e : SUCCESS in 1h 08m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/462be9e1a7564c0da3c54d35ce8251b0 : SUCCESS in 1h 09m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2be9fa46e41641509cedf132e9feb118 : SUCCESS in 1h 07m 18s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/acb864f95a824b73837538d333761619 : SUCCESS in 1h 31m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e6f3d6e5a42418eaaaaf1ec4cc9a25a : SUCCESS in 1h 14m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d4c4e775a5644dee885819b14e2ba893 : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":2},{"id":"9bac235c76c483613e8173b58373bdb79d13098c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:49:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f6a58081d5ca1307b82904b39474010a422de8f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:51:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2a329468174329bebb2aeccd98095a : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"78ff22e6bb3254859df5c90d41e31b84c70ab5b5","revisions":{"08c0f8ef37426cd402df40267564039fdd50f6af":{"kind":"REWORK","_number":1,"created":"2021-09-08 11:39:06.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/65/807765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/807765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/807765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/807765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/807765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/807765/1"}}},"commit":{"parents":[{"commit":"6a761dc42da99625bb18b4aabf4e2b340396c78c","subject":"Merge \"Don\u0027t use singleton in routes.middleware.RoutesMiddleware\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a761dc42da99625bb18b4aabf4e2b340396c78c"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-03 07:29:14.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-08 11:39:06.000000000","tz":0},"subject":"Ensure net dict has provider info on precommit delete","message":"Ensure net dict has provider info on precommit delete\n\nCommit 80eddc40390e63c9c1102b827997054708f2618b optimized\nnet delete by including net info into notification payload,\nhowever ML2 plugin needs provider info as well.\n\nCloses-Bug: #1942469\nChange-Id: I9f753be0ce5ae7870afb9b3cb74f89be8482356e\n(cherry picked from commit 27edf6b6d311d7d334090c89c4cee54d63162a55)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08c0f8ef37426cd402df40267564039fdd50f6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08c0f8ef37426cd402df40267564039fdd50f6af"}]},"branch":"refs/heads/stable/ussuri"},"78ff22e6bb3254859df5c90d41e31b84c70ab5b5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-08 14:26:14.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/65/807765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/807765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/807765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/807765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/807765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/807765/2"}}},"commit":{"parents":[{"commit":"6a761dc42da99625bb18b4aabf4e2b340396c78c","subject":"Merge \"Don\u0027t use singleton in routes.middleware.RoutesMiddleware\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a761dc42da99625bb18b4aabf4e2b340396c78c"}]}],"author":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-03 07:29:14.000000000","tz":180},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-09-08 14:26:14.000000000","tz":0},"subject":"Ensure net dict has provider info on precommit delete","message":"Ensure net dict has provider info on precommit delete\n\nCommit 80eddc40390e63c9c1102b827997054708f2618b optimized\nnet delete by including net info into notification payload,\nhowever ML2 plugin needs provider info as well.\n\nConflicts:\n    neutron/plugins/ml2/plugin.py\n\nCloses-Bug: #1942469\nChange-Id: I9f753be0ce5ae7870afb9b3cb74f89be8482356e\n(cherry picked from commit 27edf6b6d311d7d334090c89c4cee54d63162a55)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78ff22e6bb3254859df5c90d41e31b84c70ab5b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78ff22e6bb3254859df5c90d41e31b84c70ab5b5"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
