)]}'
{"id":"openstack%2Fneutron~809219","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Ide7e9771d08eb623dd75941e425813d9b857b4c6","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1927868-stable/wallaby","hashtags":[],"change_id":"Ide7e9771d08eb623dd75941e425813d9b857b4c6","subject":"Revert \"[L3][HA] Retry when setting HA router GW status.\"","status":"MERGED","created":"2021-09-16 08:05:43.000000000","updated":"2021-09-16 21:37:04.000000000","submitted":"2021-09-16 21:35:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809219-bug/1927868-stable/wallaby","cherry_pick_of_change":805366,"cherry_pick_of_patch_set":1,"meta_rev_id":"e6eb44061d0f3ca25aa5d7f993050abce13110ec","_number":809219,"virtual_id_number":809219,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-16 21:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-09-16 19:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-09-16 11:29:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-09-16 19:44:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 08:05:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"},{"updated":"2021-09-16 09:05:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-16 10:20:58.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-09-16 10:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-16 11:29:08.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-09-16 11:53:28.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-16 19:44:04.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"6801f8193c7d187a50329d6d8641eab382908c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-16 08:05:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"62f26e81a5477156b1b3d080acb1e62507fe5122","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 09:05:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/037547c3868f4e3dafd461aeb16775d7 : SUCCESS in 49m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e31cca411cd84b98a111d8daf20d2ae6 : SUCCESS in 48m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ab22c83d0ef62fce074cf0bfee385ca1538d898","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-16 10:20:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/809219/1 : SUCCESS in 2h 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4b8ba112b6b0a781ab857a977c4bd38ce3a1af28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:29:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/7855e628403747368ee6c345335f891a : SUCCESS in 2h 15m 06s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/2a802964854d4a86a343b8d100b18dc4 : SUCCESS in 1h 52m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/10d5d9acd9584a6fb29b3b75c5224dda : SUCCESS in 1h 33m 41s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/15ed8a53c67f4289aa42db53ba45b0ec : SUCCESS in 2h 11m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc329298875a4a77aaabf50de79d1f1b : SUCCESS in 36m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2828ca47e54176b1918cb21607e345 : SUCCESS in 14m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3dda2a4c7894d23a0f64e5b43ef8f7a : SUCCESS in 32m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c55bca3c31df49d19518ddbc7623cc13 : SUCCESS in 30m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a304dda8849e4a928ee55a6e7a4ad84b : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a0578e442f34e06a2d4a804842a41a0 : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d988f72ccbbd4398a0b798d1625984a1 : SUCCESS in 48m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1af43a814db1448089411d4a6703f7fc : SUCCESS in 1h 23m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e58df8e30f2041abb66cb30bc5463e50 : SUCCESS in 1h 13m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b0dc4aeffe441d1bebe1fcffdf1ac28 : SUCCESS in 59m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/16a0a00bda124330852769dce50a2577 : FAILURE in 49m 46s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5b39133ac6614c90a8fc11bad80fc51f : SUCCESS in 1h 23m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/51bfbaa5fd0c4933a2f489b09140657d : SUCCESS in 1h 31m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3919a18e0bdd4b8f87ba95252a389d4a : SUCCESS in 1h 42m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/826dd1e6104345f19196b520bf9e4f41 : SUCCESS in 1h 00m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/af26d5d68bf048b580227efeb9d5270d : SUCCESS in 1h 02m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cff2e5aca98c4893a94b55033f3fe067 : SUCCESS in 33m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aa8cd0ee68e242928fa6c3c958bcd170 : SUCCESS in 1h 09m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9fc49e7fa1f44d8993bc8d5f7c5f60c1 : SUCCESS in 1h 26m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0ff63aab73e862dc6e063561ec4b2bbe39887528","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-16 11:29:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a599120e48246507686e60528c2ab095a5de944","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-16 11:53:28.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/19/809219/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/19/809219/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d5fab1effc346e14233489ee95becff31bd0e3c8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-09-16 19:44:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88a2dd5d25a1c07c368943c0905b1c4a4ac4ee50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 19:44:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6680d27e1e92110ac4e2a672e039306e3d56a461","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 21:35:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a55951683454d5d869128407f6aa2ab : SUCCESS in 17m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de94c49373304257842826ab1d57137e : SUCCESS in 36m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbce8fc0de4f4ad08d0ef2fa9b727ecf : SUCCESS in 29m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e9aebb1544a46848d29a43521eb3e36 : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6c9d48d22474e7bb0cf8063b4a4ab3a : SUCCESS in 48m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31cd89fc7a2a43cbbea1efda5b3f5371 : SUCCESS in 1h 13m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5043b75f40ff4423b9a45d58013ff39b : SUCCESS in 1h 20m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3bb15acbacd46ecad7f444707151454 : SUCCESS in 1h 11m 19s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20d7cc6785c641f9a083b53aef401ae0 : SUCCESS in 1h 44m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b77840f6bba844a9b6ba330d97a816bc : SUCCESS in 1h 00m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1ba9898292bc4e09bd84f94aa3ef9196 : SUCCESS in 1h 24m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a520d0820c634eebbc54d22d842e7761 : SUCCESS in 39m 29s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9d89ff30a9734ef28cc0d8fab0bb4f77 : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b2ba85b7d3651417474d0ef68c764cacb8be0a70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 21:35:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e6eb44061d0f3ca25aa5d7f993050abce13110ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 21:37:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e305b24a04484326bab72e8fc4b9c74a : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3a70fe75315fba061ecb7d6ac1d50a04768ec13","revisions":{"b3a70fe75315fba061ecb7d6ac1d50a04768ec13":{"kind":"REWORK","_number":1,"created":"2021-09-16 08:05:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/809219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/809219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/809219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/809219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/809219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/809219/1"}}},"commit":{"parents":[{"commit":"0aa5ab2dcb7d72cb49a16aa718154cee20bd9001","subject":"Merge \"Skip FIP check if VALIDATE_MIGRATION is not True\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aa5ab2dcb7d72cb49a16aa718154cee20bd9001"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-08-20 11:25:04.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 08:05:43.000000000","tz":0},"subject":"Revert \"[L3][HA] Retry when setting HA router GW status.\"","message":"Revert \"[L3][HA] Retry when setting HA router GW status.\"\n\nIn short this patch can cause the privsep reader thread to\ndie resulting in the l3 agent getting stuck and e.g. not\nprocessing any router updates. See related LP bug for full\nexplanation.\n\nCloses-Bug: #1927868\n\nThis reverts commit 662f483120972a373e19bde52f16392e2ccb9c82.\n\nChange-Id: Ide7e9771d08eb623dd75941e425813d9b857b4c6\n(cherry picked from commit 344fc0c8d2ce7d942606c834a54cb81f0b47aa37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3a70fe75315fba061ecb7d6ac1d50a04768ec13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3a70fe75315fba061ecb7d6ac1d50a04768ec13"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
