)]}'
{"id":"openstack%2Fneutron~810704","triplet_id":"openstack%2Fneutron~master~Ibe08bfbc2efc55b9d628cdd0605941b7486186b6","project":"openstack/neutron","branch":"master","topic":"bug/1944708-segment-tag","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-08-09 14:54:15.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-08-10 14:48:49.000000000","reason":"Change was submitted"},"33493":{"account":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"last_update":"2022-08-10 14:48:49.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2022-08-10 14:48:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe08bfbc2efc55b9d628cdd0605941b7486186b6","subject":"[OVN] Fix updating network segmentation ID","status":"MERGED","created":"2021-09-23 15:06:31.000000000","updated":"2022-08-29 12:26:51.000000000","submitted":"2022-08-10 14:48:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810704-bug/1944708-segment-tag","meta_rev_id":"efab5c834d48b7077f2a2d817814c7edf0c201ab","_number":810704,"virtual_id_number":810704,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-08-10 09:47:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 14:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-08-10 12:56:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-10 10:34:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-08-10 12:56:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"}],"CC":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 15:06:58.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-09-23 15:06:58.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-09-23 15:53:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-23 17:05:17.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-09-23 17:26:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 23:06:10.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-05 12:11:51.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 10:08:45.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"},{"updated":"2022-08-09 10:37:25.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"CC"},{"updated":"2022-08-09 21:11:14.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-08-10 06:34:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"50d628c507ebf4fffbfa231f5d93b799b06545f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-09-23 15:06:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2d3bf77220e20a17deee3becfff09968c44f951","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 15:53:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b16b0f13d197452290646ad0da1cdeca : SUCCESS in 34m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/628aa9ff4bb0404aa248efd066256076 : SUCCESS in 44m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d062bb4e2099658cbb1c7a4b2d6ce61c02b8c324","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-23 16:46:01.000000000","message":"Patch Set 1: Code-Review-1\n\nSo far, only networks without ports or ports:\n- unbound\n- failed state\n- OVS bound ports\n\ncan change the segmentation ID. Error when changing segID in a network with OVN driver bound ports: https://paste.opendev.org/show/809538/ (same as with any other driver except OVS).\n\nYou can check https://review.opendev.org/q/topic:%22bug%252F1806052%22+(status:open%20OR%20status:merged)\n\nThe feature implementation server patch:  https://review.opendev.org/c/openstack/neutron/+/633165\nThe OVS mech driver patch: https://review.opendev.org/c/openstack/neutron/+/632984 --\u003e you need https://review.opendev.org/c/openstack/neutron/+/632984/28/neutron/plugins/ml2/drivers/openvswitch/mech_driver/mech_openvswitch.py (along with the driver changes you implemented in this patch)\n\nPing me if you need more details on this","accounts_in_message":[],"_revision_number":1},{"id":"a4d8cad146d08eb2867c930b0c2c5cf0d9b4a836","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-23 17:05:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/810704/1 : SUCCESS in 1h 58m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f3d675ec8fe4418354307d8aff180737452f3b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 17:26:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/74125a20dc534fe7a44d8e396aec815c : FAILURE in 59m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/12f5fd8b88b344d4893e857acb0115f6 : FAILURE in 2h 16m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/684c2ff26fdf41489e5657da661787ea : SUCCESS in 39m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e981a64b564a4ae99d9a8b3128fa19c5 : SUCCESS in 44m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01158acc6f4c4302b3cd7ad08681b9b0 : SUCCESS in 15m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a26bd6ca34343ce8569043b6d7101e6 : SUCCESS in 28m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8d18c9672c54c49bf2981ef90ab5213 : SUCCESS in 28m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c117ad63604445db62e1003c0aa0877 : SUCCESS in 25m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f13ef76f417a452f82e66b4690956f9d : SUCCESS in 11m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc873f555dbf4545aa3d4a94d0115fbf : FAILURE in 49m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/be1c9d85603541568ecfc064a8b7ce8c : FAILURE in 52m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/883d5cc3add7464ba44daae39b26320f : FAILURE in 36m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/46c94c613321460b8298fe92524227ee : FAILURE in 1h 12m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e5a4b965150b4395932b6fe3f013cd52 : FAILURE in 34m 08s","accounts_in_message":[],"_revision_number":1},{"id":"98f7b30576435d7d570628d3200dadbb0d367c81","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-23 23:06:10.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/04/810704/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4a9f071dffef5c6d7f3e266baf01b1ac9e53a234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-09-24 09:09:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4370c2388a338d9adc2c8595bcc15bc90a6407eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 09:51:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d2bc8d9a57a40439d94bf785c84a2e4 : SUCCESS in 37m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4ad12a171ee45f2ab0c1f7670ebe491 : SUCCESS in 40m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1682d7e792fa6a4cec6dd4cd1b3bc684d0065c68","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-24 10:39:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/810704/2 : SUCCESS in 1h 29m 43s","accounts_in_message":[],"_revision_number":2},{"id":"407cf3d5e707e58d757e923d4a20ddd6da30bf7f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-24 10:44:38.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/04/810704/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"736da626aa1d34d263aeaaa4d9975e20c72c4fd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 11:41:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9e74da28cdb842d3ad48c26c7445df4b : SUCCESS in 43m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c925901fbca4474aba710dec650823ca : SUCCESS in 2h 27m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5149fc44858b483595efa3e7236ef049 : SUCCESS in 34m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f8ad468421b455a9fc667299da4f96d : SUCCESS in 41m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c905ce6d56a6404e8387eb3e6ea27f03 : SUCCESS in 14m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8751cc0ef77c4b89aaa4f005a3be02f1 : SUCCESS in 25m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8c3b3b9df2047438d0a1d97a5b68305 : SUCCESS in 25m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/237a471a4218433d8030eb24a27e9e62 : SUCCESS in 23m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b567bd6c95cd40cfac34241264d80963 : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a62154b7e074102b6cc738041a4ab45 : SUCCESS in 46m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8effbbdcea474a93b923560eb4781beb : SUCCESS in 55m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9b9cc2d1358b4425b6263faf2063915a : SUCCESS in 32m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d7b9dc99444a418095d3caf4b22a01ac : SUCCESS in 1h 10m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7c6ecb2a0dd949858014fce9bb02a0a2 : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":2},{"id":"26fae809d22b4dc181f3690e022a075cea56021b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-24 13:03:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"203fe1ddf61aac3344d27c44d40e574ca863815c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-05 11:16:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ae98fb38b369615cd2c33b7cba4c44acb9286116","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-05 11:16:28.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"97b76e5045259a18cd2d0d9f2d41e48007560f2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 11:58:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d829e1817dc1423db8616b2fe9107d68\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae90d818d6c24601a7eecff806e078e9 : SUCCESS in 30m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e734be5d306d47dc8f175c93f5286331 : SUCCESS in 40m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5780a2930155fb70a8cc86f3e8d52e440e7c59d0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-05 12:11:51.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/04/810704/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"dd955b27c9456cd7afd89c7345ba4639d42fa58c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 12:52:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/946ec5d8505847c7bddffa14382b77f2\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0a3e77c9eaff4fa0a03e698345ddb851 : SUCCESS in 1h 29m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/248eacaa75a54b65921c527b8997bd81 : SUCCESS in 1h 16m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/63905f311e6747499faec2322a6108a9 : SUCCESS in 1h 06m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6a5b4878a9744bce9293fbb94f9b6e99 : SUCCESS in 1h 17m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00fb808b38b24fa69227147d1fbc1d45 : SUCCESS in 42m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2a3f7e78f7b430496ef67ee3bef5372 : SUCCESS in 29m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2e6cdbdcca94f43800121124a046b31 : SUCCESS in 11m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bcf19bc44cf47b1bbbc2ba3dbf3036e : SUCCESS in 23m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93438419b74e4bbeaa7cc91333843ced : SUCCESS in 24m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7693874a090b4888b187a39e0cc151a8 : SUCCESS in 23m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f0347a13d845a1b464272a1b06b55c : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0f2f4646fa047459ac87c7cbcb5871f : SUCCESS in 1h 02m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d652527c83bb4252985e0b59e68b18e0 : SUCCESS in 1h 34m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/501bc09b99be43f4acb446ec1a5d3387 : SUCCESS in 1h 04m 36s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c79fbc7f20c4bd183bf0952cac6cccd : SUCCESS in 1h 02m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/49f3e4258c7a4529990289e645d07155 : SUCCESS in 1h 03m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9eb8c2b4656a4e8484cbbbb6db10090d : SUCCESS in 1h 06m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0fd7d05711e34cdda726b8d1997df5bb : SUCCESS in 1h 19m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3aa91ad45f28457ea806a9a35a5fab27 : SUCCESS in 1h 09m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27c9bb7e2135446fb36b7f81411cd57e : SUCCESS in 1h 06m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5bf5e923ba4e5def3558f93f1bb24e44322530ba","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-05 12:55:21.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/810704/3 : FAILURE in 1h 38m 39s","accounts_in_message":[],"_revision_number":3},{"id":"e5343099b630a0e6df3f67bea204c5d91c8e7ecf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-09 09:16:41.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ae1f52ae6fd9239bd954e5320f44b0fd4c9fcb71","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-09 09:17:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"85ca140c1a14034450195a73f707fe1f421d21f0","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-08-09 10:08:45.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0b67fdfe8952f7b2c68b9b3d15349d88116f48e3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-08-09 10:37:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"677f0b7b1e72a0a2561760b8631d80658cf36c6a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-09 10:51:23.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4f9bf01fd8461b971bb15f7e40ef5ff94ddc4c43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-09 11:03:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5204d06d3edbce6cf4d76817673721424e2a24d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-09 11:31:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d72eb6fc19312e1f40dd2b201534fc300928a6ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 11:55:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c85f6f54ef7e41c9ba8e8cbbfee64a1b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b30a2db2490b43d59c1932e96b5abafc : SUCCESS in 45m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1aecbe1dde2c41819362b13ca643a32f : SUCCESS in 44m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1d77c622ef7ad80c5bbdef02ebfb6fc76919bf6a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-09 11:56:29.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/04/810704/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"df011e0c9b0fb6e159c6c8805fdf9ceb5f68f787","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-09 12:14:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"073e7c8c3187ccd7d21cbaf44ab7a32de56b6214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-09 12:18:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4a31e0c8667a214fb246575ed229c0073af9c5cf","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-09 13:04:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8655d6e21be9a8a90b12dc8c0597ae1b133a8637","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 13:21:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/903410505583471ea88df4f640aced5b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e1d4acb69d6403ea11ea1bb6f293912 : FAILURE in 49m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7043956a64b64a6cb6658e004035e545 : FAILURE in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2340afe25471d1d31f16613658db6d9838e39ced","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-09 13:35:03.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/04/810704/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"833b88a5c6042e15cb6c0a231e7e1b1619515865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 14:08:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/551d2ae4d7a44621ad33808ad0fbc2cd\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/85c7242c3dd1490e883ebf0959ec011d : SUCCESS in 1h 27m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eaf39616e30d41a79af41ec4dd8ebfb5 : SUCCESS in 1h 45m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd5f04142f7e4d62bbdcd282d028679f : SUCCESS in 1h 40m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fea1485813b64b0eaa97a909f44f7293 : SUCCESS in 1h 24m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d682dabd82cb425caf56024e218da42e : SUCCESS in 36m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54e965997ea4462baaf4ffa0ab6a6b5e : SUCCESS in 29m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ebe29d09994f15845831dd296a4bbd : SUCCESS in 13m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c6009734d3749598ca1b11868539df9 : SUCCESS in 24m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/336edec9dd41419d9544f4e895f93944 : SUCCESS in 26m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c65e9614df043c5aad6eee9b8666427 : SUCCESS in 24m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a273db79444a4492556860805850ab : SUCCESS in 14m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c9c75cf60fc40c789b18425459d09c4 : SUCCESS in 8m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/339019f229e04376b7001cddf8ead0d2 : SUCCESS in 1h 05m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/191fb5b7f64e488094cbcb2ec4c18064 : SUCCESS in 1h 38m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dc166a0cef4f47e8afc97dd694430627 : SUCCESS in 1h 15m 37s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8bda335e396244baa21ece0f988143f9 : SUCCESS in 1h 04m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ddf3655c6e3b4c718be86e344bdc81cb : SUCCESS in 1h 06m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/51c454fcb25f46f4b7f4c188e83b4d13 : SUCCESS in 1h 21m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2579b811d7c8433da06c07fbc85c5ee3 : SUCCESS in 1h 26m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6244b4d23291465b877ade74a8d41eca : SUCCESS in 1h 07m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/891ef816ce1c400d9786cbc619ed57e1 : SUCCESS in 1h 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db372794a143b8b76527e11c480e070a2812a9ed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-09 14:54:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0f8a6a7506e2cc7afe389b8e988607132c769f6e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-09 17:46:30.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/810704/5 : SUCCESS in 2h 24m 35s","accounts_in_message":[],"_revision_number":5},{"id":"ac26969c9c79e08f7201261c413c90baba1eb7f8","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-08-09 21:11:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2e1d382d30631dca1c0ac7cb1dad62ef938612c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 21:11:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c0f19687d6a4546c76edda637d2d16db0309048b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 22:50:52.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5383701daffb426694a311ab8215e659\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/188d79d138654e8aa1457c19418f8590 : FAILURE in 28m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5739cebc08f64005abab878b8a9617a3 : FAILURE in 27m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f3ad1c33813f4668baf632ead3a53c98 : SUCCESS in 1h 14m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/81a60253be6d47728cc4fc91ae7db22b : SUCCESS in 1h 05m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/105bacbd17a04af0829c8b8745677669 : SUCCESS in 14m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/857101ed19b347d19f2c0b4a6ab4e032 : SUCCESS in 27m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e7d7516fc2e4dcaad43b694d7b2487b : SUCCESS in 28m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d3a59e64c0415d968c35a1870672fa : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bed18ea412f14ff0ab7c93171516492d : SUCCESS in 8m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/05e58633a33e4d97bb93dc23c2877d1e : SUCCESS in 1h 00m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/73749ee37dcc4cc1b93b78f47fd46281 : SUCCESS in 1h 32m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/188c17f63ad84bf9ba0a9ff5e46bc403 : SUCCESS in 1h 18m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccbb1d2ab83c4574a5207657592d1d02 : SUCCESS in 1h 04m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b59d93256ce84c34849dbf6ee6f0f479 : SUCCESS in 1h 07m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/384d4c00d2cb4c789490759ae21104b3 : SUCCESS in 1h 13m 29s","accounts_in_message":[],"_revision_number":5},{"id":"7a3f23b50988c3e8c229cad9cc184e806e97453b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-08-10 06:34:51.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6d0341c075c147f74e109e3e6e4655b892dc1ef5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 07:12:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/181d0ce682dd46a39d7a6ca4037c4dc7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecd75d0146404690a8e75d4c32f75ea8 : SUCCESS in 32m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c52cb5f6c784dcfb7126839572e1d5b : SUCCESS in 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"94287d712c90358268edccc2639e82d9931de5ea","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-10 07:53:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bcf4c556abdb119c0651a09edd9e1325578eabc7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-10 08:18:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/810704/5 : SUCCESS in 1h 43m 15s","accounts_in_message":[],"_revision_number":5},{"id":"4a383f513392d17067dc7c28bd3156a8ce089850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 08:19:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/429b302ef0e54074a4b74ca716d56fd8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/234161cce3ca4abc860966ab4036e753 : SUCCESS in 1h 17m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8f6da557f221439d9ae472d10907e234 : SUCCESS in 1h 42m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2844e8ff5edc4caf9c083de335559d4f : SUCCESS in 1h 07m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e4f7f00c0d5f4c51adca6ae4ddeacc49 : SUCCESS in 1h 04m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3405cffd2f4d4b30a79f161543cf9f68 : SUCCESS in 36m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dbc42331056402a8e258d3483059d12 : SUCCESS in 38m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7465cc1167aa417481dcdeafd69db3e3 : SUCCESS in 13m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e12347de01b4ab5ade442a700d6c793 : SUCCESS in 29m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/398881d809674f6d9f0d3a8768d727e9 : SUCCESS in 29m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e12eb7f30d2840459281ef087468c37b : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1bdc523f6c84e80ba80ee7bc8387e81 : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53fb9ae3203440ce85be5bd067030134 : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecb6c3ff8c6449d2b034b7e13ee52aa6 : SUCCESS in 1h 04m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e6c0f54a67a471db456dfbc26840044 : SUCCESS in 1h 37m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8e5918c09fa04abe843dfb2a1ecd59ff : SUCCESS in 1h 16m 15s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f4b430ba4184290b95fd0b44b43d956 : SUCCESS in 1h 11m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e4abaceb39f94f43b6e16314f948f304 : SUCCESS in 1h 09m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2195ea01999b4486b482554183ade1fa : SUCCESS in 1h 28m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/77b73d28a2f241e88f5099fb5e52be10 : SUCCESS in 1h 26m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6cd127b669a4febaa4638c2f491bc8e : SUCCESS in 1h 07m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2d50f8facdc24af28e8b096af8fdbfc8 : SUCCESS in 1h 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5bf2ac74843b397e2a262770a4c8514020d8512c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 08:36:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e8ca68620df498d9e3ab1691e7f2632\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51b573fb69864a5c97be35c689da55f7 : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1096fa63568944228fbc216dd47619e2 : SUCCESS in 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b5d58a62afc82cfd53be6cd6f41cbe3b35b23f28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-10 08:47:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3aed26aed17b2456d339d044164b1814683d14f5","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-10 08:47:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"87247ff1d55ab00efd21dfb5b2f117068cbc00a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:30:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d09e12bcc2de4828ac37c37d42bc8f57\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c47ae8ffc0d4302a6d42093fd0ea003 : SUCCESS in 32m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c235a30444a545479f716a9eb955c653 : SUCCESS in 40m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fbec15daa50be7395044b43eda183ad1229bc3f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-10 09:47:50.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/04/810704/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"40a0407bc96d71b6a9592b4ae11ff08081bfa2a5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-10 10:27:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/810704/6 : SUCCESS in 1h 39m 10s","accounts_in_message":[],"_revision_number":6},{"id":"6614bf3ebc7eaf82f429d5c2afc9216967ecc0d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 10:31:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8784eab5b0246099329f9905b6fb5cf\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/185fd0ac44994a3782c42554209e8555 : SUCCESS in 1h 32m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/615f7b416a1d4e2182e3a19685132337 : SUCCESS in 1h 41m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f172178b340042ca88aa8550cb180915 : SUCCESS in 1h 41m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/75f092be01d744038ba6bda5be790815 : SUCCESS in 1h 16m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/76ea03972e024c0cbec3d9c9f800b5f4 : SUCCESS in 51m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b246f84c73894f54975db089552a543c : SUCCESS in 37m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c349d65b82604407834cbdcf6429c332 : SUCCESS in 9m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e05835ca0ea44e6adc0970f6d47ba15 : SUCCESS in 26m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3be5e6b70ab43bb972603aef635b146 : SUCCESS in 25m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9879f14c79c437a902fa977b8123fcf : SUCCESS in 25m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c019d2fb61ff47148a571f330b7a8511 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9730a79ae9e04ee0849c0c78bb58e26a : SUCCESS in 6m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1507cfebec3417494291e7b3d840174 : SUCCESS in 1h 23m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a5aaa325bb045058394547ed622f742 : SUCCESS in 1h 31m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/64707df6571547e4a2d6cd61a95ca144 : SUCCESS in 1h 06m 38s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2047ce32a24465dbd5fba3562609777 : SUCCESS in 50m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fbe1f393b51c4c03873538243fb7df6e : SUCCESS in 1h 10m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d10d9ee19f094712a6bc2cb98f2a45c3 : SUCCESS in 1h 33m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/695c76245da641fcb0364a8e8f3748db : SUCCESS in 1h 19m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b0638401b614aa4bf81dae3b4059b7c : SUCCESS in 1h 09m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c9d45fb4027741039c2af4ec73529ec9 : SUCCESS in 1h 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2dc0d59501ae513653ce8af8ebfe0243d33aae29","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-08-10 10:34:59.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"be119b181df9b65f6c3c01ae93b16a39c6ba1ae6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-10 12:56:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"54e0ef145ef14522839aea382425d7631826e624","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 12:57:05.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7156356c6f0821a767cfb1397776d2a17ae4b71c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 14:48:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f2609cd73db4f8ba04b180d5c25d0ba\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3a3cb98306d44ce6a10c7457e121e286 : SUCCESS in 1h 29m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3a24dc77b1ec44089a35caffc071878c : SUCCESS in 1h 10m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/11d64d3f7ecb45a3b2c0b16545cc9b77 : SUCCESS in 1h 16m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/279f8af46a7d48788adbca2cc9694151 : SUCCESS in 1h 09m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60bd4a0f150045cbb72f45e6cd05159a : SUCCESS in 14m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9078c81e69d482795d8ddafc3ee872a : SUCCESS in 24m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/342b123d24d14d2e8735403a8124eeee : SUCCESS in 23m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8866c39bddb4468194632754ee661dbd : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8229f13bd6774823922074f9a1a47cc8 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a46e7e621e2d4c9b896b7170c1de560e : SUCCESS in 1h 05m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec1f3be2edbe4450b1b7811cb3d93dd2 : SUCCESS in 1h 43m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5182f067073a46e488f7771a154e1ca5 : SUCCESS in 1h 19m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a66faca1d9e54d1697c6737385c1db96 : SUCCESS in 1h 02m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9256445d32e14107bafa9935dcb0b894 : SUCCESS in 1h 09m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea9392641e9140a2a574d687a116abd0 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":6},{"id":"4955c06b816f9e684573acb3c62e40f31562334c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 14:48:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"588ba94f63fb5977570b526225382a085e04e8c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 14:51:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/041d6d28e07944b2b9376fccfa75e19f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e66e53075f4d6f8ce3b13990b55f95 : SUCCESS in 1m 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df7a9a397ee04442aa08f6d900a060d6 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":6},{"id":"a4df7f654bc613aec1709333059612438c9a83e8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-29 12:26:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"efab5c834d48b7077f2a2d817814c7edf0c201ab","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-29 12:26:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"982c22dd46bb651c4c5c614262f1a9d68a96cd70","revisions":{"9585fd46dd3533b56833c24fa67622e070ce0e47":{"kind":"REWORK","_number":1,"created":"2021-09-23 15:06:31.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/04/810704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/810704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/810704/1"}}},"commit":{"parents":[{"commit":"d25cfaa702aca5b71fddf5aa1661a1ff1468e424","subject":"Merge \"Fix _create_dvr_floating_gw_port missing payload field\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d25cfaa702aca5b71fddf5aa1661a1ff1468e424"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-09-23 15:04:00.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-09-23 15:06:19.000000000","tz":60},"subject":"[OVN] Fix updating network segmentation ID","message":"[OVN] Fix updating network segmentation ID\n\nThe ML2/OVN driver wasn\u0027t handling updates to the segmentation ID for a\ngiven network. This patch fixes this problem.\n\nCloses-Bug: #1944708\nChange-Id: Ibe08bfbc2efc55b9d628cdd0605941b7486186b6\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9585fd46dd3533b56833c24fa67622e070ce0e47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9585fd46dd3533b56833c24fa67622e070ce0e47"}]},"branch":"refs/heads/master"},"918ac5f8d1ebd169e09b7c830f21cfd85de79308":{"kind":"REWORK","_number":2,"created":"2021-09-24 09:09:02.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/04/810704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/810704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/810704/2"}}},"commit":{"parents":[{"commit":"d25cfaa702aca5b71fddf5aa1661a1ff1468e424","subject":"Merge \"Fix _create_dvr_floating_gw_port missing payload field\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d25cfaa702aca5b71fddf5aa1661a1ff1468e424"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-09-23 15:04:00.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-09-24 08:53:29.000000000","tz":60},"subject":"[OVN] Fix updating network segmentation ID","message":"[OVN] Fix updating network segmentation ID\n\nThe ML2/OVN driver wasn\u0027t handling updates to the segmentation ID for a\ngiven network. This patch fixes this problem.\n\nCloses-Bug: #1944708\nChange-Id: Ibe08bfbc2efc55b9d628cdd0605941b7486186b6\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/918ac5f8d1ebd169e09b7c830f21cfd85de79308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/918ac5f8d1ebd169e09b7c830f21cfd85de79308"}]},"branch":"refs/heads/master"},"f97c57fd0fb93cca0a86fbee8718f9f09558fc4c":{"kind":"REWORK","_number":3,"created":"2022-08-05 11:16:04.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/04/810704/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/810704/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/810704/3"}}},"commit":{"parents":[{"commit":"b7fad3dd35ad23303a5f52ec64478860a137775e","subject":"Merge \"Implement specific tracked resource count method per quota driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7fad3dd35ad23303a5f52ec64478860a137775e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-09-23 15:04:00.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-08-05 11:15:45.000000000","tz":60},"subject":"[OVN] Fix updating network segmentation ID","message":"[OVN] Fix updating network segmentation ID\n\nThe ML2/OVN driver wasn\u0027t handling updates to the segmentation ID for a\ngiven network. This patch fixes this problem.\n\nThis patch extends the _update_segmentation_id() method to check on\ndrivers which does not inherits from AgentMechanismDriverBase, which\nis the case of OVN (which inherits from MechanismDriver). A new method\nis now called for those drivers to get a list of supported VIF types,\ncalled get_supported_vif_types().\n\nCloses-Bug: #1944708\nChange-Id: Ibe08bfbc2efc55b9d628cdd0605941b7486186b6\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f97c57fd0fb93cca0a86fbee8718f9f09558fc4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f97c57fd0fb93cca0a86fbee8718f9f09558fc4c"}]},"branch":"refs/heads/master"},"4c10e6f6baf362717c70d07354307aa13f1f70cf":{"kind":"REWORK","_number":4,"created":"2022-08-09 11:03:19.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/04/810704/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/810704/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/810704/4"}}},"commit":{"parents":[{"commit":"b7fad3dd35ad23303a5f52ec64478860a137775e","subject":"Merge \"Implement specific tracked resource count method per quota driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7fad3dd35ad23303a5f52ec64478860a137775e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-09-23 15:04:00.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-08-09 10:59:33.000000000","tz":60},"subject":"[OVN] Fix updating network segmentation ID","message":"[OVN] Fix updating network segmentation ID\n\nThe ML2/OVN driver wasn\u0027t handling updates to the segmentation ID for a\ngiven network. This patch fixes this problem.\n\nThis patch extends the _update_segmentation_id() method to check on\ndrivers which does not inherits from AgentMechanismDriverBase, which\nis the case of OVN (which inherits from MechanismDriver). A new method\nis now called for those drivers to get a list of supported VIF types,\ncalled get_supported_vif_types().\n\nCloses-Bug: #1944708\nChange-Id: Ibe08bfbc2efc55b9d628cdd0605941b7486186b6\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c10e6f6baf362717c70d07354307aa13f1f70cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c10e6f6baf362717c70d07354307aa13f1f70cf"}]},"branch":"refs/heads/master"},"e85f7dddc700f245dddedcc07ba745e8714791e7":{"kind":"REWORK","_number":5,"created":"2022-08-09 12:18:56.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/04/810704/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/810704/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/810704/5"}}},"commit":{"parents":[{"commit":"b7fad3dd35ad23303a5f52ec64478860a137775e","subject":"Merge \"Implement specific tracked resource count method per quota driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7fad3dd35ad23303a5f52ec64478860a137775e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-09-23 15:04:00.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-08-09 12:18:18.000000000","tz":60},"subject":"[OVN] Fix updating network segmentation ID","message":"[OVN] Fix updating network segmentation ID\n\nThe ML2/OVN driver wasn\u0027t handling updates to the segmentation ID for a\ngiven network. This patch fixes this problem.\n\nThis patch extends the _update_segmentation_id() method to check on\ndrivers which does not inherits from AgentMechanismDriverBase, which\nis the case of OVN (which inherits from MechanismDriver). A new method\nis now called for those drivers to get a list of supported VIF types,\ncalled get_supported_vif_types().\n\nCloses-Bug: #1944708\nChange-Id: Ibe08bfbc2efc55b9d628cdd0605941b7486186b6\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e85f7dddc700f245dddedcc07ba745e8714791e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e85f7dddc700f245dddedcc07ba745e8714791e7"}]},"branch":"refs/heads/master"},"982c22dd46bb651c4c5c614262f1a9d68a96cd70":{"kind":"REWORK","_number":6,"created":"2022-08-10 08:47:25.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/04/810704/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/810704/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/810704/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/810704/6"}}},"commit":{"parents":[{"commit":"492b68493f1fcc71ac3b09f42f9b97b1aa961521","subject":"Merge \"[OVN][Placement] Drive binding by placement allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/492b68493f1fcc71ac3b09f42f9b97b1aa961521"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-09-23 15:04:00.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-08-10 08:44:55.000000000","tz":60},"subject":"[OVN] Fix updating network segmentation ID","message":"[OVN] Fix updating network segmentation ID\n\nThe ML2/OVN driver wasn\u0027t handling updates to the segmentation ID for a\ngiven network. This patch fixes this problem.\n\nThis patch extends the _update_segmentation_id() method to check on\ndrivers which does not inherits from AgentMechanismDriverBase, which\nis the case of OVN (which inherits from MechanismDriver). A new method\nis now called for those drivers to get a list of supported VIF types,\ncalled get_supported_vif_types().\n\nCloses-Bug: #1944708\nChange-Id: Ibe08bfbc2efc55b9d628cdd0605941b7486186b6\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/982c22dd46bb651c4c5c614262f1a9d68a96cd70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/982c22dd46bb651c4c5c614262f1a9d68a96cd70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
