)]}'
{"id":"openstack%2Fneutron~811751","triplet_id":"openstack%2Fneutron~master~Ibe9178c4126977f1321e414676d67f28e5ec9b57","project":"openstack/neutron","branch":"master","topic":"bug/1945512","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-10-11 18:14:56.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2021-10-11 18:14:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe9178c4126977f1321e414676d67f28e5ec9b57","subject":"[HA] Do not add initial state change delay in HA router","status":"MERGED","created":"2021-09-29 16:37:24.000000000","updated":"2021-10-11 18:18:19.000000000","submitted":"2021-10-11 18:14:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811751-bug/1945512","meta_rev_id":"154679a7860b19f54340cb31f0fa8b31cc784d37","_number":811751,"virtual_id_number":811751,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-10-04 15:19:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-11 18:14:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-05 15:08:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-10-11 14:20:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-10-11 14:20:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-29 16:56:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-09-29 17:36:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-29 18:05:19.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-29 18:31:25.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-09-29 19:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 00:32:47.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 08:08:23.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-10-04 07:12:19.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"f92c86c68f615a4708391aedfa17883792952968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-29 16:37:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2f63164740833f0dea38e053a9e4aa6149c14cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 17:36:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0339afcd61464e20b818a0b2cbcfeaea : FAILURE in 44m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88212747a5904eb5bee09a385caff0ba : FAILURE in 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"740b7f65bb2df98c28b87315a2553d65e09133cf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-29 18:05:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/51/811751/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/51/811751/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cf758520941dd74557f5756cd1f300d063119f51","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-29 18:31:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/811751/1 : SUCCESS in 1h 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ab8763e241ee589cf2c8f26954e2b925df90b42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 19:07:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d2be226a663d47769ed4ab564611c211 : SUCCESS in 1h 41m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3af12c42fe0543adb778eec458be2f18 : SUCCESS in 2h 17m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91b89a6f4f664d7b899e88538e00bab4 : SUCCESS in 1h 19m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfc520d9eb054162aa3706e67e8daac2 : FAILURE in 31m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db8f8d887c98444592876fc5c4bdd5e9 : FAILURE in 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4edc2c7aa0347ccb419de06adbd6c4a : SUCCESS in 14m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/895501ef76884cfbb45971e103717dbc : FAILURE in 33m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98284c0056e843cd86f28c267ff78a37 : FAILURE in 26m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/643f487df550483ca6f44378bea34e54 : FAILURE in 25m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba83eae1b934d49ac7c57c23b078fcf : SUCCESS in 10m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4dc5d9acc0e429a85aeed3c11260888 : SUCCESS in 59m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/78709d0087644d72b385dbab68286a96 : SUCCESS in 1h 45m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d36c6bc9f6f24ea087be39dcc5761771 : SUCCESS in 1h 00m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/78d3c1b7fbd24b6da18fc835ca8cb5aa : SUCCESS in 58m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/21ed8ec0398f494491012c3c38582abd : SUCCESS in 57m 15s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6041a091ffe744e9a634c6fbb375ea14 : SUCCESS in 1h 07m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a023819f2401409684c733b66d0712fa : SUCCESS in 1h 07m 08s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/782c618c1f6049c2ac32076c6d09b1e3 : SUCCESS in 1h 19m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b6d983657d4db2805755efd1a63c24 : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":1},{"id":"716164905b0fbf3e91160730f6171aa26fc0a317","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-30 00:32:47.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/51/811751/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0c3d4279680b80a275b4f2fd5d09ae7b04be4ac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-30 07:53:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2a4bb5ee2f15255b6dfcf5cc4cc075f0d71360a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-30 08:08:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26f105f0299408b22b943556161f3768bd0485c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 08:39:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58f3f83690d64fe0a25bee25e035b8bb : SUCCESS in 35m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ed19c4039564c31839c630110832c11 : SUCCESS in 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b0a4a78d0b0aaccac90a8f36090841e4a6e78c4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-30 08:55:45.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/51/811751/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"723de834a9e20a9444f1fe84aaa77af9540ca04c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-30 08:57:11.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d346d2fd5cdd0e1116d0b84a624dfa2284f372c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-30 09:27:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a5836197b42d9ed5e23e81c0db74419a5c764ce","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-30 09:42:58.000000000","message":"Patch Set 3:\n\nPlease check https://bugs.launchpad.net/neutron/+bug/1945512/comments/3","accounts_in_message":[],"_revision_number":3},{"id":"37eb9a12c79cd065487fe27470ef709d256719ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:33:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5af7b38630e148698d7f15ed96e4d8b6 : SUCCESS in 55m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2e02577a26d4b90a5185427b5f7b96c : SUCCESS in 46m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f3c54e7a339f5873a7bba545ac2cb1410d815188","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-30 10:43:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"00e12a0ac9df3cef3b839a24845a8b1fd2640578","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-09-30 11:04:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/811751/3 : SUCCESS in 1h 35m 50s","accounts_in_message":[],"_revision_number":3},{"id":"7f839cca89cc4ca776ed79e322eb39b1a449487c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-09-30 11:35:56.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2021/Arista_Tempest_Testrefs/changes/51/811751/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"4a85a6f89073999b15328ea672aaae29864e5e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 11:45:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/387825cd9eb54bc185fe3b185cccaefd : SUCCESS in 2h 07m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e6f805fb88634a59a1c2abf11e6f1940 : SUCCESS in 1h 46m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ef2f742502a642229681e01e36f7950d : FAILURE in 1h 15m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98cb46d509ba46758b2d71cb61233762 : SUCCESS in 36m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/243810b8eccc4cec97ba5ad0720a0da7 : SUCCESS in 43m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/370d6661f3bd414c811b7f721b7b5372 : SUCCESS in 14m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b83a3d3a2e94daaab77c34c27b459fc : SUCCESS in 26m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7a08f18645d4f20bcee67f3fa3336f6 : SUCCESS in 24m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba7cd7ae93b942979171f69989d1e80c : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b56adbdf4ad41698ac762619c5b9f34 : SUCCESS in 12m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/823ed7014a43494eb373cc49f0fda152 : SUCCESS in 51m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/050f0b8a10ed49f6807268a909d0d1ee : SUCCESS in 1h 39m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/aa7077858f2f479385a6d39d3979a873 : SUCCESS in 1h 06m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/da52e14ad7054096bbd0c13f6837835b : SUCCESS in 53m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a412dff44a9c41ecb68b1fc49407b5d4 : SUCCESS in 54m 47s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4a050d25f5b849069d7e65b7f66acded : FAILURE in 48m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9b56dacb932640b9a0794b9e4033644c : SUCCESS in 1h 12m 39s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/ac2fa9938b1845b3b9eae7c15c007b0b : SUCCESS in 1h 26m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8975499487ad4845b207be6ca1985b4c : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":3},{"id":"70869553fbbc437a30a51458ebd251423ea90264","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-09-30 14:02:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d7542b6fe3ff76169f6ae45fcb866b7e2b850530","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-01 13:18:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f4f862ad3bd6d8a002f1567931230878622eb1e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 14:07:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69823f0c8ee546b49d9e5480870e517d : SUCCESS in 45m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a85786570c534debbe8096c3c373b60c : SUCCESS in 46m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ee07915c5de4601b93a88e97395c22dffc3ff0b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 15:41:10.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1aba9d4cc66f4ef3870fe64770144fae : SUCCESS in 2h 20m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/197701481b7d49889d31a16511759a29 : SUCCESS in 1h 48m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/95b1341c8e4c40c98b307367447342f5 : SUCCESS in 1h 32m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe123767183d4f389ffe767c8c8687ce : SUCCESS in 33m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/387d9d75db204646b5ce5f508af54afe : SUCCESS in 35m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5661ab532dd040e6977abf3e138eaef6 : SUCCESS in 12m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44a1d3a8ce92491795d4c54fa4c5e6ff : SUCCESS in 25m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d37d454141d411587c4b3c3329bfe93 : SUCCESS in 23m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3e76d42fc7449c3a98fd358522b4210 : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c43130541494174adaea14a958094d3 : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/12a0fdcea89244d4b461f04b59bb37b0 : FAILURE in 45m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2a496f5ce984ce4a1f254ca9a3cc26d : SUCCESS in 1h 38m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8bf8b4e64e4d4e1ca25a6f87ff990a41 : SUCCESS in 1h 09m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc77662950e14cdf86f7544d11d7718b : SUCCESS in 53m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9e3fc10d5819406b8a5a2669b2643ac2 : SUCCESS in 54m 27s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/002206b90864430d8708930ce573d876 : SUCCESS in 1h 08m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8779fbdb7a7242a89de27bc70e03a957 : FAILURE in 2h 04m 54s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/583db7f4174a4c33b511f09d2d215321 : SUCCESS in 1h 20m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a87bac5df9b4c0ebd1366a441d31e25 : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":3},{"id":"8fcc618e1d6c1a9ab3dbfc55635611572d6778de","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-03 06:41:34.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/811751/3 : FAILURE in 41h 22m 17s","accounts_in_message":[],"_revision_number":3},{"id":"b16f3d510f2f684cd977a06a068306693d1fe170","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-03 09:12:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6bbc19fe2783731dd76ce24b6e854ff5036b152c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 10:15:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5969c551a87f49d1a727600d19a740f4 : SUCCESS in 30m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/449837b5c37949d6ab91d993b7e355dd : SUCCESS in 42m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a11e0efd42895efdddae5d6610f06262f8929270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 11:23:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/14127899bf69488696ef783ece7d7df6 : SUCCESS in 2h 10m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d2cf47c74caf4545abe9a890eea898fe : SUCCESS in 1h 42m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a41858c1488c4d3796252077ac05cc81 : SUCCESS in 1h 42m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/249fdae3c27f4a31b0eebf68c166602b : SUCCESS in 31m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdbf0db6d50f43f991b742f596c3fce7 : SUCCESS in 37m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb41e82dcbd84e008b2ec51175bd7e52 : SUCCESS in 16m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/380a7bcba69c45268aa45905f0a2a8fb : SUCCESS in 27m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e714a43d596426baef5386259d736e0 : SUCCESS in 23m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b3a7ef8264a43ea8f7a99a25922605f : SUCCESS in 27m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0114b39b5b4f578aabed11322ff421 : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/61195e260b864636b07ceb15cf510ff3 : SUCCESS in 46m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/df80d92c68bb4f9685f3d3f2e020b0ab : SUCCESS in 1h 43m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/035a24a690224d658d0b3adb754d5e06 : SUCCESS in 1h 20m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a7a8e5e786c4c56a3769ab120e12091 : SUCCESS in 48m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cc990eefde674ba48810b4a1f5df4937 : SUCCESS in 50m 45s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fb2a5daf3d2e4278a0969e46a9412f9f : SUCCESS in 58m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/781390af8ab14f0d93be410dfe4453ed : SUCCESS in 1h 19m 34s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/baef037236a14bb2a971e427fe2777c0 : SUCCESS in 1h 33m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fe8728055d549b2a6dde98deff12101 : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":3},{"id":"850a3199af5aaf7fb10a0445e27cc346ab17ca40","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-04 06:49:45.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/811751/3 : FAILURE in 21h 36m 49s","accounts_in_message":[],"_revision_number":3},{"id":"89bc0aeafbfc65240c8f94d3f49b28b7f368dca3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-04 07:12:19.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6089039da78936371a88bd0142e19103fc85bfd9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-04 11:35:53.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d8125f45a9841c2b0b0ee3e87e1702a4a34afcc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-04 14:28:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0b41f9d2d60ef250bf8db34711cd854c1ca9d9b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 15:16:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fc7b84375fb49b2a868b9d7afe0db0f : SUCCESS in 44m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25c408ef3f954e08af5fd222c02f226c : SUCCESS in 44m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"25e726cba4eeaca35abf5946e0b70979edd7ac1b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-04 15:19:14.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/51/811751/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"cb8b476aa7b3d7a5e767716cbfb24ddea23e9c6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 17:19:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ae41e38df1674d71913d742602cf8a53 : SUCCESS in 2h 47m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/026f03f6f9894bebbf5b3ce1136eddf3 : SUCCESS in 2h 25m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e82ccd6ef6e4ec2b9d2331cf8d6184b : SUCCESS in 1h 31m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f10f9d0afec843a6998e1be4f71abae2 : SUCCESS in 41m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85fce8f0d8d5448e827471619b37ec27 : SUCCESS in 40m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3853d83a2220439d91addc179fbce559 : SUCCESS in 15m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d78032560a4140638c0dcbfd4d52de29 : SUCCESS in 32m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05ce032ceb2c4666952c1e6146935ed9 : SUCCESS in 30m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be6d4bed8d1140c191b0a94c9524e8e9 : SUCCESS in 30m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a54468b04b42478c6810a0b789e08c : SUCCESS in 10m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1664cab5ed714cf98210158f88b27f68 : SUCCESS in 54m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/493bc426af2843019d4df0d341012bc6 : SUCCESS in 1h 41m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b4a0aa8936b14aa7a12db3d2e27db6b1 : SUCCESS in 1h 12m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccaae92fe1b941c2b4758f0e27fc825f : SUCCESS in 58m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2b6786df39f347a394d8b64faf78e9ba : SUCCESS in 1h 02m 16s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/14ec799c368a4eceb21c480b3494677d : SUCCESS in 1h 12m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7a5afb3b4d2a4f02bc6afabc204f8a82 : SUCCESS in 1h 15m 10s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/9c86fcdda4c04dc29df7005ffced97bb : SUCCESS in 1h 35m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d78ce3254b74bebb74d2dae503c0a78 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":4},{"id":"8bcfa76c4521d84c5f08465e5b77c8848a0d29be","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-05 15:08:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f7ba86536a6023f012b52bee87165b9ada5a1af7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-10-05 16:30:21.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM, but i\u0027ll let Oleg look as he had some initial comments.","accounts_in_message":[],"_revision_number":4},{"id":"e308911562ff6db054846f54d662d2986a227a53","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-11 09:48:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca9a2c58132ac2365ee1f6c3c556a4f65e903e03","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-11 09:51:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aa68995507c4b93af3abadec31416fa4db1c2f39","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-10-11 14:20:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"42406c86d05fbeb3e9e5113695f0f3ec332286ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 14:21:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5062eace50d09354fdcb5149ec461b554a1c930c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 18:14:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/85769466a0aa4bdc92fed27efa63c32b : SUCCESS in 2h 38m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ed326da38e1c423baeb60e4315cc2db3 : SUCCESS in 1h 59m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/23166cb27485454aabaee522c0d8f31e : SUCCESS in 1h 26m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abaa08de945e48a4b29db02d97cdd288 : SUCCESS in 41m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56c4dbffff147609de743af3f615641 : SUCCESS in 15m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8af157bb2edd4d50add2bc888ce59396 : SUCCESS in 26m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa7e830aab654400a57ba26183a2e38d : SUCCESS in 28m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727f3f6bf14e4e5ba0750681edefe92d : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b2f5ad08d104227ae10412e153150c6 : SUCCESS in 46m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eddc8733e393496889f538c6e99685e7 : SUCCESS in 1h 41m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/338f22519fc341c68b60d87be92ac2cd : SUCCESS in 1h 15m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0d5794c155b46998b12737cc319b71e : SUCCESS in 57m 40s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/cefd7fb21c044e1380550949bba3d9ae : SUCCESS in 1h 32m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1dcb4c9a8e1043b9b7a952ba771517b7 : SUCCESS in 59m 38s","accounts_in_message":[],"_revision_number":4},{"id":"36377a07b7299da29aacf3e8f31cd8167f332043","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 18:14:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"154679a7860b19f54340cb31f0fa8b31cc784d37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 18:18:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba97eb7d8f34750a27d12770948f284 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c20f2e5136fd241f4be5c37403ab1ed54cdaefb5","revisions":{"0b651cfeae6a2ed99d0193e5e38a7db64c2e62d3":{"kind":"REWORK","_number":1,"created":"2021-09-29 16:37:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/811751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/811751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/811751/1"}}},"commit":{"parents":[{"commit":"7c87ecb679928efffe4985b46c2305e638afcbe3","subject":"Merge \"Replace cirros 0.4.0 by 0.5.2 in ovn migration create-resources.sh.j2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c87ecb679928efffe4985b46c2305e638afcbe3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-29 16:28:28.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-29 16:34:38.000000000","tz":0},"subject":"[WIP][HA] Do not add initial state change delay in HA router","message":"[WIP][HA] Do not add initial state change delay in HA router\n\nThe initial state (\"primary\", \"backup\") should be set immediately.\nin [1], a transition delay to \"primary\" was introduced. This delay\nis unnecesary when the first state happens.\n\nCloses-Bug: #1945512\n\n[1]https://review.opendev.org/q/I70037da9cdd0f8448e0af8dd96b4e3f5de5728ad\n\nChange-Id: Ibe9178c4126977f1321e414676d67f28e5ec9b57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b651cfeae6a2ed99d0193e5e38a7db64c2e62d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b651cfeae6a2ed99d0193e5e38a7db64c2e62d3"}]},"branch":"refs/heads/master"},"81f9eed31405074cbf5d3035c335b4916abc8fbb":{"kind":"REWORK","_number":2,"created":"2021-09-30 07:53:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/811751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/811751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/811751/2"}}},"commit":{"parents":[{"commit":"7c87ecb679928efffe4985b46c2305e638afcbe3","subject":"Merge \"Replace cirros 0.4.0 by 0.5.2 in ovn migration create-resources.sh.j2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c87ecb679928efffe4985b46c2305e638afcbe3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-29 16:28:28.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-30 07:53:03.000000000","tz":0},"subject":"[HA] Do not add initial state change delay in HA router","message":"[HA] Do not add initial state change delay in HA router\n\nThe initial state (\"primary\", \"backup\") should be set immediately.\nin [1], a transition delay to \"primary\" was introduced. This delay\nis unnecesary when the first state happens.\n\nCloses-Bug: #1945512\n\n[1]https://review.opendev.org/q/I70037da9cdd0f8448e0af8dd96b4e3f5de5728ad\n\nChange-Id: Ibe9178c4126977f1321e414676d67f28e5ec9b57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81f9eed31405074cbf5d3035c335b4916abc8fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81f9eed31405074cbf5d3035c335b4916abc8fbb"}]},"branch":"refs/heads/master"},"b0d420bb8577b8fa61caa35082982d49077d5c0c":{"kind":"REWORK","_number":3,"created":"2021-09-30 09:27:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/811751/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/811751/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/811751/3"}}},"commit":{"parents":[{"commit":"7c87ecb679928efffe4985b46c2305e638afcbe3","subject":"Merge \"Replace cirros 0.4.0 by 0.5.2 in ovn migration create-resources.sh.j2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c87ecb679928efffe4985b46c2305e638afcbe3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-29 16:28:28.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-30 09:24:26.000000000","tz":0},"subject":"[HA] Do not add initial state change delay in HA router","message":"[HA] Do not add initial state change delay in HA router\n\nThe initial state (\"primary\", \"backup\") should be set immediately.\nin [1], a transition delay to \"primary\" was introduced. This delay\nis unnecesary when the first state happens.\n\nCloses-Bug: #1945512\n\n[1]https://review.opendev.org/q/I70037da9cdd0f8448e0af8dd96b4e3f5de5728ad\n\nChange-Id: Ibe9178c4126977f1321e414676d67f28e5ec9b57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0d420bb8577b8fa61caa35082982d49077d5c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0d420bb8577b8fa61caa35082982d49077d5c0c"}]},"branch":"refs/heads/master"},"c20f2e5136fd241f4be5c37403ab1ed54cdaefb5":{"kind":"REWORK","_number":4,"created":"2021-10-04 14:28:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/51/811751/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/51/811751/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/51/811751/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/51/811751/4"}}},"commit":{"parents":[{"commit":"7c87ecb679928efffe4985b46c2305e638afcbe3","subject":"Merge \"Replace cirros 0.4.0 by 0.5.2 in ovn migration create-resources.sh.j2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c87ecb679928efffe4985b46c2305e638afcbe3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-09-29 16:28:28.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-10-04 14:28:22.000000000","tz":0},"subject":"[HA] Do not add initial state change delay in HA router","message":"[HA] Do not add initial state change delay in HA router\n\nThe initial state (\"primary\", \"backup\") should be set immediately.\nin [1], a transition delay to \"primary\" was introduced. This delay\nis unnecesary when the first state happens.\n\nCloses-Bug: #1945512\n\n[1]https://review.opendev.org/q/I70037da9cdd0f8448e0af8dd96b4e3f5de5728ad\n\nChange-Id: Ibe9178c4126977f1321e414676d67f28e5ec9b57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c20f2e5136fd241f4be5c37403ab1ed54cdaefb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c20f2e5136fd241f4be5c37403ab1ed54cdaefb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
