)]}'
{"id":"openstack%2Fneutron~813610","triplet_id":"openstack%2Fneutron~master~I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972","project":"openstack/neutron","branch":"master","topic":"bug/1946318-always-learn-arp","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2021-10-18 10:01:15.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2021-10-18 10:01:15.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2021-10-18 10:01:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972","subject":"[OVN] Tune OVN routers to reduce the mem footprint for ML2/OVN","status":"MERGED","created":"2021-10-12 11:22:05.000000000","updated":"2021-10-18 10:03:08.000000000","submitted":"2021-10-18 10:01:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"813610-bug/1946318-always-learn-arp","meta_rev_id":"e327a9b884f201a90e9dedc30415d1aae7b6d001","_number":813610,"virtual_id_number":813610,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-10-14 14:41:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-18 10:01:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2021-10-18 07:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-10-14 15:34:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2021-10-15 10:30:07.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2021-10-18 07:35:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 12:11:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 12:14:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-12 14:32:49.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-10-12 15:07:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 15:31:17.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-10-12 16:26:57.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2021-10-14 08:33:22.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"CC"},{"updated":"2021-10-15 10:30:07.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2021-10-18 07:35:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"2993759dfe6ff0c457cd5562bff4fed93dcf6ddb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-12 11:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14db764f377be3dbba7c278651950d4fadf5636e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-12 11:27:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e7f77c4a5cd4d81bae3c77b38870011b6afb2abf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-12 12:11:28.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"76eb17a73d6e0e596c61e54c5dc62b695b904f4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 12:14:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a179e7db4b55497cb82dd9508e0761ab : SUCCESS in 42m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b39925da9574d2ca63743161bf36cd4 : SUCCESS in 44m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4421374a6a3f844a09b29af4f7a1715276e0cc1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-12 12:45:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0db9722e0521e0158482f0432b0efeaf4c3a9559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-12 12:47:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8235461555744a4e01acc72a87ce4c34abc0f89d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-12 13:04:50.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"74a1826c83ab97a4c5bcf64cd3eee20e06ee6850","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 13:33:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17feb7c39f5e4401a26bbf0f645bab48 : SUCCESS in 44m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5e16a822ea4440d8e370fa86bbd9690 : SUCCESS in 43m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8f3984cadf00579795d1624cd9bd489f6233c948","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-12 14:32:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813610/4 : SUCCESS in 1h 44m 44s","accounts_in_message":[],"_revision_number":4},{"id":"becdfb83b6180765895e55e4f794e55c46d0797f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-12 14:55:42.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d255509fc08fe4d94697425dad0b2c8c80596e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 15:07:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4dd98e9870614f0491abedc3cfd9eee9 : SUCCESS in 43m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4fa846d1f7974c138170133e61db8406 : SUCCESS in 1h 53m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b515c87e5dfb42719eb978b396895eb4 : SUCCESS in 2h 17m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f2418cc208e34b35939ea3beeaaa56d9 : SUCCESS in 1h 43m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/03786a0e79c644018cce97571e99af17 : SUCCESS in 2h 07m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6920f6cff1c74624b470e13574f5f24d : SUCCESS in 30m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f4e261dbc9a49819119c1dd0469b8f2 : SUCCESS in 33m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30b322854ee9448b9a8e5b9a502eca8d : SUCCESS in 35m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5ec4ea0a414a2fac6458398cff4349 : SUCCESS in 14m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd7d7ed17f8a4cdfb45c2472ad4570e6 : SUCCESS in 31m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/889780fb0a2547da95d2ef1f57d3221a : SUCCESS in 30m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa185ad7f5bb4476bd2ea23cfd623aa8 : SUCCESS in 22m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0e6a2edc414c60b2beeced1c5e6532 : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a9e731aab9a4a7ba1186b4ee1d53569 : SUCCESS in 45m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf27bfb999964550934dd10b1014432f : SUCCESS in 1h 40m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb3d091390ab4c1bb5d19b1fd57283b6 : FAILURE in 1h 30m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d6cdc58de444c059939ac8ebe8240e3 : SUCCESS in 1h 02m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/824c05767bdc45d7b4b70ed2cc8feaed : SUCCESS in 56m 13s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c224f76ad07b4e61b10b25902fec61a8 : SUCCESS in 1h 00m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c001d5f209714aca9ae6b01112cfd07c : SUCCESS in 1h 17m 30s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/7389ac5a84fa484faee5d31b59c52538 : SUCCESS in 1h 32m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57a7f2d99e1849ba9d6021fa943a82b8 : SUCCESS in 56m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bab31de9d5a94ae79b7137744ba00f8b : SUCCESS in 1h 19m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6e3b2f15f52d49e8976164b0e04b0c80 : SUCCESS in 30m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7132d5a9844841be89c38aecffabb88f : SUCCESS in 1h 18m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/762b887e9652490e927361c095858aaa : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":4},{"id":"907f7b88655a2dcd17717c992f2c48daab3e6a95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-12 15:30:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a3f4b429e069e2e46c5d009afd3d39b7306dd3c4","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-12 15:31:58.000000000","message":"Patch Set 5: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"127ea783d45c5a724272e72485f293783b45a8aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-12 15:36:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7b0090435b905215cb61abaa342c12c7ae3303c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-12 15:39:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b517124bd791aca9e1087ff49f722c6da8f088ba","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-12 15:45:30.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"66092305277b515bfe8e1f644280cb7fcd45719d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:26:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8aa830045184070b055cda73dd55135 : SUCCESS in 45m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a46b715b5b924f2fbd48b1e3b5da25b3 : SUCCESS in 45m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9ca761cabd7177dfe25bd81f76ec396c8cdbdf44","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-10-12 16:26:57.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"86db7d706550158a36db3be595af250a6accd4dc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-12 16:43:42.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"87197332322188cf8e17f1d0a46ab4736be542fb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-12 17:10:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813610/7 : SUCCESS in 1h 30m 45s","accounts_in_message":[],"_revision_number":7},{"id":"8bf5126979d8ad840521016d51f6bdf1182f8177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 17:32:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4989d4dce50b4fdf9ed1b0bba5bc82d1 : FAILURE in 38m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ce0465ab058f4b6ea02d45f145b1657a : SUCCESS in 1h 37m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c214b163b69d497c913e93af27a85489 : SUCCESS in 1h 25m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2b5f943a0354eb384518b9706a3fea8 : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6c60e0bb135f4489b1f592200bb895fe : SUCCESS in 1h 51m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b5513c8f934949ccaf04d1f1c39e59c6 : SUCCESS in 28m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a4e9764d8a94199b6b8a0202caa74e0 : SUCCESS in 29m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65dd6aa9426d4aa5b1a7c3ca2d43b1d7 : SUCCESS in 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a907aff415a4f84a354e9346d345f30 : SUCCESS in 11m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/839ec4f81afd4b38afa30f8ded7bd666 : SUCCESS in 22m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6181e1f1104402caf8ecb6ca7c08e78 : SUCCESS in 21m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf70d4a650144fa4b61e9b2a7f18d7d2 : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c57dd40110c049d5ae90d1f6cba5a5ca : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1971ef2c933481da7031bad0f5c3909 : SUCCESS in 46m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad3fc83d08f04de49b05e553532f61fc : SUCCESS in 1h 37m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ae9f1b02fb23430d8683ae2dd03255b2 : FAILURE in 1h 07m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ff8e23c1fea43c99afce9fe8e64b57f : SUCCESS in 53m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e10063fb06c0402db1a88c50625b1110 : SUCCESS in 58m 32s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8089fdd02f9d4dcba844a974b7e9efb6 : SUCCESS in 1h 12m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/118a1db23a0e4c6fbfd88729012443be : SUCCESS in 1h 13m 24s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/fb5ef1f0114248339690d523157214f1 : SUCCESS in 1h 25m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f04057cdd3344a09d31dc3a78d91589 : SUCCESS in 1h 10m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d6afd672a6374321b638795d2f047ec6 : SUCCESS in 57m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0cd7bef8123b4c3db1db838482f4db31 : SUCCESS in 25m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c1e3e18ddbbb4118ae60e3422a052563 : SUCCESS in 1h 08m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b91d7c1737e74fa382139c0df5b8234c : SUCCESS in 1h 10m 28s","accounts_in_message":[],"_revision_number":7},{"id":"9caac4e8208e548f623a766536929e23c0fc5afc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-12 19:24:34.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"d5822d678a71e450851cb7749b69c01aa9ec2714","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-12 20:12:07.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"985f47d389f1a4653040a7b87dc6172c58270129","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-12 21:05:38.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"8f7076049b026a676a6a549f4f98b78bb62df9ca","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-13 08:03:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9f40e8d2d623fb1a058d4ccdb9b5c8d7e5b8d89e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-13 08:09:52.000000000","message":"Patch Set 7: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"659d6b113f74af5fffc51de2ca9635b2149e04dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-13 08:13:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4d3c69e723b1a99ec16e919c9656117eef3d3a18","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-13 09:17:42.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"f927c18f8ab22160bb13f31d6af7bd9b5c0a7dee","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-13 10:03:11.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813610/8 : SUCCESS in 1h 48m 33s","accounts_in_message":[],"_revision_number":8},{"id":"8475dc508f1b525e44661e48a442c36177741446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 10:42:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1ec550245851408d86c7dc96f311bd9d : SUCCESS in 57m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/acb83e2889a64b659078fa6179a4f8b3 : SUCCESS in 2h 05m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9ec1f4bd6e884dfc916dbe2440cf9f72 : FAILURE in 2h 22m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/34ec7865890d49d0b2daaba81adbc63e : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a949c845408644aa9c8c24c22d5d58c8 : SUCCESS in 1h 41m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d65b03f4fe444069b1597bb338c7a560 : SUCCESS in 36m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f91017f4f954fc7a2acb11356a78e6c : FAILURE in 34m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7312a82780a54e65a75418ba17338df9 : FAILURE in 39m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1b749ec74743b9b1670a39e53c5af8 : SUCCESS in 13m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/142dadb04e324c15bbcd56e4ed2e99f3 : POST_FAILURE in 26m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42b186e6fd294892b5411262b1c627e7 : FAILURE in 22m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/023a8f629d57458aa0761d671c0d28d1 : FAILURE in 23m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebc4d7cb37d4dc9a3e5f3e77e9e0f5b : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/327e2901491b4ce0b3fc43b24adeb9fd : SUCCESS in 52m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ecc3a064cb14722a4e2485bd1545e63 : SUCCESS in 1h 44m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/88137d71e6ad457f8879497d5786f463 : POST_FAILURE in 1h 06m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2dea2362bac48f3bbd48545ce2ba1ae : SUCCESS in 1h 00m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a3d5e0e719384261bcd5820b5955790b : SUCCESS in 59m 13s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4a23b1d6d42946149e19942baba11cc0 : SUCCESS in 1h 19m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/146b5de8165f41cba8f12d9d82adac1b : SUCCESS in 1h 14m 46s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/cbb53f67026e4d8e86ab94d877fe3b26 : SUCCESS in 1h 27m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b536662fcec4dcabd7b27a181675c65 : SUCCESS in 56m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/da58b945773848c0a157c45d577335be : SUCCESS in 55m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ca69a760b3ad44cb8a0a58911c3c6e8b : SUCCESS in 40m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3d5bc2749920417abd8b8236b6d33d87 : SUCCESS in 1h 08m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/04e5062be6fa43ab823c07a9dc1e1935 : SUCCESS in 1h 13m 22s","accounts_in_message":[],"_revision_number":8},{"id":"549a8759f96d5182772a04c28b58ece49eeb3737","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 10:56:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5cd8bfd77da243a6b1465993809d71ad : FAILURE in 42m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8602e255034411299fe8a8eeac20875 : FAILURE in 41m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8e63feafe30507fb03aec7e2143968464333378e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-13 10:57:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a4ad8952892fe2dcfabdf952e35739e8621806e5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-13 11:52:41.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"98dfef7378e36985951c34bc9c4e5f687c00b22e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-13 12:35:29.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813610/9 : SUCCESS in 1h 37m 10s","accounts_in_message":[],"_revision_number":9},{"id":"9f50a5c305f836e937d12d32b90d282177c7726e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 13:22:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9317fbc77d7a4642aef56a6539d4605a : SUCCESS in 45m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9484bfcef5543f28e072ba473458c1d : SUCCESS in 49m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8f5ca38b134528db447beeae318360fd63735754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 13:55:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6e1209f03c8d4d3a9c5009052fbb1216 : SUCCESS in 46m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/031ee37e1a6c4ae4afe482a14df15971 : TIMED_OUT in 2h 52m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/00cebb72f4644d569676558c55825102 : SUCCESS in 2h 32m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c03e6b992310455387295ac212f169cc : SUCCESS in 1h 30m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f45b71d2644b4ea89bca9ce55ab0c273 : SUCCESS in 2h 42m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ff81a750ef0462aa29d6f8e77396869 : SUCCESS in 35m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5ec5b2e5f4e43999849ac83a69298a3 : SUCCESS in 40m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06445eb094e04d63b59fb490bb772c08 : SUCCESS in 43m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17072709c4a24d39a108a450f6fb5332 : SUCCESS in 14m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b51b09ee87d744ef8964254960eaf86d : SUCCESS in 32m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77d14bb182f34b29850cdb342282a05e : SUCCESS in 28m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a04c33538bdc48edb6a3b0d2b5e68b4e : SUCCESS in 27m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09964ab447ea4130a86e67e2ac2b2926 : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/27c28c1b3c2847a38ef4659019502b22 : SUCCESS in 52m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3163ade6dd84a90a8db3d7ed4f20a20 : SUCCESS in 1h 52m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4c86a007648e4bbbbe1693cc9f969006 : SUCCESS in 1h 14m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ae57eb8c85a43b08eb9251b3f932457 : SUCCESS in 1h 01m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cb20ca042c5f4917aafbefa1b8aed195 : SUCCESS in 1h 00m 27s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1dac345212ee4be2bc8bd954b33b84bd : FAILURE in 59m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4334a6841d6b4570878fa31db1c7af0b : SUCCESS in 1h 14m 03s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/22fc45acf2be4e428055e4404cbab0f5 : SUCCESS in 1h 30m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c7a03f68dce499db8d224cd04b63b41 : SUCCESS in 57m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4430491f90884453be19e65075afaba3 : SUCCESS in 58m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3dd432902b85496d8378c42d528acdd5 : SUCCESS in 32m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b4325eb21ea4ef1bbc8537f495c4be3 : SUCCESS in 1h 13m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fed01695ac6b4c5b842dde73624ecc8f : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":9},{"id":"76247bc3b1eb4fb8cc4f1cda1fc65845a07c97db","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-10-13 16:02:09.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f8943803d92821daefb8f7fcc7a7a62eacb9c866","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-10-13 16:02:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"72680f38f64477ce09cbaac7845598ad98a0aacd","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-14 08:26:09.000000000","message":"Patch Set 9: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"02171e4374b6416231e2a19243d9b7326dc0af2f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-10-14 08:33:22.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c29ff11ea29f2dd20d6bbb0c0edd4b114bba33ac","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-10-14 09:05:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"276f21d6e58b79f36ceefd4c705eb8604cb8bc68","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-10-14 09:08:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9948d57dee163285ef4afb510e1150ed20cc66bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-14 13:48:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e7edc3b0a92d61e203d7cef91d52c18baf966812","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-10-14 14:01:52.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"27037018f8c683b4ab66fb3874b800e0280df212","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 14:37:43.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11e0fb0857d74351a32eb2900a331d19 : SUCCESS in 31m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29c0b932174e41eb9f86ec95d5151ecc : SUCCESS in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b29ffb667a06453500b0d60075cd11758b4b7f82","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-14 14:41:49.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/10/813610/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"f5a68e54b039600d8b125b5c2ece003763cf1873","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-14 15:26:02.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e93f7ecacc25eb2f1a8d232092430c74df13031d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 15:33:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5a867bf550da42c797f1f7bcc9b271c0 : SUCCESS in 46m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f25b144f61d24c6fb7705fbb2da873c2 : SUCCESS in 1h 35m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06d6502270a342a88cb39f944973b50b : SUCCESS in 29m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e80df83d6e046678e1211829fa426ff : SUCCESS in 30m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2051b1a04da415bb96c33ce7108269b : SUCCESS in 17m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88ba6ff13fea47009e73124f92952bfe : SUCCESS in 46m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f5ff50b5e1c4cc2934a315bd5c272e3 : SUCCESS in 21m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8cd8f37f9934352ac6cd7362e8ac347 : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/182ced04455f4f8dbaf1ef7e470a6d39 : SUCCESS in 9m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c657aaaf3cc46fd82c11a73c75be76f : SUCCESS in 52m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a0ad603db0a248b0b50dca2240ed34e5 : SUCCESS in 51m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e63b42faa496404c888994cd4590ef83 : SUCCESS in 26m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5d027b49c83741a3b1a7b3d575bc49c0 : SUCCESS in 1h 37m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e13b1bf33f614b54ad9445980edaa283 : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":10},{"id":"ab2d36e461e906e5a4875f405aba7346a085f038","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-10-14 15:34:28.000000000","message":"Patch Set 10: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6f8999d90a0a78870d3b30a4b80622294a0fb15d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-14 15:34:54.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813610/10 : SUCCESS in 1h 45m 21s","accounts_in_message":[],"_revision_number":10},{"id":"766218897d5ca7156da1253e1f5ea6bc7d0a86ba","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-14 16:10:11.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0feedc3ac99734518bb6da5d6a9c1d86fac520ee","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-10-15 10:30:07.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3703ecf62ada775c3920e1b03c2592fad1c1caa3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-18 07:35:22.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"2cc7986cd49fca0873887818ee3f20a3728cdfa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 07:36:02.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"00976c9aebf6c7106875c31b75b2921cf0add694","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 10:01:14.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/918ff36640d84a17acc66e5767f28d0d : SUCCESS in 40m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dbfd4411f5754bb0b0ed2c4c2dec670e : SUCCESS in 2h 11m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7dec0fe74a1b419d9b2c959d7a0e3331 : SUCCESS in 33m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c8fa0b574f64d46b77616fa8995a45a : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3eff00b8f80441caff0d244ae9ae93c : SUCCESS in 24m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10ea2eed122b46808daf47b0e1c57fc4 : SUCCESS in 27m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ad188e5aeb047a09dcd9d757d097796 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/57bc211157ac426fbae0e9db251a0ceb : SUCCESS in 53m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2fb9971f7d8248cf89b75552ad9698cf : SUCCESS in 56m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dabf082bdf294d668650e6b87d22e887 : SUCCESS in 28m 13s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6e0718c953a4449ca29c7efc3ffa5a4b : SUCCESS in 1h 10m 59s","accounts_in_message":[],"_revision_number":10},{"id":"a84b98a2ec07b7383e785e7f970f96d97ef0b84f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 10:01:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"e327a9b884f201a90e9dedc30415d1aae7b6d001","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 10:03:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af3043633ce48edaa595a570eec2f61 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a278c5ba789c014ec777a75fc9538179d6707202","revisions":{"0086d631fa71b888a4d230e13111b581d7f371a2":{"kind":"REWORK","_number":1,"created":"2021-10-12 11:22:05.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/1"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 11:21:55.000000000","tz":60},"subject":"Allow setting always_learn_from_arp_request for routers","message":"Allow setting always_learn_from_arp_request for routers\n\nThis patch adds a new configuration option for ML2/OVN called\n\"router_always_learn_from_arp_request\". This new configuration is\na passthru for the Logical Router \"always_learn_from_arp_request\"\noption in core OVN and  allow operators to avoid having a MAC_Binding\ntable explosion (helping lowering the memory footprint) in core OVN\nby setting this to false or true (default) based on their deployment.\n\nFor more information on this option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0086d631fa71b888a4d230e13111b581d7f371a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0086d631fa71b888a4d230e13111b581d7f371a2"}]},"branch":"refs/heads/master"},"50bd615cfea8a511038c9d68764c56c00b24dd5c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-12 11:27:53.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/2"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 11:27:39.000000000","tz":60},"subject":"[OVN] Allow setting always_learn_from_arp_request for routers","message":"[OVN] Allow setting always_learn_from_arp_request for routers\n\nThis patch adds a new configuration option for ML2/OVN called\n\"router_always_learn_from_arp_request\". This new configuration is\na passthru for the Logical Router \"always_learn_from_arp_request\"\noption in core OVN and  allow operators to avoid having a MAC_Binding\ntable explosion (helping lowering the memory footprint) in core OVN\nby setting this to false or true (default) based on their deployment.\n\nFor more information on this option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50bd615cfea8a511038c9d68764c56c00b24dd5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50bd615cfea8a511038c9d68764c56c00b24dd5c"}]},"branch":"refs/heads/master"},"304db11a2afb94d5e9a182f6e6331674955524c0":{"kind":"REWORK","_number":3,"created":"2021-10-12 12:45:05.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/3"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 12:44:56.000000000","tz":60},"subject":"[OVN] Allow setting always_learn_from_arp_request for routers","message":"[OVN] Allow setting always_learn_from_arp_request for routers\n\nThis patch adds a new configuration option for ML2/OVN called\n\"router_always_learn_from_arp_request\". This new configuration is\na passthru for the Logical Router \"always_learn_from_arp_request\"\noption in core OVN and  allow operators to avoid having a MAC_Binding\ntable explosion (helping lowering the memory footprint) in core OVN\nby setting this to false or true (default) based on their deployment.\n\nFor more information on this option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/304db11a2afb94d5e9a182f6e6331674955524c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/304db11a2afb94d5e9a182f6e6331674955524c0"}]},"branch":"refs/heads/master"},"f83e9ec2789ddfb9abb91dff247b87a7609c41bb":{"kind":"REWORK","_number":4,"created":"2021-10-12 12:47:25.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/4"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 12:47:14.000000000","tz":60},"subject":"[OVN] Allow setting always_learn_from_arp_request for routers","message":"[OVN] Allow setting always_learn_from_arp_request for routers\n\nThis patch adds a new configuration option for ML2/OVN called\n\"router_always_learn_from_arp_request\". This new configuration is\na passthru for the Logical Router \"always_learn_from_arp_request\"\noption in core OVN and  allow operators to avoid having a MAC_Binding\ntable explosion (helping lowering the memory footprint) in core OVN\nby setting this to false or true (default) based on their deployment.\n\nFor more information on this option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f83e9ec2789ddfb9abb91dff247b87a7609c41bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f83e9ec2789ddfb9abb91dff247b87a7609c41bb"}]},"branch":"refs/heads/master"},"e4c5a41011fe161df25e7d636b54e7f623a310ca":{"kind":"REWORK","_number":5,"created":"2021-10-12 15:30:30.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/5"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 15:18:06.000000000","tz":60},"subject":"[OVN] Allow for tuning the use of MAC_Binding","message":"[OVN] Allow for tuning the use of MAC_Binding\n\nIn order to avoid having a MAC_Binding table explosion and helping\nlowering the memory footprint when using ML2/OVN this patch is\nintroducing two new configuration options that allows for tuning how\nOVN uses the MAC_Binding table:\n\n* router_always_learn_from_arp_request: This option is a passthru for\n  the \"always_learn_from_arp_request\" option for the Logical Router\n  in OVN. This option controls the behavior of the OVN routers when\n  handling IPv4 ARP requests or IPv6 ND-NS packets.\n\n* dynamic_neigh_routers: This option is a passthru for\n  the \"dynamic_neigh_routers\" option for the Logical Router in\n  OVN. This option tells the OVN router to resolve neighbor routers\n  MAC addresses only by dynamic ARP/ND whe set to True.\n\nFor more information on these option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4c5a41011fe161df25e7d636b54e7f623a310ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4c5a41011fe161df25e7d636b54e7f623a310ca"}]},"branch":"refs/heads/master"},"dd2dca3088f65cdaec25858ff3760656fb01d30d":{"kind":"REWORK","_number":6,"created":"2021-10-12 15:36:29.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/6"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 15:36:21.000000000","tz":60},"subject":"[OVN] Allow for tuning the use of MAC_Binding","message":"[OVN] Allow for tuning the use of MAC_Binding\n\nIn order to avoid having a MAC_Binding table explosion and helping\nlowering the memory footprint when using ML2/OVN this patch is\nintroducing two new configuration options that allows for tuning how\nOVN uses the MAC_Binding table:\n\n* router_always_learn_from_arp_request: This option is a passthru for\n  the \"always_learn_from_arp_request\" option for the Logical Router\n  in OVN. This option controls the behavior of the OVN routers when\n  handling IPv4 ARP requests or IPv6 ND-NS packets.\n\n* dynamic_neigh_routers: This option is a passthru for\n  the \"dynamic_neigh_routers\" option for the Logical Router in\n  OVN. This option tells the OVN router to resolve neighbor routers\n  MAC addresses only by dynamic ARP/ND whe set to True.\n\nFor more information on these option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd2dca3088f65cdaec25858ff3760656fb01d30d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd2dca3088f65cdaec25858ff3760656fb01d30d"}]},"branch":"refs/heads/master"},"8c3d5da02a391b6d27cb8009a6d2545ef36091df":{"kind":"REWORK","_number":7,"created":"2021-10-12 15:39:10.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/7"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 15:38:59.000000000","tz":60},"subject":"[OVN] Allow for tuning the use of MAC_Binding","message":"[OVN] Allow for tuning the use of MAC_Binding\n\nIn order to avoid having a MAC_Binding table explosion and helping\nlowering the memory footprint when using ML2/OVN this patch is\nintroducing two new configuration options that allows for tuning how\nOVN uses the MAC_Binding table:\n\n* router_always_learn_from_arp_request: This option is a passthru for\n  the \"always_learn_from_arp_request\" option for the Logical Router\n  in OVN. This option controls the behavior of the OVN routers when\n  handling IPv4 ARP requests or IPv6 ND-NS packets.\n\n* dynamic_neigh_routers: This option is a passthru for\n  the \"dynamic_neigh_routers\" option for the Logical Router in\n  OVN. This option tells the OVN router to resolve neighbor routers\n  MAC addresses only by dynamic ARP/ND whe set to True.\n\nFor more information on these option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c3d5da02a391b6d27cb8009a6d2545ef36091df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c3d5da02a391b6d27cb8009a6d2545ef36091df"}]},"branch":"refs/heads/master"},"c902333989d8e33a7742ae5713ef49bb5d3e050d":{"kind":"REWORK","_number":8,"created":"2021-10-13 08:13:59.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/8"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-13 08:13:44.000000000","tz":60},"subject":"[OVN] Allow for tuning the use of MAC_Binding","message":"[OVN] Allow for tuning the use of MAC_Binding\n\nIn order to avoid having a MAC_Binding table explosion and helping\nlowering the memory footprint when using ML2/OVN this patch is\nintroducing two new configuration options that allows for tuning how\nOVN uses the MAC_Binding table:\n\n* router_always_learn_from_arp_request: This option is a passthru for\n  the \"always_learn_from_arp_request\" option for the Logical Router\n  in OVN. This option controls the behavior of the OVN routers when\n  handling IPv4 ARP requests or IPv6 ND-NS packets.\n\n* dynamic_neigh_routers: This option is a passthru for\n  the \"dynamic_neigh_routers\" option for the Logical Router in\n  OVN. This option tells the OVN router to resolve neighbor routers\n  MAC addresses only by dynamic ARP/ND whe set to True.\n\nFor more information on these option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c902333989d8e33a7742ae5713ef49bb5d3e050d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c902333989d8e33a7742ae5713ef49bb5d3e050d"}]},"branch":"refs/heads/master"},"12890182d9b0a96239065e83e24b5748b711b224":{"kind":"REWORK","_number":9,"created":"2021-10-13 10:57:33.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/9"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-13 10:57:25.000000000","tz":60},"subject":"[OVN] Allow for tuning the use of MAC_Binding","message":"[OVN] Allow for tuning the use of MAC_Binding\n\nIn order to avoid having a MAC_Binding table explosion and helping\nlowering the memory footprint when using ML2/OVN this patch is\nintroducing two new configuration options that allows for tuning how\nOVN uses the MAC_Binding table:\n\n* router_always_learn_from_arp_request: This option is a passthru for\n  the \"always_learn_from_arp_request\" option for the Logical Router\n  in OVN. This option controls the behavior of the OVN routers when\n  handling IPv4 ARP requests or IPv6 ND-NS packets.\n\n* dynamic_neigh_routers: This option is a passthru for\n  the \"dynamic_neigh_routers\" option for the Logical Router in\n  OVN. This option tells the OVN router to resolve neighbor routers\n  MAC addresses only by dynamic ARP/ND whe set to True.\n\nFor more information on these option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12890182d9b0a96239065e83e24b5748b711b224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12890182d9b0a96239065e83e24b5748b711b224"}]},"branch":"refs/heads/master"},"a278c5ba789c014ec777a75fc9538179d6707202":{"kind":"REWORK","_number":10,"created":"2021-10-14 13:48:54.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/813610/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/813610/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/813610/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/813610/10"}}},"commit":{"parents":[{"commit":"7dd2f0044512dd1d3f067e54f66d627a5697f29f","subject":"Merge \"[Policy] Set scope_types for BaseCheck rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dd2f0044512dd1d3f067e54f66d627a5697f29f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-12 10:42:39.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-14 13:47:34.000000000","tz":60},"subject":"[OVN] Tune OVN routers to reduce the mem footprint for ML2/OVN","message":"[OVN] Tune OVN routers to reduce the mem footprint for ML2/OVN\n\nIn order to avoid having a MAC_Binding table explosion and helping\nlowering the memory footprint when using ML2/OVN this patch is setting\ntwo options to the OVN routers:\n\n* always_learn_from_arp_request: By setting this to False we\n  avoid learning from ARP replies observed in the network. Only the\n  ARP requests sent by OVN will generate a MAC_Binding entry in the\n  OVSDB database. For larger broadcasts domains this avoids having a\n  MAC_Binding table explosion, reduce the DB size and memory footprint\n  of ML2/OVN.\n\n* dynamic_neigh_routers: By setting this to True we avoid\n  pre-populating flows for router to router communication, reduding\n  the number of flows, DB size and memory footprint of ML2/OVN.\n\nFor more information on these option for core OVN please refer to:\nhttps://www.ovn.org/support/dist-docs/ovn-nb.5.html\n\nThis patch also includes a new maintenance task to include these options\nto existing routers in the system.\n\nRelated-Bug: #1946318\nChange-Id: I056acdec9b6ee2341d2bc4f7bd9a678f3bf91972\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a278c5ba789c014ec777a75fc9538179d6707202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a278c5ba789c014ec777a75fc9538179d6707202"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
