)]}'
{"id":"openstack%2Fneutron~813883","triplet_id":"openstack%2Fneutron~master~I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6","project":"openstack/neutron","branch":"master","topic":"bug/1946666","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-18 22:45:05.000000000","reason":"Change was abandoned"},"10338":{"account":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"last_update":"2021-11-02 13:22:31.000000000","reason":"Daniel Speichert replied on the change"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2021-11-18 22:45:05.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-18 22:45:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6","subject":"Make OVN PG ACL sync compare known fields","status":"ABANDONED","created":"2021-10-13 18:42:30.000000000","updated":"2021-11-18 22:45:05.000000000","total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"2dd3479b63dc9538f0891c5e2ae7aa378a87de5a","_number":813883,"virtual_id_number":813883,"owner":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-02 03:42:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 18:53:16.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-13 19:27:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-13 19:30:25.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 20:13:48.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-10-13 21:06:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-14 10:36:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-10-15 09:24:54.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-10-20 15:13:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-02 13:22:32.000000000","updated_by":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2021-11-02 13:22:32.000000000","updated_by":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2021-11-02 13:22:32.000000000","updated_by":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REMOVED"},{"updated":"2021-11-03 08:20:08.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"90202d6652af98bcb040a4d4c8b855433ebda6ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-13 18:42:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b686f3e49e33d8c81685d3e52792cda1e2beac0f","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-13 18:49:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ebe6e289d8424fce4b1ea5efec2c6e4a53e9073","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-13 18:53:16.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/83/813883/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/83/813883/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fc14ce37512fb56c6ef8eb80a49f183e5e0417ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 19:27:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/724ad314b42a4b4ea8d9ebbc88882927 : SUCCESS in 43m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2cc883d926e345d1bbc85a9f070d8c7a : SUCCESS in 41m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de964aa37ae2d60b480e3cf42f403d6e104e80d3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-13 19:30:25.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/83/813883/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"38593865eff88e085cc5c9ea2a58358d0941e1f9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-13 20:13:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813883/1 : SUCCESS in 1h 30m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4262a2db918ac06554079cddbe8c7aedcf6484db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 21:06:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fc763861b5e54e22884b178064d54dea : SUCCESS in 50m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6730b587ff6244d48cb8eafd3485c9a4 : SUCCESS in 2h 17m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39e5bf40bf3346b6ba42a9393a9c178e : SUCCESS in 31m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b3de367a07546c19405dd227fe7cf70 : SUCCESS in 30m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58b6655550eb47f79693a799251abbd6 : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf77857bb5604e8bacb4168f3c832486 : SUCCESS in 23m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ecd44cde8c1429cbdcc18c8bbb28962 : SUCCESS in 21m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18f2ea3c4990442fa9fab14f27d1d335 : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7266d95062849259a1bc45f39e37168 : SUCCESS in 10m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3b92449308d41bb8df562b0e7a22a2b : SUCCESS in 43m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/15c6bf5797124c279e013a1a7c423c51 : SUCCESS in 54m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e6ac177b996b45c2843cc48b6d076869 : SUCCESS in 28m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/63cc7b5bec81497ab1e3e8082b79900e : SUCCESS in 1h 04m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/345c0c039f4c4fea8e9c18735fdb7812 : SUCCESS in 1h 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"509e0beea20b135ce7610b95b9a01c56a6f32a4b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-14 10:36:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e283448d1c7e17fcf50bbbb77f1dd4afe8ccf1a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-14 14:41:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5e8f2216a366bdf0ca7a8a1565d327de22019fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-14 14:55:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/83/813883/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/83/813883/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e74b27adf529db79fa6c4c4af5ccd4c744350ca8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 15:20:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b73753792ef84a1cb59181689338987e : SUCCESS in 33m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5edb28b641d441c8ad85df7c342ab6e : SUCCESS in 32m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00d113629abb1a880eb3eaf8c0d214a2692797f6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-14 15:31:46.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/83/813883/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"11e96d63c565b14004167ad5a20db02d2a3794db","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-14 16:12:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813883/2 : SUCCESS in 1h 29m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2c7279684f627b310fc25d089aaa38bf84c4333e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 16:28:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8ff1325bfc4a4340a5dbfd342d68fadc : SUCCESS in 42m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8e27486cb42a406cb8eeee7556b5ef10 : SUCCESS in 1h 43m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/625f27cf09f441b7ae9fe6060ca4fd41 : SUCCESS in 29m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9c4527046d74ea5ae2b79ef897e172d : SUCCESS in 32m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d31a1247d64024af654213bbe61623 : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e2426dd0a6343b1b37806f1d16e89ce : SUCCESS in 22m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17504fa31daf4f0bb9254ad011efc3b2 : SUCCESS in 26m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10f55515e0434235b1307d7a33b3609d : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97208f90c9c483b94837915cdf31189 : SUCCESS in 11m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4bddb1b4c55f4194b312fbc0f38b3616 : SUCCESS in 50m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1fc184dab8f74243a479950344bfad8c : SUCCESS in 51m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/80317a51d4304271bfadd2c3e9993216 : SUCCESS in 31m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74b52697237c4a92b179e6eef5154f01 : SUCCESS in 1h 09m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9fdf209bfba4475cb24f0cccf47d06a1 : SUCCESS in 1h 16m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b2ed2133856a69933ae3ff157b8145e4d2b50bbf","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-14 20:48:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"adb029d81b01f84503f6c38ec1d13f1a95290a87","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-10-15 09:24:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"caf5774178c2ebd98766fb2e3f2acf5edd53d6c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-15 18:16:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6868f691ee1bcce08c81a2be3eb185801a9a5e34","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-15 18:27:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81c9ff8ffa93c2f737de9955872b8ab87731c62f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-15 18:33:17.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b52ead69eef83d631688dd08fa532c09e5bb22a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-15 18:33:39.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/83/813883/3/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/83/813883/3/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c0c0ce65ec65c83bca9ec9b2865e70abf1c3bf76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 19:11:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/520f05510ac7430f8bf79dea94c9527c : SUCCESS in 33m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6651e5598ea349b7a99b4e9228cce992 : SUCCESS in 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b5fe9db6ddae49e53b4a0d79c43bfcc8b7648eed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-15 19:27:32.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/83/813883/4/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/83/813883/4/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"cee7683acc5a4a735d2ab23580bf7371a745b63e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-15 20:22:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813883/4 : SUCCESS in 1h 48m 59s","accounts_in_message":[],"_revision_number":4},{"id":"d345074fd71f5d7008bc68579e647f44320337d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 20:23:30.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/04d7640ad2e24caaac0624b263ca5f9e : SUCCESS in 43m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/37c5324c18aa4130914e5d195e01c26e : SUCCESS in 1h 42m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecec6e6187794103bbad72a9ee35472c : SUCCESS in 47m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2103889d415c4367a3abdc71054367bc : SUCCESS in 41m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f94d13a5be641dbbf07082c77b24743 : FAILURE in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21af2efbedef4cbb92f01be55f3d797f : SUCCESS in 27m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c27933d880040eabe984c29be2c8ca3 : SUCCESS in 34m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/014d034beac4478fbacc9cff8b970828 : SUCCESS in 24m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce1fe3143ea411e8a72e487e2793d6c : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/61330cf7ada7456482eab8cbc4637be9 : SUCCESS in 50m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f6e7b121d585487c8b4541b0a23fa4f2 : SUCCESS in 1h 02m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3fd90f54737e47c6a259f8aec5617d29 : SUCCESS in 32m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/049002cb794446d0ae47e2f64f7979cf : SUCCESS in 1h 09m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/579f03b9904547f5831965b9753799bd : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":4},{"id":"c5cd3331df5eac7b1e876d19ad39e19f1beb0f8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-15 20:29:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fdeedd22aeaa13140a0c47983756536d30b90bac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 21:12:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2f0a32241be4222919c6e6d6c0456bd : SUCCESS in 40m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e1cd27ce4044b45a4f85858b5a6a60f : SUCCESS in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f59096316d44b011e8fa674a462358ec662b1d09","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-15 21:32:56.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/83/813883/5/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/83/813883/5/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"766d5dbff7fc982538006ef95c24579aa34e3ce6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-15 21:56:21.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/813883/5 : FAILURE in 1h 26m 01s","accounts_in_message":[],"_revision_number":5},{"id":"a49318d6ab16d2518f1fe581d97bc7ecf55e505f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-15 22:13:24.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/83/813883/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b851894ec9222e3390b9079666da6b45df5034bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-15 22:21:42.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/83/813883/5/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/83/813883/5/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"21ebd5d8ecaa90e4eade215dc2451a9582bf5c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 22:32:02.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/52426844a0e14c90bd7b0cf96fe8bf72 : SUCCESS in 41m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ee8438bfdd02483f9e535dffb71595e0 : SUCCESS in 1h 55m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a2220ae0fa84429a0b0147db3c47e60 : SUCCESS in 36m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5425354015044ff49d62e1ac1b60ba32 : SUCCESS in 36m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3256121dde040cf957939d858735a31 : FAILURE in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee63e5b3d3ae4fe7a5923a41fb46e27a : SUCCESS in 28m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76124ecf41c041c8b864f09c99485098 : SUCCESS in 25m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a896a6a420f445dbd890838e4b23111 : SUCCESS in 27m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ab31b36a6f4d5585dbd84dabf375ad : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/049d6182c66447f1bfb35cd1697676b8 : SUCCESS in 51m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e3f4cd57297445109e7de2856197d6fd : SUCCESS in 59m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b3f5140808ab4283b5760c9acaf8304a : SUCCESS in 33m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c55d3b78bc85409a9b300ef333248c31 : SUCCESS in 1h 09m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3379e9c8e8d04ba18e9ee486e498e714 : SUCCESS in 1h 14m 52s","accounts_in_message":[],"_revision_number":5},{"id":"11d1dd9cdf326552ec10d4f9ae46873a673d4052","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-15 23:07:48.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/83/813883/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"9a7228ce0c90d2f0b2fe69cfb70c8a874ebf4dce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-15 23:56:07.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/83/813883/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"b0777fa28374cc3cd0058358957cb7ca04dac28f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-17 18:01:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dc317523204ecd0aa96b8eb315792b29dac92ccc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 18:46:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c48d7735c8ab4e758a0e966505e15e03 : SUCCESS in 33m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d268b1380c8b4884a496e75678b98ab9 : SUCCESS in 42m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9188ac98ddf01967eb16c3eacb870d689ac26ce1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-17 18:51:18.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/83/813883/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"09449bbcb96ed12a9f39aa7245bbe372ba85a864","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-17 18:58:23.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/83/813883/6/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/83/813883/6/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"d8d605ec979de2fcade433aa37fceff7ad1ae913","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-17 19:46:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813883/6 : SUCCESS in 1h 44m 31s","accounts_in_message":[],"_revision_number":6},{"id":"14ef6f6a29b0a903e6407e34dd95cb3eb08f2cad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 20:17:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f074cfc910fd47d0ac8198f2efee9825 : SUCCESS in 32m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e47186d5cf48491a949ecae8c895af94 : SUCCESS in 2h 08m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3de8d1160b824d34a9274e80d9b8d278 : SUCCESS in 53m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/058ac4564dbd498587313124458b28c5 : SUCCESS in 43m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be8c1b7311d4166a4d5596b231fa532 : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc9af67154764b288280201dcdd60ec1 : SUCCESS in 43m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de00dc8a8beb491bae8c44d7deb34310 : SUCCESS in 27m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22598801c4e642bfbd980852fb8c52fb : SUCCESS in 40m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff22bf990314d79bd34f1a44e89f053 : SUCCESS in 10m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77ecb4bdcb7e4307b8d3c647270ecfee : SUCCESS in 47m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/63e09a8e717d464aa79a5af5fe12cbd1 : SUCCESS in 1h 02m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c5022770318f49358d3c05b66c90d4bc : SUCCESS in 32m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c393bfd71307486991e46259221a8999 : SUCCESS in 1h 17m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2e7cfe8b2caa473fbc93f8b50451b973 : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":6},{"id":"886f2365ae15f6f64245d3023462b3fe07cc321c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-18 07:44:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e7fb5d110b119f59c40e185f71bf888a0752c10c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-20 15:18:19.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7453ca28c5a8cdf01b12277f5c402e148de2b7dd","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-10-20 17:32:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9da9e1f56b84e23abbf6cf37d73138a80cba6f87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-11-01 22:52:30.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6df0d004485d4c1b47d9042366dbe48b5f75e032","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 23:54:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c338fc1a261c41168dbb15cef24bb192 : SUCCESS in 59m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e03b6a2bae9b467a9d948beb50163b1d : SUCCESS in 57m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9724f980acde7234427db48b71fa11727025cc18","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-02 01:00:23.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813883/7 : SUCCESS in 2h 07m 08s","accounts_in_message":[],"_revision_number":7},{"id":"f2075a4b0bfaf5a50ae6b092a583d26496858a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 01:07:57.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bfb07dc66766487b878a1d695db0a0e9 : SUCCESS in 39m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a8d7f638693c492aa7c88e690b3d29fb : FAILURE in 2h 13m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f64912eab62741b8b7bb698df9df485f : SUCCESS in 46m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c371c694711448d485d22ec9bcda1f36 : SUCCESS in 48m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4593d77836f141a188a431fc9d0fb460 : FAILURE in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d0d83d8b81941c2977e57d441934758 : SUCCESS in 38m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f75d83f9577a4ae094d5f97a0b8c56f8 : SUCCESS in 39m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/690ff9f882bd471ca96a416cb4349bed : SUCCESS in 38m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65dc20313b03481c99be8bbfbafc2412 : SUCCESS in 10m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8560caa88da849f48a5aa1594f0e1b63 : SUCCESS in 50m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cc62ce320a52480c9be38444ac7be97f : SUCCESS in 49m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5dd69865852b4930bdb12228618e92ec : SUCCESS in 28m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c59beb7cc67413383ede5c69da9666f : SUCCESS in 1h 04m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/10501bad752b4715a33500af8a572f62 : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":7},{"id":"f397b5949d928bee589abd1d8ff8cd9a939b4aca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-11-02 01:43:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"af79a64c3172e4b1f56fe6c529537ea29fab1dfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 02:42:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0ed4fa42582494e9de77c5bf339e294 : SUCCESS in 54m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01931c43a8a544d3a515ec6649cbbf6f : SUCCESS in 56m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c82b483aff881504dba52d5c9c944064ae864dbd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-02 03:29:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/813883/8 : SUCCESS in 1h 45m 49s","accounts_in_message":[],"_revision_number":8},{"id":"dbf8a9bff78e5640e164887d4f5a976ca5ae2936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 03:42:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0c8d6c044c4a49aaa7bc81a20725715d : SUCCESS in 44m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8dfb338e932c4aec87d10b26495a4630 : SUCCESS in 1h 56m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dda383553f4412abc72aa1c7c4de12a : SUCCESS in 55m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc4913b837784e97965d15ea9b8aab3a : SUCCESS in 57m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db980c60c394bb6be7e32f44615e2f1 : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b1e5d363adb4404985eb747bac60fbc : SUCCESS in 47m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3c73989b91f4160bee4fa819ca9a299 : SUCCESS in 37m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/409c81943f64429a8e58ead4547399a5 : SUCCESS in 39m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514ad513a4d34cf0b28836c397414900 : SUCCESS in 10m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef3e6e8190ab407086ed1b398e04a2f0 : SUCCESS in 50m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4043cbc5526b4693b62b9d5041576eab : SUCCESS in 55m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ee9b0a8e4bac43469cc5e2110d2111c1 : SUCCESS in 29m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0ce584d99cb418a92db9548726119a0 : SUCCESS in 1h 13m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3cf4c54c50cc4477842d84dde15d675e : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":8},{"id":"797f64b57a0ca13a226ece83821bfa875cdbea9a","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-11-02 13:22:31.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"94b23e380c206a1fd083d21fdde393a0d2bb11c8","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-11-02 13:22:32.000000000","message":"Removed reviewer Mellanox CI.","accounts_in_message":[],"_revision_number":8},{"id":"d1a3ec487c75defc3440797acd16d29f6f4ee102","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-11-02 13:22:32.000000000","message":"Removed reviewer Arista CI.","accounts_in_message":[],"_revision_number":8},{"id":"2911568d39c3831fa0bae42a49e3ef4a1a1632b2","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-11-02 13:22:32.000000000","message":"Removed reviewer Cloudbase Neutron Hyper-V CI.","accounts_in_message":[],"_revision_number":8},{"id":"02694378fcfc968d34a1ebe474d00099968667fc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-03 08:20:08.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/83/813883/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d87257b20da29af672368e0e4415fae997ad6b99","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-03 09:17:51.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/83/813883/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"2dd3479b63dc9538f0891c5e2ae7aa378a87de5a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"date":"2021-11-18 22:45:05.000000000","message":"Abandoned\n\nNo longer needed, see https://review.opendev.org/c/openstack/neutron/+/818255","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a2389f7ec9ad14b601933af9603dbddfff465804","revisions":{"37948d0c4f513fa9645e761330b93e8642817abd":{"kind":"REWORK","_number":1,"created":"2021-10-13 18:42:30.000000000","uploader":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"ref":"refs/changes/83/813883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/813883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/813883/1"}}},"commit":{"parents":[{"commit":"aeaca468d2b5e140729fe5a2042aea26d643a711","subject":"Merge \"Log OvsdbAppException as warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeaca468d2b5e140729fe5a2042aea26d643a711"}]}],"author":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-13 18:37:43.000000000","tz":-240},"committer":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-13 18:42:20.000000000","tz":-240},"subject":"Fix OVN sync with new ACL label field","message":"Fix OVN sync with new ACL label field\n\nCloses-bug: #1946666\nChange-Id: I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37948d0c4f513fa9645e761330b93e8642817abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37948d0c4f513fa9645e761330b93e8642817abd"}]},"branch":"refs/heads/master"},"d987e43a03c10038f105b1be607a54e890bdc88b":{"kind":"REWORK","_number":2,"created":"2021-10-14 14:41:53.000000000","uploader":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"ref":"refs/changes/83/813883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/813883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/813883/2"}}},"commit":{"parents":[{"commit":"aeaca468d2b5e140729fe5a2042aea26d643a711","subject":"Merge \"Log OvsdbAppException as warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeaca468d2b5e140729fe5a2042aea26d643a711"}]}],"author":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-13 18:37:43.000000000","tz":-240},"committer":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-14 14:40:42.000000000","tz":-240},"subject":"Fix OVN sync with new ACL label field","message":"Fix OVN sync with new ACL label field\n\nCloses-bug: #1946666\nChange-Id: I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d987e43a03c10038f105b1be607a54e890bdc88b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d987e43a03c10038f105b1be607a54e890bdc88b"}]},"branch":"refs/heads/master"},"eede501ea011cfa1bb4b80903e81096ca74c7131":{"kind":"REWORK","_number":3,"created":"2021-10-15 18:16:55.000000000","uploader":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"ref":"refs/changes/83/813883/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/813883/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/813883/3"}}},"commit":{"parents":[{"commit":"aeaca468d2b5e140729fe5a2042aea26d643a711","subject":"Merge \"Log OvsdbAppException as warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeaca468d2b5e140729fe5a2042aea26d643a711"}]}],"author":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-13 18:37:43.000000000","tz":-240},"committer":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-15 18:15:59.000000000","tz":-240},"subject":"Fix OVN sync with new ACL label field","message":"Fix OVN sync with new ACL label field\n\nCloses-bug: #1946666\nChange-Id: I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eede501ea011cfa1bb4b80903e81096ca74c7131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eede501ea011cfa1bb4b80903e81096ca74c7131"}]},"branch":"refs/heads/master"},"a671b3e697a26a9d4e1b6263bd5b2e2bd33c21df":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-10-15 18:33:17.000000000","uploader":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"ref":"refs/changes/83/813883/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/813883/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/813883/4"}}},"commit":{"parents":[{"commit":"aeaca468d2b5e140729fe5a2042aea26d643a711","subject":"Merge \"Log OvsdbAppException as warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeaca468d2b5e140729fe5a2042aea26d643a711"}]}],"author":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-13 18:37:43.000000000","tz":-240},"committer":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-15 18:31:43.000000000","tz":-240},"subject":"Make OVN ACL sync compare specific fields","message":"Make OVN ACL sync compare specific fields\n\nWith OVN introducting new fields, previous method was prone to detect\nfalse-positive differences and crash when adding identical ACLs.\n\nCloses-bug: #1946666\nChange-Id: I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a671b3e697a26a9d4e1b6263bd5b2e2bd33c21df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a671b3e697a26a9d4e1b6263bd5b2e2bd33c21df"}]},"branch":"refs/heads/master"},"92d94991ea57e6e5ce30b2f47b393d0a6b7c9425":{"kind":"REWORK","_number":5,"created":"2021-10-15 20:29:39.000000000","uploader":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"ref":"refs/changes/83/813883/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/813883/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/813883/5"}}},"commit":{"parents":[{"commit":"aeaca468d2b5e140729fe5a2042aea26d643a711","subject":"Merge \"Log OvsdbAppException as warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeaca468d2b5e140729fe5a2042aea26d643a711"}]}],"author":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-13 18:37:43.000000000","tz":-240},"committer":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-15 20:29:32.000000000","tz":-240},"subject":"Make OVN ACL sync compare specific fields","message":"Make OVN ACL sync compare specific fields\n\nWith OVN introducting new fields, previous method was prone to detect\nfalse-positive differences and crash when adding identical ACLs.\n\nCloses-bug: #1946666\nChange-Id: I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92d94991ea57e6e5ce30b2f47b393d0a6b7c9425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92d94991ea57e6e5ce30b2f47b393d0a6b7c9425"}]},"branch":"refs/heads/master"},"ec8a796f6b431ac139d09e7464a9d77a63bcbea6":{"kind":"REWORK","_number":6,"created":"2021-10-17 18:01:16.000000000","uploader":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"ref":"refs/changes/83/813883/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/813883/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/813883/6"}}},"commit":{"parents":[{"commit":"aeaca468d2b5e140729fe5a2042aea26d643a711","subject":"Merge \"Log OvsdbAppException as warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeaca468d2b5e140729fe5a2042aea26d643a711"}]}],"author":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-13 18:37:43.000000000","tz":-240},"committer":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-17 18:01:07.000000000","tz":-240},"subject":"Make OVN ACL sync compare specific fields","message":"Make OVN ACL sync compare specific fields\n\nWith OVN introducting new fields, previous method was prone to detect\nfalse-positive differences and crash when adding identical ACLs.\n\nCloses-bug: #1946666\nChange-Id: I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec8a796f6b431ac139d09e7464a9d77a63bcbea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec8a796f6b431ac139d09e7464a9d77a63bcbea6"}]},"branch":"refs/heads/master"},"8c3725ec270f9528b229d72bb1ef393b0b6329b8":{"kind":"REWORK","_number":7,"created":"2021-11-01 22:52:30.000000000","uploader":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"ref":"refs/changes/83/813883/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/813883/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/813883/7"}}},"commit":{"parents":[{"commit":"aeaca468d2b5e140729fe5a2042aea26d643a711","subject":"Merge \"Log OvsdbAppException as warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeaca468d2b5e140729fe5a2042aea26d643a711"}]}],"author":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-13 18:37:43.000000000","tz":-240},"committer":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-11-01 22:51:51.000000000","tz":-240},"subject":"Make OVN PG ACL sync compare known fields","message":"Make OVN PG ACL sync compare known fields\n\nOVN may introduce new fields. Other places in code\nalready relied on a list of fields defined in constants.\n\nUnused and untested get_acls() is removed.\n\nCloses-bug: #1946666\nChange-Id: I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c3725ec270f9528b229d72bb1ef393b0b6329b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c3725ec270f9528b229d72bb1ef393b0b6329b8"}]},"branch":"refs/heads/master"},"a2389f7ec9ad14b601933af9603dbddfff465804":{"kind":"REWORK","_number":8,"created":"2021-11-02 01:43:22.000000000","uploader":{"_account_id":10338,"name":"Daniel Speichert","email":"daniel@speichert.pl","username":"daniel"},"ref":"refs/changes/83/813883/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/813883/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/813883/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/813883/8"}}},"commit":{"parents":[{"commit":"aeaca468d2b5e140729fe5a2042aea26d643a711","subject":"Merge \"Log OvsdbAppException as warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeaca468d2b5e140729fe5a2042aea26d643a711"}]}],"author":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-10-13 18:37:43.000000000","tz":-240},"committer":{"name":"Daniel Speichert","email":"Daniel_Speichert@comcast.com","date":"2021-11-02 01:43:15.000000000","tz":-240},"subject":"Make OVN PG ACL sync compare known fields","message":"Make OVN PG ACL sync compare known fields\n\nOVN may introduce new fields. Other places in code\nalready relied on a list of fields defined in constants.\n\nUnused and untested get_acls() is removed.\n\nCloses-bug: #1946666\nChange-Id: I53f2c7d352fa7b5e20a0271d00bdcd472f911cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2389f7ec9ad14b601933af9603dbddfff465804"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2389f7ec9ad14b601933af9603dbddfff465804"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
