)]}'
{"id":"openstack%2Fneutron~815280","triplet_id":"openstack%2Fneutron~master~Ie93cf3e47e09b3e5051be1ffad512251775b0492","project":"openstack/neutron","branch":"master","topic":"bug/1948466","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-10-26 19:06:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie93cf3e47e09b3e5051be1ffad512251775b0492","subject":"Check subnet in \"_remove_subnet_dhcp_options\"","status":"MERGED","created":"2021-10-25 11:16:17.000000000","updated":"2021-10-26 19:08:28.000000000","submitted":"2021-10-26 19:06:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815280-bug/1948466","meta_rev_id":"5b24735fbdb41672fba1f54be1952f843475bf46","_number":815280,"virtual_id_number":815280,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-26 19:06:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-26 13:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-10-26 00:15:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-10-25 11:19:44.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-26 06:01:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-26 13:25:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-25 11:17:42.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-25 11:19:44.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-10-25 12:15:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-25 12:46:43.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-10-25 13:31:20.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-25 14:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-26 00:15:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-10-26 06:01:06.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2021-10-26 13:25:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"6a6e47efa74f21b360a1c1fa6350e862749be48a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-25 11:16:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59154be6e615756873868300ce4a8e3beb4ae843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-25 11:17:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1ca58626ca77c66335491ed9e0a5f2241017a879","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-25 11:17:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/815280/1/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/815280/1/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"ea9f1eaf31ad289d416b534d2d7257ec13de2090","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-10-25 11:19:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5cdbc43fe12cde3c540f568e08ea79286afdd0e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-25 12:14:58.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/815280/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/815280/2/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"884edaf338539eec507335bbd4a9926826b2598e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 12:15:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/025d449eabd74d86ace830c3b46e2d70 : SUCCESS in 45m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51fe3744d7114d8fa1ecb978797553e0 : SUCCESS in 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7606d90ba5397d93a07ee7697259c3391b4b994","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-25 12:46:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815280/2 : SUCCESS in 1h 28m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c1b407e8193e2198652aec545e2b83fa4228e80a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-25 13:31:20.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/80/815280/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4dafc97fe324ba8cdd451debfd91de671bb1b7ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 14:11:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/275ab603470c4709aa5e434bc8b4d1e6 : SUCCESS in 33m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f1a6e6111fa2485bac845ba17b3c6166 : SUCCESS in 2h 43m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/547fdbe4e0284e3fa20d94a854e42210 : SUCCESS in 29m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f2e16d437394ff185b7c9914968edc5 : SUCCESS in 39m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a8d99c33f24b288de1bdf777ab71f9 : SUCCESS in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3afc9e27636e4c7299ac2f515fa7b540 : SUCCESS in 27m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed83763170484fe7bba78d936554206d : SUCCESS in 28m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26e0001bd6fe4d74973bf4fec275bba3 : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f234ee5bfec4536b45896f9a69694bc : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e46e5a66535c44a9b04bd83051bd91aa : SUCCESS in 48m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0cefc736299547e8b2116c41ae79746d : SUCCESS in 53m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8dd1bd1cfadb456f857d49db705f8006 : FAILURE in 13m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f248da7efa804342b5192419774dfd9c : SUCCESS in 1h 15m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/eaea39cba4f74f19bf9912d0d2af63d6 : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":2},{"id":"e9623f22451024e6043da1c4dd9b97dd8cfbe812","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-10-25 14:12:46.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2021/Arista_Tempest_Testrefs/changes/80/815280/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7d947d45118d7bd87cae6882ef26142a8fb9f6ed","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-10-26 00:15:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac998b81539b1cda2725f14f8d47c13d0116610e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 00:58:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b41c3e9144d43b99074cebde11f7f0e : SUCCESS in 41m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69f7d1d5285c40c980604f34a11b38da : SUCCESS in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9172fd09d4370227f30ba2359f9624170bfba840","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-26 01:17:50.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/815280/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/815280/2/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b0132e716240404d10fefac541bec9c632e6b475","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-26 01:58:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815280/2 : SUCCESS in 1h 42m 42s","accounts_in_message":[],"_revision_number":2},{"id":"084d11e95d2689d59e1f7b589f6e5f8aba4ba0bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 02:11:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/42d2c7dedde744dd9dd07ab9224f9ea9 : SUCCESS in 42m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9779adb591fe45949bb853478d4c2a86 : SUCCESS in 1h 49m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/651feb41c3564b00a1ddc13a58386a49 : SUCCESS in 30m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9e2e1ebee88446388fd201f5730750d : SUCCESS in 36m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ac3dfb34157483babb9a7764f1d8cb4 : SUCCESS in 12m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81d291885bcc4283934b89e82a7b49b9 : SUCCESS in 21m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b129b28592754460a0c99908eb23eb6b : SUCCESS in 21m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ede28f3f11b041b28fcc3067dbb00077 : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6747accb9843eeb187778767c92955 : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eff69a3c9b164c30b830a007da85e6f9 : FAILURE in 46m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/01c3e601c9dd43f6b425819b4416dd6f : SUCCESS in 1h 13m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/266a54c361484030bf4785a856e6724c : SUCCESS in 25m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3f44092e009c439d8027ce6edfcb7d56 : SUCCESS in 1h 02m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8d7dc8d0f4e64422b72bf0db6584d98f : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":2},{"id":"065fa5b469b93409d16976c2dcb80f879e84eede","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-10-26 06:01:06.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c266a79433956c2ff86ca78d7273b4e2cd8a9692","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 06:42:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95fe5b9bc08a47ea8cd08e3479b495b5 : SUCCESS in 31m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f346f7fadd84926b5f707af88352fd8 : SUCCESS in 36m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"352825c45a9ae4abad3dfa047980c928a7688938","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-26 07:02:22.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/80/815280/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/80/815280/2/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0d092c2273740b6911d08156e822a30042e85d85","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-26 07:33:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815280/2 : SUCCESS in 1h 31m 43s","accounts_in_message":[],"_revision_number":2},{"id":"2f0e25860af5c44177f3fbca60bc1f6359adf531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 08:24:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/736d97e05bd74971b72ff3bd33b645e2 : SUCCESS in 43m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/44dcc861afdf41e294f81315680dccf2 : SUCCESS in 2h 18m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd2e19e06b814426b77f26e3e13ae4ef : SUCCESS in 46m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0ac82afb8c343d59d6c46d5ae4b8a6f : SUCCESS in 36m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4ed96c7af745a1b8ebf34b92251574 : SUCCESS in 13m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7fc65e06b464f8fb59ce0940869c180 : SUCCESS in 33m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bda4c76e4a0944b4b0227df58e716aa5 : SUCCESS in 23m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40a52c9cf6c74a72a2d378139c16102b : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70cb55f753934c1eb1850cd6034b8999 : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/611057d7f1734e158ba268f81f9746b3 : SUCCESS in 47m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f1ef52933d9c45f6b2a5f1582b8c8eae : SUCCESS in 1h 01m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b51950f349b84a7b9d15f2570b9a044a : SUCCESS in 36m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f9e26316f7242e293f817e36f9414c0 : SUCCESS in 1h 08m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/52de233ed5394e528f6e36df5142b45f : SUCCESS in 1h 21m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1849d16872ea98ae525c422a1cca1732208cc807","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-26 13:25:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e3cd674069a411993ba5f92e390b3dcbef4d7f85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 13:27:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"46bdb79a981a84dce83d38f7ecdc5fc94296fa57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 19:06:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a0d319a1676d4ea7b8f46997ffe1732c : SUCCESS in 46m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1ac938713b854a378164b32d7f141057 : SUCCESS in 2h 23m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/514c2388b3ea4dfeaa5afe3b28348901 : SUCCESS in 43m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa040121e31247c7b5d42072ee10170e : SUCCESS in 17m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f74e0a6d0cdc41a2a539f88f670875ab : SUCCESS in 26m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3701a394ca4e40e9940cc42b2bcb074f : SUCCESS in 21m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728e5121b27a4615bd0458eeac3e86a0 : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/18761c144ff34121b84c82095ede3e7e : SUCCESS in 51m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f0662128c76044e6a840133575847d29 : SUCCESS in 58m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c6e3e4d6f09c4808ac7bcb9b4c573e11 : SUCCESS in 30m 59s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/240eb7441ec0421ca2d38a8962d0642f : SUCCESS in 1h 23m 01s","accounts_in_message":[],"_revision_number":2},{"id":"7520352b4733ad309df8418fbf4d186ad662f81e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 19:06:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5b24735fbdb41672fba1f54be1952f843475bf46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 19:08:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4887479413e41f282cd25513184a7a8 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"95c2801da855cada741693b0dc31192e7baa111c","revisions":{"70093078f10ce3314a7cc1b4eebeaaaf46bee2a1":{"kind":"REWORK","_number":1,"created":"2021-10-25 11:16:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/815280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/815280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/815280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/815280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/815280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/815280/1"}}},"commit":{"parents":[{"commit":"6d8e830859cd4ac9708701b8e344fdc68cbcaebb","subject":"Merge \"Fix OVN migration workload creation order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d8e830859cd4ac9708701b8e344fdc68cbcaebb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-10-25 11:12:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-10-25 11:16:05.000000000","tz":0},"subject":"Check subnet UUID in \"_remove_subnet_dhcp_options\"","message":"Check subnet UUID in \"_remove_subnet_dhcp_options\"\n\nSince [1], \"get_subnet_dhcp_options\" returns always a dictionary\nin \"subnet\" instead of None. This patch checks not only that \"subnet\"\nis None but also the dictionary is not empty.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/807692\n\nCloses-Bug: #1948466\n\nChange-Id: Ie93cf3e47e09b3e5051be1ffad512251775b0492\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70093078f10ce3314a7cc1b4eebeaaaf46bee2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70093078f10ce3314a7cc1b4eebeaaaf46bee2a1"}]},"branch":"refs/heads/master"},"95c2801da855cada741693b0dc31192e7baa111c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-25 11:17:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/815280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/815280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/815280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/815280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/815280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/815280/2"}}},"commit":{"parents":[{"commit":"6d8e830859cd4ac9708701b8e344fdc68cbcaebb","subject":"Merge \"Fix OVN migration workload creation order\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d8e830859cd4ac9708701b8e344fdc68cbcaebb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-10-25 11:12:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-10-25 11:17:17.000000000","tz":0},"subject":"Check subnet in \"_remove_subnet_dhcp_options\"","message":"Check subnet in \"_remove_subnet_dhcp_options\"\n\nSince [1], \"get_subnet_dhcp_options\" returns always a dictionary\nin \"subnet\" instead of None. This patch checks not only that \"subnet\"\nis None but also the dictionary is not empty.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/807692\n\nCloses-Bug: #1948466\n\nChange-Id: Ie93cf3e47e09b3e5051be1ffad512251775b0492\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95c2801da855cada741693b0dc31192e7baa111c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95c2801da855cada741693b0dc31192e7baa111c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
