)]}'
{"id":"openstack%2Fneutron~815465","triplet_id":"openstack%2Fneutron~master~I657881c319d425470277885545240d6a8b66a1f6","project":"openstack/neutron","branch":"master","topic":"improve-neutron-ci","attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-02-03 06:14:50.000000000","reason":"yatin replied on the change"}},"removed_from_attention_set":{"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2021-11-17 23:20:56.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2021-11-17 14:22:59.000000000","reason":"Oleg Bondarev replied on the change"}},"hashtags":[],"change_id":"I657881c319d425470277885545240d6a8b66a1f6","subject":"Remove some scenario jobs from the check and gate queues","status":"MERGED","created":"2021-10-26 10:26:03.000000000","updated":"2022-02-03 06:14:50.000000000","submitted":"2021-11-17 23:20:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"815465-improve-neutron-ci","meta_rev_id":"7bce2b1c5b3fba1c4c4434e42bbfe83ca2107990","_number":815465,"virtual_id_number":815465,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-17 23:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-11-17 15:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-11-17 15:46:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-11-17 14:22:59.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-11-17 15:46:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 11:01:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-10-26 11:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-26 12:19:45.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-10-26 13:29:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-26 15:28:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-10-26 15:30:59.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-10-26 16:14:33.000000000","updated_by":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"reviewer":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"state":"CC"},{"updated":"2021-11-17 15:37:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-17 15:46:56.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-02-03 06:14:50.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"6de1c0292518272a959fe2eebfd11ada1c5692d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-26 10:26:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d48109372f98b1577f355952c9e1899bd27d815","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-26 11:01:07.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b699f2ae9232799366a66dad45621c999cfc3f09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 11:19:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f15c0cab7af4289b048b896cb0772c3 : SUCCESS in 31m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/619ab8d747274cc7a114bc151cfcae23 : SUCCESS in 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d97845f1c354b1189592f4cea4b13977c899a023","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-26 12:19:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/1 : SUCCESS in 1h 52m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2a0a144500d2d6238027049c630a60ae30e3a6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 13:29:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3de57089b2264152b0e7170c7ff3f545 : SUCCESS in 50m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d0a7b5da4b0049639da77bce6c455cfc : SUCCESS in 2h 10m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fb42c11c63054b259fb35039096a8bd9 : SUCCESS in 2h 11m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd30994c7a4943f2a2620877bdc80f00 : SUCCESS in 1h 49m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a6c3fd5e05b64035a90ed34960fb6368 : SUCCESS in 2h 17m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec9f33f09774402d802a40fea9f89371 : SUCCESS in 41m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6069be4243f43e7ae1ad17ac20e01cf : SUCCESS in 36m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dd32f0b31124ebba513d49f8e5badda : SUCCESS in 35m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cabc1f09ce7f4079aa9ea121f1ad54e0 : SUCCESS in 14m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bac6063a8fb7403895887430b6ca80cb : SUCCESS in 25m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4095d47da9da4c7d8be8ebc325e0ce24 : SUCCESS in 24m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8ff629fd1934f4c8f6be63be28d0900 : SUCCESS in 37m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74442417a01a40e8a4d6294c08e61eaf : SUCCESS in 12m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d022b8b5b3d848f094bae74d28a13bdf : SUCCESS in 51m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bce995bc60c54fce9644cac14f593b50 : SUCCESS in 1h 31m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bfa1d513d2304bd599668c9930546e91 : SUCCESS in 1h 21m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e75232e0b314e97bd2b00bbb7ae28a4 : SUCCESS in 1h 02m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/077da5d456534b2d88369305e1d63e75 : SUCCESS in 1h 05m 16s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7d8d9ed317a840d984942ed5bc1533f7 : SUCCESS in 1h 05m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7b869153a85249a4aabf452fb27de069 : SUCCESS in 1h 22m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8edeae135fba4a329c96908ffb13e225 : SUCCESS in 57m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6bfd39e294974e19804247c1582122b2 : SUCCESS in 1h 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca2360f0762cd9e6affdbd310b00ce07bb4991be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-26 15:47:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f9e15d75ad1f57f7e4eef6094c188b99d18fde2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:36:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca5dc514d6a348548217ab70a9143cf8 : SUCCESS in 41m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88f64a7e0f0e438eaeaa27a3140455d2 : SUCCESS in 37m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab26a01828c411a1f369c1cf1675e96acdcd5d42","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-26 17:35:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/2 : SUCCESS in 1h 47m 31s","accounts_in_message":[],"_revision_number":2},{"id":"dca7d2e3f46dcbfbdcbc3ede59641a63eec5c9b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 18:20:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/71e67d85eb3c4de8ac9c643f381f6d8a : SUCCESS in 55m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cc81296b6dff4b759e9185810bc02d7d : SUCCESS in 2h 17m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d7fd1b68e35f4b3ebbf69bc49fa1f170 : SUCCESS in 2h 23m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7be2f2a0481547a9a246d7fbea66b44d : FAILURE in 1h 34m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d93ea10eeb7d4d438e03e89e42531e6e : SUCCESS in 2h 21m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4b4b66a85e2449348e0bade0f8b9c617 : SUCCESS in 36m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79600f378687474c895e732a46fb77d6 : SUCCESS in 32m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/936749925c584ed9ab4eed66964df4ac : SUCCESS in 38m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d89efe614f45428d342045a31b1142 : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/110d44ab341646cf9f236d1123a0b3b9 : SUCCESS in 27m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/359b79f00d21414da0225029dcc1e346 : SUCCESS in 23m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a356d690c00142178367281b90ffd2e1 : SUCCESS in 26m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994b9d773557422783d5a825ede947ab : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/56a8c769b6294c4fbc93b9bce566495d : SUCCESS in 54m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3430d1bf1a7b4317b3a832fe19635740 : SUCCESS in 1h 26m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c44a8e86e6e74924836b6fcd908c9ec2 : SUCCESS in 59m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d0d38a9d4cb49adb7e399b1b1101cf7 : SUCCESS in 58m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/02a123250691471682ef731a88d367a1 : SUCCESS in 59m 24s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0751cfd331e04417b2bb811ec15b091b : SUCCESS in 1h 15m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/77c313f76f56451e9f901073a2c54e67 : SUCCESS in 1h 09m 12s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67d255c8f9cb4c3e82b470908eb90907 : FAILURE in 1h 02m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b93e8e3e642c4290a39bad976e2c6e10 : SUCCESS in 1h 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24989f9e552c94428af7f207b23b4e827277860a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-27 14:31:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab4f98973ca3381ffccb64d3ca9a3e278adbdc5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 15:18:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/418cbb29d7f8442193dacd7d628fd5db : SUCCESS in 37m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8f233d346bc4f06af52e5f506843993 : SUCCESS in 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"26fad3a046064e87ce8ff52120307f48c4f6c9ff","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-27 17:14:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/2 : SUCCESS in 1h 35m 23s","accounts_in_message":[],"_revision_number":2},{"id":"be6c6bebd963b55b43aa87b992f689416dd1ec9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 18:08:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a2d312969b5e4db8b4758b298743a6ce : SUCCESS in 49m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2944102aec7a44a3baaad73374467d2e : SUCCESS in 2h 37m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bbfefe4c2df74732be5768766aad3c02 : SUCCESS in 2h 26m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/900dc2c32d7e4dd1a2141d0cd01f4930 : SUCCESS in 1h 28m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b51c709af6d6438fa6d9e63cb67be5dc : SUCCESS in 1h 47m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a2f5033d20ca45ada7a9dcebb019038c : SUCCESS in 42m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cf776cd67c5427f86f2c50211cec59e : SUCCESS in 34m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b58158c77e428a81a5c0256334421a : SUCCESS in 40m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f4a39905d5c4b74a3bd91eda1466be8 : SUCCESS in 13m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/774b325cd33741b887cee0e489029b5b : SUCCESS in 24m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75622a6e48ea420a8166ea167bf1642e : SUCCESS in 30m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95a99002958f41758b9824cf076424a5 : SUCCESS in 27m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d01a7ad11cd4f11a72f2efe34147b44 : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4d189bfc70140499e30595e76c5e1f9 : SUCCESS in 1h 01m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/204d25cd74b04115b7ed72ef10c830fb : SUCCESS in 1h 31m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c27e2a4055e0427bbfc882e99c927b05 : SUCCESS in 1h 02m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c49be22c3174fb590fb18960474b3e6 : SUCCESS in 1h 01m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a9044ec963d43aaaf25f671552f7025 : SUCCESS in 1h 05m 14s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/62779c78aadc43eab71bae9420478d3b : SUCCESS in 1h 13m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/691f5a4872cf40b38bac5ff698d4129b : SUCCESS in 1h 16m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ab625c7eb3340a5a94994dbf6c1ea5a : FAILURE in 1h 13m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b243bb424a6744908916b8b3b3ce694e : SUCCESS in 1h 06m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6514250baec8fbda63cf88ebc09097799a18172","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-10-28 07:13:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"259b856824ce430a70648d7f1569d98a9fb8831a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 07:58:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/092cf32e85574024b414d699041ab0b6 : SUCCESS in 42m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be6551b6c750490e8e6c362d0dee6243 : SUCCESS in 42m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"548b1e7325b64fb91ba2f2ea2d17f19e18135485","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-10-28 09:06:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/2 : SUCCESS in 1h 52m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d4a1e939a0c7c34159e8bcc6c9cc52c5bbdfd110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:12:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c36a632637df4f5dba0f6a000f20a159 : SUCCESS in 41m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8314601c5ee14dbd96834bca3348f381 : SUCCESS in 1h 43m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0cd3eb81a5774994b6b1a6a9ce76b503 : SUCCESS in 1h 57m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b99fc8c2e76440f7b617d47ca370cfb8 : SUCCESS in 1h 24m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6cc3dcf9bfdf464bb6ad754cdaea15ec : SUCCESS in 1h 52m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b6130774d72e44d7b2d2f34a85f0f7bc : SUCCESS in 33m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98de51d6c5b0459e888c9a45bcaa90ae : SUCCESS in 30m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01023fd3aa1b436c90c0aed71ddb0ea8 : SUCCESS in 31m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11162210e67f44f492c94f508497518b : SUCCESS in 11m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d910d9ad7dc54f25b9fec4903f629590 : SUCCESS in 24m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ef21a3d0c7242079a0586f53ac723f1 : SUCCESS in 21m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69a68898b4a1435a812e73e527009ad4 : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef6346ca77b4f688ecb7d1299ef12dd : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb4ba0bb75744f3fa38b0b3795ce150e : SUCCESS in 47m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98daa45e48634860ba1b69a02eb6ca0f : SUCCESS in 1h 34m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d4b82a07b12640a8a1cca3fbdcc12ed9 : SUCCESS in 1h 04m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/97516e6388a94322b597f3eaa73e1060 : SUCCESS in 56m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/57c0c99dcbf64cfa9686cb863b7f868b : SUCCESS in 57m 26s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7e01075ab0694a77ba7dcb19336b87bf : SUCCESS in 1h 13m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/873e8edd551e4a2c8cf35dcb15e82f27 : SUCCESS in 1h 11m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1810dc84ad84e58ab2767f93333c347 : FAILURE in 1h 05m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6cdddebd5d2d4beeb1e206d92c8e1804 : SUCCESS in 1h 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89a7a69eb0ad11242d115945ec9dcf3b2bd78933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-02 08:39:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"29a4041a9a1dda7b22df029e772531dc6d35d07c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 09:36:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e88b04c16bf4baa8250673c338cbf82 : SUCCESS in 45m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9310cdc53bda4f26b2e17a7e69dac25a : SUCCESS in 52m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2a49906c4a5ed5be769fcafe14ccfe9717957217","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-02 10:09:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/3 : SUCCESS in 1h 29m 50s","accounts_in_message":[],"_revision_number":3},{"id":"ef34e2dac941452b21f4bba5190b8a7b3cdd3672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 11:37:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9f47104dc9ef479bb707ae09dcddc467 : SUCCESS in 1h 10m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a516b66520b849c781b1d759220c9774 : TIMED_OUT in 2h 51m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/de020ac1be9748fcb94c4ee3b3c08ce2 : TIMED_OUT in 2h 51m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ca83c5ec70e14e479321dde5a6e3261e : SUCCESS in 1h 28m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2c5b54689abd45fba48ba891f3d627a6 : FAILURE in 2h 18m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e06cc87c6c24238b4018ffab5bdadf8 : SUCCESS in 34m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fac70fb29c464a4b9436c262d3a7d9f5 : SUCCESS in 46m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/183a0443e34a496b8e79c872ad46a61b : SUCCESS in 1h 00m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab25ff3b0ce4801b9eb14d8dfd10126 : SUCCESS in 15m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c391ada98c494a99a98e51aefc57a390 : SUCCESS in 40m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c612fc845f4e4727aa810552c1f905eb : SUCCESS in 43m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1878885ae0d4e7d84ea00cbb398f9de : TIMED_OUT in 41m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b29d124bb24311aa49fc07e98eea64 : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/93bb6811e13d4cb89a40feff7b4ebbd2 : SUCCESS in 46m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d92cda7fbae4fc4a3ffaa99f89d47f8 : SUCCESS in 1h 26m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f2b170604caf43db8baa7d2d5f827bbc : SUCCESS in 1h 07m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0cf51c4e96f14fe09e650afa9666bc95 : SUCCESS in 1h 08m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6dd720470ffd47c9883efdf59672577b : SUCCESS in 51m 07s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/891a0f06f8e44d30b4d07bb36a532ace : SUCCESS in 1h 33m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dcdddc6beafd48d2a4f3c85739cb40a6 : SUCCESS in 1h 40m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/292741bdcf4a415a9c0068468a5a07b1 : FAILURE in 1h 15m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/68153879bec44ceabec72efb8bd353b1 : SUCCESS in 58m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd8f435b360db849864cf3d9879b6dbe85f28ca2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-02 16:52:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"df24973c210c41e13591b8f494cb875176aeaf5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:00:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12544960f5c64c6e950261a7f2802960 : TIMED_OUT in 1h 01m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22a54cc7abcc4cddb197177eb96e5c9e : TIMED_OUT in 1h 01m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e5c1e86ae73e07f9c03d1421c37bba80541a1981","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-02 18:33:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/3 : SUCCESS in 1h 40m 31s","accounts_in_message":[],"_revision_number":3},{"id":"9adcdb884da0a80ea5bd7b964eab6f11a50da8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 22:53:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/be257b59d1a648258ee79eacd3ecb06f : SUCCESS in 59m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/964da4d67e04408f9a0515e769e05bde : SUCCESS in 2h 48m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fd2d7aceb7c743a1b373f9c6bc61c467 : TIMED_OUT in 2h 51m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d1c3253e5e0449d89a9beccc62ddb62 : SUCCESS in 1h 40m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/74aad5464f3245d3aba98744b8e098d6 : SUCCESS in 2h 23m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9436669d2a4c4875a2bbf545d47cb62c : SUCCESS in 40m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38d6245e56b74ad387138f65f9b7716d : SUCCESS in 50m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a4f3df131d244528f77a7ae022bf362 : TIMED_OUT in 1h 01m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1df5040486f34d598fdad901fcae2d0f : SUCCESS in 14m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6436f76ebf154c598c55b8b173818494 : SUCCESS in 42m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/970488f5dc4c438fa5fbc49bc836fa70 : SUCCESS in 44m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d792e04db40a45b4bfc1dedd98d252ec : TIMED_OUT in 41m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada160b587cb4d82a9c6e39fefad9730 : SUCCESS in 11m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17908d8ac2c24a0abe41805a38b56ca5 : SUCCESS in 52m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1bd12847a844d66a8ac79a5a534eec5 : SUCCESS in 1h 35m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/21a3cfa3dc5745a79f0fc83305a002f4 : SUCCESS in 1h 10m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72a225a224e44df68eb8468ecfea8599 : SUCCESS in 52m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f8663708edf94f8982e72726cf3d349d : SUCCESS in 1h 08m 27s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/228505f83b484ef5b28fa74b1b772cb1 : SUCCESS in 1h 38m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0670fc6126b44db79dd4be7b9c6d35cd : SUCCESS in 1h 20m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24e6fae398c74c2186c233544ea2892d : FAILURE in 1h 11m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c9b6a2140c2f4857ba5c18fe289c706d : SUCCESS in 1h 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"22490808bddd4af7622ffa54b1baf9d9224556c1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-03 07:25:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c0d133b336284fc887f08cc349678ae25f7c38c6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-04 10:03:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"61c93663358e31ea7af8debba7df98daf71c8c07","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-04 10:04:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d38441f0625f8f6c4ec3fb14521fcf23cabf3d5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 11:07:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24a1415861dd40e7ac8de06e8c84d5e8 : SUCCESS in 57m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d46aee12c04e4a638d76e48e0ebfcd4a : SUCCESS in 58m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e45d97e1ed1dd550cdfdafe001ca3f1b9b411656","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-04 11:58:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/3 : SUCCESS in 1h 31m 36s","accounts_in_message":[],"_revision_number":3},{"id":"410fbe63020e83f2f7bea43cfdf096c65c64c431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 12:54:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d0609744351d496ab109ee8a8772546c : SUCCESS in 44m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/816526b870944491b5146005f7b7fb50 : SUCCESS in 2h 43m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d9943733bddc4621b96896e69994bdb8 : SUCCESS in 2h 26m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0af6e110e92a47878946995673e55076 : SUCCESS in 1h 43m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0329fc4558ff432e9acc88c0325859be : SUCCESS in 2h 28m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02129258401843c194cc2aea3cc966fd : SUCCESS in 35m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/598cca1de59a4a32b4856b8961166909 : SUCCESS in 1h 05m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26990eb15587446da33e51e16da8b3de : SUCCESS in 51m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8552f68e72d4b5099ec09a229aea13a : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1584ba1cd7da4a8f835cfadd8d6bd375 : SUCCESS in 50m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6a66d525a1f472db193014e8f831750 : SUCCESS in 38m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa3ebf43c53243348fcb6f87fdaa6704 : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718e320bc1114d2799f5432dc9bd2c96 : SUCCESS in 10m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ed5b20e88474d8c9150a1178d2d9ed0 : SUCCESS in 52m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/61783dcda16e4567912e4d9965b28e4a : SUCCESS in 1h 30m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a7d6fe5ced9347e899b9938f846f67eb : SUCCESS in 1h 10m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f085d5d65400488385cceff733ae9428 : SUCCESS in 1h 04m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b56631da41da4cadba771cd0c12c180f : SUCCESS in 59m 14s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/613322cda6044c07855616ceff06c3ca : SUCCESS in 1h 09m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/da6a7c5daf724859bc24af312a6ec592 : SUCCESS in 1h 18m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03dd6b7771884404a412ed0aa8041ddd : FAILURE in 1h 05m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8ddfbc9919ba4888a7dfdc2906b4e6b6 : SUCCESS in 1h 16m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"66964781ce28a1fd8616af84c6376e7491748099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-08 09:33:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5d143de6e3088f1295dd239f8c0133bae3371130","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 10:33:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f55c640d53874fd6ac1fb65c621cfc38 : SUCCESS in 55m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1dd54e60520f4b3e876355e64b41e939 : SUCCESS in 51m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"10a6f35fe2a34453e5f98de3aece7c894cd8c553","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-08 11:38:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/4 : SUCCESS in 2h 04m 15s","accounts_in_message":[],"_revision_number":4},{"id":"cb99d861a3cd892ceba6f017a689be9a5170a7fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-08 12:15:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3b0b38c6fbf709c25b34ca198e571d9943724957","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 13:17:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/536254a760cc49698a3092aa81c45c11 : SUCCESS in 58m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9694186ed574a9d8873a10e4bd34b38 : SUCCESS in 50m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"59b60f6ffac684e51a3faed1e15adf4342233376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-08 13:20:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"74ba2e8d0f8b18bd31ad2cdd20b0a90d002351f4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-08 15:32:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/6 : SUCCESS in 2h 11m 09s","accounts_in_message":[],"_revision_number":6},{"id":"cc64470843667d80f13a9598755e2aafeac99a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-08 15:38:08.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"135ffb1658b78041d9d9f491b03ee79d1d5f2210","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 16:39:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79da46612a604c96ba57d3c0b56de225 : SUCCESS in 57m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5f091944da443deb25b3b11b60fe338 : SUCCESS in 46m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8f30c6b5a64d6558c7d74aacdb43bc52da4eaa53","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-08 17:12:02.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/7 : SUCCESS in 1h 33m 09s","accounts_in_message":[],"_revision_number":7},{"id":"8640018a9963c6cd0928829cee173f541c3f97af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 20:53:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5c17bf450b049eebd2382c7626a0416 : SUCCESS in 1h 03m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d3bc9174c361417db62451a816113628 : TIMED_OUT in 2h 52m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/328001f1f50544ab9e86649b2d6ec6a4 : SUCCESS in 2h 02m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e4697c6f417c45f8a03f8d9340108863 : SUCCESS in 1h 38m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/761f7eae93724d178c10d6581362a374 : SUCCESS in 2h 43m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d37c877ec8254265820cef055f2e19e6 : SUCCESS in 50m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f29543456f064c9c977ea5f00589c15c : SUCCESS in 57m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae99043d35fd4dce99cb0a7b38c5d1ff : SUCCESS in 58m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5a6d008e734882b918010bb5f8a742 : SUCCESS in 16m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2ba76ddbe654eeebdc81faefe8e1198 : SUCCESS in 44m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93bbac92ad0040a89723326886407643 : SUCCESS in 37m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/530cd551b16f44f7a538970a02f969f7 : SUCCESS in 38m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aafab16869a4beb8bdfeb9fcd7d1a07 : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c939f1134fcd4f44b1c2c49e1e02087d : SUCCESS in 47m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b92bddb2df9e42b39103455e372b26c5 : SUCCESS in 1h 24m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cefa240a62ee411295f353e46652d534 : SUCCESS in 1h 04m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba1c2ea73f954f58992f5e7824c91d61 : SUCCESS in 58m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/911a6dcdacd14fe983a3959df5ee1baa : SUCCESS in 1h 01m 43s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5f798e760f3246fc91df13d346f58c28 : SUCCESS in 1h 01m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e7b7463b9e0745079dc281f47844df86 : SUCCESS in 1h 13m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/191078db46a147a59dd2d0a8693affd2 : FAILURE in 1h 17m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd70b7b80ba84d3ba8b0dab5e039941c : SUCCESS in 1h 18m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3f647450786b311f31b31c80db931f819019c95d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-17 10:46:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d320f1090b066f65e9c72aba5630749350137088","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 11:50:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82cad9358211409786642a2f4f68d748 : SUCCESS in 44m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e9e19f7bd37459d8b1e92560b0918a9 : TIMED_OUT in 1h 01m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"feef2cbc6408d262b5646de6599b105fd9eb982a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-17 12:28:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/8 : SUCCESS in 1h 41m 33s","accounts_in_message":[],"_revision_number":8},{"id":"e1082d61d735663871a0ddeeb709b65e254ee3ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 13:05:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd711c2d2f5c48daaaa22fcedb8a5951 : SUCCESS in 42m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2262d9470a084d0dbda3ea230ab586b3 : SUCCESS in 2h 01m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a25be8b70d664e89bbe3db28f04cfcb1 : SUCCESS in 2h 00m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06d3a520c9104e659d433a3bbb493daa : SUCCESS in 2h 08m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5e83004966e44f66955d591460bf2559 : SUCCESS in 2h 13m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e026a63a6e3b4cdaaf1570c1210f4d59 : SUCCESS in 40m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e108bb125d9243cf8a850b3f2979174c : SUCCESS in 48m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27d8c873dc654c93946853fa52c64349 : SUCCESS in 49m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9eaa5e144f34025a46f71b8c835d8df : SUCCESS in 16m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5b8f18b8cf443d588c1ee336f5fea57 : SUCCESS in 49m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4daaca20a3db4f5d83de3bf69664d2ed : SUCCESS in 35m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e468fc2e1cd04de4a0211dfcce6384fc : SUCCESS in 36m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba4a0f9b99f34a7180c97dedc169e6e5 : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab12f30479fd4256b78b6c460b907c70 : SUCCESS in 1h 10m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/345ec1de5ac44025b31d314827a570e2 : SUCCESS in 1h 20m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e2f72e33eda64dfc924c811053e1e5f5 : SUCCESS in 1h 04m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0b4bde061134be3a35035316f178c24 : SUCCESS in 47m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fb840db317c846429fa98a899a4e55eb : SUCCESS in 1h 02m 16s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b2dce1627b604f3685e9187aa468a2ed : SUCCESS in 1h 12m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5775cf84ce924fd083d44ea67a1d3b12 : SUCCESS in 1h 11m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fe532929e094a25b00fbbb448218db5 : SUCCESS in 51m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad1636ad09134f33b586313db95f0108 : SUCCESS in 1h 07m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6190d4e598347069ca70b95195aeb60 : SUCCESS in 46m 22s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcda07838a79476fa21b3e2e8969bc08 : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":8},{"id":"e65bc7d74fd8ca8a736ff08b78567669d2388e3a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-17 14:22:59.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"ddb72be2f7930285ee5e59fda3cac119f3ffe7cb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-17 15:37:33.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8916c51c2d1554f33c8a6acc2d7f7b7fff2b0708","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-17 15:46:56.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2de36e56ceb432e3d8411271dd5ffc1411fce885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 15:49:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"cac396f70d36dced7a7ebf4422e299588251d2a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 18:50:24.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0d2175999fa54294a40d99f2ce879001 : SUCCESS in 48m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4b675cba0b3e435b9c4015d055ca272b : SUCCESS in 2h 05m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5e278c58ed184637a8aabe9d1ef96e2e : SUCCESS in 2h 14m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/941d30ba7bf149bf83333f636914604a : SUCCESS in 2h 09m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04fee70c60974460b66ee7ffc890a025 : FAILURE in 1h 45m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27e826a7526f49c7addda06ea75d8c8c : SUCCESS in 54m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee5be790d6a74e929aeb9a94d3b4702c : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26b1c401c9974c7d840f7a3c63e6d897 : SUCCESS in 39m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e326e292441144d7a1d56abb98987d74 : SUCCESS in 39m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c19b8a085249a2a084ba1f0be4a8b9 : SUCCESS in 15m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbafd37dffa443a1947f27388c41380e : SUCCESS in 49m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c2b9f3c5f9c4a0988416affe8a7bad3 : SUCCESS in 1h 25m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a0badd6d0f444e85a0f155ddfc5eb443 : SUCCESS in 1h 20m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/aac18d88451b492990c6c1b80428b12d : SUCCESS in 1h 00m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36366c0b95784b77983afd75a148814d : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":8},{"id":"0ad7b2aeb04d89cbc25d2bd45bd71c203c69e98c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-17 18:50:34.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"7124214230e171ffb49cdb43502941000fd6346e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 19:57:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac4b70ab2fc941dfa5a5a58afe468633 : SUCCESS in 45m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f65b89ecbf942afbebc1b061176eaa1 : TIMED_OUT in 1h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"de2c2d6806637b966e81009f78a6c0515a31e51f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-17 20:28:10.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/815465/8 : SUCCESS in 1h 36m 55s","accounts_in_message":[],"_revision_number":8},{"id":"62d2d87594a5387b108f584397b5eb99791f071f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 21:05:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/632f4c6e726b44f9b9ed317d0018d496 : SUCCESS in 40m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/865e72655b634b64b441fb3d935d1d0f : SUCCESS in 2h 00m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c6e9acfdb40d447daf298793624e5304 : SUCCESS in 1h 45m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7cb3486073464699b379a3d73eff129f : SUCCESS in 1h 26m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/132443e96bb84593a339d0b258fc1311 : SUCCESS in 2h 10m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35e1ca9ce7f8415fa708860598f1d245 : SUCCESS in 35m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76becf29d2fa4001837a9c77b2f34a4b : SUCCESS in 43m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bd8fba1e9a94d2b9d55307a7c782210 : SUCCESS in 50m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae686b98d5ef42b48a04c8dea132f736 : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a5a1c75a769478d8b005aa0913ac371 : SUCCESS in 35m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70e0b45c596140789f281d0b95aa0884 : SUCCESS in 42m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ccc0aa483a849f5b3fd494439119722 : SUCCESS in 36m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21647cfb26cc41bfad99d69beb599079 : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c193f2aa9bf4a8e87f73b16534cb4b6 : SUCCESS in 42m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/609de101e1794903b7175241a25624c5 : SUCCESS in 1h 26m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e637763a0cf048c5b70bb7cd20ef5293 : SUCCESS in 1h 02m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3916cbfb871045c0803c12c81c37cd24 : SUCCESS in 54m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3fdd3bf7c2b640678ccf9c9d65eedf9f : SUCCESS in 56m 18s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3b082e4691d74cb181c8ce0a74779943 : SUCCESS in 1h 06m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bc502114b3d34be297b04c70d8dfc5be : SUCCESS in 1h 10m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b308b21df148452ca3ef17f71b0f9c98 : SUCCESS in 52m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eb5bf3cab4a2426abec0b8a9959b0a9f : SUCCESS in 1h 12m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1860d4a857943a98a2f1306b3c55362 : SUCCESS in 48m 00s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/422cf920ba0a416cb059d78d60ccd721 : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":8},{"id":"e673fe59a4c4c5b8dbc2edee36e069bbc0fa0d62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 21:06:28.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6dd2346517f0cf41401c83599d1881501336b8e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 23:20:55.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/871fa53be5bc48249c4869aa9698d2ee : SUCCESS in 48m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ee9a3aad32bf4af68ffeba8dd7a97d05 : SUCCESS in 2h 12m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9cfa54375afc48ef8d79dc62ed8df7e2 : SUCCESS in 1h 58m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/970c599fb0624f87874bc5ac6e822e4e : SUCCESS in 2h 07m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b3e903ebc86f4443b33bf81991e239ea : SUCCESS in 1h 46m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b8a38c580343d4ab307fd103377823 : SUCCESS in 55m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca0cf509f7348618dade9c2dfca97f7 : SUCCESS in 12m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86e3a284082f498fa50d6aa14afabd12 : SUCCESS in 47m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5a03e92486f47d4b69c5573c5924155 : SUCCESS in 34m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01f8d3e2627435d861450ebb3e1a837 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e21e02d12d84e6fa9b8b300b2378c2d : SUCCESS in 51m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/58d15b232289468fa043501a000db174 : SUCCESS in 1h 43m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/45f238328ed245558672c342cfd0df09 : SUCCESS in 1h 03m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9bf32f19eac40d9b371b999935b600d : SUCCESS in 53m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5f5ac79a1af4663a43971309a1907fc : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":8},{"id":"62c513aec66e3bda7fb7d5e75ecb3c4637bdba67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 23:20:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"be74d13345d4c176daf32ae2e4fdb7baf93740e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 23:27:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1239b07f7a8040d9a608593f031f1dbd : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":8},{"id":"7bce2b1c5b3fba1c4c4434e42bbfe83ca2107990","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 06:14:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5357689002501b22a1027add7b93c59c2a02d54e","revisions":{"c56f06db2925c222a6cd97d44fd8aceb90cd91e0":{"kind":"REWORK","_number":1,"created":"2021-10-26 10:26:03.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/815465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/815465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/815465/1"}}},"commit":{"parents":[{"commit":"efa12cf35c17547af7c490f2627ef4cd93c6d8b6","subject":"Merge \"Check a namespace existence by checking only its own directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efa12cf35c17547af7c490f2627ef4cd93c6d8b6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 10:23:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 10:23:45.000000000","tz":120},"subject":"Remove some scenario jobs from the check and gate queues","message":"Remove some scenario jobs from the check and gate queues\n\nAfter discussing this topic again during the PTG I spent some time\nchecking our scenario jobs which runs in the check and gate queues.\nAfter analysis this patch proposes to:\n\n* remove neutron-ovs-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovs-tempest-multinode-full job,\n* remove neutron-ovn-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovn-tempest-ipv6-only job - of course this job is using IPv6\n  instead of IPv4 but I don\u0027t really think it\u0027s big issue in that case,\n* move neutron-ovn-tempest-ovs-release and\n  neutron-ovn-tempest-ovs-release-ipv6-only jobs to periodic queue - I\n  think that running those tests once per day should be enough.\n\nAdditionally this patch removes definition of the neutron-ovs-tempest-slow\nand neutron-ovn-tempest-slow jobs are those jobs aren\u0027t used anywhere now.\n\nChange-Id: I657881c319d425470277885545240d6a8b66a1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c56f06db2925c222a6cd97d44fd8aceb90cd91e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c56f06db2925c222a6cd97d44fd8aceb90cd91e0"}]},"branch":"refs/heads/master"},"d7c56e32c9f7bf68bcdbfd56de1a750d5a509c20":{"kind":"REWORK","_number":2,"created":"2021-10-26 15:47:17.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/815465/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/815465/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/815465/2"}}},"commit":{"parents":[{"commit":"efa12cf35c17547af7c490f2627ef4cd93c6d8b6","subject":"Merge \"Check a namespace existence by checking only its own directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efa12cf35c17547af7c490f2627ef4cd93c6d8b6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 10:23:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 15:47:05.000000000","tz":120},"subject":"Remove some scenario jobs from the check and gate queues","message":"Remove some scenario jobs from the check and gate queues\n\nAfter discussing this topic again during the PTG I spent some time\nchecking our scenario jobs which runs in the check and gate queues.\nAfter analysis this patch proposes to:\n\n* remove neutron-ovs-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovs-tempest-multinode-full job,\n* remove neutron-ovn-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovn-tempest-ipv6-only job - of course this job is using IPv6\n  instead of IPv4 but I don\u0027t really think it\u0027s big issue in that case,\n  As neutron-ovn-tempest-slow job was multinode job, this patch also\n  changes neutron-ovn-tempest-ipv6-only job to be multinode job,\n* move neutron-ovn-tempest-ovs-release and\n  neutron-ovn-tempest-ovs-release-ipv6-only jobs to periodic queue - I\n  think that running those tests once per day should be enough.\n\nAdditionally this patch removes definition of the neutron-ovs-tempest-slow\nand neutron-ovn-tempest-slow jobs are those jobs aren\u0027t used anywhere now.\n\nChange-Id: I657881c319d425470277885545240d6a8b66a1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7c56e32c9f7bf68bcdbfd56de1a750d5a509c20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7c56e32c9f7bf68bcdbfd56de1a750d5a509c20"}]},"branch":"refs/heads/master"},"73d0bdc12d49aaad07232b162fb1f934ad0fe219":{"kind":"REWORK","_number":3,"created":"2021-11-02 08:39:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/815465/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/815465/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/815465/3"}}},"commit":{"parents":[{"commit":"81f8524527b8fcbfc5d380700b97e72b73bd8cd1","subject":"Merge \"[DVR] Set arp entries only for IPs from the correct subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81f8524527b8fcbfc5d380700b97e72b73bd8cd1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 10:23:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-02 08:38:48.000000000","tz":60},"subject":"Remove some scenario jobs from the check and gate queues","message":"Remove some scenario jobs from the check and gate queues\n\nAfter discussing this topic again during the PTG I spent some time\nchecking our scenario jobs which runs in the check and gate queues.\nAfter analysis this patch proposes to:\n\n* remove neutron-ovs-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovs-tempest-multinode-full job,\n* remove neutron-ovn-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovn-tempest-ipv6-only job - of course this job is using IPv6\n  instead of IPv4 but I don\u0027t really think it\u0027s big issue in that case,\n  As neutron-ovn-tempest-slow job was multinode job, this patch also\n  changes neutron-ovn-tempest-ipv6-only job to be multinode job,\n* move neutron-ovn-tempest-ovs-release and\n  neutron-ovn-tempest-ovs-release-ipv6-only jobs to periodic queue - I\n  think that running those tests once per day should be enough.\n\nAdditionally this patch removes definition of the neutron-ovs-tempest-slow\nand neutron-ovn-tempest-slow jobs are those jobs aren\u0027t used anywhere now.\n\nChange-Id: I657881c319d425470277885545240d6a8b66a1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73d0bdc12d49aaad07232b162fb1f934ad0fe219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73d0bdc12d49aaad07232b162fb1f934ad0fe219"}]},"branch":"refs/heads/master"},"11a8b70136337a2658c84cf1be99cd647a1a813a":{"kind":"REWORK","_number":4,"created":"2021-11-08 09:33:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/815465/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/815465/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/815465/4"}}},"commit":{"parents":[{"commit":"81f8524527b8fcbfc5d380700b97e72b73bd8cd1","subject":"Merge \"[DVR] Set arp entries only for IPs from the correct subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81f8524527b8fcbfc5d380700b97e72b73bd8cd1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 10:23:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-08 09:33:10.000000000","tz":60},"subject":"Remove some scenario jobs from the check and gate queues","message":"Remove some scenario jobs from the check and gate queues\n\nAfter discussing this topic again during the PTG I spent some time\nchecking our scenario jobs which runs in the check and gate queues.\nAfter analysis this patch proposes to:\n\n* remove neutron-ovs-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovs-tempest-multinode-full job,\n* remove neutron-ovn-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovn-tempest-ipv6-only job - of course this job is using IPv6\n  instead of IPv4 but I don\u0027t really think it\u0027s big issue in that case,\n  As neutron-ovn-tempest-slow job was multinode job, this patch also\n  changes neutron-ovn-tempest-ipv6-only job to be multinode job,\n* move neutron-ovn-tempest-ovs-release and\n  neutron-ovn-tempest-ovs-release-ipv6-only jobs to periodic queue - I\n  think that running those tests once per day should be enough.\n\nAdditionally this patch removes definition of the neutron-ovs-tempest-slow\nand neutron-ovn-tempest-slow jobs are those jobs aren\u0027t used anywhere now.\n\nChange-Id: I657881c319d425470277885545240d6a8b66a1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11a8b70136337a2658c84cf1be99cd647a1a813a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11a8b70136337a2658c84cf1be99cd647a1a813a"}]},"branch":"refs/heads/master"},"6ebe8e48e72483981cc8f2e532b264163b414d00":{"kind":"REWORK","_number":5,"created":"2021-11-08 12:15:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/815465/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/815465/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/815465/5"}}},"commit":{"parents":[{"commit":"81f8524527b8fcbfc5d380700b97e72b73bd8cd1","subject":"Merge \"[DVR] Set arp entries only for IPs from the correct subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81f8524527b8fcbfc5d380700b97e72b73bd8cd1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 10:23:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-08 12:14:52.000000000","tz":60},"subject":"Remove some scenario jobs from the check and gate queues","message":"Remove some scenario jobs from the check and gate queues\n\nAfter discussing this topic again during the PTG I spent some time\nchecking our scenario jobs which runs in the check and gate queues.\nAfter analysis this patch proposes to:\n\n* remove neutron-ovs-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovs-tempest-multinode-full job,\n* remove neutron-ovn-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovn-tempest-ipv6-only job - of course this job is using IPv6\n  instead of IPv4 but I don\u0027t really think it\u0027s big issue in that case,\n  As neutron-ovn-tempest-slow job was multinode job, this patch also\n  changes neutron-ovn-tempest-ipv6-only job to be multinode job,\n* move neutron-ovn-tempest-ovs-release and\n  neutron-ovn-tempest-ovs-release-ipv6-only jobs to periodic queue - I\n  think that running those tests once per day should be enough.\n\nAdditionally this patch removes definition of the neutron-ovs-tempest-slow\nand neutron-ovn-tempest-slow jobs are those jobs aren\u0027t used anywhere now.\n\nChange-Id: I657881c319d425470277885545240d6a8b66a1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ebe8e48e72483981cc8f2e532b264163b414d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ebe8e48e72483981cc8f2e532b264163b414d00"}]},"branch":"refs/heads/master"},"914bb65e2d9105da0a96f46ea8ade4fa2fdc0f4d":{"kind":"REWORK","_number":6,"created":"2021-11-08 13:20:55.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/815465/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/815465/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/815465/6"}}},"commit":{"parents":[{"commit":"81f8524527b8fcbfc5d380700b97e72b73bd8cd1","subject":"Merge \"[DVR] Set arp entries only for IPs from the correct subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81f8524527b8fcbfc5d380700b97e72b73bd8cd1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 10:23:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-08 13:20:43.000000000","tz":60},"subject":"Remove some scenario jobs from the check and gate queues","message":"Remove some scenario jobs from the check and gate queues\n\nAfter discussing this topic again during the PTG I spent some time\nchecking our scenario jobs which runs in the check and gate queues.\nAfter analysis this patch proposes to:\n\n* remove neutron-ovs-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovs-tempest-multinode-full job,\n* remove neutron-ovn-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovn-tempest-ipv6-only job - of course this job is using IPv6\n  instead of IPv4 but I don\u0027t really think it\u0027s big issue in that case,\n  As neutron-ovn-tempest-slow job was multinode job, this patch also\n  changes neutron-ovn-tempest-ipv6-only job to be multinode job,\n* move neutron-ovn-tempest-ovs-release and\n  neutron-ovn-tempest-ovs-release-ipv6-only jobs to periodic queue - I\n  think that running those tests once per day should be enough.\n\nAdditionally this patch removes definition of the neutron-ovs-tempest-slow\nand neutron-ovn-tempest-slow jobs are those jobs aren\u0027t used anywhere now.\n\nChange-Id: I657881c319d425470277885545240d6a8b66a1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/914bb65e2d9105da0a96f46ea8ade4fa2fdc0f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/914bb65e2d9105da0a96f46ea8ade4fa2fdc0f4d"}]},"branch":"refs/heads/master"},"d45b8d2e7d013b9fee626796cf029c78f6ff2a0f":{"kind":"REWORK","_number":7,"created":"2021-11-08 15:38:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/815465/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/815465/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/815465/7"}}},"commit":{"parents":[{"commit":"81f8524527b8fcbfc5d380700b97e72b73bd8cd1","subject":"Merge \"[DVR] Set arp entries only for IPs from the correct subnet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81f8524527b8fcbfc5d380700b97e72b73bd8cd1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 10:23:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-08 15:38:08.000000000","tz":0},"subject":"Remove some scenario jobs from the check and gate queues","message":"Remove some scenario jobs from the check and gate queues\n\nAfter discussing this topic again during the PTG I spent some time\nchecking our scenario jobs which runs in the check and gate queues.\nAfter analysis this patch proposes to:\n\n* remove neutron-ovs-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovs-tempest-multinode-full job,\n* remove neutron-ovn-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovn-tempest-ipv6-only job - of course this job is using IPv6\n  instead of IPv4 but I don\u0027t really think it\u0027s big issue in that case,\n  As neutron-ovn-tempest-slow job was multinode job, this patch also\n  changes neutron-ovn-tempest-ipv6-only job to be multinode job,\n* move neutron-ovn-tempest-ovs-release and\n  neutron-ovn-tempest-ovs-release-ipv6-only jobs to periodic queue - I\n  think that running those tests once per day should be enough.\n\nAdditionally this patch removes definition of the neutron-ovs-tempest-slow\nand neutron-ovn-tempest-slow jobs are those jobs aren\u0027t used anywhere now.\n\nChange-Id: I657881c319d425470277885545240d6a8b66a1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d45b8d2e7d013b9fee626796cf029c78f6ff2a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d45b8d2e7d013b9fee626796cf029c78f6ff2a0f"}]},"branch":"refs/heads/master"},"5357689002501b22a1027add7b93c59c2a02d54e":{"kind":"REWORK","_number":8,"created":"2021-11-17 10:46:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/815465/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/815465/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/815465/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/815465/8"}}},"commit":{"parents":[{"commit":"322ed0131d7b55411edf69903e15a3dc010f1b84","subject":"Merge \"Replace \"tenant_id\" with \"project_id\" in OVO base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/322ed0131d7b55411edf69903e15a3dc010f1b84"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-26 10:23:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-17 10:46:17.000000000","tz":60},"subject":"Remove some scenario jobs from the check and gate queues","message":"Remove some scenario jobs from the check and gate queues\n\nAfter discussing this topic again during the PTG I spent some time\nchecking our scenario jobs which runs in the check and gate queues.\nAfter analysis this patch proposes to:\n\n* remove neutron-ovs-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovs-tempest-multinode-full job,\n* remove neutron-ovn-tempest-slow job from both check and gate queue as\n  slow tests are already run also in the\n  neutron-ovn-tempest-ipv6-only job - of course this job is using IPv6\n  instead of IPv4 but I don\u0027t really think it\u0027s big issue in that case,\n  neutron-ovn-tempest-slow job was multinode job, unfortunately\n  neutron-ovn-tempest-ipv6-only is single node job and for now it isn\u0027t\n  possible to make ipv6-only job to be multinode job so we will keep it\n  like single node job and hopefully move to be multinode job when zuul\n  will provide required data in the job\u0027s inventory,\n* move neutron-ovn-tempest-ovs-release and\n  neutron-ovn-tempest-ovs-release-ipv6-only jobs to periodic queue - I\n  think that running those tests once per day should be enough.\n\nAdditionally this patch removes definition of the neutron-ovs-tempest-slow\nand neutron-ovn-tempest-slow jobs are those jobs aren\u0027t used anywhere now.\n\nChange-Id: I657881c319d425470277885545240d6a8b66a1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5357689002501b22a1027add7b93c59c2a02d54e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5357689002501b22a1027add7b93c59c2a02d54e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
