)]}'
{"id":"openstack%2Fneutron~816349","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Idfa763df8c60d8ae46cd6351d1b6dc7d950b4c67","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1948642-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-08 09:09:53.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-08 12:46:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idfa763df8c60d8ae46cd6351d1b6dc7d950b4c67","subject":"Don\u0027t setup bridge controller if it is already set","status":"MERGED","created":"2021-11-03 11:13:26.000000000","updated":"2021-11-08 12:48:21.000000000","submitted":"2021-11-08 12:46:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816349-bug/1948642-stable/victoria-stable/ussuri","cherry_pick_of_change":816457,"cherry_pick_of_patch_set":1,"meta_rev_id":"4771ece1b2b03534e7e3e937cdbe1839efbfa73a","_number":816349,"virtual_id_number":816349,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-08 12:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-11-08 09:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-11-08 07:53:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-11-04 09:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-11-08 09:09:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-03 13:40:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-03 14:15:29.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-03 14:52:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 09:21:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-11-08 07:53:07.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"bdbf2afbcd8a48e94f7af643caa0c3a53a3dc38a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-03 11:13:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"45e8dc3a39fad7dcd1169f7c2e91008a40cb8ae2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-03 14:15:29.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/816349/1 : FAILURE in 2h 23m 40s","accounts_in_message":[],"_revision_number":1},{"id":"687a527b8ca276851cba849bbfd42d1b39d634be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 14:52:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/a16fbf262e94458b91a4dfd4a0876a7c : SUCCESS in 1h 05m 49s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/350a515284d64c0487a9da86e31919e8 : SUCCESS in 2h 17m 05s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/4e8c41ef1e6e47c0a82a9ab600b72ef8 : SUCCESS in 2h 12m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/7e94b81695ff482d828d179f3cf3d660 : SUCCESS in 1h 55m 19s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/c457cc3ff55f4c689e4fa29e8b051247 : SUCCESS in 2h 19m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d32c367e2cb94d689cc6b9b6dfbb6a35 : SUCCESS in 44m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8777bdb8b2b5420a91a9ca7ae030af9b : SUCCESS in 15m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fb4ae515c654baea9632b68bb21b3bf : SUCCESS in 25m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9cd18c7f352405da877ced9f6bfe4b5 : SUCCESS in 27m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cef1a8340a8470593d5b423d79797f4 : SUCCESS in 31m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93429e316053477c9b5ca28a56a6644a : SUCCESS in 11m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d0bed23bc6644779a1acf3728f3700d8 : SUCCESS in 1h 02m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3a22dcdd7692416db9b4e55885300488 : SUCCESS in 1h 23m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccc13b71ce3f4e4f8ae943c61fbdc597 : SUCCESS in 1h 25m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e73e2462d4684bd2bf12f4d025e0fbf5 : SUCCESS in 1h 29m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ee2940ca620434cb4d49108fea97486 : SUCCESS in 1h 33m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dfc2373e3b6b403fa32cd4eadc218802 : SUCCESS in 1h 13m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0d1ecc88fceb4c0d9d20a908a9bc2b34 : SUCCESS in 1h 12m 01s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5230a0f498cf4addab9c016b81123cc7 : SUCCESS in 1h 05m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f1037da2022c42c4bcedeb256ca0c37b : SUCCESS in 1h 00m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7fad483ed77e40b287e2ec462a274c18 : FAILURE in 1h 28m 55s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0e3cfe7739d34512853007bb6acb1494 : SUCCESS in 1h 17m 47s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b7c3eb2df3444aa4960103c5984dac7f : SUCCESS in 1h 12m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/74455c1eff164bfa8921f94747e1073b : SUCCESS in 59m 37s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d89f3dbe87764fd1acb62677f15c8404 : SUCCESS in 1h 36m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4af4dd65b823494fa649fef988b3220a : SUCCESS in 1h 02m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4f07c14353e1456f9cfd5a1c335f39fa : SUCCESS in 1h 16m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/70f00d0bf97e41ce8ed627854bd74229 : SUCCESS in 38m 30s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/aebdf49225dd4801bd5d46e6ee97713d : SUCCESS in 1h 19m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e68b8edc93034af1838d54a1a7aabbfe : SUCCESS in 57m 37s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cc60186232b848aa90677c2e648bc544 : SUCCESS in 3h 36m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e9c1285e107d49aebe589354c8185b09 : SUCCESS in 1h 21m 24s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d72f795026764b43b2e70b026c0f39a4 : SUCCESS in 1h 43m 06s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7f9f8587c281431db33ab11c67ae9a3d : SUCCESS in 2h 09m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5871a5a70d1a4c4d9bbf3725657f4b5b : FAILURE in 1h 46m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d495f22d24b74b7b417c71a4f017530903231c1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-11-04 09:21:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26b0504545ed76d9a5d0ce539bd81ed32558a2f3","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-11-08 07:53:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a3e16b39065f98e7e33fa2fead4e903ec339b5f5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-08 09:09:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e80081e70c7a95ed5677546eab6f06b9925c67e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 09:20:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b023260f5191c22b1d24f98e0c218e9fbbfceb89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 12:46:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/51d9d5e7af3146eea03cebd5d7893883 : SUCCESS in 38m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3a5f8e3eee4498e9cdda7174e1dd8e9 : SUCCESS in 15m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87c08f4873f94a1083b86ce906d5f1d6 : SUCCESS in 21m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b61ae853427447a8a32aa2e7004d401a : SUCCESS in 24m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed530b98ab34d8a9051cc531fbf11e9 : SUCCESS in 10m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cefb420c31c2495188d655bd5c581555 : SUCCESS in 47m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0e2a4a5d0aa44178897dbebbc413adc5 : SUCCESS in 1h 38m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf4a1c914750494890fae29f0aae2413 : SUCCESS in 1h 11m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7152718e677f4c83b723cc78af81e3db : SUCCESS in 59m 52s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c433c02604364cf5b4b9f9f021c98834 : SUCCESS in 58m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5764c4ae52874b3aa1f8f2087273b676 : SUCCESS in 1h 05m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f5d56819370a4050ad2e6debf4abd398 : SUCCESS in 1h 06m 49s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70fd0529a6874fd7934d897f359e2417 : SUCCESS in 1h 22m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7da31386fe84dadb82ff864e7270579 : SUCCESS in 52m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ef2b65a28786422690c9390a73249db8 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"aa53655ad1795a723ba157134abc7602799e3320","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 12:46:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4771ece1b2b03534e7e3e937cdbe1839efbfa73a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 12:48:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13989541689a4a199829950cec102625 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11fe2bff17d8ad27c0c49037fcd36203f9baa32d","revisions":{"11fe2bff17d8ad27c0c49037fcd36203f9baa32d":{"kind":"REWORK","_number":1,"created":"2021-11-03 11:13:26.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/49/816349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/816349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/816349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/816349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/816349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/816349/1"}}},"commit":{"parents":[{"commit":"72fb5944735be97e6bb0dad0b153f2a3ccd6d63a","subject":"Merge \"Fix OVN migration workload creation order\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72fb5944735be97e6bb0dad0b153f2a3ccd6d63a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-25 08:59:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-03 11:13:26.000000000","tz":0},"subject":"Don\u0027t setup bridge controller if it is already set","message":"Don\u0027t setup bridge controller if it is already set\n\nSetting new controller for bridge every time when neutron-ovs-agent\nis restarted or is doing full-sync may cause some short data plane\nconnectivity loss and is not needed if same controller is already\nconfigured for the bridge.\nWith this patch neutron-ovs-agent will first check if controller is\nconfigured for the bridge and if it\u0027s the same as what should be\nconfigured, it will skip setting it up.\n\nWith this patch also protocols added to the bridge will be first checked\nif they aren\u0027t already there and only missing ones will be added if\nnecessary.\n\nSetting of the connectivity mode and inactivity probe is\nalways performed as this don\u0027t cause connectivity issues and is cheap\nso we can always ensure that those parameters are configured properly.\n\nConflicts:\n    neutron/agent/common/ovs_lib.py\n    neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/ovs_bridge.py\n    neutron/tests/unit/plugins/ml2/drivers/openvswitch/agent/openflow/native/ovs_bridge_test_base.py\n\nCloses-Bug: #1948642\nChange-Id: Idfa763df8c60d8ae46cd6351d1b6dc7d950b4c67\n(cherry picked from commit 11d166be683d36ba51e3f85ddea45cb0a1fcf8db)\n(cherry picked from commit b1eccf5a2deaf6d18ff18dfacfbcf7ef7fdb781f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11fe2bff17d8ad27c0c49037fcd36203f9baa32d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11fe2bff17d8ad27c0c49037fcd36203f9baa32d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
