)]}'
{"id":"openstack%2Fneutron~816612","triplet_id":"openstack%2Fneutron~master~I85f7749e7e6186b9e3a7534bce6c306b18702333","project":"openstack/neutron","branch":"master","topic":"stateful-update-acl","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-30 16:47:33.000000000","reason":"Rodolfo Alonso replied on the change"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2021-11-30 19:44:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85f7749e7e6186b9e3a7534bce6c306b18702333","subject":"ovn: enable stateless-security-group api","status":"MERGED","created":"2021-11-04 03:04:06.000000000","updated":"2021-12-01 09:44:35.000000000","submitted":"2021-11-30 19:44:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816612-stateful-update-acl","meta_rev_id":"896cb7a3579bc49ef228fda3e1f9e6b797cc8d70","_number":816612,"virtual_id_number":816612,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-12-01 09:44:35.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-30 19:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-01 09:44:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-12-01 09:44:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-11-30 13:23:44.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-30 01:08:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-01 09:44:35.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-04 04:26:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 04:49:34.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-04 06:09:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 06:23:34.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 08:39:49.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-11-04 13:48:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-09 20:41:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-12 08:41:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-11-30 01:08:41.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-11-30 13:23:44.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"}],"messages":[{"id":"e053c0fbb5ef1ab24ec1bb6d5386e9aeaead7b0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-11-04 03:04:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9831cd9a5f0f0419d42fbac788074332c791847d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-11-04 03:08:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3caabd7825f55a9f0df9bbeb5087f542a45aa944","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-11-04 03:17:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f29a7902dcc39c47d9d5bf7b81c73cb27bb6870e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 04:26:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f14b9a6d84644dea0e5f9c1c08df5f3 : TIMED_OUT in 1h 01m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da13bd7e0b1b4ab088f7cc389181923d : TIMED_OUT in 1h 01m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c34ca18c52f89a0fe6ec5c58d3d001f4373cc06f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-04 04:49:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/816612/3 : SUCCESS in 1h 30m 06s","accounts_in_message":[],"_revision_number":3},{"id":"ddf67cc4d9a6f056f9e0209e96ddae46aef13b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 06:09:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/20e56b9a516d414495a1fdc107985d49 : SUCCESS in 49m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dd880b62580442459d2a2c68b7997e20 : SUCCESS in 1h 59m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0763012d5fd3471eab906e9edd8fc29e : SUCCESS in 2h 23m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e939d7d15994d8d9b87480d2d9cc2fb : SUCCESS in 1h 30m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1a4d505295534e79b6013304b860f5ca : SUCCESS in 2h 42m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5ee03a23dd90426fb155d1b2ef2b09f5 : SUCCESS in 33m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77583d674eff4228a6a9f97b19bc18ba : SUCCESS in 48m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3038dba347e4420eba8693789d3881a6 : SUCCESS in 56m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37bd7cb41a254df3a01e89c96067c602 : SUCCESS in 12m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50a44491b16040cbaa14e1e60e35f681 : SUCCESS in 43m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44b46191f6604ca884967b7648eb40a7 : SUCCESS in 39m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d94b82b69744c9a87dbe7e2773788cf : SUCCESS in 41m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/822bf178c9ce433387ab34f3f759d3a8 : SUCCESS in 11m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73d56a06523e41708249d233dd0689dc : RETRY_LIMIT in 15m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7aef48e743884a26a0ac35ff1f40c01c : RETRY_LIMIT in 17m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c687c2b048c8433cb4bdd672dc6e7299 : SUCCESS in 1h 14m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c88fb111c2241d3b8e71cb59c6d2252 : SUCCESS in 56m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c314c795c524406fbc2206cc53b8aefe : SUCCESS in 58m 45s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e8542ce9de9483eae86085142d1bdac : SUCCESS in 1h 08m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2b4c72b836cb46cc82932888b216fa62 : SUCCESS in 1h 17m 47s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/a7219a2cdcfd4d85b0da435216dde056 : SUCCESS in 1h 25m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7d89b94bbb148b3b0c60c2062031bd7 : SUCCESS in 1h 01m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/503b462aabdd4820b3a9aaa8c6cdff10 : SUCCESS in 1h 00m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e5cfaee5fdb84f3a8a494ca102a6d70a : SUCCESS in 30m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9a13339230c34858a47a5144c2d654d1 : SUCCESS in 1h 09m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b3648069ca4644e4b477a0b641c9a648 : SUCCESS in 1h 20m 05s","accounts_in_message":[],"_revision_number":3},{"id":"8938ab7b70c8733bbcac9cb48855c9b85f680897","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-04 06:23:34.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/816612/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/816612/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"4aac431f61cc08621e23debf29da55396c8ec653","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-04 08:39:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9cdbb82ca0aa6b0588d2fe49d2f80b7c15434857","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 09:39:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2768e8ca13843eeaae58c9b72a7c019 : SUCCESS in 55m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5db3c802c204c31895d0d27a141ffc4 : SUCCESS in 55m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9df6bb646ed42dcf2f940ccf05c91bc38ccc96bd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-04 09:46:28.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/816612/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/816612/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"211ab8c9eae2be5c430f264d2870af5d06b3c74a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-04 10:27:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/816612/3 : SUCCESS in 1h 47m 01s","accounts_in_message":[],"_revision_number":3},{"id":"f207cf49a490a56ad8724426694051ba6d79d8a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 11:13:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e4b3a7409ac7480499957095d5c92744 : SUCCESS in 43m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/eb75c90848814a38bbfebd99e9a1ffb6 : SUCCESS in 2h 28m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5beda44a803e48fe822ad3a3d366bc60 : SUCCESS in 1h 38m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/594395c298004f639229d3decf67ffd4 : SUCCESS in 1h 37m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/363d2578e6604f428aea5214f97f979b : SUCCESS in 2h 02m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4cea4a257fc94e72b5d99a5f517c0210 : SUCCESS in 36m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18cc1420ef2542dc963c28b4074669c9 : SUCCESS in 1h 01m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fccb79c97aff446bacbd03d544220f4c : TIMED_OUT in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6982182b38464000b7b09ddacd229c08 : SUCCESS in 13m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2542a022249f412f9bcbdbd39e248d70 : SUCCESS in 41m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/892145cdd22f41c391fb972d88f9143f : SUCCESS in 39m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e605c4f0b3f5479b92da34dad31887bb : SUCCESS in 34m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b28fc1300c6a494eadf2cc914f8b473b : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/02eaf2fcc583442b9d8b1551f3325174 : RETRY_LIMIT in 16m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d43af257868449ba17864f2dfbc2745 : RETRY_LIMIT in 16m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4d0b576b5a034cc6a97e4fd1912a545f : SUCCESS in 1h 02m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/48504211875742f8a7a1aabf38401cae : SUCCESS in 1h 00m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/11849f27aca34694a5a7803cad81cb97 : SUCCESS in 59m 58s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/754ea49d40014365bdc9c1e83ad5b9ed : SUCCESS in 1h 15m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9a0e3852ec1a4e648ff55bab511279cb : SUCCESS in 1h 23m 44s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/f57da78e4fd949c2900fa5afdc8b62eb : SUCCESS in 1h 41m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20b9e22e0fdc4d768856df4aafedd21a : SUCCESS in 1h 02m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d2fbb1ffdb4949a3ba0b4f0ad7f19e01 : SUCCESS in 56m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0ed97f4a9ec94c579a620fa12448ccc7 : SUCCESS in 32m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e350967754c452d8e9af064cd898047 : SUCCESS in 1h 16m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bbaa9de4705f4275b8a43e258517a504 : SUCCESS in 1h 19m 13s","accounts_in_message":[],"_revision_number":3},{"id":"207aa84ac1e0e8c75b475741d11585804403766d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-11-04 13:40:36.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"d0ca806ddfe57a612dc23e70165ae0f6bcc97f79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 15:04:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b42a2507f0e43c587077c5b15c5b114 : TIMED_OUT in 1h 02m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3de7c63730704063a8ee72018aa66cf8 : SUCCESS in 48m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bc09d4adb188a9e3d8a573529344753d5a0bc9f5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-04 15:41:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/816612/4 : SUCCESS in 1h 49m 48s","accounts_in_message":[],"_revision_number":4},{"id":"a1295fa3537164edc99c5f51a6201c63160aef03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-04 15:59:30.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/816612/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/816612/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"6ced4b69c1e39cf93a46ed0eb5240df789204204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:36:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5e3b10aa0d024e05b994078d360891c4 : FAILURE in 46m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3e9b4eb92dba44be9e6be2d35cdad82b : SUCCESS in 2h 21m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b9f5f7a6b10647d6b435750aba393740 : SUCCESS in 2h 19m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d60ca513cac741f083d0d0bcfe3c38c1 : SUCCESS in 1h 56m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5954360c32414adcaa8c4f44df554aa6 : SUCCESS in 2h 27m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/46cc9f3a0cc4428985ce4c25a1b80fe3 : SUCCESS in 45m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/087396e9075a4dcfb57d83916611919e : SUCCESS in 44m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95be852006de4a6b9b07a358b4b718e8 : SUCCESS in 49m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e701b599223455184242de394a31369 : SUCCESS in 15m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dd3d365e5944329a79963f9cd2ba929 : SUCCESS in 41m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f451d0547fdd4198b020b44cebc70628 : SUCCESS in 48m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d00385c6e9804e46b14e1e4131bb7b9c : TIMED_OUT in 42m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf02db556cf4328b006d679272c9e45 : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48a548d46a45419687bb1bde6f33ffb7 : SUCCESS in 1h 05m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/758e209e5bae4f7d865926367e58c042 : SUCCESS in 1h 30m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9788c160387349c6ae2e15ed64cf5601 : SUCCESS in 1h 30m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/14b91167fc3d43e0a81dd05a40111f3a : SUCCESS in 1h 13m 48s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/03be1527f8cf4bb6a6d1d6ca9791e01c : SUCCESS in 1h 19m 07s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9192228d5f0c4fed802dc6543f7369e9 : SUCCESS in 1h 23m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/adc0b10d35f543e299796b520cc42cef : SUCCESS in 1h 21m 19s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/dd483388ac9b48db83b65108ebc983f6 : SUCCESS in 1h 33m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85a8175787df48b98be3928b93d91719 : SUCCESS in 1h 15m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d6f3bccd71a94434b7777d4710582c22 : SUCCESS in 1h 07m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/31823f50234a43b1bc95175b1790d5a3 : SUCCESS in 31m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b83b2426c4434cdbb532f2130a4de051 : SUCCESS in 1h 24m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/46451783457a4fe58a5b84d9d621942e : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":4},{"id":"f20f332fb465bfdc1458677f3b853266b7891e64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-11-04 16:46:12.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"ceefd8677e3b91745ce027eb9a032d3df45cfa2c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-04 17:01:52.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/816612/5 : FAILURE in 13m 47s","accounts_in_message":[],"_revision_number":5},{"id":"8c1f64359f726bc1ee12e7b3deb2595623dde586","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 18:03:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6abd1b1280e4e97863c5700dee79153 : SUCCESS in 54m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/256c646990204653af35c7507f6330eb : TIMED_OUT in 1h 01m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d32afbd3d7a33a49400a28b649ea5e90bef50b91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-04 19:19:27.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/816612/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/816612/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"f79c2c0197435dbfbb9dbd50dffee451f0ed20c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 19:31:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e9332db04b6a4454b15fa047176b2584 : SUCCESS in 38m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8fe7cc05103d4e08a303ff88de960f65 : SUCCESS in 1h 51m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bd7ae9792007426cb75c48f02eaa9417 : SUCCESS in 1h 52m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3c50bb90d0dd46fe8625ddbf95317276 : SUCCESS in 1h 36m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4842650258294b8f8310af517c982375 : SUCCESS in 2h 20m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d67c421faa434d5982bdd8d08ae8bec4 : SUCCESS in 42m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42430a7c111b4e3b97809f246e2c3d1f : SUCCESS in 44m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a27c4bf1456c491d904f21aef3916edc : SUCCESS in 55m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd2c797a0f849f694d36d8c499a015e : SUCCESS in 14m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a7c956f2bce4088bc51cd1e6d06d835 : SUCCESS in 41m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11b010042f064e74a4d600664ffa83ec : SUCCESS in 39m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad5a0dc50d9a491f9dfd13667d333223 : SUCCESS in 41m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a4d22bee974a1db7c659899ee71897 : SUCCESS in 12m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a538c856d3fe427db9a05a76a54e2ecf : SUCCESS in 48m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/06b6f72a9d704c4eb1d47a336ccb2c64 : SUCCESS in 1h 32m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/98b95f7e0a2046d8b1e09dc0e6c747c5 : FAILURE in 29m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ffcde3351d204f25bcb2d6b66ea6a099 : SUCCESS in 56m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5087d309b2f342c69a44669d475ba831 : SUCCESS in 59m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2b33977b30634cc3a4d60c900d6874a0 : SUCCESS in 1h 14m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/653cc48494a64250a96933def16469a5 : SUCCESS in 1h 29m 42s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/fdd42ad2309b433ab5bf9b0a0627a52c : SUCCESS in 1h 45m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3336678f06dd4010a562b067a1dcc608 : SUCCESS in 59m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eec1cd369ed144b7bacc0b6cfc77a508 : SUCCESS in 58m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/20056dd396494a0187befbe647609b34 : SUCCESS in 34m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b297cdb233d24adaafd6fa2f87102b00 : FAILURE in 26m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e6614314e6c7415381f0686bc2a1c90b : SUCCESS in 1h 22m 08s","accounts_in_message":[],"_revision_number":5},{"id":"956c41da1cd132a659f32af943257bd9183c6396","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-11-05 00:30:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7fe73bbfff5087769cd899eb35083e02ca53ac4f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-05 00:45:22.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/816612/5 : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":5},{"id":"afb29fa7d6714c4d2bc96df4db694d49f1879599","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 01:31:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36c54fa396574ac7bc1276d0b883b6f2 : SUCCESS in 58m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2995b9073314c169c1c8cf6d32c0378 : SUCCESS in 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3c568771e985a14d7c8a8fdc86703dabfa11bc6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 02:45:06.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bfc8e6a024aa443b8f2549ff56ea9d1d : SUCCESS in 39m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/126b615d542e4ac4a8a4fe5cd4ffdcd1 : SUCCESS in 2h 11m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3bb029d5398344aebc2c1736d0219cc7 : SUCCESS in 2h 09m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7d82becedeeb459db0be4fa3d9ce3078 : SUCCESS in 1h 47m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b965d8cfcdea40fcb9ca943e1488eeef : SUCCESS in 2h 11m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b25ae9527064317a0dee07d69a2c13c : SUCCESS in 34m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03f693f799cb4fae97bc56c3046713fc : SUCCESS in 46m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e660173aeaa448d790e53d1ceadb3341 : SUCCESS in 54m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2d6aa228484a64be4a7cc86b2a9402 : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/860091990d034ece9f4a4ca12f5153dc : SUCCESS in 37m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b07db70012c4c28a4c5f0844c39a6fb : SUCCESS in 35m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f500f63d60404898b663b1a28d657763 : SUCCESS in 38m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edda26fcdedd4bb9865c65111b9e8cb1 : SUCCESS in 9m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0b9a61465d14c1e84ad4315c58ecde2 : SUCCESS in 51m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc96617eec964ef6bc86a1b5db6e56a9 : SUCCESS in 1h 29m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4f558369f4ff4d4b91767967e60a8338 : FAILURE in 29m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fda2208490914037af70322fc1b7e6ca : SUCCESS in 55m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/113c9e71c9294bc8b4cd65586c45ed14 : SUCCESS in 57m 20s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ecc0df8c40f0461db69c6ef01e5850eb : SUCCESS in 1h 05m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d3ac9b3d83294da793024dee22e34a0f : SUCCESS in 1h 07m 41s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/13690b81a73c4bd1b2a82450477fbefa : SUCCESS in 1h 25m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c97e6191330418481af2b4a6c30161a : SUCCESS in 53m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/60c758b28ca44d9a8caedb290e284463 : SUCCESS in 59m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/03204233dc794c7e88d8872c9f86501d : SUCCESS in 28m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5bbb6926b031410d94c747231cfa23dd : FAILURE in 27m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/88898239b107468c94399bc760b49feb : SUCCESS in 1h 12m 19s","accounts_in_message":[],"_revision_number":5},{"id":"ca31d2267b8d4c2cf45a5f1df00cb615be1c62d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-05 06:35:18.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/816612/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/816612/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"08cd6d59848f557ee7a069db244e3ca5a7ac86bf","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-11-08 18:29:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5e794ab9adb85178e7605703109ee4a20f796fb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 18:31:20.000000000","message":"Patch Set 5:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":5},{"id":"f6d925501e2267bf838064a366d9b43e31e690a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-08 19:43:28.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/816612/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/816612/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"c038111b1678255347ad443f9891f4368e3e4c00","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-08 20:05:23.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/816612/5 : SUCCESS in 1h 34m 32s","accounts_in_message":[],"_revision_number":5},{"id":"7ef85f7786e4c83802e20ce8794759571fba3a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:01:29.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6b4c426adba9409399c047d806014b38 : SUCCESS in 45m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/820d515092494fd987114e1bc0b1ea76 : SUCCESS in 2h 14m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/776e31a3b12d45ab85f77bfa7322a995 : SUCCESS in 2h 26m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/340072f6092343a3b325388b68bc4079 : SUCCESS in 1h 37m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b22ad449f2784285bd3f7653a28aa18d : SUCCESS in 1h 51m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8ebe22809e0f4776ba429a05e578b90a : SUCCESS in 47m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0557409c72b54c6daf5e91ec9b7e6c49 : SUCCESS in 44m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62508ff6bc3142e2824a9f3656bcfcbd : SUCCESS in 55m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759fd607a57e4c10964f6b79a045410b : SUCCESS in 14m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/852039a625f144878daacf5d7c09c23a : SUCCESS in 40m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f17d1c2d67734ce58bf63157cd816040 : SUCCESS in 38m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c99e03cbceea42589a2105968a07fa63 : SUCCESS in 39m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df2f0f2f44745b3a473e2ffba5d13a9 : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/afb83a488843401d9fc2a346b9c24807 : SUCCESS in 53m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a78fa527f0ed47c6b1940d1bad7909fa : SUCCESS in 1h 32m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2c9520f049444f1a8d76af13ead937a5 : SUCCESS in 1h 09m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5e486be17d9459d979cc99d2c4aa7de : POST_FAILURE in 1h 02m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/401e350a0f924b29be313ff8b993fe62 : SUCCESS in 1h 03m 11s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/391b3d5e28aa4514aa9f345592b65037 : SUCCESS in 1h 15m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b64aff392da1432a939945cede197a82 : SUCCESS in 1h 16m 16s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/657935481e61462b87111c222bf11e10 : SUCCESS in 1h 38m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aad9cf39f29b4f51988f2d4f835e0b5d : SUCCESS in 54m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fb8c86ce8e5f45d5a979001cf9800646 : SUCCESS in 1h 00m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8c03dda579f447e68fdf9ffbf7edf5ac : SUCCESS in 32m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/993031a7167f43078b5c3d74b34ccea5 : SUCCESS in 1h 10m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dae02c455c3646b8aac22fd45da2fe47 : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":5},{"id":"b73f66756708e22318159534c691666b9c3c5e02","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-11-08 21:07:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cf983c9f59bead475ccf67fa3980f69b4192d4d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 22:05:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9efc976995904c9a893199552caf57f4 : SUCCESS in 51m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c029d1c3797440b28ccf5e6851838553 : SUCCESS in 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"17ddeed94e9b86e134726dbf1a1895fbe7f94e98","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-08 22:21:16.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/816612/5/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/816612/5/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"1fea24ee59962ec725d99907fbdbe513c1b4dc31","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-08 22:59:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/816612/5 : SUCCESS in 1h 50m 26s","accounts_in_message":[],"_revision_number":5},{"id":"9dc3dd32f10f9b1b66b994480deee9ec913f6823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 23:20:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/69147f49ab9643adbc2514174ee0e2cd : SUCCESS in 44m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1337e15f38344d12af1cc36649784351 : SUCCESS in 1h 54m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3c931e8ab027456080cb2059291130aa : SUCCESS in 1h 54m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b527359fc3b4589abd689ed18aa63bc : SUCCESS in 1h 24m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/20342d43a01f46b1b190d68493d2c627 : SUCCESS in 2h 07m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bb0827801c704bdbbb64bc93f45b46f6 : SUCCESS in 38m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2cfe58626c741af8497d6d92e94c44c : SUCCESS in 48m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3f48167be9543bca87b882e24ab798a : SUCCESS in 46m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ce80c68f344514b80c5f69ddd60748 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39dffab5a5fa441bb4ad8550abaddf94 : SUCCESS in 40m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c00643067bc040b1a9d0ca8dc64d6ae7 : SUCCESS in 34m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce090a4e88b34adcb0240ccfd7b33c67 : SUCCESS in 40m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85626d27fed84b82b5e6f7f490ecc951 : SUCCESS in 11m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8735c81d51b4116b20df761c561ec4e : SUCCESS in 51m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff3ab74c437447f388e0dfe099c40594 : SUCCESS in 1h 44m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/361e73e074c846f19a1a6216e54266c2 : SUCCESS in 1h 05m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f66643c13e944e4b5efd8f31b3b21b2 : SUCCESS in 55m 14s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d15885db93643f9b460d1c9b46d2460 : SUCCESS in 1h 00m 40s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/62a8bd1021a140d5a9049d06c8ee48a4 : SUCCESS in 1h 21m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b50bffda6dc479389fd4e8ca17f9fde : SUCCESS in 1h 12m 40s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/68fdb236fef44f39b5c71d034af3b77c : SUCCESS in 1h 29m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd6dcdcbb766406496f0440854458c0e : SUCCESS in 1h 03m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d9c42b366adb4ac196a3a2d6eed5a196 : SUCCESS in 1h 01m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1288ec106824476fa99b9da33f963b58 : SUCCESS in 30m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0bde979201a9410da5698fd9a4df4e84 : SUCCESS in 1h 36m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/30ff6a04021744e0a872f7058688f341 : SUCCESS in 1h 26m 34s","accounts_in_message":[],"_revision_number":5},{"id":"cc695a7f63791045500ba189e8f682c3762b811b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-09 20:41:53.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/12/816612/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"929db9fc34969f626053987cb20343fbd35f81a5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-10 04:37:25.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/12/816612/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"0e35e229f82b4250d5273f3fe9a81a58758e4a33","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-10 09:23:53.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/12/816612/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"4e6dd71e05e5ece2915d8b811bd9c9473c85da0c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-10 12:39:15.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/12/816612/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"aae98cebbd850d923652c8a0bdd1a37d643bd4be","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-10 13:37:10.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/12/816612/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"1c0bb51f8e497e26ebe7316a5de7f7b3682ca753","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-12 08:41:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1a6a0f1a59c20d193e961a18abaf2d6fcf4b7681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2021-11-29 18:35:35.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"18ba5c3aec3b952ac98367364d6c7c906d26bdd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 19:19:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2622a47a5f864521b8f4ac3715270342 : SUCCESS in 41m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f39d3555757a4cd7ae3af2316ef17bc0 : SUCCESS in 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"53cebfb8349a45b4b78f67dfe910ba365356ea71","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-29 19:36:06.000000000","message":"Patch Set 6:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/12/816612/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"7d2880597b87bc6ac3a042bac0eab9092a5b109c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-29 20:26:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/816612/6 : SUCCESS in 1h 50m 14s","accounts_in_message":[],"_revision_number":6},{"id":"e3590a4cccb81578e9bc9edc15ac97e1660a6963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 21:03:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0b707f99c52047e4a822384f30bd457d : SUCCESS in 48m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/74d4dee5b0704f32af10b30ef755924f : SUCCESS in 2h 19m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1d3803c072c84aee8506c7d0993a16e0 : SUCCESS in 1h 54m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8829d9c44f4f43349248c5e172064b3f : SUCCESS in 1h 40m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7d00cff270ea4ac5b65d87436f66e124 : SUCCESS in 2h 18m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1e95c80846d4a30bc0b045e72b96dbd : SUCCESS in 47m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af138beea7e84940ab377ea459cd155b : SUCCESS in 29m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e5fab42c43340a3aea6adaf94754068 : SUCCESS in 27m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c028343b8997418784e161c5bf5c2cd3 : SUCCESS in 13m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18eb4dac6e2044618d87f98902731fbf : SUCCESS in 24m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b045c4912e0c468694de6b0623779ce1 : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2096f8fbd224e51b2d6ac2807cb650c : SUCCESS in 10m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/18ebc85a3d654e84b0d30c3244954680 : SUCCESS in 45m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a37c7a98e8c4339bec8469d2f55e7f8 : SUCCESS in 1h 30m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e5fc98021102486fb15dd1fba77ac32e : SUCCESS in 1h 04m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6c95c9cb11944dd826e38c0226a67bd : SUCCESS in 52m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/11b175e3694f4aa1998af8cec29ac052 : SUCCESS in 1h 00m 28s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4ecd9c29d7b54314bbdbbd9f361114cc : SUCCESS in 1h 11m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c63ac6c219d94ef3810953fdc04d3c03 : SUCCESS in 1h 13m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4313a6bef46143c7affd75472483c40c : SUCCESS in 54m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/882641c9e9c14f47947bdd8c7ba4f9ef : SUCCESS in 1h 06m 57s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec37dcea205c435991ff1f5f587392a4 : SUCCESS in 33m 40s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae806527e3a246198c9ea5e7a5d2f18a : SUCCESS in 38m 58s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a9d96f1b86e4e969d58189c0218b783 : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":6},{"id":"a9c162d0fade7e4262e41210b183109dec3f1f96","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-30 00:30:03.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/816612/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/816612/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"9b2bd90ad818eaa1f2a7ed6944cfa63993be7f10","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-11-30 01:08:41.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"83801e3ed3c30013abfc32dfa2b1b4b4160568d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 09:03:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2dd0979b36a1b2d1fbdd1c526e4f4eb7a7cd6725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 11:31:28.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e56e5c4e80564d49b3ef882bb79c16ca : SUCCESS in 44m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/11eeadc559c345b38d583adbda61e7f3 : FAILURE in 2h 02m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ffef0a821dc447c09b95d5dad06d6520 : SUCCESS in 2h 13m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ee687446f6f24d6485c424dcd428a7d9 : SUCCESS in 2h 15m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/79baa3452d1f4b5b9666d0f55cdc2ab6 : SUCCESS in 1h 31m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ce33d127c5a4cdf95a211df546f5f2d : SUCCESS in 37m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea3f0008bd84a7fbc273232478e6a9d : SUCCESS in 14m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a86fc8e84fd04bc28e96ca9c638c4f50 : SUCCESS in 24m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/437d9c64d514457f9e2986970c283221 : SUCCESS in 27m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f7eff7d23b245b9903ae8fc5286f1f5 : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a259e12bed541e4b5a5a07a2248e5a1 : FAILURE in 45m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6cce1b97d23a427faafe9e62006e3940 : SUCCESS in 1h 27m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e7626218c9d949d187e9476da383eed4 : SUCCESS in 57m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/02429d164952427496ee027ba5ec2081 : SUCCESS in 46m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d5f98d61ffb4bbbb95059118cbb84d5 : SUCCESS in 58m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d572615f48a46619d823006984a84fd : SUCCESS in 23m 50s","accounts_in_message":[],"_revision_number":6},{"id":"6a01eff0a61540a2a5b336ba1773970eb39b525f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-11-30 13:23:44.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"720e5129cb12676497215e95289e9a2ac2cbf367","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-30 13:37:59.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/816612/6 : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":6},{"id":"71413bf0780d2d81a5d86be0695905f16956227e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 14:35:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8402d7359084d27a4bab1be8f7c87d6 : SUCCESS in 40m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a14ec8d6e2854a40be850ea992db7960 : SUCCESS in 39m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2836f4a16f33c855c5c6aff73613b9f6d9128fed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-30 16:47:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"acc754ae3405f62c2413e3271195b8daa0b98abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 16:56:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa1471eb3a294ef8aebf1db8ce1a706e : SUCCESS in 52m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7c442b40d1244420bd079cc3767d921d : SUCCESS in 2h 04m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a22cd9b44aea41f28211fe83809be9d8 : SUCCESS in 2h 04m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0cf04b6aff5a4650b0b950f9e9040a4d : SUCCESS in 1h 27m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5d4353e58e5947a3ad2c59856337c895 : SUCCESS in 2h 12m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d56ab28e9ea46adaabdaa693e62fd3a : SUCCESS in 40m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab851c638ff24761aa4e94e9bbc898e4 : SUCCESS in 32m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f656907014b44494a555a791d66b481b : SUCCESS in 29m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4344d9b62ba14c338857f7e7ea803581 : SUCCESS in 14m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a9cf7433d6c4345a5b9b246b9fb6fe8 : SUCCESS in 29m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3bfc06d5dbc4133b33fc0bbcaf90bc7 : SUCCESS in 25m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/403b4c77a7934241b3926482a13f4332 : SUCCESS in 13m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d47aca6ec57f4595b1645df21e8ab2eb : SUCCESS in 54m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2289b7213dfa41e4b4c1995223f566bb : SUCCESS in 1h 24m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dba7f546a4f14a5f8d1dc4f7230baaf1 : SUCCESS in 1h 07m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f4d17eb3d3440f88523243fe13362bc : SUCCESS in 57m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ca0a7d7a710f4f36af5dd2124ae82288 : SUCCESS in 54m 50s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ec48996b36454fc0a96028b85f685861 : SUCCESS in 1h 03m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7cbc438228b948cc9f6ac346f0c96076 : SUCCESS in 1h 16m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5153ff669b92478797d5dd7527d585ea : SUCCESS in 54m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8edf32f3633a415ebf5972d2f0fe1d69 : SUCCESS in 1h 07m 48s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d95086d5a574cbcbe9c94b71ebe3734 : SUCCESS in 29m 20s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb293398f6994f438df8509d8896acbc : SUCCESS in 31m 53s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03f845039b534180b6bcd64c265ac410 : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":6},{"id":"14c4393d3cc02ac48c939317d228229fc0742aa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 16:57:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2ca6eb57dfac6cfe2c85b5123681fda1da1e68a7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-30 17:01:51.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/816612/6 : FAILURE in 13m 45s","accounts_in_message":[],"_revision_number":6},{"id":"1167b22fefe3268c360118280cd6ce553e8f8772","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:53:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e697634e1f54cf4ab7bff039b303879 : SUCCESS in 38m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fe7eba3c2ed44b9afffcb1ad3a703c2 : TIMED_OUT in 1h 01m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cbfaac27c2c44f4d2b2a67c34af2aabdb3499bd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 19:44:53.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/52bb2be6d5a5480c8a82f08631396721 : SUCCESS in 56m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d77782b7a20a489a9d639144984482fb : SUCCESS in 2h 41m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/88412abe0b41496ba599896725b89031 : SUCCESS in 2h 17m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/af1184838ffc4b7cb99c59116393fbb6 : SUCCESS in 2h 36m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f2cc0e675b274ced84089757df995353 : SUCCESS in 2h 29m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f8cf17e102b4a4db23521d5d373d76b : SUCCESS in 34m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3affa96c4914c87be3108b22939998f : SUCCESS in 16m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/143402aafa804e4dbf5cf7b4e63ebdaa : SUCCESS in 27m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5bdf6ae816f412c916f82b25fc8c6aa : SUCCESS in 34m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e8258845494e008a9008b6f7c357b3 : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96fb484d6fc543ddbbfc96d5f61eb7e9 : SUCCESS in 51m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/94df460166d24a0f892c70228b0d229c : SUCCESS in 1h 34m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0b06106aa2914bc181afe008cd1ba0ac : SUCCESS in 1h 06m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dfba6a25f1bf41c5aac7bbe01dab1401 : SUCCESS in 52m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ce9cb1045fd49fd980777e905e212fd : SUCCESS in 58m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0133959092b8482ca450b8bb79c6f77e : SUCCESS in 25m 22s","accounts_in_message":[],"_revision_number":6},{"id":"8ffa85dae5dfd3aa5510d0c895f964d15762a699","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 19:44:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1b9c218b78d09fef8fe698a1ca11ddf05d4f867d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 19:48:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc301a0a731c48fe992a82800d0e0051 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":6},{"id":"7b5cbad0cb65a469913648945c3d54a15bb1f455","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-01 09:44:35.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/12/816612/6/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/12/816612/6/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c95ccf0a43f8f40ed6988d88fae01b225b68b43e","revisions":{"f1ad05576192c53effabc1f9244fe6921b84d272":{"kind":"REWORK","_number":1,"created":"2021-11-04 03:04:06.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/12/816612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/816612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/816612/1"}}},"commit":{"parents":[{"commit":"e012eb2008cca4dfd6c42c539abae50d062c38ec","subject":"ovn: update ACL actions on stateful field change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e012eb2008cca4dfd6c42c539abae50d062c38ec"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 03:02:03.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 03:03:02.000000000","tz":0},"subject":"ovn: enable stateless-security-group api","message":"ovn: enable stateless-security-group api\n\nThe feature is long implemented but the list of extensions was not\nupdated accordingly. Because of missing tempest test cases for the\nfield, it was not caught before.\n\nChange-Id: I85f7749e7e6186b9e3a7534bce6c306b18702333\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1ad05576192c53effabc1f9244fe6921b84d272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1ad05576192c53effabc1f9244fe6921b84d272"}]},"branch":"refs/heads/master"},"384cf9c9146eb7526067af98e8118500635e253d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-04 03:08:37.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/12/816612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/816612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/816612/2"}}},"commit":{"parents":[{"commit":"e012eb2008cca4dfd6c42c539abae50d062c38ec","subject":"ovn: update ACL actions on stateful field change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e012eb2008cca4dfd6c42c539abae50d062c38ec"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 03:02:03.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 03:07:52.000000000","tz":0},"subject":"ovn: enable stateless-security-group api","message":"ovn: enable stateless-security-group api\n\nThe feature is long implemented but the list of extensions was not\nupdated accordingly. Because of missing tempest test cases for the\nfield, it was not caught before.\n\nCloses-Bug: #1949451\nChange-Id: I85f7749e7e6186b9e3a7534bce6c306b18702333\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/384cf9c9146eb7526067af98e8118500635e253d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/384cf9c9146eb7526067af98e8118500635e253d"}]},"branch":"refs/heads/master"},"14edab078f47017391e7da5617b9fc8a98081cde":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-04 03:17:54.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/12/816612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/816612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/816612/3"}}},"commit":{"parents":[{"commit":"3ee9f7a68aa30006477d00d86b3b1c720d63eab1","subject":"ovn: update ACL actions on stateful field change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ee9f7a68aa30006477d00d86b3b1c720d63eab1"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 03:02:03.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 03:17:44.000000000","tz":0},"subject":"ovn: enable stateless-security-group api","message":"ovn: enable stateless-security-group api\n\nThe feature is long implemented but the list of extensions was not\nupdated accordingly. Because of missing tempest test cases for the\nfield, it was not caught before.\n\nCloses-Bug: #1949451\nChange-Id: I85f7749e7e6186b9e3a7534bce6c306b18702333\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14edab078f47017391e7da5617b9fc8a98081cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14edab078f47017391e7da5617b9fc8a98081cde"}]},"branch":"refs/heads/master"},"60598f25c4975b2b142426eb56207cfc6136efcf":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-11-04 13:40:36.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/12/816612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/816612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/816612/4"}}},"commit":{"parents":[{"commit":"c251b5e1d4f4b823429a0a823ee3d662c394240a","subject":"ovn: update ACL actions on stateful field change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c251b5e1d4f4b823429a0a823ee3d662c394240a"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 03:02:03.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 13:40:29.000000000","tz":0},"subject":"ovn: enable stateless-security-group api","message":"ovn: enable stateless-security-group api\n\nThe feature is long implemented but the list of extensions was not\nupdated accordingly. Because of missing tempest test cases for the\nfield, it was not caught before.\n\nCloses-Bug: #1949451\nChange-Id: I85f7749e7e6186b9e3a7534bce6c306b18702333\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60598f25c4975b2b142426eb56207cfc6136efcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60598f25c4975b2b142426eb56207cfc6136efcf"}]},"branch":"refs/heads/master"},"5ee86273b33d3c7413fa1ebf0fffececcd2e66e3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-11-04 16:46:12.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/12/816612/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/816612/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/816612/5"}}},"commit":{"parents":[{"commit":"499da731cb53adf6eed189eaec28dd95722cfea5","subject":"ovn: update ACL actions on stateful field change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/499da731cb53adf6eed189eaec28dd95722cfea5"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 03:02:03.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 16:46:03.000000000","tz":0},"subject":"ovn: enable stateless-security-group api","message":"ovn: enable stateless-security-group api\n\nThe feature is long implemented but the list of extensions was not\nupdated accordingly. Because of missing tempest test cases for the\nfield, it was not caught before.\n\nCloses-Bug: #1949451\nChange-Id: I85f7749e7e6186b9e3a7534bce6c306b18702333\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee86273b33d3c7413fa1ebf0fffececcd2e66e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee86273b33d3c7413fa1ebf0fffececcd2e66e3"}]},"branch":"refs/heads/master"},"c95ccf0a43f8f40ed6988d88fae01b225b68b43e":{"kind":"NO_CHANGE","_number":6,"created":"2021-11-29 18:35:35.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/12/816612/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/816612/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/816612/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/816612/6"}}},"commit":{"parents":[{"commit":"278af8512374dcc75bb4884bd364ac20f2bb9e69","subject":"ovn: update ACL actions on stateful field change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/278af8512374dcc75bb4884bd364ac20f2bb9e69"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-04 03:02:03.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2021-11-29 18:35:35.000000000","tz":0},"subject":"ovn: enable stateless-security-group api","message":"ovn: enable stateless-security-group api\n\nThe feature is long implemented but the list of extensions was not\nupdated accordingly. Because of missing tempest test cases for the\nfield, it was not caught before.\n\nCloses-Bug: #1949451\nChange-Id: I85f7749e7e6186b9e3a7534bce6c306b18702333\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c95ccf0a43f8f40ed6988d88fae01b225b68b43e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c95ccf0a43f8f40ed6988d88fae01b225b68b43e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
