)]}'
{"id":"openstack%2Fneutron~817006","triplet_id":"openstack%2Fneutron~master~I215923f6d1e2bcf7f9091289f42c9e792981986a","project":"openstack/neutron","branch":"master","topic":"improve-neutron-ci","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-09 10:47:16.000000000","reason":"Slawek Kaplonski replied on the change"}},"hashtags":[],"change_id":"I215923f6d1e2bcf7f9091289f42c9e792981986a","subject":"DNM FIP - only testing why neutron LB fullstack tests are failing","status":"ABANDONED","created":"2021-11-08 13:50:22.000000000","updated":"2021-11-15 09:17:42.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"66553e3e5238e2688f1f52bf9b67350b4d044e05","_number":817006,"virtual_id_number":817006,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-11-09 05:46:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-08 17:07:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-08 14:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-08 15:47:22.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-08 16:52:28.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-08 17:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-09 05:46:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"237368671474d429d4a9f8f386f95d42b686ff43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-08 13:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e943a3f69442cbecf349bd093debc34c1760d4c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-08 13:50:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d32a8fb059213ea1ece49b94ad653bdc4dfca1d1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-08 14:27:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4db02e5de257df55a4eb52fbdf0eba85f76a80b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-08 14:28:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc38e55ff75ac42928e634a1fcb4073a2c5bd6a5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 14:29:16.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"3d477d254593c51a1c5dda512ad657b8996ab1f2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-08 15:14:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78ea5370e46d8a09e21a208dae2bb6cc4defe9fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 15:31:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d3367acd14e4a64a3d5484072c61a9c : TIMED_OUT in 1h 01m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/023cccc7fa1a46a7a0edd71e1c68fbc3 : TIMED_OUT in 1h 01m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9075751815f1817bbb11ed41237e9191cb8ccd4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-08 15:47:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/817006/1 : SUCCESS in 1h 55m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9be2fe876fe3d5648322c0cf127bcb9c55e59170","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-08 16:52:28.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/06/817006/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/06/817006/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7a40b8654160af3377335d6bbf3449b520de4e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 17:07:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d4db0942f85a42d9b892068ddcf5b55c : SUCCESS in 50m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/94bebb265e6e4d4a986f0de90a00c314 : FAILURE in 2h 32m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d6ea1d240c64c479ac8e300fb911589 : SUCCESS in 50m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/321e3b6934124721ad755a1f630ba679 : SUCCESS in 59m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87eb8771c7bf4a138f81abbbe1ad258a : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb331c1a47e54d52a2e6465a032ed8c2 : SUCCESS in 40m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13d2a602606647cb91271d64875ba33a : SUCCESS in 41m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b907b4137024c41b70a9e15da0887a8 : SUCCESS in 35m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc74561dcf34746ba6ccd1409000d2b : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a23db06bc9547a7b0b89cedfb32575d : SUCCESS in 49m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2241f87a0d1c4b6ca0cd8c972260868e : SUCCESS in 1h 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e0ebb3996f8b01ff823a4503b0acb68f19c02a35","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 17:20:04.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- neutron-functional-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/c1d15f75bb5148d8b85a9bff8a46a271 : SUCCESS in 54m 35s\n- neutron-fullstack-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/c5e186320a6e45ac9ec0e999f99d5624 : FAILURE in 2h 02m 19s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1aee8c057677422bba833647a7b9d7d4 : SUCCESS in 55m 50s\n- neutron-fullstack-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e6314aea67c945f5a7e1d2f1c1b391be : SUCCESS in 1h 19m 53s\n- neutron-functional-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c60b7bd64dc74671b49dbeda35afb67c : SUCCESS in 41m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6a502a1fd18088afe63c3edac9b929e9d9733b5c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-09 05:46:52.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/06/817006/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"622071f6da5647d0d4d59be838f49aa3219dc749","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-09 10:47:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd3d34568d588ea49ac1249b864f59218f3133ee","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 13:09:50.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- neutron-functional-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/0021f3f71e17437da91d3ebd47e3ea49 : FAILURE in 59m 56s\n- neutron-fullstack-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/f068f881e7844f8dab54f03753aef65d : TIMED_OUT in 2h 17m 21s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0eac2b7648db4bd5906e68a3d0f99b97 : SUCCESS in 59m 25s\n- neutron-fullstack-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4bd948d69f2f4c7f9651b2d6d1408489 : SUCCESS in 1h 58m 57s\n- neutron-functional-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f02dbdfb6f174f98be0474f171d89362 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":1},{"id":"66553e3e5238e2688f1f52bf9b67350b4d044e05","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-15 09:17:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21baef48c9261c6f12b32a58972e6e0aa18655e7","revisions":{"21baef48c9261c6f12b32a58972e6e0aa18655e7":{"kind":"REWORK","_number":1,"created":"2021-11-08 13:50:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/06/817006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/817006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/817006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/817006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/817006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/817006/1"}}},"commit":{"parents":[{"commit":"95839aef17ce20fa1afded66fb83d063c483c59b","subject":"Add functional and fullstack jobs with FIPS enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95839aef17ce20fa1afded66fb83d063c483c59b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-08 13:49:44.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-08 13:49:44.000000000","tz":60},"subject":"DNM FIP - only testing why neutron LB fullstack tests are failing","message":"DNM FIP - only testing why neutron LB fullstack tests are failing\n\nChange-Id: I215923f6d1e2bcf7f9091289f42c9e792981986a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21baef48c9261c6f12b32a58972e6e0aa18655e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21baef48c9261c6f12b32a58972e6e0aa18655e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
